)]}'
{"id":"openstack%2Foctavia~652679","triplet_id":"openstack%2Foctavia~master~Ie8bbcbe073644137b8eb04c42d9457f507513007","project":"openstack/octavia","branch":"master","topic":"bionic-amp","hashtags":[],"change_id":"Ie8bbcbe073644137b8eb04c42d9457f507513007","subject":"Make Ubuntu bionic the default for amphora images","status":"MERGED","created":"2019-04-15 14:50:15.000000000","updated":"2019-07-27 03:46:17.000000000","submitted":"2019-07-27 03:46:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652679-1564199177540-3342d8ab","meta_rev_id":"e475426e4eaf3bbe5dcbd49761049802e6e63571","_number":652679,"virtual_id_number":652679,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2019-07-26 23:33:06.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-07-27 03:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-07-26 21:36:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":2,"date":"2019-07-23 21:55:03.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-25 16:20:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-07-26 21:36:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 13:12:16.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-07-23 21:55:03.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-07-25 16:20:19.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-07-26 21:36:00.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-07-27 03:46:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"846cf1ce7fe16d8ac60b591145cb9db4fef57715","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-15 14:50:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80fcf2ee21367a3587a33b8f973fec179b18a70b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 17:22:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/79/652679/1/check/openstack-tox-cover/3149d08/cover/ : SUCCESS in 9m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/79/652679/1/check/openstack-tox-lower-constraints/56bb391/ : SUCCESS in 8m 47s\n- openstack-tox-pep8 http://logs.openstack.org/79/652679/1/check/openstack-tox-pep8/1faa54c/ : SUCCESS in 7m 23s\n- openstack-tox-py27 http://logs.openstack.org/79/652679/1/check/openstack-tox-py27/f1f4e77/ : SUCCESS in 7m 05s\n- openstack-tox-py35 http://logs.openstack.org/79/652679/1/check/openstack-tox-py35/1678dc8/ : SUCCESS in 6m 08s\n- openstack-tox-py36 http://logs.openstack.org/79/652679/1/check/openstack-tox-py36/599725e/ : SUCCESS in 6m 33s\n- openstack-tox-docs http://logs.openstack.org/79/652679/1/check/openstack-tox-docs/873134c/html/ : SUCCESS in 7m 06s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/79/652679/1/check/neutron-lbaasv2-dsvm-api/4003071/ : SUCCESS in 1h 22m 31s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/79/652679/1/check/neutron-lbaasv2-dsvm-py3x-api/63bbf1d/ : SUCCESS in 1h 48m 28s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/79/652679/1/check/neutron-lbaasv2-dsvm-scenario/cc94069/ : SUCCESS in 1h 25m 08s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/79/652679/1/check/neutron-lbaasv2-dsvm-py3x-scenario/b4e1dfe/ : SUCCESS in 1h 24m 25s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/79/652679/1/check/openstack-tox-functional/7143960/ : SUCCESS in 7m 23s\n- openstack-tox-functional-py35 http://logs.openstack.org/79/652679/1/check/openstack-tox-functional-py35/a191fb4/ : SUCCESS in 8m 37s\n- openstack-tox-functional-py36 http://logs.openstack.org/79/652679/1/check/openstack-tox-functional-py36/7b8b56e/ : SUCCESS in 7m 55s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/79/652679/1/check/octavia-v1-dsvm-scenario/ff3195e/ : FAILURE in 56m 44s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/79/652679/1/check/octavia-v1-dsvm-py3x-scenario/dac1b43/ : SUCCESS in 59m 59s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/79/652679/1/check/octavia-v2-dsvm-noop-api/69211e7/ : SUCCESS in 46m 51s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/79/652679/1/check/octavia-v2-dsvm-noop-py2-api/aa2eb3d/ : SUCCESS in 43m 35s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/79/652679/1/check/octavia-v2-dsvm-scenario/fa9cde7/ : SUCCESS in 1h 37m 37s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/79/652679/1/check/octavia-v2-dsvm-py2-scenario/b53aa1e/ : SUCCESS in 1h 48m 32s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/79/652679/1/check/octavia-v2-dsvm-py2-scenario-centos-7/69697c4/ : SUCCESS in 2h 05m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/79/652679/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/c09f21d/ : SUCCESS in 1h 43m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/79/652679/1/check/octavia-v2-dsvm-scenario-two-node/133a08e/ : SUCCESS in 2h 09m 13s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/79/652679/1/check/octavia-v2-dsvm-py2-scenario-two-node/1885fe9/ : FAILURE in 1h 48m 04s (non-voting)\n- octavia-grenade http://logs.openstack.org/79/652679/1/check/octavia-grenade/ba65225/ : SUCCESS in 1h 13m 14s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/79/652679/1/check/octavia-v2-dsvm-tls-barbican/647bbcd/ : SUCCESS in 56m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f8ef5f24cda90ab5a1b43eb19624015b0fe30628","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-22 13:05:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"38e44c7964ea96c4370a0cdc3cc3293a47ea980c","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-04-22 13:12:16.000000000","message":"Patch Set 1:\n\nfailure does not seem related to the change: http://logs.openstack.org/79/652679/1/check/octavia-v1-dsvm-scenario/ff3195e/logs/testr_results.html.gz","accounts_in_message":[],"_revision_number":1},{"id":"7688a0dded7e7aa46692ca1fd7097d6a2f0fc2aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 15:27:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/79/652679/1/check/openstack-tox-cover/ac78e8d/cover/ : SUCCESS in 9m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/79/652679/1/check/openstack-tox-lower-constraints/607f18e/ : SUCCESS in 7m 11s\n- openstack-tox-pep8 http://logs.openstack.org/79/652679/1/check/openstack-tox-pep8/c8eeaaf/ : SUCCESS in 6m 48s\n- openstack-tox-py27 http://logs.openstack.org/79/652679/1/check/openstack-tox-py27/3ff920e/ : SUCCESS in 5m 53s\n- openstack-tox-py35 http://logs.openstack.org/79/652679/1/check/openstack-tox-py35/a5c6380/ : SUCCESS in 5m 47s\n- openstack-tox-py36 http://logs.openstack.org/79/652679/1/check/openstack-tox-py36/25d79c9/ : SUCCESS in 5m 23s\n- openstack-tox-docs http://logs.openstack.org/79/652679/1/check/openstack-tox-docs/9a79fe5/html/ : SUCCESS in 6m 36s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/79/652679/1/check/neutron-lbaasv2-dsvm-api/751b481/ : SUCCESS in 1h 19m 12s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/79/652679/1/check/neutron-lbaasv2-dsvm-py3x-api/c41c408/ : SUCCESS in 1h 14m 45s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/79/652679/1/check/neutron-lbaasv2-dsvm-scenario/1565765/ : SUCCESS in 1h 37m 05s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/79/652679/1/check/neutron-lbaasv2-dsvm-py3x-scenario/56ba424/ : SUCCESS in 1h 00m 46s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/79/652679/1/check/openstack-tox-functional/b537b46/ : SUCCESS in 9m 53s\n- openstack-tox-functional-py35 http://logs.openstack.org/79/652679/1/check/openstack-tox-functional-py35/bdf16ac/ : SUCCESS in 10m 37s\n- openstack-tox-functional-py36 http://logs.openstack.org/79/652679/1/check/openstack-tox-functional-py36/46c49a5/ : SUCCESS in 7m 15s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/79/652679/1/check/octavia-v1-dsvm-scenario/a62f668/ : SUCCESS in 1h 57m 11s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/79/652679/1/check/octavia-v1-dsvm-py3x-scenario/744f8b6/ : SUCCESS in 57m 00s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/79/652679/1/check/octavia-v2-dsvm-noop-api/9828ddf/ : SUCCESS in 46m 51s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/79/652679/1/check/octavia-v2-dsvm-noop-py2-api/6adcbdb/ : SUCCESS in 1h 03m 43s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/79/652679/1/check/octavia-v2-dsvm-scenario/fd42651/ : SUCCESS in 1h 52m 40s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/79/652679/1/check/octavia-v2-dsvm-py2-scenario/ac8e8c9/ : SUCCESS in 1h 52m 44s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/79/652679/1/check/octavia-v2-dsvm-py2-scenario-centos-7/878f895/ : FAILURE in 2h 08m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/79/652679/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/0106b36/ : SUCCESS in 1h 36m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/79/652679/1/check/octavia-v2-dsvm-scenario-two-node/022af7b/ : SUCCESS in 1h 49m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/79/652679/1/check/octavia-v2-dsvm-py2-scenario-two-node/717ab42/ : SUCCESS in 1h 30m 35s (non-voting)\n- octavia-grenade http://logs.openstack.org/79/652679/1/check/octavia-grenade/9ba7cfa/ : SUCCESS in 1h 22m 27s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/79/652679/1/check/octavia-v2-dsvm-tls-barbican/79d07d8/ : SUCCESS in 1h 03m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c42ca3470b04bd2a16c8713fd3d008bbcbdce9de","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-18 22:54:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fb8af0671d18cec61e442ab67f67abe319905c44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 01:23:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/79/652679/2/check/openstack-tox-cover/efb14ce/cover/ : SUCCESS in 7m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/79/652679/2/check/openstack-tox-lower-constraints/eded9f6/ : SUCCESS in 7m 49s\n- openstack-tox-pep8 http://logs.openstack.org/79/652679/2/check/openstack-tox-pep8/4614717/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/79/652679/2/check/openstack-tox-py27/eb0f213/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/79/652679/2/check/openstack-tox-py36/a75aef4/ : SUCCESS in 4m 05s\n- openstack-tox-py37 http://logs.openstack.org/79/652679/2/check/openstack-tox-py37/d425aac/ : SUCCESS in 5m 05s\n- openstack-tox-docs http://logs.openstack.org/79/652679/2/check/openstack-tox-docs/f97f059/html/ : SUCCESS in 5m 12s\n- openstack-tox-functional http://logs.openstack.org/79/652679/2/check/openstack-tox-functional/2220cd1/ : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 http://logs.openstack.org/79/652679/2/check/openstack-tox-functional-py36/dd4451e/ : SUCCESS in 5m 39s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/79/652679/2/check/octavia-v2-dsvm-noop-api/f86721b/ : SUCCESS in 44m 09s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/79/652679/2/check/octavia-v2-dsvm-noop-py2-api/ce28ec5/ : SUCCESS in 47m 19s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/79/652679/2/check/octavia-v2-dsvm-scenario/af63b22/ : SUCCESS in 1h 26m 22s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/79/652679/2/check/octavia-v2-dsvm-py2-scenario/5f8a13a/ : SUCCESS in 1h 47m 07s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/79/652679/2/check/octavia-v2-dsvm-py2-scenario-centos-7/6b856d4/ : FAILURE in 1h 56m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/79/652679/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/54e4a16/ : SUCCESS in 1h 40m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/79/652679/2/check/octavia-v2-dsvm-scenario-two-node/52efc06/ : SUCCESS in 2h 22m 48s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/79/652679/2/check/octavia-v2-dsvm-py2-scenario-two-node/4426881/ : FAILURE in 1h 16m 08s (non-voting)\n- octavia-grenade http://logs.openstack.org/79/652679/2/check/octavia-grenade/5525e1c/ : SUCCESS in 1h 21m 57s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/79/652679/2/check/octavia-v2-dsvm-tls-barbican/0e025f5/ : SUCCESS in 53m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/79/652679/2/check/octavia-v2-dsvm-scenario-amphora-v2/42d178c/ : SUCCESS in 1h 35m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"32b477f4bd34586f18df61d6925a12eeb3875955","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-22 20:27:40.000000000","message":"Patch Set 2: Code-Review-1\n\nThe README file should be updated too.","accounts_in_message":[],"_revision_number":2},{"id":"0a283b3b849c3b5750c1a00df60bf6e9ae5ed145","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-23 21:52:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5fa7e40b395c805b4d3137b7268a3d8c8bf656cf","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-23 21:55:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"03b60a691b50fbb41ef2cba52bc5c902208b6d05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 00:14:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/79/652679/3/check/openstack-tox-cover/cbce84c/cover/ : SUCCESS in 8m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/79/652679/3/check/openstack-tox-lower-constraints/9525903/ : SUCCESS in 7m 04s\n- openstack-tox-pep8 http://logs.openstack.org/79/652679/3/check/openstack-tox-pep8/b754ec7/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/79/652679/3/check/openstack-tox-py27/0938bcf/ : SUCCESS in 4m 31s\n- openstack-tox-py36 http://logs.openstack.org/79/652679/3/check/openstack-tox-py36/4495f75/ : SUCCESS in 4m 24s\n- openstack-tox-py37 http://logs.openstack.org/79/652679/3/check/openstack-tox-py37/5dd39f1/ : SUCCESS in 6m 28s\n- openstack-tox-docs http://logs.openstack.org/79/652679/3/check/openstack-tox-docs/b63929c/html/ : SUCCESS in 5m 25s\n- openstack-tox-functional http://logs.openstack.org/79/652679/3/check/openstack-tox-functional/26814f1/ : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 http://logs.openstack.org/79/652679/3/check/openstack-tox-functional-py36/5f17585/ : SUCCESS in 5m 40s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-noop-api/b4f1f1a/ : SUCCESS in 42m 12s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-noop-py2-api/b1f6d89/ : SUCCESS in 50m 20s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario/e44e86c/ : SUCCESS in 1h 40m 31s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-py2-scenario/2af02cc/ : FAILURE in 1h 34m 29s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-py2-scenario-centos-7/6689bb2/ : FAILURE in 1h 55m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/c703b20/ : SUCCESS in 1h 49m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario-two-node/5b3f320/ : FAILURE in 1h 54m 15s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-py2-scenario-two-node/d58a5c7/ : SUCCESS in 1h 51m 49s (non-voting)\n- octavia-grenade http://logs.openstack.org/79/652679/3/check/octavia-grenade/a18952b/ : SUCCESS in 1h 32m 00s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-tls-barbican/6e91d4e/ : SUCCESS in 1h 03m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario-amphora-v2/58a84c5/ : SUCCESS in 1h 29m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2eff460809ae5661dbf7cc6699427be11d28582c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-07-25 16:20:19.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"77e07f099a335fb464bb399507d004bc446e8de5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-25 16:24:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ebd13e38cfcbeed198973d6b414787c6f4c63f68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 18:49:52.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/79/652679/3/check/openstack-tox-cover/12f126a/cover/ : SUCCESS in 8m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/79/652679/3/check/openstack-tox-lower-constraints/cf7a669/ : SUCCESS in 7m 39s\n- openstack-tox-pep8 http://logs.openstack.org/79/652679/3/check/openstack-tox-pep8/cba10e6/ : SUCCESS in 7m 26s\n- openstack-tox-py27 http://logs.openstack.org/79/652679/3/check/openstack-tox-py27/12aeaec/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/79/652679/3/check/openstack-tox-py36/f5f2150/ : SUCCESS in 5m 08s\n- openstack-tox-py37 http://logs.openstack.org/79/652679/3/check/openstack-tox-py37/b9f0e8e/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/79/652679/3/check/openstack-tox-docs/43f0e5b/html/ : SUCCESS in 8m 20s\n- openstack-tox-functional http://logs.openstack.org/79/652679/3/check/openstack-tox-functional/3e42e36/ : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 http://logs.openstack.org/79/652679/3/check/openstack-tox-functional-py36/02457d4/ : SUCCESS in 6m 08s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-noop-api/693c11b/ : SUCCESS in 50m 07s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-noop-py2-api/837d5e6/ : SUCCESS in 52m 02s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario/bf7d1c7/ : SUCCESS in 2h 04m 40s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-py2-scenario/1eddcac/ : FAILURE in 30m 28s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-py2-scenario-centos-7/f205368/ : FAILURE in 1h 51m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/d828e0b/ : SUCCESS in 1h 52m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario-two-node/06d95ba/ : SUCCESS in 1h 28m 46s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-py2-scenario-two-node/46ad358/ : SUCCESS in 2h 16m 51s (non-voting)\n- octavia-grenade http://logs.openstack.org/79/652679/3/check/octavia-grenade/127f45d/ : SUCCESS in 1h 30m 57s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-tls-barbican/b15ab53/ : SUCCESS in 1h 00m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario-amphora-v2/d391987/ : SUCCESS in 1h 43m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1eb4235cfa49608dc64c2b79b9f9e9749d704a22","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-25 19:20:30.000000000","message":"Patch Set 3:\n\nrecheck\n\nfortnebula mirror issues","accounts_in_message":[],"_revision_number":3},{"id":"f5d09bfbb97b4fbec1788742137ac08f2f9218bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 21:28:14.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/79/652679/3/check/openstack-tox-cover/7274523/cover/ : SUCCESS in 9m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/79/652679/3/check/openstack-tox-lower-constraints/37b2601/ : SUCCESS in 6m 46s\n- openstack-tox-pep8 http://logs.openstack.org/79/652679/3/check/openstack-tox-pep8/f635844/ : SUCCESS in 5m 07s\n- openstack-tox-py27 http://logs.openstack.org/79/652679/3/check/openstack-tox-py27/a4eeb82/ : SUCCESS in 4m 38s\n- openstack-tox-py36 http://logs.openstack.org/79/652679/3/check/openstack-tox-py36/d590993/ : SUCCESS in 5m 27s\n- openstack-tox-py37 http://logs.openstack.org/79/652679/3/check/openstack-tox-py37/ef07e29/ : SUCCESS in 6m 03s\n- openstack-tox-docs http://logs.openstack.org/79/652679/3/check/openstack-tox-docs/af1c3f8/html/ : SUCCESS in 5m 39s\n- openstack-tox-functional http://logs.openstack.org/79/652679/3/check/openstack-tox-functional/0de2fee/ : SUCCESS in 5m 53s\n- openstack-tox-functional-py36 http://logs.openstack.org/79/652679/3/check/openstack-tox-functional-py36/37d9366/ : SUCCESS in 5m 25s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-noop-api/942d9f4/ : SUCCESS in 42m 23s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-noop-py2-api/6a3d4e4/ : SUCCESS in 45m 01s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario/85b55e1/ : SUCCESS in 1h 35m 58s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-py2-scenario/8e5d596/ : FAILURE in 1h 49m 17s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-py2-scenario-centos-7/f4bb69a/ : SUCCESS in 2h 04m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/e849f20/ : SUCCESS in 1h 44m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario-two-node/74b481c/ : SUCCESS in 2h 03m 47s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-py2-scenario-two-node/de25669/ : FAILURE in 1h 54m 50s (non-voting)\n- octavia-grenade http://logs.openstack.org/79/652679/3/check/octavia-grenade/954984f/ : FAILURE in 1h 06m 43s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-tls-barbican/ec568d7/ : SUCCESS in 54m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario-amphora-v2/086256a/ : SUCCESS in 1h 40m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"472cda396d59f1c0c70c8caad27dfabc8d5b6089","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-26 21:00:06.000000000","message":"Patch Set 3:\n\nrecheck\n\ntimeouts","accounts_in_message":[],"_revision_number":3},{"id":"63b94e69b87e99168a0154f8dc3a70a2d10701d1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-26 21:36:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3e89e7615338359fd88f7052a5cd81d2cdc6eff7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 23:27:00.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/79/652679/3/check/openstack-tox-cover/c9d909d/cover/ : SUCCESS in 6m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/79/652679/3/check/openstack-tox-lower-constraints/70a5363/ : SUCCESS in 7m 56s\n- openstack-tox-pep8 http://logs.openstack.org/79/652679/3/check/openstack-tox-pep8/03d4c4f/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/79/652679/3/check/openstack-tox-py27/e39dfab/ : SUCCESS in 4m 03s\n- openstack-tox-py36 http://logs.openstack.org/79/652679/3/check/openstack-tox-py36/30f9c97/ : SUCCESS in 4m 21s\n- openstack-tox-py37 http://logs.openstack.org/79/652679/3/check/openstack-tox-py37/1f22b63/ : SUCCESS in 4m 07s\n- openstack-tox-docs http://logs.openstack.org/79/652679/3/check/openstack-tox-docs/9a8ff97/html/ : SUCCESS in 4m 57s\n- openstack-tox-functional http://logs.openstack.org/79/652679/3/check/openstack-tox-functional/a68e8c3/ : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 http://logs.openstack.org/79/652679/3/check/openstack-tox-functional-py36/633c89e/ : SUCCESS in 6m 03s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-noop-api/bb7bfea/ : SUCCESS in 51m 12s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-noop-py2-api/b2cb682/ : SUCCESS in 49m 19s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario/11f78ae/ : SUCCESS in 1h 46m 36s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-py2-scenario/64890cc/ : FAILURE in 1h 42m 05s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-py2-scenario-centos-7/d54ea57/ : SUCCESS in 2h 13m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/3d6012a/ : SUCCESS in 1h 41m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario-two-node/be4acb3/ : FAILURE in 1h 29m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-py2-scenario-two-node/1982abc/ : SUCCESS in 1h 56m 16s (non-voting)\n- octavia-grenade http://logs.openstack.org/79/652679/3/check/octavia-grenade/a24bc46/ : SUCCESS in 1h 18m 59s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-tls-barbican/db06320/ : SUCCESS in 56m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario-amphora-v2/35e1886/ : SUCCESS in 1h 41m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"527e8c559aac06a6e7f2aa4a51613817bf2eda41","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-26 23:33:06.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"51e02588395a68730e2b46f5dd7c0f8a9ddbe921","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 01:59:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/79/652679/3/check/openstack-tox-cover/a08e6cb/cover/ : SUCCESS in 9m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/79/652679/3/check/openstack-tox-lower-constraints/3640b1c/ : SUCCESS in 7m 04s\n- openstack-tox-pep8 http://logs.openstack.org/79/652679/3/check/openstack-tox-pep8/3681113/ : SUCCESS in 5m 48s\n- openstack-tox-py27 http://logs.openstack.org/79/652679/3/check/openstack-tox-py27/ef98872/ : SUCCESS in 5m 32s\n- openstack-tox-py36 http://logs.openstack.org/79/652679/3/check/openstack-tox-py36/0dd7f9f/ : SUCCESS in 4m 42s\n- openstack-tox-py37 http://logs.openstack.org/79/652679/3/check/openstack-tox-py37/1211126/ : SUCCESS in 5m 54s\n- openstack-tox-docs http://logs.openstack.org/79/652679/3/check/openstack-tox-docs/894960f/html/ : SUCCESS in 6m 05s\n- openstack-tox-functional http://logs.openstack.org/79/652679/3/check/openstack-tox-functional/51744ba/ : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 http://logs.openstack.org/79/652679/3/check/openstack-tox-functional-py36/c3a33e0/ : SUCCESS in 6m 22s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-noop-api/3daab52/ : SUCCESS in 55m 15s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-noop-py2-api/3a9ee2d/ : SUCCESS in 53m 26s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario/f0b49d5/ : SUCCESS in 1h 37m 30s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-py2-scenario/0d7ea2b/ : SUCCESS in 1h 51m 13s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-py2-scenario-centos-7/98aa758/ : FAILURE in 2h 10m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/6b90e81/ : SUCCESS in 1h 39m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario-two-node/ba6c1d9/ : FAILURE in 1h 46m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-py2-scenario-two-node/dd6e207/ : SUCCESS in 2h 04m 10s (non-voting)\n- octavia-grenade http://logs.openstack.org/79/652679/3/check/octavia-grenade/8f87630/ : SUCCESS in 1h 36m 21s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-tls-barbican/cd9f6d9/ : SUCCESS in 57m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/79/652679/3/check/octavia-v2-dsvm-scenario-amphora-v2/7ee3587/ : SUCCESS in 1h 43m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/79/652679/3/check/octavia-v2-act-stdby-iptables-dsvm-scenario/8392171/ : SUCCESS in 1h 08m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/79/652679/3/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/3b2957a/ : FAILURE in 57m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/79/652679/3/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/a72248e/ : FAILURE in 1h 15m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cc396fc61e42188ce76a9a9ca007c2a856db8478","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 01:59:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"50d039d4b23c99baa1224f230ed9c209a1023931","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 03:46:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a7d7815bba3d0b48e4fcf50b8a3d1fe1f409673c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 03:46:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/79/652679/3/gate/openstack-tox-lower-constraints/44d1dab/ : SUCCESS in 6m 46s\n- openstack-tox-pep8 http://logs.openstack.org/79/652679/3/gate/openstack-tox-pep8/c0de632/ : SUCCESS in 4m 38s\n- openstack-tox-py27 http://logs.openstack.org/79/652679/3/gate/openstack-tox-py27/7b031a8/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/79/652679/3/gate/openstack-tox-py36/610283e/ : SUCCESS in 3m 30s\n- openstack-tox-py37 http://logs.openstack.org/79/652679/3/gate/openstack-tox-py37/5c42a2a/ : SUCCESS in 4m 24s\n- openstack-tox-docs http://logs.openstack.org/79/652679/3/gate/openstack-tox-docs/bbd7993/html/ : SUCCESS in 5m 11s\n- openstack-tox-functional http://logs.openstack.org/79/652679/3/gate/openstack-tox-functional/c2c43bb/ : SUCCESS in 5m 20s\n- openstack-tox-functional-py36 http://logs.openstack.org/79/652679/3/gate/openstack-tox-functional-py36/f3e2508/ : SUCCESS in 5m 00s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/79/652679/3/gate/octavia-v2-dsvm-noop-api/95512fd/ : SUCCESS in 44m 32s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/79/652679/3/gate/octavia-v2-dsvm-noop-py2-api/56f18e5/ : SUCCESS in 38m 56s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/79/652679/3/gate/octavia-v2-dsvm-scenario/d22fb65/ : SUCCESS in 1h 40m 53s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/79/652679/3/gate/octavia-v2-dsvm-py2-scenario/48bca19/ : SUCCESS in 1h 33m 56s\n- octavia-grenade http://logs.openstack.org/79/652679/3/gate/octavia-grenade/554b3de/ : SUCCESS in 1h 16m 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"24fc7d615ea1e5ae29d22e5d10b4df29c7803de0","revisions":{"12b9d78e2515577a338b4367e09aaa5f2b53cc2b":{"kind":"REWORK","_number":1,"created":"2019-04-15 14:50:15.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/79/652679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/652679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/652679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/652679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/652679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/652679/1"}}},"commit":{"parents":[{"commit":"bd07eada87635320be8d1b42cc22ab5b4af6ba27","subject":"Merge \"Add error-handling for permission of /etc/octavia\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd07eada87635320be8d1b42cc22ab5b4af6ba27"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-04-15 14:49:12.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-04-15 14:50:13.000000000","tz":-420},"subject":"Make Ubuntu bionic the default for amphora images","message":"Make Ubuntu bionic the default for amphora images\n\nThis patch updates the diskimage-create script to select Ubuntu\nbionic as the default version of Ubuntu for the amphora image.\n\nChange-Id: Ie8bbcbe073644137b8eb04c42d9457f507513007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12b9d78e2515577a338b4367e09aaa5f2b53cc2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12b9d78e2515577a338b4367e09aaa5f2b53cc2b"}]},"branch":"refs/heads/master"},"17b771e2ea013b8546c5795f12f4cc49bf2f34a6":{"kind":"REWORK","_number":2,"created":"2019-07-18 22:54:11.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/79/652679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/652679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/652679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/652679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/652679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/652679/2"}}},"commit":{"parents":[{"commit":"2bb2f918ee4fb38b7a17fb6f8781042d204b05e8","subject":"Fix cryptsetup --pbkdf-memory failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2bb2f918ee4fb38b7a17fb6f8781042d204b05e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-04-15 14:49:12.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-18 22:54:10.000000000","tz":-420},"subject":"Make Ubuntu bionic the default for amphora images","message":"Make Ubuntu bionic the default for amphora images\n\nThis patch updates the diskimage-create script to select Ubuntu\nbionic as the default version of Ubuntu for the amphora image.\n\nChange-Id: Ie8bbcbe073644137b8eb04c42d9457f507513007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/17b771e2ea013b8546c5795f12f4cc49bf2f34a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/17b771e2ea013b8546c5795f12f4cc49bf2f34a6"}]},"branch":"refs/heads/master"},"24fc7d615ea1e5ae29d22e5d10b4df29c7803de0":{"kind":"REWORK","_number":3,"created":"2019-07-23 21:52:50.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/79/652679/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/652679/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/652679/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/652679/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/652679/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/652679/3"}}},"commit":{"parents":[{"commit":"2bb2f918ee4fb38b7a17fb6f8781042d204b05e8","subject":"Fix cryptsetup --pbkdf-memory failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2bb2f918ee4fb38b7a17fb6f8781042d204b05e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-04-15 14:49:12.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-23 21:52:50.000000000","tz":-420},"subject":"Make Ubuntu bionic the default for amphora images","message":"Make Ubuntu bionic the default for amphora images\n\nThis patch updates the diskimage-create script to select Ubuntu\nbionic as the default version of Ubuntu for the amphora image.\n\nChange-Id: Ie8bbcbe073644137b8eb04c42d9457f507513007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24fc7d615ea1e5ae29d22e5d10b4df29c7803de0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24fc7d615ea1e5ae29d22e5d10b4df29c7803de0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
