)]}'
{"id":"openstack%2Foctavia~653357","triplet_id":"openstack%2Foctavia~master~I9c9002486a9a547a215514c24ccb95f039b23ec2","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I9c9002486a9a547a215514c24ccb95f039b23ec2","subject":"Fix spare amphora failover","status":"MERGED","created":"2019-04-17 08:47:34.000000000","updated":"2019-04-29 11:49:47.000000000","submitted":"2019-04-26 04:29:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653357-1556252978148-3a5dd909","meta_rev_id":"309d3e779aca3108743a3f2ff89d14f876ac6949","_number":653357,"virtual_id_number":653357,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-04-25 22:47:53.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-04-26 04:29:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":2,"date":"2019-04-25 11:58:47.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-04-24 22:13:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":1,"date":"2019-04-25 11:58:47.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 22:13:41.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-04-25 11:58:47.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-04-25 22:47:53.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-04-26 04:29:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0fed6b731901d2bf784e73f8358d34222c1fe62b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-04-17 08:47:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e8f5f7d5f631fd855ff1b077f1d7728604fe52c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 12:37:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/57/653357/1/check/openstack-tox-cover/1fea38a/cover/ : SUCCESS in 13m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/653357/1/check/openstack-tox-lower-constraints/82be8b2/ : SUCCESS in 8m 38s\n- openstack-tox-pep8 http://logs.openstack.org/57/653357/1/check/openstack-tox-pep8/a460c91/ : SUCCESS in 7m 17s\n- openstack-tox-py27 http://logs.openstack.org/57/653357/1/check/openstack-tox-py27/003369f/ : SUCCESS in 6m 04s\n- openstack-tox-py35 http://logs.openstack.org/57/653357/1/check/openstack-tox-py35/9b0e035/ : SUCCESS in 7m 22s\n- openstack-tox-py36 http://logs.openstack.org/57/653357/1/check/openstack-tox-py36/15daaab/ : SUCCESS in 7m 17s\n- openstack-tox-docs http://logs.openstack.org/57/653357/1/check/openstack-tox-docs/30f1e3f/html/ : SUCCESS in 7m 52s\n- build-openstack-releasenotes http://logs.openstack.org/57/653357/1/check/build-openstack-releasenotes/f1baebf/html/ : SUCCESS in 5m 37s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/57/653357/1/check/neutron-lbaasv2-dsvm-api/5ff8542/ : SUCCESS in 1h 23m 40s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/57/653357/1/check/neutron-lbaasv2-dsvm-py3x-api/8219862/ : SUCCESS in 1h 21m 45s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/57/653357/1/check/neutron-lbaasv2-dsvm-scenario/e7fd928/ : SUCCESS in 1h 28m 34s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/57/653357/1/check/neutron-lbaasv2-dsvm-py3x-scenario/6247b9a/ : SUCCESS in 1h 33m 00s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/57/653357/1/check/openstack-tox-functional/f3943b8/ : SUCCESS in 7m 23s\n- openstack-tox-functional-py35 http://logs.openstack.org/57/653357/1/check/openstack-tox-functional-py35/5eeee5c/ : SUCCESS in 7m 52s\n- openstack-tox-functional-py36 http://logs.openstack.org/57/653357/1/check/openstack-tox-functional-py36/aed71b1/ : SUCCESS in 9m 16s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/57/653357/1/check/octavia-v1-dsvm-scenario/aea206b/ : SUCCESS in 1h 08m 53s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/57/653357/1/check/octavia-v1-dsvm-py3x-scenario/e0d3b4d/ : SUCCESS in 1h 13m 43s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/57/653357/1/check/octavia-v2-dsvm-noop-api/776e88a/ : SUCCESS in 46m 12s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/57/653357/1/check/octavia-v2-dsvm-noop-py2-api/1c5f6fd/ : SUCCESS in 46m 11s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/57/653357/1/check/octavia-v2-dsvm-scenario/f991917/ : SUCCESS in 1h 31m 35s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/57/653357/1/check/octavia-v2-dsvm-py2-scenario/95a6aca/ : SUCCESS in 1h 25m 01s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/57/653357/1/check/octavia-v2-dsvm-py2-scenario-centos-7/3ed029e/ : FAILURE in 1h 53m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/57/653357/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/ba7b1eb/ : SUCCESS in 1h 36m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/57/653357/1/check/octavia-v2-dsvm-scenario-two-node/2261dae/ : FAILURE in 2h 33m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/57/653357/1/check/octavia-v2-dsvm-py2-scenario-two-node/33a0075/ : FAILURE in 1h 49m 00s (non-voting)\n- octavia-grenade http://logs.openstack.org/57/653357/1/check/octavia-grenade/2f3a0c9/ : SUCCESS in 1h 28m 56s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/57/653357/1/check/octavia-v2-dsvm-tls-barbican/85e93f7/ : SUCCESS in 56m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9b323b2df5e77eaf5465798eeb05702e978230e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-24 22:13:41.000000000","message":"Patch Set 1: Code-Review+2\n\nSigh, thanks!  LGTM","accounts_in_message":[],"_revision_number":1},{"id":"f2e0436b81121609f234e3284c76ec81f76adae2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-24 22:15:11.000000000","message":"Patch Set 1:\n\nBonus points for a tempest test covering this.","accounts_in_message":[],"_revision_number":1},{"id":"61dbb66726b0d0e09d5b3c8170b4a0bddece3b37","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-25 11:58:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nGreg seems to working on that tempest test which would be code in octavia-tempest-plugin repo so we can merge this one.","accounts_in_message":[],"_revision_number":1},{"id":"a5b248061ed0012b655194aefd33fdbb6eff80e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 11:59:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8f6215a5db6a7b6e855e172c2b5b2f5968976f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 14:06:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/57/653357/1/gate/openstack-tox-lower-constraints/4f2ec96/ : SUCCESS in 7m 32s\n- openstack-tox-pep8 http://logs.openstack.org/57/653357/1/gate/openstack-tox-pep8/9601e04/ : SUCCESS in 6m 47s\n- openstack-tox-py27 http://logs.openstack.org/57/653357/1/gate/openstack-tox-py27/17cf9c2/ : SUCCESS in 5m 01s\n- openstack-tox-py35 http://logs.openstack.org/57/653357/1/gate/openstack-tox-py35/fbd66c3/ : SUCCESS in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/57/653357/1/gate/openstack-tox-py36/f1c5a83/ : SUCCESS in 7m 42s\n- openstack-tox-docs http://logs.openstack.org/57/653357/1/gate/openstack-tox-docs/dccfba4/html/ : SUCCESS in 8m 02s\n- build-openstack-releasenotes http://logs.openstack.org/57/653357/1/gate/build-openstack-releasenotes/69b29c3/html/ : SUCCESS in 4m 55s\n- openstack-tox-functional http://logs.openstack.org/57/653357/1/gate/openstack-tox-functional/97cc101/ : SUCCESS in 8m 11s\n- openstack-tox-functional-py35 http://logs.openstack.org/57/653357/1/gate/openstack-tox-functional-py35/c1a610e/ : SUCCESS in 7m 43s\n- openstack-tox-functional-py36 http://logs.openstack.org/57/653357/1/gate/openstack-tox-functional-py36/bb53ad3/ : SUCCESS in 6m 41s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/57/653357/1/gate/octavia-v1-dsvm-scenario/d392efe/ : SUCCESS in 1h 08m 28s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/57/653357/1/gate/octavia-v1-dsvm-py3x-scenario/7e4b87b/ : SUCCESS in 1h 03m 23s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/57/653357/1/gate/neutron-lbaasv2-dsvm-api/b792c7a/ : SUCCESS in 1h 19m 05s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/57/653357/1/gate/neutron-lbaasv2-dsvm-py3x-api/2867564/ : SUCCESS in 1h 20m 40s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/57/653357/1/gate/octavia-v2-dsvm-noop-api/133a15a/ : SUCCESS in 38m 34s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/57/653357/1/gate/octavia-v2-dsvm-noop-py2-api/5e0472f/ : SUCCESS in 36m 52s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/57/653357/1/gate/octavia-v2-dsvm-scenario/d77d2f2/ : FAILURE in 1h 38m 40s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/57/653357/1/gate/octavia-v2-dsvm-py2-scenario/f28e645/ : SUCCESS in 1h 36m 39s\n- octavia-grenade http://logs.openstack.org/57/653357/1/gate/octavia-grenade/22f9e40/ : SUCCESS in 1h 14m 33s","accounts_in_message":[],"_revision_number":1},{"id":"9ac69d26f446fcbe5857acfcfc71b5bd3dcc4046","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-04-25 22:47:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"18b2234d06376d748e593f527726364181912a80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 02:33:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/57/653357/1/check/openstack-tox-cover/2dd00af/cover/ : SUCCESS in 8m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/653357/1/check/openstack-tox-lower-constraints/24c2c50/ : SUCCESS in 7m 31s\n- openstack-tox-pep8 http://logs.openstack.org/57/653357/1/check/openstack-tox-pep8/84a7922/ : SUCCESS in 7m 32s\n- openstack-tox-py27 http://logs.openstack.org/57/653357/1/check/openstack-tox-py27/9c20dda/ : SUCCESS in 4m 45s\n- openstack-tox-py35 http://logs.openstack.org/57/653357/1/check/openstack-tox-py35/bb17cac/ : SUCCESS in 5m 20s\n- openstack-tox-py36 http://logs.openstack.org/57/653357/1/check/openstack-tox-py36/aef4428/ : SUCCESS in 6m 03s\n- openstack-tox-docs http://logs.openstack.org/57/653357/1/check/openstack-tox-docs/f4ce78b/html/ : SUCCESS in 6m 40s\n- build-openstack-releasenotes http://logs.openstack.org/57/653357/1/check/build-openstack-releasenotes/423cb7b/html/ : SUCCESS in 5m 29s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/57/653357/1/check/neutron-lbaasv2-dsvm-api/bef8b2f/ : SUCCESS in 1h 16m 32s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/57/653357/1/check/neutron-lbaasv2-dsvm-py3x-api/ab01b2a/ : SUCCESS in 1h 18m 30s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/57/653357/1/check/neutron-lbaasv2-dsvm-scenario/23d1822/ : SUCCESS in 1h 23m 50s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/57/653357/1/check/neutron-lbaasv2-dsvm-py3x-scenario/199e166/ : SUCCESS in 1h 24m 24s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/57/653357/1/check/openstack-tox-functional/f576a6f/ : SUCCESS in 6m 30s\n- openstack-tox-functional-py35 http://logs.openstack.org/57/653357/1/check/openstack-tox-functional-py35/c14a402/ : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 http://logs.openstack.org/57/653357/1/check/openstack-tox-functional-py36/d6bcf5f/ : SUCCESS in 7m 19s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/57/653357/1/check/octavia-v1-dsvm-scenario/f65ab24/ : SUCCESS in 1h 06m 38s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/57/653357/1/check/octavia-v1-dsvm-py3x-scenario/7b9e830/ : SUCCESS in 46m 48s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/57/653357/1/check/octavia-v2-dsvm-noop-api/9217989/ : SUCCESS in 43m 18s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/57/653357/1/check/octavia-v2-dsvm-noop-py2-api/0a44159/ : SUCCESS in 45m 59s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/57/653357/1/check/octavia-v2-dsvm-scenario/e9dfa53/ : SUCCESS in 1h 42m 54s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/57/653357/1/check/octavia-v2-dsvm-py2-scenario/ae0e308/ : SUCCESS in 1h 44m 20s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/57/653357/1/check/octavia-v2-dsvm-py2-scenario-centos-7/88755a3/ : FAILURE in 37m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/57/653357/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/9314b3a/ : SUCCESS in 1h 48m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/57/653357/1/check/octavia-v2-dsvm-scenario-two-node/19dfd73/ : SUCCESS in 1h 10m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/57/653357/1/check/octavia-v2-dsvm-py2-scenario-two-node/f12b6b4/ : SUCCESS in 1h 38m 13s (non-voting)\n- octavia-grenade http://logs.openstack.org/57/653357/1/check/octavia-grenade/43c4a9a/ : SUCCESS in 1h 20m 25s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/57/653357/1/check/octavia-v2-dsvm-tls-barbican/840e369/ : SUCCESS in 53m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b726bf62ad2597b36cb13f0b6a058212fe308f55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 02:33:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1715e569fb42d3150259120eddb723c377bf079","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 04:29:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/57/653357/1/gate/openstack-tox-lower-constraints/054e537/ : SUCCESS in 7m 21s\n- openstack-tox-pep8 http://logs.openstack.org/57/653357/1/gate/openstack-tox-pep8/8d2b7e8/ : SUCCESS in 6m 38s\n- openstack-tox-py27 http://logs.openstack.org/57/653357/1/gate/openstack-tox-py27/e2871e4/ : SUCCESS in 5m 01s\n- openstack-tox-py35 http://logs.openstack.org/57/653357/1/gate/openstack-tox-py35/e2a5064/ : SUCCESS in 6m 05s\n- openstack-tox-py36 http://logs.openstack.org/57/653357/1/gate/openstack-tox-py36/d2e0413/ : SUCCESS in 5m 36s\n- openstack-tox-docs http://logs.openstack.org/57/653357/1/gate/openstack-tox-docs/4c8c4cc/html/ : SUCCESS in 6m 31s\n- build-openstack-releasenotes http://logs.openstack.org/57/653357/1/gate/build-openstack-releasenotes/13d8125/html/ : SUCCESS in 4m 42s\n- openstack-tox-functional http://logs.openstack.org/57/653357/1/gate/openstack-tox-functional/3ee75c2/ : SUCCESS in 6m 58s\n- openstack-tox-functional-py35 http://logs.openstack.org/57/653357/1/gate/openstack-tox-functional-py35/ab09677/ : SUCCESS in 7m 58s\n- openstack-tox-functional-py36 http://logs.openstack.org/57/653357/1/gate/openstack-tox-functional-py36/6ffdad2/ : SUCCESS in 7m 04s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/57/653357/1/gate/octavia-v1-dsvm-scenario/664032b/ : SUCCESS in 1h 14m 59s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/57/653357/1/gate/octavia-v1-dsvm-py3x-scenario/e9c2ed8/ : SUCCESS in 1h 06m 12s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/57/653357/1/gate/neutron-lbaasv2-dsvm-api/57968d8/ : SUCCESS in 1h 11m 09s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/57/653357/1/gate/neutron-lbaasv2-dsvm-py3x-api/633ce54/ : SUCCESS in 1h 14m 32s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/57/653357/1/gate/octavia-v2-dsvm-noop-api/95aac8e/ : SUCCESS in 39m 11s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/57/653357/1/gate/octavia-v2-dsvm-noop-py2-api/6520134/ : SUCCESS in 42m 50s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/57/653357/1/gate/octavia-v2-dsvm-scenario/032a924/ : SUCCESS in 1h 48m 12s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/57/653357/1/gate/octavia-v2-dsvm-py2-scenario/6e254f2/ : SUCCESS in 1h 45m 08s\n- octavia-grenade http://logs.openstack.org/57/653357/1/gate/octavia-grenade/f9d1010/ : SUCCESS in 1h 22m 02s","accounts_in_message":[],"_revision_number":1},{"id":"754efebc1639f81c2cfa5512cf9993e6255bf9c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 04:29:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"309d3e779aca3108743a3f2ff89d14f876ac6949","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-04-29 11:49:47.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit aa259d12234b33e1a0061e4f87bb0038e3e25b40","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a509710a7faf5c643377d1593a2591d37f3e8390","revisions":{"a509710a7faf5c643377d1593a2591d37f3e8390":{"kind":"REWORK","_number":1,"created":"2019-04-17 08:47:34.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/57/653357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/653357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/653357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/653357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/653357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/653357/1"}}},"commit":{"parents":[{"commit":"66656bceb6657ac0a4d0d1ce86b6ac2925cf043e","subject":"Merge \"Update documentation about rotating amphorae\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66656bceb6657ac0a4d0d1ce86b6ac2925cf043e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-04-17 08:46:07.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-04-17 08:46:24.000000000","tz":120},"subject":"Fix spare amphora failover","message":"Fix spare amphora failover\n\nFailover of spare amphorae failed with an exception,\nbecause they\u0027re not attached to any loadbalancer.\n\nThis commit fixes this issue, allowing to rotate/update\nspare amphora images.\n\nStory: 2005470\nTask: 30549\n\nChange-Id: I9c9002486a9a547a215514c24ccb95f039b23ec2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a509710a7faf5c643377d1593a2591d37f3e8390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a509710a7faf5c643377d1593a2591d37f3e8390"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
