)]}'
{"id":"openstack%2Foctavia~653488","triplet_id":"openstack%2Foctavia~master~I8897278b1ac8b76d564f45bd0c7cbc26b29a8e5d","project":"openstack/octavia","branch":"master","topic":"enable-ppc64le-amphora","hashtags":[],"change_id":"I8897278b1ac8b76d564f45bd0c7cbc26b29a8e5d","subject":"Adding changes to support ppc64le amphora builds with CentOS","status":"MERGED","created":"2019-04-17 16:32:55.000000000","updated":"2019-04-22 17:22:47.000000000","submitted":"2019-04-22 17:22:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":33,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653488-1555953767144-4f1731f4","meta_rev_id":"3167276a7787453b243671f4846b6d7be3182160","_number":653488,"virtual_id_number":653488,"owner":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"date":"2019-04-22 13:13:15.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-04-22 17:22:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-04-18 21:34:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":2,"date":"2019-04-20 13:53:53.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":1,"date":"2019-04-20 13:53:53.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-18 07:12:18.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-04-18 14:48:39.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-04-18 21:34:56.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-04-20 13:53:53.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-04-22 13:13:15.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-04-22 17:22:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca21767fa46fc4a06070c4380436b42497fa316c","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-04-17 16:32:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cdc08d2b917b3895367229f49d2c4a5322c87b2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-17 17:23:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ee2ab5ee708d51e861ed2ccbc5f63e0642715b7d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-17 17:25:17.000000000","message":"Patch Set 1:\n\nYou also need to update the README.rst","accounts_in_message":[],"_revision_number":1},{"id":"329e78b8b89787859a7468a8a9babd2718d305f8","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-04-17 17:42:52.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3001e376f9e6a320286d80a29527ae920b20e58d","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-04-17 17:43:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ed1589ef15e75dd9ff5790d16343be692f030018","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-17 17:46:35.000000000","message":"Patch Set 2: Code-Review-1\n\nStill three open issues, remarking -1","accounts_in_message":[],"_revision_number":2},{"id":"9c3fb1984cea390313807e60f341f74a0a8ae7b6","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-04-17 19:08:23.000000000","message":"Patch Set 2: Workflow-1\n\nWas asked to mark it WIP until I can get a test run done of the generated image.","accounts_in_message":[],"_revision_number":2},{"id":"5c1083878bf23a1bdd5cece7ad23b743905fc59f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 20:01:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/88/653488/2/check/openstack-tox-cover/ff9662c/cover/ : SUCCESS in 10m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/653488/2/check/openstack-tox-lower-constraints/2c23a59/ : SUCCESS in 8m 27s\n- openstack-tox-pep8 http://logs.openstack.org/88/653488/2/check/openstack-tox-pep8/f617bb7/ : SUCCESS in 7m 51s\n- openstack-tox-py27 http://logs.openstack.org/88/653488/2/check/openstack-tox-py27/be3a12e/ : SUCCESS in 7m 10s\n- openstack-tox-py35 http://logs.openstack.org/88/653488/2/check/openstack-tox-py35/6377c2b/ : SUCCESS in 5m 58s\n- openstack-tox-py36 http://logs.openstack.org/88/653488/2/check/openstack-tox-py36/2b0fb62/ : SUCCESS in 5m 43s\n- openstack-tox-docs http://logs.openstack.org/88/653488/2/check/openstack-tox-docs/74f3921/html/ : SUCCESS in 7m 04s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/88/653488/2/check/neutron-lbaasv2-dsvm-api/734c28a/ : SUCCESS in 1h 22m 44s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/88/653488/2/check/neutron-lbaasv2-dsvm-py3x-api/428370e/ : SUCCESS in 1h 23m 25s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/88/653488/2/check/neutron-lbaasv2-dsvm-scenario/8d20061/ : SUCCESS in 1h 27m 46s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/88/653488/2/check/neutron-lbaasv2-dsvm-py3x-scenario/48dc73e/ : SUCCESS in 1h 20m 46s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/88/653488/2/check/openstack-tox-functional/5256ae5/ : SUCCESS in 7m 34s\n- openstack-tox-functional-py35 http://logs.openstack.org/88/653488/2/check/openstack-tox-functional-py35/eae8386/ : SUCCESS in 8m 14s\n- openstack-tox-functional-py36 http://logs.openstack.org/88/653488/2/check/openstack-tox-functional-py36/e966c57/ : SUCCESS in 8m 08s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/88/653488/2/check/octavia-v1-dsvm-scenario/042c3d0/ : SUCCESS in 1h 06m 54s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/88/653488/2/check/octavia-v1-dsvm-py3x-scenario/3870e45/ : SUCCESS in 1h 09m 15s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/653488/2/check/octavia-v2-dsvm-noop-api/5e9394f/ : SUCCESS in 47m 41s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/653488/2/check/octavia-v2-dsvm-noop-py2-api/21be090/ : SUCCESS in 46m 36s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/653488/2/check/octavia-v2-dsvm-scenario/6c3a23e/ : SUCCESS in 1h 37m 43s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/653488/2/check/octavia-v2-dsvm-py2-scenario/1f66e61/ : SUCCESS in 1h 36m 26s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/88/653488/2/check/octavia-v2-dsvm-py2-scenario-centos-7/e31b880/ : FAILURE in 1h 56m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/653488/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/b76fdc3/ : SUCCESS in 1h 37m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/653488/2/check/octavia-v2-dsvm-scenario-two-node/24c5035/ : SUCCESS in 1h 44m 16s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/653488/2/check/octavia-v2-dsvm-py2-scenario-two-node/0077740/ : FAILURE in 1h 35m 58s (non-voting)\n- octavia-grenade http://logs.openstack.org/88/653488/2/check/octavia-grenade/0ce02cf/ : SUCCESS in 1h 19m 11s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/653488/2/check/octavia-v2-dsvm-tls-barbican/5860651/ : SUCCESS in 53m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a1e0665e43f600f5f11903d397848a54ecb93384","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-04-17 21:19:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"eb6e4bd6aeead4ad14f6bb762e9659e711865abe","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-17 21:20:00.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c1d7e569641f739ed71179115e2f3692744d406b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-17 21:52:00.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"19b133ac7ebe21eaf6e5d0f307eac7b2f9d43a9c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-04-17 22:35:49.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"974570b46003a282546307d077ed90fd505f59db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 23:09:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/88/653488/3/check/openstack-tox-cover/e9bef88/cover/ : SUCCESS in 10m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/653488/3/check/openstack-tox-lower-constraints/209e755/ : SUCCESS in 7m 54s\n- openstack-tox-pep8 http://logs.openstack.org/88/653488/3/check/openstack-tox-pep8/57d67cf/ : SUCCESS in 8m 12s\n- openstack-tox-py27 http://logs.openstack.org/88/653488/3/check/openstack-tox-py27/f4ea6ce/ : SUCCESS in 5m 55s\n- openstack-tox-py35 http://logs.openstack.org/88/653488/3/check/openstack-tox-py35/b50199a/ : SUCCESS in 6m 39s\n- openstack-tox-py36 http://logs.openstack.org/88/653488/3/check/openstack-tox-py36/646decb/ : SUCCESS in 6m 09s\n- openstack-tox-docs http://logs.openstack.org/88/653488/3/check/openstack-tox-docs/7a912fe/html/ : SUCCESS in 7m 33s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/88/653488/3/check/neutron-lbaasv2-dsvm-api/e8deb46/ : SUCCESS in 1h 24m 00s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/88/653488/3/check/neutron-lbaasv2-dsvm-py3x-api/e6d451c/ : SUCCESS in 1h 28m 25s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/88/653488/3/check/neutron-lbaasv2-dsvm-scenario/493c9fd/ : SUCCESS in 1h 41m 48s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/88/653488/3/check/neutron-lbaasv2-dsvm-py3x-scenario/44f1055/ : SUCCESS in 1h 33m 26s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/88/653488/3/check/openstack-tox-functional/cd2aed9/ : SUCCESS in 9m 04s\n- openstack-tox-functional-py35 http://logs.openstack.org/88/653488/3/check/openstack-tox-functional-py35/37ae708/ : SUCCESS in 8m 16s\n- openstack-tox-functional-py36 http://logs.openstack.org/88/653488/3/check/openstack-tox-functional-py36/eecf7cd/ : SUCCESS in 9m 00s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/88/653488/3/check/octavia-v1-dsvm-scenario/d8010fc/ : SUCCESS in 1h 09m 06s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/88/653488/3/check/octavia-v1-dsvm-py3x-scenario/d4f8376/ : SUCCESS in 1h 03m 53s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/653488/3/check/octavia-v2-dsvm-noop-api/e8d92ac/ : SUCCESS in 46m 17s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/653488/3/check/octavia-v2-dsvm-noop-py2-api/02b9439/ : SUCCESS in 42m 49s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/653488/3/check/octavia-v2-dsvm-scenario/8e6628b/ : SUCCESS in 1h 33m 16s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/653488/3/check/octavia-v2-dsvm-py2-scenario/0473a8f/ : SUCCESS in 1h 28m 25s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/88/653488/3/check/octavia-v2-dsvm-py2-scenario-centos-7/568d9fb/ : FAILURE in 39m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/653488/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/ab783b5/ : SUCCESS in 1h 34m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/653488/3/check/octavia-v2-dsvm-scenario-two-node/b94fb29/ : FAILURE in 1h 35m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/653488/3/check/octavia-v2-dsvm-py2-scenario-two-node/c4297c4/ : FAILURE in 1h 26m 39s (non-voting)\n- octavia-grenade http://logs.openstack.org/88/653488/3/check/octavia-grenade/327da3c/ : SUCCESS in 1h 09m 10s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/653488/3/check/octavia-v2-dsvm-tls-barbican/415589c/ : SUCCESS in 52m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e12d0420a6a0ebf807a743aeedfd8a051044a33f","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-18 07:12:18.000000000","message":"Patch Set 3:\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a0fdd4f37002e1434ae24552a3948cf34468b359","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-04-18 09:30:13.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b015cc2fa71416b43a006838acd9fbe0f92bf256","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-04-18 14:10:33.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nSo do I need to pull the tmpfs change out of this review, or is it minimal enough to leave in?  Also, I didn\u0027t really see a consensus about the ppc64le repo link, so I didn\u0027t change anything.","accounts_in_message":[],"_revision_number":3},{"id":"0f8d45c14838439a9727915ee9bc49ab6566ee56","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-04-18 14:30:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a6162e73c0b1358f6cb0a802fe23195244b9675a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-18 14:48:39.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"affb103cbc5408e661ea537278602d09c28d917e","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-04-18 16:11:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4cbbc5eaddf8e7e7e97ab89d30ceb929c120c8cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 18:35:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/88/653488/5/check/openstack-tox-cover/bd1b16a/cover/ : SUCCESS in 9m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/653488/5/check/openstack-tox-lower-constraints/3db0a50/ : SUCCESS in 9m 01s\n- openstack-tox-pep8 http://logs.openstack.org/88/653488/5/check/openstack-tox-pep8/59dcd4b/ : SUCCESS in 7m 29s\n- openstack-tox-py27 http://logs.openstack.org/88/653488/5/check/openstack-tox-py27/a24c22a/ : SUCCESS in 7m 14s\n- openstack-tox-py35 http://logs.openstack.org/88/653488/5/check/openstack-tox-py35/f43fa9c/ : SUCCESS in 9m 19s\n- openstack-tox-py36 http://logs.openstack.org/88/653488/5/check/openstack-tox-py36/af8d8e2/ : SUCCESS in 6m 29s\n- openstack-tox-docs http://logs.openstack.org/88/653488/5/check/openstack-tox-docs/138d4bd/html/ : SUCCESS in 7m 30s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/88/653488/5/check/neutron-lbaasv2-dsvm-api/4c2a172/ : SUCCESS in 1h 21m 07s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/88/653488/5/check/neutron-lbaasv2-dsvm-py3x-api/7fe4a10/ : SUCCESS in 1h 20m 15s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/88/653488/5/check/neutron-lbaasv2-dsvm-scenario/7f7376b/ : SUCCESS in 1h 25m 10s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/88/653488/5/check/neutron-lbaasv2-dsvm-py3x-scenario/6755a95/ : SUCCESS in 1h 29m 25s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/88/653488/5/check/openstack-tox-functional/a9f723a/ : SUCCESS in 9m 06s\n- openstack-tox-functional-py35 http://logs.openstack.org/88/653488/5/check/openstack-tox-functional-py35/7cd6ec1/ : SUCCESS in 8m 19s\n- openstack-tox-functional-py36 http://logs.openstack.org/88/653488/5/check/openstack-tox-functional-py36/d30c866/ : SUCCESS in 9m 01s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/88/653488/5/check/octavia-v1-dsvm-scenario/3101c7b/ : SUCCESS in 1h 17m 46s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/88/653488/5/check/octavia-v1-dsvm-py3x-scenario/87eaa91/ : SUCCESS in 1h 10m 14s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/653488/5/check/octavia-v2-dsvm-noop-api/8944b68/ : SUCCESS in 45m 53s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/653488/5/check/octavia-v2-dsvm-noop-py2-api/b5619cf/ : SUCCESS in 49m 28s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/653488/5/check/octavia-v2-dsvm-scenario/21a4694/ : SUCCESS in 1h 36m 49s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/653488/5/check/octavia-v2-dsvm-py2-scenario/adad09d/ : SUCCESS in 1h 34m 12s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/88/653488/5/check/octavia-v2-dsvm-py2-scenario-centos-7/b5b2979/ : FAILURE in 35m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/653488/5/check/octavia-v2-dsvm-scenario-ubuntu-bionic/18a8f51/ : SUCCESS in 2h 20m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/653488/5/check/octavia-v2-dsvm-scenario-two-node/3e56507/ : SUCCESS in 1h 26m 06s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/653488/5/check/octavia-v2-dsvm-py2-scenario-two-node/5c15c2e/ : FAILURE in 2h 03m 35s (non-voting)\n- octavia-grenade http://logs.openstack.org/88/653488/5/check/octavia-grenade/1b646cb/ : SUCCESS in 1h 19m 54s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/653488/5/check/octavia-v2-dsvm-tls-barbican/b931ce9/ : SUCCESS in 53m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"74fb954640b8f66a9c7978a5312a48cfea58c76c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-04-18 21:31:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a09aa8034b063c8aab2461d18b6386cc06fb284d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-04-18 21:34:56.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)\n\nSuper-nit: I feel like the commit message violates the style rules, but it looks like we don\u0027t flag that stuff anymore in the pep8 checks?\nAlso I hate the letter we chose for the tmpfs disable thing, but given how many are reserved already, I figure this is the best we\u0027re going to do with only supporting short-opts. \u003e_\u003e","accounts_in_message":[],"_revision_number":5},{"id":"bded0f08575c0d24fa1cc17d9519967468d01c9b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-20 13:53:53.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(2 comments)\n\nThis is okay as is for me and Trevor addressed latest comments. The includepkgs can be revisited if proven to be causing trouble.","accounts_in_message":[],"_revision_number":5},{"id":"9c64ba11608bab25b4ef2de61ac5224f9e6c6b09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 13:54:05.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6dab377356c2aa1e933805e466767da264238f21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 15:29:37.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/88/653488/5/gate/openstack-tox-lower-constraints/86acdfc/ : SUCCESS in 7m 23s\n- openstack-tox-pep8 http://logs.openstack.org/88/653488/5/gate/openstack-tox-pep8/824a381/ : SUCCESS in 6m 57s\n- openstack-tox-py27 http://logs.openstack.org/88/653488/5/gate/openstack-tox-py27/ed081f1/ : SUCCESS in 5m 07s\n- openstack-tox-py35 http://logs.openstack.org/88/653488/5/gate/openstack-tox-py35/980f4c7/ : SUCCESS in 7m 07s\n- openstack-tox-py36 http://logs.openstack.org/88/653488/5/gate/openstack-tox-py36/095aac1/ : SUCCESS in 7m 51s\n- openstack-tox-docs http://logs.openstack.org/88/653488/5/gate/openstack-tox-docs/900f50f/html/ : SUCCESS in 6m 23s\n- openstack-tox-functional http://logs.openstack.org/88/653488/5/gate/openstack-tox-functional/4dbd9e1/ : SUCCESS in 5m 56s\n- openstack-tox-functional-py35 http://logs.openstack.org/88/653488/5/gate/openstack-tox-functional-py35/793135a/ : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 http://logs.openstack.org/88/653488/5/gate/openstack-tox-functional-py36/4f5ed7e/ : SUCCESS in 7m 12s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/88/653488/5/gate/octavia-v1-dsvm-scenario/f15a139/ : FAILURE in 4m 45s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/88/653488/5/gate/octavia-v1-dsvm-py3x-scenario/d53ce3b/ : FAILURE in 13m 16s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/88/653488/5/gate/neutron-lbaasv2-dsvm-api/24091f4/ : FAILURE in 4m 21s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/88/653488/5/gate/neutron-lbaasv2-dsvm-py3x-api/6a44fcc/ : FAILURE in 9m 17s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/653488/5/gate/octavia-v2-dsvm-noop-api/eb997d5/ : SUCCESS in 43m 09s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/653488/5/gate/octavia-v2-dsvm-noop-py2-api/600f702/ : SUCCESS in 33m 38s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/653488/5/gate/octavia-v2-dsvm-scenario/145f48b/ : SUCCESS in 1h 27m 06s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/653488/5/gate/octavia-v2-dsvm-py2-scenario/ab46fb6/ : SUCCESS in 1h 34m 39s\n- octavia-grenade http://logs.openstack.org/88/653488/5/gate/octavia-grenade/23b51aa/ : FAILURE in 9m 31s","accounts_in_message":[],"_revision_number":5},{"id":"66b869e87a10d2f2b685a15438c4ce1edc7dcb15","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-04-22 13:13:15.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"0af604812978d9dd1bf00931c267fd1fe2309565","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 15:38:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/88/653488/5/check/openstack-tox-cover/c6dd32e/cover/ : SUCCESS in 10m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/653488/5/check/openstack-tox-lower-constraints/18d51fc/ : SUCCESS in 8m 08s\n- openstack-tox-pep8 http://logs.openstack.org/88/653488/5/check/openstack-tox-pep8/1907882/ : SUCCESS in 9m 21s\n- openstack-tox-py27 http://logs.openstack.org/88/653488/5/check/openstack-tox-py27/b976890/ : SUCCESS in 11m 31s\n- openstack-tox-py35 http://logs.openstack.org/88/653488/5/check/openstack-tox-py35/b84d90f/ : SUCCESS in 8m 16s\n- openstack-tox-py36 http://logs.openstack.org/88/653488/5/check/openstack-tox-py36/17ad379/ : SUCCESS in 6m 04s\n- openstack-tox-docs http://logs.openstack.org/88/653488/5/check/openstack-tox-docs/bceddaa/html/ : SUCCESS in 7m 23s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/88/653488/5/check/neutron-lbaasv2-dsvm-api/26d53db/ : SUCCESS in 1h 20m 18s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/88/653488/5/check/neutron-lbaasv2-dsvm-py3x-api/0dc0064/ : SUCCESS in 1h 21m 06s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/88/653488/5/check/neutron-lbaasv2-dsvm-scenario/5cd66b4/ : SUCCESS in 1h 44m 52s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/88/653488/5/check/neutron-lbaasv2-dsvm-py3x-scenario/e2c0b5b/ : FAILURE in 10m 51s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/88/653488/5/check/openstack-tox-functional/107c4c2/ : SUCCESS in 6m 58s\n- openstack-tox-functional-py35 http://logs.openstack.org/88/653488/5/check/openstack-tox-functional-py35/72a43d8/ : SUCCESS in 10m 40s\n- openstack-tox-functional-py36 http://logs.openstack.org/88/653488/5/check/openstack-tox-functional-py36/f10d3bc/ : SUCCESS in 7m 23s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/88/653488/5/check/octavia-v1-dsvm-scenario/a57a027/ : SUCCESS in 1h 07m 52s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/88/653488/5/check/octavia-v1-dsvm-py3x-scenario/8ade172/ : SUCCESS in 1h 07m 15s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/653488/5/check/octavia-v2-dsvm-noop-api/70aefa3/ : SUCCESS in 41m 54s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/653488/5/check/octavia-v2-dsvm-noop-py2-api/74b1388/ : SUCCESS in 37m 33s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/653488/5/check/octavia-v2-dsvm-scenario/d9e8817/ : SUCCESS in 1h 37m 42s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/653488/5/check/octavia-v2-dsvm-py2-scenario/da7c654/ : SUCCESS in 1h 35m 17s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/88/653488/5/check/octavia-v2-dsvm-py2-scenario-centos-7/f8cdcd3/ : FAILURE in 38m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/653488/5/check/octavia-v2-dsvm-scenario-ubuntu-bionic/2a10cda/ : SUCCESS in 1h 36m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/653488/5/check/octavia-v2-dsvm-scenario-two-node/8ac3bca/ : SUCCESS in 1h 51m 59s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/653488/5/check/octavia-v2-dsvm-py2-scenario-two-node/e560503/ : SUCCESS in 1h 43m 15s (non-voting)\n- octavia-grenade http://logs.openstack.org/88/653488/5/check/octavia-grenade/a539cf6/ : SUCCESS in 1h 16m 56s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/653488/5/check/octavia-v2-dsvm-tls-barbican/d9f8de4/ : SUCCESS in 57m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5a11fc5639c92250b268b0cc86596c0a128b7e87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 15:38:20.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2d30dfd1be0ba2f4fe9e69391cce41567f2bc688","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:22:46.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/88/653488/5/gate/openstack-tox-lower-constraints/fd0bdf5/ : SUCCESS in 7m 36s\n- openstack-tox-pep8 http://logs.openstack.org/88/653488/5/gate/openstack-tox-pep8/5a5f0aa/ : SUCCESS in 6m 44s\n- openstack-tox-py27 http://logs.openstack.org/88/653488/5/gate/openstack-tox-py27/f9d0fb8/ : SUCCESS in 5m 43s\n- openstack-tox-py35 http://logs.openstack.org/88/653488/5/gate/openstack-tox-py35/4983731/ : SUCCESS in 5m 51s\n- openstack-tox-py36 http://logs.openstack.org/88/653488/5/gate/openstack-tox-py36/629502c/ : SUCCESS in 5m 13s\n- openstack-tox-docs http://logs.openstack.org/88/653488/5/gate/openstack-tox-docs/c942484/html/ : SUCCESS in 6m 58s\n- openstack-tox-functional http://logs.openstack.org/88/653488/5/gate/openstack-tox-functional/64d5a9b/ : SUCCESS in 7m 02s\n- openstack-tox-functional-py35 http://logs.openstack.org/88/653488/5/gate/openstack-tox-functional-py35/c4d387f/ : SUCCESS in 8m 41s\n- openstack-tox-functional-py36 http://logs.openstack.org/88/653488/5/gate/openstack-tox-functional-py36/3af6f52/ : SUCCESS in 7m 52s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/88/653488/5/gate/octavia-v1-dsvm-scenario/2b461a2/ : SUCCESS in 59m 54s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/88/653488/5/gate/octavia-v1-dsvm-py3x-scenario/fe1bb1b/ : SUCCESS in 1h 07m 23s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/88/653488/5/gate/neutron-lbaasv2-dsvm-api/cdb21b2/ : SUCCESS in 1h 20m 21s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/88/653488/5/gate/neutron-lbaasv2-dsvm-py3x-api/3e995cb/ : SUCCESS in 1h 30m 31s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/653488/5/gate/octavia-v2-dsvm-noop-api/b5af011/ : SUCCESS in 36m 50s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/653488/5/gate/octavia-v2-dsvm-noop-py2-api/9a4c453/ : SUCCESS in 46m 19s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/653488/5/gate/octavia-v2-dsvm-scenario/718cdda/ : SUCCESS in 1h 35m 42s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/653488/5/gate/octavia-v2-dsvm-py2-scenario/c6d62c6/ : SUCCESS in 1h 35m 13s\n- octavia-grenade http://logs.openstack.org/88/653488/5/gate/octavia-grenade/1d1fba7/ : SUCCESS in 1h 02m 57s","accounts_in_message":[],"_revision_number":5},{"id":"2c61446a32af70c39c451b141d226be90e9cdf00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:22:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4c9097c681c937531132456a5f7443175b591b17","revisions":{"c52ce929d6df1770f171310251d30e2d1940a63e":{"kind":"REWORK","_number":1,"created":"2019-04-17 16:32:55.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/88/653488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/653488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/653488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/653488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/653488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/653488/1"}}},"commit":{"parents":[{"commit":"66656bceb6657ac0a4d0d1ce86b6ac2925cf043e","subject":"Merge \"Update documentation about rotating amphorae\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66656bceb6657ac0a4d0d1ce86b6ac2925cf043e"}]}],"author":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-04-17 16:22:28.000000000","tz":-300},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-04-17 16:32:55.000000000","tz":-300},"subject":"Adding changes to support ppc64le amphora builds with CentOS","message":"Adding changes to support ppc64le amphora builds with CentOS\n\nUpdated diskimage create script to include an argument for disabling the temporary filesystem.\nUpdated diskimage create to support ppc64le as an argument.\nUpdated backports to support a properly configured PaaS repository for CentOS on power.\n\nChange-Id: I8897278b1ac8b76d564f45bd0c7cbc26b29a8e5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c52ce929d6df1770f171310251d30e2d1940a63e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c52ce929d6df1770f171310251d30e2d1940a63e"}]},"branch":"refs/heads/master"},"3b9f4caae8715e34afe9aa73986aaf2a2a354826":{"kind":"REWORK","_number":2,"created":"2019-04-17 17:43:19.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/88/653488/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/653488/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/653488/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/653488/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/653488/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/653488/2"}}},"commit":{"parents":[{"commit":"66656bceb6657ac0a4d0d1ce86b6ac2925cf043e","subject":"Merge \"Update documentation about rotating amphorae\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66656bceb6657ac0a4d0d1ce86b6ac2925cf043e"}]}],"author":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-04-17 16:22:28.000000000","tz":-300},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-04-17 17:43:13.000000000","tz":-300},"subject":"Adding changes to support ppc64le amphora builds with CentOS","message":"Adding changes to support ppc64le amphora builds with CentOS\n\nUpdated diskimage create script to include an argument for disabling the temporary filesystem.\nUpdated diskimage create to support ppc64le as an argument.\nUpdated backports to support a properly configured PaaS repository for CentOS on power.\n\nChange-Id: I8897278b1ac8b76d564f45bd0c7cbc26b29a8e5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b9f4caae8715e34afe9aa73986aaf2a2a354826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b9f4caae8715e34afe9aa73986aaf2a2a354826"}]},"branch":"refs/heads/master"},"a525a469ca35334c66cea7d8590995187588edac":{"kind":"REWORK","_number":3,"created":"2019-04-17 21:19:47.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/88/653488/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/653488/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/653488/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/653488/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/653488/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/653488/3"}}},"commit":{"parents":[{"commit":"66656bceb6657ac0a4d0d1ce86b6ac2925cf043e","subject":"Merge \"Update documentation about rotating amphorae\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66656bceb6657ac0a4d0d1ce86b6ac2925cf043e"}]}],"author":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-04-17 16:22:28.000000000","tz":-300},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-04-17 21:19:41.000000000","tz":-300},"subject":"Adding changes to support ppc64le amphora builds with CentOS","message":"Adding changes to support ppc64le amphora builds with CentOS\n\nUpdated diskimage create script to include an argument for disabling the temporary filesystem.\nUpdated diskimage create to support ppc64le as an argument.\nUpdated backports to support a properly configured PaaS repository for CentOS on power.\n\nChange-Id: I8897278b1ac8b76d564f45bd0c7cbc26b29a8e5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a525a469ca35334c66cea7d8590995187588edac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a525a469ca35334c66cea7d8590995187588edac"}]},"branch":"refs/heads/master"},"b236fdb71271726bffc14400d4ce854df71d1e18":{"kind":"REWORK","_number":4,"created":"2019-04-18 14:30:30.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/88/653488/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/653488/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/653488/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/653488/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/653488/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/653488/4"}}},"commit":{"parents":[{"commit":"66656bceb6657ac0a4d0d1ce86b6ac2925cf043e","subject":"Merge \"Update documentation about rotating amphorae\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66656bceb6657ac0a4d0d1ce86b6ac2925cf043e"}]}],"author":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-04-17 16:22:28.000000000","tz":-300},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-04-18 14:30:22.000000000","tz":-300},"subject":"Adding changes to support ppc64le amphora builds with CentOS","message":"Adding changes to support ppc64le amphora builds with CentOS\n\nUpdated diskimage create script to include an argument for disabling the temporary filesystem.\nUpdated diskimage create to support ppc64le as an argument.\nUpdated backports to support a properly configured PaaS repository for CentOS on power.\n\nChange-Id: I8897278b1ac8b76d564f45bd0c7cbc26b29a8e5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b236fdb71271726bffc14400d4ce854df71d1e18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b236fdb71271726bffc14400d4ce854df71d1e18"}]},"branch":"refs/heads/master"},"4c9097c681c937531132456a5f7443175b591b17":{"kind":"REWORK","_number":5,"created":"2019-04-18 16:11:19.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/88/653488/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/653488/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/653488/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/653488/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/653488/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/653488/5"}}},"commit":{"parents":[{"commit":"66656bceb6657ac0a4d0d1ce86b6ac2925cf043e","subject":"Merge \"Update documentation about rotating amphorae\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66656bceb6657ac0a4d0d1ce86b6ac2925cf043e"}]}],"author":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-04-17 16:22:28.000000000","tz":-300},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-04-18 16:11:09.000000000","tz":-300},"subject":"Adding changes to support ppc64le amphora builds with CentOS","message":"Adding changes to support ppc64le amphora builds with CentOS\n\nUpdated diskimage create script to include an argument for disabling the temporary filesystem.\nUpdated diskimage create to support ppc64le as an argument.\nUpdated backports to support a properly configured PaaS repository for CentOS on power.\n\nChange-Id: I8897278b1ac8b76d564f45bd0c7cbc26b29a8e5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c9097c681c937531132456a5f7443175b591b17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c9097c681c937531132456a5f7443175b591b17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
