)]}'
{"id":"openstack%2Foctavia~653872","triplet_id":"openstack%2Foctavia~master~Idae8cfc6f5f2884631cf1070b0c931781b17b3a9","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Idae8cfc6f5f2884631cf1070b0c931781b17b3a9","subject":"Handle non-graceful shutdown","status":"ABANDONED","created":"2019-04-18 22:28:06.000000000","updated":"2019-05-07 21:23:47.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"74ce10485b35a0c1c14ec9a21969bd115a2abd9f","_number":653872,"virtual_id_number":653872,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-04-24 15:21:55.000000000","_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":-1,"date":"2019-04-23 01:17:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-04-22 23:05:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 13:29:23.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-04-23 01:17:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-24 15:21:55.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"1e527e3454aa9621fd294f6ea36bac0e0d2fb686","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-18 22:28:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0c38c6321c4deed0113c429e48eef8a7fc1b2a7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-18 22:28:38.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"71bbe1320d7980f073e95dcf3fad7237d34d4208","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-18 22:30:22.000000000","message":"Patch Set 1:\n\nStill needs work around amphroa failover.\nNeeds some test cleanup.","accounts_in_message":[],"_revision_number":1},{"id":"ac661ee885b8951c4b5d4fdc354cdf07c99d9578","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-18 22:40:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce1df68216368632ae9aac852b92184e32942aa2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 02:51:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/72/653872/1/check/openstack-tox-cover/112bd98/cover/ : SUCCESS in 10m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/72/653872/1/check/openstack-tox-lower-constraints/51c80db/ : SUCCESS in 6m 29s\n- openstack-tox-pep8 http://logs.openstack.org/72/653872/1/check/openstack-tox-pep8/75ead00/ : SUCCESS in 9m 38s\n- openstack-tox-py27 http://logs.openstack.org/72/653872/1/check/openstack-tox-py27/cc91377/ : SUCCESS in 7m 18s\n- openstack-tox-py35 http://logs.openstack.org/72/653872/1/check/openstack-tox-py35/c2047d4/ : SUCCESS in 6m 18s\n- openstack-tox-py36 http://logs.openstack.org/72/653872/1/check/openstack-tox-py36/2cf9c82/ : SUCCESS in 8m 44s\n- openstack-tox-docs http://logs.openstack.org/72/653872/1/check/openstack-tox-docs/71eefec/html/ : SUCCESS in 9m 09s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/72/653872/1/check/neutron-lbaasv2-dsvm-api/dc38749/ : SUCCESS in 1h 35m 15s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/72/653872/1/check/neutron-lbaasv2-dsvm-py3x-api/d127f4b/ : SUCCESS in 1h 33m 14s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/72/653872/1/check/neutron-lbaasv2-dsvm-scenario/949c967/ : FAILURE in 1h 16m 46s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/72/653872/1/check/neutron-lbaasv2-dsvm-py3x-scenario/bc9cabf/ : SUCCESS in 1h 29m 05s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/72/653872/1/check/openstack-tox-functional/af89574/ : SUCCESS in 10m 03s\n- openstack-tox-functional-py35 http://logs.openstack.org/72/653872/1/check/openstack-tox-functional-py35/f78b3e9/ : SUCCESS in 9m 41s\n- openstack-tox-functional-py36 http://logs.openstack.org/72/653872/1/check/openstack-tox-functional-py36/fbbdc78/ : SUCCESS in 10m 51s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/72/653872/1/check/octavia-v1-dsvm-scenario/2db75d4/ : SUCCESS in 1h 11m 35s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/72/653872/1/check/octavia-v1-dsvm-py3x-scenario/5e5a4db/ : SUCCESS in 1h 12m 57s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/72/653872/1/check/octavia-v2-dsvm-noop-api/cf003b8/ : SUCCESS in 48m 06s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/72/653872/1/check/octavia-v2-dsvm-noop-py2-api/51749f2/ : SUCCESS in 57m 22s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/72/653872/1/check/octavia-v2-dsvm-scenario/63cf221/ : SUCCESS in 1h 24m 43s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/72/653872/1/check/octavia-v2-dsvm-py2-scenario/c65c5b2/ : SUCCESS in 1h 32m 46s\n- octavia-v2-dsvm-py2-scenario-centos-7 finger://ze01.openstack.org/3aa1f52f0dad4801a12270ff24e38b1e : RETRY_LIMIT in 30m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic finger://ze08.openstack.org/9dc229d1b2f34da887f638845094b4e4 : RETRY_LIMIT in 22m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/72/653872/1/check/octavia-v2-dsvm-scenario-two-node/f15b0a3/ : FAILURE in 33m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node finger://ze08.openstack.org/8a07dabf893948db81207479f4080add : RETRY_LIMIT in 30m 40s (non-voting)\n- octavia-grenade http://logs.openstack.org/72/653872/1/check/octavia-grenade/217238c/ : SUCCESS in 1h 35m 17s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/72/653872/1/check/octavia-v2-dsvm-tls-barbican/8e383f4/ : POST_FAILURE in 1h 13m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e800f27b5d73d28ad4f63135ee4d156a370b074b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-22 22:57:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0dd0c942f5490c5411cbf99883e0158cee5a23e7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-22 23:05:55.000000000","message":"Patch Set 2: Workflow-1\n\nStill need to look at the failover paths","accounts_in_message":[],"_revision_number":2},{"id":"c8224e7ead24c9561cbd777944c30e6ede73df90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 01:17:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/72/653872/2/check/openstack-tox-cover/d56864b/cover/ : SUCCESS in 8m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/72/653872/2/check/openstack-tox-lower-constraints/3ac6649/ : SUCCESS in 7m 49s\n- openstack-tox-pep8 http://logs.openstack.org/72/653872/2/check/openstack-tox-pep8/d409fea/ : SUCCESS in 7m 09s\n- openstack-tox-py27 http://logs.openstack.org/72/653872/2/check/openstack-tox-py27/9af1ef9/ : SUCCESS in 5m 41s\n- openstack-tox-py35 http://logs.openstack.org/72/653872/2/check/openstack-tox-py35/4fc8fa8/ : SUCCESS in 5m 39s\n- openstack-tox-py36 http://logs.openstack.org/72/653872/2/check/openstack-tox-py36/380b81d/ : SUCCESS in 6m 48s\n- openstack-tox-docs http://logs.openstack.org/72/653872/2/check/openstack-tox-docs/75cfde5/html/ : SUCCESS in 6m 43s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/72/653872/2/check/neutron-lbaasv2-dsvm-api/9812916/ : SUCCESS in 1h 21m 27s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/72/653872/2/check/neutron-lbaasv2-dsvm-py3x-api/0b00dd3/ : SUCCESS in 1h 14m 05s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/72/653872/2/check/neutron-lbaasv2-dsvm-scenario/9d93772/ : SUCCESS in 1h 32m 47s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/72/653872/2/check/neutron-lbaasv2-dsvm-py3x-scenario/bd86393/ : SUCCESS in 1h 35m 24s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/72/653872/2/check/openstack-tox-functional/918da73/ : SUCCESS in 6m 58s\n- openstack-tox-functional-py35 http://logs.openstack.org/72/653872/2/check/openstack-tox-functional-py35/15d30e4/ : SUCCESS in 9m 07s\n- openstack-tox-functional-py36 http://logs.openstack.org/72/653872/2/check/openstack-tox-functional-py36/91f901f/ : SUCCESS in 7m 16s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/72/653872/2/check/octavia-v1-dsvm-scenario/b1f2916/ : SUCCESS in 1h 11m 45s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/72/653872/2/check/octavia-v1-dsvm-py3x-scenario/d5fe2ac/ : FAILURE in 13m 37s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/72/653872/2/check/octavia-v2-dsvm-noop-api/c541f75/ : SUCCESS in 48m 33s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/72/653872/2/check/octavia-v2-dsvm-noop-py2-api/53f3080/ : SUCCESS in 41m 51s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/72/653872/2/check/octavia-v2-dsvm-scenario/130e336/ : SUCCESS in 1h 46m 31s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/72/653872/2/check/octavia-v2-dsvm-py2-scenario/bdee84d/ : SUCCESS in 1h 33m 36s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/72/653872/2/check/octavia-v2-dsvm-py2-scenario-centos-7/0da9bd4/ : FAILURE in 38m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/72/653872/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/c35980d/ : SUCCESS in 1h 47m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/72/653872/2/check/octavia-v2-dsvm-scenario-two-node/b763b73/ : SUCCESS in 2h 05m 13s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/72/653872/2/check/octavia-v2-dsvm-py2-scenario-two-node/105becb/ : SUCCESS in 1h 51m 49s (non-voting)\n- octavia-grenade http://logs.openstack.org/72/653872/2/check/octavia-grenade/fc26682/ : FAILURE in 39m 55s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/72/653872/2/check/octavia-v2-dsvm-tls-barbican/cce5a80/ : SUCCESS in 50m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9875e91bfe1d5ef2a9c2ff37a45bd6e69d156d2e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-04-24 15:21:55.000000000","message":"Patch Set 2:\n\nIf we are going to have flow resumption sooner or later, do we need this change at all? As this change adds some complexity(unfortunately, complexity means possible races, etc) and with flows resumption will be superseded. So, it will be cleaned up later or what is the long-term plan here?\nI hope my comment does not sound offending, I just try to understand the big picture.","accounts_in_message":[],"_revision_number":2},{"id":"74ce10485b35a0c1c14ec9a21969bd115a2abd9f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-07 21:23:47.000000000","message":"Abandoned\n\nAbandoning in favor of jobboard work","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c317a957133e15c0f858bfa85ee4a3540c850af3","revisions":{"95b3edbc6a405991656cff6ff5b6ac5e41f7b32e":{"kind":"REWORK","_number":1,"created":"2019-04-18 22:28:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/72/653872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/653872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/653872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/653872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/653872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/653872/1"}}},"commit":{"parents":[{"commit":"b274355e8ecedee8eb89bd90e4f417bbe68f77dc","subject":"Constraint networkx to \u003c2.3 for Python 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b274355e8ecedee8eb89bd90e4f417bbe68f77dc"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-04-18 22:26:34.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-04-18 22:28:05.000000000","tz":-420},"subject":"Handle non-graceful shutdown","message":"Handle non-graceful shutdown\n\nThis patch provides an interim (until we have flow resumption) fix\nto objects getting stuck in PENDING_* states when the processes are\nnot gracefully shutdown (power loss, kill -9, etc.).\n\nChange-Id: Idae8cfc6f5f2884631cf1070b0c931781b17b3a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95b3edbc6a405991656cff6ff5b6ac5e41f7b32e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95b3edbc6a405991656cff6ff5b6ac5e41f7b32e"}]},"branch":"refs/heads/master"},"c317a957133e15c0f858bfa85ee4a3540c850af3":{"kind":"REWORK","_number":2,"created":"2019-04-22 22:57:16.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/72/653872/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/653872/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/653872/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/653872/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/653872/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/653872/2"}}},"commit":{"parents":[{"commit":"b274355e8ecedee8eb89bd90e4f417bbe68f77dc","subject":"Constraint networkx to \u003c2.3 for Python 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b274355e8ecedee8eb89bd90e4f417bbe68f77dc"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-04-18 22:26:34.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-04-22 22:57:04.000000000","tz":-420},"subject":"Handle non-graceful shutdown","message":"Handle non-graceful shutdown\n\nThis patch provides an interim (until we have flow resumption) fix\nto objects getting stuck in PENDING_* states when the processes are\nnot gracefully shutdown (power loss, kill -9, etc.).\n\nChange-Id: Idae8cfc6f5f2884631cf1070b0c931781b17b3a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c317a957133e15c0f858bfa85ee4a3540c850af3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c317a957133e15c0f858bfa85ee4a3540c850af3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
