)]}'
{"id":"openstack%2Foctavia~655042","triplet_id":"openstack%2Foctavia~master~I291eb706df84489398b4f4b2736d7c5ab2e84747","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I291eb706df84489398b4f4b2736d7c5ab2e84747","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 11:39:24.000000000","updated":"2019-05-14 01:58:30.000000000","submitted":"2019-05-14 01:58:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655042-1557799110482-c6786d84","meta_rev_id":"6731d6936cccc45f5d1ec9be04d1a76cecbf0dea","_number":655042,"virtual_id_number":655042,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-05-14 01:58:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":2,"date":"2019-05-14 00:19:41.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-05-13 20:07:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":1,"date":"2019-05-14 00:19:41.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 03:34:39.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-05-13 20:07:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-05-14 00:19:41.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-05-14 01:58:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d9ef14d85ca3517b687c61b810910d9d54b6952","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 11:39:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22eebc4b47ee2c23e199166085b26a29f9cad11b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 15:40:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/42/655042/1/check/test-release-openstack/bc396c2/ : SUCCESS in 5m 16s\n- openstack-tox-cover http://logs.openstack.org/42/655042/1/check/openstack-tox-cover/06c5743/cover/ : SUCCESS in 9m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/42/655042/1/check/openstack-tox-lower-constraints/e5503d5/ : SUCCESS in 8m 55s\n- openstack-tox-pep8 http://logs.openstack.org/42/655042/1/check/openstack-tox-pep8/b8b582d/ : SUCCESS in 7m 49s\n- openstack-tox-py27 http://logs.openstack.org/42/655042/1/check/openstack-tox-py27/7c6b39a/ : SUCCESS in 6m 08s\n- openstack-tox-py35 http://logs.openstack.org/42/655042/1/check/openstack-tox-py35/aeacf25/ : SUCCESS in 5m 50s\n- openstack-tox-py36 http://logs.openstack.org/42/655042/1/check/openstack-tox-py36/498d4f1/ : SUCCESS in 6m 37s\n- openstack-tox-docs http://logs.openstack.org/42/655042/1/check/openstack-tox-docs/8a0905f/html/ : SUCCESS in 7m 58s\n- build-openstack-releasenotes http://logs.openstack.org/42/655042/1/check/build-openstack-releasenotes/6a3e376/html/ : SUCCESS in 6m 02s\n- neutron-lbaasv2-dsvm-api finger://ze03.openstack.org/7d3e388e9c574cc69a38b2f9c3b1f967 : RETRY_LIMIT in 48m 01s\n- neutron-lbaasv2-dsvm-py3x-api finger://ze06.openstack.org/1e6534b8332945a8a96696edf2b60cb2 : RETRY_LIMIT in 56m 04s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/42/655042/1/check/neutron-lbaasv2-dsvm-scenario/5f6df67/ : FAILURE in 34m 22s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/42/655042/1/check/neutron-lbaasv2-dsvm-py3x-scenario/df9492d/ : FAILURE in 35m 27s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/42/655042/1/check/openstack-tox-functional/fb354a0/ : SUCCESS in 7m 59s\n- openstack-tox-functional-py35 http://logs.openstack.org/42/655042/1/check/openstack-tox-functional-py35/213390a/ : SUCCESS in 10m 01s\n- openstack-tox-functional-py36 http://logs.openstack.org/42/655042/1/check/openstack-tox-functional-py36/797683b/ : SUCCESS in 10m 19s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/42/655042/1/check/octavia-v1-dsvm-scenario/c4cde0b/ : FAILURE in 36m 44s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/42/655042/1/check/octavia-v1-dsvm-py3x-scenario/5cc6ae8/ : FAILURE in 34m 44s\n- octavia-v2-dsvm-noop-api finger://ze05.openstack.org/97f802ab62684fc4a6df3d64b2b88f00 : RETRY_LIMIT in 46m 21s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/42/655042/1/check/octavia-v2-dsvm-noop-py2-api/2fd0e39/ : SUCCESS in 44m 14s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/42/655042/1/check/octavia-v2-dsvm-scenario/3fa2b7a/ : FAILURE in 33m 20s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/42/655042/1/check/octavia-v2-dsvm-py2-scenario/1424056/ : FAILURE in 30m 27s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/42/655042/1/check/octavia-v2-dsvm-py2-scenario-centos-7/979f580/ : POST_FAILURE in 38m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/42/655042/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/5184f6a/ : FAILURE in 30m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/42/655042/1/check/octavia-v2-dsvm-scenario-two-node/0b13a28/ : POST_FAILURE in 1h 05m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/42/655042/1/check/octavia-v2-dsvm-py2-scenario-two-node/6545013/ : FAILURE in 39m 39s (non-voting)\n- octavia-grenade finger://ze03.openstack.org/9cd8b4ecf4dc4c57805bab4dc372d7f4 : RETRY_LIMIT in 44m 41s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/42/655042/1/check/octavia-v2-dsvm-tls-barbican/aeda49b/ : FAILURE in 38m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"edae3b59cbd6c95e72c5761f4d2806ca8909d056","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-24 03:34:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"814cbeb99ae13afb027a1eb283b806b288b96e84","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-24 07:01:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ac6359adc8a313b9aef204d617870c6739a07cbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 09:09:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/42/655042/2/check/test-release-openstack/35599f6/ : SUCCESS in 3m 19s\n- openstack-tox-cover http://logs.openstack.org/42/655042/2/check/openstack-tox-cover/5e57120/cover/ : SUCCESS in 10m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/42/655042/2/check/openstack-tox-lower-constraints/a94f764/ : SUCCESS in 8m 06s\n- openstack-tox-pep8 http://logs.openstack.org/42/655042/2/check/openstack-tox-pep8/e775484/ : SUCCESS in 7m 01s\n- openstack-tox-py27 http://logs.openstack.org/42/655042/2/check/openstack-tox-py27/5ceadea/ : SUCCESS in 5m 18s\n- openstack-tox-py35 http://logs.openstack.org/42/655042/2/check/openstack-tox-py35/b0583d3/ : SUCCESS in 6m 44s\n- openstack-tox-py36 http://logs.openstack.org/42/655042/2/check/openstack-tox-py36/ca67133/ : SUCCESS in 5m 47s\n- openstack-tox-docs http://logs.openstack.org/42/655042/2/check/openstack-tox-docs/5b476c3/html/ : SUCCESS in 6m 27s\n- build-openstack-releasenotes http://logs.openstack.org/42/655042/2/check/build-openstack-releasenotes/8a43d44/html/ : SUCCESS in 5m 14s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/42/655042/2/check/neutron-lbaasv2-dsvm-api/651b61c/ : SUCCESS in 1h 19m 26s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/42/655042/2/check/neutron-lbaasv2-dsvm-py3x-api/c28a249/ : SUCCESS in 1h 25m 16s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/42/655042/2/check/neutron-lbaasv2-dsvm-scenario/29de4e8/ : FAILURE in 28m 27s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/42/655042/2/check/neutron-lbaasv2-dsvm-py3x-scenario/e782d59/ : SUCCESS in 1h 17m 08s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/42/655042/2/check/openstack-tox-functional/b099d45/ : SUCCESS in 6m 54s\n- openstack-tox-functional-py35 http://logs.openstack.org/42/655042/2/check/openstack-tox-functional-py35/e693345/ : SUCCESS in 7m 06s\n- openstack-tox-functional-py36 http://logs.openstack.org/42/655042/2/check/openstack-tox-functional-py36/026b00b/ : SUCCESS in 6m 56s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/42/655042/2/check/octavia-v1-dsvm-scenario/648c1fe/ : SUCCESS in 57m 35s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/42/655042/2/check/octavia-v1-dsvm-py3x-scenario/6774887/ : SUCCESS in 1h 03m 38s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/42/655042/2/check/octavia-v2-dsvm-noop-api/3de8bee/ : SUCCESS in 38m 17s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/42/655042/2/check/octavia-v2-dsvm-noop-py2-api/5fe94f8/ : SUCCESS in 43m 58s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/42/655042/2/check/octavia-v2-dsvm-scenario/4b15b7e/ : SUCCESS in 1h 16m 44s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/42/655042/2/check/octavia-v2-dsvm-py2-scenario/923fd88/ : SUCCESS in 1h 23m 50s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/42/655042/2/check/octavia-v2-dsvm-py2-scenario-centos-7/5e1665c/ : FAILURE in 45m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/42/655042/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/d878a5b/ : SUCCESS in 1h 30m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/42/655042/2/check/octavia-v2-dsvm-scenario-two-node/79986f5/ : SUCCESS in 1h 40m 40s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/42/655042/2/check/octavia-v2-dsvm-py2-scenario-two-node/448f959/ : SUCCESS in 1h 33m 23s (non-voting)\n- octavia-grenade http://logs.openstack.org/42/655042/2/check/octavia-grenade/e8d43e4/ : FAILURE in 2h 03m 19s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/42/655042/2/check/octavia-v2-dsvm-tls-barbican/57a37e8/ : SUCCESS in 57m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f79d3e266ab2d2dfe66c445f345766a1f29b3b3f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-13 19:59:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"6c1119c6d8a5d227c59ce8174c270612e0e3596b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-13 20:07:53.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"694da026cd51ea166dfcb06ae46299e40e31feba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 22:18:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/42/655042/3/check/test-release-openstack/db7ef18/ : SUCCESS in 8m 54s\n- openstack-tox-cover http://logs.openstack.org/42/655042/3/check/openstack-tox-cover/8ee3492/cover/ : SUCCESS in 17m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/42/655042/3/check/openstack-tox-lower-constraints/b410ea7/ : SUCCESS in 8m 03s\n- openstack-tox-pep8 http://logs.openstack.org/42/655042/3/check/openstack-tox-pep8/c19ef21/ : SUCCESS in 6m 51s\n- openstack-tox-py27 http://logs.openstack.org/42/655042/3/check/openstack-tox-py27/9aa5396/ : SUCCESS in 6m 39s\n- openstack-tox-py36 http://logs.openstack.org/42/655042/3/check/openstack-tox-py36/bcfd131/ : SUCCESS in 5m 51s\n- openstack-tox-docs http://logs.openstack.org/42/655042/3/check/openstack-tox-docs/f465ef9/html/ : SUCCESS in 7m 14s\n- build-openstack-releasenotes http://logs.openstack.org/42/655042/3/check/build-openstack-releasenotes/557d60f/html/ : SUCCESS in 12m 17s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/42/655042/3/check/neutron-lbaasv2-dsvm-api/e0c5154/ : SUCCESS in 1h 18m 56s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/42/655042/3/check/neutron-lbaasv2-dsvm-py3x-api/3427754/ : SUCCESS in 1h 11m 30s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/42/655042/3/check/neutron-lbaasv2-dsvm-scenario/ed44cab/ : SUCCESS in 1h 34m 53s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/42/655042/3/check/neutron-lbaasv2-dsvm-py3x-scenario/def89aa/ : FAILURE in 44m 31s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/42/655042/3/check/openstack-tox-functional/5be12bd/ : SUCCESS in 7m 26s\n- openstack-tox-functional-py36 http://logs.openstack.org/42/655042/3/check/openstack-tox-functional-py36/8d2eaad/ : SUCCESS in 7m 05s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/42/655042/3/check/octavia-v1-dsvm-scenario/868e3ec/ : SUCCESS in 1h 13m 48s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/42/655042/3/check/octavia-v1-dsvm-py3x-scenario/f5a5bba/ : SUCCESS in 1h 25m 44s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/42/655042/3/check/octavia-v2-dsvm-noop-api/a046d04/ : SUCCESS in 43m 59s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/42/655042/3/check/octavia-v2-dsvm-noop-py2-api/8b74ac8/ : SUCCESS in 48m 44s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/42/655042/3/check/octavia-v2-dsvm-scenario/76e8466/ : SUCCESS in 1h 30m 34s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/42/655042/3/check/octavia-v2-dsvm-py2-scenario/bc2c048/ : SUCCESS in 1h 34m 38s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/42/655042/3/check/octavia-v2-dsvm-py2-scenario-centos-7/8e8de21/ : FAILURE in 2h 10m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/42/655042/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/668c35f/ : SUCCESS in 1h 45m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/42/655042/3/check/octavia-v2-dsvm-scenario-two-node/cf5653f/ : SUCCESS in 1h 48m 42s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/42/655042/3/check/octavia-v2-dsvm-py2-scenario-two-node/5210f31/ : SUCCESS in 1h 47m 19s (non-voting)\n- octavia-grenade http://logs.openstack.org/42/655042/3/check/octavia-grenade/a5c587b/ : SUCCESS in 1h 35m 11s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/42/655042/3/check/octavia-v2-dsvm-tls-barbican/ad5e703/ : SUCCESS in 55m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"71fd51edff8482001fece5d71391e7e944742d6e","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-05-14 00:19:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"eb16b8a9753ffd4fb7c70ec8d2db1af1742d5d89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 00:19:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a90aad315ab022b95fb8c5dd5d365b2a9d61d430","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 01:58:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/42/655042/3/gate/test-release-openstack/bcb3561/ : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/42/655042/3/gate/openstack-tox-lower-constraints/16aeb92/ : SUCCESS in 7m 30s\n- openstack-tox-pep8 http://logs.openstack.org/42/655042/3/gate/openstack-tox-pep8/ce5f722/ : SUCCESS in 6m 02s\n- openstack-tox-py27 http://logs.openstack.org/42/655042/3/gate/openstack-tox-py27/063d4e1/ : SUCCESS in 6m 07s\n- openstack-tox-py36 http://logs.openstack.org/42/655042/3/gate/openstack-tox-py36/a403c3f/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/42/655042/3/gate/openstack-tox-docs/5b624a6/html/ : SUCCESS in 5m 54s\n- build-openstack-releasenotes http://logs.openstack.org/42/655042/3/gate/build-openstack-releasenotes/6f3468c/html/ : SUCCESS in 4m 59s\n- openstack-tox-functional http://logs.openstack.org/42/655042/3/gate/openstack-tox-functional/a725c9d/ : SUCCESS in 7m 23s\n- openstack-tox-functional-py36 http://logs.openstack.org/42/655042/3/gate/openstack-tox-functional-py36/c3a8fea/ : SUCCESS in 8m 13s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/42/655042/3/gate/octavia-v2-dsvm-noop-api/278a78f/ : SUCCESS in 41m 15s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/42/655042/3/gate/octavia-v2-dsvm-noop-py2-api/1606511/ : SUCCESS in 44m 53s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/42/655042/3/gate/octavia-v2-dsvm-scenario/fc6f3d8/ : SUCCESS in 1h 26m 51s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/42/655042/3/gate/octavia-v2-dsvm-py2-scenario/019c419/ : SUCCESS in 1h 33m 12s\n- octavia-grenade http://logs.openstack.org/42/655042/3/gate/octavia-grenade/e587004/ : SUCCESS in 1h 16m 17s","accounts_in_message":[],"_revision_number":3},{"id":"6731d6936cccc45f5d1ec9be04d1a76cecbf0dea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 01:58:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1910f024a395908085ef7e0d8673110eb9a2eaeb","revisions":{"9a3358bc1214825b2b470094bfd423440cc3c5a5":{"kind":"REWORK","_number":1,"created":"2019-04-23 11:39:24.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/42/655042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/655042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/655042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/655042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/655042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/655042/1"}}},"commit":{"parents":[{"commit":"fd1a7e50848dda8381705ae2a1f70530797cd8da","subject":"Merge \"Adding changes to support ppc64le amphora builds with CentOS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd1a7e50848dda8381705ae2a1f70530797cd8da"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:00.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:21.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I291eb706df84489398b4f4b2736d7c5ab2e84747\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a3358bc1214825b2b470094bfd423440cc3c5a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a3358bc1214825b2b470094bfd423440cc3c5a5"}]},"branch":"refs/heads/master"},"2b1a4a427b4579a6bd05b50162a9f1a6cfe32547":{"kind":"REWORK","_number":2,"created":"2019-04-24 07:01:35.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/42/655042/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/655042/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/655042/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/655042/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/655042/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/655042/2"}}},"commit":{"parents":[{"commit":"fd1a7e50848dda8381705ae2a1f70530797cd8da","subject":"Merge \"Adding changes to support ppc64le amphora builds with CentOS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd1a7e50848dda8381705ae2a1f70530797cd8da"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:00.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-24 07:01:12.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I291eb706df84489398b4f4b2736d7c5ab2e84747\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2b1a4a427b4579a6bd05b50162a9f1a6cfe32547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2b1a4a427b4579a6bd05b50162a9f1a6cfe32547"}]},"branch":"refs/heads/master"},"1910f024a395908085ef7e0d8673110eb9a2eaeb":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-13 19:59:59.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/42/655042/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/655042/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/655042/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/655042/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/655042/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/655042/3"}}},"commit":{"parents":[{"commit":"28fe8ed45e237a6f6e44ad4d337f723303d37e17","subject":"Merge \"Ignore .vscode directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28fe8ed45e237a6f6e44ad4d337f723303d37e17"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:00.000000000","tz":480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-13 19:59:59.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I291eb706df84489398b4f4b2736d7c5ab2e84747\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1910f024a395908085ef7e0d8673110eb9a2eaeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1910f024a395908085ef7e0d8673110eb9a2eaeb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
