)]}'
{"id":"openstack%2Foctavia~655245","triplet_id":"openstack%2Foctavia~master~I97fd1de040a8bf7706b78fe1825e48761c2073ee","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I97fd1de040a8bf7706b78fe1825e48761c2073ee","subject":"Removing an erroneous comment","status":"MERGED","created":"2019-04-23 20:32:36.000000000","updated":"2019-04-25 18:27:36.000000000","submitted":"2019-04-25 18:27:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655245-1556216856305-36e30e48","meta_rev_id":"9e54304a7605b4f2f568906bdc516ea501a44c0a","_number":655245,"virtual_id_number":655245,"owner":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2019-04-25 14:20:09.000000000","_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"value":2,"date":"2019-04-25 18:27:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-04-23 20:40:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-04-23 20:42:32.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-04-23 20:42:32.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 20:40:39.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-04-23 20:42:32.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-04-25 18:27:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"036732c3a6e2be9e2f079c8a4b9d0c5f9493b594","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-04-23 20:32:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"346dbdb3698fdb455937e5d0d808e9aa9a8fbda1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-04-23 20:40:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c03e18b0d73cf7ac94c5e895d942dd0bfcce19da","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-23 20:42:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8df0e66c6db1e29dbc8d643387ae8005517d55a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 00:01:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/45/655245/1/check/openstack-tox-cover/35e3257/cover/ : SUCCESS in 8m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/655245/1/check/openstack-tox-lower-constraints/3acb5d2/ : SUCCESS in 7m 40s\n- openstack-tox-pep8 http://logs.openstack.org/45/655245/1/check/openstack-tox-pep8/34edacb/ : SUCCESS in 6m 25s\n- openstack-tox-py27 http://logs.openstack.org/45/655245/1/check/openstack-tox-py27/0c57535/ : POST_FAILURE in 7m 38s\n- openstack-tox-py35 http://logs.openstack.org/45/655245/1/check/openstack-tox-py35/2317184/ : SUCCESS in 6m 06s\n- openstack-tox-py36 http://logs.openstack.org/45/655245/1/check/openstack-tox-py36/5e06b4f/ : SUCCESS in 5m 05s\n- openstack-tox-docs http://logs.openstack.org/45/655245/1/check/openstack-tox-docs/26565db/html/ : SUCCESS in 6m 50s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/45/655245/1/check/neutron-lbaasv2-dsvm-api/3914a79/ : SUCCESS in 1h 21m 01s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/45/655245/1/check/neutron-lbaasv2-dsvm-py3x-api/eb54098/ : SUCCESS in 1h 21m 49s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/45/655245/1/check/neutron-lbaasv2-dsvm-scenario/d1d41a9/ : SUCCESS in 1h 38m 27s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/45/655245/1/check/neutron-lbaasv2-dsvm-py3x-scenario/03e065e/ : SUCCESS in 1h 31m 26s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/45/655245/1/check/openstack-tox-functional/9ce22e2/ : SUCCESS in 7m 22s\n- openstack-tox-functional-py35 http://logs.openstack.org/45/655245/1/check/openstack-tox-functional-py35/2220f90/ : SUCCESS in 7m 36s\n- openstack-tox-functional-py36 http://logs.openstack.org/45/655245/1/check/openstack-tox-functional-py36/fc49375/ : SUCCESS in 10m 30s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/45/655245/1/check/octavia-v1-dsvm-scenario/e84527e/ : SUCCESS in 1h 08m 50s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/45/655245/1/check/octavia-v1-dsvm-py3x-scenario/f6b281a/ : SUCCESS in 52m 28s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/45/655245/1/check/octavia-v2-dsvm-noop-api/8652a55/ : SUCCESS in 42m 36s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/45/655245/1/check/octavia-v2-dsvm-noop-py2-api/a5283a9/ : SUCCESS in 1h 00m 44s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/45/655245/1/check/octavia-v2-dsvm-scenario/943ec4a/ : SUCCESS in 1h 39m 16s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/45/655245/1/check/octavia-v2-dsvm-py2-scenario/1dfd059/ : POST_FAILURE in 7m 50s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/45/655245/1/check/octavia-v2-dsvm-py2-scenario-centos-7/ec98e20/ : FAILURE in 37m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/45/655245/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/6e728b2/ : FAILURE in 41m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/45/655245/1/check/octavia-v2-dsvm-scenario-two-node/0c2d4fc/ : SUCCESS in 1h 58m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/45/655245/1/check/octavia-v2-dsvm-py2-scenario-two-node/57d7f2d/ : SUCCESS in 1h 26m 47s (non-voting)\n- octavia-grenade http://logs.openstack.org/45/655245/1/check/octavia-grenade/538f603/ : SUCCESS in 1h 14m 42s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/45/655245/1/check/octavia-v2-dsvm-tls-barbican/6878b91/ : SUCCESS in 59m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5fdc0d45ae58eb70c8ca69d5d8dc406fb2ae8f95","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-04-25 14:20:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cdc63c8d21b1154fa831a2bea5689278f0d5fd55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 16:35:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/45/655245/1/check/openstack-tox-cover/7ccfb2c/cover/ : SUCCESS in 9m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/655245/1/check/openstack-tox-lower-constraints/1a042fa/ : SUCCESS in 10m 41s\n- openstack-tox-pep8 http://logs.openstack.org/45/655245/1/check/openstack-tox-pep8/0e5965e/ : SUCCESS in 6m 36s\n- openstack-tox-py27 http://logs.openstack.org/45/655245/1/check/openstack-tox-py27/45637ba/ : SUCCESS in 5m 42s\n- openstack-tox-py35 http://logs.openstack.org/45/655245/1/check/openstack-tox-py35/264df5d/ : SUCCESS in 5m 37s\n- openstack-tox-py36 http://logs.openstack.org/45/655245/1/check/openstack-tox-py36/d7a7f36/ : SUCCESS in 5m 37s\n- openstack-tox-docs http://logs.openstack.org/45/655245/1/check/openstack-tox-docs/d0de5e3/html/ : SUCCESS in 6m 13s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/45/655245/1/check/neutron-lbaasv2-dsvm-api/df8183e/ : SUCCESS in 1h 26m 04s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/45/655245/1/check/neutron-lbaasv2-dsvm-py3x-api/a4df670/ : SUCCESS in 1h 48m 37s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/45/655245/1/check/neutron-lbaasv2-dsvm-scenario/052b771/ : FAILURE in 1h 14m 40s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/45/655245/1/check/neutron-lbaasv2-dsvm-py3x-scenario/71df065/ : FAILURE in 1h 31m 27s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/45/655245/1/check/openstack-tox-functional/cff091a/ : SUCCESS in 7m 18s\n- openstack-tox-functional-py35 http://logs.openstack.org/45/655245/1/check/openstack-tox-functional-py35/18d1366/ : SUCCESS in 7m 25s\n- openstack-tox-functional-py36 http://logs.openstack.org/45/655245/1/check/openstack-tox-functional-py36/b2d04aa/ : SUCCESS in 7m 58s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/45/655245/1/check/octavia-v1-dsvm-scenario/6443099/ : SUCCESS in 1h 01m 55s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/45/655245/1/check/octavia-v1-dsvm-py3x-scenario/405487c/ : SUCCESS in 1h 05m 52s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/45/655245/1/check/octavia-v2-dsvm-noop-api/7b649eb/ : SUCCESS in 45m 53s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/45/655245/1/check/octavia-v2-dsvm-noop-py2-api/ad17185/ : SUCCESS in 45m 51s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/45/655245/1/check/octavia-v2-dsvm-scenario/d340078/ : SUCCESS in 1h 36m 27s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/45/655245/1/check/octavia-v2-dsvm-py2-scenario/b86bc26/ : SUCCESS in 1h 31m 05s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/45/655245/1/check/octavia-v2-dsvm-py2-scenario-centos-7/6e26c50/ : FAILURE in 32m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/45/655245/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/3530544/ : SUCCESS in 1h 42m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/45/655245/1/check/octavia-v2-dsvm-scenario-two-node/d7bdf8d/ : SUCCESS in 1h 55m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/45/655245/1/check/octavia-v2-dsvm-py2-scenario-two-node/833cacb/ : SUCCESS in 1h 43m 02s (non-voting)\n- octavia-grenade http://logs.openstack.org/45/655245/1/check/octavia-grenade/f49c289/ : SUCCESS in 1h 29m 10s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/45/655245/1/check/octavia-v2-dsvm-tls-barbican/8bc19f8/ : SUCCESS in 54m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"09f9defe30b0d6cc1253794134ab5865ed6ba6c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 16:35:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4310a3141fba441e6433b13a2d06ba0c246be2f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 18:27:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/45/655245/1/gate/openstack-tox-lower-constraints/268102c/ : SUCCESS in 8m 16s\n- openstack-tox-pep8 http://logs.openstack.org/45/655245/1/gate/openstack-tox-pep8/329cb58/ : SUCCESS in 6m 31s\n- openstack-tox-py27 http://logs.openstack.org/45/655245/1/gate/openstack-tox-py27/57b1b5b/ : SUCCESS in 5m 11s\n- openstack-tox-py35 http://logs.openstack.org/45/655245/1/gate/openstack-tox-py35/7fdcb80/ : SUCCESS in 5m 41s\n- openstack-tox-py36 http://logs.openstack.org/45/655245/1/gate/openstack-tox-py36/df96f7f/ : SUCCESS in 5m 16s\n- openstack-tox-docs http://logs.openstack.org/45/655245/1/gate/openstack-tox-docs/d0c3039/html/ : SUCCESS in 6m 33s\n- openstack-tox-functional http://logs.openstack.org/45/655245/1/gate/openstack-tox-functional/b14f26c/ : SUCCESS in 7m 46s\n- openstack-tox-functional-py35 http://logs.openstack.org/45/655245/1/gate/openstack-tox-functional-py35/51009c5/ : SUCCESS in 7m 43s\n- openstack-tox-functional-py36 http://logs.openstack.org/45/655245/1/gate/openstack-tox-functional-py36/fc123fa/ : SUCCESS in 8m 23s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/45/655245/1/gate/octavia-v1-dsvm-scenario/ea55c5b/ : SUCCESS in 1h 04m 26s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/45/655245/1/gate/octavia-v1-dsvm-py3x-scenario/3330b3a/ : SUCCESS in 1h 17m 56s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/45/655245/1/gate/neutron-lbaasv2-dsvm-api/0b7d4f0/ : SUCCESS in 1h 17m 37s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/45/655245/1/gate/neutron-lbaasv2-dsvm-py3x-api/cf2b2d2/ : SUCCESS in 1h 22m 32s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/45/655245/1/gate/octavia-v2-dsvm-noop-api/dd921c2/ : SUCCESS in 47m 14s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/45/655245/1/gate/octavia-v2-dsvm-noop-py2-api/c3569f6/ : SUCCESS in 51m 52s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/45/655245/1/gate/octavia-v2-dsvm-scenario/b85b6b9/ : SUCCESS in 1h 43m 49s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/45/655245/1/gate/octavia-v2-dsvm-py2-scenario/4cc8659/ : SUCCESS in 1h 29m 28s\n- octavia-grenade http://logs.openstack.org/45/655245/1/gate/octavia-grenade/6a9f951/ : SUCCESS in 1h 16m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9e54304a7605b4f2f568906bdc516ea501a44c0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 18:27:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b1f3d949c6f326c0224abfa0f47c3b27cd847fe5","revisions":{"b1f3d949c6f326c0224abfa0f47c3b27cd847fe5":{"kind":"REWORK","_number":1,"created":"2019-04-23 20:32:36.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/45/655245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/45/655245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/655245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/655245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/655245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/45/655245/1"}}},"commit":{"parents":[{"commit":"fd1a7e50848dda8381705ae2a1f70530797cd8da","subject":"Merge \"Adding changes to support ppc64le amphora builds with CentOS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd1a7e50848dda8381705ae2a1f70530797cd8da"}]}],"author":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-04-23 20:31:47.000000000","tz":-300},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-04-23 20:31:47.000000000","tz":-300},"subject":"Removing an erroneous comment","message":"Removing an erroneous comment\n\nLeft a code-comment in a change by accident, updating to remove it.\n\nChange-Id: I97fd1de040a8bf7706b78fe1825e48761c2073ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b1f3d949c6f326c0224abfa0f47c3b27cd847fe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b1f3d949c6f326c0224abfa0f47c3b27cd847fe5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
