)]}'
{"id":"openstack%2Foctavia~655770","triplet_id":"openstack%2Foctavia~stable%2Frocky~I41c9bedd8a3452513af3d409fbacd65ea287f02a","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"I41c9bedd8a3452513af3d409fbacd65ea287f02a","subject":"Fix listener API handling of None/null updates","status":"MERGED","created":"2019-04-25 19:17:58.000000000","updated":"2019-04-26 11:42:10.000000000","submitted":"2019-04-26 11:42:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655770-1556278930496-0411a076","meta_rev_id":"106bd5b359b6f9dcda2316e4ae3781b47a18748e","_number":655770,"virtual_id_number":655770,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-04-26 11:42:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":2,"date":"2019-04-25 19:24:56.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2019-04-25 20:54:03.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":1,"date":"2019-04-25 21:39:26.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-25 19:24:56.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2019-04-25 21:39:26.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-04-26 11:42:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"28f3a4d019fddbb9ac920d039f9f4ede7be76cb1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-25 19:17:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ed2900b2dbe6c02101353e48339eb42bff3eeaa","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2019-04-25 19:24:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fcb22f8ab1118cbcbb9fb9a2e0b7cc325e36eb08","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-25 20:54:03.000000000","message":"Patch Set 1: Code-Review+2\n\nCI jobs still running and look okay thus far.","accounts_in_message":[],"_revision_number":1},{"id":"b8a6fc34891bf9e85fa2ebbe232cc6753cd0ca44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 21:21:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/70/655770/1/check/openstack-tox-lower-constraints/f914eb7/ : SUCCESS in 7m 08s\n- openstack-tox-pep8 http://logs.openstack.org/70/655770/1/check/openstack-tox-pep8/7f073fe/ : SUCCESS in 7m 58s\n- openstack-tox-py27 http://logs.openstack.org/70/655770/1/check/openstack-tox-py27/d533668/ : SUCCESS in 6m 14s\n- openstack-tox-py35 http://logs.openstack.org/70/655770/1/check/openstack-tox-py35/e747901/ : SUCCESS in 7m 30s\n- openstack-tox-docs http://logs.openstack.org/70/655770/1/check/openstack-tox-docs/14a9281/html/ : SUCCESS in 6m 07s\n- build-openstack-releasenotes http://logs.openstack.org/70/655770/1/check/build-openstack-releasenotes/7c80d49/html/ : SUCCESS in 5m 30s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/70/655770/1/check/neutron-lbaasv2-dsvm-api/28fb20d/ : SUCCESS in 1h 15m 29s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/70/655770/1/check/neutron-lbaasv2-dsvm-py3x-api/dc7d7d9/ : SUCCESS in 1h 14m 44s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/70/655770/1/check/neutron-lbaasv2-dsvm-scenario/e4bd7db/ : SUCCESS in 1h 28m 04s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/70/655770/1/check/neutron-lbaasv2-dsvm-py3x-scenario/590c1c4/ : SUCCESS in 1h 31m 05s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/70/655770/1/check/openstack-tox-functional/c31194f/ : SUCCESS in 6m 20s\n- openstack-tox-functional-py35 http://logs.openstack.org/70/655770/1/check/openstack-tox-functional-py35/97d71d9/ : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 http://logs.openstack.org/70/655770/1/check/openstack-tox-functional-py36/3f2e08a/ : SUCCESS in 9m 33s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/70/655770/1/check/octavia-v1-dsvm-scenario/2ff6253/ : SUCCESS in 1h 06m 13s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/70/655770/1/check/octavia-v1-dsvm-py3x-scenario/d27fd91/ : SUCCESS in 49m 30s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/70/655770/1/check/octavia-v1-dsvm-scenario-multinode/625939f/ : FAILURE in 1h 28m 08s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/70/655770/1/check/octavia-v1-dsvm-py3x-scenario-multinode/72bdc50/ : FAILURE in 1h 26m 07s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/70/655770/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/859c300/ : SUCCESS in 1h 11m 24s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/70/655770/1/check/octavia-v2-dsvm-noop-api/4842482/ : SUCCESS in 45m 08s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/70/655770/1/check/octavia-v2-dsvm-noop-py35-api/1d933d8/ : SUCCESS in 40m 52s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/70/655770/1/check/octavia-v2-dsvm-scenario/af27c69/ : SUCCESS in 1h 41m 36s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/70/655770/1/check/octavia-v2-dsvm-py35-scenario/c25e58e/ : SUCCESS in 1h 36m 09s (non-voting)\n- octavia-grenade http://logs.openstack.org/70/655770/1/check/octavia-grenade/d383555/ : SUCCESS in 1h 08m 13s\n- openstack-tox-cover http://logs.openstack.org/70/655770/1/check/openstack-tox-cover/536f22a/cover/ : SUCCESS in 9m 07s","accounts_in_message":[],"_revision_number":1},{"id":"94ff1083ba28ae1fcc65ad7418882249842c8c3b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-25 21:39:26.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"15177a44ff604d09551a0398dd445a7da44a757e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 21:39:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f2e69a226987984bb31210cb61dd698810321ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 23:13:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/70/655770/1/gate/openstack-tox-lower-constraints/f5d25a1/ : SUCCESS in 8m 24s\n- openstack-tox-pep8 http://logs.openstack.org/70/655770/1/gate/openstack-tox-pep8/6f0822d/ : SUCCESS in 7m 47s\n- openstack-tox-py27 http://logs.openstack.org/70/655770/1/gate/openstack-tox-py27/380eec1/ : SUCCESS in 6m 07s\n- openstack-tox-py35 http://logs.openstack.org/70/655770/1/gate/openstack-tox-py35/a726a34/ : SUCCESS in 6m 29s\n- openstack-tox-docs http://logs.openstack.org/70/655770/1/gate/openstack-tox-docs/35947c9/html/ : SUCCESS in 6m 12s\n- build-openstack-releasenotes http://logs.openstack.org/70/655770/1/gate/build-openstack-releasenotes/635841e/html/ : SUCCESS in 5m 09s\n- openstack-tox-functional http://logs.openstack.org/70/655770/1/gate/openstack-tox-functional/1e56cb4/ : SUCCESS in 7m 44s\n- openstack-tox-functional-py35 http://logs.openstack.org/70/655770/1/gate/openstack-tox-functional-py35/fbacb44/ : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 http://logs.openstack.org/70/655770/1/gate/openstack-tox-functional-py36/a8995af/ : SUCCESS in 6m 50s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/70/655770/1/gate/octavia-v1-dsvm-scenario/8e2b3d3/ : SUCCESS in 1h 04m 15s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/70/655770/1/gate/octavia-v1-dsvm-py3x-scenario/1c21b43/ : FAILURE in 43m 02s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/70/655770/1/gate/octavia-v1-dsvm-scenario-multinode/dd4b3b3/ : FAILURE in 1h 23m 22s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/70/655770/1/gate/octavia-v1-dsvm-py3x-scenario-multinode/bd36c04/ : FAILURE in 1h 28m 36s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/70/655770/1/gate/neutron-lbaasv2-dsvm-py3x-api/e386a9a/ : SUCCESS in 1h 23m 50s\n- octavia-grenade http://logs.openstack.org/70/655770/1/gate/octavia-grenade/9d3b80d/ : SUCCESS in 1h 11m 51s\n- openstack-tox-cover http://logs.openstack.org/70/655770/1/gate/openstack-tox-cover/e529f16/cover/ : SUCCESS in 9m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a3dc5683f320dc3f35af4c1bd26425bb81d7d3ce","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-26 07:08:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fb73c09bbfb1024b586ba20201ce5dd22fd90077","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 08:55:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/70/655770/1/check/openstack-tox-lower-constraints/56bcf9b/ : SUCCESS in 7m 54s\n- openstack-tox-pep8 http://logs.openstack.org/70/655770/1/check/openstack-tox-pep8/70ce588/ : SUCCESS in 8m 17s\n- openstack-tox-py27 http://logs.openstack.org/70/655770/1/check/openstack-tox-py27/b7c0bf1/ : SUCCESS in 7m 00s\n- openstack-tox-py35 http://logs.openstack.org/70/655770/1/check/openstack-tox-py35/f8747c7/ : SUCCESS in 6m 41s\n- openstack-tox-docs http://logs.openstack.org/70/655770/1/check/openstack-tox-docs/8576a60/html/ : SUCCESS in 6m 25s\n- build-openstack-releasenotes http://logs.openstack.org/70/655770/1/check/build-openstack-releasenotes/4eba6eb/html/ : SUCCESS in 5m 03s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/70/655770/1/check/neutron-lbaasv2-dsvm-api/0a9e51e/ : SUCCESS in 1h 16m 03s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/70/655770/1/check/neutron-lbaasv2-dsvm-py3x-api/d4fd90b/ : SUCCESS in 1h 15m 23s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/70/655770/1/check/neutron-lbaasv2-dsvm-scenario/723db02/ : SUCCESS in 1h 32m 49s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/70/655770/1/check/neutron-lbaasv2-dsvm-py3x-scenario/92802dd/ : SUCCESS in 1h 21m 13s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/70/655770/1/check/openstack-tox-functional/4ab819a/ : SUCCESS in 6m 29s\n- openstack-tox-functional-py35 http://logs.openstack.org/70/655770/1/check/openstack-tox-functional-py35/6bec54f/ : SUCCESS in 6m 47s\n- openstack-tox-functional-py36 http://logs.openstack.org/70/655770/1/check/openstack-tox-functional-py36/26d3999/ : SUCCESS in 6m 26s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/70/655770/1/check/octavia-v1-dsvm-scenario/f104477/ : SUCCESS in 1h 03m 58s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/70/655770/1/check/octavia-v1-dsvm-py3x-scenario/bc603db/ : SUCCESS in 1h 03m 53s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/70/655770/1/check/octavia-v1-dsvm-scenario-multinode/dbc5005/ : FAILURE in 1h 19m 45s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/70/655770/1/check/octavia-v1-dsvm-py3x-scenario-multinode/7fbd26c/ : FAILURE in 1h 18m 49s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/70/655770/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/5689d9b/ : SUCCESS in 1h 12m 20s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/70/655770/1/check/octavia-v2-dsvm-noop-api/b4b51f7/ : SUCCESS in 42m 38s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/70/655770/1/check/octavia-v2-dsvm-noop-py35-api/4385086/ : SUCCESS in 44m 54s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/70/655770/1/check/octavia-v2-dsvm-scenario/e474d40/ : SUCCESS in 1h 23m 02s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/70/655770/1/check/octavia-v2-dsvm-py35-scenario/251ada9/ : SUCCESS in 1h 33m 59s (non-voting)\n- octavia-grenade http://logs.openstack.org/70/655770/1/check/octavia-grenade/b7c2685/ : SUCCESS in 1h 17m 23s\n- openstack-tox-cover http://logs.openstack.org/70/655770/1/check/openstack-tox-cover/59756cc/cover/ : SUCCESS in 12m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d5506472e524ec19e52bc5c07c4316f13327c5ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 08:55:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5bfb3449df89905c00441bb954663c6cc34e34e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 11:42:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/70/655770/1/gate/openstack-tox-lower-constraints/b1456b7/ : SUCCESS in 11m 20s\n- openstack-tox-pep8 http://logs.openstack.org/70/655770/1/gate/openstack-tox-pep8/cc5b5af/ : SUCCESS in 7m 36s\n- openstack-tox-py27 http://logs.openstack.org/70/655770/1/gate/openstack-tox-py27/4ee9cfc/ : SUCCESS in 7m 26s\n- openstack-tox-py35 http://logs.openstack.org/70/655770/1/gate/openstack-tox-py35/71cf1b4/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/70/655770/1/gate/openstack-tox-docs/6dcea3b/html/ : SUCCESS in 6m 16s\n- build-openstack-releasenotes http://logs.openstack.org/70/655770/1/gate/build-openstack-releasenotes/7f80ff4/html/ : SUCCESS in 5m 08s\n- openstack-tox-functional http://logs.openstack.org/70/655770/1/gate/openstack-tox-functional/84b6faa/ : SUCCESS in 7m 34s\n- openstack-tox-functional-py35 http://logs.openstack.org/70/655770/1/gate/openstack-tox-functional-py35/cfb3d22/ : SUCCESS in 5m 45s\n- openstack-tox-functional-py36 http://logs.openstack.org/70/655770/1/gate/openstack-tox-functional-py36/481d537/ : SUCCESS in 7m 51s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/70/655770/1/gate/octavia-v1-dsvm-scenario/1460ae4/ : SUCCESS in 59m 13s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/70/655770/1/gate/octavia-v1-dsvm-py3x-scenario/e1b55b8/ : SUCCESS in 58m 54s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/70/655770/1/gate/octavia-v1-dsvm-scenario-multinode/00d1d77/ : FAILURE in 1h 39m 39s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/70/655770/1/gate/octavia-v1-dsvm-py3x-scenario-multinode/d3d652d/ : FAILURE in 1h 24m 21s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/70/655770/1/gate/neutron-lbaasv2-dsvm-py3x-api/42effd6/ : SUCCESS in 1h 08m 38s\n- octavia-grenade http://logs.openstack.org/70/655770/1/gate/octavia-grenade/b329186/ : SUCCESS in 1h 16m 34s\n- openstack-tox-cover http://logs.openstack.org/70/655770/1/gate/openstack-tox-cover/a77381a/cover/ : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":1},{"id":"106bd5b359b6f9dcda2316e4ae3781b47a18748e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 11:42:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"85584435a561034f7ae5385f14d03df8a14cb354","revisions":{"85584435a561034f7ae5385f14d03df8a14cb354":{"kind":"REWORK","_number":1,"created":"2019-04-25 19:17:58.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/70/655770/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/655770/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/655770/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/655770/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/655770/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/655770/1"}}},"commit":{"parents":[{"commit":"6da36f03c24be087d42567b549baa73389adea90","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6da36f03c24be087d42567b549baa73389adea90"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-04-04 21:29:55.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-04-25 19:17:55.000000000","tz":-420},"subject":"Fix listener API handling of None/null updates","message":"Fix listener API handling of None/null updates\n\nThe current listener API does not properly handle clearing/reseting\nvalues on update. Some integer only fields, such as connection-limit,\nwill accept null, but will store the value as \"None\". These will\nwill cause failures updating the amphora configuration.\n\nThis patch corrects this to appropriately handle None/null updates\nto the listener parameters.\n\nChange-Id: I41c9bedd8a3452513af3d409fbacd65ea287f02a\nStory: 2005374\nTask: 30352\n(cherry picked from commit 930a3236bf7fd341b32ba41b2c99cd3ad832d709)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85584435a561034f7ae5385f14d03df8a14cb354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85584435a561034f7ae5385f14d03df8a14cb354"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
