)]}'
{"id":"openstack%2Foctavia~656491","triplet_id":"openstack%2Foctavia~master~Icd68236a5b28e8cd2f1059828aae4ca5ea8cf3a2","project":"openstack/octavia","branch":"master","topic":"add-debian-as-supported-amphora","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2022-09-12 14:08:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icd68236a5b28e8cd2f1059828aae4ca5ea8cf3a2","subject":"Add support for Debian in the Amphorae class.","status":"MERGED","created":"2019-04-30 10:32:32.000000000","updated":"2022-09-12 14:09:57.000000000","submitted":"2022-09-12 14:08:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656491-add-debian-as-supported-amphora","meta_rev_id":"c07938d4d6ebf83da7b9bf5f354029f925a5946c","_number":656491,"virtual_id_number":656491,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-12 14:08:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"disliked":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"all":[{"value":-1,"date":"2020-05-07 23:48:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":2,"date":"2021-04-26 08:27:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":2,"date":"2022-09-12 11:16:18.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":1,"date":"2022-09-12 11:16:18.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-30 10:42:49.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2020-05-08 01:40:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-26 08:27:09.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-09-12 11:16:18.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"0df5509ebd6f2982642029f47c17e70f83a61245","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2019-04-30 10:32:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36465e528a5eaa7fe4313e059598d9bf78368b73","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-04-30 10:42:49.000000000","message":"Patch Set 1: Code-Review-1\n\nThank you for your contribution, Tomas.\n\nFew things are missing here:\n1. Tests, please update test_server.py, test_osutils.py and test_plug.py.\n2. Please update diskimage-create.sh to reflect that new option.\n3. You will possibly needto update diskimage-builder elements. either in this code base or on diskimage-builder itself\n4. Please add a release note.\n\nI would also advise having a job to thoroughly test this, to get an indication of how this works.\nYou can either add it as a followup patch or in here.\nAs for whether or not we keep the job is a matter of discussion with the core team, but at least we will have a better picture of this addition while you develop it.\n\nPlease let me know if you have any questions.","accounts_in_message":[],"_revision_number":1},{"id":"756b37d1d09bd10053eb08c2d4c4cd4e3278d3d8","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2019-04-30 11:34:30.000000000","message":"Patch Set 1:\n\nNir,\n\nI am the Debian package maintainer, and also maintainer of OCI:\nhttps://salsa.debian.org/openstack-team/debian/openstack-cluster-installer\n\nThis is already too much for a single person, so I will *not* have time to add some tests for this.\n\nAlso, I\u0027m not a user of diskimage-create.sh, I create my own image using openstack-debian-images. This probably will end up being one of the official published Debian images for OpenStack. I will, of course, publish my scripts to do it, though really, it will never use DIB, as I insist to use only packages, not using Pip.\n\nI hope this comment helps you understand where I\u0027m at.","accounts_in_message":[],"_revision_number":1},{"id":"6e66bcdd863c9aa840085ed03ac0c6ab18714514","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 12:20:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/91/656491/1/check/openstack-tox-cover/9f6bee7/cover/ : SUCCESS in 8m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/656491/1/check/openstack-tox-lower-constraints/0c89113/ : SUCCESS in 7m 25s\n- openstack-tox-pep8 http://logs.openstack.org/91/656491/1/check/openstack-tox-pep8/b012298/ : SUCCESS in 7m 20s\n- openstack-tox-py27 http://logs.openstack.org/91/656491/1/check/openstack-tox-py27/b0de85e/ : SUCCESS in 6m 58s\n- openstack-tox-py36 http://logs.openstack.org/91/656491/1/check/openstack-tox-py36/a882159/ : SUCCESS in 6m 56s\n- openstack-tox-docs http://logs.openstack.org/91/656491/1/check/openstack-tox-docs/4e40cd9/html/ : SUCCESS in 6m 31s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/91/656491/1/check/neutron-lbaasv2-dsvm-api/be9f6c4/ : SUCCESS in 1h 21m 46s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/91/656491/1/check/neutron-lbaasv2-dsvm-py3x-api/fa5b38f/ : SUCCESS in 1h 18m 11s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/91/656491/1/check/neutron-lbaasv2-dsvm-scenario/a9f13f2/ : SUCCESS in 1h 37m 45s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/91/656491/1/check/neutron-lbaasv2-dsvm-py3x-scenario/d46726b/ : SUCCESS in 1h 21m 46s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/91/656491/1/check/openstack-tox-functional/827d519/ : SUCCESS in 8m 02s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/656491/1/check/openstack-tox-functional-py36/322e4e4/ : SUCCESS in 8m 04s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/91/656491/1/check/octavia-v1-dsvm-scenario/1255df1/ : SUCCESS in 1h 03m 44s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/91/656491/1/check/octavia-v1-dsvm-py3x-scenario/e61ea4d/ : SUCCESS in 1h 01m 30s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/91/656491/1/check/octavia-v2-dsvm-noop-api/a130dfa/ : SUCCESS in 41m 41s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/91/656491/1/check/octavia-v2-dsvm-noop-py2-api/eba59fb/ : SUCCESS in 45m 17s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/91/656491/1/check/octavia-v2-dsvm-scenario/dbcefa8/ : SUCCESS in 1h 34m 34s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/91/656491/1/check/octavia-v2-dsvm-py2-scenario/58964d3/ : SUCCESS in 1h 30m 43s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/91/656491/1/check/octavia-v2-dsvm-py2-scenario-centos-7/c10b651/ : FAILURE in 33m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/91/656491/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/3ad57fe/ : SUCCESS in 1h 40m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/91/656491/1/check/octavia-v2-dsvm-scenario-two-node/bb4e841/ : FAILURE in 1h 32m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/91/656491/1/check/octavia-v2-dsvm-py2-scenario-two-node/0af4787/ : FAILURE in 1h 27m 06s (non-voting)\n- octavia-grenade http://logs.openstack.org/91/656491/1/check/octavia-grenade/53b1232/ : FAILURE in 28m 11s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/91/656491/1/check/octavia-v2-dsvm-tls-barbican/4754207/ : SUCCESS in 53m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c8428eab8969ebafe339032c5d0c5e66aa47195","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2019-05-10 10:16:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"680b46adc7b24f337d62d64e15cb77ca47a78a67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 12:30:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/91/656491/1/check/openstack-tox-cover/b023810/cover/ : SUCCESS in 9m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/656491/1/check/openstack-tox-lower-constraints/218a263/ : SUCCESS in 8m 15s\n- openstack-tox-pep8 http://logs.openstack.org/91/656491/1/check/openstack-tox-pep8/c4d919a/ : FAILURE in 6m 40s\n- openstack-tox-py27 http://logs.openstack.org/91/656491/1/check/openstack-tox-py27/8ae6a7a/ : SUCCESS in 5m 12s\n- openstack-tox-py36 http://logs.openstack.org/91/656491/1/check/openstack-tox-py36/c99be11/ : SUCCESS in 6m 26s\n- openstack-tox-docs http://logs.openstack.org/91/656491/1/check/openstack-tox-docs/faccc36/html/ : SUCCESS in 5m 50s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/91/656491/1/check/neutron-lbaasv2-dsvm-api/37daeac/ : SUCCESS in 1h 17m 34s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/91/656491/1/check/neutron-lbaasv2-dsvm-py3x-api/bb8b0ff/ : SUCCESS in 1h 16m 04s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/91/656491/1/check/neutron-lbaasv2-dsvm-scenario/f3549cd/ : SUCCESS in 1h 25m 00s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/91/656491/1/check/neutron-lbaasv2-dsvm-py3x-scenario/869db58/ : FAILURE in 46m 43s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/91/656491/1/check/openstack-tox-functional/f92cd44/ : SUCCESS in 7m 27s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/656491/1/check/openstack-tox-functional-py36/489c532/ : SUCCESS in 6m 48s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/91/656491/1/check/octavia-v1-dsvm-scenario/48c0a61/ : SUCCESS in 1h 00m 06s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/91/656491/1/check/octavia-v1-dsvm-py3x-scenario/e59a109/ : SUCCESS in 1h 05m 38s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/91/656491/1/check/octavia-v2-dsvm-noop-api/ba6d4c2/ : SUCCESS in 40m 38s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/91/656491/1/check/octavia-v2-dsvm-noop-py2-api/a7098e7/ : SUCCESS in 44m 58s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/91/656491/1/check/octavia-v2-dsvm-scenario/78465dc/ : SUCCESS in 1h 30m 00s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/91/656491/1/check/octavia-v2-dsvm-py2-scenario/d572575/ : SUCCESS in 1h 39m 20s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/91/656491/1/check/octavia-v2-dsvm-py2-scenario-centos-7/e54cc85/ : FAILURE in 1h 56m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/91/656491/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/d7ca8c9/ : SUCCESS in 1h 41m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/91/656491/1/check/octavia-v2-dsvm-scenario-two-node/99307ea/ : SUCCESS in 1h 56m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/91/656491/1/check/octavia-v2-dsvm-py2-scenario-two-node/46afcb8/ : SUCCESS in 1h 50m 47s (non-voting)\n- octavia-grenade http://logs.openstack.org/91/656491/1/check/octavia-grenade/2690087/ : FAILURE in 30m 40s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/91/656491/1/check/octavia-v2-dsvm-tls-barbican/a42d674/ : SUCCESS in 46m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"69b364ea5cff3c8e2284445b14a979caac978e94","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-07 23:48:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7bbe61d03e1c5744c44e0c234c44a2756435bd89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 01:40:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4224a0a9dfa1463eb2463c477e707282 : SUCCESS in 8m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5654f2c294aa472ba786f964d2b8b344 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f11edd26b7f4edf8720d0365087e8b3 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebefc1a3954f4d16a3ca843e51f98722 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d05da72375b4de7b5c41bc071fa9294 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/064df13ea0ab412c92c42fe58ecae28b : SUCCESS in 11m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e878c0b7b52d4e1bb0a7fb94d414135d : SUCCESS in 5m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d67bb5fada56473d9cb6774c727d1b87 : SUCCESS in 7m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b7219ac4dac240e686f16e9c633f35ea : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3dd5ce79705d48f182cb80a05dc8c754 : SUCCESS in 46m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab34b398044145e5809e2b660c17e83c : SUCCESS in 1h 51m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1c6acdc5ac4e44e58a29b08d47f7e516 : SUCCESS in 59m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/38c7969e51b94b7f89ba1b8eebe1e6db : SUCCESS in 49m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/920fbed9fa8246b6a403f5a86f85fa85 : SUCCESS in 1h 00m 05s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/71e265eb8c394921bec37766571b356d : SUCCESS in 56m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/faf22842214141679f15d158b9f3b560 : SUCCESS in 1h 47m 12s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0a3f82737c09480daff233a61b0dabe2 : SUCCESS in 1h 10m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f5e246e7c80147fa8f5acb7e318951fd : SUCCESS in 58m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2cc43674f925c1d1f9aecdd6b72ffca5aea8c01d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-26 08:27:09.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7d30f0e79bab9647f9696a6b697b7ebb609df7d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 10:19:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08d926e753f94461ab3907a9fe48c2b3 : SUCCESS in 9m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/636062d082914bc0b2e89aaa49ea05fa : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34b1e400180f49018f85e90fad36a870 : FAILURE in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c19c616ee19940919a0fca2bbb298008 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25bed3600fdd4411ba3c584dd2273c17 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0eb0df5d50748cd95ef889c4149783e : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f67694a44e80405585ea3451fd6f5117 : SUCCESS in 13m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/134ecb05e8fd4fc8a5d5a87d24702301 : SUCCESS in 7m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b4933cf7acd5494b8ea25f861946df10 : SUCCESS in 9m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/10b45308dc5a469a92328598dbdd71a4 : SUCCESS in 7m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7a5c8a59803640858eb1146c6945b6de : SUCCESS in 7m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4a44665975f146dfb0965c978ece7c5e : SUCCESS in 1h 20m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1dcc76c79095458a824940e656a67325 : SUCCESS in 1h 48m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3d0687043ece4653b41c3e5bba81f812 : SUCCESS in 42m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/acb75a0141ae49f5a46bda4b5696d3a9 : SUCCESS in 38m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6671d0a81cc645048e5b839f9f543003 : SUCCESS in 35m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c57e58c6ad254425afe518938bee6a70 : SUCCESS in 49m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b1cfa3cf2ca04bcb8bfdc62cdb9af879 : FAILURE in 18m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e22b144b0074412a8513db002312e852 : FAILURE in 41m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/20dc8207222a4edeaa5b083087da612c : SUCCESS in 1h 46m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a0236c64943f4c58813293e2e0629eb5 : SUCCESS in 1h 36m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/79edc7bf8e88425d8148474a3cd8c8be : SUCCESS in 1h 17m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/90a7335686d743d5977e266ac8663386 : FAILURE in 22m 33s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/utils/network_namespace.py\n  Comments left for invalid file octavia/amphorae/drivers/health/heartbeat_udp.py\n  Comments left for invalid file octavia/controller/healthmanager/health_manager.py\n  Comments left for invalid file octavia/common/base_taskflow.py","accounts_in_message":[],"_revision_number":2},{"id":"a5ece8f980d480ee0824ccf4771bc4c12d0d50e8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-26 09:43:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a967dffe52398350ee7f2f7b275b2651291f55f2","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-10-26 11:16:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e21a742533e935ece488f243ef75e6439c64d7d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 11:58:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e606bbfd8484d89b5f6e497f9c1433b : SUCCESS in 11m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c8d6fb6d2fe4012889f702c36a9852b : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c902fb15ccfd4034a5eb91b3f1b268ad : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9aa1992fc3544ed3a812d4cefd5c6241 : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b79163a9f24487292ab4a76ace238cd : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/663d67f6d0204be9afc6bceae2f22d62 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e41e33a462641cdaeb60d9997ea3ca7 : SUCCESS in 12m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/202c9d1c9fbf4b74a8b196a2df38c6a3 : SUCCESS in 6m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2c25b0002b71465ba0d5047abe30d049 : SUCCESS in 11m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5e1f29fd95ec4bbfa1f13bfb838a1f3b : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e8566652cc944e2ba6ffe269dbe6f2f2 : SUCCESS in 10m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/db32959b9d024c65ab7693084ce43ba4 : SUCCESS in 2h 03m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/958a931d489048a2a2e316fdc199966a : SUCCESS in 1h 24m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0f1bb4aac6654619987d600c744e955c : SUCCESS in 37m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/333cac4f4bb746e3a9dd952f1067ade5 : SUCCESS in 51m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/559d4464df0a4eda8cf9178eb6411965 : SUCCESS in 36m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/822627379f794e9b815e6c1827634033 : SUCCESS in 1h 39m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1a3f2f711415400fbc76dc45b9587f81 : SUCCESS in 1h 39m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2f161b6485be486ca237c0e4815c294d : SUCCESS in 1h 49m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/46e23eabba7d48ff806042b26b43a71a : SUCCESS in 1h 43m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/af0f56c3205a4f6eb03a9d50e8971353 : SUCCESS in 1h 28m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fdeffb94b011424a80438d0d28408e79 : FAILURE in 19m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9e11350fb832069909f5660fe62f747633977464","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 11:16:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5d4d31a932d8fd278732b821eaf69833a31b5896","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 11:16:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"930a796afcecc9de318ee6c507926e8c282e87a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 14:08:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9ebb6a6dfec4488ae4d3602ed7f9c41\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99c1087042e6405e9eac83b1dd97bb32 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/416a97c21b2a44f0be6c12f48693c6da : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26bea6f82cd34d8781b5442bdfe97535 : SUCCESS in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11d3c403e6f74c788b0860e4bc2a26af : SUCCESS in 12m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/98feffc9e37944b89bab53750af49104 : SUCCESS in 5m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/73832a01a7704ce0a198fc9866d8e9d9 : SUCCESS in 8m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/77bd2044825a489e990f4f0790189819 : SUCCESS in 2h 50m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b80d6fc887c4e618ae86b28fb13770f : SUCCESS in 2h 20m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c2c49bab864f4f99b08328b1e354e95b : SUCCESS in 59m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a1c3217aa8f54a0fadb308d17c77af69 : SUCCESS in 1h 10m 44s","accounts_in_message":[],"_revision_number":2},{"id":"60600d4827fad7d97eb49ee44a85b1a96f7b723b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 14:08:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c07938d4d6ebf83da7b9bf5f354029f925a5946c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 14:09:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/483db2f9c3974a22b886c34e7dca0a4a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2e96fc51be44b388b55aeb173cf12cc : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5162eec7bc75ce7778049967e333c51e74a24b2e","revisions":{"6d6161e3545ec35e699f169c4c12c9f88135e5cb":{"kind":"REWORK","_number":1,"created":"2019-04-30 10:32:32.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/91/656491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/656491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/656491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/656491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/656491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/656491/1"}}},"commit":{"parents":[{"commit":"6008859476a267a1a38f8a70a4d14fa874dc99cd","subject":"Merge \"Fix the loss of access to barbican secrets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6008859476a267a1a38f8a70a4d14fa874dc99cd"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2019-04-30 10:31:52.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2019-04-30 10:32:29.000000000","tz":120},"subject":"Add support for Debian in the Amphorae class.","message":"Add support for Debian in the Amphorae class.\n\nChange-Id: Icd68236a5b28e8cd2f1059828aae4ca5ea8cf3a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d6161e3545ec35e699f169c4c12c9f88135e5cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d6161e3545ec35e699f169c4c12c9f88135e5cb"}]},"branch":"refs/heads/master"},"5162eec7bc75ce7778049967e333c51e74a24b2e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-07 23:48:22.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/91/656491/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/656491/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/656491/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/656491/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/656491/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/656491/2"}}},"commit":{"parents":[{"commit":"837b7bf94092438e486d3e18c5b9a21bb1298f98","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/837b7bf94092438e486d3e18c5b9a21bb1298f98"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2019-04-30 10:31:52.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"thomas@goirand.fr","date":"2020-05-07 23:48:22.000000000","tz":0},"subject":"Add support for Debian in the Amphorae class.","message":"Add support for Debian in the Amphorae class.\n\nChange-Id: Icd68236a5b28e8cd2f1059828aae4ca5ea8cf3a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5162eec7bc75ce7778049967e333c51e74a24b2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5162eec7bc75ce7778049967e333c51e74a24b2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
