)]}'
{"id":"openstack%2Foctavia~656916","triplet_id":"openstack%2Foctavia~master~Ie81de8e85753fde1516aea0b084df6a0c513ad7b","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ie81de8e85753fde1516aea0b084df6a0c513ad7b","subject":"Make sure amphora logging works the same on py2 and py3","status":"MERGED","created":"2019-05-03 04:06:03.000000000","updated":"2019-05-15 19:46:54.000000000","submitted":"2019-05-15 19:46:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656916-1557949614436-ec554898","meta_rev_id":"b5fea735ba3e0b3504672995874570819fd074a5","_number":656916,"virtual_id_number":656916,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-05-15 19:46:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"date":"2019-05-04 21:42:14.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-05-14 20:04:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-15 18:00:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-15 18:00:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-04 18:48:57.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-05-14 20:04:59.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-05-15 18:00:31.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-05-15 19:46:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0eb26c6ba1bbf7e82c93eb0346feb563d182e9c8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-03 04:06:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13c308fa2d62ce123b26d833782646e17505c781","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 06:03:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/16/656916/1/check/openstack-tox-cover/4f7cb30/cover/ : SUCCESS in 7m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/656916/1/check/openstack-tox-lower-constraints/f4ba7a7/ : SUCCESS in 7m 43s\n- openstack-tox-pep8 http://logs.openstack.org/16/656916/1/check/openstack-tox-pep8/5a17c5e/ : SUCCESS in 6m 06s\n- openstack-tox-py27 http://logs.openstack.org/16/656916/1/check/openstack-tox-py27/4179ac7/ : SUCCESS in 5m 10s\n- openstack-tox-py36 http://logs.openstack.org/16/656916/1/check/openstack-tox-py36/7ce096b/ : SUCCESS in 5m 05s\n- openstack-tox-docs http://logs.openstack.org/16/656916/1/check/openstack-tox-docs/23b7446/html/ : SUCCESS in 5m 57s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/16/656916/1/check/neutron-lbaasv2-dsvm-api/85811d4/ : SUCCESS in 1h 18m 16s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/16/656916/1/check/neutron-lbaasv2-dsvm-py3x-api/836e560/ : SUCCESS in 1h 17m 51s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/16/656916/1/check/neutron-lbaasv2-dsvm-scenario/8d617fc/ : SUCCESS in 1h 25m 11s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/16/656916/1/check/neutron-lbaasv2-dsvm-py3x-scenario/b9e18c0/ : FAILURE in 1h 21m 46s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/16/656916/1/check/openstack-tox-functional/89f1272/ : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 http://logs.openstack.org/16/656916/1/check/openstack-tox-functional-py36/8ac44aa/ : SUCCESS in 6m 52s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/16/656916/1/check/octavia-v1-dsvm-scenario/c2a7b75/ : SUCCESS in 1h 02m 38s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/16/656916/1/check/octavia-v1-dsvm-py3x-scenario/c7badab/ : SUCCESS in 58m 48s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/16/656916/1/check/octavia-v2-dsvm-noop-api/f8deca6/ : SUCCESS in 43m 01s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/16/656916/1/check/octavia-v2-dsvm-noop-py2-api/a1a0fff/ : SUCCESS in 41m 25s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/16/656916/1/check/octavia-v2-dsvm-scenario/3891943/ : SUCCESS in 1h 25m 31s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/16/656916/1/check/octavia-v2-dsvm-py2-scenario/5d75e8f/ : SUCCESS in 1h 37m 13s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/16/656916/1/check/octavia-v2-dsvm-py2-scenario-centos-7/d138623/ : FAILURE in 35m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/16/656916/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/f1d7e50/ : SUCCESS in 1h 31m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/16/656916/1/check/octavia-v2-dsvm-scenario-two-node/9d27fc7/ : SUCCESS in 1h 52m 31s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/16/656916/1/check/octavia-v2-dsvm-py2-scenario-two-node/b8e4597/ : SUCCESS in 1h 49m 52s (non-voting)\n- octavia-grenade http://logs.openstack.org/16/656916/1/check/octavia-grenade/8a52a3c/ : SUCCESS in 1h 19m 03s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/16/656916/1/check/octavia-v2-dsvm-tls-barbican/e32fb50/ : SUCCESS in 56m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9194bceb360a7c366f65506a9d5b028271d3fa32","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-04 18:40:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9d0a0ba22ca3d26538683d61439712c50df0c7b6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-04 18:48:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b317dba3ac2f64889b2d4c9f622c4255f8d7cdd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-04 18:52:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fe9b758793e4db0767b6cc4d5f82629393b6f979","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-04 19:03:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"baa3117fba950c35662803a7ec38cd1939dd0e96","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-04 19:04:53.000000000","message":"Patch Set 4: Code-Review-1\n\nThe logging changes need to be split out into a separate patch","accounts_in_message":[],"_revision_number":4},{"id":"37025428c79a5d28b5e29163f8e2b94ba3446f81","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-04 19:05:50.000000000","message":"Patch Set 4:\n\n\u003e The logging changes need to be split out into a separate patch\n\nI just threw them in to get some actual working tests, but I can remove them.","accounts_in_message":[],"_revision_number":4},{"id":"e532f46a5f6cfec79427ed699fa22ecf105416db","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-04 19:07:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e8563ab51ad36cb859f8f0b60f61b9259a4bb888","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-04 19:09:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"86aed7db02e09b1d9b4b1a79b01dc095e947e9db","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-04 19:19:35.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"7cfc634c78798ae9660c9b5dab286ea2e395d70c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 21:06:21.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/16/656916/7/check/openstack-tox-cover/c0270be/cover/ : SUCCESS in 10m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/656916/7/check/openstack-tox-lower-constraints/2a6fd57/ : SUCCESS in 6m 56s\n- openstack-tox-pep8 http://logs.openstack.org/16/656916/7/check/openstack-tox-pep8/ed193b8/ : SUCCESS in 7m 43s\n- openstack-tox-py27 http://logs.openstack.org/16/656916/7/check/openstack-tox-py27/31de348/ : SUCCESS in 5m 50s\n- openstack-tox-py36 http://logs.openstack.org/16/656916/7/check/openstack-tox-py36/fd7db87/ : SUCCESS in 6m 47s\n- openstack-tox-docs http://logs.openstack.org/16/656916/7/check/openstack-tox-docs/d7d9be1/html/ : SUCCESS in 6m 10s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/16/656916/7/check/neutron-lbaasv2-dsvm-api/ff4b467/ : SUCCESS in 1h 17m 31s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/16/656916/7/check/neutron-lbaasv2-dsvm-py3x-api/48d35be/ : SUCCESS in 1h 41m 25s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/16/656916/7/check/neutron-lbaasv2-dsvm-scenario/90cfffb/ : FAILURE in 1h 26m 30s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/16/656916/7/check/neutron-lbaasv2-dsvm-py3x-scenario/ed30265/ : FAILURE in 1h 24m 57s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/16/656916/7/check/openstack-tox-functional/41b5a1c/ : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 http://logs.openstack.org/16/656916/7/check/openstack-tox-functional-py36/9a11b51/ : SUCCESS in 6m 43s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/16/656916/7/check/octavia-v1-dsvm-scenario/2ded9da/ : SUCCESS in 48m 53s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/16/656916/7/check/octavia-v1-dsvm-py3x-scenario/eabb5b1/ : FAILURE in 28m 20s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/16/656916/7/check/octavia-v2-dsvm-noop-api/1563e96/ : FAILURE in 25m 57s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/16/656916/7/check/octavia-v2-dsvm-noop-py2-api/d68110c/ : SUCCESS in 39m 39s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/16/656916/7/check/octavia-v2-dsvm-scenario/9219334/ : SUCCESS in 1h 36m 21s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/16/656916/7/check/octavia-v2-dsvm-py2-scenario/199d861/ : SUCCESS in 1h 34m 16s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/16/656916/7/check/octavia-v2-dsvm-py2-scenario-centos-7/eff75b5/ : FAILURE in 34m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/16/656916/7/check/octavia-v2-dsvm-scenario-ubuntu-bionic/fee3156/ : SUCCESS in 1h 42m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/16/656916/7/check/octavia-v2-dsvm-scenario-two-node/75f84cf/ : SUCCESS in 1h 28m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/16/656916/7/check/octavia-v2-dsvm-py2-scenario-two-node/1c3e771/ : FAILURE in 1h 45m 15s (non-voting)\n- octavia-grenade http://logs.openstack.org/16/656916/7/check/octavia-grenade/64eb3d4/ : SUCCESS in 1h 15m 44s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/16/656916/7/check/octavia-v2-dsvm-tls-barbican/067a0ab/ : SUCCESS in 52m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3967cf273b155c4ced73637a902708bfb72c16d2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-04 21:42:14.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"030cc6c0c115c406ea16329d31dd611cb8d2107e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 23:50:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/16/656916/7/check/openstack-tox-cover/146173c/cover/ : SUCCESS in 9m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/656916/7/check/openstack-tox-lower-constraints/b587549/ : SUCCESS in 11m 15s\n- openstack-tox-pep8 http://logs.openstack.org/16/656916/7/check/openstack-tox-pep8/a87e095/ : SUCCESS in 7m 06s\n- openstack-tox-py27 http://logs.openstack.org/16/656916/7/check/openstack-tox-py27/cb05973/ : SUCCESS in 10m 29s\n- openstack-tox-py36 http://logs.openstack.org/16/656916/7/check/openstack-tox-py36/d97c622/ : SUCCESS in 6m 23s\n- openstack-tox-docs http://logs.openstack.org/16/656916/7/check/openstack-tox-docs/33bb1f8/html/ : SUCCESS in 7m 16s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/16/656916/7/check/neutron-lbaasv2-dsvm-api/52a4ade/ : SUCCESS in 1h 45m 44s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/16/656916/7/check/neutron-lbaasv2-dsvm-py3x-api/f9374e2/ : SUCCESS in 1h 23m 17s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/16/656916/7/check/neutron-lbaasv2-dsvm-scenario/5267717/ : SUCCESS in 1h 25m 39s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/16/656916/7/check/neutron-lbaasv2-dsvm-py3x-scenario/d578d5e/ : FAILURE in 1h 20m 23s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/16/656916/7/check/openstack-tox-functional/43520d4/ : SUCCESS in 9m 10s\n- openstack-tox-functional-py36 http://logs.openstack.org/16/656916/7/check/openstack-tox-functional-py36/23d8763/ : SUCCESS in 9m 17s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/16/656916/7/check/octavia-v1-dsvm-scenario/2e455f4/ : SUCCESS in 1h 02m 12s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/16/656916/7/check/octavia-v1-dsvm-py3x-scenario/781c4d3/ : SUCCESS in 1h 05m 12s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/16/656916/7/check/octavia-v2-dsvm-noop-api/27a763b/ : SUCCESS in 48m 42s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/16/656916/7/check/octavia-v2-dsvm-noop-py2-api/9873404/ : SUCCESS in 48m 29s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/16/656916/7/check/octavia-v2-dsvm-scenario/4c63f38/ : SUCCESS in 1h 37m 19s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/16/656916/7/check/octavia-v2-dsvm-py2-scenario/e5f010e/ : SUCCESS in 1h 24m 46s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/16/656916/7/check/octavia-v2-dsvm-py2-scenario-centos-7/595b6a2/ : FAILURE in 40m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/16/656916/7/check/octavia-v2-dsvm-scenario-ubuntu-bionic/88e6e99/ : SUCCESS in 1h 37m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/16/656916/7/check/octavia-v2-dsvm-scenario-two-node/4b89002/ : FAILURE in 1h 27m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/16/656916/7/check/octavia-v2-dsvm-py2-scenario-two-node/19e3bf1/ : FAILURE in 1h 21m 38s (non-voting)\n- octavia-grenade http://logs.openstack.org/16/656916/7/check/octavia-grenade/88218d2/ : SUCCESS in 1h 08m 32s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/16/656916/7/check/octavia-v2-dsvm-tls-barbican/96e1d04/ : SUCCESS in 1h 00m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"962d358a9b4b8f2f7ae14cf9d165e83a3cd9efaf","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-14 20:04:59.000000000","message":"Patch Set 7: Code-Review+2\n\nSeems sane to me.","accounts_in_message":[],"_revision_number":7},{"id":"92c07de9a62765072301df8282b05ad4840a0053","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-15 18:00:31.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":7},{"id":"10257f10946876000b6ab90d581c95443e2abc70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 18:00:44.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"de483373a419a3ed8b9aa622920893c973b66ea7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 19:46:54.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/16/656916/7/gate/openstack-tox-lower-constraints/eaed898/ : SUCCESS in 9m 01s\n- openstack-tox-pep8 http://logs.openstack.org/16/656916/7/gate/openstack-tox-pep8/30f1eb8/ : SUCCESS in 7m 38s\n- openstack-tox-py27 http://logs.openstack.org/16/656916/7/gate/openstack-tox-py27/8f4b4bc/ : SUCCESS in 7m 01s\n- openstack-tox-py36 http://logs.openstack.org/16/656916/7/gate/openstack-tox-py36/8ab0e2c/ : SUCCESS in 7m 07s\n- openstack-tox-docs http://logs.openstack.org/16/656916/7/gate/openstack-tox-docs/1a2349a/html/ : SUCCESS in 6m 13s\n- openstack-tox-functional http://logs.openstack.org/16/656916/7/gate/openstack-tox-functional/3384035/ : SUCCESS in 8m 00s\n- openstack-tox-functional-py36 http://logs.openstack.org/16/656916/7/gate/openstack-tox-functional-py36/fce9a47/ : SUCCESS in 6m 15s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/16/656916/7/gate/octavia-v2-dsvm-noop-api/e70870e/ : SUCCESS in 41m 58s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/16/656916/7/gate/octavia-v2-dsvm-noop-py2-api/3fd120a/ : SUCCESS in 46m 30s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/16/656916/7/gate/octavia-v2-dsvm-scenario/bfbc853/ : SUCCESS in 1h 44m 17s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/16/656916/7/gate/octavia-v2-dsvm-py2-scenario/8ccb9a4/ : SUCCESS in 1h 33m 16s\n- octavia-grenade http://logs.openstack.org/16/656916/7/gate/octavia-grenade/cbf3d17/ : SUCCESS in 1h 11m 36s","accounts_in_message":[],"_revision_number":7},{"id":"b5fea735ba3e0b3504672995874570819fd074a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 19:46:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0000412cf47b46fb876964c8596b182e56320a6a","revisions":{"52ce82e5aec906a9bd1ae0665f730fd4a42ab4e0":{"kind":"REWORK","_number":1,"created":"2019-05-03 04:06:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/656916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/656916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/656916/1"}}},"commit":{"parents":[{"commit":"57d653cc63d409b8670961e2a6a2556ce9231498","subject":"Make amphora cert validity time configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57d653cc63d409b8670961e2a6a2556ce9231498"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-03 03:54:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-03 03:56:59.000000000","tz":-420},"subject":"Fixing python 3 compatibility with amp health daemon","message":"Fixing python 3 compatibility with amp health daemon\n\nIOError and OSError is the same thing in Python 3, so\none of the code paths was unreachable.\n\nChange-Id: Ie81de8e85753fde1516aea0b084df6a0c513ad7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/52ce82e5aec906a9bd1ae0665f730fd4a42ab4e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/52ce82e5aec906a9bd1ae0665f730fd4a42ab4e0"}]},"branch":"refs/heads/master"},"b8977c10221e23c52384cbe6449c700d55cfe26d":{"kind":"REWORK","_number":2,"created":"2019-05-04 18:40:38.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/656916/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/656916/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/656916/2"}}},"commit":{"parents":[{"commit":"57d653cc63d409b8670961e2a6a2556ce9231498","subject":"Make amphora cert validity time configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57d653cc63d409b8670961e2a6a2556ce9231498"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-03 03:54:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-04 18:38:02.000000000","tz":-420},"subject":"Make sure amphora logging works the same on py2 and py3","message":"Make sure amphora logging works the same on py2 and py3\n\nIn Python 3 IOError and OSError are the same thing. This\ncauses logging in a very specific scenario to not log\nthe appropriate message.\n\nIn addition we import the StandardLogging fixture\nto allow us to easily caputre logs as part of our\ntests.\n\nChange-Id: Ie81de8e85753fde1516aea0b084df6a0c513ad7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8977c10221e23c52384cbe6449c700d55cfe26d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8977c10221e23c52384cbe6449c700d55cfe26d"}]},"branch":"refs/heads/master"},"4452765a15d54c64e71092c5304bc714a44a59b5":{"kind":"REWORK","_number":3,"created":"2019-05-04 18:52:15.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/656916/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/656916/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/656916/3"}}},"commit":{"parents":[{"commit":"57d653cc63d409b8670961e2a6a2556ce9231498","subject":"Make amphora cert validity time configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57d653cc63d409b8670961e2a6a2556ce9231498"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-03 03:54:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-04 18:51:10.000000000","tz":-420},"subject":"Make sure amphora logging works the same on py2 and py3","message":"Make sure amphora logging works the same on py2 and py3\n\nIn Python 3 IOError and OSError are the same thing. This\ncauses logging in a very specific scenario to not log\nthe appropriate message. This is fixed in this patch\nby merging the two exception paths.\n\nIn addition we import the StandardLogging fixture\nfrom nova to allow us to easily caputre logs as\npart of our tests.\n\nChange-Id: Ie81de8e85753fde1516aea0b084df6a0c513ad7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4452765a15d54c64e71092c5304bc714a44a59b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4452765a15d54c64e71092c5304bc714a44a59b5"}]},"branch":"refs/heads/master"},"fa077a58ebc1651b09a7c9d4899367534e79092e":{"kind":"REWORK","_number":4,"created":"2019-05-04 19:03:52.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/656916/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/656916/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/656916/4"}}},"commit":{"parents":[{"commit":"57d653cc63d409b8670961e2a6a2556ce9231498","subject":"Make amphora cert validity time configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57d653cc63d409b8670961e2a6a2556ce9231498"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-03 03:54:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-04 19:03:34.000000000","tz":-420},"subject":"Make sure amphora logging works the same on py2 and py3","message":"Make sure amphora logging works the same on py2 and py3\n\nIn Python 3 IOError and OSError are the same thing. This\ncauses logging in a very specific scenario to not log\nthe appropriate message. This is fixed in this patch\nby merging the two exception paths.\n\nIn addition we import the StandardLogging fixture\nfrom nova to allow us to easily caputre logs as\npart of our tests.\n\nChange-Id: Ie81de8e85753fde1516aea0b084df6a0c513ad7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa077a58ebc1651b09a7c9d4899367534e79092e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa077a58ebc1651b09a7c9d4899367534e79092e"}]},"branch":"refs/heads/master"},"6f03caa54f44a786f0cfa502e2aa50fde3ab5782":{"kind":"REWORK","_number":5,"created":"2019-05-04 19:07:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/656916/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/656916/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/656916/5"}}},"commit":{"parents":[{"commit":"57d653cc63d409b8670961e2a6a2556ce9231498","subject":"Make amphora cert validity time configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57d653cc63d409b8670961e2a6a2556ce9231498"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-03 03:54:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-04 19:07:27.000000000","tz":-420},"subject":"Make sure amphora logging works the same on py2 and py3","message":"Make sure amphora logging works the same on py2 and py3\n\nIn Python 3 IOError and OSError are the same thing. This\ncauses logging in a very specific scenario to not log\nthe appropriate message. This is fixed in this patch\nby merging the two exception paths.\n\nChange-Id: Ie81de8e85753fde1516aea0b084df6a0c513ad7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f03caa54f44a786f0cfa502e2aa50fde3ab5782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f03caa54f44a786f0cfa502e2aa50fde3ab5782"}]},"branch":"refs/heads/master"},"195828506e1388cffda8ca0ff80a9ac8803f28b3":{"kind":"REWORK","_number":6,"created":"2019-05-04 19:09:22.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/656916/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/656916/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/656916/6"}}},"commit":{"parents":[{"commit":"57d653cc63d409b8670961e2a6a2556ce9231498","subject":"Make amphora cert validity time configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57d653cc63d409b8670961e2a6a2556ce9231498"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-03 03:54:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-04 19:09:00.000000000","tz":-420},"subject":"Make sure amphora logging works the same on py2 and py3","message":"Make sure amphora logging works the same on py2 and py3\n\nIn Python 3 IOError and OSError are the same thing. This\ncauses logging in a very specific scenario to not log\nthe appropriate message. This is fixed in this patch\nby merging the two exception paths.\n\nChange-Id: Ie81de8e85753fde1516aea0b084df6a0c513ad7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/195828506e1388cffda8ca0ff80a9ac8803f28b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/195828506e1388cffda8ca0ff80a9ac8803f28b3"}]},"branch":"refs/heads/master"},"0000412cf47b46fb876964c8596b182e56320a6a":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2019-05-04 19:19:35.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/656916/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/656916/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/656916/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/656916/7"}}},"commit":{"parents":[{"commit":"57d653cc63d409b8670961e2a6a2556ce9231498","subject":"Make amphora cert validity time configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57d653cc63d409b8670961e2a6a2556ce9231498"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-03 03:54:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-04 19:19:33.000000000","tz":0},"subject":"Make sure amphora logging works the same on py2 and py3","message":"Make sure amphora logging works the same on py2 and py3\n\nIn Python 3.3 IOError is just an alias of OSError. This\ncauses logging in a very specific scenario to not log\nthe appropriate message, as one code path is unreachable. \nThis is fixed in this patch by merging the two exception paths.\n\nStory: 2005576\nTask: 30765\n\nChange-Id: Ie81de8e85753fde1516aea0b084df6a0c513ad7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0000412cf47b46fb876964c8596b182e56320a6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0000412cf47b46fb876964c8596b182e56320a6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
