)]}'
{"id":"openstack%2Foctavia~657756","triplet_id":"openstack%2Foctavia~master~I7fde54084b39c1d0529cfb0bcfb79697d63ea6ae","project":"openstack/octavia","branch":"master","topic":"performance_improvement","hashtags":[],"change_id":"I7fde54084b39c1d0529cfb0bcfb79697d63ea6ae","subject":"Performance improvement for non-udp health checks","status":"MERGED","created":"2019-05-08 08:45:06.000000000","updated":"2019-05-15 01:05:26.000000000","submitted":"2019-05-15 01:05:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657756-1557882326762-4a183efd","meta_rev_id":"56a095d3c3e6934b71623639d9797f206941c67e","_number":657756,"virtual_id_number":657756,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-05-15 01:05:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-05-14 23:14:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-05-14 20:50:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-11 05:33:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-05-14 23:14:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 20:50:12.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-05-14 23:14:34.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-05-15 01:05:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"37a723f6df3d1cd4f3de2a85f7f3a2395c41d88d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-08 08:45:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"516739ed1cc9e637d8aa0005eefb5722a8e6e633","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-08 08:52:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e3b5453d3a686683c720b450125bd70195ee857b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 11:01:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/56/657756/2/check/openstack-tox-cover/3faf1c1/ : FAILURE in 6m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/657756/2/check/openstack-tox-lower-constraints/9273649/ : FAILURE in 5m 05s\n- openstack-tox-pep8 http://logs.openstack.org/56/657756/2/check/openstack-tox-pep8/032557e/ : SUCCESS in 7m 00s\n- openstack-tox-py27 http://logs.openstack.org/56/657756/2/check/openstack-tox-py27/198b143/ : FAILURE in 6m 03s\n- openstack-tox-py36 http://logs.openstack.org/56/657756/2/check/openstack-tox-py36/b8355d9/ : FAILURE in 4m 53s\n- openstack-tox-docs http://logs.openstack.org/56/657756/2/check/openstack-tox-docs/fa721c2/html/ : SUCCESS in 6m 57s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/56/657756/2/check/neutron-lbaasv2-dsvm-api/bce9ef7/ : SUCCESS in 1h 18m 54s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/56/657756/2/check/neutron-lbaasv2-dsvm-py3x-api/97bf2d3/ : SUCCESS in 1h 18m 27s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/56/657756/2/check/neutron-lbaasv2-dsvm-scenario/8f139f5/ : SUCCESS in 1h 24m 50s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/56/657756/2/check/neutron-lbaasv2-dsvm-py3x-scenario/ceecba5/ : SUCCESS in 1h 24m 56s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/56/657756/2/check/openstack-tox-functional/5aeec24/ : FAILURE in 6m 47s\n- openstack-tox-functional-py36 http://logs.openstack.org/56/657756/2/check/openstack-tox-functional-py36/6424ed9/ : FAILURE in 8m 52s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/56/657756/2/check/octavia-v1-dsvm-scenario/3b5f37d/ : SUCCESS in 1h 02m 07s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/56/657756/2/check/octavia-v1-dsvm-py3x-scenario/4e99682/ : SUCCESS in 1h 00m 28s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/56/657756/2/check/octavia-v2-dsvm-noop-api/c4e6fb5/ : SUCCESS in 40m 19s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/56/657756/2/check/octavia-v2-dsvm-noop-py2-api/7eed666/ : SUCCESS in 39m 06s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/56/657756/2/check/octavia-v2-dsvm-scenario/39f627c/ : SUCCESS in 1h 28m 17s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/56/657756/2/check/octavia-v2-dsvm-py2-scenario/3f9dae3/ : SUCCESS in 1h 33m 02s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/56/657756/2/check/octavia-v2-dsvm-py2-scenario-centos-7/c2ee391/ : FAILURE in 2h 01m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/56/657756/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/769601b/ : FAILURE in 1h 36m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/56/657756/2/check/octavia-v2-dsvm-scenario-two-node/9d6f5d7/ : SUCCESS in 1h 45m 56s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/56/657756/2/check/octavia-v2-dsvm-py2-scenario-two-node/6977d3e/ : SUCCESS in 1h 44m 23s (non-voting)\n- octavia-grenade http://logs.openstack.org/56/657756/2/check/octavia-grenade/19ac1a3/ : SUCCESS in 1h 13m 06s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/56/657756/2/check/octavia-v2-dsvm-tls-barbican/f8c3cd2/ : SUCCESS in 50m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ed62332a4fcf7fbf20f7a6ddb0e4e3842a9f71cc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-08 15:46:02.000000000","message":"Patch Set 2: Code-Review-1\n\nI\u0027m not sure this is necessary as we updated the amphora-agent to send versioned heartbeats, which will bypass this method. It is just here for backwards compatibility purposes. I think the non-versioned heartbeat code was only released as a milestone (I will go check).","accounts_in_message":[],"_revision_number":2},{"id":"4a3501fdd0e311e9555d9161afcffb3d7fd94ad4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-08 15:50:16.000000000","message":"Patch Set 2:\n\nConfirmed, the current Rocky release has the versioned heartbeat fix.","accounts_in_message":[],"_revision_number":2},{"id":"76d15a44d06ea968cee05e3a6fffe4e7732b48e0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-08 15:58:53.000000000","message":"Patch Set 2:\n\n\u003e Confirmed, the current Rocky release has the versioned heartbeat\n \u003e fix.\n\nThis is actually because your stresstool\nhttps://github.com/johnsom/stresshm/blob/master/stresshm.py\n\nI was benchmarking late last night and couldn\u0027t figure out why it was so slow. I realized this morning that it probably is because your stress testing tool does not support version yet.","accounts_in_message":[],"_revision_number":2},{"id":"41cf7fe74c23c129b3cd9a50b82310a724f92ac5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-08 16:00:58.000000000","message":"Abandoned\n\nThis isn\u0027t actually needed. Just need to update our stress test tool to pass on ver.","accounts_in_message":[],"_revision_number":2},{"id":"01d69baffc06b7ff9f3cfb77e524cbb5df1e4c49","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-09 07:02:17.000000000","message":"Restored\n\nGoing to fix this to be useful","accounts_in_message":[],"_revision_number":2},{"id":"92f533eabba4d9126fc1fcb053b832abeb0dc8ec","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-09 07:02:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"843b7baa0c38b0d8bf4ceadc1a82c47141e8fb44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 09:14:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/56/657756/3/check/openstack-tox-cover/13e3894/cover/ : SUCCESS in 9m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/657756/3/check/openstack-tox-lower-constraints/255378b/ : SUCCESS in 7m 24s\n- openstack-tox-pep8 http://logs.openstack.org/56/657756/3/check/openstack-tox-pep8/7a072f8/ : SUCCESS in 6m 09s\n- openstack-tox-py27 http://logs.openstack.org/56/657756/3/check/openstack-tox-py27/00412c3/ : SUCCESS in 4m 45s\n- openstack-tox-py36 http://logs.openstack.org/56/657756/3/check/openstack-tox-py36/e402cbc/ : SUCCESS in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/56/657756/3/check/openstack-tox-docs/b7bdc80/html/ : SUCCESS in 6m 09s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/56/657756/3/check/neutron-lbaasv2-dsvm-api/be56d63/ : SUCCESS in 1h 21m 17s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/56/657756/3/check/neutron-lbaasv2-dsvm-py3x-api/22c0df6/ : SUCCESS in 1h 14m 10s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/56/657756/3/check/neutron-lbaasv2-dsvm-scenario/5a5cef0/ : SUCCESS in 1h 10m 32s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/56/657756/3/check/neutron-lbaasv2-dsvm-py3x-scenario/b3eed1c/ : FAILURE in 1h 15m 23s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/56/657756/3/check/openstack-tox-functional/7a744c7/ : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 http://logs.openstack.org/56/657756/3/check/openstack-tox-functional-py36/49ace7d/ : SUCCESS in 6m 52s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/56/657756/3/check/octavia-v1-dsvm-scenario/8d5f676/ : SUCCESS in 1h 06m 55s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/56/657756/3/check/octavia-v1-dsvm-py3x-scenario/1bb3013/ : SUCCESS in 1h 06m 26s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/56/657756/3/check/octavia-v2-dsvm-noop-api/a259755/ : SUCCESS in 41m 33s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/56/657756/3/check/octavia-v2-dsvm-noop-py2-api/15a31a0/ : SUCCESS in 42m 35s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/56/657756/3/check/octavia-v2-dsvm-scenario/7496179/ : FAILURE in 1h 51m 17s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/56/657756/3/check/octavia-v2-dsvm-py2-scenario/c501271/ : FAILURE in 1h 43m 36s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/56/657756/3/check/octavia-v2-dsvm-py2-scenario-centos-7/dbfa0ea/ : FAILURE in 1h 54m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/56/657756/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/37561f2/ : FAILURE in 1h 47m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/56/657756/3/check/octavia-v2-dsvm-scenario-two-node/be76a86/ : FAILURE in 1h 07m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/56/657756/3/check/octavia-v2-dsvm-py2-scenario-two-node/24a4672/ : FAILURE in 1h 45m 52s (non-voting)\n- octavia-grenade http://logs.openstack.org/56/657756/3/check/octavia-grenade/8582571/ : FAILURE in 1h 04m 52s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/56/657756/3/check/octavia-v2-dsvm-tls-barbican/880e530/ : SUCCESS in 51m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7dbee8e9c92680068bb7e05f2abe9deeedb0a24f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-09 13:48:24.000000000","message":"Patch Set 3: Code-Review-1\n\nIf we do this, it should only be on the stable/rocky branch.","accounts_in_message":[],"_revision_number":3},{"id":"23f5f923a29dd9cf3f6b8d928a8a69abfce732ec","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-09 16:14:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"04f489e8483b2b648c9f3a25f33f8092cb243e13","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-09 18:15:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1ece60ca9958782d663712e6bd3ba98af366936e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-09 18:27:19.000000000","message":"Patch Set 4:\n\n@Michael, I think we should still merge this down the chain into stable branches as normal, to avoid having different code for no reason (impacting other future backports) and also because this could be important for upgrading control-planes with older amphora.","accounts_in_message":[],"_revision_number":4},{"id":"1e956cd911494afb67d404c00b6bc438ab5182e1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-09 18:41:07.000000000","message":"Patch Set 4: Code-Review-1\n\nHave you benchmarked the slowdown from this change?","accounts_in_message":[],"_revision_number":4},{"id":"26b9ef76d6a228727059f6b427c94e9b3f22e2a9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-09 18:41:24.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"67dd6fe9b9efd40c63fd5f0501284df0532de67e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-09 18:43:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fdf61ce305b0b39b61b18d396071f402aedebd08","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-09 18:49:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7621a05cb8b94c435a0671e3005c74235cfa183a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-09 18:50:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"36fac85f337e3cbcf8a474b2da1efaf70fdaa5a7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-09 20:46:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a3bf96260aceae63953d8b8e1e1da846b371449b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 21:15:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/56/657756/5/check/openstack-tox-cover/dc18d81/ : FAILURE in 8m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/657756/5/check/openstack-tox-lower-constraints/fa69b00/ : FAILURE in 8m 58s\n- openstack-tox-pep8 http://logs.openstack.org/56/657756/5/check/openstack-tox-pep8/fa36b1e/ : FAILURE in 6m 42s\n- openstack-tox-py27 http://logs.openstack.org/56/657756/5/check/openstack-tox-py27/61feff8/ : SUCCESS in 6m 45s\n- openstack-tox-py36 http://logs.openstack.org/56/657756/5/check/openstack-tox-py36/43db155/ : SUCCESS in 4m 51s\n- openstack-tox-docs http://logs.openstack.org/56/657756/5/check/openstack-tox-docs/ed17bb7/html/ : SUCCESS in 6m 01s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/56/657756/5/check/neutron-lbaasv2-dsvm-api/3e6e65d/ : SUCCESS in 1h 20m 53s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/56/657756/5/check/neutron-lbaasv2-dsvm-py3x-api/74d7ca2/ : SUCCESS in 1h 18m 56s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/56/657756/5/check/neutron-lbaasv2-dsvm-scenario/e070874/ : FAILURE in 1h 27m 44s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/56/657756/5/check/neutron-lbaasv2-dsvm-py3x-scenario/28b49d6/ : SUCCESS in 50m 40s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/56/657756/5/check/openstack-tox-functional/0550f50/ : FAILURE in 6m 58s\n- openstack-tox-functional-py36 http://logs.openstack.org/56/657756/5/check/openstack-tox-functional-py36/df74bb4/ : FAILURE in 7m 07s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/56/657756/5/check/octavia-v1-dsvm-scenario/4d6add2/ : SUCCESS in 1h 06m 02s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/56/657756/5/check/octavia-v1-dsvm-py3x-scenario/9c47271/ : SUCCESS in 1h 04m 39s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/56/657756/5/check/octavia-v2-dsvm-noop-api/7734295/ : SUCCESS in 44m 48s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/56/657756/5/check/octavia-v2-dsvm-noop-py2-api/b5948ee/ : SUCCESS in 42m 49s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/56/657756/5/check/octavia-v2-dsvm-scenario/138d967/ : SUCCESS in 1h 32m 24s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/56/657756/5/check/octavia-v2-dsvm-py2-scenario/3d9af6a/ : SUCCESS in 1h 37m 16s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/56/657756/5/check/octavia-v2-dsvm-py2-scenario-centos-7/575d6a8/ : FAILURE in 2h 19m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/56/657756/5/check/octavia-v2-dsvm-scenario-ubuntu-bionic/073052c/ : SUCCESS in 1h 38m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/56/657756/5/check/octavia-v2-dsvm-scenario-two-node/b89d0a9/ : FAILURE in 28m 10s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/56/657756/5/check/octavia-v2-dsvm-py2-scenario-two-node/b47c1c2/ : SUCCESS in 1h 46m 18s (non-voting)\n- octavia-grenade http://logs.openstack.org/56/657756/5/check/octavia-grenade/fb5e09b/ : SUCCESS in 1h 24m 06s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/56/657756/5/check/octavia-v2-dsvm-tls-barbican/914c542/ : SUCCESS in 55m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7b3f9a57f77c82d28aca833eca2675a4c1c27175","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-09 21:41:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f38216ca1a4d93d31cfc04fea467a55e2c6e54ae","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-09 21:45:26.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"467d616eb1fef40124edfb99208309ac5b68cdcf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 23:56:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/56/657756/7/check/openstack-tox-cover/59d4d26/cover/ : SUCCESS in 10m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/657756/7/check/openstack-tox-lower-constraints/984bf01/ : SUCCESS in 8m 07s\n- openstack-tox-pep8 http://logs.openstack.org/56/657756/7/check/openstack-tox-pep8/30eb0bd/ : FAILURE in 6m 22s\n- openstack-tox-py27 http://logs.openstack.org/56/657756/7/check/openstack-tox-py27/6d99e71/ : SUCCESS in 5m 08s\n- openstack-tox-py36 http://logs.openstack.org/56/657756/7/check/openstack-tox-py36/d46b8b0/ : SUCCESS in 4m 55s\n- openstack-tox-docs http://logs.openstack.org/56/657756/7/check/openstack-tox-docs/ee6825a/html/ : SUCCESS in 10m 17s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/56/657756/7/check/neutron-lbaasv2-dsvm-api/a2bb6cf/ : SUCCESS in 1h 24m 08s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/56/657756/7/check/neutron-lbaasv2-dsvm-py3x-api/df48fa6/ : SUCCESS in 1h 17m 00s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/56/657756/7/check/neutron-lbaasv2-dsvm-scenario/67af02c/ : FAILURE in 44m 34s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/56/657756/7/check/neutron-lbaasv2-dsvm-py3x-scenario/02a6e65/ : FAILURE in 29m 15s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/56/657756/7/check/openstack-tox-functional/f26e521/ : SUCCESS in 7m 25s\n- openstack-tox-functional-py36 http://logs.openstack.org/56/657756/7/check/openstack-tox-functional-py36/1977b2e/ : SUCCESS in 7m 26s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/56/657756/7/check/octavia-v1-dsvm-scenario/8f97877/ : SUCCESS in 1h 07m 51s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/56/657756/7/check/octavia-v1-dsvm-py3x-scenario/5448057/ : SUCCESS in 1h 06m 52s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/56/657756/7/check/octavia-v2-dsvm-noop-api/7d1e392/ : SUCCESS in 45m 21s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/56/657756/7/check/octavia-v2-dsvm-noop-py2-api/dafc208/ : SUCCESS in 42m 18s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/56/657756/7/check/octavia-v2-dsvm-scenario/af9a7d5/ : SUCCESS in 1h 28m 20s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/56/657756/7/check/octavia-v2-dsvm-py2-scenario/f7c15bd/ : SUCCESS in 1h 44m 21s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/56/657756/7/check/octavia-v2-dsvm-py2-scenario-centos-7/a3ffece/ : FAILURE in 2h 04m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/56/657756/7/check/octavia-v2-dsvm-scenario-ubuntu-bionic/45ced85/ : SUCCESS in 1h 34m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/56/657756/7/check/octavia-v2-dsvm-scenario-two-node/4488187/ : SUCCESS in 1h 41m 27s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/56/657756/7/check/octavia-v2-dsvm-py2-scenario-two-node/7a53d1d/ : FAILURE in 1h 38m 19s (non-voting)\n- octavia-grenade http://logs.openstack.org/56/657756/7/check/octavia-grenade/fc5d4f8/ : SUCCESS in 1h 14m 15s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/56/657756/7/check/octavia-v2-dsvm-tls-barbican/e353b54/ : SUCCESS in 48m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"99dc0495aad4cb29449ebfc3e1b4095d3051228f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-10 20:32:38.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"9b006498432dc635d92f741421d4f4d472da508e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 22:44:34.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/56/657756/8/check/openstack-tox-cover/b33d94c/cover/ : SUCCESS in 10m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/657756/8/check/openstack-tox-lower-constraints/7fdf73f/ : SUCCESS in 6m 52s\n- openstack-tox-pep8 http://logs.openstack.org/56/657756/8/check/openstack-tox-pep8/7307db4/ : SUCCESS in 6m 31s\n- openstack-tox-py27 http://logs.openstack.org/56/657756/8/check/openstack-tox-py27/4c765a9/ : SUCCESS in 5m 05s\n- openstack-tox-py36 http://logs.openstack.org/56/657756/8/check/openstack-tox-py36/aafe3b3/ : SUCCESS in 5m 40s\n- openstack-tox-docs http://logs.openstack.org/56/657756/8/check/openstack-tox-docs/5f410dc/html/ : SUCCESS in 6m 23s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/56/657756/8/check/neutron-lbaasv2-dsvm-api/f5ffb4d/ : SUCCESS in 1h 15m 18s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/56/657756/8/check/neutron-lbaasv2-dsvm-py3x-api/5b0e076/ : SUCCESS in 1h 17m 13s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/56/657756/8/check/neutron-lbaasv2-dsvm-scenario/b60860d/ : SUCCESS in 1h 39m 59s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/56/657756/8/check/neutron-lbaasv2-dsvm-py3x-scenario/aec0a84/ : SUCCESS in 1h 18m 38s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/56/657756/8/check/openstack-tox-functional/04c08b0/ : SUCCESS in 7m 49s\n- openstack-tox-functional-py36 http://logs.openstack.org/56/657756/8/check/openstack-tox-functional-py36/7a66e20/ : SUCCESS in 7m 24s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/56/657756/8/check/octavia-v1-dsvm-scenario/fcb7f41/ : SUCCESS in 1h 06m 35s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/56/657756/8/check/octavia-v1-dsvm-py3x-scenario/7c72c1b/ : SUCCESS in 1h 12m 27s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/56/657756/8/check/octavia-v2-dsvm-noop-api/5f7f054/ : SUCCESS in 41m 05s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/56/657756/8/check/octavia-v2-dsvm-noop-py2-api/e127131/ : SUCCESS in 43m 33s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/56/657756/8/check/octavia-v2-dsvm-scenario/d6ef266/ : FAILURE in 1h 53m 31s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/56/657756/8/check/octavia-v2-dsvm-py2-scenario/4d54de3/ : SUCCESS in 1h 30m 28s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/56/657756/8/check/octavia-v2-dsvm-py2-scenario-centos-7/f8ee5bb/ : FAILURE in 1h 57m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/56/657756/8/check/octavia-v2-dsvm-scenario-ubuntu-bionic/9e45180/ : SUCCESS in 1h 33m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/56/657756/8/check/octavia-v2-dsvm-scenario-two-node/795485c/ : SUCCESS in 1h 51m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/56/657756/8/check/octavia-v2-dsvm-py2-scenario-two-node/679e5cf/ : SUCCESS in 2h 04m 22s (non-voting)\n- octavia-grenade http://logs.openstack.org/56/657756/8/check/octavia-grenade/d7700fe/ : SUCCESS in 1h 30m 58s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/56/657756/8/check/octavia-v2-dsvm-tls-barbican/5f4e31f/ : SUCCESS in 52m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e1f8b8426f98b5937dda6391ad22337b6c51e9d7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-10 23:07:24.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"bae0df4a8ac437169add4291efc7e22750796255","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 01:22:33.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/56/657756/8/check/openstack-tox-cover/28f353e/cover/ : SUCCESS in 8m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/657756/8/check/openstack-tox-lower-constraints/5beaa9b/ : SUCCESS in 8m 11s\n- openstack-tox-pep8 http://logs.openstack.org/56/657756/8/check/openstack-tox-pep8/c3174eb/ : SUCCESS in 7m 33s\n- openstack-tox-py27 http://logs.openstack.org/56/657756/8/check/openstack-tox-py27/6803c42/ : SUCCESS in 6m 11s\n- openstack-tox-py36 http://logs.openstack.org/56/657756/8/check/openstack-tox-py36/b6e103d/ : SUCCESS in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/56/657756/8/check/openstack-tox-docs/6057e97/html/ : SUCCESS in 7m 47s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/56/657756/8/check/neutron-lbaasv2-dsvm-api/391072c/ : SUCCESS in 1h 20m 41s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/56/657756/8/check/neutron-lbaasv2-dsvm-py3x-api/d6aa686/ : SUCCESS in 1h 28m 44s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/56/657756/8/check/neutron-lbaasv2-dsvm-scenario/38c3efa/ : SUCCESS in 1h 28m 19s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/56/657756/8/check/neutron-lbaasv2-dsvm-py3x-scenario/f9cc477/ : SUCCESS in 46m 53s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/56/657756/8/check/openstack-tox-functional/8a20db7/ : SUCCESS in 7m 52s\n- openstack-tox-functional-py36 http://logs.openstack.org/56/657756/8/check/openstack-tox-functional-py36/1dab314/ : SUCCESS in 8m 21s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/56/657756/8/check/octavia-v1-dsvm-scenario/0daa96f/ : SUCCESS in 1h 03m 41s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/56/657756/8/check/octavia-v1-dsvm-py3x-scenario/8f98735/ : SUCCESS in 1h 04m 57s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/56/657756/8/check/octavia-v2-dsvm-noop-api/efd173c/ : SUCCESS in 44m 44s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/56/657756/8/check/octavia-v2-dsvm-noop-py2-api/feb83b7/ : SUCCESS in 34m 00s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/56/657756/8/check/octavia-v2-dsvm-scenario/52b5770/ : SUCCESS in 1h 36m 49s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/56/657756/8/check/octavia-v2-dsvm-py2-scenario/455644f/ : SUCCESS in 1h 41m 11s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/56/657756/8/check/octavia-v2-dsvm-py2-scenario-centos-7/ab55613/ : FAILURE in 2h 12m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/56/657756/8/check/octavia-v2-dsvm-scenario-ubuntu-bionic/f06f94f/ : SUCCESS in 1h 36m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/56/657756/8/check/octavia-v2-dsvm-scenario-two-node/a3277c4/ : FAILURE in 1h 38m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/56/657756/8/check/octavia-v2-dsvm-py2-scenario-two-node/75b8558/ : FAILURE in 1h 50m 12s (non-voting)\n- octavia-grenade http://logs.openstack.org/56/657756/8/check/octavia-grenade/4b8316d/ : SUCCESS in 1h 13m 12s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/56/657756/8/check/octavia-v2-dsvm-tls-barbican/d3b5b07/ : SUCCESS in 51m 47s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a1bc8d013e11186dbb66872570e84ef0567272ec","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-11 05:33:55.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"d3054b7fe4b01fbc590908ad8f4325cc56a2b8de","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-12 04:33:49.000000000","message":"Patch Set 8:\n\nI know Michael isn\u0027t a fan, but this patch is day and night for Rocky. A single VM with 4 cores can power many thousands of amps with it, without it it breaks at a few hundreds. While upgrading is the obvious solution, many people will be running Rocky for the foreseeable future, and this will make sure the two versions are consistent on performance at least.","accounts_in_message":[],"_revision_number":8},{"id":"da17e241c2ef90a47247007da9eacd8f9e6b14ed","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-14 20:50:12.000000000","message":"Patch Set 8: Code-Review+2\n\nEven though I hate pulling back more data from the database, that will not be needed most of the time and we are adding very niche code, I can see that people running old amp images could benefit from this. Maybe.\nIn general this is an important optimization for stable/rocky.","accounts_in_message":[],"_revision_number":8},{"id":"ac1e22cd03116b28644289d7068861e765c8a92d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-14 23:14:34.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"00dc17078436e6b91ee2e6b6fbba9fc5b6982222","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 23:14:47.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"91150f67aeceb592c3359ce1c768ad9409e36d89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 01:05:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"721e02a65bedbf2749952ac07cb4c86878f9e1a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 01:05:26.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/56/657756/8/gate/openstack-tox-lower-constraints/ecc8586/ : SUCCESS in 7m 16s\n- openstack-tox-pep8 http://logs.openstack.org/56/657756/8/gate/openstack-tox-pep8/ff7659c/ : SUCCESS in 6m 26s\n- openstack-tox-py27 http://logs.openstack.org/56/657756/8/gate/openstack-tox-py27/086dadd/ : SUCCESS in 4m 52s\n- openstack-tox-py36 http://logs.openstack.org/56/657756/8/gate/openstack-tox-py36/215adc2/ : SUCCESS in 6m 27s\n- openstack-tox-docs http://logs.openstack.org/56/657756/8/gate/openstack-tox-docs/7b67077/html/ : SUCCESS in 5m 45s\n- openstack-tox-functional http://logs.openstack.org/56/657756/8/gate/openstack-tox-functional/8d75f3f/ : SUCCESS in 11m 58s\n- openstack-tox-functional-py36 http://logs.openstack.org/56/657756/8/gate/openstack-tox-functional-py36/198a1a9/ : SUCCESS in 9m 19s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/56/657756/8/gate/octavia-v2-dsvm-noop-api/9513015/ : SUCCESS in 43m 01s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/56/657756/8/gate/octavia-v2-dsvm-noop-py2-api/92dfe2d/ : SUCCESS in 45m 40s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/56/657756/8/gate/octavia-v2-dsvm-scenario/079036d/ : SUCCESS in 1h 33m 51s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/56/657756/8/gate/octavia-v2-dsvm-py2-scenario/0f874f5/ : SUCCESS in 1h 19m 17s\n- octavia-grenade http://logs.openstack.org/56/657756/8/gate/octavia-grenade/96da43c/ : SUCCESS in 1h 25m 45s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"24e77d606dce174acf16fadcf32ba4b5182aa49a","revisions":{"a18b58c296d684f0c2db3e9c41556a6ae2a7e097":{"kind":"REWORK","_number":1,"created":"2019-05-08 08:45:06.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/56/657756/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/657756/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/657756/1"}}},"commit":{"parents":[{"commit":"28f4f9e44a07c224ea4da74169470c4e5cd2b17e","subject":"Fix missing INCLUDEPKGS var in haproxy element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28f4f9e44a07c224ea4da74169470c4e5cd2b17e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-08 08:44:23.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-08 08:44:23.000000000","tz":-420},"subject":"[WIP] Testing performance improvement on non-udp health checks","message":"[WIP] Testing performance improvement on non-udp health checks\n\nChange-Id: I7fde54084b39c1d0529cfb0bcfb79697d63ea6ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a18b58c296d684f0c2db3e9c41556a6ae2a7e097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a18b58c296d684f0c2db3e9c41556a6ae2a7e097"}]},"branch":"refs/heads/master"},"90a65bdc89be6e1db4e20e22b2f5e6043ab59c7d":{"kind":"REWORK","_number":2,"created":"2019-05-08 08:52:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/56/657756/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/657756/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/657756/2"}}},"commit":{"parents":[{"commit":"28f4f9e44a07c224ea4da74169470c4e5cd2b17e","subject":"Fix missing INCLUDEPKGS var in haproxy element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28f4f9e44a07c224ea4da74169470c4e5cd2b17e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-08 08:44:23.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-08 08:52:40.000000000","tz":-420},"subject":"[WIP] Testing performance improvement on non-udp health checks","message":"[WIP] Testing performance improvement on non-udp health checks\n\nChange-Id: I7fde54084b39c1d0529cfb0bcfb79697d63ea6ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/90a65bdc89be6e1db4e20e22b2f5e6043ab59c7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/90a65bdc89be6e1db4e20e22b2f5e6043ab59c7d"}]},"branch":"refs/heads/master"},"ce46f8af3f2c8fba04b0cd4060784b6a64283a9d":{"kind":"REWORK","_number":3,"created":"2019-05-09 07:02:33.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/56/657756/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/657756/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/657756/3"}}},"commit":{"parents":[{"commit":"28f4f9e44a07c224ea4da74169470c4e5cd2b17e","subject":"Fix missing INCLUDEPKGS var in haproxy element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28f4f9e44a07c224ea4da74169470c4e5cd2b17e"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-09 06:59:05.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-09 07:00:47.000000000","tz":-420},"subject":"Performance improvement for non-udp health checks","message":"Performance improvement for non-udp health checks\n\nWhen no UDP listeners are present, skip the UDP health-check code\nbranch, which prevents expensive and unnecessary DB calls.\nAlso optimise the UDP health-check code so it only fetches information\nfor relevant listeners.\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I7fde54084b39c1d0529cfb0bcfb79697d63ea6ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce46f8af3f2c8fba04b0cd4060784b6a64283a9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce46f8af3f2c8fba04b0cd4060784b6a64283a9d"}]},"branch":"refs/heads/master"},"63a4c87c511c0d8a86076343246c5548fba29f69":{"kind":"REWORK","_number":4,"created":"2019-05-09 18:15:35.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/56/657756/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/657756/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/657756/4"}}},"commit":{"parents":[{"commit":"28f4f9e44a07c224ea4da74169470c4e5cd2b17e","subject":"Fix missing INCLUDEPKGS var in haproxy element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28f4f9e44a07c224ea4da74169470c4e5cd2b17e"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-09 06:59:05.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-09 18:15:22.000000000","tz":-420},"subject":"Performance improvement for non-udp health checks","message":"Performance improvement for non-udp health checks\n\nWhen no UDP listeners are present, skip the UDP health-check code\nbranch, which prevents expensive and unnecessary DB calls.\nAlso optimise the UDP health-check code so it only fetches information\nfor relevant listeners.\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I7fde54084b39c1d0529cfb0bcfb79697d63ea6ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63a4c87c511c0d8a86076343246c5548fba29f69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63a4c87c511c0d8a86076343246c5548fba29f69"}]},"branch":"refs/heads/master"},"5702da1d625a9a968c6cf2c206c3d1657e7ad31d":{"kind":"REWORK","_number":5,"created":"2019-05-09 18:49:46.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/56/657756/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/657756/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/657756/5"}}},"commit":{"parents":[{"commit":"28f4f9e44a07c224ea4da74169470c4e5cd2b17e","subject":"Fix missing INCLUDEPKGS var in haproxy element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28f4f9e44a07c224ea4da74169470c4e5cd2b17e"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-09 06:59:05.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-09 18:49:34.000000000","tz":-420},"subject":"Performance improvement for non-udp health checks","message":"Performance improvement for non-udp health checks\n\nWhen no UDP listeners are present, skip the UDP health-check code\nbranch, which prevents expensive and unnecessary DB calls.\nAlso optimise the UDP health-check code so it only fetches information\nfor relevant listeners.\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I7fde54084b39c1d0529cfb0bcfb79697d63ea6ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5702da1d625a9a968c6cf2c206c3d1657e7ad31d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5702da1d625a9a968c6cf2c206c3d1657e7ad31d"}]},"branch":"refs/heads/master"},"03fe6e34e0ce317b54960e3a7a05280ed6a7e0ab":{"kind":"REWORK","_number":6,"created":"2019-05-09 21:41:24.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/56/657756/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/657756/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/657756/6"}}},"commit":{"parents":[{"commit":"28f4f9e44a07c224ea4da74169470c4e5cd2b17e","subject":"Fix missing INCLUDEPKGS var in haproxy element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28f4f9e44a07c224ea4da74169470c4e5cd2b17e"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-09 06:59:05.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-09 21:41:08.000000000","tz":-420},"subject":"Performance improvement for non-udp health checks","message":"Performance improvement for non-udp health checks\n\nWhen no UDP listeners are present, skip the UDP health-check code\nbranch, which prevents expensive and unnecessary DB calls.\nAlso optimise the UDP health-check code so it only fetches information\nfor relevant listeners.\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I7fde54084b39c1d0529cfb0bcfb79697d63ea6ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/03fe6e34e0ce317b54960e3a7a05280ed6a7e0ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/03fe6e34e0ce317b54960e3a7a05280ed6a7e0ab"}]},"branch":"refs/heads/master"},"553753bfcc5c06c31cdd8ef138fcec02c8d6ea45":{"kind":"REWORK","_number":7,"created":"2019-05-09 21:45:26.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/56/657756/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/657756/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/657756/7"}}},"commit":{"parents":[{"commit":"28f4f9e44a07c224ea4da74169470c4e5cd2b17e","subject":"Fix missing INCLUDEPKGS var in haproxy element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28f4f9e44a07c224ea4da74169470c4e5cd2b17e"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-09 06:59:05.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-09 21:45:13.000000000","tz":-420},"subject":"Performance improvement for non-udp health checks","message":"Performance improvement for non-udp health checks\n\nWhen no UDP listeners are present, skip the UDP health-check code\nbranch, which prevents expensive and unnecessary DB calls.\nAlso optimise the UDP health-check code so it only fetches information\nfor relevant listeners.\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I7fde54084b39c1d0529cfb0bcfb79697d63ea6ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/553753bfcc5c06c31cdd8ef138fcec02c8d6ea45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/553753bfcc5c06c31cdd8ef138fcec02c8d6ea45"}]},"branch":"refs/heads/master"},"24e77d606dce174acf16fadcf32ba4b5182aa49a":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-05-10 20:32:38.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/56/657756/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/657756/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/657756/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/657756/8"}}},"commit":{"parents":[{"commit":"1a87298ac774d921709ab9cbe5016414db65783e","subject":"Bandit test exclusions syntax change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a87298ac774d921709ab9cbe5016414db65783e"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-09 06:59:05.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-10 20:32:38.000000000","tz":0},"subject":"Performance improvement for non-udp health checks","message":"Performance improvement for non-udp health checks\n\nWhen no UDP listeners are present, skip the UDP health-check code\nbranch, which prevents expensive and unnecessary DB calls.\nAlso optimise the UDP health-check code so it only fetches information\nfor relevant listeners.\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I7fde54084b39c1d0529cfb0bcfb79697d63ea6ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24e77d606dce174acf16fadcf32ba4b5182aa49a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24e77d606dce174acf16fadcf32ba4b5182aa49a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
