)]}'
{"id":"openstack%2Foctavia~658476","triplet_id":"openstack%2Foctavia~master~I02fbc02ce8d623b75a7d568ea48bb5e5e6c66b58","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I02fbc02ce8d623b75a7d568ea48bb5e5e6c66b58","subject":"Bandit test exclusions syntax change","status":"MERGED","created":"2019-05-10 20:31:50.000000000","updated":"2019-05-15 13:49:31.000000000","submitted":"2019-05-11 08:02:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658476-1557561749196-96febb6b","meta_rev_id":"84cd08598155d918140b32626ad64b610652dfd3","_number":658476,"virtual_id_number":658476,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-05-11 08:02:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-05-10 21:00:24.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-05-10 20:44:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-05-11 01:56:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 20:44:54.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-05-10 21:00:24.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-05-11 08:02:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08fe86428c9d1f0b231c06a26b02176f451b2850","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-10 20:31:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0347ff676d35cb1c7d9a61d58b0a771de3ab13a5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-10 20:44:54.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"37d78198593ee08e80fd14b043953f9eb35fcb4c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-10 21:00:24.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM. https://github.com/PyCQA/bandit/pull/450 and released in bandit 1.6.0 (May 9, 2019).","accounts_in_message":[],"_revision_number":1},{"id":"e89755908169bb92a756b0fbf5c3ab07dc974745","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 22:32:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/76/658476/1/check/openstack-tox-cover/ef0198d/cover/ : SUCCESS in 9m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/658476/1/check/openstack-tox-lower-constraints/ad98de9/ : SUCCESS in 8m 03s\n- openstack-tox-pep8 http://logs.openstack.org/76/658476/1/check/openstack-tox-pep8/59ca89e/ : SUCCESS in 6m 23s\n- openstack-tox-py27 http://logs.openstack.org/76/658476/1/check/openstack-tox-py27/80b1f85/ : SUCCESS in 6m 15s\n- openstack-tox-py36 http://logs.openstack.org/76/658476/1/check/openstack-tox-py36/630ff89/ : SUCCESS in 5m 42s\n- openstack-tox-docs http://logs.openstack.org/76/658476/1/check/openstack-tox-docs/3b9e171/html/ : SUCCESS in 6m 11s\n- build-openstack-releasenotes http://logs.openstack.org/76/658476/1/check/build-openstack-releasenotes/45d67e3/html/ : SUCCESS in 4m 58s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/76/658476/1/check/neutron-lbaasv2-dsvm-api/4a43d4a/ : SUCCESS in 1h 26m 04s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/76/658476/1/check/neutron-lbaasv2-dsvm-py3x-api/b66ea66/ : SUCCESS in 1h 18m 17s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/76/658476/1/check/neutron-lbaasv2-dsvm-scenario/e7b450a/ : SUCCESS in 1h 33m 01s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/76/658476/1/check/neutron-lbaasv2-dsvm-py3x-scenario/7bfd9bf/ : SUCCESS in 1h 34m 01s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/76/658476/1/check/openstack-tox-functional/2f8a34e/ : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 http://logs.openstack.org/76/658476/1/check/openstack-tox-functional-py36/006ed65/ : SUCCESS in 7m 21s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/76/658476/1/check/octavia-v1-dsvm-scenario/f440b22/ : SUCCESS in 1h 04m 08s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/76/658476/1/check/octavia-v1-dsvm-py3x-scenario/e8bf546/ : SUCCESS in 1h 10m 58s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-noop-api/6e07380/ : SUCCESS in 48m 40s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-noop-py2-api/fdcd86c/ : SUCCESS in 44m 41s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-scenario/43d367f/ : FAILURE in 1h 40m 16s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-py2-scenario/a41d13d/ : SUCCESS in 1h 35m 43s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-py2-scenario-centos-7/f28256d/ : FAILURE in 1h 58m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/74b997f/ : SUCCESS in 1h 52m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-scenario-two-node/af8285f/ : SUCCESS in 1h 47m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-py2-scenario-two-node/11de1a0/ : SUCCESS in 1h 54m 02s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-tls-barbican/8f2d4d3/ : SUCCESS in 52m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9ae4b21d70c111ffc102ae26372c68d25af16da1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-10 23:00:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7fa061cf0c4d08275b9e087e36170627792db358","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 01:14:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/76/658476/1/check/openstack-tox-cover/bfa9cde/cover/ : SUCCESS in 8m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/658476/1/check/openstack-tox-lower-constraints/99b82c5/ : SUCCESS in 7m 00s\n- openstack-tox-pep8 http://logs.openstack.org/76/658476/1/check/openstack-tox-pep8/3629694/ : SUCCESS in 6m 28s\n- openstack-tox-py27 http://logs.openstack.org/76/658476/1/check/openstack-tox-py27/0ced85f/ : SUCCESS in 5m 00s\n- openstack-tox-py36 http://logs.openstack.org/76/658476/1/check/openstack-tox-py36/b1d4a0f/ : SUCCESS in 5m 07s\n- openstack-tox-docs http://logs.openstack.org/76/658476/1/check/openstack-tox-docs/7aa739c/html/ : SUCCESS in 5m 59s\n- build-openstack-releasenotes http://logs.openstack.org/76/658476/1/check/build-openstack-releasenotes/27f9471/html/ : SUCCESS in 4m 43s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/76/658476/1/check/neutron-lbaasv2-dsvm-api/709fa42/ : SUCCESS in 1h 18m 01s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/76/658476/1/check/neutron-lbaasv2-dsvm-py3x-api/f38e6e7/ : SUCCESS in 1h 09m 31s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/76/658476/1/check/neutron-lbaasv2-dsvm-scenario/b68a302/ : SUCCESS in 1h 34m 20s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/76/658476/1/check/neutron-lbaasv2-dsvm-py3x-scenario/5a23a82/ : FAILURE in 1h 15m 56s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/76/658476/1/check/openstack-tox-functional/3abf95e/ : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 http://logs.openstack.org/76/658476/1/check/openstack-tox-functional-py36/5ae944e/ : SUCCESS in 7m 22s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/76/658476/1/check/octavia-v1-dsvm-scenario/e230d5b/ : SUCCESS in 59m 11s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/76/658476/1/check/octavia-v1-dsvm-py3x-scenario/1687acd/ : SUCCESS in 58m 04s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-noop-api/5aa4db5/ : SUCCESS in 42m 53s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-noop-py2-api/65f64da/ : FAILURE in 35m 45s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-scenario/94da99f/ : SUCCESS in 1h 27m 27s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-py2-scenario/a315105/ : SUCCESS in 1h 41m 30s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-py2-scenario-centos-7/3d2069c/ : FAILURE in 2h 02m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/d60dd02/ : SUCCESS in 1h 41m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-scenario-two-node/2eed1f8/ : FAILURE in 1h 24m 13s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-py2-scenario-two-node/f3674c0/ : SUCCESS in 1h 50m 01s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-tls-barbican/a2798a9/ : SUCCESS in 53m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bef4f5fcd721524d41fafb1cd6d0cc458d9e3816","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-11 01:56:28.000000000","message":"Patch Set 1: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"83e532a41d0522838a6a441a35196d938dd9e24a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 04:09:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/76/658476/1/check/openstack-tox-cover/3536233/cover/ : SUCCESS in 8m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/658476/1/check/openstack-tox-lower-constraints/597ec91/ : SUCCESS in 7m 06s\n- openstack-tox-pep8 http://logs.openstack.org/76/658476/1/check/openstack-tox-pep8/9aa108d/ : SUCCESS in 6m 10s\n- openstack-tox-py27 http://logs.openstack.org/76/658476/1/check/openstack-tox-py27/a225105/ : SUCCESS in 4m 53s\n- openstack-tox-py36 http://logs.openstack.org/76/658476/1/check/openstack-tox-py36/9187c4a/ : SUCCESS in 5m 43s\n- openstack-tox-docs http://logs.openstack.org/76/658476/1/check/openstack-tox-docs/90bce0b/html/ : SUCCESS in 5m 35s\n- build-openstack-releasenotes http://logs.openstack.org/76/658476/1/check/build-openstack-releasenotes/809cd1c/html/ : SUCCESS in 4m 27s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/76/658476/1/check/neutron-lbaasv2-dsvm-api/d15b1ab/ : SUCCESS in 1h 15m 43s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/76/658476/1/check/neutron-lbaasv2-dsvm-py3x-api/59a56c0/ : SUCCESS in 1h 09m 23s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/76/658476/1/check/neutron-lbaasv2-dsvm-scenario/2accd38/ : SUCCESS in 1h 26m 18s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/76/658476/1/check/neutron-lbaasv2-dsvm-py3x-scenario/155bfe1/ : SUCCESS in 1h 20m 02s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/76/658476/1/check/openstack-tox-functional/34c5d0e/ : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 http://logs.openstack.org/76/658476/1/check/openstack-tox-functional-py36/38c99fa/ : SUCCESS in 6m 37s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/76/658476/1/check/octavia-v1-dsvm-scenario/df42bcb/ : FAILURE in 57m 10s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/76/658476/1/check/octavia-v1-dsvm-py3x-scenario/00e5db5/ : SUCCESS in 1h 01m 19s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-noop-api/dd5f509/ : SUCCESS in 32m 34s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-noop-py2-api/910bcd4/ : SUCCESS in 41m 51s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-scenario/6072ee1/ : SUCCESS in 1h 40m 01s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-py2-scenario/fcd75d3/ : SUCCESS in 1h 29m 22s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-py2-scenario-centos-7/6b5f0c0/ : FAILURE in 2h 00m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/d1a4674/ : FAILURE in 35m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-scenario-two-node/40de4d5/ : SUCCESS in 1h 46m 19s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-py2-scenario-two-node/d8cbef7/ : SUCCESS in 1h 02m 53s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-tls-barbican/9eabbd5/ : SUCCESS in 43m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34f6f9876974bdca5816943971ee03d447c4ccb1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-11 04:10:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"81cbdb4f76bab43df3ee4c34c31e7d1b50611a40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 06:19:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/76/658476/1/check/openstack-tox-cover/041a86f/cover/ : SUCCESS in 8m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/658476/1/check/openstack-tox-lower-constraints/d9d2950/ : SUCCESS in 7m 00s\n- openstack-tox-pep8 http://logs.openstack.org/76/658476/1/check/openstack-tox-pep8/cbe6572/ : SUCCESS in 6m 06s\n- openstack-tox-py27 http://logs.openstack.org/76/658476/1/check/openstack-tox-py27/2e26097/ : SUCCESS in 4m 50s\n- openstack-tox-py36 http://logs.openstack.org/76/658476/1/check/openstack-tox-py36/617ce4e/ : SUCCESS in 5m 30s\n- openstack-tox-docs http://logs.openstack.org/76/658476/1/check/openstack-tox-docs/c5bba1b/html/ : SUCCESS in 5m 53s\n- build-openstack-releasenotes http://logs.openstack.org/76/658476/1/check/build-openstack-releasenotes/29f0a87/html/ : SUCCESS in 4m 47s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/76/658476/1/check/neutron-lbaasv2-dsvm-api/db3af52/ : SUCCESS in 1h 17m 47s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/76/658476/1/check/neutron-lbaasv2-dsvm-py3x-api/3df944b/ : SUCCESS in 1h 18m 06s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/76/658476/1/check/neutron-lbaasv2-dsvm-scenario/4ab23c2/ : SUCCESS in 1h 25m 38s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/76/658476/1/check/neutron-lbaasv2-dsvm-py3x-scenario/ec99bfe/ : SUCCESS in 1h 22m 42s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/76/658476/1/check/openstack-tox-functional/6c7801b/ : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 http://logs.openstack.org/76/658476/1/check/openstack-tox-functional-py36/9a12857/ : SUCCESS in 6m 30s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/76/658476/1/check/octavia-v1-dsvm-scenario/b54ded9/ : SUCCESS in 1h 15m 58s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/76/658476/1/check/octavia-v1-dsvm-py3x-scenario/4ccc7b9/ : SUCCESS in 1h 10m 51s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-noop-api/13aba09/ : SUCCESS in 39m 16s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-noop-py2-api/e7321a8/ : SUCCESS in 43m 13s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-scenario/1494b75/ : SUCCESS in 1h 26m 01s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-py2-scenario/998f412/ : SUCCESS in 1h 34m 23s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-py2-scenario-centos-7/1dcd631/ : FAILURE in 2h 03m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/1859218/ : SUCCESS in 1h 41m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-scenario-two-node/8a0d7e8/ : SUCCESS in 1h 53m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-py2-scenario-two-node/a04281a/ : SUCCESS in 1h 40m 26s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/76/658476/1/check/octavia-v2-dsvm-tls-barbican/e084e1d/ : SUCCESS in 1h 01m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96cdf6ec67e676b9ccc00a0b311367660b0fd137","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 06:19:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4726298a0f431a4ebf479a28c608fbf92618e0f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 08:02:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/76/658476/1/gate/openstack-tox-lower-constraints/a413d2c/ : SUCCESS in 7m 24s\n- openstack-tox-pep8 http://logs.openstack.org/76/658476/1/gate/openstack-tox-pep8/e5a08a5/ : SUCCESS in 5m 54s\n- openstack-tox-py27 http://logs.openstack.org/76/658476/1/gate/openstack-tox-py27/0b0a6ec/ : SUCCESS in 4m 47s\n- openstack-tox-py36 http://logs.openstack.org/76/658476/1/gate/openstack-tox-py36/4dfb6ad/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/76/658476/1/gate/openstack-tox-docs/80771de/html/ : SUCCESS in 6m 49s\n- build-openstack-releasenotes http://logs.openstack.org/76/658476/1/gate/build-openstack-releasenotes/aa471cf/html/ : SUCCESS in 4m 54s\n- openstack-tox-functional http://logs.openstack.org/76/658476/1/gate/openstack-tox-functional/c6288b0/ : SUCCESS in 6m 38s\n- openstack-tox-functional-py36 http://logs.openstack.org/76/658476/1/gate/openstack-tox-functional-py36/a562283/ : SUCCESS in 6m 55s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/76/658476/1/gate/octavia-v1-dsvm-scenario/4138978/ : SUCCESS in 1h 06m 10s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/76/658476/1/gate/octavia-v1-dsvm-py3x-scenario/1c7e748/ : SUCCESS in 1h 07m 05s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/76/658476/1/gate/neutron-lbaasv2-dsvm-api/0b16c89/ : SUCCESS in 1h 19m 03s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/76/658476/1/gate/neutron-lbaasv2-dsvm-py3x-api/23a48cf/ : SUCCESS in 1h 19m 53s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/76/658476/1/gate/octavia-v2-dsvm-noop-api/42da564/ : SUCCESS in 40m 25s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/76/658476/1/gate/octavia-v2-dsvm-noop-py2-api/7fde899/ : SUCCESS in 42m 04s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/76/658476/1/gate/octavia-v2-dsvm-scenario/1795c0a/ : SUCCESS in 1h 30m 36s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/76/658476/1/gate/octavia-v2-dsvm-py2-scenario/8ee3fb1/ : SUCCESS in 1h 37m 44s","accounts_in_message":[],"_revision_number":1},{"id":"bae6202d07c43365123b8353c7f9a00c2147e64d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 08:02:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"170bdc8b9512ceb394320aae64cd9722d799806e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-13 09:13:44.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 30a4feb820baa04a5a2a1b7f65f80688e8238604","accounts_in_message":[],"_revision_number":1},{"id":"84cd08598155d918140b32626ad64b610652dfd3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-05-15 13:49:31.000000000","message":"Patch Set 1:\n\nI think this might cause a breakage when 1.6.1 gets released, as this behavior was a bug introduced in 1.6.0.  We just blacklisted 1.6.0 in neutron instead, https://review.opendev.org/#/c/658233/\n\n[0] https://github.com/PyCQA/bandit/issues/488\n[1] https://github.com/PyCQA/bandit/pull/489","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1a87298ac774d921709ab9cbe5016414db65783e","revisions":{"1a87298ac774d921709ab9cbe5016414db65783e":{"kind":"REWORK","_number":1,"created":"2019-05-10 20:31:50.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/76/658476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/658476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/658476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/658476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/658476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/658476/1"}}},"commit":{"parents":[{"commit":"28f4f9e44a07c224ea4da74169470c4e5cd2b17e","subject":"Fix missing INCLUDEPKGS var in haproxy element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28f4f9e44a07c224ea4da74169470c4e5cd2b17e"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-10 20:30:32.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-10 20:30:32.000000000","tz":-420},"subject":"Bandit test exclusions syntax change","message":"Bandit test exclusions syntax change\n\nNewer versions of bandit broke their handling of test exclusion\ndirectories. It is now necessary to glob the contents of the directory\nrather than just list the base path.\n\nChange-Id: I02fbc02ce8d623b75a7d568ea48bb5e5e6c66b58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a87298ac774d921709ab9cbe5016414db65783e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a87298ac774d921709ab9cbe5016414db65783e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
