)]}'
{"id":"openstack%2Foctavia~658764","triplet_id":"openstack%2Foctavia~stable%2Frocky~I02fbc02ce8d623b75a7d568ea48bb5e5e6c66b58","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"I02fbc02ce8d623b75a7d568ea48bb5e5e6c66b58","subject":"Bandit test exclusions syntax change","status":"MERGED","created":"2019-05-13 09:13:51.000000000","updated":"2019-05-14 01:51:04.000000000","submitted":"2019-05-14 01:51:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658764-1557798664354-88673ecb","meta_rev_id":"bd552253f5bac7e6264fd0247037a87267a98733","_number":658764,"virtual_id_number":658764,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-05-14 01:51:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-05-14 00:11:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-05-14 00:11:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 00:11:26.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-05-14 01:51:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ecd8a930ac31d2d06d1ec0043c11e27beb896eed","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-13 09:13:51.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"3478b314791ef1dc55cbc550d60782ac5570e684","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-13 09:14:05.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit a3a141167d2553596c1d1662c110fa589517627d","accounts_in_message":[],"_revision_number":1},{"id":"64a39ac3dd4bf8bf4b7b878e9620fae40bc30a97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 10:53:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/64/658764/1/check/openstack-tox-lower-constraints/b97fdc7/ : SUCCESS in 10m 45s\n- openstack-tox-pep8 http://logs.openstack.org/64/658764/1/check/openstack-tox-pep8/8f51c12/ : SUCCESS in 11m 11s\n- openstack-tox-py27 http://logs.openstack.org/64/658764/1/check/openstack-tox-py27/0c45162/ : SUCCESS in 7m 47s\n- openstack-tox-py35 http://logs.openstack.org/64/658764/1/check/openstack-tox-py35/c982456/ : SUCCESS in 6m 32s\n- openstack-tox-docs http://logs.openstack.org/64/658764/1/check/openstack-tox-docs/39247f7/html/ : SUCCESS in 7m 37s\n- build-openstack-releasenotes http://logs.openstack.org/64/658764/1/check/build-openstack-releasenotes/ca878dc/html/ : SUCCESS in 5m 30s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/64/658764/1/check/neutron-lbaasv2-dsvm-api/dee6b04/ : SUCCESS in 1h 12m 49s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/64/658764/1/check/neutron-lbaasv2-dsvm-py3x-api/2490ac5/ : SUCCESS in 1h 15m 34s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/64/658764/1/check/neutron-lbaasv2-dsvm-scenario/556e82a/ : SUCCESS in 58m 22s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/64/658764/1/check/neutron-lbaasv2-dsvm-py3x-scenario/b474c4f/ : SUCCESS in 1h 22m 12s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/64/658764/1/check/openstack-tox-functional/02d73ab/ : SUCCESS in 5m 53s\n- openstack-tox-functional-py35 http://logs.openstack.org/64/658764/1/check/openstack-tox-functional-py35/9e5d024/ : SUCCESS in 7m 53s\n- openstack-tox-functional-py36 http://logs.openstack.org/64/658764/1/check/openstack-tox-functional-py36/78f7f2c/ : SUCCESS in 6m 14s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/64/658764/1/check/octavia-v1-dsvm-scenario/0bbaedb/ : SUCCESS in 1h 00m 19s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/64/658764/1/check/octavia-v1-dsvm-py3x-scenario/bb021b2/ : SUCCESS in 1h 00m 36s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/64/658764/1/check/octavia-v1-dsvm-scenario-multinode/0ec426e/ : FAILURE in 1h 17m 15s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/64/658764/1/check/octavia-v1-dsvm-py3x-scenario-multinode/d08b3b4/ : FAILURE in 1h 14m 58s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/64/658764/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/34393e0/ : SUCCESS in 1h 09m 14s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/64/658764/1/check/octavia-v2-dsvm-noop-api/7f48fad/ : SUCCESS in 40m 57s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/64/658764/1/check/octavia-v2-dsvm-noop-py35-api/6d53423/ : SUCCESS in 46m 33s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/64/658764/1/check/octavia-v2-dsvm-scenario/85cd266/ : SUCCESS in 1h 34m 42s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/64/658764/1/check/octavia-v2-dsvm-py35-scenario/6ebcede/ : SUCCESS in 1h 23m 32s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/64/658764/1/check/openstack-tox-cover/6910408/cover/ : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":1},{"id":"829ae84a6a78db27bce110392346322675457d45","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-14 00:11:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c64598db3be5c6b4a59af9bef68933ce44e2a8af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 00:11:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c594ad8f8b3250da913277f459a36540c8758b0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 01:51:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bd552253f5bac7e6264fd0247037a87267a98733","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 01:51:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/64/658764/1/gate/openstack-tox-lower-constraints/4b488d4/ : SUCCESS in 11m 26s\n- openstack-tox-pep8 http://logs.openstack.org/64/658764/1/gate/openstack-tox-pep8/257488b/ : SUCCESS in 8m 46s\n- openstack-tox-py27 http://logs.openstack.org/64/658764/1/gate/openstack-tox-py27/dc5bb59/ : SUCCESS in 6m 59s\n- openstack-tox-py35 http://logs.openstack.org/64/658764/1/gate/openstack-tox-py35/5184382/ : SUCCESS in 7m 28s\n- openstack-tox-docs http://logs.openstack.org/64/658764/1/gate/openstack-tox-docs/6918cd7/html/ : SUCCESS in 7m 52s\n- build-openstack-releasenotes http://logs.openstack.org/64/658764/1/gate/build-openstack-releasenotes/9c20ced/html/ : SUCCESS in 5m 46s\n- openstack-tox-functional http://logs.openstack.org/64/658764/1/gate/openstack-tox-functional/3adad17/ : SUCCESS in 8m 01s\n- openstack-tox-functional-py35 http://logs.openstack.org/64/658764/1/gate/openstack-tox-functional-py35/a9a7087/ : SUCCESS in 9m 04s\n- openstack-tox-functional-py36 http://logs.openstack.org/64/658764/1/gate/openstack-tox-functional-py36/9bfa5a1/ : SUCCESS in 8m 29s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/64/658764/1/gate/octavia-v1-dsvm-scenario/05b9432/ : SUCCESS in 58m 08s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/64/658764/1/gate/octavia-v1-dsvm-py3x-scenario/830518b/ : SUCCESS in 1h 00m 40s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/64/658764/1/gate/octavia-v1-dsvm-scenario-multinode/54c053d/ : FAILURE in 1h 19m 54s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/64/658764/1/gate/octavia-v1-dsvm-py3x-scenario-multinode/850d24a/ : FAILURE in 1h 17m 29s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/64/658764/1/gate/neutron-lbaasv2-dsvm-py3x-api/3ade005/ : SUCCESS in 1h 23m 30s\n- openstack-tox-cover http://logs.openstack.org/64/658764/1/gate/openstack-tox-cover/9261cd7/cover/ : SUCCESS in 10m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2bf15e86691c676ad0ebd6890488d0cbf615c21d","revisions":{"2bf15e86691c676ad0ebd6890488d0cbf615c21d":{"kind":"REWORK","_number":1,"created":"2019-05-13 09:13:51.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/64/658764/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/658764/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/658764/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/658764/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/658764/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/658764/1"}}},"commit":{"parents":[{"commit":"cedb553b144688b0de6440e0e1a5eed0ff838b51","subject":"Fix IPv6 in Active/Standby topology on CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cedb553b144688b0de6440e0e1a5eed0ff838b51"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-10 20:30:32.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-13 09:13:51.000000000","tz":0},"subject":"Bandit test exclusions syntax change","message":"Bandit test exclusions syntax change\n\nNewer versions of bandit broke their handling of test exclusion\ndirectories. It is now necessary to glob the contents of the directory\nrather than just list the base path.\n\nChange-Id: I02fbc02ce8d623b75a7d568ea48bb5e5e6c66b58\n(cherry picked from commit 1a87298ac774d921709ab9cbe5016414db65783e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2bf15e86691c676ad0ebd6890488d0cbf615c21d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2bf15e86691c676ad0ebd6890488d0cbf615c21d"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
