)]}'
{"id":"openstack%2Foctavia~659020","triplet_id":"openstack%2Foctavia~master~Ic533412ef876eb76702c972d87b4c2af140accbd","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ic533412ef876eb76702c972d87b4c2af140accbd","subject":"Blacklist bandit 1.6.0","status":"ABANDONED","created":"2019-05-14 10:11:25.000000000","updated":"2019-05-23 05:37:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2f552fbe908950a1ce9bdf908f77f1d752e8786b","_number":659020,"virtual_id_number":659020,"owner":{"_account_id":27809,"name":"hanxiang gao","email":"gaohanxiang@126.com","username":"gao.hanxiang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":-1,"date":"2019-05-14 12:26:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":-2,"date":"2019-05-14 10:14:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 10:14:07.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-05-14 12:26:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e9e978c4eb6e500075194dee8c69a98b6b3a6fd","author":{"_account_id":27809,"name":"hanxiang gao","email":"gaohanxiang@126.com","username":"gao.hanxiang"},"date":"2019-05-14 10:11:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3253c4c8d1274bd618252689ecd74a177a66c2a","author":{"_account_id":27809,"name":"hanxiang gao","email":"gaohanxiang@126.com","username":"gao.hanxiang"},"date":"2019-05-14 10:11:55.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 80bd128d7c7867629aafd02625cdbcd3a98eace5","accounts_in_message":[],"_revision_number":1},{"id":"b0d66d763499951d6ae7d9165616be47bed9d2c6","author":{"_account_id":27809,"name":"hanxiang gao","email":"gaohanxiang@126.com","username":"gao.hanxiang"},"date":"2019-05-14 10:12:05.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 797cdd34ae658439e406560bdc87b6ef099de551","accounts_in_message":[],"_revision_number":1},{"id":"774fea2bda0313d66216ea5ec8fd652775a3166b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-14 10:14:07.000000000","message":"Patch Set 1: Code-Review-2\n\nThere\u0027s no reason to do this, as we\u0027ve simply resolved the issue.","accounts_in_message":[],"_revision_number":1},{"id":"494565bdde0fd6dd415974c86c731c7b0de13a25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 12:26:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/20/659020/1/check/build-openstack-api-ref/69c25aa/html/ : SUCCESS in 5m 09s\n- requirements-check http://logs.openstack.org/20/659020/1/check/requirements-check/4242b91/ : FAILURE in 3m 46s\n- openstack-tox-cover http://logs.openstack.org/20/659020/1/check/openstack-tox-cover/8685486/cover/ : SUCCESS in 11m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/659020/1/check/openstack-tox-lower-constraints/45bcc96/ : SUCCESS in 7m 04s\n- openstack-tox-pep8 http://logs.openstack.org/20/659020/1/check/openstack-tox-pep8/d9ebc36/ : FAILURE in 9m 12s\n- openstack-tox-py27 http://logs.openstack.org/20/659020/1/check/openstack-tox-py27/b3eb51b/ : SUCCESS in 5m 18s\n- openstack-tox-py36 http://logs.openstack.org/20/659020/1/check/openstack-tox-py36/bfc9982/ : SUCCESS in 8m 14s\n- openstack-tox-docs http://logs.openstack.org/20/659020/1/check/openstack-tox-docs/d1a6247/html/ : SUCCESS in 5m 53s\n- build-openstack-releasenotes http://logs.openstack.org/20/659020/1/check/build-openstack-releasenotes/9aa4538/html/ : SUCCESS in 5m 27s\n- openstack-tox-functional http://logs.openstack.org/20/659020/1/check/openstack-tox-functional/87e9121/ : SUCCESS in 6m 46s\n- openstack-tox-functional-py36 http://logs.openstack.org/20/659020/1/check/openstack-tox-functional-py36/cd92c7f/ : SUCCESS in 9m 39s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/20/659020/1/check/octavia-v2-dsvm-noop-api/316935a/ : SUCCESS in 35m 09s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/20/659020/1/check/octavia-v2-dsvm-noop-py2-api/08099f8/ : SUCCESS in 42m 56s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/20/659020/1/check/octavia-v2-dsvm-scenario/257b31d/ : SUCCESS in 1h 30m 43s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/20/659020/1/check/octavia-v2-dsvm-py2-scenario/8dd5b1b/ : SUCCESS in 1h 25m 18s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/20/659020/1/check/octavia-v2-dsvm-py2-scenario-centos-7/bbb1051/ : FAILURE in 2h 14m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/20/659020/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/40058cc/ : SUCCESS in 1h 40m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/20/659020/1/check/octavia-v2-dsvm-scenario-two-node/8a5d58d/ : FAILURE in 1h 28m 19s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/20/659020/1/check/octavia-v2-dsvm-py2-scenario-two-node/645f7f5/ : SUCCESS in 1h 48m 26s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/20/659020/1/check/octavia-v2-dsvm-tls-barbican/94c8882/ : SUCCESS in 50m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f552fbe908950a1ce9bdf908f77f1d752e8786b","author":{"_account_id":27809,"name":"hanxiang gao","email":"gaohanxiang@126.com","username":"gao.hanxiang"},"date":"2019-05-23 05:37:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5c55cf179ed350698288a666a7e182bcc7204a5","revisions":{"a5c55cf179ed350698288a666a7e182bcc7204a5":{"kind":"REWORK","_number":1,"created":"2019-05-14 10:11:25.000000000","uploader":{"_account_id":27809,"name":"hanxiang gao","email":"gaohanxiang@126.com","username":"gao.hanxiang"},"ref":"refs/changes/20/659020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/659020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/659020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/659020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/659020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/659020/1"}}},"commit":{"parents":[{"commit":"987a6b3f1dc83fcaa86c7c02418b7f33afdd59b3","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/987a6b3f1dc83fcaa86c7c02418b7f33afdd59b3"}]}],"author":{"name":"gao.hanxiang","email":"gao.hanxiang@99cloud.net","date":"2019-05-14 10:10:53.000000000","tz":480},"committer":{"name":"gao.hanxiang","email":"gao.hanxiang@99cloud.net","date":"2019-05-14 10:11:23.000000000","tz":480},"subject":"Blacklist bandit 1.6.0","message":"Blacklist bandit 1.6.0\n\nThere\u0027s a regression[0] in bandit 1.6.0 which causes bandit to stop\nrespecting excluded directories, and our tests throw a bunch of\nviolations. Blacklist this version, but allow newer versions as there is\nalready a pull request[1] to fix it, and I expect it will be included in\nthe next release.\n\n[0] https://github.com/PyCQA/bandit/issues/488\n[1] https://github.com/PyCQA/bandit/pull/489\n\nChange-Id: Ic533412ef876eb76702c972d87b4c2af140accbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5c55cf179ed350698288a666a7e182bcc7204a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5c55cf179ed350698288a666a7e182bcc7204a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
