)]}'
{"id":"openstack%2Foctavia~659022","triplet_id":"openstack%2Foctavia~stable%2Fstein~Ic533412ef876eb76702c972d87b4c2af140accbd","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"Ic533412ef876eb76702c972d87b4c2af140accbd","subject":"Blacklist bandit 1.6.0","status":"ABANDONED","created":"2019-05-14 10:12:05.000000000","updated":"2019-05-23 05:37:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7b206f2e81381049f5c94d81b2f636f9bb87ffda","_number":659022,"virtual_id_number":659022,"owner":{"_account_id":27809,"name":"hanxiang gao","email":"gaohanxiang@126.com","username":"gao.hanxiang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":-1,"date":"2019-05-14 12:17:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-2,"date":"2019-05-14 12:01:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 12:01:51.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-05-14 12:17:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"900526102f1de61d56fe2556aa54c8dc08a3e622","author":{"_account_id":27809,"name":"hanxiang gao","email":"gaohanxiang@126.com","username":"gao.hanxiang"},"date":"2019-05-14 10:12:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"285541096493f0136f416e3740ea171f2b936d3d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-14 12:01:51.000000000","message":"Patch Set 1: Code-Review-2\n\nAdam\u0027s comment on the master patch: \"There\u0027s no reason to do this, as we\u0027ve simply resolved the issue.\"","accounts_in_message":[],"_revision_number":1},{"id":"f84b8be604219e1259a4f8a345784719a07db351","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 12:17:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/22/659022/1/check/requirements-check/71bad78/ : FAILURE in 3m 39s\n- openstack-tox-cover http://logs.openstack.org/22/659022/1/check/openstack-tox-cover/1f8eafc/cover/ : SUCCESS in 8m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/22/659022/1/check/openstack-tox-lower-constraints/f864878/ : SUCCESS in 8m 06s\n- openstack-tox-pep8 http://logs.openstack.org/22/659022/1/check/openstack-tox-pep8/3cc418b/ : FAILURE in 6m 29s\n- openstack-tox-py27 http://logs.openstack.org/22/659022/1/check/openstack-tox-py27/c49d80a/ : SUCCESS in 5m 33s\n- openstack-tox-py35 http://logs.openstack.org/22/659022/1/check/openstack-tox-py35/9ab90b7/ : SUCCESS in 6m 09s\n- openstack-tox-py36 http://logs.openstack.org/22/659022/1/check/openstack-tox-py36/8605008/ : SUCCESS in 7m 03s\n- openstack-tox-docs http://logs.openstack.org/22/659022/1/check/openstack-tox-docs/b9df4bd/html/ : SUCCESS in 7m 33s\n- build-openstack-releasenotes http://logs.openstack.org/22/659022/1/check/build-openstack-releasenotes/30abae2/html/ : SUCCESS in 5m 40s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/22/659022/1/check/neutron-lbaasv2-dsvm-api/027f4e7/ : SUCCESS in 1h 16m 44s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/22/659022/1/check/neutron-lbaasv2-dsvm-py3x-api/4b9492c/ : SUCCESS in 1h 18m 00s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/22/659022/1/check/neutron-lbaasv2-dsvm-scenario/2111ad4/ : SUCCESS in 1h 22m 15s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/22/659022/1/check/neutron-lbaasv2-dsvm-py3x-scenario/cab15a0/ : SUCCESS in 1h 30m 45s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/22/659022/1/check/openstack-tox-functional/13342f4/ : SUCCESS in 6m 53s\n- openstack-tox-functional-py35 http://logs.openstack.org/22/659022/1/check/openstack-tox-functional-py35/c1fa043/ : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 http://logs.openstack.org/22/659022/1/check/openstack-tox-functional-py36/2dacab4/ : SUCCESS in 6m 43s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/22/659022/1/check/octavia-v1-dsvm-scenario/afb08b7/ : SUCCESS in 58m 50s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/22/659022/1/check/octavia-v1-dsvm-py3x-scenario/96b77f2/ : SUCCESS in 1h 00m 57s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/22/659022/1/check/octavia-v2-dsvm-noop-api/488ad5c/ : SUCCESS in 41m 43s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/22/659022/1/check/octavia-v2-dsvm-noop-py2-api/dfb1b44/ : SUCCESS in 46m 57s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/22/659022/1/check/octavia-v2-dsvm-scenario/39cec2b/ : SUCCESS in 1h 32m 45s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/22/659022/1/check/octavia-v2-dsvm-py2-scenario/847c02e/ : SUCCESS in 1h 37m 29s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/22/659022/1/check/octavia-v2-dsvm-py2-scenario-centos-7/5ccc07f/ : FAILURE in 1h 57m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/22/659022/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/1b5ad56/ : SUCCESS in 1h 40m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/22/659022/1/check/octavia-v2-dsvm-scenario-two-node/2740995/ : SUCCESS in 1h 47m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/22/659022/1/check/octavia-v2-dsvm-py2-scenario-two-node/80be898/ : SUCCESS in 1h 41m 52s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/22/659022/1/check/octavia-v2-dsvm-tls-barbican/c5c083e/ : SUCCESS in 49m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b206f2e81381049f5c94d81b2f636f9bb87ffda","author":{"_account_id":27809,"name":"hanxiang gao","email":"gaohanxiang@126.com","username":"gao.hanxiang"},"date":"2019-05-23 05:37:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"797cdd34ae658439e406560bdc87b6ef099de551","revisions":{"797cdd34ae658439e406560bdc87b6ef099de551":{"kind":"REWORK","_number":1,"created":"2019-05-14 10:12:05.000000000","uploader":{"_account_id":27809,"name":"hanxiang gao","email":"gaohanxiang@126.com","username":"gao.hanxiang"},"ref":"refs/changes/22/659022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/659022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/659022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/659022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/659022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/659022/1"}}},"commit":{"parents":[{"commit":"30a4feb820baa04a5a2a1b7f65f80688e8238604","subject":"Bandit test exclusions syntax change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30a4feb820baa04a5a2a1b7f65f80688e8238604"}]}],"author":{"name":"gao.hanxiang","email":"gao.hanxiang@99cloud.net","date":"2019-05-14 10:10:53.000000000","tz":480},"committer":{"name":"Hanxiang Gao","email":"gao.hanxiang@99cloud.net","date":"2019-05-14 10:12:05.000000000","tz":0},"subject":"Blacklist bandit 1.6.0","message":"Blacklist bandit 1.6.0\n\nThere\u0027s a regression[0] in bandit 1.6.0 which causes bandit to stop\nrespecting excluded directories, and our tests throw a bunch of\nviolations. Blacklist this version, but allow newer versions as there is\nalready a pull request[1] to fix it, and I expect it will be included in\nthe next release.\n\n[0] https://github.com/PyCQA/bandit/issues/488\n[1] https://github.com/PyCQA/bandit/pull/489\n\nChange-Id: Ic533412ef876eb76702c972d87b4c2af140accbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/797cdd34ae658439e406560bdc87b6ef099de551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/797cdd34ae658439e406560bdc87b6ef099de551"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
