)]}'
{"id":"openstack%2Foctavia~659283","triplet_id":"openstack%2Foctavia~master~I665503aa4f69308ff31be53824b26faea13d2ad1","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I665503aa4f69308ff31be53824b26faea13d2ad1","subject":"Remove tests/tempest path from devstack and tox","status":"MERGED","created":"2019-05-15 12:35:57.000000000","updated":"2019-05-15 20:13:15.000000000","submitted":"2019-05-15 20:13:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659283-1557951195699-ec8a5968","meta_rev_id":"dce62e08047105febaf4bd8c4a6b0148895bf64a","_number":659283,"virtual_id_number":659283,"owner":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-05-15 20:13:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-05-15 17:29:01.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"value":2,"date":"2019-05-15 18:05:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"value":1,"date":"2019-05-15 18:05:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-15 17:29:01.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-05-15 18:05:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-05-15 20:13:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47adb6905c66729d1398e1fd43bc30b065908b0a","author":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"date":"2019-05-15 12:35:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fb3ccfd51e4d59bb650e8a0bace64f2e35475e0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-15 12:42:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2f00de560510a0a61d6b86c91ff4436b060cf31a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 14:45:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/83/659283/1/check/openstack-tox-cover/a130ab6/cover/ : SUCCESS in 8m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/83/659283/1/check/openstack-tox-lower-constraints/7484120/ : SUCCESS in 7m 23s\n- openstack-tox-pep8 http://logs.openstack.org/83/659283/1/check/openstack-tox-pep8/fd87c21/ : SUCCESS in 7m 05s\n- openstack-tox-py27 http://logs.openstack.org/83/659283/1/check/openstack-tox-py27/fe96f7d/ : SUCCESS in 6m 35s\n- openstack-tox-py36 http://logs.openstack.org/83/659283/1/check/openstack-tox-py36/66e2d9a/ : SUCCESS in 7m 26s\n- openstack-tox-docs http://logs.openstack.org/83/659283/1/check/openstack-tox-docs/651a166/html/ : SUCCESS in 6m 13s\n- build-openstack-releasenotes http://logs.openstack.org/83/659283/1/check/build-openstack-releasenotes/a4ca556/html/ : SUCCESS in 5m 55s\n- openstack-tox-functional http://logs.openstack.org/83/659283/1/check/openstack-tox-functional/fcf3574/ : SUCCESS in 7m 57s\n- openstack-tox-functional-py36 http://logs.openstack.org/83/659283/1/check/openstack-tox-functional-py36/bddc389/ : SUCCESS in 6m 37s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/83/659283/1/check/octavia-v2-dsvm-noop-api/599aadb/ : SUCCESS in 44m 04s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/83/659283/1/check/octavia-v2-dsvm-noop-py2-api/b4254bb/ : SUCCESS in 43m 52s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/83/659283/1/check/octavia-v2-dsvm-scenario/1609d2d/ : FAILURE in 1h 36m 11s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/83/659283/1/check/octavia-v2-dsvm-py2-scenario/5ef7b27/ : SUCCESS in 1h 31m 42s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/83/659283/1/check/octavia-v2-dsvm-py2-scenario-centos-7/d4b7370/ : FAILURE in 2h 04m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/83/659283/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/b8b18ee/ : SUCCESS in 1h 31m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/83/659283/1/check/octavia-v2-dsvm-scenario-two-node/364830a/ : SUCCESS in 1h 46m 40s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/83/659283/1/check/octavia-v2-dsvm-py2-scenario-two-node/94adc0c/ : SUCCESS in 1h 48m 47s (non-voting)\n- octavia-grenade http://logs.openstack.org/83/659283/1/check/octavia-grenade/a4e9fde/ : SUCCESS in 1h 11m 39s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/83/659283/1/check/octavia-v2-dsvm-tls-barbican/6deba5f/ : SUCCESS in 50m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc49f157ac2b43296f309b963dea3db829e34cc1","author":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"date":"2019-05-15 14:52:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"55d7b3f0b5a82e6ff9eee71e8ca09174c241fc8a","author":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"date":"2019-05-15 14:52:30.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"07d8ef715e16e014bfde4041ced3a858cbd162ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 17:10:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/83/659283/2/check/openstack-tox-cover/e8d5c76/cover/ : SUCCESS in 8m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/83/659283/2/check/openstack-tox-lower-constraints/f714d80/ : SUCCESS in 7m 32s\n- openstack-tox-pep8 http://logs.openstack.org/83/659283/2/check/openstack-tox-pep8/0da9f7d/ : SUCCESS in 6m 53s\n- openstack-tox-py27 http://logs.openstack.org/83/659283/2/check/openstack-tox-py27/54049c4/ : SUCCESS in 6m 37s\n- openstack-tox-py36 http://logs.openstack.org/83/659283/2/check/openstack-tox-py36/7eb22b4/ : SUCCESS in 8m 23s\n- openstack-tox-docs http://logs.openstack.org/83/659283/2/check/openstack-tox-docs/e3e8d82/html/ : SUCCESS in 6m 50s\n- build-openstack-releasenotes http://logs.openstack.org/83/659283/2/check/build-openstack-releasenotes/339f8e5/html/ : SUCCESS in 6m 39s\n- openstack-tox-functional http://logs.openstack.org/83/659283/2/check/openstack-tox-functional/eb13673/ : SUCCESS in 6m 17s\n- openstack-tox-functional-py36 http://logs.openstack.org/83/659283/2/check/openstack-tox-functional-py36/e04d3bf/ : SUCCESS in 8m 47s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/83/659283/2/check/octavia-v2-dsvm-noop-api/6154771/ : SUCCESS in 44m 45s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/83/659283/2/check/octavia-v2-dsvm-noop-py2-api/7225fce/ : SUCCESS in 46m 13s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/83/659283/2/check/octavia-v2-dsvm-scenario/04c6f11/ : SUCCESS in 1h 26m 31s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/83/659283/2/check/octavia-v2-dsvm-py2-scenario/3026c09/ : SUCCESS in 1h 34m 33s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/83/659283/2/check/octavia-v2-dsvm-py2-scenario-centos-7/ecd5441/ : FAILURE in 2h 03m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/83/659283/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/5e97994/ : SUCCESS in 1h 37m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/83/659283/2/check/octavia-v2-dsvm-scenario-two-node/2c08033/ : SUCCESS in 1h 53m 33s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/83/659283/2/check/octavia-v2-dsvm-py2-scenario-two-node/353d8e6/ : SUCCESS in 2h 04m 11s (non-voting)\n- octavia-grenade http://logs.openstack.org/83/659283/2/check/octavia-grenade/4caceb1/ : SUCCESS in 1h 19m 35s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/83/659283/2/check/octavia-v2-dsvm-tls-barbican/0f280cf/ : SUCCESS in 50m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8596ff8f8b3fca1da9cbc612b59029985da53292","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-15 17:29:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2e4da97d6246861ea48d451d2730d9dd3ffe5c8a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-15 18:05:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"3eaafb4207ebf0cce2e3408af62b6e967c6fd002","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 18:06:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"59fe471b8c1917dfc44d3eff823d54c3761ab2b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 20:13:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dce62e08047105febaf4bd8c4a6b0148895bf64a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 20:13:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/83/659283/2/gate/openstack-tox-lower-constraints/0a8a56d/ : SUCCESS in 6m 52s\n- openstack-tox-pep8 http://logs.openstack.org/83/659283/2/gate/openstack-tox-pep8/71ac156/ : SUCCESS in 6m 07s\n- openstack-tox-py27 http://logs.openstack.org/83/659283/2/gate/openstack-tox-py27/aa2a08b/ : SUCCESS in 7m 12s\n- openstack-tox-py36 http://logs.openstack.org/83/659283/2/gate/openstack-tox-py36/801271e/ : SUCCESS in 5m 59s\n- openstack-tox-docs http://logs.openstack.org/83/659283/2/gate/openstack-tox-docs/77c9d59/html/ : SUCCESS in 7m 31s\n- build-openstack-releasenotes http://logs.openstack.org/83/659283/2/gate/build-openstack-releasenotes/48f4a3f/html/ : SUCCESS in 6m 45s\n- openstack-tox-functional http://logs.openstack.org/83/659283/2/gate/openstack-tox-functional/b62e89e/ : SUCCESS in 8m 15s\n- openstack-tox-functional-py36 http://logs.openstack.org/83/659283/2/gate/openstack-tox-functional-py36/2cf5c04/ : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/83/659283/2/gate/octavia-v2-dsvm-noop-api/fc3ac97/ : SUCCESS in 42m 18s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/83/659283/2/gate/octavia-v2-dsvm-noop-py2-api/546165d/ : SUCCESS in 47m 10s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/83/659283/2/gate/octavia-v2-dsvm-scenario/3413e59/ : SUCCESS in 2h 02m 33s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/83/659283/2/gate/octavia-v2-dsvm-py2-scenario/dc2b390/ : SUCCESS in 1h 42m 04s\n- octavia-grenade http://logs.openstack.org/83/659283/2/gate/octavia-grenade/38acdf5/ : SUCCESS in 1h 11m 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f9455cd94fd50f8c79fbdd50a6be3293cd66b1d8","revisions":{"c03f95d1770c7513c41955ffa08be457ee3f2086":{"kind":"REWORK","_number":1,"created":"2019-05-15 12:35:57.000000000","uploader":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"ref":"refs/changes/83/659283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/659283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/659283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/659283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/659283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/659283/1"}}},"commit":{"parents":[{"commit":"0573d2c609378e206db932b78fb54cd9d7384332","subject":"Merge \"Performance improvement for non-udp health checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0573d2c609378e206db932b78fb54cd9d7384332"}]}],"author":{"name":"Adit Sarfaty","email":"asarfaty@vmware.com","date":"2019-05-15 12:34:39.000000000","tz":180},"committer":{"name":"Adit Sarfaty","email":"asarfaty@vmware.com","date":"2019-05-15 12:34:39.000000000","tz":180},"subject":"Remove tests/tempest path from devstack settings and tox","message":"Remove tests/tempest path from devstack settings and tox\n\nCommit I3082962841d3b645f3cbd1a6b41fc7fb28dcf7e6 removed\noctavia/tests/tempest directory\nThis patch removed other references to this path.\n\nChange-Id: I665503aa4f69308ff31be53824b26faea13d2ad1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c03f95d1770c7513c41955ffa08be457ee3f2086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c03f95d1770c7513c41955ffa08be457ee3f2086"}]},"branch":"refs/heads/master"},"f9455cd94fd50f8c79fbdd50a6be3293cd66b1d8":{"kind":"REWORK","_number":2,"created":"2019-05-15 14:52:06.000000000","uploader":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"ref":"refs/changes/83/659283/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/659283/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/659283/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/659283/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/659283/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/659283/2"}}},"commit":{"parents":[{"commit":"0573d2c609378e206db932b78fb54cd9d7384332","subject":"Merge \"Performance improvement for non-udp health checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0573d2c609378e206db932b78fb54cd9d7384332"}]}],"author":{"name":"Adit Sarfaty","email":"asarfaty@vmware.com","date":"2019-05-15 12:34:39.000000000","tz":180},"committer":{"name":"Adit Sarfaty","email":"asarfaty@vmware.com","date":"2019-05-15 14:51:14.000000000","tz":180},"subject":"Remove tests/tempest path from devstack and tox","message":"Remove tests/tempest path from devstack and tox\n\nCommit I3082962841d3b645f3cbd1a6b41fc7fb28dcf7e6 removed\noctavia/tests/tempest directory\nThis patch removed other references to this path.\n\nChange-Id: I665503aa4f69308ff31be53824b26faea13d2ad1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9455cd94fd50f8c79fbdd50a6be3293cd66b1d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9455cd94fd50f8c79fbdd50a6be3293cd66b1d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
