)]}'
{"id":"openstack%2Foctavia~659538","triplet_id":"openstack%2Foctavia~master~I6570f9f82173a1258da76899de8f1410ff5fc52a","project":"openstack/octavia","branch":"master","topic":"jobboard","hashtags":[],"change_id":"I6570f9f82173a1258da76899de8f1410ff5fc52a","subject":"[Jobboard] Importable flow functions","status":"ABANDONED","created":"2019-05-16 11:52:09.000000000","updated":"2019-10-16 16:11:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"086aac9b59ba79e117bc2ba8bc5a8004ec4442f2","_number":659538,"virtual_id_number":659538,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-08-07 15:47:04.000000000","_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-08-07 18:18:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 16:28:27.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-08-07 18:18:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d7a4167eb9d03097f11797dcf587970f0e5ad38","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-05-16 11:52:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33e41028e1d46d792e3e877187db8bc9df5251cd","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-05-16 12:26:18.000000000","message":"Topic set to jobboard","accounts_in_message":[],"_revision_number":1},{"id":"98472d0aea689346428714652b2f169683510f5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 14:10:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/38/659538/1/check/openstack-tox-cover/fb7191f/cover/ : SUCCESS in 7m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/659538/1/check/openstack-tox-lower-constraints/e5afe80/ : SUCCESS in 8m 21s\n- openstack-tox-pep8 http://logs.openstack.org/38/659538/1/check/openstack-tox-pep8/ad4d64c/ : SUCCESS in 7m 47s\n- openstack-tox-py27 http://logs.openstack.org/38/659538/1/check/openstack-tox-py27/ae9debc/ : SUCCESS in 6m 34s\n- openstack-tox-py36 http://logs.openstack.org/38/659538/1/check/openstack-tox-py36/fb25b77/ : SUCCESS in 15m 20s\n- openstack-tox-py37 http://logs.openstack.org/38/659538/1/check/openstack-tox-py37/ccfb9c6/ : SUCCESS in 7m 14s\n- openstack-tox-docs http://logs.openstack.org/38/659538/1/check/openstack-tox-docs/e5a609e/html/ : SUCCESS in 6m 48s\n- openstack-tox-functional http://logs.openstack.org/38/659538/1/check/openstack-tox-functional/5c78a6a/ : SUCCESS in 6m 23s\n- openstack-tox-functional-py36 http://logs.openstack.org/38/659538/1/check/openstack-tox-functional-py36/a94a808/ : SUCCESS in 9m 12s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/38/659538/1/check/octavia-v2-dsvm-noop-api/860c80c/ : SUCCESS in 44m 56s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/38/659538/1/check/octavia-v2-dsvm-noop-py2-api/3beddbc/ : SUCCESS in 43m 55s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/38/659538/1/check/octavia-v2-dsvm-scenario/7c432d5/ : SUCCESS in 1h 27m 59s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/38/659538/1/check/octavia-v2-dsvm-py2-scenario/617a78f/ : SUCCESS in 1h 31m 06s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/38/659538/1/check/octavia-v2-dsvm-py2-scenario-centos-7/cdbb2eb/ : FAILURE in 2h 06m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/38/659538/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/7581086/ : SUCCESS in 1h 33m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/38/659538/1/check/octavia-v2-dsvm-scenario-two-node/b6ebea4/ : FAILURE in 1h 31m 19s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/38/659538/1/check/octavia-v2-dsvm-py2-scenario-two-node/61527aa/ : FAILURE in 1h 32m 39s (non-voting)\n- octavia-grenade http://logs.openstack.org/38/659538/1/check/octavia-grenade/70e9276/ : FAILURE in 51m 54s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/38/659538/1/check/octavia-v2-dsvm-tls-barbican/13f526e/ : SUCCESS in 51m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a860f205478ee79a83e11cd920eb04f7e7b55877","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-05-16 14:38:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"62ae7eacd05de0a2b608465c22ee24f98e835f35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 18:48:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/38/659538/1/check/openstack-tox-cover/0036fa7/cover/ : SUCCESS in 8m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/659538/1/check/openstack-tox-lower-constraints/ce13465/ : SUCCESS in 8m 19s\n- openstack-tox-pep8 http://logs.openstack.org/38/659538/1/check/openstack-tox-pep8/25ffd44/ : SUCCESS in 7m 19s\n- openstack-tox-py27 http://logs.openstack.org/38/659538/1/check/openstack-tox-py27/6babdc3/ : SUCCESS in 6m 44s\n- openstack-tox-py36 http://logs.openstack.org/38/659538/1/check/openstack-tox-py36/c0cf64d/ : SUCCESS in 6m 38s\n- openstack-tox-py37 http://logs.openstack.org/38/659538/1/check/openstack-tox-py37/694ad9a/ : SUCCESS in 7m 09s\n- openstack-tox-docs http://logs.openstack.org/38/659538/1/check/openstack-tox-docs/a7d47ad/html/ : SUCCESS in 7m 31s\n- openstack-tox-functional http://logs.openstack.org/38/659538/1/check/openstack-tox-functional/eabee7d/ : SUCCESS in 8m 40s\n- openstack-tox-functional-py36 http://logs.openstack.org/38/659538/1/check/openstack-tox-functional-py36/9bcd3b0/ : SUCCESS in 8m 47s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/38/659538/1/check/octavia-v2-dsvm-noop-api/6a04f1d/ : SUCCESS in 45m 40s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/38/659538/1/check/octavia-v2-dsvm-noop-py2-api/e113fec/ : SUCCESS in 42m 42s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/38/659538/1/check/octavia-v2-dsvm-scenario/253faeb/ : SUCCESS in 1h 27m 00s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/38/659538/1/check/octavia-v2-dsvm-py2-scenario/2703c59/ : SUCCESS in 1h 47m 48s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/38/659538/1/check/octavia-v2-dsvm-py2-scenario-centos-7/ff42f54/ : FAILURE in 1h 56m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/38/659538/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/756e689/ : SUCCESS in 1h 42m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/38/659538/1/check/octavia-v2-dsvm-scenario-two-node/05103e4/ : SUCCESS in 1h 54m 36s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/38/659538/1/check/octavia-v2-dsvm-py2-scenario-two-node/bdd78c5/ : SUCCESS in 2h 06m 21s (non-voting)\n- octavia-grenade http://logs.openstack.org/38/659538/1/check/octavia-grenade/6a94bab/ : SUCCESS in 1h 12m 15s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/38/659538/1/check/octavia-v2-dsvm-tls-barbican/8ce88fc/ : SUCCESS in 53m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"758babd0fb1a445004344ac8ee5ab33b169913cb","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-05-22 08:09:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b4dbc1580bc74682f82399dfd0248223ac28c42e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-05-22 08:35:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3eeed857e68d211f3a6db0d23cf0e18d66f193ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 10:44:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/38/659538/3/check/openstack-tox-cover/1bafb72/cover/ : SUCCESS in 7m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/659538/3/check/openstack-tox-lower-constraints/2e862f9/ : SUCCESS in 7m 03s\n- openstack-tox-pep8 http://logs.openstack.org/38/659538/3/check/openstack-tox-pep8/4cb9534/ : SUCCESS in 6m 18s\n- openstack-tox-py27 http://logs.openstack.org/38/659538/3/check/openstack-tox-py27/fe30a43/ : SUCCESS in 5m 33s\n- openstack-tox-py36 http://logs.openstack.org/38/659538/3/check/openstack-tox-py36/3518d01/ : SUCCESS in 5m 44s\n- openstack-tox-py37 http://logs.openstack.org/38/659538/3/check/openstack-tox-py37/16f9aff/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/38/659538/3/check/openstack-tox-docs/ad02d23/html/ : SUCCESS in 6m 20s\n- openstack-tox-functional http://logs.openstack.org/38/659538/3/check/openstack-tox-functional/70f6226/ : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 http://logs.openstack.org/38/659538/3/check/openstack-tox-functional-py36/a21ae6d/ : SUCCESS in 6m 28s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/38/659538/3/check/octavia-v2-dsvm-noop-api/e813e5b/ : SUCCESS in 40m 54s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/38/659538/3/check/octavia-v2-dsvm-noop-py2-api/23176f8/ : SUCCESS in 41m 14s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/38/659538/3/check/octavia-v2-dsvm-scenario/98d6462/ : FAILURE in 34m 30s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/38/659538/3/check/octavia-v2-dsvm-py2-scenario/56b1f85/ : FAILURE in 28m 15s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/38/659538/3/check/octavia-v2-dsvm-py2-scenario-centos-7/612f081/ : FAILURE in 2h 00m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/38/659538/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/4a7b3a9/ : FAILURE in 31m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/38/659538/3/check/octavia-v2-dsvm-scenario-two-node/edb3153/ : FAILURE in 35m 23s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/38/659538/3/check/octavia-v2-dsvm-py2-scenario-two-node/e016e8b/ : FAILURE in 35m 31s (non-voting)\n- octavia-grenade http://logs.openstack.org/38/659538/3/check/octavia-grenade/8fe785f/ : FAILURE in 1h 09m 50s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/38/659538/3/check/octavia-v2-dsvm-tls-barbican/4ff5528/ : FAILURE in 38m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/38/659538/3/check/octavia-v2-dsvm-scenario-amphora-v2/9d008c6/ : FAILURE in 33m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b671dafd6c21bf5de248c88e786c3326678cbf2b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-05-22 15:56:51.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2ba61bc6923c62818eeb88dbafb150f6a7f57348","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 20:11:44.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/38/659538/3/check/openstack-tox-cover/b8cf321/cover/ : SUCCESS in 9m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/659538/3/check/openstack-tox-lower-constraints/8ce8154/ : SUCCESS in 8m 24s\n- openstack-tox-pep8 http://logs.openstack.org/38/659538/3/check/openstack-tox-pep8/d2e7144/ : SUCCESS in 7m 27s\n- openstack-tox-py27 http://logs.openstack.org/38/659538/3/check/openstack-tox-py27/5e09355/ : SUCCESS in 7m 29s\n- openstack-tox-py36 http://logs.openstack.org/38/659538/3/check/openstack-tox-py36/d617d55/ : SUCCESS in 5m 57s\n- openstack-tox-py37 http://logs.openstack.org/38/659538/3/check/openstack-tox-py37/270eca5/ : SUCCESS in 7m 00s\n- openstack-tox-docs http://logs.openstack.org/38/659538/3/check/openstack-tox-docs/a328b9b/html/ : SUCCESS in 7m 54s\n- openstack-tox-functional http://logs.openstack.org/38/659538/3/check/openstack-tox-functional/813ee2e/ : SUCCESS in 7m 54s\n- openstack-tox-functional-py36 http://logs.openstack.org/38/659538/3/check/openstack-tox-functional-py36/bb41651/ : SUCCESS in 7m 49s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/38/659538/3/check/octavia-v2-dsvm-noop-api/70285ac/ : SUCCESS in 45m 03s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/38/659538/3/check/octavia-v2-dsvm-noop-py2-api/b75868e/ : SUCCESS in 43m 49s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/38/659538/3/check/octavia-v2-dsvm-scenario/3293fd4/ : SUCCESS in 1h 27m 56s\n- octavia-v2-dsvm-py2-scenario finger://ze10.openstack.org/515ef61090e1498fbe888774d3dcbb54 : RETRY_LIMIT in 25m 24s\n- octavia-v2-dsvm-py2-scenario-centos-7 finger://ze05.openstack.org/6e6072af8226403b97c2ce205f3a8f19 : RETRY_LIMIT in 25m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic finger://ze02.openstack.org/811ae4eb715145b28cf2818a00ed3e17 : RETRY_LIMIT in 26m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node finger://ze08.openstack.org/149e69a953284885961b81908aae116c : RETRY_LIMIT in 26m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node finger://ze08.openstack.org/47441ec6fd2740f08ae4eb496dbd3a02 : RETRY_LIMIT in 31m 44s (non-voting)\n- octavia-grenade http://logs.openstack.org/38/659538/3/check/octavia-grenade/8f50144/ : FAILURE in 49m 05s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/38/659538/3/check/octavia-v2-dsvm-tls-barbican/36a188f/ : SUCCESS in 54m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 finger://ze02.openstack.org/132fb883ce06439692d8c6cf38b39511 : RETRY_LIMIT in 25m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"18366dfc0d5203ee04f419e0d7829bc2ad07ba48","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-05-23 10:32:56.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"937abfb546d5017f04dcd606a55b2ed4cec5430f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 12:55:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/38/659538/4/check/openstack-tox-cover/15d87ac/cover/ : SUCCESS in 9m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/659538/4/check/openstack-tox-lower-constraints/a86b2c1/ : SUCCESS in 6m 55s\n- openstack-tox-pep8 http://logs.openstack.org/38/659538/4/check/openstack-tox-pep8/570453f/ : SUCCESS in 6m 41s\n- openstack-tox-py27 http://logs.openstack.org/38/659538/4/check/openstack-tox-py27/4cab349/ : SUCCESS in 5m 20s\n- openstack-tox-py36 http://logs.openstack.org/38/659538/4/check/openstack-tox-py36/ad89e82/ : SUCCESS in 5m 56s\n- openstack-tox-py37 http://logs.openstack.org/38/659538/4/check/openstack-tox-py37/33038be/ : SUCCESS in 11m 06s\n- openstack-tox-docs http://logs.openstack.org/38/659538/4/check/openstack-tox-docs/e6d7b00/html/ : SUCCESS in 9m 39s\n- openstack-tox-functional http://logs.openstack.org/38/659538/4/check/openstack-tox-functional/410504e/ : SUCCESS in 5m 58s\n- openstack-tox-functional-py36 http://logs.openstack.org/38/659538/4/check/openstack-tox-functional-py36/2882f27/ : SUCCESS in 11m 26s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-noop-api/e414ae9/ : SUCCESS in 44m 58s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-noop-py2-api/1a30db5/ : SUCCESS in 45m 17s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-scenario/3f62c43/ : SUCCESS in 1h 26m 59s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-py2-scenario/abe3889/ : FAILURE in 1h 45m 15s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-py2-scenario-centos-7/aa3b77b/ : FAILURE in 2h 09m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-scenario-ubuntu-bionic/9873f80/ : SUCCESS in 1h 38m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-scenario-two-node/a1851d6/ : SUCCESS in 1h 49m 07s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-py2-scenario-two-node/0e44f94/ : SUCCESS in 1h 43m 09s (non-voting)\n- octavia-grenade http://logs.openstack.org/38/659538/4/check/octavia-grenade/7da2d58/ : SUCCESS in 1h 13m 48s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-tls-barbican/bf88fd9/ : SUCCESS in 52m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-scenario-amphora-v2/c3608f0/ : SUCCESS in 1h 20m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b31f34d2540c7691d9a82d4ca12b7de31d7de35c","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-05-24 07:15:14.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a45df4a270db640b702b96d29ffabcf9d4d2ccb0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 09:30:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/38/659538/4/check/openstack-tox-cover/fe097eb/cover/ : SUCCESS in 9m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/659538/4/check/openstack-tox-lower-constraints/f31c904/ : SUCCESS in 8m 59s\n- openstack-tox-pep8 http://logs.openstack.org/38/659538/4/check/openstack-tox-pep8/792a0bb/ : SUCCESS in 7m 09s\n- openstack-tox-py27 http://logs.openstack.org/38/659538/4/check/openstack-tox-py27/9a29db3/ : SUCCESS in 5m 21s\n- openstack-tox-py36 http://logs.openstack.org/38/659538/4/check/openstack-tox-py36/743ff9b/ : SUCCESS in 9m 16s\n- openstack-tox-py37 http://logs.openstack.org/38/659538/4/check/openstack-tox-py37/6ef161f/ : SUCCESS in 5m 43s\n- openstack-tox-docs http://logs.openstack.org/38/659538/4/check/openstack-tox-docs/9fac350/html/ : SUCCESS in 7m 06s\n- openstack-tox-functional http://logs.openstack.org/38/659538/4/check/openstack-tox-functional/219f56b/ : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 http://logs.openstack.org/38/659538/4/check/openstack-tox-functional-py36/fd77112/ : SUCCESS in 10m 19s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-noop-api/386e54d/ : SUCCESS in 44m 32s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-noop-py2-api/f2e8996/ : SUCCESS in 42m 42s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-scenario/849d314/ : SUCCESS in 1h 29m 32s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-py2-scenario/440fd40/ : SUCCESS in 1h 45m 07s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-py2-scenario-centos-7/e6d31a1/ : FAILURE in 2h 09m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-scenario-ubuntu-bionic/89b3e67/ : SUCCESS in 1h 36m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-scenario-two-node/0351bb8/ : SUCCESS in 1h 48m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-py2-scenario-two-node/f55ee7c/ : SUCCESS in 1h 56m 28s (non-voting)\n- octavia-grenade http://logs.openstack.org/38/659538/4/check/octavia-grenade/8b5e99e/ : SUCCESS in 1h 14m 43s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-tls-barbican/738791f/ : SUCCESS in 59m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/38/659538/4/check/octavia-v2-dsvm-scenario-amphora-v2/87f1c79/ : SUCCESS in 1h 31m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"24e9ca7441ae5763da60dd15dd887c4a68a3ba60","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-06-06 13:11:40.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"90ccb48bc9f7af42a33e75851ab9768971f737da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 15:32:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/38/659538/5/check/openstack-tox-cover/2bc7fb7/cover/ : SUCCESS in 8m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/659538/5/check/openstack-tox-lower-constraints/9323c7f/ : SUCCESS in 6m 54s\n- openstack-tox-pep8 http://logs.openstack.org/38/659538/5/check/openstack-tox-pep8/a9df836/ : SUCCESS in 9m 15s\n- openstack-tox-py27 http://logs.openstack.org/38/659538/5/check/openstack-tox-py27/3dbad15/ : SUCCESS in 6m 58s\n- openstack-tox-py36 http://logs.openstack.org/38/659538/5/check/openstack-tox-py36/d501ff7/ : SUCCESS in 5m 05s\n- openstack-tox-py37 http://logs.openstack.org/38/659538/5/check/openstack-tox-py37/efe2f17/ : SUCCESS in 6m 07s\n- openstack-tox-docs http://logs.openstack.org/38/659538/5/check/openstack-tox-docs/f835897/html/ : SUCCESS in 9m 19s\n- openstack-tox-functional http://logs.openstack.org/38/659538/5/check/openstack-tox-functional/42171a8/ : SUCCESS in 9m 16s\n- openstack-tox-functional-py36 http://logs.openstack.org/38/659538/5/check/openstack-tox-functional-py36/9896ef8/ : SUCCESS in 7m 49s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/38/659538/5/check/octavia-v2-dsvm-noop-api/97c0f19/ : SUCCESS in 41m 09s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/38/659538/5/check/octavia-v2-dsvm-noop-py2-api/06570be/ : SUCCESS in 39m 15s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/38/659538/5/check/octavia-v2-dsvm-scenario/def2a5b/ : SUCCESS in 1h 48m 33s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/38/659538/5/check/octavia-v2-dsvm-py2-scenario/5ceb7fe/ : SUCCESS in 1h 54m 38s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/38/659538/5/check/octavia-v2-dsvm-py2-scenario-centos-7/3340fa9/ : FAILURE in 2h 11m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/38/659538/5/check/octavia-v2-dsvm-scenario-ubuntu-bionic/c7572f8/ : SUCCESS in 1h 46m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/38/659538/5/check/octavia-v2-dsvm-scenario-two-node/b9d0600/ : SUCCESS in 1h 59m 00s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/38/659538/5/check/octavia-v2-dsvm-py2-scenario-two-node/19b5de4/ : SUCCESS in 1h 51m 11s (non-voting)\n- octavia-grenade http://logs.openstack.org/38/659538/5/check/octavia-grenade/03489e1/ : SUCCESS in 1h 25m 26s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/38/659538/5/check/octavia-v2-dsvm-tls-barbican/2d949bd/ : SUCCESS in 51m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/38/659538/5/check/octavia-v2-dsvm-scenario-amphora-v2/9dc5d31/ : SUCCESS in 1h 29m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b918c45429ede65911fb047adfe6b62ca7962330","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-05 16:28:27.000000000","message":"Patch Set 5: Code-Review+1\n\nIt\u0027s hard for me to review this without seeing it actually being used for something -- right now this is just ... kinda dead code? I get what it\u0027s for I guess, but I\u0027d feel more comfortable merging this as part of a chain. You should just be able to base your next steps on this patch, then we can see it in action.","accounts_in_message":[],"_revision_number":5},{"id":"03415edb7fedd1bb58ea942637cbe4575e926d56","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-05 16:29:23.000000000","message":"Patch Set 5:\n\nAh, you are doing that, it just isn\u0027t working yet. So, yeah, I\u0027ll be able to say +2 on this once I see it obviously working in the next patch.","accounts_in_message":[],"_revision_number":5},{"id":"552c93a72a4fc8fa0687f8beb55d442e9ee0e266","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-07-08 07:06:40.000000000","message":"Patch Set 5:\n\nTo implement jobboard required two things:\n- have importable flow (current change)\n- flow passing dicts not db objects (https://review.opendev.org/#/q/status:open+project:openstack/octavia+branch:master+topic:jobboard_dicts)\n\nAs I have a lot of changes I tried to make each as small and not conflicting with others as possible.\nI can add import of flow_utils in v2/controller_worker.py in current change - this will work, but I will have conflicts with all my other changes.\nThis is dead code for now, but it is just refactor, so I think it is safe to land it now in this variant.\nIf you still think I should add usage of flow_util I will do in the next patch set.","accounts_in_message":[],"_revision_number":5},{"id":"eb6d7f0bb0fad527896895cd15a32020f42518ba","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-07-17 14:26:01.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"02c25ff384d95c590240f656f187ebb4a124df25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 16:45:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/38/659538/6/check/openstack-tox-cover/4105f4c/cover/ : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/659538/6/check/openstack-tox-lower-constraints/0d3a46a/ : SUCCESS in 6m 37s\n- openstack-tox-pep8 http://logs.openstack.org/38/659538/6/check/openstack-tox-pep8/36961b4/ : SUCCESS in 5m 28s\n- openstack-tox-py27 http://logs.openstack.org/38/659538/6/check/openstack-tox-py27/354bd76/ : SUCCESS in 4m 24s\n- openstack-tox-py36 http://logs.openstack.org/38/659538/6/check/openstack-tox-py36/b5a62b8/ : SUCCESS in 5m 53s\n- openstack-tox-py37 http://logs.openstack.org/38/659538/6/check/openstack-tox-py37/4fe2040/ : SUCCESS in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/38/659538/6/check/openstack-tox-docs/d529cec/html/ : SUCCESS in 5m 37s\n- openstack-tox-functional http://logs.openstack.org/38/659538/6/check/openstack-tox-functional/854f866/ : SUCCESS in 4m 58s\n- openstack-tox-functional-py36 http://logs.openstack.org/38/659538/6/check/openstack-tox-functional-py36/469a206/ : SUCCESS in 6m 17s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-noop-api/faacc6a/ : SUCCESS in 55m 05s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-noop-py2-api/83ba671/ : SUCCESS in 45m 42s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-scenario/23c17f5/ : FAILURE in 2h 16m 09s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-py2-scenario/88e5718/ : SUCCESS in 1h 40m 55s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-py2-scenario-centos-7/537dfac/ : FAILURE in 1h 58m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-scenario-ubuntu-bionic/a31b0f6/ : SUCCESS in 1h 28m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-scenario-two-node/183a91c/ : SUCCESS in 1h 43m 08s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-py2-scenario-two-node/deb6197/ : SUCCESS in 1h 40m 51s (non-voting)\n- octavia-grenade http://logs.openstack.org/38/659538/6/check/octavia-grenade/3f2965f/ : SUCCESS in 1h 21m 14s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-tls-barbican/4138654/ : SUCCESS in 1h 02m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-scenario-amphora-v2/573f243/ : SUCCESS in 1h 28m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9248ecafb0c03b39016f6819d3b0086078e88699","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-07-24 16:05:47.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"84deab56780e86af2e91ab89b4d0845e551beae9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 18:26:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/38/659538/6/check/openstack-tox-cover/137beca/cover/ : SUCCESS in 9m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/659538/6/check/openstack-tox-lower-constraints/9170996/ : SUCCESS in 7m 18s\n- openstack-tox-pep8 http://logs.openstack.org/38/659538/6/check/openstack-tox-pep8/acc2122/ : SUCCESS in 6m 49s\n- openstack-tox-py27 http://logs.openstack.org/38/659538/6/check/openstack-tox-py27/9c2fb7d/ : SUCCESS in 4m 59s\n- openstack-tox-py36 http://logs.openstack.org/38/659538/6/check/openstack-tox-py36/64250a4/ : SUCCESS in 5m 28s\n- openstack-tox-py37 http://logs.openstack.org/38/659538/6/check/openstack-tox-py37/02e60bf/ : SUCCESS in 4m 24s\n- openstack-tox-docs http://logs.openstack.org/38/659538/6/check/openstack-tox-docs/cf10ab9/html/ : SUCCESS in 6m 18s\n- openstack-tox-functional http://logs.openstack.org/38/659538/6/check/openstack-tox-functional/b1be984/ : SUCCESS in 5m 30s\n- openstack-tox-functional-py36 http://logs.openstack.org/38/659538/6/check/openstack-tox-functional-py36/be2b41d/ : SUCCESS in 5m 45s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-noop-api/525d674/ : SUCCESS in 48m 15s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-noop-py2-api/33f986b/ : SUCCESS in 43m 41s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-scenario/18d17be/ : SUCCESS in 1h 41m 24s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-py2-scenario/a49eb87/ : SUCCESS in 1h 35m 58s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-py2-scenario-centos-7/257b5de/ : FAILURE in 2h 14m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-scenario-ubuntu-bionic/63b1b54/ : FAILURE in 1h 36m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-scenario-two-node/42acaff/ : SUCCESS in 2h 03m 55s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-py2-scenario-two-node/84d189f/ : SUCCESS in 1h 34m 07s (non-voting)\n- octavia-grenade http://logs.openstack.org/38/659538/6/check/octavia-grenade/9bebff7/ : SUCCESS in 1h 24m 28s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-tls-barbican/a4ec6e6/ : SUCCESS in 1h 06m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/38/659538/6/check/octavia-v2-dsvm-scenario-amphora-v2/4b19575/ : SUCCESS in 1h 42m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f81dc455c4365f63212243c3abd4f9516b33c9ed","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-08-06 10:12:43.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3e3bd49d636ece97aeadc0f764b5a717ea70809a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 13:03:00.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/38/659538/7/check/openstack-tox-cover/1e469cd/ : FAILURE in 6m 49s\n- openstack-tox-lower-constraints https://logs.opendev.org/38/659538/7/check/openstack-tox-lower-constraints/695879c/ : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://logs.opendev.org/38/659538/7/check/openstack-tox-pep8/20580f8/ : SUCCESS in 5m 32s\n- openstack-tox-py27 https://logs.opendev.org/38/659538/7/check/openstack-tox-py27/626cd32/ : FAILURE in 7m 27s\n- openstack-tox-py36 https://logs.opendev.org/38/659538/7/check/openstack-tox-py36/3e2e9a6/ : FAILURE in 6m 49s\n- openstack-tox-py37 https://logs.opendev.org/38/659538/7/check/openstack-tox-py37/cc30662/ : FAILURE in 7m 52s\n- openstack-tox-docs https://logs.opendev.org/38/659538/7/check/openstack-tox-docs/292b50c/html/ : SUCCESS in 6m 09s\n- openstack-tox-functional https://logs.opendev.org/38/659538/7/check/openstack-tox-functional/d502c34/ : SUCCESS in 5m 36s\n- openstack-tox-functional-py36 https://logs.opendev.org/38/659538/7/check/openstack-tox-functional-py36/6c39f46/ : SUCCESS in 6m 08s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/38/659538/7/check/octavia-v2-dsvm-noop-api/25d7303/ : SUCCESS in 48m 02s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/38/659538/7/check/octavia-v2-dsvm-noop-py2-api/7b9ce2d/ : SUCCESS in 42m 43s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/38/659538/7/check/octavia-v2-dsvm-scenario/abb3d55/ : SUCCESS in 1h 46m 37s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/38/659538/7/check/octavia-v2-dsvm-py2-scenario/7f8ae49/ : SUCCESS in 1h 49m 38s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/38/659538/7/check/octavia-v2-dsvm-py2-scenario-centos-7/a977b8d/ : TIMED_OUT in 2h 35m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/38/659538/7/check/octavia-v2-dsvm-scenario-two-node/267e691/ : SUCCESS in 2h 05m 56s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/38/659538/7/check/octavia-v2-dsvm-py2-scenario-two-node/ce3d965/ : SUCCESS in 1h 52m 53s (non-voting)\n- octavia-grenade https://logs.opendev.org/38/659538/7/check/octavia-grenade/dbd9ce3/ : SUCCESS in 1h 06m 31s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/38/659538/7/check/octavia-v2-dsvm-tls-barbican/dfada12/ : SUCCESS in 1h 05m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/38/659538/7/check/octavia-v2-dsvm-scenario-amphora-v2/45d6937/ : SUCCESS in 1h 52m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/38/659538/7/check/octavia-v2-act-stdby-iptables-dsvm-scenario/cf7c29e/ : FAILURE in 1h 35m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/38/659538/7/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/976aa93/ : FAILURE in 1h 15m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/38/659538/7/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/0e76909/ : FAILURE in 56m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"09d0a4e42755536cd9c3ab052f90f25e8539daf1","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-08-07 15:47:04.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"4df57bf88f3f6d359e895eaf3c9054b42b702df2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 18:18:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/38/659538/7/check/openstack-tox-cover/a33b27d/cover/ : SUCCESS in 6m 49s\n- openstack-tox-lower-constraints https://logs.opendev.org/38/659538/7/check/openstack-tox-lower-constraints/3326365/ : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://logs.opendev.org/38/659538/7/check/openstack-tox-pep8/1f2b96e/ : SUCCESS in 5m 18s\n- openstack-tox-py27 https://logs.opendev.org/38/659538/7/check/openstack-tox-py27/7d5aef9/ : SUCCESS in 5m 21s\n- openstack-tox-py36 https://logs.opendev.org/38/659538/7/check/openstack-tox-py36/b7db90c/ : SUCCESS in 4m 54s\n- openstack-tox-py37 https://logs.opendev.org/38/659538/7/check/openstack-tox-py37/edd3229/ : SUCCESS in 5m 13s\n- openstack-tox-docs https://logs.opendev.org/38/659538/7/check/openstack-tox-docs/28efe8f/html/ : SUCCESS in 5m 29s\n- openstack-tox-functional https://logs.opendev.org/38/659538/7/check/openstack-tox-functional/3725c63/ : SUCCESS in 6m 47s\n- openstack-tox-functional-py36 https://logs.opendev.org/38/659538/7/check/openstack-tox-functional-py36/1217686/ : SUCCESS in 6m 14s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/38/659538/7/check/octavia-v2-dsvm-noop-api/8000122/ : SUCCESS in 45m 48s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/38/659538/7/check/octavia-v2-dsvm-noop-py2-api/93a192c/ : SUCCESS in 49m 36s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/38/659538/7/check/octavia-v2-dsvm-scenario/2ba5703/ : SUCCESS in 1h 33m 57s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/38/659538/7/check/octavia-v2-dsvm-py2-scenario/48b3acd/ : SUCCESS in 1h 39m 56s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/38/659538/7/check/octavia-v2-dsvm-py2-scenario-centos-7/6088be4/ : FAILURE in 2h 14m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/38/659538/7/check/octavia-v2-dsvm-scenario-two-node/5bd1640/ : SUCCESS in 2h 08m 50s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/38/659538/7/check/octavia-v2-dsvm-py2-scenario-two-node/e138fe6/ : SUCCESS in 2h 10m 54s (non-voting)\n- octavia-grenade https://logs.opendev.org/38/659538/7/check/octavia-grenade/1a53baf/ : SUCCESS in 1h 22m 00s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/38/659538/7/check/octavia-v2-dsvm-tls-barbican/290408e/ : SUCCESS in 57m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/38/659538/7/check/octavia-v2-dsvm-scenario-amphora-v2/0540fc4/ : SUCCESS in 1h 42m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/38/659538/7/check/octavia-v2-act-stdby-iptables-dsvm-scenario/63e1e5e/ : FAILURE in 1h 11m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/38/659538/7/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/08462c3/ : FAILURE in 1h 02m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/38/659538/7/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/0e457d5/ : SUCCESS in 1h 33m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"086aac9b59ba79e117bc2ba8bc5a8004ec4442f2","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-10-16 16:11:14.000000000","message":"Abandoned\n\nIncluded in https://review.opendev.org/#/c/647406","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"086516818cc77dbab89427ba4a5c85fe7d549422","revisions":{"db340b799c008d7e298332fa1b0591bc866687ab":{"kind":"REWORK","_number":1,"created":"2019-05-16 11:52:09.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/38/659538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/659538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/659538/1"}}},"commit":{"parents":[{"commit":"09020b6bfc06c615f6d01d550188b1e6d7ed9d21","subject":"Merge \"Add Python 3.7 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09020b6bfc06c615f6d01d550188b1e6d7ed9d21"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-05-16 10:07:25.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-05-16 11:51:50.000000000","tz":240},"subject":"[Jobboard] Importable flow functions","message":"[Jobboard] Importable flow functions\n\nPart of required refactor for implementation of taskflow jobboard\nbased controller is creation of importable flow functions.\n\nChange-Id: I6570f9f82173a1258da76899de8f1410ff5fc52a\nStory: 2005072\nTask: 30816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db340b799c008d7e298332fa1b0591bc866687ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db340b799c008d7e298332fa1b0591bc866687ab"}]},"branch":"refs/heads/master"},"65e7137b3cce289b0f0134c29ace84d6c4972125":{"kind":"REWORK","_number":2,"created":"2019-05-22 08:09:15.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/38/659538/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/659538/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/659538/2"}}},"commit":{"parents":[{"commit":"a2aaf5a1f63b4ac9ec0b731449c2b864bcfcb9a6","subject":"Create Amphora V2 provider driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a2aaf5a1f63b4ac9ec0b731449c2b864bcfcb9a6"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-05-16 10:07:25.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-05-22 08:08:23.000000000","tz":240},"subject":"[Jobboard] Importable flow functions","message":"[Jobboard] Importable flow functions\n\nPart of required refactor for implementation of taskflow jobboard\nbased controller is creation of importable flow functions.\n\nChange-Id: I6570f9f82173a1258da76899de8f1410ff5fc52a\nStory: 2005072\nTask: 30816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/65e7137b3cce289b0f0134c29ace84d6c4972125"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/65e7137b3cce289b0f0134c29ace84d6c4972125"}]},"branch":"refs/heads/master"},"0dd43ba3ddf9fb36f0521b0ef222b252427a42ff":{"kind":"REWORK","_number":3,"created":"2019-05-22 08:35:56.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/38/659538/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/659538/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/659538/3"}}},"commit":{"parents":[{"commit":"a2aaf5a1f63b4ac9ec0b731449c2b864bcfcb9a6","subject":"Create Amphora V2 provider driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a2aaf5a1f63b4ac9ec0b731449c2b864bcfcb9a6"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-05-16 10:07:25.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-05-22 08:34:40.000000000","tz":240},"subject":"[Jobboard] Importable flow functions","message":"[Jobboard] Importable flow functions\n\nPart of required refactor for implementation of taskflow jobboard\nbased controller is creation of importable flow functions.\n\nChange-Id: I6570f9f82173a1258da76899de8f1410ff5fc52a\nStory: 2005072\nTask: 30816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0dd43ba3ddf9fb36f0521b0ef222b252427a42ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0dd43ba3ddf9fb36f0521b0ef222b252427a42ff"}]},"branch":"refs/heads/master"},"f6793be3aa105dbbf5745664e412bd2aa051a420":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-05-23 10:32:56.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/38/659538/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/659538/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/659538/4"}}},"commit":{"parents":[{"commit":"841cd348a444af66ad1953ed1fa8e3d17aebdfaf","subject":"Create Amphora V2 provider driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/841cd348a444af66ad1953ed1fa8e3d17aebdfaf"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-05-16 10:07:25.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-05-23 10:32:56.000000000","tz":0},"subject":"[Jobboard] Importable flow functions","message":"[Jobboard] Importable flow functions\n\nPart of required refactor for implementation of taskflow jobboard\nbased controller is creation of importable flow functions.\n\nChange-Id: I6570f9f82173a1258da76899de8f1410ff5fc52a\nStory: 2005072\nTask: 30816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f6793be3aa105dbbf5745664e412bd2aa051a420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f6793be3aa105dbbf5745664e412bd2aa051a420"}]},"branch":"refs/heads/master"},"d0febe3c40627574911d112ef0023d02d06d1dea":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-06-06 13:11:40.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/38/659538/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/659538/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/659538/5"}}},"commit":{"parents":[{"commit":"ff4680eb71cf03e4eae1a58e7a66321ddadcbead","subject":"Merge \"Create Amphora V2 provider driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff4680eb71cf03e4eae1a58e7a66321ddadcbead"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-05-16 10:07:25.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-06 13:11:40.000000000","tz":0},"subject":"[Jobboard] Importable flow functions","message":"[Jobboard] Importable flow functions\n\nPart of required refactor for implementation of taskflow jobboard\nbased controller is creation of importable flow functions.\n\nChange-Id: I6570f9f82173a1258da76899de8f1410ff5fc52a\nStory: 2005072\nTask: 30816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0febe3c40627574911d112ef0023d02d06d1dea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0febe3c40627574911d112ef0023d02d06d1dea"}]},"branch":"refs/heads/master"},"bcb6a0a5d906b6ecd5fa61ff95e9dbd0ca0eaf76":{"kind":"REWORK","_number":6,"created":"2019-07-17 14:26:01.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/38/659538/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/659538/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/659538/6"}}},"commit":{"parents":[{"commit":"ff4680eb71cf03e4eae1a58e7a66321ddadcbead","subject":"Merge \"Create Amphora V2 provider driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff4680eb71cf03e4eae1a58e7a66321ddadcbead"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-05-16 10:07:25.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-07-17 14:25:40.000000000","tz":240},"subject":"[Jobboard] Importable flow functions","message":"[Jobboard] Importable flow functions\n\nPart of required refactor for implementation of taskflow jobboard\nbased controller is creation of importable flow functions.\n\nChange-Id: I6570f9f82173a1258da76899de8f1410ff5fc52a\nStory: 2005072\nTask: 30816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bcb6a0a5d906b6ecd5fa61ff95e9dbd0ca0eaf76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bcb6a0a5d906b6ecd5fa61ff95e9dbd0ca0eaf76"}]},"branch":"refs/heads/master"},"086516818cc77dbab89427ba4a5c85fe7d549422":{"kind":"REWORK","_number":7,"created":"2019-08-06 10:12:43.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/38/659538/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/659538/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/659538/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/659538/7"}}},"commit":{"parents":[{"commit":"ff4680eb71cf03e4eae1a58e7a66321ddadcbead","subject":"Merge \"Create Amphora V2 provider driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff4680eb71cf03e4eae1a58e7a66321ddadcbead"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-05-16 10:07:25.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-08-06 10:12:08.000000000","tz":240},"subject":"[Jobboard] Importable flow functions","message":"[Jobboard] Importable flow functions\n\nPart of required refactor for implementation of taskflow jobboard\nbased controller is creation of importable flow functions.\n\nChange-Id: I6570f9f82173a1258da76899de8f1410ff5fc52a\nStory: 2005072\nTask: 30816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/086516818cc77dbab89427ba4a5c85fe7d549422"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/086516818cc77dbab89427ba4a5c85fe7d549422"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
