)]}'
{"id":"openstack%2Foctavia~659626","triplet_id":"openstack%2Foctavia~master~Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6","project":"openstack/octavia","branch":"master","topic":"vip-acl","hashtags":[],"change_id":"Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6","subject":"Add VIP access control list","status":"MERGED","created":"2019-05-16 17:22:05.000000000","updated":"2019-09-13 17:10:47.000000000","submitted":"2019-09-13 17:09:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":67,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659626-1568394567290-e46e57b4","meta_rev_id":"bacc50eec40eeadc492790a99d9ab70cc7d8d51e","_number":659626,"virtual_id_number":659626,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2019-09-13 08:09:51.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-09-13 17:09:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-13 08:12:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-13 15:04:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-13 15:04:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2019-09-13 15:04:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 17:49:20.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-09-05 06:46:41.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-09-13 08:12:01.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-13 15:04:54.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-13 17:09:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"099f50a28ce043f6ca00fa2811bde486b78e5c63","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-16 17:22:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c45ce7bdb34f3c6ca71d8fab9a3a6b18c9e50f7f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-16 17:22:37.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9af674d07857fea679495d4126a757ad183489cb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-16 17:24:36.000000000","message":"Topic changed from acl to vip-acl","accounts_in_message":[],"_revision_number":2},{"id":"0904d4f1816c086b6863071e06d7c9218fb5a8db","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-16 17:28:59.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"236d0bba40520f8cb40ea42dc4db6119302846b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 19:48:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/26/659626/2/check/openstack-tox-cover/e8bcf0a/ : FAILURE in 5m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/26/659626/2/check/openstack-tox-lower-constraints/52da88b/ : FAILURE in 5m 29s\n- openstack-tox-pep8 http://logs.openstack.org/26/659626/2/check/openstack-tox-pep8/b68ffd8/ : SUCCESS in 8m 11s\n- openstack-tox-py27 http://logs.openstack.org/26/659626/2/check/openstack-tox-py27/7b145e5/ : FAILURE in 5m 13s\n- openstack-tox-py36 http://logs.openstack.org/26/659626/2/check/openstack-tox-py36/52d6af9/ : FAILURE in 5m 37s\n- openstack-tox-py37 http://logs.openstack.org/26/659626/2/check/openstack-tox-py37/64a2f14/ : FAILURE in 8m 52s\n- openstack-tox-docs http://logs.openstack.org/26/659626/2/check/openstack-tox-docs/35f65ec/html/ : SUCCESS in 6m 04s\n- openstack-tox-functional http://logs.openstack.org/26/659626/2/check/openstack-tox-functional/df312f6/ : FAILURE in 6m 15s\n- openstack-tox-functional-py36 http://logs.openstack.org/26/659626/2/check/openstack-tox-functional-py36/561f5b0/ : FAILURE in 7m 07s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/26/659626/2/check/octavia-v2-dsvm-noop-api/8d61531/ : FAILURE in 43m 45s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/26/659626/2/check/octavia-v2-dsvm-noop-py2-api/0c41f32/ : FAILURE in 41m 33s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/26/659626/2/check/octavia-v2-dsvm-scenario/345fb72/ : FAILURE in 1h 17m 57s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/26/659626/2/check/octavia-v2-dsvm-py2-scenario/369608d/ : FAILURE in 1h 20m 56s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/26/659626/2/check/octavia-v2-dsvm-py2-scenario-centos-7/d75568e/ : FAILURE in 2h 13m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/26/659626/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/03d1996/ : FAILURE in 1h 42m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/26/659626/2/check/octavia-v2-dsvm-scenario-two-node/8ea56a6/ : FAILURE in 1h 49m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/26/659626/2/check/octavia-v2-dsvm-py2-scenario-two-node/a247641/ : FAILURE in 1h 36m 55s (non-voting)\n- octavia-grenade http://logs.openstack.org/26/659626/2/check/octavia-grenade/44ed4fb/ : FAILURE in 1h 06m 17s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/26/659626/2/check/octavia-v2-dsvm-tls-barbican/a5e2ae9/ : SUCCESS in 52m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f5c533dbc1c20ea0c70418df11a138fdbc9ffcf0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-25 21:40:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"929b9a260f81d63dc2caef1e0c138949775ae10f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-25 21:40:29.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"2ee74fb767e04bb125fd24d510fd2f93cd2635a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 23:46:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/26/659626/3/check/openstack-tox-cover/62bd999/ : FAILURE in 4m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/26/659626/3/check/openstack-tox-lower-constraints/0b1f6f2/ : FAILURE in 4m 12s\n- openstack-tox-pep8 http://logs.openstack.org/26/659626/3/check/openstack-tox-pep8/07a758d/ : SUCCESS in 6m 23s\n- openstack-tox-py27 http://logs.openstack.org/26/659626/3/check/openstack-tox-py27/decdd22/ : FAILURE in 4m 06s\n- openstack-tox-py36 http://logs.openstack.org/26/659626/3/check/openstack-tox-py36/ee9ef26/ : FAILURE in 5m 05s\n- openstack-tox-py37 http://logs.openstack.org/26/659626/3/check/openstack-tox-py37/936fc1e/ : FAILURE in 4m 10s\n- openstack-tox-docs http://logs.openstack.org/26/659626/3/check/openstack-tox-docs/7936911/html/ : SUCCESS in 5m 27s\n- openstack-tox-functional http://logs.openstack.org/26/659626/3/check/openstack-tox-functional/6baf9ed/ : FAILURE in 4m 56s\n- openstack-tox-functional-py36 http://logs.openstack.org/26/659626/3/check/openstack-tox-functional-py36/36bb51e/ : FAILURE in 5m 18s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/26/659626/3/check/octavia-v2-dsvm-noop-api/b42dc4d/ : FAILURE in 30m 21s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/26/659626/3/check/octavia-v2-dsvm-noop-py2-api/619106b/ : FAILURE in 31m 26s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/26/659626/3/check/octavia-v2-dsvm-scenario/2458067/ : FAILURE in 44m 15s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/26/659626/3/check/octavia-v2-dsvm-py2-scenario/5d6738a/ : FAILURE in 39m 52s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/26/659626/3/check/octavia-v2-dsvm-py2-scenario-centos-7/645449b/ : FAILURE in 1h 57m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/26/659626/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/99a1936/ : FAILURE in 44m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/26/659626/3/check/octavia-v2-dsvm-scenario-two-node/dc98f71/ : FAILURE in 48m 21s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/26/659626/3/check/octavia-v2-dsvm-py2-scenario-two-node/0e7680f/ : SUCCESS in 1h 08m 33s (non-voting)\n- octavia-grenade http://logs.openstack.org/26/659626/3/check/octavia-grenade/34a55ea/ : FAILURE in 1h 06m 55s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/26/659626/3/check/octavia-v2-dsvm-tls-barbican/46ae93e/ : SUCCESS in 58m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/26/659626/3/check/octavia-v2-dsvm-scenario-amphora-v2/f9c9259/ : FAILURE in 43m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e6067559c9e72c3ba680c6bc26713ffdbfa3b85f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-16 09:53:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c3ae399aa6549a69903c8b873e52cd39ef0301ff","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-16 09:58:51.000000000","message":"Patch Set 4: Workflow-1\n\noctavia/controller/worker/v2/flows/listener_flows.py","accounts_in_message":[],"_revision_number":4},{"id":"e40a012106789122a1c64a7e926d85f6419101b8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-16 11:45:16.000000000","message":"Patch Set 4:\n\n2019-07-16 11:26:58.446 | + /opt/stack/new/octavia/devstack/upgrade/resources.sh:destroy:102 :   openstack loadbalancer delete --cascade lb1\n2019-07-16 11:27:06.408 | __init__() got an unexpected keyword argument \u0027allowed_cidrs\u0027 (HTTP 500) (Request-ID: req-c3de601d-895f-4245-92e4-336e0dced42a)","accounts_in_message":[],"_revision_number":4},{"id":"6d4952ebffb88f36e05c2502ec6a900b6b131d53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 12:14:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/26/659626/4/check/build-openstack-api-ref/1bde969/html/ : SUCCESS in 3m 19s\n- openstack-tox-cover http://logs.openstack.org/26/659626/4/check/openstack-tox-cover/4b70f84/ : FAILURE in 3m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/26/659626/4/check/openstack-tox-lower-constraints/7efe301/ : FAILURE in 3m 49s\n- openstack-tox-pep8 http://logs.openstack.org/26/659626/4/check/openstack-tox-pep8/bbea628/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/26/659626/4/check/openstack-tox-py27/f608415/ : FAILURE in 3m 39s\n- openstack-tox-py36 http://logs.openstack.org/26/659626/4/check/openstack-tox-py36/fc52eb3/ : FAILURE in 3m 53s\n- openstack-tox-py37 http://logs.openstack.org/26/659626/4/check/openstack-tox-py37/6503120/ : FAILURE in 4m 01s\n- openstack-tox-docs http://logs.openstack.org/26/659626/4/check/openstack-tox-docs/4b65ba4/html/ : SUCCESS in 6m 21s\n- build-openstack-releasenotes http://logs.openstack.org/26/659626/4/check/build-openstack-releasenotes/cb5da1e/html/ : SUCCESS in 4m 07s\n- openstack-tox-functional http://logs.openstack.org/26/659626/4/check/openstack-tox-functional/93de5ac/ : FAILURE in 4m 31s\n- openstack-tox-functional-py36 http://logs.openstack.org/26/659626/4/check/openstack-tox-functional-py36/1c85269/ : FAILURE in 5m 12s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/26/659626/4/check/octavia-v2-dsvm-noop-api/82ef3a0/ : SUCCESS in 46m 05s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/26/659626/4/check/octavia-v2-dsvm-noop-py2-api/e4677e0/ : SUCCESS in 41m 06s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/26/659626/4/check/octavia-v2-dsvm-scenario/29ac561/ : SUCCESS in 1h 27m 26s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/26/659626/4/check/octavia-v2-dsvm-py2-scenario/6d43103/ : SUCCESS in 1h 23m 38s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/26/659626/4/check/octavia-v2-dsvm-py2-scenario-centos-7/eecb3fb/ : FAILURE in 1h 58m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/26/659626/4/check/octavia-v2-dsvm-scenario-ubuntu-bionic/ac8dff6/ : SUCCESS in 1h 43m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/26/659626/4/check/octavia-v2-dsvm-scenario-two-node/5760a65/ : SUCCESS in 2h 12m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/26/659626/4/check/octavia-v2-dsvm-py2-scenario-two-node/4f8c2c6/ : SUCCESS in 1h 39m 40s (non-voting)\n- octavia-grenade http://logs.openstack.org/26/659626/4/check/octavia-grenade/1a00ecb/ : FAILURE in 1h 33m 35s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/26/659626/4/check/octavia-v2-dsvm-tls-barbican/7c41699/ : SUCCESS in 57m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/26/659626/4/check/octavia-v2-dsvm-scenario-amphora-v2/98bf108/ : SUCCESS in 1h 35m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e2b85cbae175a62fd73bb367ba821bf13cdd1e29","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-16 16:35:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"06b4181c087e5224e4b3f578e9b3dfa54aa9a8ed","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-16 16:37:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a32b97ed8a13a571cd881f82dd71b82878d3777c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 18:42:41.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/204ea852d92e4ffd8c6163cb350e1882 : SUCCESS in 4m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f3259b2ffbf433a853b4fab64eb37c4 : SUCCESS in 8m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc0503ab8f5545599a1d6ba366231069 : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65dfc3423e3247fb8265ff9a0156d6c3 : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/79e2b01bd229477aa4092287e86d4111 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc38dce0fbb14cfdb555c2393d9a212f : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1359c0d9d83c4a8db48c677cac53cfc9 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2fcc0f3eb88406b926b512233e12dac : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b1ae5fcde6a4fbabc947bdb5db36c25 : SUCCESS in 5m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0097adad75ca4622b871378ac92925ad : FAILURE in 4m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b69bfc9d6e8d4b3c925845de377a40b6 : SUCCESS in 5m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/45a80406bd7948c8800e7b7726b81775 : SUCCESS in 46m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c50dd1a2bd20468a9211ab4dd139bc2c : SUCCESS in 41m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15914b55158048f0a9c7982d409ceda9 : FAILURE in 1h 37m 00s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b2846593d8e14d02a7fb6518e17f8421 : SUCCESS in 1h 34m 31s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4093ffda306b48cea3f1af93146216bc : FAILURE in 6m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5a86786f3f844200965867c0a40c3128 : FAILURE in 1h 52m 55s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0d85b664dffe4ae3af609a0b2a1c9b1b : POST_FAILURE in 1h 08m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9202b121b53f4320bf0cf7d42ee31651 : SUCCESS in 1h 19m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/59f2a923cb7c47b38084f1580a3db7fa : SUCCESS in 54m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/922fdf58055c4229907e91229340d4e0 : SUCCESS in 1h 42m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b75cc41297e1457db5f5c55e939c2f1c : SUCCESS in 1h 07m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3de8fc592c874c7db82987a768765a02 : SUCCESS in 1h 15m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/98e28fb122664a2e87c4fbc45c9f66af : SUCCESS in 1h 33m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9e83eeb4e7445517a2732389545f9bd578c7b658","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-17 10:19:25.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"ab27f9e0bc5c5ec2d23d1a77cd5f76fea3ac2a2b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-20 20:54:56.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d8e736dd68dbe1a6eab326b243c52487bd40b7b4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-20 20:57:27.000000000","message":"Patch Set 7: Workflow-1\n\nThe patch is open for review. I am Workflow-1 because there are some code paths not covered by tests yet.","accounts_in_message":[],"_revision_number":7},{"id":"166a75fe31b6f047c38b67ad112f1ba3d04f2d2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 23:08:43.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e201db3a707341efb0d1a03bc1b59a13 : SUCCESS in 7m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a2f93c7aec345d09492c7c9cf202b85 : SUCCESS in 8m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f244e7d351db41ecbc31f464a25c696e : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2de07786f4cf4caf8ea1f19eea5f1a44 : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5760e61abb54495186c380795ea473e1 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3eae2d837fe4f5a8feb1954dcf46761 : SUCCESS in 6m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66b6cc7052264e20abbb3abf00ec0514 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/261b7238538f4623a7e0d6af41230d82 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00d63bdcddc64e3ca5d74397f3e1d2b9 : SUCCESS in 6m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e92fe3dc233a492a97147c6bf5fcd0d0 : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/58ceda4657514b3c82a603d1358da23a : SUCCESS in 5m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/62ca9185fdea4580a073a48137d12c71 : SUCCESS in 54m 49s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/90f50825bb824bb1851f6eba9307782d : SUCCESS in 54m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0ad46c0dffc4043b23e15b6fed1776d : SUCCESS in 1h 50m 10s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/87ba86edf7b841be92d89f158ced9726 : SUCCESS in 1h 35m 03s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1a124832680f4d85b78c3940c65aec57 : SUCCESS in 1h 51m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b5f7bd4414144ccc8ec62a7c02066497 : FAILURE in 1h 31m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/17281c9dbabb48d88b04f3b0cccfe15c : FAILURE in 1h 58m 28s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e86446b9a7d04a3e987b9068db311426 : SUCCESS in 1h 15m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cf60846ad9cd4d5b85aade253b695c39 : SUCCESS in 1h 02m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7fd7e021563e46c5aacf0c9fb2d98e61 : SUCCESS in 2h 04m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11791748c2ba4da7893dc1228268d0b0 : SUCCESS in 1h 18m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4e48b53ea1e94037a229634908c3cd88 : SUCCESS in 1h 07m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6b62e443a57a491ebc133a2cc4f81095 : SUCCESS in 1h 26m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"06d631171fc08c6a6e9c60fc52654d007ed9e21f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-22 16:16:26.000000000","message":"Patch Set 7: -Workflow","accounts_in_message":[],"_revision_number":7},{"id":"b2b10e9ec95dbf361d75913e36de6cc9ca1c8dff","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-23 00:26:39.000000000","message":"Patch Set 7: Code-Review-1\n\n(20 comments)\n\nLooks pretty good, mostly minor comments.","accounts_in_message":[],"_revision_number":7},{"id":"df7063110a6d3887bd80f8ef21ff7fe363e72708","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-23 00:27:20.000000000","message":"Patch Set 7:\n\nAlso, please update the provider driver guide for the new parameter.","accounts_in_message":[],"_revision_number":7},{"id":"4bd51e5fd1eca3246d5da47e4571e4c7ed0ed9e3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 13:03:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"74c26af3eeb5b4181d57a5e94536d428fed00ea4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 13:04:28.000000000","message":"Patch Set 7:\n\n(20 comments)\n\nForgot to update provider driver guide in PS8.","accounts_in_message":[],"_revision_number":7},{"id":"03163eddf4cbb6be050e5faab6e8da2f29e08d7a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 13:31:23.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e659de87814dead8aeff027b39ff066044037439","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 13:49:27.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"726e7063d5eff9f8fc3761c359a73dc290bf7b5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:33:47.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/403e87df8d6d48ee96109b95cb7060a7 : SUCCESS in 4m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/088b93b6aaab4b4a84be0052231040ae : SUCCESS in 7m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e3ee59d03064713abe7cef44eea0f95 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb32b2d0e07a448a8b769d6737f54d8d : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5386aba42e8944218b9ea2203034f3e1 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3dbc8d0e6ac4f8fa4800e08dbea1e29 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a4231d4ebbc495f96c2561ad78859bd : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a5ae6fddeab4c1e819fe8d688fd4cd7 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7bcf57bd34044e18679a28cc2f046fb : SUCCESS in 5m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/acb8c351dc3940f882cac4aa540dadf9 : SUCCESS in 5m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4e103295df864359a05c5390ea4b0c31 : SUCCESS in 7m 29s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ec67b92157734cd08505ea13dc1cf28d : SUCCESS in 6m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1c37c0d1f70545dc9d524c982fed7a6f : SUCCESS in 6m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a8ed5a3d333842ddad03acd3de7367cd : SUCCESS in 51m 23s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3fce4097c93f47c2a76209277d93820a : SUCCESS in 47m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/07cc3e02d99743f8b3805748f810aa6c : SUCCESS in 2h 06m 13s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c0726212ed134acaaa474d4af1f78591 : SUCCESS in 1h 54m 55s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/01342377f85c4782a699ed002e24c268 : TIMED_OUT in 2h 37m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f3d73efdae504ed6b722451362022fd3 : SUCCESS in 2h 05m 59s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ac4fd14054d1432d87580f14cc811d73 : SUCCESS in 1h 58m 30s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/473e0d9c8c3e48398881898802763557 : SUCCESS in 1h 14m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f1a6f3333de84de1a126f5cf89efa8e0 : SUCCESS in 52m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6b2b5c6e10324776840447d743fa9f5e : SUCCESS in 1h 35m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/56d2d8b14e2b45539cde67642496b20b : FAILURE in 1h 40m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/45af7b2e5b1d4b839ab5dff762b645a1 : SUCCESS in 1h 11m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4b54882500ba45bca55b7b00761d136b : FAILURE in 1h 23m 37s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3703e16d3c934a93da80a005b5da8f3aba9fc115","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-26 17:27:32.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8fcea278caca4c527fe6222b39f6c263846398f2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-27 08:35:49.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"51ace954f24b430b80c97e437be377ba01bf2326","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-08-27 09:01:52.000000000","message":"Patch Set 10: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":10},{"id":"2e4a464385d473d1881668c1ec91d459439627e9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-27 09:05:14.000000000","message":"Patch Set 10:\n\n(9 comments)","accounts_in_message":[],"_revision_number":10},{"id":"8722f4f4800054982bac33703f471afc821ccf07","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-27 09:05:14.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"d09f731701a0d6d45645233b8a4b5716620af9fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 11:47:49.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0402dca5424a445d810f9ef983ba43a2 : SUCCESS in 6m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e1c6edf002f47d3b9524dc32adf1589 : SUCCESS in 8m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d211da7a8aea4e72934228f87a094686 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faf058a485f44de8b650542c6241d813 : SUCCESS in 5m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1232a2d11bc841d8b4bc2f8538b286e6 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1afdbc12995740d7bd9f8bbb18755629 : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa227f621e5c44bb9a51feb5e3f15539 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/052dfc8b3d8247ed85d991b5cf0abd1c : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/134ee7825957429885d6454ae480b20f : SUCCESS in 5m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bd7531014ab64d0b90057255da1aad2b : SUCCESS in 5m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/326e221ffc3c4613a3a3eef9df8a0c14 : SUCCESS in 7m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/60ad88428d67414e94c6eff55a998bb4 : SUCCESS in 8m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b5f88f4aaa844254bb55a15e07d81171 : SUCCESS in 5m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/11c0a0d031c947048c453056c80e014a : SUCCESS in 47m 41s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/02e79814063a40d3b2943604f8aa1119 : SUCCESS in 48m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2dc46b13f5764e55a34b307e30b650de : SUCCESS in 1h 36m 17s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3f91833e13f640c799cfbb98ae449c28 : SUCCESS in 1h 49m 37s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/20222f830c514d35a5a230caeeb9ec2b : TIMED_OUT in 2h 36m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a7ef8dc03b0643b198b90dbeddc48cd0 : SUCCESS in 1h 23m 01s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/b7cdd35054284409ba9998be911d7749 : SUCCESS in 1h 57m 28s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/216bb233940c4babbf0280759ea1d423 : SUCCESS in 1h 17m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b215d1996f7c40b4961b0b5712f9bac9 : SUCCESS in 1h 02m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/400e79b9ccc84db9bce0965c003c02a7 : SUCCESS in 1h 35m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31ce39a813bd47b89822bdd1902c7419 : SUCCESS in 1h 05m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7ff00f59b3524df4bee01fbef6851951 : SUCCESS in 1h 26m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a07a719b43504109b63cb81bea82a5e6 : FAILURE in 1h 11m 33s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"58726bcc9a53e92cd3b3e0909f63d87989b3ede9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-08-27 13:07:41.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"65fd8e569725e8dbb899ef8375b579995d6a8344","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-08-27 15:37:40.000000000","message":"Patch Set 11: Code-Review-1\n\nAny issue exists when using 2 listeners with acls on the same LB. When adding the 2nd listener, acls from the 1st listener are removed.\nIt seems that ListenerCidr are not correctly added in the DB.","accounts_in_message":[],"_revision_number":11},{"id":"b3c8b47af492488c85eeca7880383d2a74acae75","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-27 16:01:50.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"3e6610a906656909e9db86b9f1e2c5b80ff1b288","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-27 16:01:57.000000000","message":"Patch Set 11:\n\nFixed. Thank you, Greg!","accounts_in_message":[],"_revision_number":11},{"id":"d963bd113c9b8603f851313ed86438cd304b5100","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-08-27 17:49:20.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"f13b2bc6941a858911e5896d954f4328ffbb251f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 18:25:19.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/349ea470842944269e549cdff246ae1e : SUCCESS in 4m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d285bbc5cc624077bd532606b9a14a6b : SUCCESS in 7m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45b0168c4b954498bd26012278fc6e5b : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63be48122b6449f39fb23fdefcd9c9fb : SUCCESS in 7m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/deb4ce12f1b948ada5046913a9ddac68 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f31e25abc4c43398b971d1c401a3c9f : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c17a87a207614d9bb57dfc8db705346b : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae73185bed714addae5c6ea851d86b9f : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c381fd4e71244f73a0d5c53d7b44cbf3 : SUCCESS in 6m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/123d074385924fe893d0fc658caae5f7 : SUCCESS in 5m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ada45a0fcf2249e79ce6cb3a3b32ffa9 : SUCCESS in 7m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a5452c72e2974f9fa3e45cec3aa0d1da : SUCCESS in 7m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8f19bcaa98ec44e7af8a719611dbe2bc : SUCCESS in 5m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3026fdaf969640a7bae63c822ffdc154 : SUCCESS in 42m 18s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/42d8f83c668b43dcab97f1a4608121c9 : SUCCESS in 49m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e303a6347224d2ea2abd1600acab630 : SUCCESS in 1h 44m 58s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3e0353d66f2544eeb3ee67bbe0b4a3ba : SUCCESS in 1h 44m 28s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ad22a91991e344279f3762e845d7e9f4 : SUCCESS in 2h 12m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fd9b0f4be23745cdb57041b6ff4f3539 : SUCCESS in 2h 07m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5e233279bab248dc91a85e63209bb790 : SUCCESS in 2h 07m 28s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d5c1d77a35f74f2fa935136cf43f7a5e : SUCCESS in 1h 36m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/86eb8f9a3d3d498cb2ee0ec656b768c4 : SUCCESS in 58m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2f41a3692fbb46c6b035b84199194a5c : FAILURE in 1h 36m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d6a5e2f78b464264bb3d10db53ad2656 : SUCCESS in 1h 16m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f4482fc9908e4a44856fc505548af8c2 : SUCCESS in 1h 17m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0c1e4e48c0004772814a25d005a977ca : FAILURE in 1h 10m 38s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"74a870fb2292a0bf20a122e36a582b9403e3d8f9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-27 21:57:29.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ac01ccc4bd523e3470dc7fef5bb77d83c4819d62","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-03 20:25:27.000000000","message":"Patch Set 12: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":12},{"id":"97f0472cbcd5e5fd886ab84910e3b52e19145494","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-09-05 06:46:41.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"0a5a9f8b8c95377403449e62426e341eb190ddbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 06:47:00.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"9f66944b40bea9f1e84f7e0b239c02c14ac8862e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 11:56:12.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3b3a0b31ac5e4f499d7589d21e7ca549 : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9bc10f22d93840c4be0ef2b6eb570232 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca3c3eee02604b92912201f35b3e30fb : SUCCESS in 6m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4995b8a8e1c94df1bd6c9cf844edb0e7 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ff7b0665ef34f00b3829b9aad167984 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c550111e907c4740ac149ba83dec8c64 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbda57fbceea4b1b94b3a803fb1354a8 : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d6aca3f01b44fb7badabb9adb74cc35 : SUCCESS in 5m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4c66e637f53d4a22934427cfba7dffde : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8062db82f6574fca85a1bbbb7944039c : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3044a864067c45f1aab7b9342c148107 : CANCELED\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9797fb1a2d3b4c378e14655e19634463 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a69ed853d8cf4c07b5e15514ddb93028 : CANCELED\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/79d9fe3041aa409995294600f36cf3c2 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/14b7d05c160a433daf32a6f87ace9fc8 : FAILURE in 30m 49s","accounts_in_message":[],"_revision_number":12},{"id":"3bd4f8d5bfeb729136f8876e98067d7cbe4ed7e4","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-09-08 13:53:15.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"84d8bf7bd6b4c8d6655bae909677a583c2dfbda2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-08 18:32:21.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/19cd00db56964bcda2104c3cfddf1542 : SUCCESS in 3m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/336497ee6c68410d90dbccac54b53dd4 : FAILURE in 3m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8de068a616c4ba4a26c374fd70bbce5 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/427789a184c843ea958798722c75855a : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5c16b1d3e52b47e596ae9fbd3494562d : FAILURE in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22ff2b4c1bcb4fecb2d539aed2ee34f2 : FAILURE in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9bf559b3f7144d10ad36b39eb6bc6c0a : FAILURE in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2977230bc3d94e2593ca8f982e7a566b : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82ac3c3948d74904a14c14af38dd1a54 : SUCCESS in 4m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5ad12b5983244e98aa98e52b5646a6a1 : FAILURE in 4m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6a1bc136b34549698c801e2a3601b08a : SUCCESS in 5m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c8f2e2ae21ff4cf59d703de687af5fd9 : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/59ed995965d24a25b43955c9ff64e09f : SUCCESS in 4m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c1e158f187af4005a69f46310b9a7c90 : SUCCESS in 41m 41s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ad87c794a1814bbb976dab462100379d : SUCCESS in 42m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a49b5353635a4c04b6b8938563dc9243 : SUCCESS in 1h 49m 54s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0deea412c7c44c6faf2a386ad042a222 : SUCCESS in 1h 40m 11s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/08e80c8d2bd54955913e60b0ddf35f5f : SUCCESS in 2h 16m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8f0e89ad3c584a9eb65d125ee5d498c6 : SUCCESS in 2h 01m 51s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a9ce1e8d749c432295206e58b61ecfbf : SUCCESS in 1h 48m 49s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/992a552ef6f4464d8fc7927a6ec8036d : SUCCESS in 1h 21m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/02153ba1fb6046b99b128ae0873cc3ee : SUCCESS in 51m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f1de75a9249c469fb9a3362ac898f092 : RETRY_LIMIT in 1h 00m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9cc02bbd2ca14649926e21dd6cac3441 : SUCCESS in 1h 08m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/821fead0d97f4cb9bbe48968f42b8cdf : FAILURE in 38m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7a3e07d7d21844edb0a498587d618d7b : FAILURE in 1h 19m 57s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"16a4d0d33de239c6179009502ddaa705302e58e0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-10 10:53:50.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"db7cf469c06bc49df01cea19d9a2c4c2fd7b844f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 13:43:47.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ec599f05dc614ccd8ad10920826176fb : SUCCESS in 4m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41d5817651f7489696d87f7549199f38 : SUCCESS in 8m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/394674e62fbc41318ccd352bb2863eaf : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6324bb3a3404c859edb493626005462 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e5a393c248649e7935dc758e3904de6 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af1273d9d75c400d9d7fc0bb20a19fba : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea42fafe0f2a45feb7ece618d3c7defe : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/433c6ac68c014d2d84fbc429afe579b0 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bbc3f1d7bc24a5981ceb7de6b69eff0 : SUCCESS in 5m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e3d031c4927940ae9f0075c33651affc : SUCCESS in 5m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b2446a9844a54a72996da6a6f761ab3b : SUCCESS in 6m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/29d39f5e748140f3ab81514796c8ef2e : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf1ac2f48b5741b29a7a018db071c7c3 : SUCCESS in 5m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9cfb22f34e8a43e9817af9b34fed972a : SUCCESS in 43m 59s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/05822815ad414018b52f67fe73f1694e : SUCCESS in 50m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/87c94ebeaf20448184f5ddd6daa24537 : RETRY_LIMIT in 55m 58s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7541f52c39cc4e328d7f88730c541055 : SUCCESS in 1h 40m 03s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/016f75bcfdd240e3bcc124b3d22c7157 : SUCCESS in 2h 21m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3d9dc65ad5634d7f8c2054fef02e42df : SUCCESS in 2h 26m 01s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/634dcd1a12ee40569c30a1bd5a435865 : SUCCESS in 1h 36m 15s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c279d9f6610442a3a84f6042b02ec1cd : SUCCESS in 1h 33m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0e852135783643b68d7b36619de06890 : SUCCESS in 1h 03m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8765627da8dd44259693fdda4a08e7d5 : POST_FAILURE in 1h 38m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e88341b23404b399c93a58064e3cd7b : SUCCESS in 1h 08m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0ccde9e8df9c4c6e8941813cdf385d42 : FAILURE in 1h 06m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0cd64726452b48919d93599157fd43c4 : FAILURE in 1h 36m 34s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"04bc01a5c36db4ba17400c8e51179e012b97f5b7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-10 14:06:59.000000000","message":"Patch Set 13: Backport-Candidate-2","accounts_in_message":[],"_revision_number":13},{"id":"1e215368ca8f2256c255c06876d2c543bb963a9c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-10 19:53:03.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"981ea28d53853d03b01e8421e8d197061cf9f9c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 23:14:12.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/18c8fdc2056f4460924a83130f49fecc : SUCCESS in 5m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78245d6a7dbc4968b7b18c233d4c536a : FAILURE in 7m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcf645e7964a464490af28735df70f23 : FAILURE in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/471b8b39a71f4dbdab088bf8f0e0414a : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57a25ca737d14c41860e35c349d8019e : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9917acdcf8d49a297533dcae33a560a : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03b43cddb7a348bcae1d4062fd974c10 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5880c00afd2e4fb8a9f164d793807711 : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f1730a54f084b848da0164aeae1e39e : SUCCESS in 5m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/07d21f58354c4c76b075907b63eea626 : SUCCESS in 4m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fc73a2961ccf45c7944c33f79b71d03e : FAILURE in 8m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c4587f4c4d8f40cb80a74f33058d136b : FAILURE in 6m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c59f3e85768b40759dbb43bfe1c08189 : FAILURE in 6m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b13ad85603b743fd820704127225418d : SUCCESS in 45m 03s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/85b0f6b408554fb9800fe395da21b860 : SUCCESS in 48m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/620554da88d14687932a229c89cd7757 : SUCCESS in 1h 43m 07s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0c55a1a69db74468ad701555ebed2fd0 : SUCCESS in 1h 36m 36s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7254c2c4307448a2b6a3dd04ca80daba : SUCCESS in 2h 31m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ff7dd93bca2c40b794a88532bd31fa41 : SUCCESS in 2h 27m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/897c528692b04daeaa996401b914aef2 : FAILURE in 1h 38m 48s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/28841485b21945bfae715909b7114e8c : SUCCESS in 1h 29m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4eed6d7c24fb4709895c14e62e2f42cc : SUCCESS in 50m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6e96f64b6023477b9182f5c19e4a6fd4 : SUCCESS in 1h 27m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/969d430cd2b146e18b4d5d25b8fd82e5 : SUCCESS in 1h 17m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d8328f407655490ba0b61f118c1a4903 : FAILURE in 1h 49m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a41f9a4ee57542439e655bf89c1844a6 : SUCCESS in 1h 33m 13s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"09c2b58f796673d4334362a87e2f402bd2bd158d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-11 21:40:34.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"f1e4f30ca2fe91776aac624dbf24083fbb38b3fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 23:52:55.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e2c19dbf62c3402da616fbc8fb8916bb : SUCCESS in 3m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88f92613a00f4b1c8d5a9fa9c18bcaa3 : FAILURE in 8m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2e13ee6aecf46c39290bf5ef03a8241 : FAILURE in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc88b838466c440da56342f253ab340e : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/edc0be2fe38c4401b8b644ceb8bfa8f5 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b8599c52d064edbbaa4f67425f7adb9 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b0ed0c90ffeb4fffbdd7061753cfe992 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ada2ed46d5e84f86a9c91d57e901f408 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d25cf7e76cb4892998adcefeee27972 : SUCCESS in 4m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7f429a2080e445fd81d4bd9e69c48cfd : SUCCESS in 4m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/51b8d1775750450295920eb130976a92 : FAILURE in 7m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6f88c49cb1cb45fb86c2bb0ce9c3c8fb : FAILURE in 5m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e6abc78f5fac473dbfad79579c40b7b1 : FAILURE in 5m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/94ed47691e8340758b38a0515960cf46 : SUCCESS in 46m 57s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/91a79cbb27504d49bf133c03f0f4500f : SUCCESS in 1h 15m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ed9932a9fc6f492f95577eaca94c9228 : FAILURE in 1h 38m 33s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3c87fca265a9409ab31d5ec32a38666e : SUCCESS in 1h 57m 02s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6170778bbf5741c182171e975a6d24f5 : SUCCESS in 1h 51m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/70960aa8558b4ca08ab94a59d6b47248 : SUCCESS in 2h 03m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4f62198a4fec454f8d29fe0122026bec : SUCCESS in 1h 53m 53s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a5b1dbfded2744e5b6d1edf774861ecf : SUCCESS in 1h 32m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/456a13745d82482fb2dda909363c6038 : SUCCESS in 1h 00m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1ff7280b7fa94b17a8b29d60248f0069 : SUCCESS in 1h 29m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08ce40815a76463a835252f74ecee572 : FAILURE in 1h 37m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5b2df78d73864773ab3d196e2bcff1dd : SUCCESS in 1h 11m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/69e86b38f13d485eaa83187de8cdc0df : SUCCESS in 1h 36m 44s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"15b9de28fbed571d02782df61f4570b78cded0eb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-12 04:06:38.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"3eed9f6c5cd202a1af5f8215ef9bed43a8aa2e5b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-12 04:08:36.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"60d0b63829726095e55d723a0066493bf9d40cb5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 06:01:53.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0e36209e7d0b46739501653f6604a714 : SUCCESS in 5m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73836459da4d415e9fe26c6cf5e2ff26 : SUCCESS in 8m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8dd168571d764a1799fe3cf432cf2873 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f9d780b786e4aca96af7599a853240b : FAILURE in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d963688f4b384ca38b8c6beded29ca89 : FAILURE in 2m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e03cf673b7149b89ad0f3a602595a85 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30e597e3e7a4428f92a2c392ce8f1775 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af378d1f83cb474ab082cfa6f14f6dfa : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98d3948419b64eb6ba0b9e4c54be316f : SUCCESS in 5m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c725e588c82f489b8dfc3c744317154c : SUCCESS in 4m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/680b98771b3c41ebb401e239d96e36a8 : SUCCESS in 7m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a3b703c813014987ad799d660865449f : FAILURE in 3m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/de9d43c9b4de4b198dc09c58afdbcf28 : SUCCESS in 8m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/be1131bfe11f44bfb05ee141a743ef32 : FAILURE in 18m 47s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1ec082c6f85546c681eafd9c0c288701 : FAILURE in 14m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28fda1e533324029a6bdc106d80a4d7c : FAILURE in 18m 58s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8db0d9f57f60463387ec70f8d46cd9b4 : FAILURE in 17m 52s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/612b5a9f28e940959627e2bbec866836 : FAILURE in 19m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9e6709a595cd4e3597e8994e00729525 : SUCCESS in 1h 45m 12s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/1f50e21c21564447a44570ce909007ae : FAILURE in 48m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/272c15f999634b5e9f8364fdfbff7c5c : FAILURE in 1h 03m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/60c84a7ab47845bf98e62d6320341456 : FAILURE in 21m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3accad5b8e8244e09e22e544e8431c29 : FAILURE in 18m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4227bd2f9acf46a4a515918d0f5ea396 : FAILURE in 20m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/207af7ab81c64d9da2032b5924a041d3 : FAILURE in 13m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f51d4f3b791442b685e0c5356b7899a3 : FAILURE in 15m 55s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"da7c9ec9fa0177b7129388eaa16f71fcc693759b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-12 07:37:07.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"57368923f80907c7dbe5cd81b92722a1b3b15434","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-12 08:18:47.000000000","message":"Patch Set 18: Patch Set 17 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"b48ed7901659bd7fb791f99aa577eed021ee2667","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-12 08:19:12.000000000","message":"Patch Set 18: Backport-Candidate-2","accounts_in_message":[],"_revision_number":18},{"id":"c4f64766043588ce2b3c3d9475887bb43bbedc3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 10:54:14.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c044a9957f5f47829039f13bdc0aeeab : SUCCESS in 4m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0da53484411a435e8ab68fc7cd54f4c6 : SUCCESS in 8m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69cb18c879394a0a99deffaa6ca0d1ed : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a2e8e1849e44813aae6403a40ee27c0 : SUCCESS in 6m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ee068dce8cbf44c899529ff2dbd74470 : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae98451c6aa347a694dab62e507bd254 : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/efd9b887e319437b854242296f61b124 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9680b1558b4b463aa8766582059475e8 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e1f761f3f6c43798110e437a866ca72 : SUCCESS in 4m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f954806b2bc74498b1b64c176c1bd30b : SUCCESS in 5m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/438400748e2e4c45a412989f012a249a : SUCCESS in 6m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/adba5a7da09a46ff86761479455801d9 : TIMED_OUT in 30m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8d47a4a3946745e0a3b186fd3fec0098 : SUCCESS in 5m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1945c94b20eb4da8a692829eb95c1667 : SUCCESS in 42m 56s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9afb211cace84792b4fba898a4170fc6 : SUCCESS in 50m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/70ec6b8a3a8943e584714bc2246b4491 : FAILURE in 1h 47m 22s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/cf517960cf4546ecbacad0cda0ea7c1e : SUCCESS in 1h 42m 00s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/801f707bd9894c669e06b0ab7193c3d4 : SUCCESS in 2h 13m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e2fe23dbf2664bf286b6ec149b78a95c : SUCCESS in 1h 22m 52s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/08fcd6eaa59a408ea46242ff5283b3e8 : FAILURE in 29m 39s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e025c29916264cfebbb9e9d3c11f5238 : FAILURE in 28m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e8eebde5ac47484dab475991cf54dfdc : FAILURE in 28m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0d2b8ab0f2ff4f24b1551fba1e839fbc : FAILURE in 26m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6373b03efb914b75a7b4e860f80c02f2 : TIMED_OUT in 2h 32m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6934e45ec7d74faf871fa464c36de203 : FAILURE in 21m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a13f2592277a43d5a28eb81e2e39b1b7 : FAILURE in 1h 02m 54s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"fb5ed0bc93f9dc535ea589be35d8d30ea7a61e02","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-12 11:41:28.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"bb486f0e2ae7f25b78f27cc38b9541c3c8ad13e3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-12 15:15:08.000000000","message":"Patch Set 19: Backport-Candidate-2 Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":19},{"id":"95824a47826daebecb19f0b9eededfc07ae94114","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 15:49:05.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5b3d2f77b3414928b1b79125d8a50a99 : SUCCESS in 4m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d6578658457427b8ee2a69af7c4c0a6 : SUCCESS in 7m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7af5c476b1d4770af62f4cd829b0b6c : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcc9f69748a84e58b230470aa01f2eb5 : SUCCESS in 6m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4afc8825a6534dffae8602d58a2fdee9 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d5708f058f147c1b4730d7d3ff4568b : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86daeeb8868c479197bcc0a2d5593bbb : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa5047447a7140e4ae8da55b661276c4 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff52b876e59d4c33971b47d1e562dce6 : SUCCESS in 4m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/36dfc48ad0924977924839fd656a0fee : SUCCESS in 3m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ecea81bdf8324197800dace76a5bbfe3 : SUCCESS in 6m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3c1d04e1a34142f1a46e8c2e0f5660a3 : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5f1ab1cfacc84539a035d6237ac0f6f5 : SUCCESS in 6m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/07efef25af1d40bfbce97e7d2edc4380 : SUCCESS in 53m 01s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/2b04f91d35eb4fec93b98cae639d945b : SUCCESS in 51m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/33f37d603b3645a89a85d8fe42b6a9c2 : SUCCESS in 1h 21m 16s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/970b74543b2e4728b71377ecf87fd557 : SUCCESS in 1h 48m 26s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/84d17c29b6c14fd9b724f5bcfb0199ec : SUCCESS in 2h 16m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6be3a9645eda4b31964042432bed7b26 : FAILURE in 1h 32m 32s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0072dc79172349c7b920661feb7ff54c : FAILURE in 1h 28m 44s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e4d51a28a75c4153b72a18c830b38c5d : FAILURE in 2h 05m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4f6402a4e191443b9d345acccd540c83 : SUCCESS in 50m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c554d73095ec4d3eb3a1c3d4d32f77ad : FAILURE in 1h 36m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9709e51bd2714d8ca07ab07d7319b4fa : SUCCESS in 1h 20m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/374d150e275f4433a02d9cfc6c92c477 : SUCCESS in 1h 10m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/bc02aa2600854f99bf77158c396fc198 : FAILURE in 1h 56m 16s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"6f7273a38323c510efba167d4e2347f136faa8d8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-12 15:50:32.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"bc8940d81ce7f4d2b27bf09af068114250c5a2e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 18:11:23.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c58077306ca749f89b50919e02b60e68 : SUCCESS in 5m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3c734c5c0704328a1f7ccdef36e2b72 : SUCCESS in 8m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e2b94f504c340be91f679e87a978673 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2cfb5d586d44b6b8523b4cd1c6a45a3 : SUCCESS in 6m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6861ebadbbde4bf396209c34a805b2e8 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85c7bf7421bf4e67aab4083b8d0fdb64 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c02d342fbab24080983345cf646ad273 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e7abc97a5442fbb2ac7368238b41e9 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bffc4ff258ae474b91bc644fbd087500 : SUCCESS in 5m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/41d68dd3625149489c3d27e8f6d5d114 : SUCCESS in 4m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b4369b85717647979e505e920ddfd8f6 : SUCCESS in 7m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6c286fca09034dae8c66798caa9000dc : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b8f6cb18d3494aed8b975214a1cd932a : SUCCESS in 6m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3f01dd202dd643adbddcc257c9d5e34b : SUCCESS in 54m 19s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3be741fcf532467bb063b1d6e1d00d63 : SUCCESS in 46m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/741d6e4e0f074ab8993dd86910b11f80 : FAILURE in 34m 02s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5a18776743b142c6ab8185e5afcd1773 : SUCCESS in 1h 40m 33s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/39fcf940c63d46669d7f0124fdf57738 : SUCCESS in 2h 13m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e493b307e9f24341a20bff9360ce4f94 : SUCCESS in 2h 08m 13s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f2ada13780484cb3aff2d71a9c811893 : SUCCESS in 1h 57m 35s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d486654f34204b318af3b70f41099e5d : SUCCESS in 1h 16m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9159238edd0f48aa80cd2d491e603607 : SUCCESS in 58m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/84cfb05f4d214e00a558de86dd039851 : FAILURE in 38m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/32049d2f6e2a49ceba42e1d56be7d9cc : SUCCESS in 1h 23m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/62b4b2cc86be44b597679ac896762875 : SUCCESS in 1h 17m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a9a8c272960f4f3b99b19c9ba60be3b3 : FAILURE in 1h 17m 19s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"97eba2cf9de5ce9838b78ee1d94a7c2446905a0a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-12 19:55:16.000000000","message":"Patch Set 19:\n\nrecheck\n\nRAX mirror issues","accounts_in_message":[],"_revision_number":19},{"id":"779128cd948513d7bc9aa50e0d4255c46101c0b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 22:35:33.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/417a125ec8c04aef90a4e62b5b0fb989 : SUCCESS in 5m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85cc4c417844485fa1a0fbf329519a2d : SUCCESS in 9m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c567ca37f554f90bdd5ca89a7147224 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f08a154d92e476c896f9d0bb10f5902 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/805bfe1ef86848b5a35cab024738a4de : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d478ccc751c431694034c0d65f8ead2 : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/99f5559ad9d64564a68a6721627f50f5 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff53254078f947e686927d188fc65494 : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a23826b5e47c44d9bed5d22567169ebf : SUCCESS in 5m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c8b12404690c49f581c44446aa644a6e : SUCCESS in 4m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fb4319b1563c4679ab9cdd640f519fe6 : SUCCESS in 5m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/01390986e331439cb82edb5f25947579 : SUCCESS in 5m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d0d1c9b41f544d56b77c88ad053abf97 : SUCCESS in 5m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f204eb631d1a4123ab1075bd2e708bd6 : SUCCESS in 47m 48s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e19a15fe8e14485eb59469cbd926a574 : SUCCESS in 46m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/314a1112607e45c7a722f2eda939e205 : SUCCESS in 1h 49m 57s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/efd34abb5d52496c95d617ac2394dee1 : SUCCESS in 1h 44m 21s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/92fa2fa88cc048c0b347c5eb92b75a84 : SUCCESS in 2h 28m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9fb8734586d14c53beb56dcbaf08e574 : FAILURE in 1h 28m 23s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ee08d4ebb9624bebba402f9f9d237f7e : SUCCESS in 2h 05m 34s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/33d12ccfb4324420ba47cc1eb2396e5f : SUCCESS in 1h 18m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/50d9684020b4446682ca0efc029c5683 : SUCCESS in 1h 00m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2a82ae13e9d142eb817891abeef7d5c6 : SUCCESS in 1h 48m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2f16758927404d769ae68aa87a111011 : SUCCESS in 1h 34m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/be16ea7213eb4ce8a0b3e78827c11e6b : SUCCESS in 1h 11m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8190fee76e8a4116be833efb8ded0e58 : SUCCESS in 1h 39m 23s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"824dee2feef91cf12185c85fafad4fb09b844e00","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-13 07:50:34.000000000","message":"Patch Set 19: Code-Review-1\n\n(3 comments)\n\nCould you quickly answer my question on line 544 of listeners.py? The other two comments are just nits and don\u0027t really matter.","accounts_in_message":[],"_revision_number":19},{"id":"05d31cc306e6062166b0d0a376ca90403894b313","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-13 08:07:07.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"45281b4807a950a7acf751b594752e62303cb910","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 08:07:20.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"52cff80af6fe9f24cf09cfc55a6b7076b8ea8e7a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-13 08:09:37.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"832db701241370047480981cd683f81a2b8d05d6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-13 08:09:51.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"f75adb2ad075bcb605050c3a3b73218cb8ba17e5","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-13 08:12:01.000000000","message":"Patch Set 20: Code-Review+2","accounts_in_message":[],"_revision_number":20},{"id":"d51072390a155d7976c4af5ec34870b8bacaddeb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 10:28:50.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f79350fb99ff42caadfbcd60fd5169c0 : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6a397d50b1b48c7a5329abf2ef06ae5 : SUCCESS in 8m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e0346ea19044f28acad37d91fc948f5 : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e92bbe05c28848f5b659c3ea2926849c : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a68aa95053e54709b27e81fa190b1d58 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b442a8bfe554ef5908481d90bc6795d : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6be8eedb21f64ae4a084fdf89858f4db : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc63f2d0afa94e31b09d9717c33ae927 : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d6b7190cdf945a28e8f867457930030 : SUCCESS in 4m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4a4ec308e5554e08afe23f3172a5496e : SUCCESS in 5m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/51e9656493da4dc68cf9adbcb850e726 : SUCCESS in 6m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4f4c8b4541db445a9b85b7bddebea036 : SUCCESS in 6m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/88a63a17221f4fcfa4d9723b249ca479 : SUCCESS in 5m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/847de5ff2fa54e7b8e3f53fe92a8ec8a : SUCCESS in 50m 39s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/663219c91e934269a611450bec431b07 : SUCCESS in 43m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f3ef07abf9fa4a8eb6d7ea23bac7c93d : SUCCESS in 1h 46m 10s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5eea12e2300543dd9993154797d103be : SUCCESS in 1h 48m 39s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/eef4f3de0a54451a95e16696ff69c733 : SUCCESS in 2h 03m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c82330a93a654fd09db16b1c46ac1cf8 : FAILURE in 2h 05m 32s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/d10e84d78e4c49979c300820f0865274 : SUCCESS in 1h 30m 15s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2fcdfe4835df444daa67e6fd3585f90a : SUCCESS in 1h 24m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1697aa138ab142afab2791b04c4923ef : SUCCESS in 55m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0e921c8ea5584306b75ddf0855292ec2 : SUCCESS in 1h 42m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/780c8b18212e4da18623d53779c84025 : SUCCESS in 1h 03m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/70f30bfbf5e7466498bd431601e32bc2 : SUCCESS in 1h 10m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/251ae7f5eb354185b05fe223b79d184c : SUCCESS in 1h 29m 20s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"0e9d44dcab88cbe72b045834ebd3b60b0644c9dd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-13 15:04:54.000000000","message":"Patch Set 20: Backport-Candidate-2 Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":20},{"id":"5f96a66c57cc372e4500fb73b5d71e009d36430e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 15:05:08.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"997209c108eff985e06fae65047bd0d0edef4846","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 17:09:27.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c3e6a67a7f22412a85550ee641c790cc : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c8896d5618a48a995509ef008a65387 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad6fe0c8d3a5412694d91adb9a60f311 : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9ab0260f57474db9a32aae19be6f2449 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79bf1a0ae1834962a705a491d6aae662 : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/847b4762467e4f2da17d8895d6f5828b : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d0a926ff18949118b9e355f04600494 : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7029653b76e649ed8b37400f20808821 : SUCCESS in 4m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/55b5d91bb5174e028a478fd9918d2014 : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a051a4f56fe3418092b982986ed4c260 : SUCCESS in 5m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cc0e2b404f79487bacbfc3bca4703b70 : SUCCESS in 50m 26s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/eee49fed42234aec8e601687be1ecfae : SUCCESS in 48m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1adc1a3384d94364acf7595bc79b12ac : SUCCESS in 1h 33m 16s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a9cbaa7ff0e64810b043ef83e53def3f : SUCCESS in 1h 58m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7607a51cd0f848f3a0ab444f8c88d367 : SUCCESS in 1h 20m 24s","accounts_in_message":[],"_revision_number":20},{"id":"84bad622507c0d69d7f64beaf61fe3156acf72d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 17:09:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":20},{"id":"bacc50eec40eeadc492790a99d9ab70cc7d8d51e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 17:10:47.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/cf5a10cfb8fc4418a5459d1d90c260bd : SUCCESS in 48s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fabfe4bbc7f149dbaf424145379256eb : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ecb4b63ea5e477c92a0e48b9552372e : SUCCESS in 47s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"f3b48bc2f7547852246ba5cbac9dd42dcca06940","revisions":{"2d614ef7b40a756996da375f159e621f7ee36b42":{"kind":"REWORK","_number":1,"created":"2019-05-16 17:22:05.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/659626/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/1"}}},"commit":{"parents":[{"commit":"88c4b9bb9dd2019ba75ade3983ed4bcf75fd36d2","subject":"Remove tempest.test_plugins from setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88c4b9bb9dd2019ba75ade3983ed4bcf75fd36d2"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-16 17:21:55.000000000","tz":120},"subject":"WIP: Add VIP access control list","message":"WIP: Add VIP access control list\n\nTODO:\n- s/cidrs/allowed_cidrs/\n- PUT\n- api-ref\n- documentation\n- release note\n- unit, functional tests\n- everything else :)\n\nTask: 26210\nStory: 2003686\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d614ef7b40a756996da375f159e621f7ee36b42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d614ef7b40a756996da375f159e621f7ee36b42"}]},"branch":"refs/heads/master"},"076611a700752cf14d6de209a40edd99a8d30396":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-16 17:22:37.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/659626/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/2"}}},"commit":{"parents":[{"commit":"88c4b9bb9dd2019ba75ade3983ed4bcf75fd36d2","subject":"Remove tempest.test_plugins from setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88c4b9bb9dd2019ba75ade3983ed4bcf75fd36d2"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-16 17:22:11.000000000","tz":120},"subject":"WIP: Add VIP access control list","message":"WIP: Add VIP access control list\n\nTODO:\n- s/cidrs/allowed_cidrs/\n- PUT\n- api-ref\n- documentation\n- release note\n- unit, functional tests\n- everything else :)\n\nTask: 26210\nStory: 2003686\n\nDepends-On: https://review.opendev.org/#/c/659625/\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/076611a700752cf14d6de209a40edd99a8d30396"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/076611a700752cf14d6de209a40edd99a8d30396"}]},"branch":"refs/heads/master"},"2413505ef67330cbb74700d153274561baf08b72":{"kind":"REWORK","_number":3,"created":"2019-06-25 21:40:04.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/659626/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/3"}}},"commit":{"parents":[{"commit":"09733a0c39ad394a319f7b807518a5439ec0826e","subject":"Add bindep.txt for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09733a0c39ad394a319f7b807518a5439ec0826e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-25 21:39:52.000000000","tz":120},"subject":"WIP: Add VIP access control list","message":"WIP: Add VIP access control list\n\nTODO:\n- api-ref\n- documentation\n- release note\n- unit, functional tests\n- bump octavia-lib requirement\n- unset?\n- everything else :)\n\nTask: 26210\nStory: 2003686\n\nDepends-On: https://review.opendev.org/#/c/659625/\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2413505ef67330cbb74700d153274561baf08b72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2413505ef67330cbb74700d153274561baf08b72"}]},"branch":"refs/heads/master"},"c5290566f1cae34117297308e8ca9398bcf62786":{"kind":"REWORK","_number":4,"created":"2019-07-16 09:53:41.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/659626/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/4"}}},"commit":{"parents":[{"commit":"58a4490bbf44df30bcf1c38adfedf1923fdbaeeb","subject":"Merge \"Add warning log if auth_strategy is not keystone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58a4490bbf44df30bcf1c38adfedf1923fdbaeeb"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-16 09:53:29.000000000","tz":120},"subject":"WIP: Add VIP access control list","message":"WIP: Add VIP access control list\n\nTODO:\n- unit\n- bump octavia-lib requirement + lower constraints\n\nTask: 26210\nStory: 2003686\n\nDepends-On: https://review.opendev.org/#/c/659625/\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5290566f1cae34117297308e8ca9398bcf62786"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5290566f1cae34117297308e8ca9398bcf62786"}]},"branch":"refs/heads/master"},"e7edc11d0ac25772e7a3e8ba63fb6bdbbac389c7":{"kind":"REWORK","_number":5,"created":"2019-08-16 16:35:46.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/659626/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/5"}}},"commit":{"parents":[{"commit":"3b5a19c386bdd0cd42bcbfff167d1747cfcf1087","subject":"Standardizes terminology in Listener documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b5a19c386bdd0cd42bcbfff167d1747cfcf1087"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-16 16:32:39.000000000","tz":120},"subject":"WIP: Add VIP access control list","message":"WIP: Add VIP access control list\n\nTODO:\n- unit\n\nTask: 26210\nStory: 2003686\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e7edc11d0ac25772e7a3e8ba63fb6bdbbac389c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e7edc11d0ac25772e7a3e8ba63fb6bdbbac389c7"}]},"branch":"refs/heads/master"},"d6e4d6aaab1aad56b5b471ee5594cfc0cfc87bc4":{"kind":"REWORK","_number":6,"created":"2019-08-16 16:37:57.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/659626/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/6"}}},"commit":{"parents":[{"commit":"3b5a19c386bdd0cd42bcbfff167d1747cfcf1087","subject":"Standardizes terminology in Listener documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b5a19c386bdd0cd42bcbfff167d1747cfcf1087"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-16 16:37:48.000000000","tz":120},"subject":"WIP: Add VIP access control list","message":"WIP: Add VIP access control list\n\nTODO:\n- unit\n\nTask: 26210\nStory: 2003686\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6e4d6aaab1aad56b5b471ee5594cfc0cfc87bc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6e4d6aaab1aad56b5b471ee5594cfc0cfc87bc4"}]},"branch":"refs/heads/master"},"1946bd182008eead62d9786bc9f52fc6fbd24f47":{"kind":"REWORK","_number":7,"created":"2019-08-20 20:54:56.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/659626/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/7"}}},"commit":{"parents":[{"commit":"3b5a19c386bdd0cd42bcbfff167d1747cfcf1087","subject":"Standardizes terminology in Listener documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b5a19c386bdd0cd42bcbfff167d1747cfcf1087"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-20 20:53:23.000000000","tz":120},"subject":"Add VIP access control list","message":"Add VIP access control list\n\nTask: 26210\nStory: 2003686\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1946bd182008eead62d9786bc9f52fc6fbd24f47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1946bd182008eead62d9786bc9f52fc6fbd24f47"}]},"branch":"refs/heads/master"},"6455ac07d2134e1035fd1ccb7ddf9fb1f132c84a":{"kind":"REWORK","_number":8,"created":"2019-08-26 13:03:02.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/659626/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/8"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-26 12:57:39.000000000","tz":120},"subject":"Add VIP access control list","message":"Add VIP access control list\n\nThis patch extends the listener API to include the new parameter\n\u0027allowed_cidrs\u0027. This parameter is a list of IPv4 or IPv6 CIDRs. Leaving\nthis list unset defaults to the traditional behavior of allowing all\ningress traffic to the listener. Setting it will deny all traffic but\nall CIDRs set in the \u0027allowed_cidrs\u0027 list.\n\nNote that the API will validate that all CIDRs match the same IP version\nof the VIP. This may change later as part of work to allow multiple VIPs\nper LB (Change-Id Id7153dbf33b9616d7af685fcf13ad9a79793c06b).\n\nTask: 26210\nStory: 2003686\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6455ac07d2134e1035fd1ccb7ddf9fb1f132c84a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6455ac07d2134e1035fd1ccb7ddf9fb1f132c84a"}]},"branch":"refs/heads/master"},"fcf20014f4821fcc63ed8ec8fc12668ebd48725e":{"kind":"REWORK","_number":9,"created":"2019-08-26 13:31:23.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/659626/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/9"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-26 13:31:13.000000000","tz":120},"subject":"Add VIP access control list","message":"Add VIP access control list\n\nThis patch extends the listener API to include the new parameter\n\u0027allowed_cidrs\u0027. This parameter is a list of IPv4 or IPv6 CIDRs. Leaving\nthis list unset defaults to the traditional behavior of allowing all\ningress traffic to the listener. Setting it will deny all traffic but\nall CIDRs set in the \u0027allowed_cidrs\u0027 list.\n\nNote that the API will validate that all CIDRs match the same IP version\nof the VIP. This may change later as part of work to allow multiple VIPs\nper LB (Change-Id Id7153dbf33b9616d7af685fcf13ad9a79793c06b).\n\nTask: 26210\nStory: 2003686\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fcf20014f4821fcc63ed8ec8fc12668ebd48725e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fcf20014f4821fcc63ed8ec8fc12668ebd48725e"}]},"branch":"refs/heads/master"},"fa4e8738e0ce086caf0030e9371c007f4e69363d":{"kind":"REWORK","_number":10,"created":"2019-08-26 13:49:27.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/659626/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/10"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-26 13:46:20.000000000","tz":120},"subject":"Add VIP access control list","message":"Add VIP access control list\n\nThis patch extends the listener API to include the new parameter\n\u0027allowed_cidrs\u0027. This parameter is a list of IPv4 or IPv6 CIDRs. Leaving\nthis list unset defaults to the traditional behavior of allowing all\ningress traffic to the listener. Setting it will deny all traffic but\nall CIDRs set in the \u0027allowed_cidrs\u0027 list.\n\nNote that the API will validate that all CIDRs match the same IP version\nof the VIP. This may change later as part of work to allow multiple VIPs\nper LB (Change-Id Id7153dbf33b9616d7af685fcf13ad9a79793c06b).\n\nTask: 26210\nStory: 2003686\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa4e8738e0ce086caf0030e9371c007f4e69363d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa4e8738e0ce086caf0030e9371c007f4e69363d"}]},"branch":"refs/heads/master"},"dc18a2a17e512411716492a0596a427bd8cf2cfb":{"kind":"REWORK","_number":11,"created":"2019-08-27 09:05:14.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/659626/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/11"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-27 09:04:56.000000000","tz":120},"subject":"Add VIP access control list","message":"Add VIP access control list\n\nThis patch extends the listener API to include the new parameter\n\u0027allowed_cidrs\u0027. This parameter is a list of IPv4 or IPv6 CIDRs. Leaving\nthis list unset defaults to the traditional behavior of allowing all\ningress traffic to the listener. Setting it will deny all traffic but\nall CIDRs set in the \u0027allowed_cidrs\u0027 list.\n\nNote that the API will validate that all CIDRs match the same IP version\nof the VIP. This may change later as part of work to allow multiple VIPs\nper LB (Change-Id Id7153dbf33b9616d7af685fcf13ad9a79793c06b).\n\nTask: 26210\nStory: 2003686\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dc18a2a17e512411716492a0596a427bd8cf2cfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dc18a2a17e512411716492a0596a427bd8cf2cfb"}]},"branch":"refs/heads/master"},"97cc7b530c39d5d9a19fba8a01edbc00507c0c83":{"kind":"REWORK","_number":12,"created":"2019-08-27 16:01:50.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/659626/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/12"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-27 16:01:42.000000000","tz":120},"subject":"Add VIP access control list","message":"Add VIP access control list\n\nThis patch extends the listener API to include the new parameter\n\u0027allowed_cidrs\u0027. This parameter is a list of IPv4 or IPv6 CIDRs. Leaving\nthis list unset defaults to the traditional behavior of allowing all\ningress traffic to the listener. Setting it will deny all traffic but\nall CIDRs set in the \u0027allowed_cidrs\u0027 list.\n\nNote that the API will validate that all CIDRs match the same IP version\nof the VIP. This may change later as part of work to allow multiple VIPs\nper LB (Change-Id Id7153dbf33b9616d7af685fcf13ad9a79793c06b).\n\nTask: 26210\nStory: 2003686\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97cc7b530c39d5d9a19fba8a01edbc00507c0c83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97cc7b530c39d5d9a19fba8a01edbc00507c0c83"}]},"branch":"refs/heads/master"},"9f4ff7c030eead296c6a3576b8d405d61a2b3023":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-09-10 10:53:50.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/659626/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/13"}}},"commit":{"parents":[{"commit":"d668122e8671d14fb1c59b6f73db1893bf993b49","subject":"Remove mock patching of jsonschema","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d668122e8671d14fb1c59b6f73db1893bf993b49"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-10 10:53:50.000000000","tz":0},"subject":"Add VIP access control list","message":"Add VIP access control list\n\nThis patch extends the listener API to include the new parameter\n\u0027allowed_cidrs\u0027. This parameter is a list of IPv4 or IPv6 CIDRs. Leaving\nthis list unset defaults to the traditional behavior of allowing all\ningress traffic to the listener. Setting it will deny all traffic but\nall CIDRs set in the \u0027allowed_cidrs\u0027 list.\n\nNote that the API will validate that all CIDRs match the same IP version\nof the VIP. This may change later as part of work to allow multiple VIPs\nper LB (Change-Id Id7153dbf33b9616d7af685fcf13ad9a79793c06b).\n\nTask: 26210\nStory: 2003686\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f4ff7c030eead296c6a3576b8d405d61a2b3023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f4ff7c030eead296c6a3576b8d405d61a2b3023"}]},"branch":"refs/heads/master"},"722424135bb881f862ed956cd6f7cf22e1c2aeed":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-09-10 19:53:03.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/659626/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/14"}}},"commit":{"parents":[{"commit":"09efc2a423c6ab033f9a648549aff87590be5990","subject":"Add get method support to the driver-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09efc2a423c6ab033f9a648549aff87590be5990"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-10 19:53:03.000000000","tz":0},"subject":"Add VIP access control list","message":"Add VIP access control list\n\nThis patch extends the listener API to include the new parameter\n\u0027allowed_cidrs\u0027. This parameter is a list of IPv4 or IPv6 CIDRs. Leaving\nthis list unset defaults to the traditional behavior of allowing all\ningress traffic to the listener. Setting it will deny all traffic but\nall CIDRs set in the \u0027allowed_cidrs\u0027 list.\n\nNote that the API will validate that all CIDRs match the same IP version\nof the VIP. This may change later as part of work to allow multiple VIPs\nper LB (Change-Id Id7153dbf33b9616d7af685fcf13ad9a79793c06b).\n\nTask: 26210\nStory: 2003686\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/722424135bb881f862ed956cd6f7cf22e1c2aeed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/722424135bb881f862ed956cd6f7cf22e1c2aeed"}]},"branch":"refs/heads/master"},"dc6feb72d3b996253baadf8f5e4725902c3a6dec":{"kind":"REWORK","_number":15,"created":"2019-09-11 21:40:34.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/659626/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/15"}}},"commit":{"parents":[{"commit":"510525a92b6c6335f9ca28b0361e138edb52cd19","subject":"Merge \"Add `additive_only` parameter to Batch Member call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/510525a92b6c6335f9ca28b0361e138edb52cd19"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-11 21:37:21.000000000","tz":120},"subject":"Add VIP access control list","message":"Add VIP access control list\n\nThis patch extends the listener API to include the new parameter\n\u0027allowed_cidrs\u0027. This parameter is a list of IPv4 or IPv6 CIDRs. Leaving\nthis list unset defaults to the traditional behavior of allowing all\ningress traffic to the listener. Setting it will deny all traffic but\nall CIDRs set in the \u0027allowed_cidrs\u0027 list.\n\nNote that the API will validate that all CIDRs match the same IP version\nof the VIP. This may change later as part of work to allow multiple VIPs\nper LB (Change-Id Id7153dbf33b9616d7af685fcf13ad9a79793c06b).\n\nTask: 26210\nStory: 2003686\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dc6feb72d3b996253baadf8f5e4725902c3a6dec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dc6feb72d3b996253baadf8f5e4725902c3a6dec"}]},"branch":"refs/heads/master"},"9ba7bb9fa21bd8d4416842311426ff8d37421bb7":{"kind":"REWORK","_number":16,"created":"2019-09-12 04:06:38.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/659626/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/16"}}},"commit":{"parents":[{"commit":"510525a92b6c6335f9ca28b0361e138edb52cd19","subject":"Merge \"Add `additive_only` parameter to Batch Member call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/510525a92b6c6335f9ca28b0361e138edb52cd19"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-12 04:06:37.000000000","tz":-420},"subject":"Add VIP access control list","message":"Add VIP access control list\n\nThis patch extends the listener API to include the new parameter\n\u0027allowed_cidrs\u0027. This parameter is a list of IPv4 or IPv6 CIDRs. Leaving\nthis list unset defaults to the traditional behavior of allowing all\ningress traffic to the listener. Setting it will deny all traffic but\nall CIDRs set in the \u0027allowed_cidrs\u0027 list.\n\nNote that the API will validate that all CIDRs match the same IP version\nof the VIP. This may change later as part of work to allow multiple VIPs\nper LB (Change-Id Id7153dbf33b9616d7af685fcf13ad9a79793c06b).\n\nTask: 26210\nStory: 2003686\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ba7bb9fa21bd8d4416842311426ff8d37421bb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ba7bb9fa21bd8d4416842311426ff8d37421bb7"}]},"branch":"refs/heads/master"},"74cc78d2d7ba6de6a3e4141d7449647f22f5dd5b":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2019-09-12 04:08:36.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/659626/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/17"}}},"commit":{"parents":[{"commit":"afe648caef8fad86f4cb09e2292fab9a0ea669af","subject":"Fix a few nits from the main volume-based patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/afe648caef8fad86f4cb09e2292fab9a0ea669af"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-12 04:08:36.000000000","tz":0},"subject":"Add VIP access control list","message":"Add VIP access control list\n\nThis patch extends the listener API to include the new parameter\n\u0027allowed_cidrs\u0027. This parameter is a list of IPv4 or IPv6 CIDRs. Leaving\nthis list unset defaults to the traditional behavior of allowing all\ningress traffic to the listener. Setting it will deny all traffic but\nall CIDRs set in the \u0027allowed_cidrs\u0027 list.\n\nNote that the API will validate that all CIDRs match the same IP version\nof the VIP. This may change later as part of work to allow multiple VIPs\nper LB (Change-Id Id7153dbf33b9616d7af685fcf13ad9a79793c06b).\n\nTask: 26210\nStory: 2003686\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/74cc78d2d7ba6de6a3e4141d7449647f22f5dd5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/74cc78d2d7ba6de6a3e4141d7449647f22f5dd5b"}]},"branch":"refs/heads/master"},"93246428433c1e30e1b322cb3f1b0973d8ed5b1b":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2019-09-12 08:18:47.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/659626/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/18"}}},"commit":{"parents":[{"commit":"e189acdfc48387c1cd84dfe6241ec22aade935f7","subject":"Fix a few nits from the main volume-based patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e189acdfc48387c1cd84dfe6241ec22aade935f7"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-12 08:18:47.000000000","tz":0},"subject":"Add VIP access control list","message":"Add VIP access control list\n\nThis patch extends the listener API to include the new parameter\n\u0027allowed_cidrs\u0027. This parameter is a list of IPv4 or IPv6 CIDRs. Leaving\nthis list unset defaults to the traditional behavior of allowing all\ningress traffic to the listener. Setting it will deny all traffic but\nall CIDRs set in the \u0027allowed_cidrs\u0027 list.\n\nNote that the API will validate that all CIDRs match the same IP version\nof the VIP. This may change later as part of work to allow multiple VIPs\nper LB (Change-Id Id7153dbf33b9616d7af685fcf13ad9a79793c06b).\n\nTask: 26210\nStory: 2003686\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93246428433c1e30e1b322cb3f1b0973d8ed5b1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93246428433c1e30e1b322cb3f1b0973d8ed5b1b"}]},"branch":"refs/heads/master"},"d14f85eaac742ca35f85cb03c20ae0430f09484a":{"kind":"NO_CHANGE","_number":19,"created":"2019-09-12 11:41:28.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/659626/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/19"}}},"commit":{"parents":[{"commit":"05f5257075dda076b0960a319d49c970b0015022","subject":"Fix a few nits from the main volume-based patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05f5257075dda076b0960a319d49c970b0015022"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-12 11:41:28.000000000","tz":0},"subject":"Add VIP access control list","message":"Add VIP access control list\n\nThis patch extends the listener API to include the new parameter\n\u0027allowed_cidrs\u0027. This parameter is a list of IPv4 or IPv6 CIDRs. Leaving\nthis list unset defaults to the traditional behavior of allowing all\ningress traffic to the listener. Setting it will deny all traffic but\nall CIDRs set in the \u0027allowed_cidrs\u0027 list.\n\nNote that the API will validate that all CIDRs match the same IP version\nof the VIP. This may change later as part of work to allow multiple VIPs\nper LB (Change-Id Id7153dbf33b9616d7af685fcf13ad9a79793c06b).\n\nTask: 26210\nStory: 2003686\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d14f85eaac742ca35f85cb03c20ae0430f09484a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d14f85eaac742ca35f85cb03c20ae0430f09484a"}]},"branch":"refs/heads/master"},"f3b48bc2f7547852246ba5cbac9dd42dcca06940":{"kind":"REWORK","_number":20,"created":"2019-09-13 08:09:37.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/659626/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/659626/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/659626/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/659626/20"}}},"commit":{"parents":[{"commit":"05f5257075dda076b0960a319d49c970b0015022","subject":"Fix a few nits from the main volume-based patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05f5257075dda076b0960a319d49c970b0015022"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-11-23 11:02:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-13 08:09:25.000000000","tz":120},"subject":"Add VIP access control list","message":"Add VIP access control list\n\nThis patch extends the listener API to include the new parameter\n\u0027allowed_cidrs\u0027. This parameter is a list of IPv4 or IPv6 CIDRs. Leaving\nthis list unset defaults to the traditional behavior of allowing all\ningress traffic to the listener. Setting it will deny all traffic but\nall CIDRs set in the \u0027allowed_cidrs\u0027 list.\n\nNote that the API will validate that all CIDRs match the same IP version\nof the VIP. This may change later as part of work to allow multiple VIPs\nper LB (Change-Id Id7153dbf33b9616d7af685fcf13ad9a79793c06b).\n\nTask: 26210\nStory: 2003686\n\nChange-Id: Id2b560df1cde9ce9403afbd593bbaa6cae5f06d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f3b48bc2f7547852246ba5cbac9dd42dcca06940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f3b48bc2f7547852246ba5cbac9dd42dcca06940"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
