)]}'
{"id":"openstack%2Foctavia~659677","triplet_id":"openstack%2Foctavia~master~I5e8cbe1f4eafd18e67b5b6914f5d459fb157a3a1","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I5e8cbe1f4eafd18e67b5b6914f5d459fb157a3a1","subject":"Enable fail-fast on the gate queue","status":"MERGED","created":"2019-05-16 22:05:21.000000000","updated":"2019-05-29 02:53:29.000000000","submitted":"2019-05-29 02:53:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659677-1559098409952-660a502c","meta_rev_id":"587e2a358dd247f3736f95f6b7cdf0a9cbad59a1","_number":659677,"virtual_id_number":659677,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-05-29 02:53:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-05-28 21:05:14.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-05-29 00:56:31.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-05-29 00:56:31.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-28 21:32:42.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-28 21:05:14.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-05-29 00:56:31.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-05-29 02:53:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"001f03ae39e8b3e1a14d529a156dba904c4d2f37","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-16 22:05:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc09c110e5ee961d9944a7339426fb9e3bbf8a2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 22:05:36.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  extra keys not allowed @ data[\u0027gate\u0027][\u0027fail-fast\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - periodic-stable-jobs-neutron\n        - openstack-cover-jobs\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python36-jobs\n        - openstack-python37-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-functional:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-noop-py2-api\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-py2-scenario\n          - octavia-v2-dsvm-py2-scenario-centos-7:\n              voting: false\n          - octavia-v2-dsvm-scenario-ubuntu-bionic:\n              voting: false\n          - octavia-v2-dsvm-scenario-two-node:\n              voting: false\n          - octavia-v2-dsvm-py2-scenario-two-node:\n              voting: false\n          - octavia-grenade:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tools/.*$\n                - ^(test-|)requirements.txt$\n                - ^tox.ini$\n          - octavia-v2-dsvm-tls-barbican:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n              voting: false\n      gate:\n        fail-fast: true\n        queue: octavia\n        jobs:\n          - openstack-tox-functional:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-noop-py2-api\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-py2-scenario\n          - octavia-grenade:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tools/.*$\n                - ^(test-|)requirements.txt$\n                - ^tox.ini$\n      periodic:\n        jobs:\n          - publish-openstack-octavia-amphora-image-xenial:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-bionic:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos7:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-fedora28:\n              branches: ^(?!stable/.*).*$\n\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":1},{"id":"d79dd2a5b83f4974f8eb0e9457276952dc5bb569","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-16 22:07:21.000000000","message":"Patch Set 1: Workflow-1\n\nWell, I guess this isn\u0027t deployed for OpenStack yet.","accounts_in_message":[],"_revision_number":1},{"id":"01f3b69c73b9c8d64d4d379779a8146cd281e1d4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-16 22:07:51.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7ee92481e622084ea8c9cbe6d7109ebb25fa8108","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-16 22:07:57.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"389bf704a7e3c7a48de8c8b59db31c803a205104","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 22:08:05.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  extra keys not allowed @ data[\u0027gate\u0027][\u0027fail-fast\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - periodic-stable-jobs-neutron\n        - openstack-cover-jobs\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python36-jobs\n        - openstack-python37-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-functional:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-noop-py2-api\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-py2-scenario\n          - octavia-v2-dsvm-py2-scenario-centos-7:\n              voting: false\n          - octavia-v2-dsvm-scenario-ubuntu-bionic:\n              voting: false\n          - octavia-v2-dsvm-scenario-two-node:\n              voting: false\n          - octavia-v2-dsvm-py2-scenario-two-node:\n              voting: false\n          - octavia-grenade:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tools/.*$\n                - ^(test-|)requirements.txt$\n                - ^tox.ini$\n          - octavia-v2-dsvm-tls-barbican:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n              voting: false\n      gate:\n        fail-fast: true\n        queue: octavia\n        jobs:\n          - openstack-tox-functional:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-noop-py2-api\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-py2-scenario\n          - octavia-grenade:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tools/.*$\n                - ^(test-|)requirements.txt$\n                - ^tox.ini$\n      periodic:\n        jobs:\n          - publish-openstack-octavia-amphora-image-xenial:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-bionic:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos7:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-fedora28:\n              branches: ^(?!stable/.*).*$\n\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":2},{"id":"1b0c03dc70e7168820fbaa3bfd3059a3322f3c51","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-21 16:25:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"35b24af64da9967d01acbe94caa815b44ca0a1b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 16:25:56.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  extra keys not allowed @ data[\u0027gate\u0027][\u0027fail-fast\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - periodic-stable-jobs-neutron\n        - openstack-cover-jobs\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python36-jobs\n        - openstack-python37-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-functional:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-noop-py2-api\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-py2-scenario\n          - octavia-v2-dsvm-py2-scenario-centos-7:\n              voting: false\n          - octavia-v2-dsvm-scenario-ubuntu-bionic:\n              voting: false\n          - octavia-v2-dsvm-scenario-two-node:\n              voting: false\n          - octavia-v2-dsvm-py2-scenario-two-node:\n              voting: false\n          - octavia-grenade:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tools/.*$\n                - ^(test-|)requirements.txt$\n                - ^tox.ini$\n          - octavia-v2-dsvm-tls-barbican:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n              voting: false\n      gate:\n        fail-fast: true\n        queue: octavia\n        jobs:\n          - openstack-tox-functional:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-noop-py2-api\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-py2-scenario\n          - octavia-grenade:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tools/.*$\n                - ^(test-|)requirements.txt$\n                - ^tox.ini$\n      periodic:\n        jobs:\n          - publish-openstack-octavia-amphora-image-xenial:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-bionic:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos7:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-fedora28:\n              branches: ^(?!stable/.*).*$\n\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":2},{"id":"3dc42b14674ec93cdda79dfcd3049d93ddecc082","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-28 20:08:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"655356d3729137ed99c7be1130c9e6f4f6969a5b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-28 21:05:14.000000000","message":"Patch Set 2: Code-Review+2\n\nZuul parses fail-fast now. LGTM.","accounts_in_message":[],"_revision_number":2},{"id":"783e832fb2ad3e86e8cebfa4726c7f597358f86d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-28 21:32:42.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"5e914a72ebbb43d47de8718ae9d84a6fd2b3b63b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 22:32:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/77/659677/2/check/openstack-tox-cover/974b6ee/cover/ : SUCCESS in 8m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/659677/2/check/openstack-tox-lower-constraints/74724c9/ : SUCCESS in 6m 34s\n- openstack-tox-pep8 http://logs.openstack.org/77/659677/2/check/openstack-tox-pep8/036c007/ : SUCCESS in 6m 12s\n- openstack-tox-py27 http://logs.openstack.org/77/659677/2/check/openstack-tox-py27/0e91907/ : SUCCESS in 6m 05s\n- openstack-tox-py36 http://logs.openstack.org/77/659677/2/check/openstack-tox-py36/76e8e82/ : SUCCESS in 6m 53s\n- openstack-tox-py37 http://logs.openstack.org/77/659677/2/check/openstack-tox-py37/2daf50f/ : SUCCESS in 7m 37s\n- openstack-tox-docs http://logs.openstack.org/77/659677/2/check/openstack-tox-docs/e63ba52/html/ : SUCCESS in 7m 02s\n- openstack-tox-functional http://logs.openstack.org/77/659677/2/check/openstack-tox-functional/f99eb27/ : SUCCESS in 8m 16s\n- openstack-tox-functional-py36 http://logs.openstack.org/77/659677/2/check/openstack-tox-functional-py36/b492e3d/ : SUCCESS in 6m 45s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/77/659677/2/check/octavia-v2-dsvm-noop-api/a5f9ad9/ : SUCCESS in 45m 45s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/77/659677/2/check/octavia-v2-dsvm-noop-py2-api/64f68e8/ : SUCCESS in 39m 33s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/77/659677/2/check/octavia-v2-dsvm-scenario/2400329/ : SUCCESS in 1h 34m 52s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/77/659677/2/check/octavia-v2-dsvm-py2-scenario/ac25a3d/ : SUCCESS in 1h 39m 44s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/77/659677/2/check/octavia-v2-dsvm-py2-scenario-centos-7/d77e3fc/ : FAILURE in 2h 12m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/77/659677/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/a578684/ : SUCCESS in 1h 49m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/77/659677/2/check/octavia-v2-dsvm-scenario-two-node/1916a75/ : SUCCESS in 1h 41m 17s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/77/659677/2/check/octavia-v2-dsvm-py2-scenario-two-node/fe40fac/ : SUCCESS in 1h 43m 50s (non-voting)\n- octavia-grenade http://logs.openstack.org/77/659677/2/check/octavia-grenade/b482097/ : SUCCESS in 1h 10m 39s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/77/659677/2/check/octavia-v2-dsvm-tls-barbican/94b9f76/ : SUCCESS in 52m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e6d6f0a8aae27251d162d5869e2899849c6b2d84","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-05-29 00:56:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c6d9f8b2fe22e4d04889e63af08b938f5f3a3a3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 00:56:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1b847648c7df65876165023889b8a761d28877a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 02:53:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"587e2a358dd247f3736f95f6b7cdf0a9cbad59a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 02:53:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/77/659677/2/gate/openstack-tox-lower-constraints/44d2e76/ : SUCCESS in 7m 16s\n- openstack-tox-pep8 http://logs.openstack.org/77/659677/2/gate/openstack-tox-pep8/130a0ad/ : SUCCESS in 6m 33s\n- openstack-tox-py27 http://logs.openstack.org/77/659677/2/gate/openstack-tox-py27/f94baee/ : SUCCESS in 5m 18s\n- openstack-tox-py36 http://logs.openstack.org/77/659677/2/gate/openstack-tox-py36/3a8dab6/ : SUCCESS in 5m 35s\n- openstack-tox-py37 http://logs.openstack.org/77/659677/2/gate/openstack-tox-py37/d7814d6/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/77/659677/2/gate/openstack-tox-docs/4d27c36/html/ : SUCCESS in 6m 17s\n- openstack-tox-functional http://logs.openstack.org/77/659677/2/gate/openstack-tox-functional/b2bd68c/ : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 http://logs.openstack.org/77/659677/2/gate/openstack-tox-functional-py36/7a07b56/ : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/77/659677/2/gate/octavia-v2-dsvm-noop-api/6e55b23/ : SUCCESS in 44m 20s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/77/659677/2/gate/octavia-v2-dsvm-noop-py2-api/ed23077/ : SUCCESS in 48m 17s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/77/659677/2/gate/octavia-v2-dsvm-scenario/f8ad443/ : SUCCESS in 1h 38m 02s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/77/659677/2/gate/octavia-v2-dsvm-py2-scenario/c4cf252/ : SUCCESS in 1h 40m 52s\n- octavia-grenade http://logs.openstack.org/77/659677/2/gate/octavia-grenade/71535f3/ : SUCCESS in 1h 11m 27s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9ecc7abfc53417e8e65b1115102e36af61179698","revisions":{"4c596e7de1c37bab71f6cd2c26bcbed7aaf20bcf":{"kind":"REWORK","_number":1,"created":"2019-05-16 22:05:21.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/77/659677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/659677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/659677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/659677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/659677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/659677/1"}}},"commit":{"parents":[{"commit":"88c4b9bb9dd2019ba75ade3983ed4bcf75fd36d2","subject":"Remove tempest.test_plugins from setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88c4b9bb9dd2019ba75ade3983ed4bcf75fd36d2"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-16 22:01:55.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-16 22:05:20.000000000","tz":-420},"subject":"Enable fail-fast on the gate queue","message":"Enable fail-fast on the gate queue\n\nZuul recently added fail-fast[1] as a new option for project\npipelines. With this set, if any one of the voting jobs fails, zuul\nwill report a build failure immediately and abort the remaining\njobs.\n\nThis patch enables fail-fast on the gate pipeline as any one\nfailing job means the patch will not merge.\n\nChange-Id: I5e8cbe1f4eafd18e67b5b6914f5d459fb157a3a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c596e7de1c37bab71f6cd2c26bcbed7aaf20bcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c596e7de1c37bab71f6cd2c26bcbed7aaf20bcf"}]},"branch":"refs/heads/master"},"9ecc7abfc53417e8e65b1115102e36af61179698":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-16 22:07:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/77/659677/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/659677/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/659677/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/659677/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/659677/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/659677/2"}}},"commit":{"parents":[{"commit":"88c4b9bb9dd2019ba75ade3983ed4bcf75fd36d2","subject":"Remove tempest.test_plugins from setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88c4b9bb9dd2019ba75ade3983ed4bcf75fd36d2"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-16 22:01:55.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-16 22:07:49.000000000","tz":0},"subject":"Enable fail-fast on the gate queue","message":"Enable fail-fast on the gate queue\n\nZuul recently added fail-fast[1] as a new option for project\npipelines. With this set, if any one of the voting jobs fails, zuul\nwill report a build failure immediately and abort the remaining\njobs.\n\nThis patch enables fail-fast on the gate pipeline as any one\nfailing job means the patch will not merge.\n\n[1] https://review.opendev.org/652764\n\nChange-Id: I5e8cbe1f4eafd18e67b5b6914f5d459fb157a3a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ecc7abfc53417e8e65b1115102e36af61179698"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ecc7abfc53417e8e65b1115102e36af61179698"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
