)]}'
{"id":"openstack%2Foctavia~660227","triplet_id":"openstack%2Foctavia~master~I196b21460789b4295ea2dc4ac297b218e26a1e43","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I196b21460789b4295ea2dc4ac297b218e26a1e43","subject":"Fix tox for functional py36 and py37","status":"MERGED","created":"2019-05-20 23:12:00.000000000","updated":"2019-05-23 15:54:21.000000000","submitted":"2019-05-23 15:54:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660227-1558626861293-80094cdc","meta_rev_id":"f0021d8678d44efc8ae31c5f44e6918e1ad956cf","_number":660227,"virtual_id_number":660227,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"date":"2019-05-23 05:56:25.000000000","_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"date":"2019-05-23 13:54:53.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-05-23 15:54:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"recommended":{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},"all":[{"value":1,"date":"2019-05-23 06:11:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"value":2,"date":"2019-05-23 06:07:27.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-05-23 14:03:15.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-05-23 14:03:15.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 05:56:25.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-05-23 06:07:27.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-05-23 06:11:42.000000000","updated_by":{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},"reviewer":{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},"state":"REVIEWER"},{"updated":"2019-05-23 14:03:15.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-05-23 15:54:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a37921ef5877657003cc3f4b2534ae7ee6896125","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-20 23:12:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e8753cd4f5bf36dd70cabffd3806c3fc5a814bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 01:26:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/27/660227/1/check/openstack-tox-cover/d0a2ee1/cover/ : SUCCESS in 7m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/660227/1/check/openstack-tox-lower-constraints/1451edb/ : SUCCESS in 6m 44s\n- openstack-tox-pep8 http://logs.openstack.org/27/660227/1/check/openstack-tox-pep8/e68fa59/ : SUCCESS in 6m 22s\n- openstack-tox-py27 http://logs.openstack.org/27/660227/1/check/openstack-tox-py27/1b4e85e/ : SUCCESS in 5m 35s\n- openstack-tox-py36 http://logs.openstack.org/27/660227/1/check/openstack-tox-py36/e076f64/ : SUCCESS in 5m 24s\n- openstack-tox-py37 http://logs.openstack.org/27/660227/1/check/openstack-tox-py37/fd8da36/ : SUCCESS in 5m 28s\n- openstack-tox-docs http://logs.openstack.org/27/660227/1/check/openstack-tox-docs/dea1165/html/ : SUCCESS in 6m 03s\n- build-openstack-releasenotes http://logs.openstack.org/27/660227/1/check/build-openstack-releasenotes/272cf05/html/ : SUCCESS in 6m 03s\n- openstack-tox-functional http://logs.openstack.org/27/660227/1/check/openstack-tox-functional/b5f9466/ : SUCCESS in 10m 21s\n- openstack-tox-functional-py36 http://logs.openstack.org/27/660227/1/check/openstack-tox-functional-py36/79274b7/ : SUCCESS in 9m 57s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-noop-api/116e4dd/ : SUCCESS in 54m 45s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-noop-py2-api/5b32c87/ : SUCCESS in 40m 30s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-scenario/cbb5be0/ : SUCCESS in 1h 38m 19s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-py2-scenario/5061924/ : FAILURE in 1h 51m 58s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-py2-scenario-centos-7/4154dc1/ : FAILURE in 2h 06m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/2515fbb/ : FAILURE in 1h 51m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-scenario-two-node/69b2558/ : SUCCESS in 1h 53m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-py2-scenario-two-node/9553c51/ : FAILURE in 1h 24m 30s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-tls-barbican/5604c11/ : SUCCESS in 55m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3eff43b5f9f6a4777091d799086d3fc0311e159","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-21 15:12:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"64806625789d1519068a83db2b377fc249510717","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 17:26:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/27/660227/1/check/openstack-tox-cover/9dddd9b/cover/ : SUCCESS in 8m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/660227/1/check/openstack-tox-lower-constraints/c8e93e0/ : SUCCESS in 6m 52s\n- openstack-tox-pep8 http://logs.openstack.org/27/660227/1/check/openstack-tox-pep8/5573e34/ : SUCCESS in 6m 59s\n- openstack-tox-py27 http://logs.openstack.org/27/660227/1/check/openstack-tox-py27/78d1825/ : SUCCESS in 5m 20s\n- openstack-tox-py36 http://logs.openstack.org/27/660227/1/check/openstack-tox-py36/dfff3a8/ : SUCCESS in 5m 19s\n- openstack-tox-py37 http://logs.openstack.org/27/660227/1/check/openstack-tox-py37/9a0c362/ : SUCCESS in 5m 51s\n- openstack-tox-docs http://logs.openstack.org/27/660227/1/check/openstack-tox-docs/cc17698/html/ : SUCCESS in 5m 48s\n- build-openstack-releasenotes http://logs.openstack.org/27/660227/1/check/build-openstack-releasenotes/b04c85a/html/ : SUCCESS in 5m 15s\n- openstack-tox-functional http://logs.openstack.org/27/660227/1/check/openstack-tox-functional/41484db/ : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 http://logs.openstack.org/27/660227/1/check/openstack-tox-functional-py36/8962933/ : SUCCESS in 6m 38s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-noop-api/2db426d/ : SUCCESS in 42m 14s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-noop-py2-api/26c464e/ : POST_FAILURE in 14m 34s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-scenario/1cc6a73/ : FAILURE in 31m 58s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-py2-scenario/b8cc0a6/ : FAILURE in 31m 17s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-py2-scenario-centos-7/ddb8e0a/ : FAILURE in 2h 01m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/73f3a02/ : FAILURE in 32m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-scenario-two-node/c8c6b80/ : FAILURE in 37m 35s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-py2-scenario-two-node/811cda4/ : FAILURE in 33m 05s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-tls-barbican/2c96d37/ : FAILURE in 34m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0dc61e36b791584bd40b683bd85bc14b24bd982","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-22 14:51:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"922b2f5d5029dda4ae67c09b852c63e0403d567b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 18:48:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/27/660227/1/check/openstack-tox-cover/4e4b171/cover/ : SUCCESS in 8m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/660227/1/check/openstack-tox-lower-constraints/1d1f7ff/ : SUCCESS in 6m 51s\n- openstack-tox-pep8 http://logs.openstack.org/27/660227/1/check/openstack-tox-pep8/a978cb9/ : SUCCESS in 6m 41s\n- openstack-tox-py27 http://logs.openstack.org/27/660227/1/check/openstack-tox-py27/fdcfae9/ : SUCCESS in 5m 22s\n- openstack-tox-py36 http://logs.openstack.org/27/660227/1/check/openstack-tox-py36/de72454/ : SUCCESS in 5m 38s\n- openstack-tox-py37 http://logs.openstack.org/27/660227/1/check/openstack-tox-py37/1333300/ : SUCCESS in 7m 26s\n- openstack-tox-docs http://logs.openstack.org/27/660227/1/check/openstack-tox-docs/fc5d02a/html/ : SUCCESS in 7m 11s\n- build-openstack-releasenotes http://logs.openstack.org/27/660227/1/check/build-openstack-releasenotes/e2345ee/html/ : SUCCESS in 5m 07s\n- openstack-tox-functional http://logs.openstack.org/27/660227/1/check/openstack-tox-functional/55aed59/ : SUCCESS in 6m 40s\n- openstack-tox-functional-py36 http://logs.openstack.org/27/660227/1/check/openstack-tox-functional-py36/4ba39fa/ : SUCCESS in 9m 25s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-noop-api/93a836e/ : SUCCESS in 42m 52s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-noop-py2-api/3a56da4/ : SUCCESS in 44m 27s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-scenario/427e8b6/ : FAILURE in 29m 18s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-py2-scenario/569e3e9/ : FAILURE in 29m 17s\n- octavia-v2-dsvm-py2-scenario-centos-7 finger://ze01.openstack.org/dd22d268ee784274bf1d07e2a1af7a6d : RETRY_LIMIT in 40m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic finger://ze09.openstack.org/2b4d4a1ed90d4244ac9747b414b2d7b0 : RETRY_LIMIT in 33m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-scenario-two-node/0847ae9/ : FAILURE in 34m 13s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node finger://ze09.openstack.org/dadadd47246447ad8e05400e49fc4112 : RETRY_LIMIT in 35m 47s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-tls-barbican/ce4be1a/ : SUCCESS in 55m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8bdf09f9f2bc19e7b4d31e8073f70ff728742d7a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-22 22:01:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"660b67bfc3a65c781c2601f2d2f6ff4614352a2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 01:52:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/27/660227/1/check/openstack-tox-cover/da05d03/cover/ : SUCCESS in 8m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/660227/1/check/openstack-tox-lower-constraints/6a13733/ : SUCCESS in 6m 42s\n- openstack-tox-pep8 http://logs.openstack.org/27/660227/1/check/openstack-tox-pep8/9b10871/ : SUCCESS in 5m 56s\n- openstack-tox-py27 http://logs.openstack.org/27/660227/1/check/openstack-tox-py27/4876260/ : SUCCESS in 5m 29s\n- openstack-tox-py36 http://logs.openstack.org/27/660227/1/check/openstack-tox-py36/475d3c8/ : SUCCESS in 7m 20s\n- openstack-tox-py37 http://logs.openstack.org/27/660227/1/check/openstack-tox-py37/aed7396/ : SUCCESS in 6m 16s\n- openstack-tox-docs http://logs.openstack.org/27/660227/1/check/openstack-tox-docs/9fae222/html/ : SUCCESS in 6m 03s\n- build-openstack-releasenotes http://logs.openstack.org/27/660227/1/check/build-openstack-releasenotes/111ac7a/html/ : SUCCESS in 5m 06s\n- openstack-tox-functional http://logs.openstack.org/27/660227/1/check/openstack-tox-functional/677bf3b/ : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 http://logs.openstack.org/27/660227/1/check/openstack-tox-functional-py36/1bc5b11/ : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-noop-api/921d0c7/ : SUCCESS in 48m 59s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-noop-py2-api/c48fe8c/ : SUCCESS in 37m 11s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-scenario/72adfc5/ : SUCCESS in 1h 31m 52s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-py2-scenario/7fe2ca6/ : SUCCESS in 1h 32m 20s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-py2-scenario-centos-7/ed3e3b7/ : FAILURE in 1h 58m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/9bf35b1/ : SUCCESS in 1h 41m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-scenario-two-node/0225464/ : FAILURE in 1h 50m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-py2-scenario-two-node/ad0b49d/ : FAILURE in 1h 45m 34s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/27/660227/1/check/octavia-v2-dsvm-tls-barbican/3ec2fef/ : SUCCESS in 54m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c7220d1fa75bb459b43bcefae767fcb42b4f3e1","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-05-23 05:56:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab743fd1b3983234c88255acd6d1c2af1bc49536","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-23 06:07:27.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b21ee4f975b6d6432fdda3e1b81c092b1c42c5f","author":{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},"date":"2019-05-23 06:11:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f127c2040e205968fd84b6c240db2e25bb49a9bc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-23 13:54:53.000000000","message":"Patch Set 1:\n\nYes, python3.6 is the default on Ubuntu Bionic which is a primary OS for OpenStack Train.","accounts_in_message":[],"_revision_number":1},{"id":"1ca40f43126fe4a153f5c1a3d9cc4b3dfc8e116b","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-05-23 14:03:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7c4ff448468041422ca6174e08e54e5180c5e5e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 14:03:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1cd502843aee54a54e9dfe24ae0fe05f87c08feb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 15:54:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f0021d8678d44efc8ae31c5f44e6918e1ad956cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 15:54:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/27/660227/1/gate/openstack-tox-lower-constraints/638a828/ : SUCCESS in 6m 18s\n- openstack-tox-pep8 http://logs.openstack.org/27/660227/1/gate/openstack-tox-pep8/632820a/ : SUCCESS in 6m 49s\n- openstack-tox-py27 http://logs.openstack.org/27/660227/1/gate/openstack-tox-py27/213b4b6/ : SUCCESS in 4m 57s\n- openstack-tox-py36 http://logs.openstack.org/27/660227/1/gate/openstack-tox-py36/c88c047/ : SUCCESS in 5m 54s\n- openstack-tox-py37 http://logs.openstack.org/27/660227/1/gate/openstack-tox-py37/67120ca/ : SUCCESS in 7m 09s\n- openstack-tox-docs http://logs.openstack.org/27/660227/1/gate/openstack-tox-docs/44db6d1/html/ : SUCCESS in 6m 50s\n- build-openstack-releasenotes http://logs.openstack.org/27/660227/1/gate/build-openstack-releasenotes/6099ffe/html/ : SUCCESS in 5m 44s\n- openstack-tox-functional http://logs.openstack.org/27/660227/1/gate/openstack-tox-functional/6f91710/ : SUCCESS in 6m 43s\n- openstack-tox-functional-py36 http://logs.openstack.org/27/660227/1/gate/openstack-tox-functional-py36/650c90c/ : SUCCESS in 7m 05s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/27/660227/1/gate/octavia-v2-dsvm-noop-api/9c45df8/ : SUCCESS in 38m 35s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/27/660227/1/gate/octavia-v2-dsvm-noop-py2-api/9fef7ed/ : SUCCESS in 40m 57s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/27/660227/1/gate/octavia-v2-dsvm-scenario/bbb0911/ : SUCCESS in 1h 33m 49s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/27/660227/1/gate/octavia-v2-dsvm-py2-scenario/dbac5e9/ : SUCCESS in 1h 43m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"135d999f5e358770edc41e0c9c351f8e5e714bda","revisions":{"135d999f5e358770edc41e0c9c351f8e5e714bda":{"kind":"REWORK","_number":1,"created":"2019-05-20 23:12:00.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/660227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/660227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/660227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/660227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/660227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/660227/1"}}},"commit":{"parents":[{"commit":"4da644136e5fa5295373a0c0d0df895ee77eef2d","subject":"Merge \"Delete unused files in tests/contrib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4da644136e5fa5295373a0c0d0df895ee77eef2d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-20 23:10:21.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-20 23:11:59.000000000","tz":-420},"subject":"Fix tox for functional py36 and py37","message":"Fix tox for functional py36 and py37\n\nThe default tox run had the wrong environment names. functional-36\nvs. functional-py36.\n\nChange-Id: I196b21460789b4295ea2dc4ac297b218e26a1e43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/135d999f5e358770edc41e0c9c351f8e5e714bda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/135d999f5e358770edc41e0c9c351f8e5e714bda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
