)]}'
{"id":"openstack%2Foctavia~662280","triplet_id":"openstack%2Foctavia~master~Ida1d544933aec9e5cd556aef57a06e9f19f1b255","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ida1d544933aec9e5cd556aef57a06e9f19f1b255","subject":"Fix health monitor API handling of None updates","status":"MERGED","created":"2019-05-30 18:15:37.000000000","updated":"2019-09-19 09:55:25.000000000","submitted":"2019-06-24 04:32:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662280-1561350773135-f1e7b0d6","meta_rev_id":"78bcea7fe681d6bfa676dc8795ccf692b715baad","_number":662280,"virtual_id_number":662280,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-06-24 04:32:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-06-18 09:19:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-06-24 02:55:58.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-31 01:06:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-06-24 02:55:58.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-31 01:06:00.000000000","updated_by":{"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"},"reviewer":{"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"},"state":"REVIEWER"},{"updated":"2019-06-18 09:19:44.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-06-24 02:55:58.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-06-24 04:32:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b9ff6e63f976b35210428e74ee6731169cfdac75","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-30 18:15:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"845f3ebb287f2bff95cfdc29f69b98e1bb5ddd9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 20:26:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/80/662280/1/check/openstack-tox-cover/d4ccdd6/cover/ : SUCCESS in 9m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/80/662280/1/check/openstack-tox-lower-constraints/5d67b33/ : SUCCESS in 8m 36s\n- openstack-tox-pep8 http://logs.openstack.org/80/662280/1/check/openstack-tox-pep8/3db9c70/ : SUCCESS in 6m 19s\n- openstack-tox-py27 http://logs.openstack.org/80/662280/1/check/openstack-tox-py27/b3b10d7/ : SUCCESS in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/80/662280/1/check/openstack-tox-py36/06b7549/ : SUCCESS in 5m 06s\n- openstack-tox-py37 http://logs.openstack.org/80/662280/1/check/openstack-tox-py37/3c4aea0/ : SUCCESS in 6m 04s\n- openstack-tox-docs http://logs.openstack.org/80/662280/1/check/openstack-tox-docs/7ebb5c7/html/ : SUCCESS in 6m 01s\n- openstack-tox-functional http://logs.openstack.org/80/662280/1/check/openstack-tox-functional/7fad2be/ : SUCCESS in 6m 41s\n- openstack-tox-functional-py36 http://logs.openstack.org/80/662280/1/check/openstack-tox-functional-py36/88e8148/ : SUCCESS in 6m 20s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/80/662280/1/check/octavia-v2-dsvm-noop-api/4db5611/ : SUCCESS in 43m 22s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/80/662280/1/check/octavia-v2-dsvm-noop-py2-api/4afc31b/ : SUCCESS in 35m 57s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/80/662280/1/check/octavia-v2-dsvm-scenario/7e7e786/ : SUCCESS in 1h 38m 53s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/80/662280/1/check/octavia-v2-dsvm-py2-scenario/b238b13/ : SUCCESS in 1h 38m 17s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/80/662280/1/check/octavia-v2-dsvm-py2-scenario-centos-7/03f65c7/ : FAILURE in 2h 06m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/80/662280/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/858b448/ : SUCCESS in 1h 45m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/80/662280/1/check/octavia-v2-dsvm-scenario-two-node/35c5304/ : SUCCESS in 1h 55m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/80/662280/1/check/octavia-v2-dsvm-py2-scenario-two-node/7b5603b/ : SUCCESS in 1h 55m 28s (non-voting)\n- octavia-grenade http://logs.openstack.org/80/662280/1/check/octavia-grenade/78fe2ef/ : SUCCESS in 1h 13m 17s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/80/662280/1/check/octavia-v2-dsvm-tls-barbican/a05f1ee/ : SUCCESS in 57m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8c0d66d123592240fe6af439ae9f067af16a95a","author":{"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"},"date":"2019-05-31 01:06:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"75ffcfd6661350d357e582309fb91ead9549dfb2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-18 09:19:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"85fe618dedd4f9127fcda2b65859a4e09c2aaf9e","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-06-24 02:55:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"02a4d7a4c087ce645268a84d412535895346a3bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 02:56:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2c44d6930472c2fb932a038e82bcc1be8508c96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 04:32:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/80/662280/1/gate/openstack-tox-lower-constraints/1c75cff/ : SUCCESS in 7m 21s\n- openstack-tox-pep8 http://logs.openstack.org/80/662280/1/gate/openstack-tox-pep8/fe11870/ : SUCCESS in 6m 15s\n- openstack-tox-py27 http://logs.openstack.org/80/662280/1/gate/openstack-tox-py27/c3e5925/ : SUCCESS in 5m 20s\n- openstack-tox-py36 http://logs.openstack.org/80/662280/1/gate/openstack-tox-py36/4ad8efd/ : SUCCESS in 5m 10s\n- openstack-tox-py37 http://logs.openstack.org/80/662280/1/gate/openstack-tox-py37/a67cfd7/ : SUCCESS in 5m 15s\n- openstack-tox-docs http://logs.openstack.org/80/662280/1/gate/openstack-tox-docs/daf5590/html/ : SUCCESS in 6m 18s\n- openstack-tox-functional http://logs.openstack.org/80/662280/1/gate/openstack-tox-functional/a9800bd/ : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 http://logs.openstack.org/80/662280/1/gate/openstack-tox-functional-py36/fc83f38/ : SUCCESS in 6m 03s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/80/662280/1/gate/octavia-v2-dsvm-noop-api/41baa02/ : SUCCESS in 40m 29s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/80/662280/1/gate/octavia-v2-dsvm-noop-py2-api/30ca0a8/ : SUCCESS in 40m 42s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/80/662280/1/gate/octavia-v2-dsvm-scenario/b5206d8/ : SUCCESS in 1h 31m 03s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/80/662280/1/gate/octavia-v2-dsvm-py2-scenario/24ba29c/ : SUCCESS in 1h 27m 26s\n- octavia-grenade http://logs.openstack.org/80/662280/1/gate/octavia-grenade/0bcf731/ : SUCCESS in 1h 16m 55s","accounts_in_message":[],"_revision_number":1},{"id":"87e9d0ef1c3e36c9b6268e00a4d8771d98a602d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 04:32:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"78bcea7fe681d6bfa676dc8795ccf692b715baad","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-19 09:55:25.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit b82589cbabd5123d6182ed248a9300cde9cbce16","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a95f7481565447f4512743a91ff3973af6f052bb","revisions":{"a95f7481565447f4512743a91ff3973af6f052bb":{"kind":"REWORK","_number":1,"created":"2019-05-30 18:15:37.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/80/662280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/662280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/662280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/662280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/662280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/662280/1"}}},"commit":{"parents":[{"commit":"21a54d6b78d002c76ca2e68b3aad34681d51686f","subject":"Merge \"Rename review.openstack.org to review.opendev.org\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21a54d6b78d002c76ca2e68b3aad34681d51686f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-30 18:13:42.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-30 18:15:36.000000000","tz":-420},"subject":"Fix health monitor API handling of None updates","message":"Fix health monitor API handling of None updates\n\nThe current health monitor API does not properly handle\nclearing/reseting values on update. Some integer only fields,\nsuch as max_retries_down, will accept null, but will store the\nvalue as \"None\". These will will cause failures updating the\namphora configuration.\n\nThis patch corrects this to appropriately handle None/null updates\nto the health monitor parameters.\n\nChange-Id: Ida1d544933aec9e5cd556aef57a06e9f19f1b255\nStory: 2005374\nTask: 33533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a95f7481565447f4512743a91ff3973af6f052bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a95f7481565447f4512743a91ff3973af6f052bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
