)]}'
{"id":"openstack%2Foctavia~662344","triplet_id":"openstack%2Foctavia~master~I5253c48871a8bb3bf91f82aa7791585cc4a6d529","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I5253c48871a8bb3bf91f82aa7791585cc4a6d529","subject":"Fix flavor profile API handling of None updates","status":"MERGED","created":"2019-05-30 22:12:54.000000000","updated":"2019-08-26 14:05:57.000000000","submitted":"2019-08-01 01:43:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662344-1564623787803-c33df9d0","meta_rev_id":"6dc41439a8c42420438e8a79dfd7171bffc8c5a4","_number":662344,"virtual_id_number":662344,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-08-01 01:43:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-07-31 23:44:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-07-29 21:50:17.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-07-31 23:44:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-07-29 21:50:17.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-25 16:15:38.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-07-29 21:50:17.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-07-31 23:44:26.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-08-01 01:43:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"afa587bbba795c65f8cb9d1ec4c6462587684671","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-30 22:12:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cbcfe972fb2e732d693a06fb1c3199d7647849ff","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-30 22:13:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3a0757df6f6b4a77759e785673acda39f68d07e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 00:42:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/44/662344/2/check/openstack-tox-cover/867125f/cover/ : SUCCESS in 7m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/44/662344/2/check/openstack-tox-lower-constraints/26687fb/ : SUCCESS in 6m 36s\n- openstack-tox-pep8 http://logs.openstack.org/44/662344/2/check/openstack-tox-pep8/125889b/ : SUCCESS in 6m 20s\n- openstack-tox-py27 http://logs.openstack.org/44/662344/2/check/openstack-tox-py27/6491ce2/ : SUCCESS in 5m 10s\n- openstack-tox-py36 http://logs.openstack.org/44/662344/2/check/openstack-tox-py36/c59daa2/ : SUCCESS in 5m 34s\n- openstack-tox-py37 http://logs.openstack.org/44/662344/2/check/openstack-tox-py37/1d47637/ : SUCCESS in 6m 48s\n- openstack-tox-docs http://logs.openstack.org/44/662344/2/check/openstack-tox-docs/423bddc/html/ : SUCCESS in 5m 44s\n- openstack-tox-functional http://logs.openstack.org/44/662344/2/check/openstack-tox-functional/03449be/ : SUCCESS in 6m 02s\n- openstack-tox-functional-py36 http://logs.openstack.org/44/662344/2/check/openstack-tox-functional-py36/dd4581e/ : SUCCESS in 6m 15s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/44/662344/2/check/octavia-v2-dsvm-noop-api/de8df47/ : SUCCESS in 42m 46s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/44/662344/2/check/octavia-v2-dsvm-noop-py2-api/fcbb5dc/ : SUCCESS in 45m 55s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/44/662344/2/check/octavia-v2-dsvm-scenario/e60dc51/ : SUCCESS in 1h 37m 25s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/44/662344/2/check/octavia-v2-dsvm-py2-scenario/7a614c4/ : SUCCESS in 1h 32m 33s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/44/662344/2/check/octavia-v2-dsvm-py2-scenario-centos-7/025fceb/ : FAILURE in 2h 13m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/44/662344/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/0415a44/ : FAILURE in 1h 57m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/44/662344/2/check/octavia-v2-dsvm-scenario-two-node/554385e/ : SUCCESS in 1h 52m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/44/662344/2/check/octavia-v2-dsvm-py2-scenario-two-node/f816b19/ : SUCCESS in 1h 55m 21s (non-voting)\n- octavia-grenade http://logs.openstack.org/44/662344/2/check/octavia-grenade/e5632de/ : SUCCESS in 1h 11m 36s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/44/662344/2/check/octavia-v2-dsvm-tls-barbican/13462da/ : SUCCESS in 53m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ff115d135f00d38152bee56b2d2aeff0a25da68a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-31 22:55:13.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d61c6cc30a904295e4768979556d2decaf634dbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 01:26:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/44/662344/3/check/openstack-tox-cover/f299cad/cover/ : SUCCESS in 8m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/44/662344/3/check/openstack-tox-lower-constraints/47720ef/ : SUCCESS in 7m 12s\n- openstack-tox-pep8 http://logs.openstack.org/44/662344/3/check/openstack-tox-pep8/e757d7e/ : SUCCESS in 6m 41s\n- openstack-tox-py27 http://logs.openstack.org/44/662344/3/check/openstack-tox-py27/80f8c8f/ : SUCCESS in 5m 24s\n- openstack-tox-py36 http://logs.openstack.org/44/662344/3/check/openstack-tox-py36/8fe5a34/ : SUCCESS in 5m 35s\n- openstack-tox-py37 http://logs.openstack.org/44/662344/3/check/openstack-tox-py37/36ffbc5/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/44/662344/3/check/openstack-tox-docs/115e485/html/ : SUCCESS in 9m 38s\n- openstack-tox-functional http://logs.openstack.org/44/662344/3/check/openstack-tox-functional/e66f5b7/ : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 http://logs.openstack.org/44/662344/3/check/openstack-tox-functional-py36/5904dce/ : SUCCESS in 7m 22s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/44/662344/3/check/octavia-v2-dsvm-noop-api/d327e70/ : SUCCESS in 53m 56s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/44/662344/3/check/octavia-v2-dsvm-noop-py2-api/654ed10/ : SUCCESS in 39m 35s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/44/662344/3/check/octavia-v2-dsvm-scenario/3075dca/ : SUCCESS in 1h 30m 03s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/44/662344/3/check/octavia-v2-dsvm-py2-scenario/604028c/ : SUCCESS in 1h 22m 03s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/44/662344/3/check/octavia-v2-dsvm-py2-scenario-centos-7/8568c08/ : FAILURE in 1h 55m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/44/662344/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/7330df1/ : SUCCESS in 1h 47m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/44/662344/3/check/octavia-v2-dsvm-scenario-two-node/8b7bc30/ : SUCCESS in 1h 42m 16s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/44/662344/3/check/octavia-v2-dsvm-py2-scenario-two-node/d47be81/ : SUCCESS in 1h 51m 35s (non-voting)\n- octavia-grenade http://logs.openstack.org/44/662344/3/check/octavia-grenade/2c04a6a/ : SUCCESS in 1h 18m 41s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/44/662344/3/check/octavia-v2-dsvm-tls-barbican/f9d892d/ : SUCCESS in 56m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0e5926f4ce59ce2c8a3acaf75975d514998c477f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-07-25 16:15:38.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f53fa9842df7d9f48bd0eb19788e3038750a5bba","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-25 16:36:16.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nA release note should be added as I think we want to backport this.","accounts_in_message":[],"_revision_number":3},{"id":"d6a4f5ad00b5a734e7fdbefe7321baaad716fef3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-26 22:05:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bf68415c9fe9a57ef7cf775519e2e8ca9b6825af","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-27 21:59:33.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nLGTM, just a release note is missing.","accounts_in_message":[],"_revision_number":3},{"id":"0dc7b8d06ed2b0f385656ecf103939247400022d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-29 21:49:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"19c5ade80b3f00f0f1a4c5e2dcef138a16ea2df2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-29 21:50:17.000000000","message":"Patch Set 4: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9c2ccd395acb34a0a44b18da53bc2f782d533d42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 23:49:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/44/662344/4/check/openstack-tox-cover/f8df8ad/cover/ : SUCCESS in 6m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/44/662344/4/check/openstack-tox-lower-constraints/3833b84/ : SUCCESS in 6m 04s\n- openstack-tox-pep8 http://logs.openstack.org/44/662344/4/check/openstack-tox-pep8/5366cdf/ : SUCCESS in 5m 59s\n- openstack-tox-py27 http://logs.openstack.org/44/662344/4/check/openstack-tox-py27/0b53ecc/ : SUCCESS in 5m 22s\n- openstack-tox-py36 http://logs.openstack.org/44/662344/4/check/openstack-tox-py36/ce20d0c/ : SUCCESS in 3m 54s\n- openstack-tox-py37 http://logs.openstack.org/44/662344/4/check/openstack-tox-py37/17fcdc6/ : SUCCESS in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/44/662344/4/check/openstack-tox-docs/fc04258/html/ : SUCCESS in 5m 03s\n- build-openstack-releasenotes http://logs.openstack.org/44/662344/4/check/build-openstack-releasenotes/40d2af8/html/ : SUCCESS in 4m 03s\n- openstack-tox-functional http://logs.openstack.org/44/662344/4/check/openstack-tox-functional/9a73be5/ : SUCCESS in 5m 22s\n- openstack-tox-functional-py36 http://logs.openstack.org/44/662344/4/check/openstack-tox-functional-py36/a542e9a/ : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/44/662344/4/check/octavia-v2-dsvm-noop-api/7a87332/ : SUCCESS in 40m 09s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/44/662344/4/check/octavia-v2-dsvm-noop-py2-api/88b092d/ : SUCCESS in 49m 17s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/44/662344/4/check/octavia-v2-dsvm-scenario/68bf91a/ : FAILURE in 1h 27m 45s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/44/662344/4/check/octavia-v2-dsvm-py2-scenario/ae28d73/ : SUCCESS in 1h 32m 07s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/44/662344/4/check/octavia-v2-dsvm-py2-scenario-centos-7/bee9f07/ : FAILURE in 1h 48m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/44/662344/4/check/octavia-v2-dsvm-scenario-two-node/ea94cc6/ : SUCCESS in 1h 48m 07s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/44/662344/4/check/octavia-v2-dsvm-py2-scenario-two-node/94cc1e0/ : SUCCESS in 1h 39m 09s (non-voting)\n- octavia-grenade http://logs.openstack.org/44/662344/4/check/octavia-grenade/b5cf608/ : SUCCESS in 1h 23m 34s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/44/662344/4/check/octavia-v2-dsvm-tls-barbican/ae9006b/ : SUCCESS in 50m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/44/662344/4/check/octavia-v2-dsvm-scenario-amphora-v2/7123c8d/ : FAILURE in 1h 36m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/44/662344/4/check/octavia-v2-act-stdby-iptables-dsvm-scenario/cd66ffa/ : SUCCESS in 1h 09m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/44/662344/4/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/d71cc0f/ : FAILURE in 1h 16m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/44/662344/4/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/32bb704/ : SUCCESS in 1h 20m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"36b98cd76bea150004dd313187e1ba3535f5e928","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-30 13:37:34.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"63ab2aa3cb9ea5dcc4ab05c4cb60aed4ebf5ab6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 17:17:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/44/662344/4/check/openstack-tox-cover/b209728/cover/ : SUCCESS in 9m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/44/662344/4/check/openstack-tox-lower-constraints/f956546/ : SUCCESS in 6m 47s\n- openstack-tox-pep8 http://logs.openstack.org/44/662344/4/check/openstack-tox-pep8/b2a41ef/ : SUCCESS in 5m 45s\n- openstack-tox-py27 http://logs.openstack.org/44/662344/4/check/openstack-tox-py27/f2e5a7e/ : SUCCESS in 5m 58s\n- openstack-tox-py36 http://logs.openstack.org/44/662344/4/check/openstack-tox-py36/97ef5ef/ : SUCCESS in 6m 23s\n- openstack-tox-py37 http://logs.openstack.org/44/662344/4/check/openstack-tox-py37/6fc162d/ : SUCCESS in 5m 23s\n- openstack-tox-docs http://logs.openstack.org/44/662344/4/check/openstack-tox-docs/ae49ced/html/ : SUCCESS in 5m 32s\n- build-openstack-releasenotes http://logs.openstack.org/44/662344/4/check/build-openstack-releasenotes/f28485b/html/ : SUCCESS in 4m 52s\n- openstack-tox-functional http://logs.openstack.org/44/662344/4/check/openstack-tox-functional/34a96d1/ : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 http://logs.openstack.org/44/662344/4/check/openstack-tox-functional-py36/82cf751/ : SUCCESS in 5m 42s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/44/662344/4/check/octavia-v2-dsvm-noop-api/56f7aa7/ : SUCCESS in 49m 19s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/44/662344/4/check/octavia-v2-dsvm-noop-py2-api/d004882/ : SUCCESS in 44m 53s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/44/662344/4/check/octavia-v2-dsvm-scenario/1c0b8fa/ : SUCCESS in 1h 28m 25s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/44/662344/4/check/octavia-v2-dsvm-py2-scenario/061d629/ : SUCCESS in 1h 44m 06s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/44/662344/4/check/octavia-v2-dsvm-py2-scenario-centos-7/9dd0b09/ : FAILURE in 1h 56m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/44/662344/4/check/octavia-v2-dsvm-scenario-two-node/804044e/ : SUCCESS in 1h 59m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/44/662344/4/check/octavia-v2-dsvm-py2-scenario-two-node/f02f436/ : SUCCESS in 1h 53m 58s (non-voting)\n- octavia-grenade http://logs.openstack.org/44/662344/4/check/octavia-grenade/29ac5fa/ : SUCCESS in 1h 15m 52s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/44/662344/4/check/octavia-v2-dsvm-tls-barbican/7e52e66/ : SUCCESS in 52m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/44/662344/4/check/octavia-v2-dsvm-scenario-amphora-v2/b9e5917/ : SUCCESS in 1h 41m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/44/662344/4/check/octavia-v2-act-stdby-iptables-dsvm-scenario/15b436b/ : FAILURE in 55m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/44/662344/4/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/8c1f59c/ : FAILURE in 1h 19m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/44/662344/4/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/1a82e8b/ : FAILURE in 1h 36m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9d7ad89dc33180cebee1c88160bab10a3b30cc80","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-31 23:44:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c3b9b95b8cbaa864dced38f7d548192051aba3b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 23:44:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"45e95c55ba932cb68b2f21abb84ffd2bf38e4012","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 01:43:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/44/662344/4/gate/openstack-tox-lower-constraints/3dd8352/ : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://logs.opendev.org/44/662344/4/gate/openstack-tox-pep8/1b3a071/ : SUCCESS in 6m 02s\n- openstack-tox-py27 https://logs.opendev.org/44/662344/4/gate/openstack-tox-py27/af4838e/ : SUCCESS in 5m 16s\n- openstack-tox-py36 https://logs.opendev.org/44/662344/4/gate/openstack-tox-py36/20c5a0a/ : SUCCESS in 4m 59s\n- openstack-tox-py37 https://logs.opendev.org/44/662344/4/gate/openstack-tox-py37/031772b/ : SUCCESS in 4m 25s\n- openstack-tox-docs https://logs.opendev.org/44/662344/4/gate/openstack-tox-docs/ef528cf/html/ : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://logs.opendev.org/44/662344/4/gate/build-openstack-releasenotes/3c9958c/html/ : SUCCESS in 5m 01s\n- openstack-tox-functional https://logs.opendev.org/44/662344/4/gate/openstack-tox-functional/0431c3a/ : SUCCESS in 6m 38s\n- openstack-tox-functional-py36 https://logs.opendev.org/44/662344/4/gate/openstack-tox-functional-py36/da0a714/ : SUCCESS in 6m 30s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/44/662344/4/gate/octavia-v2-dsvm-noop-api/0cf6c55/ : SUCCESS in 46m 51s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/44/662344/4/gate/octavia-v2-dsvm-noop-py2-api/8e67722/ : SUCCESS in 39m 05s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/44/662344/4/gate/octavia-v2-dsvm-scenario/afff944/ : SUCCESS in 1h 47m 54s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/44/662344/4/gate/octavia-v2-dsvm-py2-scenario/5ee895a/ : SUCCESS in 1h 51m 42s\n- octavia-grenade https://logs.opendev.org/44/662344/4/gate/octavia-grenade/0cf48f9/ : SUCCESS in 1h 18m 24s","accounts_in_message":[],"_revision_number":4},{"id":"a2d65a34161561d7b08974560dce69d3f848a4b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 01:43:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"6dc41439a8c42420438e8a79dfd7171bffc8c5a4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 14:05:57.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 3f1b38908cca70f1e3dcd12be1da807fed77e858","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ab8a263674a918bf5a4f1c7ebf20594baad1d65a","revisions":{"e5b4630a8fb5c73bd9684091cfd8efc4c26ad96a":{"kind":"REWORK","_number":1,"created":"2019-05-30 22:12:54.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/44/662344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/662344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/662344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/662344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/662344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/662344/1"}}},"commit":{"parents":[{"commit":"21a54d6b78d002c76ca2e68b3aad34681d51686f","subject":"Merge \"Rename review.openstack.org to review.opendev.org\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21a54d6b78d002c76ca2e68b3aad34681d51686f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-30 22:10:53.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-30 22:12:52.000000000","tz":-420},"subject":"Fix flavor profile API handling of None updates","message":"Fix flavor profile API handling of None updates\n\nThe current flavor profile API does not properly handle\nclearing/reseting values on update. Some mandatory fields would\nreturn a database \"cannot be Null\" error. This patch raises the\nproper invalid option execption.\n\nChange-Id: I5253c48871a8bb3bf91f82aa7791585cc4a6d529\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5b4630a8fb5c73bd9684091cfd8efc4c26ad96a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5b4630a8fb5c73bd9684091cfd8efc4c26ad96a"}]},"branch":"refs/heads/master"},"22f50f116c31db3c164b7f74bd41270d47e5f034":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-30 22:13:31.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/44/662344/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/662344/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/662344/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/662344/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/662344/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/662344/2"}}},"commit":{"parents":[{"commit":"21a54d6b78d002c76ca2e68b3aad34681d51686f","subject":"Merge \"Rename review.openstack.org to review.opendev.org\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21a54d6b78d002c76ca2e68b3aad34681d51686f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-30 22:10:53.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-30 22:13:29.000000000","tz":0},"subject":"Fix flavor profile API handling of None updates","message":"Fix flavor profile API handling of None updates\n\nThe current flavor profile API does not properly handle\nclearing/reseting values on update. Some mandatory fields would\nreturn a database \"cannot be Null\" error. This patch raises the\nproper invalid option execption.\n\nStory: 2005374\nTask: 33539\n\nChange-Id: I5253c48871a8bb3bf91f82aa7791585cc4a6d529\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/22f50f116c31db3c164b7f74bd41270d47e5f034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/22f50f116c31db3c164b7f74bd41270d47e5f034"}]},"branch":"refs/heads/master"},"8bdce107f8d7aae2d05363fab87c5db30f7bae57":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-05-31 22:55:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/44/662344/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/662344/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/662344/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/662344/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/662344/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/662344/3"}}},"commit":{"parents":[{"commit":"21a54d6b78d002c76ca2e68b3aad34681d51686f","subject":"Merge \"Rename review.openstack.org to review.opendev.org\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21a54d6b78d002c76ca2e68b3aad34681d51686f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-30 22:10:53.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-31 22:55:10.000000000","tz":0},"subject":"Fix flavor profile API handling of None updates","message":"Fix flavor profile API handling of None updates\n\nThe current flavor profile API does not properly handle\nclearing/reseting values on update. Some mandatory fields would\nreturn a database \"cannot be Null\" error. This patch raises the\nproper invalid option execption.\n\nStory: 2005374\nTask: 33542\n\nChange-Id: I5253c48871a8bb3bf91f82aa7791585cc4a6d529\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8bdce107f8d7aae2d05363fab87c5db30f7bae57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8bdce107f8d7aae2d05363fab87c5db30f7bae57"}]},"branch":"refs/heads/master"},"ab8a263674a918bf5a4f1c7ebf20594baad1d65a":{"kind":"REWORK","_number":4,"created":"2019-07-29 21:49:23.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/44/662344/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/662344/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/662344/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/662344/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/662344/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/662344/4"}}},"commit":{"parents":[{"commit":"21a54d6b78d002c76ca2e68b3aad34681d51686f","subject":"Merge \"Rename review.openstack.org to review.opendev.org\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21a54d6b78d002c76ca2e68b3aad34681d51686f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-30 22:10:53.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-29 21:49:22.000000000","tz":-420},"subject":"Fix flavor profile API handling of None updates","message":"Fix flavor profile API handling of None updates\n\nThe current flavor profile API does not properly handle\nclearing/reseting values on update. Some mandatory fields would\nreturn a database \"cannot be Null\" error. This patch raises the\nproper invalid option execption.\n\nStory: 2005374\nTask: 33542\n\nChange-Id: I5253c48871a8bb3bf91f82aa7791585cc4a6d529\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ab8a263674a918bf5a4f1c7ebf20594baad1d65a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ab8a263674a918bf5a4f1c7ebf20594baad1d65a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
