)]}'
{"id":"openstack%2Foctavia~662791","triplet_id":"openstack%2Foctavia~master~I2cf241ea965ad56ed70ebde83632ab855f5d859e","project":"openstack/octavia","branch":"master","topic":"jobboard","hashtags":[],"change_id":"I2cf241ea965ad56ed70ebde83632ab855f5d859e","subject":"Use retry for AmphoraComputeConnectivityWait","status":"MERGED","created":"2019-06-03 15:48:38.000000000","updated":"2019-12-03 03:26:47.000000000","submitted":"2019-12-03 03:25:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":36,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662791-1575343502123-7c5e976c","meta_rev_id":"b42294884f442090f4dba00d318f9caaa170f26c","_number":662791,"virtual_id_number":662791,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-12-02 19:53:22.000000000","_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-03 03:25:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2019-12-03 00:36:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-12-03 00:58:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-12-03 01:52:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":1,"date":"2019-12-03 01:52:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":-2,"date":"2019-12-03 01:52:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-19 06:55:37.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-10-31 06:59:20.000000000","updated_by":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"reviewer":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"state":"REVIEWER"},{"updated":"2019-12-03 00:36:18.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-12-03 00:58:30.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-12-03 01:52:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-12-03 03:25:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2e6f77b7a7dce9af0e6fbbbbcb4618ce53648f9","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-06-03 15:48:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66af78d26ed7efcaf5e597cc6ef360cfa4bbbf7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 17:15:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/91/662791/1/check/openstack-tox-cover/5bfdb5b/ : FAILURE in 7m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662791/1/check/openstack-tox-lower-constraints/072fd94/ : FAILURE in 8m 26s\n- openstack-tox-pep8 http://logs.openstack.org/91/662791/1/check/openstack-tox-pep8/181bad4/ : FAILURE in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/91/662791/1/check/openstack-tox-py27/8a04544/ : FAILURE in 5m 39s\n- openstack-tox-py36 http://logs.openstack.org/91/662791/1/check/openstack-tox-py36/8404e19/ : FAILURE in 6m 28s\n- openstack-tox-py37 http://logs.openstack.org/91/662791/1/check/openstack-tox-py37/f97d4e9/ : FAILURE in 6m 06s\n- openstack-tox-docs http://logs.openstack.org/91/662791/1/check/openstack-tox-docs/734467a/html/ : SUCCESS in 6m 31s\n- openstack-tox-functional http://logs.openstack.org/91/662791/1/check/openstack-tox-functional/0182b3a/ : SUCCESS in 6m 24s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/662791/1/check/openstack-tox-functional-py36/580aec8/ : SUCCESS in 6m 52s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/91/662791/1/check/octavia-v2-dsvm-noop-api/f8b875e/ : SUCCESS in 41m 40s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/91/662791/1/check/octavia-v2-dsvm-noop-py2-api/86b303c/ : SUCCESS in 41m 55s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/91/662791/1/check/octavia-v2-dsvm-scenario/0926a98/ : FAILURE in 50m 06s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/91/662791/1/check/octavia-v2-dsvm-py2-scenario/0d345cc/ : FAILURE in 57m 07s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/91/662791/1/check/octavia-v2-dsvm-py2-scenario-centos-7/0bdfb63/ : FAILURE in 1h 03m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/91/662791/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/6c6824c/ : FAILURE in 48m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/91/662791/1/check/octavia-v2-dsvm-scenario-two-node/26c4da1/ : FAILURE in 1h 15m 04s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/91/662791/1/check/octavia-v2-dsvm-py2-scenario-two-node/4f10a2b/ : FAILURE in 1h 03m 46s (non-voting)\n- octavia-grenade http://logs.openstack.org/91/662791/1/check/octavia-grenade/9c8137d/ : FAILURE in 45m 47s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/91/662791/1/check/octavia-v2-dsvm-tls-barbican/270572b/ : FAILURE in 49m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/91/662791/1/check/octavia-v2-dsvm-scenario-amphora-v2/0078a75/ : FAILURE in 49m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db19291de5a1cc8cce77240bc6830956d8ac70e7","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-06-05 12:55:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cad2608acdaad3ebcc6e61e64bdb280303a935ee","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-06-05 12:57:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"66f0766d96aa15301679e643af4cb50d7f02fc9f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-06-05 13:22:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"86be69825002268b434f3241a1284cd7913c93f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 15:38:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/91/662791/4/check/openstack-tox-cover/cc1c162/cover/ : SUCCESS in 7m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662791/4/check/openstack-tox-lower-constraints/f4a90e7/ : SUCCESS in 6m 22s\n- openstack-tox-pep8 http://logs.openstack.org/91/662791/4/check/openstack-tox-pep8/0f5c921/ : FAILURE in 5m 12s\n- openstack-tox-py27 http://logs.openstack.org/91/662791/4/check/openstack-tox-py27/d2b8e84/ : SUCCESS in 5m 04s\n- openstack-tox-py36 http://logs.openstack.org/91/662791/4/check/openstack-tox-py36/6d38e70/ : SUCCESS in 5m 19s\n- openstack-tox-py37 http://logs.openstack.org/91/662791/4/check/openstack-tox-py37/6d9a232/ : SUCCESS in 5m 45s\n- openstack-tox-docs http://logs.openstack.org/91/662791/4/check/openstack-tox-docs/8802da6/html/ : SUCCESS in 6m 29s\n- openstack-tox-functional http://logs.openstack.org/91/662791/4/check/openstack-tox-functional/6f27074/ : SUCCESS in 9m 00s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/662791/4/check/openstack-tox-functional-py36/a1d788b/ : SUCCESS in 6m 16s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/91/662791/4/check/octavia-v2-dsvm-noop-api/971c629/ : SUCCESS in 41m 58s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/91/662791/4/check/octavia-v2-dsvm-noop-py2-api/a56a380/ : SUCCESS in 42m 05s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/91/662791/4/check/octavia-v2-dsvm-scenario/9aca999/ : SUCCESS in 1h 37m 49s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/91/662791/4/check/octavia-v2-dsvm-py2-scenario/33c45bc/ : SUCCESS in 1h 35m 54s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/91/662791/4/check/octavia-v2-dsvm-py2-scenario-centos-7/6f65342/ : FAILURE in 1h 58m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/91/662791/4/check/octavia-v2-dsvm-scenario-ubuntu-bionic/e14fffb/ : SUCCESS in 1h 40m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/91/662791/4/check/octavia-v2-dsvm-scenario-two-node/b5a4742/ : SUCCESS in 2h 03m 31s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/91/662791/4/check/octavia-v2-dsvm-py2-scenario-two-node/bd2dde5/ : SUCCESS in 1h 36m 07s (non-voting)\n- octavia-grenade http://logs.openstack.org/91/662791/4/check/octavia-grenade/eded2b5/ : SUCCESS in 1h 05m 32s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/91/662791/4/check/octavia-v2-dsvm-tls-barbican/8ad8424/ : SUCCESS in 1h 19m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/91/662791/4/check/octavia-v2-dsvm-scenario-amphora-v2/e48371e/ : FAILURE in 54m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"be0a4cef6369cd9e539a2f212112b8dc36c154ab","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-06-05 16:05:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b1ce0f617e47c7176a852809223b39c1d14de8c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 18:43:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/91/662791/5/check/openstack-tox-cover/f1bdf0a/cover/ : SUCCESS in 7m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662791/5/check/openstack-tox-lower-constraints/69312c3/ : SUCCESS in 8m 36s\n- openstack-tox-pep8 http://logs.openstack.org/91/662791/5/check/openstack-tox-pep8/9a027b8/ : SUCCESS in 7m 30s\n- openstack-tox-py27 http://logs.openstack.org/91/662791/5/check/openstack-tox-py27/4f0156a/ : SUCCESS in 6m 03s\n- openstack-tox-py36 http://logs.openstack.org/91/662791/5/check/openstack-tox-py36/f9dc114/ : SUCCESS in 5m 38s\n- openstack-tox-py37 http://logs.openstack.org/91/662791/5/check/openstack-tox-py37/7fb6daf/ : SUCCESS in 7m 28s\n- openstack-tox-docs http://logs.openstack.org/91/662791/5/check/openstack-tox-docs/3606253/html/ : SUCCESS in 6m 51s\n- openstack-tox-functional http://logs.openstack.org/91/662791/5/check/openstack-tox-functional/c40cf0e/ : SUCCESS in 6m 19s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/662791/5/check/openstack-tox-functional-py36/06f70a2/ : SUCCESS in 6m 06s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/91/662791/5/check/octavia-v2-dsvm-noop-api/260a9c4/ : SUCCESS in 42m 23s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/91/662791/5/check/octavia-v2-dsvm-noop-py2-api/73b1c0f/ : SUCCESS in 42m 45s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/91/662791/5/check/octavia-v2-dsvm-scenario/1626947/ : SUCCESS in 1h 45m 17s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/91/662791/5/check/octavia-v2-dsvm-py2-scenario/9222f7c/ : SUCCESS in 1h 30m 30s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/91/662791/5/check/octavia-v2-dsvm-py2-scenario-centos-7/2090819/ : FAILURE in 2h 09m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/91/662791/5/check/octavia-v2-dsvm-scenario-ubuntu-bionic/4654d70/ : SUCCESS in 1h 34m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/91/662791/5/check/octavia-v2-dsvm-scenario-two-node/9cd9349/ : FAILURE in 1h 36m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/91/662791/5/check/octavia-v2-dsvm-py2-scenario-two-node/626c00f/ : SUCCESS in 2h 04m 59s (non-voting)\n- octavia-grenade http://logs.openstack.org/91/662791/5/check/octavia-grenade/8944b5e/ : SUCCESS in 1h 18m 48s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/91/662791/5/check/octavia-v2-dsvm-tls-barbican/4a779f9/ : SUCCESS in 51m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/91/662791/5/check/octavia-v2-dsvm-scenario-amphora-v2/e446dd2/ : FAILURE in 48m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8357a7f37dc154ade2c66fb8becc78f91be20c30","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-06-06 13:15:48.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3bcb597bf9ad14887fa5aefc052e7f5f0b82c0f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 15:34:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/91/662791/6/check/openstack-tox-cover/b36ac18/cover/ : SUCCESS in 8m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662791/6/check/openstack-tox-lower-constraints/42316db/ : SUCCESS in 6m 52s\n- openstack-tox-pep8 http://logs.openstack.org/91/662791/6/check/openstack-tox-pep8/7cb9a08/ : SUCCESS in 5m 56s\n- openstack-tox-py27 http://logs.openstack.org/91/662791/6/check/openstack-tox-py27/eb29eeb/ : SUCCESS in 5m 20s\n- openstack-tox-py36 http://logs.openstack.org/91/662791/6/check/openstack-tox-py36/4209e86/ : SUCCESS in 5m 11s\n- openstack-tox-py37 http://logs.openstack.org/91/662791/6/check/openstack-tox-py37/99d66de/ : SUCCESS in 5m 37s\n- openstack-tox-docs http://logs.openstack.org/91/662791/6/check/openstack-tox-docs/121717b/html/ : SUCCESS in 5m 57s\n- openstack-tox-functional http://logs.openstack.org/91/662791/6/check/openstack-tox-functional/942ce68/ : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/662791/6/check/openstack-tox-functional-py36/f60d06d/ : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/91/662791/6/check/octavia-v2-dsvm-noop-api/cfeb7ee/ : SUCCESS in 39m 35s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/91/662791/6/check/octavia-v2-dsvm-noop-py2-api/84c98bf/ : SUCCESS in 41m 57s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/91/662791/6/check/octavia-v2-dsvm-scenario/b5eaa3f/ : SUCCESS in 1h 21m 47s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/91/662791/6/check/octavia-v2-dsvm-py2-scenario/98987ba/ : SUCCESS in 1h 32m 26s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/91/662791/6/check/octavia-v2-dsvm-py2-scenario-centos-7/5bf57b2/ : FAILURE in 1h 55m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/91/662791/6/check/octavia-v2-dsvm-scenario-ubuntu-bionic/5d9bcb7/ : SUCCESS in 1h 47m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/91/662791/6/check/octavia-v2-dsvm-scenario-two-node/b5710a8/ : FAILURE in 1h 43m 34s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/91/662791/6/check/octavia-v2-dsvm-py2-scenario-two-node/c83da51/ : SUCCESS in 1h 52m 59s (non-voting)\n- octavia-grenade http://logs.openstack.org/91/662791/6/check/octavia-grenade/6892bc3/ : SUCCESS in 1h 42m 08s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/91/662791/6/check/octavia-v2-dsvm-tls-barbican/637cfca/ : SUCCESS in 51m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/91/662791/6/check/octavia-v2-dsvm-scenario-amphora-v2/07ee23f/ : FAILURE in 49m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7bd0467782d50401624bed8f571470b93fb914c1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-07 20:45:52.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"97a1ca49949a32cba8b17a0bd0d20ae86a4f34a3","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-06-10 08:35:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"89799c63448fe4d3acac45330892df351ddc08e5","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-06-10 09:51:19.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b1f3fb1b40a6d24e5e186a3b80f7faf0589895c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 11:57:22.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/91/662791/8/check/openstack-tox-cover/f495b33/cover/ : SUCCESS in 9m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662791/8/check/openstack-tox-lower-constraints/9ce9395/ : SUCCESS in 7m 36s\n- openstack-tox-pep8 http://logs.openstack.org/91/662791/8/check/openstack-tox-pep8/49e085b/ : SUCCESS in 6m 51s\n- openstack-tox-py27 http://logs.openstack.org/91/662791/8/check/openstack-tox-py27/3bb3f47/ : SUCCESS in 5m 39s\n- openstack-tox-py36 http://logs.openstack.org/91/662791/8/check/openstack-tox-py36/4bafb28/ : SUCCESS in 5m 18s\n- openstack-tox-py37 http://logs.openstack.org/91/662791/8/check/openstack-tox-py37/5b43a34/ : SUCCESS in 5m 38s\n- openstack-tox-docs http://logs.openstack.org/91/662791/8/check/openstack-tox-docs/303d33b/html/ : SUCCESS in 6m 22s\n- openstack-tox-functional http://logs.openstack.org/91/662791/8/check/openstack-tox-functional/b42a678/ : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/662791/8/check/openstack-tox-functional-py36/f64ae4f/ : SUCCESS in 6m 23s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/91/662791/8/check/octavia-v2-dsvm-noop-api/c40416c/ : SUCCESS in 43m 31s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/91/662791/8/check/octavia-v2-dsvm-noop-py2-api/5b9ad8a/ : SUCCESS in 42m 55s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/91/662791/8/check/octavia-v2-dsvm-scenario/c616b0a/ : SUCCESS in 1h 41m 15s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/91/662791/8/check/octavia-v2-dsvm-py2-scenario/2d55396/ : SUCCESS in 1h 44m 59s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/91/662791/8/check/octavia-v2-dsvm-py2-scenario-centos-7/80bc767/ : FAILURE in 2h 00m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/91/662791/8/check/octavia-v2-dsvm-scenario-ubuntu-bionic/7b51043/ : SUCCESS in 1h 49m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/91/662791/8/check/octavia-v2-dsvm-scenario-two-node/355a19a/ : SUCCESS in 1h 39m 34s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/91/662791/8/check/octavia-v2-dsvm-py2-scenario-two-node/823fd4b/ : SUCCESS in 1h 43m 52s (non-voting)\n- octavia-grenade http://logs.openstack.org/91/662791/8/check/octavia-grenade/d5e6cd2/ : SUCCESS in 1h 19m 42s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/91/662791/8/check/octavia-v2-dsvm-tls-barbican/811b10b/ : SUCCESS in 57m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/91/662791/8/check/octavia-v2-dsvm-scenario-amphora-v2/c4200d6/ : SUCCESS in 1h 30m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"aab699ebdb696a4e7beda4024afe74105baa5188","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-06-10 15:22:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"10f5f04e41b57cc3050c7dfe4ff0dab53b89d351","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:50:07.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/91/662791/9/check/openstack-tox-cover/169fe23/cover/ : SUCCESS in 10m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662791/9/check/openstack-tox-lower-constraints/123be5b/ : SUCCESS in 7m 46s\n- openstack-tox-pep8 http://logs.openstack.org/91/662791/9/check/openstack-tox-pep8/47f261a/ : SUCCESS in 6m 07s\n- openstack-tox-py27 http://logs.openstack.org/91/662791/9/check/openstack-tox-py27/95406cb/ : SUCCESS in 7m 19s\n- openstack-tox-py36 http://logs.openstack.org/91/662791/9/check/openstack-tox-py36/9b5bfef/ : SUCCESS in 5m 19s\n- openstack-tox-py37 http://logs.openstack.org/91/662791/9/check/openstack-tox-py37/2d332b5/ : SUCCESS in 6m 33s\n- openstack-tox-docs http://logs.openstack.org/91/662791/9/check/openstack-tox-docs/fe62a14/html/ : SUCCESS in 6m 00s\n- openstack-tox-functional http://logs.openstack.org/91/662791/9/check/openstack-tox-functional/3f62641/ : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/662791/9/check/openstack-tox-functional-py36/89bd232/ : SUCCESS in 8m 49s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/91/662791/9/check/octavia-v2-dsvm-noop-api/1cfce2e/ : SUCCESS in 47m 10s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/91/662791/9/check/octavia-v2-dsvm-noop-py2-api/906c819/ : SUCCESS in 48m 37s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/91/662791/9/check/octavia-v2-dsvm-scenario/9e93cfb/ : SUCCESS in 1h 35m 07s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/91/662791/9/check/octavia-v2-dsvm-py2-scenario/651b4f3/ : SUCCESS in 1h 36m 49s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/91/662791/9/check/octavia-v2-dsvm-py2-scenario-centos-7/8da4ccd/ : FAILURE in 2h 02m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/91/662791/9/check/octavia-v2-dsvm-scenario-ubuntu-bionic/bc8e781/ : SUCCESS in 1h 37m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/91/662791/9/check/octavia-v2-dsvm-scenario-two-node/2591825/ : SUCCESS in 1h 53m 01s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/91/662791/9/check/octavia-v2-dsvm-py2-scenario-two-node/4762fe8/ : SUCCESS in 1h 52m 52s (non-voting)\n- octavia-grenade http://logs.openstack.org/91/662791/9/check/octavia-grenade/12207a2/ : SUCCESS in 1h 22m 08s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/91/662791/9/check/octavia-v2-dsvm-tls-barbican/825ccb7/ : SUCCESS in 52m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/91/662791/9/check/octavia-v2-dsvm-scenario-amphora-v2/cab6d5a/ : SUCCESS in 1h 32m 12s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"90af6760b0112e22197e2fbdcf4910d8bab3602c","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-06-11 05:12:47.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"650455f6bfd7c6bf24d85763464c344d8d7124e7","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-06-11 07:16:47.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"f6a5c545fca2081cad3bb0ec8d569e78a6e1cf1b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-06-11 08:00:33.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"71e019d16afbc6a259bb7e96388a0cd9f8aef4a0","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-06-11 08:19:19.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"d8ed46a7f9178876636fdbb7f51010326bda2b63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 10:50:40.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/91/662791/11/check/openstack-tox-cover/7eac781/ : FAILURE in 5m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662791/11/check/openstack-tox-lower-constraints/cb0a60c/ : FAILURE in 5m 25s\n- openstack-tox-pep8 http://logs.openstack.org/91/662791/11/check/openstack-tox-pep8/ed58857/ : SUCCESS in 6m 38s\n- openstack-tox-py27 http://logs.openstack.org/91/662791/11/check/openstack-tox-py27/223157c/ : FAILURE in 5m 05s\n- openstack-tox-py36 http://logs.openstack.org/91/662791/11/check/openstack-tox-py36/b30d96d/ : FAILURE in 5m 39s\n- openstack-tox-py37 http://logs.openstack.org/91/662791/11/check/openstack-tox-py37/105905f/ : FAILURE in 5m 55s\n- openstack-tox-docs http://logs.openstack.org/91/662791/11/check/openstack-tox-docs/0ea5726/html/ : SUCCESS in 6m 28s\n- openstack-tox-functional http://logs.openstack.org/91/662791/11/check/openstack-tox-functional/a63b8e8/ : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/662791/11/check/openstack-tox-functional-py36/3e8c0ad/ : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/91/662791/11/check/octavia-v2-dsvm-noop-api/08e0226/ : SUCCESS in 39m 41s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/91/662791/11/check/octavia-v2-dsvm-noop-py2-api/39f4c97/ : SUCCESS in 47m 39s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/91/662791/11/check/octavia-v2-dsvm-scenario/ad08d52/ : SUCCESS in 1h 29m 18s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/91/662791/11/check/octavia-v2-dsvm-py2-scenario/dfa9bb4/ : SUCCESS in 1h 28m 48s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/91/662791/11/check/octavia-v2-dsvm-py2-scenario-centos-7/d5c96d4/ : FAILURE in 2h 04m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/91/662791/11/check/octavia-v2-dsvm-scenario-ubuntu-bionic/948e93f/ : FAILURE in 2h 29m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/91/662791/11/check/octavia-v2-dsvm-scenario-two-node/c3aa649/ : SUCCESS in 1h 23m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/91/662791/11/check/octavia-v2-dsvm-py2-scenario-two-node/85be8f1/ : FAILURE in 1h 44m 34s (non-voting)\n- octavia-grenade http://logs.openstack.org/91/662791/11/check/octavia-grenade/1bbc5ea/ : SUCCESS in 1h 10m 06s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/91/662791/11/check/octavia-v2-dsvm-tls-barbican/fb78466/ : SUCCESS in 52m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/91/662791/11/check/octavia-v2-dsvm-scenario-amphora-v2/caa73e8/ : SUCCESS in 1h 32m 35s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"af7a58eee855a99dd3b5f010c5c8feb2d8564065","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-06-11 14:08:05.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"d39a456e6e3e497520eba27f2abeacf6a77cd65d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 16:43:49.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/91/662791/12/check/openstack-tox-cover/e3de2ad/cover/ : SUCCESS in 7m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662791/12/check/openstack-tox-lower-constraints/521eb8d/ : SUCCESS in 9m 02s\n- openstack-tox-pep8 http://logs.openstack.org/91/662791/12/check/openstack-tox-pep8/75b448c/ : SUCCESS in 6m 31s\n- openstack-tox-py27 http://logs.openstack.org/91/662791/12/check/openstack-tox-py27/d3a3eb5/ : SUCCESS in 5m 45s\n- openstack-tox-py36 http://logs.openstack.org/91/662791/12/check/openstack-tox-py36/5107391/ : SUCCESS in 6m 17s\n- openstack-tox-py37 http://logs.openstack.org/91/662791/12/check/openstack-tox-py37/9f2d2e9/ : SUCCESS in 6m 45s\n- openstack-tox-docs http://logs.openstack.org/91/662791/12/check/openstack-tox-docs/b253dd8/html/ : SUCCESS in 6m 18s\n- openstack-tox-functional http://logs.openstack.org/91/662791/12/check/openstack-tox-functional/173995f/ : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/662791/12/check/openstack-tox-functional-py36/d00583a/ : SUCCESS in 8m 26s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/91/662791/12/check/octavia-v2-dsvm-noop-api/90113a3/ : SUCCESS in 37m 06s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/91/662791/12/check/octavia-v2-dsvm-noop-py2-api/d025396/ : SUCCESS in 50m 39s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/91/662791/12/check/octavia-v2-dsvm-scenario/87ad728/ : SUCCESS in 1h 18m 41s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/91/662791/12/check/octavia-v2-dsvm-py2-scenario/bf4dd81/ : SUCCESS in 1h 32m 55s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/91/662791/12/check/octavia-v2-dsvm-py2-scenario-centos-7/56e0daa/ : FAILURE in 2h 22m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/91/662791/12/check/octavia-v2-dsvm-scenario-ubuntu-bionic/ce43dc8/ : SUCCESS in 1h 35m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/91/662791/12/check/octavia-v2-dsvm-scenario-two-node/78e63da/ : SUCCESS in 1h 43m 43s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/91/662791/12/check/octavia-v2-dsvm-py2-scenario-two-node/17065e9/ : SUCCESS in 1h 50m 43s (non-voting)\n- octavia-grenade http://logs.openstack.org/91/662791/12/check/octavia-grenade/decca0a/ : SUCCESS in 1h 12m 07s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/91/662791/12/check/octavia-v2-dsvm-tls-barbican/bdb9539/ : SUCCESS in 54m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/91/662791/12/check/octavia-v2-dsvm-scenario-amphora-v2/2d9ab6d/ : SUCCESS in 1h 25m 41s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"9d63644c8788272a7cac6666ae90dfccdf6ab4c0","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-06-13 04:58:57.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"04c930587522ee3648db94926b4c7203b58783c6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-18 09:14:37.000000000","message":"Patch Set 12: Code-Review-1\n\n(3 comments)\n\nJust one minor formatting thing and then I think this looks fine.","accounts_in_message":[],"_revision_number":12},{"id":"936e207d9421dd03cc09efa17117df7626c8f45b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-06-18 11:33:52.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"6438cbe02c8468f2d006447ba75b90022348368b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-06-18 11:41:18.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"bf3c48d69ac3e84efd7b25c176335cd5aeb716c4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-18 12:35:06.000000000","message":"Patch Set 13: Code-Review+2\n\nlooks fine to me.","accounts_in_message":[],"_revision_number":13},{"id":"30f36eeeb3135d916bbc91ddfbc12b3d4655d416","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 14:41:08.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/91/662791/13/check/openstack-tox-cover/dad91cf/cover/ : SUCCESS in 11m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662791/13/check/openstack-tox-lower-constraints/b9cc1a6/ : SUCCESS in 7m 20s\n- openstack-tox-pep8 http://logs.openstack.org/91/662791/13/check/openstack-tox-pep8/8e0afa0/ : SUCCESS in 8m 12s\n- openstack-tox-py27 http://logs.openstack.org/91/662791/13/check/openstack-tox-py27/1cbea84/ : SUCCESS in 8m 40s\n- openstack-tox-py36 http://logs.openstack.org/91/662791/13/check/openstack-tox-py36/0a89bd4/ : SUCCESS in 5m 33s\n- openstack-tox-py37 http://logs.openstack.org/91/662791/13/check/openstack-tox-py37/7a1d923/ : SUCCESS in 7m 38s\n- openstack-tox-docs http://logs.openstack.org/91/662791/13/check/openstack-tox-docs/a4bd108/html/ : SUCCESS in 6m 28s\n- openstack-tox-functional http://logs.openstack.org/91/662791/13/check/openstack-tox-functional/e20a18c/ : SUCCESS in 7m 57s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/662791/13/check/openstack-tox-functional-py36/73d456c/ : SUCCESS in 6m 52s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/91/662791/13/check/octavia-v2-dsvm-noop-api/4e3be3a/ : SUCCESS in 42m 50s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/91/662791/13/check/octavia-v2-dsvm-noop-py2-api/1833201/ : SUCCESS in 46m 29s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/91/662791/13/check/octavia-v2-dsvm-scenario/9e4df1c/ : SUCCESS in 1h 37m 54s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/91/662791/13/check/octavia-v2-dsvm-py2-scenario/8deecd9/ : SUCCESS in 1h 27m 40s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/91/662791/13/check/octavia-v2-dsvm-py2-scenario-centos-7/4aa17fc/ : FAILURE in 2h 02m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/91/662791/13/check/octavia-v2-dsvm-scenario-ubuntu-bionic/556e79e/ : SUCCESS in 1h 44m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/91/662791/13/check/octavia-v2-dsvm-scenario-two-node/3be36f9/ : FAILURE in 1h 45m 06s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/91/662791/13/check/octavia-v2-dsvm-py2-scenario-two-node/13d6774/ : SUCCESS in 1h 46m 56s (non-voting)\n- octavia-grenade http://logs.openstack.org/91/662791/13/check/octavia-grenade/d7255bf/ : SUCCESS in 1h 14m 16s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/91/662791/13/check/octavia-v2-dsvm-tls-barbican/be94464/ : SUCCESS in 54m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/91/662791/13/check/octavia-v2-dsvm-scenario-amphora-v2/2499883/ : SUCCESS in 1h 29m 08s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"196c7a5816981e02c85ade118a7b1cb07eab6959","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-06-19 06:55:37.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"f3c557f80361cd4988883b5515b0790e2e19be37","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-11 20:13:27.000000000","message":"Patch Set 13: Code-Review-1\n\n(4 comments)\n\nThis looks great. My only ask is that we change the sub-flow names to not conflict with the parent/other flow names.","accounts_in_message":[],"_revision_number":13},{"id":"97b5ceb4e0aa7d1d255f90afe66111648e6e41e1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-11 20:51:58.000000000","message":"Patch Set 13:\n\n(3 comments)\n\nActually, this has more issues:\nhttp://logs.openstack.org/91/662791/13/check/octavia-v2-dsvm-scenario-amphora-v2/2499883/controller/logs/screen-o-cw.txt.gz#_Jun_18_12_40_59_773645\nIs there a way we can not log these as ERRORs?\nReally I don\u0027t think this should be reverting, maybe a revert method on a task needs to be updated?","accounts_in_message":[],"_revision_number":13},{"id":"a181b3ac57dccc1a67f86ff8ce836dcdc082677c","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-07-12 13:01:44.000000000","message":"Patch Set 13:\n\n(7 comments)\n\nI was looking to how not to trace all of that as error, but didn\u0027t find how. As it is throwing an exception and then it is retried the task... May be taskflow should be expanded for such cases.\n\nActually the description of RETRY is\n``RETRY``: when the controlling flow must be reverted and restarted again  https://github.com/openstack/taskflow/blob/master/taskflow/retry.py#L219\nSo, it is expected that it is shown as reverted and then restarted.\nThere is no option without revert to do retry...","accounts_in_message":[],"_revision_number":13},{"id":"deada78068d47d91fd2f78d09791a41ac93f3467","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-07-12 13:16:43.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"20d79763349b93111e93220c96721c25c27f8347","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 15:26:08.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/91/662791/14/check/openstack-tox-cover/e072b58/cover/ : SUCCESS in 6m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662791/14/check/openstack-tox-lower-constraints/ad08ab2/ : SUCCESS in 7m 03s\n- openstack-tox-pep8 http://logs.openstack.org/91/662791/14/check/openstack-tox-pep8/6a9f7b9/ : SUCCESS in 5m 24s\n- openstack-tox-py27 http://logs.openstack.org/91/662791/14/check/openstack-tox-py27/139107e/ : SUCCESS in 4m 12s\n- openstack-tox-py36 http://logs.openstack.org/91/662791/14/check/openstack-tox-py36/0534b19/ : SUCCESS in 3m 44s\n- openstack-tox-py37 http://logs.openstack.org/91/662791/14/check/openstack-tox-py37/f16a3ca/ : SUCCESS in 4m 45s\n- openstack-tox-docs http://logs.openstack.org/91/662791/14/check/openstack-tox-docs/b0bf2f5/html/ : SUCCESS in 5m 32s\n- openstack-tox-functional http://logs.openstack.org/91/662791/14/check/openstack-tox-functional/d4399de/ : SUCCESS in 5m 31s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/662791/14/check/openstack-tox-functional-py36/92f1cd7/ : SUCCESS in 5m 13s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/91/662791/14/check/octavia-v2-dsvm-noop-api/8b7dc74/ : SUCCESS in 41m 44s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/91/662791/14/check/octavia-v2-dsvm-noop-py2-api/4589d10/ : SUCCESS in 41m 09s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/91/662791/14/check/octavia-v2-dsvm-scenario/07ea389/ : SUCCESS in 1h 30m 27s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/91/662791/14/check/octavia-v2-dsvm-py2-scenario/98098c4/ : SUCCESS in 1h 29m 09s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/91/662791/14/check/octavia-v2-dsvm-py2-scenario-centos-7/c3d9c2c/ : FAILURE in 2h 07m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/91/662791/14/check/octavia-v2-dsvm-scenario-ubuntu-bionic/7f6831c/ : SUCCESS in 1h 43m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/91/662791/14/check/octavia-v2-dsvm-scenario-two-node/4cc79d9/ : SUCCESS in 1h 13m 05s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/91/662791/14/check/octavia-v2-dsvm-py2-scenario-two-node/562036a/ : SUCCESS in 1h 41m 22s (non-voting)\n- octavia-grenade http://logs.openstack.org/91/662791/14/check/octavia-grenade/3a9afac/ : SUCCESS in 1h 19m 05s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/91/662791/14/check/octavia-v2-dsvm-tls-barbican/1c9c169/ : SUCCESS in 1h 03m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/91/662791/14/check/octavia-v2-dsvm-scenario-amphora-v2/1d781b7/ : SUCCESS in 1h 38m 13s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"8dd24cad4595e828cb53e0627eb0e518467c3ce7","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-07-25 12:21:16.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"1d618087239bb9c52e722014f2c9df280d597936","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-07-26 11:39:29.000000000","message":"Patch Set 14: Workflow-1\n\nThe problem that I described in inline comment now I can describe as a blocker for using taskflow retry mechanism and using Redis backend as a result.\n\nI filed a bug for taskflow about this issue https://bugs.launchpad.net/taskflow/+bug/1838015\n\nI will mark this change as WIP for now and will continue development using zookeper backend. On the next team meeting will discuss this issue more.","accounts_in_message":[],"_revision_number":14},{"id":"4856dfe93796137be6f6ed09608665125ee6632e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-08-08 08:06:43.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"e97a54cf346e4110fb2ad6d7ea8a91d4576a640e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-08-08 10:37:54.000000000","message":"Patch Set 15: Workflow-1","accounts_in_message":[],"_revision_number":15},{"id":"6984a9d759210f57ca6a35f7be43bb670a779ce3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 10:47:09.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/91/662791/15/check/openstack-tox-cover/ce5d6af/cover/ : SUCCESS in 7m 53s\n- openstack-tox-lower-constraints https://logs.opendev.org/91/662791/15/check/openstack-tox-lower-constraints/18027b8/ : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://logs.opendev.org/91/662791/15/check/openstack-tox-pep8/d296062/ : FAILURE in 4m 41s\n- openstack-tox-py27 https://logs.opendev.org/91/662791/15/check/openstack-tox-py27/a6e90da/ : SUCCESS in 5m 41s\n- openstack-tox-py36 https://logs.opendev.org/91/662791/15/check/openstack-tox-py36/1fe956d/ : SUCCESS in 5m 24s\n- openstack-tox-py37 https://logs.opendev.org/91/662791/15/check/openstack-tox-py37/c0d6db9/ : SUCCESS in 5m 46s\n- openstack-tox-docs https://logs.opendev.org/91/662791/15/check/openstack-tox-docs/f522ec5/html/ : SUCCESS in 6m 19s\n- openstack-tox-functional https://logs.opendev.org/91/662791/15/check/openstack-tox-functional/0a0ad6d/ : SUCCESS in 5m 07s\n- openstack-tox-functional-py36 https://logs.opendev.org/91/662791/15/check/openstack-tox-functional-py36/16571ab/ : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/91/662791/15/check/octavia-v2-dsvm-noop-api/5348a9e/ : SUCCESS in 45m 54s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/91/662791/15/check/octavia-v2-dsvm-noop-py2-api/33c4cac/ : SUCCESS in 44m 10s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/91/662791/15/check/octavia-v2-dsvm-scenario/0f569f0/ : FAILURE in 1h 17m 29s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/91/662791/15/check/octavia-v2-dsvm-py2-scenario/4b80a47/ : FAILURE in 1h 20m 06s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/91/662791/15/check/octavia-v2-dsvm-py2-scenario-centos-7/a34934d/ : FAILURE in 1h 54m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/91/662791/15/check/octavia-v2-dsvm-scenario-two-node/a3c5fdf/ : FAILURE in 1h 42m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/91/662791/15/check/octavia-v2-dsvm-py2-scenario-two-node/f5e4045/ : FAILURE in 27m 27s (non-voting)\n- octavia-grenade https://logs.opendev.org/91/662791/15/check/octavia-grenade/3fbd978/ : SUCCESS in 1h 19m 15s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/91/662791/15/check/octavia-v2-dsvm-tls-barbican/275de97/ : FAILURE in 56m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/91/662791/15/check/octavia-v2-dsvm-scenario-amphora-v2/e4a36cd/ : TIMED_OUT in 2h 33m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/91/662791/15/check/octavia-v2-act-stdby-iptables-dsvm-scenario/c776656/ : FAILURE in 48m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/91/662791/15/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/33b10c6/ : FAILURE in 48m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/91/662791/15/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/d051c8c/ : FAILURE in 1h 01m 29s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"05875cd2d4c90975303e06f1a723284035fde4ef","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-08-08 13:27:26.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"2eec4526d85fd9bccc5790ce3a5f498c97ae4dfc","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-08-08 13:27:55.000000000","message":"Patch Set 16: Workflow-1","accounts_in_message":[],"_revision_number":16},{"id":"918899c28ee077d7119615596d374be87b5ef4bb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-08 15:50:38.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"3764decaed074b64bde304fafe1118a9224f50c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 15:52:03.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/91/662791/16/check/openstack-tox-cover/307857d/cover/ : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints https://logs.opendev.org/91/662791/16/check/openstack-tox-lower-constraints/693d6b0/ : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://logs.opendev.org/91/662791/16/check/openstack-tox-pep8/25e2f81/ : SUCCESS in 5m 24s\n- openstack-tox-py27 https://logs.opendev.org/91/662791/16/check/openstack-tox-py27/2498134/ : SUCCESS in 5m 23s\n- openstack-tox-py36 https://logs.opendev.org/91/662791/16/check/openstack-tox-py36/0101323/ : SUCCESS in 6m 50s\n- openstack-tox-py37 https://logs.opendev.org/91/662791/16/check/openstack-tox-py37/896f51f/ : SUCCESS in 7m 36s\n- openstack-tox-docs https://logs.opendev.org/91/662791/16/check/openstack-tox-docs/a24513c/html/ : SUCCESS in 6m 49s\n- openstack-tox-functional https://logs.opendev.org/91/662791/16/check/openstack-tox-functional/7a355e9/ : SUCCESS in 8m 16s\n- openstack-tox-functional-py36 https://logs.opendev.org/91/662791/16/check/openstack-tox-functional-py36/39f8c42/ : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/91/662791/16/check/octavia-v2-dsvm-noop-api/9357018/ : SUCCESS in 1h 06m 54s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/91/662791/16/check/octavia-v2-dsvm-noop-py2-api/142fe60/ : SUCCESS in 42m 16s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/91/662791/16/check/octavia-v2-dsvm-scenario/3a28a50/ : SUCCESS in 1h 44m 30s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/91/662791/16/check/octavia-v2-dsvm-py2-scenario/b2f992f/ : SUCCESS in 1h 46m 20s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/91/662791/16/check/octavia-v2-dsvm-py2-scenario-centos-7/c22e561/ : SUCCESS in 2h 05m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/91/662791/16/check/octavia-v2-dsvm-scenario-two-node/432a52e/ : FAILURE in 1h 31m 08s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/91/662791/16/check/octavia-v2-dsvm-py2-scenario-two-node/d474a20/ : SUCCESS in 1h 59m 51s (non-voting)\n- octavia-grenade https://logs.opendev.org/91/662791/16/check/octavia-grenade/02b0d28/ : SUCCESS in 1h 40m 18s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/91/662791/16/check/octavia-v2-dsvm-tls-barbican/ef71942/ : SUCCESS in 53m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/91/662791/16/check/octavia-v2-dsvm-scenario-amphora-v2/5fc6c23/ : SUCCESS in 1h 25m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/91/662791/16/check/octavia-v2-act-stdby-iptables-dsvm-scenario/941e0ca/ : SUCCESS in 1h 07m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/91/662791/16/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/336f18d/ : FAILURE in 1h 00m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/91/662791/16/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/d321266/ : SUCCESS in 1h 21m 05s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"3126edcb762bec1db55584bc65a6d6f063f1dbd0","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-08-21 06:53:15.000000000","message":"Patch Set 16: -Workflow\n\nThis change is not blocked from review and merging now, as issue was fixed on the taskflow side.\nAs currently we do not store any tasks details it is safe to merge as it is. Newer taskflow version will be required for persistence enablement in upcoming changes.","accounts_in_message":[],"_revision_number":16},{"id":"c8850634ffa2724fb199eeb3db64e0562010a312","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-09-12 10:51:15.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"ef984ec17b60b9b5d664b7e7bbc9740c65c773ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 14:10:13.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a35ee4f9f074142815819b7698518be : SUCCESS in 8m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c1509a4b7d340c282f7921bbca6489d : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04cf2a4150e440d285b656d3d82a008d : FAILURE in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f803bf0310d454f8953ed69aa72b909 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ec970bfb4984a12b37044c2c0020bb9 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f05d498da4645d586d11b31eea94894 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a532a6f1c244d86b9f8c1e4a2b74c32 : SUCCESS in 8m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/584872f530bf42f288810c2ee36013c5 : SUCCESS in 6m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c57dd925b6ec43708c1f149f6fc2470f : SUCCESS in 6m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2bff7f28f1934e4fba0ac848657ea9a7 : SUCCESS in 5m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1ce5b45bba194d899d93fe1ba046cf8c : SUCCESS in 5m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f53674f5559b4960aef44cca7a48d2b5 : SUCCESS in 54m 27s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/31ea6f8d1e48437890206c8602a3945a : SUCCESS in 45m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/64ee32b4520548b4b2f7c7bec5a84c99 : SUCCESS in 1h 36m 58s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/289f3d0e57d744529d26d17cbb9d6889 : SUCCESS in 1h 44m 37s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/260ff4903090433a96510a95d3067796 : SUCCESS in 2h 06m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/75d70896499f45e4839b5834c819a12d : FAILURE in 1h 54m 05s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/d5f7d211201f4dcf85a385b963b00ccc : FAILURE in 1h 45m 29s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dd3f70fc303a4b59a551fcc7a4a1e0a2 : SUCCESS in 1h 35m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/542523a2daf341dbb8a9300a19e0221f : SUCCESS in 1h 11m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/efb1dfbd817a4ddbb7947624ce4ca548 : FAILURE in 1h 43m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e63eab4aa64b4e058f531fb98b868a99 : SUCCESS in 1h 04m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/67be3e8241f646d88e4e080534cf5267 : SUCCESS in 1h 12m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c574c4a1a3644338951612faca6fbf8f : FAILURE in 51m 15s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"68b0bac862dd4ff45e57b46f6f068cc9c11b3a45","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-09-18 14:34:35.000000000","message":"Patch Set 18: Patch Set 17 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"e45e3605b3e5550abeb983ac76c70bae71f6c67a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 18:41:00.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd0fced4ca214970aa6c938aeb6d16a9 : SUCCESS in 7m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/378e69d1e2154f8db521c8e4352ac9d5 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09a212900e5d4951bb96384c0d54e548 : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba9f9e82a8d4424f86b07d8af51d3244 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1958a3b18ea146129019c203bbb47e82 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c1b1f90cc7a434e8145f42749d9a639 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc58483b40342b7b152f41baad109b7 : SUCCESS in 7m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b528decdb8c246a783561cee11ceef82 : SUCCESS in 4m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a6d43f640d554df99cb8cbce511f4fec : SUCCESS in 5m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fae00c6f27d64863985f10c183361331 : SUCCESS in 5m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/55cf11aabe074722aa2d765bf358ee55 : SUCCESS in 5m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cd9f94cddb5b4d8eadafc5b6743cc743 : SUCCESS in 47m 57s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ec9cb1780ffe48498bf0c970ac1d50a3 : SUCCESS in 37m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e15118cc95884f35b2f93b6156ff4532 : SUCCESS in 1h 33m 42s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5433e35a92b14ee1a976f9c867414d5e : SUCCESS in 1h 53m 35s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/73ce1baf48ee40f485a9c627c7424214 : SUCCESS in 2h 05m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c2b9b4216cef4d2c9209ae3f9034e327 : SUCCESS in 1h 54m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/074f44eaa024471695465c956b8625a6 : TIMED_OUT in 2h 34m 27s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b24d16e1e0994b50a53037fd57def0b7 : SUCCESS in 1h 24m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b87a835222e14d599e5733fc7e210e6b : SUCCESS in 58m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3113b90d0afd47258548fa61a4f4451d : SUCCESS in 1h 50m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3957e5f5e5fe483c890dac3f35248dcc : SUCCESS in 1h 15m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c8439889fc3d45f8bcb95aefbf3fbe10 : SUCCESS in 1h 27m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/864d4e7d0cdb4c6586b057207020812f : SUCCESS in 1h 37m 00s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"34f44cb190ee7a95c59d1636833617901bd42338","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-10-17 10:54:23.000000000","message":"Patch Set 19: Commit message was updated.","accounts_in_message":[],"_revision_number":19},{"id":"afca7e322ed9a5836df93565a1dbd75df7cd63b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 13:39:24.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7acb8a3c6fba4049ba6ab8a783d9c69e : FAILURE in 3m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/842b2aba959048ec8d9059b41326ea34 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/479376bd19ff419b9688ac61900be2d3 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d89811d70c744516bc36af418108de7f : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcc2bf00356e43c0816f744d8e4a236a : FAILURE in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31950a503e474cd08c95b7888e726878 : FAILURE in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab1143987a2f4599a831bf279d371308 : SUCCESS in 10m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/51504a935f0248bc82dff6810d179af6 : FAILURE in 5m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8ca46c990e5a47aba981a1f74d262175 : SUCCESS in 6m 24s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a938a80d87854e40816ae6ee34bd34c8 : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/928d2146ee914dfc82a8ccac411762b0 : SUCCESS in 5m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/117baba7bd3a484884d3dd28597435e6 : SUCCESS in 45m 23s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4fe6c2b12b4540bf98177ffb226fb8c2 : SUCCESS in 43m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fdcf7933cd404df5847f90fd98476bcd : SUCCESS in 1h 35m 45s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/842e959eecb14482989645391a1f1eeb : TIMED_OUT in 2h 37m 34s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/efce29e4eb33430c89c21ff1dda38996 : SUCCESS in 2h 16m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9da83dd6ce8442c58253f54599ffdac6 : SUCCESS in 1h 54m 31s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0e90435bc9df433cb6aff82e44bf5ddf : TIMED_OUT in 2h 34m 16s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4810a5f7ccfc4b0588144f288812975f : SUCCESS in 1h 23m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e55f8fe29bae4059bd261840297dafd8 : SUCCESS in 50m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/525936fcc0e546dcb7356e72398f7d80 : SUCCESS in 1h 47m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/39861579de3949ab83abfccc2fd42c50 : SUCCESS in 1h 13m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6c04504b3079493498e15e1febbe0bbe : SUCCESS in 1h 12m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/cd0923c789c94cee9b78eaecb0fcf2bf : SUCCESS in 1h 28m 51s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"381a91018879cb12ef3104bca5a84a221f3eab76","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-10-22 09:51:49.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"4d54451757ec6ab75e701a96f18bd8fe1cbc0a58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 12:03:04.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99c2b18708bd4cb2872beb1df40fbb11 : SUCCESS in 8m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36cc4bca60234e05b9f4caec4341206b : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05410bb94b48401c8e91ac2c74279731 : SUCCESS in 6m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4f0adda12b6a480bbca50b67eea51e93 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/388b5eadcc4145e9abe5711e6fd31eea : SUCCESS in 6m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19a4ccccd6294d9185310602c1331ab4 : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a69c227a48c24ac2b1213ae2d1192e0c : SUCCESS in 16m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/db49711d843f4bc281b38ae11bc0dcf3 : SUCCESS in 8m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2dabaa58120b42298caef14630c09841 : SUCCESS in 10m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d7a837a20cb64e84a02cbeaf6b5f9053 : SUCCESS in 5m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/925314bbee1744018a6be9d7ad7112a6 : SUCCESS in 6m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/971a049bd8904721ac46ba497d4e7909 : SUCCESS in 49m 01s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0cca2ef377ea412193c1eeee6d67c61c : SUCCESS in 46m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f006a6ac0cb4292a1cfbba5ecca1690 : SUCCESS in 1h 36m 10s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/09d94c3a5b03498aa379934fb187ba53 : SUCCESS in 1h 34m 05s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/919e5d09ef3f457bb8c54119d9497d07 : SUCCESS in 2h 06m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fe098ada1d4445f79d50bbf967bd5a1d : SUCCESS in 2h 00m 17s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/10cca6f6d1034498b08a0d44a3c11d2a : SUCCESS in 1h 50m 21s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ffb4efa8bab14b5298d7b9f0c3857ab6 : SUCCESS in 1h 20m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b9a4381253b74b008e43f5f0f6c95d13 : SUCCESS in 54m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/29deb16bd036427f8112412398a4c198 : SUCCESS in 1h 30m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7967724652254d259cd880f3f61ac0fe : SUCCESS in 1h 05m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7f0c5b708d51477f9aaf31839b3e853a : SUCCESS in 1h 07m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3428476ca32a4484bf60261425af3073 : FAILURE in 22m 47s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"62a159a4df6404e0b28ef60d9793b52377f3e0ed","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-10-23 05:09:31.000000000","message":"Patch Set 20: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"464858efbf644e24a9fb772f32f108bdd7ca7bd2","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-10-23 12:41:38.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"6595860a6a40cf52099933c3820080503c7c7a89","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-10-23 12:41:59.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"11a987027073b86a853d6a769ae940fbf9a9d2a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 15:29:34.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3586f3c370ed4ae69ed0bc2c6375ffd1 : SUCCESS in 9m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6adb71cdcdf498ba7eca189b481ac3e : POST_FAILURE in 35m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97d3561a0ed542cfa2475c7e85bb0a31 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/76489dfca2994653b311f3de5c480eb3 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33e75b5ddbf64aae848e498e44e24490 : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bfcdd81aa00443c803d3311a4ad9f24 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b09a30c2b0d245419dfcddcdf68c1455 : SUCCESS in 11m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3023616edd254de59be42d2e9d2fb345 : SUCCESS in 6m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d831cbb47bd9446c9bd36d04e63bb968 : SUCCESS in 6m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/da19d530980e425295f537703c3a738f : SUCCESS in 5m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/419575331c614755b3277cc218621616 : SUCCESS in 7m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/04deac5b0d2b44a0b1d9b81fd191798c : SUCCESS in 55m 18s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c0bc0f2d9a694a81a5b5ef5d03898e16 : SUCCESS in 42m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2db478a4604845b48363894279297479 : SUCCESS in 1h 31m 46s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2bce07b72cb34b7a9a1603cd8c9e3027 : TIMED_OUT in 2h 34m 02s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/db1b9e5ffa3145289cb40b5905dbbb7d : SUCCESS in 2h 24m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5fafd7dd035a4d27bd6a24c760418778 : FAILURE in 1h 44m 28s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/9c407489737d493ba11752136a80eff0 : SUCCESS in 2h 03m 56s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5df3e88a8b9c48b39ccdb3f919310018 : SUCCESS in 1h 33m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f88a79cb3d9548adbf7b3a2ba3c08c0c : SUCCESS in 49m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ec209d29aaca47f2886b9cc6fa722b5a : SUCCESS in 1h 41m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee74db30e8d146e8b7eba52a396916f9 : SUCCESS in 1h 14m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/937144c0df184f3f8961e92a51480e97 : FAILURE in 1h 11m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9ecce3312d834f0da8016578d3e620a3 : SUCCESS in 1h 33m 08s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"ad7f0ba6c85a6623dbde89324bbfc89c9fceb85d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-10-23 15:59:49.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"757f5ea3e5a124f907f8973d7c8928f29f725f5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 18:41:45.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01bed446a4714019a3c807f2eac8f3ff : SUCCESS in 8m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b51c52e9daf43488295d9907c1452df : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53cd68070aa14e35975332085a118349 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7280daee7388445ab07a5df60dfb6283 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5583316024714511a8981848d958b9ae : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24e7a6a7a90148429094548d71300ede : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4692472aad464151bc9fda370a827e4b : SUCCESS in 9m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/790e57fda8d6450b99a9094df659eb87 : SUCCESS in 7m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5f24b121038b458f94fcff15419daea3 : SUCCESS in 8m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/465052168cfe4f8098df086eac2b50b8 : SUCCESS in 6m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dee435b7f2854b95887c289b2a28b159 : SUCCESS in 5m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/324f23b567b240cf8971eb64c986867a : SUCCESS in 45m 56s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/12ed70acf10a4833b65d4bdd2b33defb : SUCCESS in 48m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1d4cb2aaa604aeea913bb36c8b8be04 : SUCCESS in 1h 29m 19s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ad79fb64688443ee823615e23640eb5a : SUCCESS in 1h 54m 13s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/81e2a4e93c364fbc9f27180e17ccafdf : TIMED_OUT in 2h 37m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ca54a58e71034bc2b54a81d5ae2979d2 : SUCCESS in 1h 45m 34s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/87a809a9fe0f4ab6bc5f40065a00cbf0 : SUCCESS in 1h 15m 11s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cd5fabaff62f40b79565bc1dddfb80d9 : SUCCESS in 1h 25m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9394ea248d1a406286cd0331a8110a76 : SUCCESS in 52m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d26706819914470897c821408db30612 : SUCCESS in 1h 27m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67e14d3b3e234335b3851edab24e925d : SUCCESS in 1h 05m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/22cf95c6e19f47698e8824f989e9b50b : SUCCESS in 1h 00m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2dc610e2886a40df9e0f2587df4ff582 : SUCCESS in 1h 27m 21s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"5caabdb66fcfd41374599440810eafe80c34e952","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-10-31 06:59:20.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"e644315756b81073e0c9b58f47095fe0a1f7a45f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-11-25 18:03:53.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"cd25cb341ae3e12a92d2941e39bf07b93ae1c280","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 19:04:46.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08592999434245f0a5ae396845c71394 : SUCCESS in 8m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1643b0f8e3b4e48ba0e26ab915bb4fc : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2848ea48877248b697c6fa9b95ca8621 : SUCCESS in 7m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/180c26ef81b24cbe92d7de4bb1962b35 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59d532bfd11242b3ab38dd00b4870088 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/482a97c7919744cf916239400c84d01e : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4849659e2a6428ba3190cb6da791ab6 : SUCCESS in 11m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b9f2ce52d7ef423b90e589a8912c5077 : RETRY_LIMIT in 2m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e8692a0089d44220aa5ca2f007266a0c : SUCCESS in 8m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/309dd933cff04d52ba82461cdc70721d : RETRY_LIMIT in 3m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a94c9c02e78d4bb38268ae421798ca35 : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2dc3b01c0479440aa91ac1990d5a5b88 : SUCCESS in 44m 16s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/30ea5aba1a0443f4a990d3bd6dfdf803 : SUCCESS in 46m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3999454be3c2479f89d06295b5a31c19 : RETRY_LIMIT in 4m 03s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8e45fadebafd487f8465999fbab670e9 : FAILURE in 42m 23s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/28cf78619ca840fc9fba55d6d2eaa391 : FAILURE in 39m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5e5027603e624fccb52631c1aa2c0c2c : FAILURE in 33m 28s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/784b138fc83c4ca996308c83e62b95f7 : FAILURE in 35m 39s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4c6a67ff19744eafbc8965f3fa3a244a : FAILURE in 39m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7ac94583c3a14080947d60a1dd3c239d : FAILURE in 35m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/93012bc303324e9fabbabdfc5d5afd26 : FAILURE in 32m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ed9d580aec0437bb948a3a4b7096283 : FAILURE in 39m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c8e5b5b688204d389b6a732a9f0503ae : FAILURE in 35m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/29b278f00d1c402baa1ffc9f4a06d090 : FAILURE in 41m 37s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"45f4291bfab83f19b4126218ef5f21b04005d348","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-11-25 22:25:47.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"7d394a49369fb940e2a87e79fb92614dc6044790","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 23:27:05.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51da9456c53d4ca28f9da8eb949dec84 : SUCCESS in 8m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00ee54a00a0a4131938377b2c1d548da : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbfa39c64bda48cc8d305e075b369a0b : SUCCESS in 7m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/687b23534dfc4fff9ad4339711ea173f : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d999c7fcdd4c407d9ad00da179bb7628 : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc274a73fa6941b79aa7d00436ed3564 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b5c03b21c7c46c4b051a80e91fd041a : SUCCESS in 14m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6aa848a8ae374d178c76c1a2f6a26723 : SUCCESS in 5m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6821bc7a08914d3db9cbf056bc9b63b3 : SUCCESS in 7m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1cfe3cbd582f44bea1205af6351be195 : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8cb8c25ab758481682924c9bc26d29e9 : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/816afdbcc33c457b98bd0cb2cf65ec13 : SUCCESS in 42m 09s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4f515541372c41ad9afee185239361de : SUCCESS in 49m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7795c743dc2247a387e3cc77b1c0fec3 : FAILURE in 30m 35s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e56d0431884f47afb9a95389d739d86f : FAILURE in 31m 43s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b53cb19bcd3e4b27a2b756f7d5b31529 : FAILURE in 39m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a51ceda93d1e429ebdd2c3ab40aa7b43 : FAILURE in 36m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/94657046b4364c7f8353182bdb5d65da : FAILURE in 35m 05s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ea29e930d87c4e39b329369e9e20da4a : FAILURE in 38m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e29b72432e594d82997a531fb6efad80 : FAILURE in 38m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c200afbdf38e46a8abe81e2a08b5ac42 : FAILURE in 31m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49cf3116a08a4e28b031d1e68d7dcbc5 : FAILURE in 31m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6c458f81999b4ee29c0801bc83be3055 : FAILURE in 33m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d3b24f8fa11c40cfbe00e1c9ecf7958b : FAILURE in 36m 50s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"e6c9e431ff85105cd0a3460ab4ec77b356524f2c","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-11-27 16:07:00.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"b8e2730a92cc8bd67d8767ad70dda687eb34f092","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 18:51:57.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a521320563c74f43b2de0fd7d423fce6 : SUCCESS in 8m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36a2aa18d0ee4d5ab79e0ce286ba3ed1 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d35cf70e8a844bb3abad13a98bad740f : SUCCESS in 6m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/055494406cbb4f35a9b919b91c289823 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac6e13aa3663472db24b772956b22de1 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a23848bf85a4cadbe798595a92220d6 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ded975f6513d41da9482d49fd8032124 : SUCCESS in 12m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8ca8699a885a4b8c8d86f8560157c4ac : SUCCESS in 5m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f9a69a957da843ea8d70ac3d70f3d2bb : SUCCESS in 7m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6b1529c65ebf4d9e909d54c385275ad6 : SUCCESS in 9m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ba2969a4e15b4e69a0667175b8a9de24 : SUCCESS in 8m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/434408f6dda54cde8324bca849e2d514 : SUCCESS in 44m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d9b54bba3d014bafbbd62307007cbc65 : SUCCESS in 49m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a983cfc42094779ae3120cba17ba855 : FAILURE in 41m 27s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0577cec7ecad44898bf636f439a9e900 : SUCCESS in 1h 34m 27s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/51664d0d6a834ef9bb831180e9730172 : SUCCESS in 2h 30m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/47ed671253e649b0bafc2bb3aa925194 : FAILURE in 39m 16s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/dd44604649424f259da74ce21c46e741 : SUCCESS in 1h 18m 03s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d1cbd394d8c24662821866791cd64bbf : FAILURE in 58m 34s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ae227c1795a54f208c5395d2f8728da3 : SUCCESS in 50m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/89496e1c172b4546b07b7922640c65a1 : SUCCESS in 1h 38m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7e1ad7defc1b4bf5aaff9db7aefc959b : SUCCESS in 1h 09m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8b1b386d747d4e7c9774f1e99d3ebb5e : SUCCESS in 1h 11m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fa35ec1fe9eb4755837461a58160b7c9 : SUCCESS in 1h 27m 26s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"5f6eaa17e4c005a3c1358449b54ac8a4f7f78afa","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-11-28 04:50:07.000000000","message":"Patch Set 23: Patch Set 22 was rebased","accounts_in_message":[],"_revision_number":23},{"id":"4e97fe38446c0e6c4cb3044f11aafcea85cc7d99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 07:25:28.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e626bdc645f46918d74650b7e13138e : SUCCESS in 6m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9da9a53ceaf42aa8e5b979ffd4b68a6 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee189ead41b54bf5beb3c94890afda2a : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f1bd38d76b3c4838889dae19493a3b80 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/237a867f0e3b4c9ba20b39feb1356d18 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a71371ea7ea4f8cbdd5008117c2bee6 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b1635b12054c589e35ff784797105a : SUCCESS in 10m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5545a5e35fad4cdc8c18d48b3db25be9 : SUCCESS in 4m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fe7e6e51eb5b4d02b2442bf009a3170b : SUCCESS in 7m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ad39f7c7ffb545e8ac30fe029fc99322 : SUCCESS in 4m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0ac8b06cbda14f0e9998f85975ed846d : SUCCESS in 5m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5255cd7ae8524b83b092192fe6ec9a23 : SUCCESS in 43m 58s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/7a0cf7341a7d489486e3d69a304f0ad3 : SUCCESS in 44m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54425586fe774521a7f88b5adab49942 : FAILURE in 41m 59s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/53b5aa3376e9492da54d80dc4a92188d : SUCCESS in 1h 38m 10s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c721a05293a74fc494c8b799d002598a : SUCCESS in 2h 03m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f2af94ca01594489accf60d069cb9631 : FAILURE in 32m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/b954ab99f0714caa8afe3381430a4f6b : SUCCESS in 1h 58m 10s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5a3a86e3731b4d86b1a601445540089b : FAILURE in 59m 50s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/747538cf273942a3934f664cdb7626dc : SUCCESS in 52m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7c88b4f5397b4a309e3c4955e6d183b1 : SUCCESS in 1h 29m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afb344ac8d1a47da9994b92dcf4d6ccf : SUCCESS in 1h 02m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/62419315148e463dbe538c6d7ae23ca5 : SUCCESS in 1h 02m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/22e0a2476fd94c30873e2ded1dff4b28 : FAILURE in 1h 30m 12s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"876aaa55db824d3d0b088e8229a1a22ce7071973","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-11-28 17:18:58.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"c8330cbb37f787af6052d8b1560354c31eabd05a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-11-28 17:33:25.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"9f8cef7ffd27af5c132847dd8764f441b8713a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 19:33:19.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d219338ed47a4c4995283b3646a828c0 : SUCCESS in 10m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5582d0f5a454035b8ecfd3407c353ef : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c7edc23d750460b8975c626b288f4c1 : SUCCESS in 6m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/393dbc4cd4b14d8c86179617e939328a : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/082a65b0c9dc42ceace9d491f75ebe0c : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8aa657f509cc46a0b12c84487c121b66 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dee84febaa444b6856c3ddb47ef5866 : SUCCESS in 10m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/81181bcff1764b029373dc2c128416f9 : SUCCESS in 5m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c4d07cdb25c844c18882fa3f72d16133 : SUCCESS in 7m 50s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8a5d92fe650646f4b7eb776bfe8b0cdc : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c176b48d6f64603a7ae5045024a23f9 : SUCCESS in 7m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d1900b915bea4fa5a5fe4f5918a312d4 : SUCCESS in 43m 04s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/80710f9061fd46c7aa30f9931ea9fe9e : SUCCESS in 45m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/978d4822a4844df1af56614298e20bc2 : SUCCESS in 1h 28m 56s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fc4f30af578e4c2ca6c0cbe6bba94117 : SUCCESS in 1h 27m 36s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/aeb99a1b890349caa35ca8e9634ede73 : SUCCESS in 1h 52m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/36a57ae264d74fc4bb191f7bc1beb3d2 : FAILURE in 31m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/2c954437b65a4959ad596b6d6738d4ea : SUCCESS in 1h 15m 11s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/38be80da5b6f4cc29fb84f9945d71c17 : FAILURE in 1h 08m 29s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d8c5c4ee1db342e286cfd2e5d0d6e8a3 : SUCCESS in 51m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1e05348ad6a740cdbc9b1ff8562edd50 : SUCCESS in 1h 41m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3167a94ceeea4110a2bda330fa089a25 : SUCCESS in 59m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/63e6ccb66ec142ea8c9fa45ae99eecd4 : SUCCESS in 1h 11m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8348dda82dab44f191754a63eae848f4 : FAILURE in 1h 15m 51s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"0bacde7746f4316108a0d270e20de321b64c77a5","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-11-29 02:55:37.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"ec49fdd7698b2df3f12dd82065ce58bbded38ffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 04:58:20.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76fb567ac43c42b09c8933b6eeaf8812 : SUCCESS in 14m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c89e2d3e07241e48020c700163b7ce5 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31d00985a1f744f0ac3d2c735efac36c : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c27399d1f3149ee905f7f83edb5556a : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e9b388f77804f7d8799209757d6852d : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6cd82dd66a0948128a89cea24307168f : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74c28af0ce3e4106a6479b439b03c041 : SUCCESS in 10m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9699d1bfb8fc46a18d8507640f6d44cc : SUCCESS in 4m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f1f16c4722234da6b60a6aba8fd02d16 : SUCCESS in 6m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/977cf6f576fb43298a6683d3d527371f : SUCCESS in 5m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8ec25e59d0424b46863ffebf53418de8 : SUCCESS in 6m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/432f79a9f2184336b3731fcb3c918c50 : SUCCESS in 41m 07s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4b89d84eaeb048b094d63a346827525a : SUCCESS in 48m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f26be31fb574cc2a505a885cf798354 : SUCCESS in 1h 27m 06s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2e709884abe64d22a035ee9227a4210d : FAILURE in 1h 39m 07s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ecccc91edd0e45c6b60129fc0002f309 : SUCCESS in 2h 01m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6ff64d2b3a94450486c9525e154aea6b : FAILURE in 36m 46s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c8932d86942c42ab89b1ddcdef477ed9 : POST_FAILURE in 1h 47m 44s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9b9f9b08fde94770830ea79904c100d7 : FAILURE in 58m 39s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/848726cf0c754d9594231d66f6ac40ba : SUCCESS in 52m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/00cc8ccdee7846ddb7644e39202acba0 : SUCCESS in 1h 27m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71f531a4f92243b79f41e9a3219eaee5 : SUCCESS in 1h 08m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/861364199f0a4baab6e8214807f5c583 : SUCCESS in 1h 04m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/cd451f25740c4909b1a6e7eaf451b490 : SUCCESS in 1h 20m 32s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"84238fc2370fc878ac681fc13f8f5b8b82ec7b34","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-02 16:02:37.000000000","message":"Patch Set 25: Code-Review+1\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":25},{"id":"99c34cd5350bc7f7fce9d077e0ed91c41b8279c3","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-12-02 18:25:05.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"4c7ca3f618435b3ddf9fd334640a589a3a077fca","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-12-02 19:53:22.000000000","message":"Patch Set 25:\n\nrecheck","accounts_in_message":[],"_revision_number":25},{"id":"8b4ed68f619be95ded616eac8f5ddba76a9cd034","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-02 20:59:55.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"269e6e9e85692bebab176e48851da414b92a75c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 22:14:23.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/767aea235c6e46d79293d5a0d5de27a0 : SUCCESS in 7m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/908bfd13f9a648418a5fd3b217189120 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83c80df33c5749c8b4487b1dfb7450a4 : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/56f10047baff41c3bfd4b2f5e5472c51 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e921e9eaf39741c68a014056a5377e0c : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b27daa77aaf428fa0b077e867b672de : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a760cb72ae1245939d261a989dad0767 : SUCCESS in 10m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/54b3ae18c6e04db3808bb7bdec4a3a16 : SUCCESS in 4m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/746fbafdc7ce4f43a01d1a30aaa496cb : SUCCESS in 6m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/043e2a516c5348679634035e8fc2428c : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ade130ec82d44b29b21ac4db72328065 : SUCCESS in 5m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/740490a7a1aa41cf986f4b82ee649ab9 : SUCCESS in 46m 11s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d641ad451acb4c65a588155a3f34e4f5 : SUCCESS in 53m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/850193a9ebfd44f88c1ab4cc46e55f3f : SUCCESS in 2h 11m 21s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7be8776d9ae1411fb761fd5d30b40375 : SUCCESS in 1h 46m 24s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8a4138528e274cbdb6a0a748a44a896a : SUCCESS in 2h 14m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f48d1513650c4834b9877681ea63dd7b : FAILURE in 40m 19s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/bf6931fc336f4588b678f03a16b6b89b : SUCCESS in 1h 50m 43s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e8dc6b8b8234412a941bc9e5ea20ffdc : FAILURE in 59m 06s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ff2f17f330944b08aa115646c87a7856 : SUCCESS in 56m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6ccbf6c562c44281b42b2ad1184b01e0 : SUCCESS in 1h 31m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d2c464c477941979d6c4355939f4a40 : SUCCESS in 1h 05m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5bb0ed564ed04c8bbca9727e347c2840 : SUCCESS in 1h 25m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/15354f67ea79424c86abc7ebbe61aa85 : FAILURE in 1h 19m 32s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"9aad0d55d2d71d91c0e2de1023af9ddb23bd1ba7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-03 00:36:18.000000000","message":"Patch Set 25: Code-Review+2","accounts_in_message":[],"_revision_number":25},{"id":"57d8eac655ebac67731cbb0a9fe9f021a405123c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-03 00:58:30.000000000","message":"Patch Set 25: Code-Review+1","accounts_in_message":[],"_revision_number":25},{"id":"dd818308be85f2fa9778dd7793ddfe869d0c4461","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-03 01:52:59.000000000","message":"Patch Set 25: Backport-Candidate-2 Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":25},{"id":"dd770af79f00e1f638bd25ba84eb4aa592ef862d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 01:53:11.000000000","message":"Patch Set 25: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"d619bd0c4f0ddfce1fe6127cc944f115b17338d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 03:25:01.000000000","message":"Patch Set 25: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93e365351e9a4aa89ba2698ad6ea7018 : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db1beb20b1d44bd59827baace45eaa08 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ec4ea4f9f7749eb9bbcd9ef9b6d382a : SUCCESS in 6m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1454efd8ee0940a693d9f3a557f5a2b3 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38f6d441c9c8466e866cbd42077150f0 : SUCCESS in 12m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2057ff36bd884127b93164fe21ea326e : SUCCESS in 7m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/76d1f8b2a7b74263bb41f5e83af82ef3 : SUCCESS in 48m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cfea72d1d44248838738aaa70b8b1580 : SUCCESS in 1h 24m 02s","accounts_in_message":[],"_revision_number":25},{"id":"898b2cfec321ad8d564fadb241ce12b49372bebb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 03:25:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":25},{"id":"bf46a9bae32c516675e96fcb8383f61c1bac30f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 03:26:47.000000000","message":"Patch Set 25:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf454100e664681b64437cdf4fa0cbb : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"314b43af9a57a698a66d46c11892380bff315166","revisions":{"24cf3dfc109ddd99c059f4b9514091d50a49fd48":{"kind":"REWORK","_number":1,"created":"2019-06-03 15:48:38.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/1"}}},"commit":{"parents":[{"commit":"841cd348a444af66ad1953ed1fa8e3d17aebdfaf","subject":"Create Amphora V2 provider driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/841cd348a444af66ad1953ed1fa8e3d17aebdfaf"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"subject":"[WIP] Use retry for AmphoraComputeConnectivityWait","message":"[WIP] Use retry for AmphoraComputeConnectivityWait\n\nMove retry out of loop. Use taskflow retry class Time.\n\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\nStory: 2005072\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24cf3dfc109ddd99c059f4b9514091d50a49fd48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24cf3dfc109ddd99c059f4b9514091d50a49fd48"}]},"branch":"refs/heads/master"},"bc0716286c4e4d38472fc7ba4b46050294058bdd":{"kind":"REWORK","_number":2,"created":"2019-06-05 12:55:13.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/2"}}},"commit":{"parents":[{"commit":"841cd348a444af66ad1953ed1fa8e3d17aebdfaf","subject":"Create Amphora V2 provider driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/841cd348a444af66ad1953ed1fa8e3d17aebdfaf"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-05 12:54:49.000000000","tz":240},"subject":"[WIP] Use retry for AmphoraComputeConnectivityWait","message":"[WIP] Use retry for AmphoraComputeConnectivityWait\n\nMove retry out of loop. Use taskflow retry class Time.\n\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\nStory: 2005072\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc0716286c4e4d38472fc7ba4b46050294058bdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc0716286c4e4d38472fc7ba4b46050294058bdd"}]},"branch":"refs/heads/master"},"f51b3e4a5acd8ab2cf6b50c954c1aa7beb6f3c41":{"kind":"REWORK","_number":3,"created":"2019-06-05 12:57:21.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/3"}}},"commit":{"parents":[{"commit":"841cd348a444af66ad1953ed1fa8e3d17aebdfaf","subject":"Create Amphora V2 provider driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/841cd348a444af66ad1953ed1fa8e3d17aebdfaf"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-05 12:57:07.000000000","tz":240},"subject":"[WIP] Use retry for AmphoraComputeConnectivityWait","message":"[WIP] Use retry for AmphoraComputeConnectivityWait\n\nMove retry out of loop. Use taskflow retry class Time.\n\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\nStory: 2005072\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f51b3e4a5acd8ab2cf6b50c954c1aa7beb6f3c41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f51b3e4a5acd8ab2cf6b50c954c1aa7beb6f3c41"}]},"branch":"refs/heads/master"},"777c23c2a434cd72804004db9e9ac04d00f25095":{"kind":"REWORK","_number":4,"created":"2019-06-05 13:22:23.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/4"}}},"commit":{"parents":[{"commit":"841cd348a444af66ad1953ed1fa8e3d17aebdfaf","subject":"Create Amphora V2 provider driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/841cd348a444af66ad1953ed1fa8e3d17aebdfaf"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-05 13:21:59.000000000","tz":240},"subject":"[WIP] Use retry for AmphoraComputeConnectivityWait","message":"[WIP] Use retry for AmphoraComputeConnectivityWait\n\nMove retry out of loop. Use taskflow retry class Time.\n\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\nStory: 2005072\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/777c23c2a434cd72804004db9e9ac04d00f25095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/777c23c2a434cd72804004db9e9ac04d00f25095"}]},"branch":"refs/heads/master"},"6cbb860de637d45917de4d9cc289b5a24afa5f00":{"kind":"REWORK","_number":5,"created":"2019-06-05 16:05:42.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/5"}}},"commit":{"parents":[{"commit":"841cd348a444af66ad1953ed1fa8e3d17aebdfaf","subject":"Create Amphora V2 provider driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/841cd348a444af66ad1953ed1fa8e3d17aebdfaf"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-05 16:03:57.000000000","tz":240},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait.\nRequired for redis jobboard implementation.\n\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\nStory: 2005072\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6cbb860de637d45917de4d9cc289b5a24afa5f00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6cbb860de637d45917de4d9cc289b5a24afa5f00"}]},"branch":"refs/heads/master"},"a3d172f075dd2cc62634db8e268f3f2eb9474f2e":{"kind":"REWORK","_number":6,"created":"2019-06-06 13:15:48.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/6"}}},"commit":{"parents":[{"commit":"ff4680eb71cf03e4eae1a58e7a66321ddadcbead","subject":"Merge \"Create Amphora V2 provider driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff4680eb71cf03e4eae1a58e7a66321ddadcbead"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-06 13:15:30.000000000","tz":240},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait.\nRequired for redis jobboard implementation.\n\nStory: 2005072\nTask: 33477\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a3d172f075dd2cc62634db8e268f3f2eb9474f2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a3d172f075dd2cc62634db8e268f3f2eb9474f2e"}]},"branch":"refs/heads/master"},"a99d9da7e2220367ba70db8ae1717f8f30bc808b":{"kind":"REWORK","_number":7,"created":"2019-06-10 08:35:36.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/7"}}},"commit":{"parents":[{"commit":"ff4680eb71cf03e4eae1a58e7a66321ddadcbead","subject":"Merge \"Create Amphora V2 provider driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff4680eb71cf03e4eae1a58e7a66321ddadcbead"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-10 08:35:12.000000000","tz":240},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait.\nRequired for redis jobboard implementation.\n\nStory: 2005072\nTask: 33477\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a99d9da7e2220367ba70db8ae1717f8f30bc808b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a99d9da7e2220367ba70db8ae1717f8f30bc808b"}]},"branch":"refs/heads/master"},"374c7f68780b650e8df4437a58d5002ab52e9f61":{"kind":"REWORK","_number":8,"created":"2019-06-10 09:51:19.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/8"}}},"commit":{"parents":[{"commit":"ff4680eb71cf03e4eae1a58e7a66321ddadcbead","subject":"Merge \"Create Amphora V2 provider driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff4680eb71cf03e4eae1a58e7a66321ddadcbead"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-10 09:51:02.000000000","tz":240},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait.\nRequired for redis jobboard implementation.\n\nStory: 2005072\nTask: 33477\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/374c7f68780b650e8df4437a58d5002ab52e9f61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/374c7f68780b650e8df4437a58d5002ab52e9f61"}]},"branch":"refs/heads/master"},"12b1e89aab37a789bcc683e0af1d5d756ccb4892":{"kind":"REWORK","_number":9,"created":"2019-06-10 15:22:31.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/9"}}},"commit":{"parents":[{"commit":"ff4680eb71cf03e4eae1a58e7a66321ddadcbead","subject":"Merge \"Create Amphora V2 provider driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff4680eb71cf03e4eae1a58e7a66321ddadcbead"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-10 15:22:11.000000000","tz":240},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait.\nRequired for redis jobboard implementation.\n\nStory: 2005072\nTask: 33477\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12b1e89aab37a789bcc683e0af1d5d756ccb4892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12b1e89aab37a789bcc683e0af1d5d756ccb4892"}]},"branch":"refs/heads/master"},"5e214fbbf3ae4c97ae297e7926dad606f82a7690":{"kind":"REWORK","_number":10,"created":"2019-06-11 08:00:33.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/10"}}},"commit":{"parents":[{"commit":"ff4680eb71cf03e4eae1a58e7a66321ddadcbead","subject":"Merge \"Create Amphora V2 provider driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff4680eb71cf03e4eae1a58e7a66321ddadcbead"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-11 08:00:13.000000000","tz":0},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait.\nRequired for redis jobboard implementation.\n\nStory: 2005072\nTask: 33477\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e214fbbf3ae4c97ae297e7926dad606f82a7690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e214fbbf3ae4c97ae297e7926dad606f82a7690"}]},"branch":"refs/heads/master"},"38e882ae65053d3662cf2217f57a05828fd4ea90":{"kind":"REWORK","_number":11,"created":"2019-06-11 08:19:19.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/11"}}},"commit":{"parents":[{"commit":"ff4680eb71cf03e4eae1a58e7a66321ddadcbead","subject":"Merge \"Create Amphora V2 provider driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff4680eb71cf03e4eae1a58e7a66321ddadcbead"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-11 08:19:08.000000000","tz":0},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait.\nRequired for redis jobboard implementation.\n\nStory: 2005072\nTask: 33477\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38e882ae65053d3662cf2217f57a05828fd4ea90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38e882ae65053d3662cf2217f57a05828fd4ea90"}]},"branch":"refs/heads/master"},"ce81ba1c725b98981f1eaaf7c365821c77a31d47":{"kind":"REWORK","_number":12,"created":"2019-06-11 14:08:05.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/12"}}},"commit":{"parents":[{"commit":"ff4680eb71cf03e4eae1a58e7a66321ddadcbead","subject":"Merge \"Create Amphora V2 provider driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff4680eb71cf03e4eae1a58e7a66321ddadcbead"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-11 14:07:06.000000000","tz":240},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait.\nRequired for redis jobboard implementation.\n\nStory: 2005072\nTask: 33477\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce81ba1c725b98981f1eaaf7c365821c77a31d47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce81ba1c725b98981f1eaaf7c365821c77a31d47"}]},"branch":"refs/heads/master"},"999a35c11e4471494ddfea51890d7f61fe88fd35":{"kind":"REWORK","_number":13,"created":"2019-06-18 11:41:18.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/13"}}},"commit":{"parents":[{"commit":"ff4680eb71cf03e4eae1a58e7a66321ddadcbead","subject":"Merge \"Create Amphora V2 provider driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff4680eb71cf03e4eae1a58e7a66321ddadcbead"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-18 11:40:59.000000000","tz":240},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait.\nRequired for redis jobboard implementation.\n\nStory: 2005072\nTask: 33477\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/999a35c11e4471494ddfea51890d7f61fe88fd35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/999a35c11e4471494ddfea51890d7f61fe88fd35"}]},"branch":"refs/heads/master"},"0fae7e68028bbec37d22783b51b2b51aa06e0d9a":{"kind":"REWORK","_number":14,"created":"2019-07-12 13:16:43.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/14"}}},"commit":{"parents":[{"commit":"ff4680eb71cf03e4eae1a58e7a66321ddadcbead","subject":"Merge \"Create Amphora V2 provider driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff4680eb71cf03e4eae1a58e7a66321ddadcbead"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-07-12 13:16:21.000000000","tz":240},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait.\nRequired for redis jobboard implementation.\n\nStory: 2005072\nTask: 33477\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0fae7e68028bbec37d22783b51b2b51aa06e0d9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0fae7e68028bbec37d22783b51b2b51aa06e0d9a"}]},"branch":"refs/heads/master"},"d1b613af0d31b136498f76c08408a3935b52a332":{"kind":"REWORK","_number":15,"created":"2019-08-08 08:06:43.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/15"}}},"commit":{"parents":[{"commit":"6e4da8506469de634a4ef57538c7f3ffed3cc937","subject":"Merge \"Re-enable the py36 tox environment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e4da8506469de634a4ef57538c7f3ffed3cc937"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-08-08 08:06:26.000000000","tz":240},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait.\nRequired for redis jobboard implementation.\n\nStory: 2005072\nTask: 33477\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1b613af0d31b136498f76c08408a3935b52a332"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1b613af0d31b136498f76c08408a3935b52a332"}]},"branch":"refs/heads/master"},"c218ddc4783e469193e0950794bd7a3764921c50":{"kind":"REWORK","_number":16,"created":"2019-08-08 13:27:26.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/16"}}},"commit":{"parents":[{"commit":"6e4da8506469de634a4ef57538c7f3ffed3cc937","subject":"Merge \"Re-enable the py36 tox environment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e4da8506469de634a4ef57538c7f3ffed3cc937"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-08-08 13:27:09.000000000","tz":240},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait.\nRequired for redis jobboard implementation.\n\nStory: 2005072\nTask: 33477\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c218ddc4783e469193e0950794bd7a3764921c50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c218ddc4783e469193e0950794bd7a3764921c50"}]},"branch":"refs/heads/master"},"5aef7a4320b50a943058dcfbabd3144de466e917":{"kind":"REWORK","_number":17,"created":"2019-09-12 10:51:15.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/17"}}},"commit":{"parents":[{"commit":"510525a92b6c6335f9ca28b0361e138edb52cd19","subject":"Merge \"Add `additive_only` parameter to Batch Member call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/510525a92b6c6335f9ca28b0361e138edb52cd19"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-09-12 10:50:56.000000000","tz":240},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait.\nRequired for redis jobboard implementation.\n\nStory: 2005072\nTask: 33477\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5aef7a4320b50a943058dcfbabd3144de466e917"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5aef7a4320b50a943058dcfbabd3144de466e917"}]},"branch":"refs/heads/master"},"d49b27d6f8a2728ca54ae4c56a7ebb191e3205a5":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2019-09-18 14:34:35.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/18"}}},"commit":{"parents":[{"commit":"38ff0be1dc74fae3cf80cb4368edaf4f88febf88","subject":"Merge \"Fix \u0027additive_only\u0027 parameter api-ref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38ff0be1dc74fae3cf80cb4368edaf4f88febf88"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-09-18 14:34:35.000000000","tz":0},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait.\nRequired for redis jobboard implementation.\n\nStory: 2005072\nTask: 33477\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d49b27d6f8a2728ca54ae4c56a7ebb191e3205a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d49b27d6f8a2728ca54ae4c56a7ebb191e3205a5"}]},"branch":"refs/heads/master"},"9a3cc6363e70aa25517aadf6f3321d07ce25f009":{"kind":"NO_CODE_CHANGE","_number":19,"created":"2019-10-17 10:54:23.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/19"}}},"commit":{"parents":[{"commit":"38ff0be1dc74fae3cf80cb4368edaf4f88febf88","subject":"Merge \"Fix \u0027additive_only\u0027 parameter api-ref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38ff0be1dc74fae3cf80cb4368edaf4f88febf88"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-10-17 10:54:15.000000000","tz":0},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait. [1]\n\nThis reqired for redis jobboard implementation as each retry expand \nclaim for job on worker. This means that worker is proccesing job and \nit should not be released for other workers to work on it.\n\nAdopted for v2 flows.\n\n[1] - https://docs.openstack.org/taskflow/latest/user/atoms.html#retry\n\nStory: 2005072\nTask: 33477\n\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a3cc6363e70aa25517aadf6f3321d07ce25f009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a3cc6363e70aa25517aadf6f3321d07ce25f009"}]},"branch":"refs/heads/master"},"e64bb14f2e036ca44ca54b82cbaf89525627e8ef":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2019-10-22 09:51:49.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/20"}}},"commit":{"parents":[{"commit":"6b58e66868cd806fff4f777b776985c2e94cb75d","subject":"Merge \"Add Gentoo platform detection in diskimage-create/diskimage-create.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6b58e66868cd806fff4f777b776985c2e94cb75d"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-10-22 09:51:49.000000000","tz":0},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait. [1]\n\nThis reqired for redis jobboard implementation as each retry expand \nclaim for job on worker. This means that worker is proccesing job and \nit should not be released for other workers to work on it.\n\nAdopted for v2 flows.\n\n[1] - https://docs.openstack.org/taskflow/latest/user/atoms.html#retry\n\nStory: 2005072\nTask: 33477\n\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e64bb14f2e036ca44ca54b82cbaf89525627e8ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e64bb14f2e036ca44ca54b82cbaf89525627e8ef"}]},"branch":"refs/heads/master"},"adec840efa4101bed618d3a5dca7b78c4bde257d":{"kind":"REWORK","_number":21,"created":"2019-10-23 12:41:38.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/21"}}},"commit":{"parents":[{"commit":"6b58e66868cd806fff4f777b776985c2e94cb75d","subject":"Merge \"Add Gentoo platform detection in diskimage-create/diskimage-create.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6b58e66868cd806fff4f777b776985c2e94cb75d"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-10-23 12:41:05.000000000","tz":0},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait. [1]\n\nThis reqired for redis jobboard implementation as each retry expand \nclaim for job on worker. This means that worker is proccesing job and \nit should not be released for other workers to work on it.\n\nAdopted for v2 flows.\n\n[1] - https://docs.openstack.org/taskflow/latest/user/atoms.html#retry\n\nStory: 2005072\nTask: 33477\n\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/adec840efa4101bed618d3a5dca7b78c4bde257d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/adec840efa4101bed618d3a5dca7b78c4bde257d"}]},"branch":"refs/heads/master"},"b5ba392e606528d9a0856a11d0136dd1f445cb82":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2019-11-25 18:03:53.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/22"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-11-25 18:03:53.000000000","tz":0},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait. [1]\n\nThis reqired for redis jobboard implementation as each retry expand \nclaim for job on worker. This means that worker is proccesing job and \nit should not be released for other workers to work on it.\n\nAdopted for v2 flows.\n\n[1] - https://docs.openstack.org/taskflow/latest/user/atoms.html#retry\n\nStory: 2005072\nTask: 33477\n\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b5ba392e606528d9a0856a11d0136dd1f445cb82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b5ba392e606528d9a0856a11d0136dd1f445cb82"}]},"branch":"refs/heads/master"},"7c8a65677c41410f453ec4e53cb0114594abd02b":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2019-11-28 04:50:07.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/23"}}},"commit":{"parents":[{"commit":"728550103f2b76a9a99847b66af54a0d40168b2e","subject":"Merge \"Remove unnecessary interface configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/728550103f2b76a9a99847b66af54a0d40168b2e"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-11-28 04:50:07.000000000","tz":0},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait. [1]\n\nThis reqired for redis jobboard implementation as each retry expand \nclaim for job on worker. This means that worker is proccesing job and \nit should not be released for other workers to work on it.\n\nAdopted for v2 flows.\n\n[1] - https://docs.openstack.org/taskflow/latest/user/atoms.html#retry\n\nStory: 2005072\nTask: 33477\n\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c8a65677c41410f453ec4e53cb0114594abd02b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c8a65677c41410f453ec4e53cb0114594abd02b"}]},"branch":"refs/heads/master"},"cbd83a3adedb1e70efd0111818c8e74bb4c0039c":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2019-11-28 17:33:25.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/24"}}},"commit":{"parents":[{"commit":"a308ad28bf6d2e2538576eedc3c19f2e6ed9a3a6","subject":"Merge \"Fix filtering with API tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a308ad28bf6d2e2538576eedc3c19f2e6ed9a3a6"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-11-28 17:31:01.000000000","tz":240},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait. [1]\n\nThis reqired for redis jobboard implementation as each retry expand \nclaim for job on worker. This means that worker is proccesing job and \nit should not be released for other workers to work on it.\n\nAdopted for v2 flows.\n\n[1] - https://docs.openstack.org/taskflow/latest/user/atoms.html#retry\n\nStory: 2005072\nTask: 33477\n\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbd83a3adedb1e70efd0111818c8e74bb4c0039c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbd83a3adedb1e70efd0111818c8e74bb4c0039c"}]},"branch":"refs/heads/master"},"314b43af9a57a698a66d46c11892380bff315166":{"kind":"REWORK","_number":25,"created":"2019-11-29 02:55:37.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/91/662791/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/662791/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/662791/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/662791/25"}}},"commit":{"parents":[{"commit":"728550103f2b76a9a99847b66af54a0d40168b2e","subject":"Merge \"Remove unnecessary interface configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/728550103f2b76a9a99847b66af54a0d40168b2e"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-06-03 15:41:47.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-11-29 02:51:47.000000000","tz":240},"subject":"Use retry for AmphoraComputeConnectivityWait","message":"Use retry for AmphoraComputeConnectivityWait\n\nUse taskflow retry for connectivity wait. [1]\n\nThis reqired for redis jobboard implementation as each retry expand\nclaim for job on worker. This means that worker is proccesing job and\nit should not be released for other workers to work on it.\n\nAdopted for v2 flows.\n\n[1] - https://docs.openstack.org/taskflow/latest/user/atoms.html#retry\n\nStory: 2005072\nTask: 33477\n\nChange-Id: I2cf241ea965ad56ed70ebde83632ab855f5d859e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/314b43af9a57a698a66d46c11892380bff315166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/314b43af9a57a698a66d46c11892380bff315166"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
