)]}'
{"id":"openstack%2Foctavia~663265","triplet_id":"openstack%2Foctavia~master~Ie810610cc7819c9ac70d4394e2f434ac11283e12","project":"openstack/octavia","branch":"master","topic":"housekeeping-log-error","hashtags":[],"change_id":"Ie810610cc7819c9ac70d4394e2f434ac11283e12","subject":"housekeeping: Log exception with error level and exc_info","status":"NEW","created":"2019-06-05 07:59:03.000000000","updated":"2019-10-29 18:58:15.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9eea416adffdd7f652d607cb4c3fa0fcc9be5738","_number":663265,"virtual_id_number":663265,"owner":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-29 18:58:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-10-29 15:35:50.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"all":[{"value":-1,"date":"2019-10-29 15:18:27.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-29 15:35:50.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-10-29 18:58:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60c24ef826b685583a9a907b9f37859a0e3e88e9","author":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"date":"2019-06-05 07:59:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1916aca0b084d5b26c32bda883abb14f069b5c38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 10:19:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/65/663265/1/check/openstack-tox-cover/fdee9b3/cover/ : SUCCESS in 7m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/663265/1/check/openstack-tox-lower-constraints/3267757/ : SUCCESS in 7m 18s\n- openstack-tox-pep8 http://logs.openstack.org/65/663265/1/check/openstack-tox-pep8/37d15f1/ : SUCCESS in 6m 21s\n- openstack-tox-py27 http://logs.openstack.org/65/663265/1/check/openstack-tox-py27/fe9cf5f/ : SUCCESS in 5m 52s\n- openstack-tox-py36 http://logs.openstack.org/65/663265/1/check/openstack-tox-py36/5ccd721/ : SUCCESS in 5m 58s\n- openstack-tox-py37 http://logs.openstack.org/65/663265/1/check/openstack-tox-py37/8c0f90d/ : SUCCESS in 5m 41s\n- openstack-tox-docs http://logs.openstack.org/65/663265/1/check/openstack-tox-docs/a6bd0a0/html/ : SUCCESS in 6m 46s\n- openstack-tox-functional http://logs.openstack.org/65/663265/1/check/openstack-tox-functional/5c39fe6/ : SUCCESS in 7m 35s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/663265/1/check/openstack-tox-functional-py36/c4d7eb9/ : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/65/663265/1/check/octavia-v2-dsvm-noop-api/45e02b4/ : SUCCESS in 47m 00s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/65/663265/1/check/octavia-v2-dsvm-noop-py2-api/025eb05/ : SUCCESS in 43m 38s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/65/663265/1/check/octavia-v2-dsvm-scenario/b79e349/ : FAILURE in 1h 58m 27s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/65/663265/1/check/octavia-v2-dsvm-py2-scenario/59de53d/ : SUCCESS in 1h 23m 13s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/65/663265/1/check/octavia-v2-dsvm-py2-scenario-centos-7/d028bc2/ : FAILURE in 1h 52m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/65/663265/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/d8075af/ : SUCCESS in 1h 34m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/65/663265/1/check/octavia-v2-dsvm-scenario-two-node/0213ed5/ : FAILURE in 1h 43m 21s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/65/663265/1/check/octavia-v2-dsvm-py2-scenario-two-node/035daa1/ : SUCCESS in 1h 45m 39s (non-voting)\n- octavia-grenade http://logs.openstack.org/65/663265/1/check/octavia-grenade/86a46af/ : SUCCESS in 1h 32m 36s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/65/663265/1/check/octavia-v2-dsvm-tls-barbican/adc410c/ : SUCCESS in 52m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf46654dc147efa9973302a59fc7453f4b277241","author":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"date":"2019-06-05 14:46:07.000000000","message":"Patch Set 1: Workflow-1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2d1901ab0dcbbdc95ef88c025b6a8375a516db03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 17:04:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/65/663265/1/check/openstack-tox-cover/146da64/cover/ : SUCCESS in 7m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/663265/1/check/openstack-tox-lower-constraints/fce9197/ : SUCCESS in 7m 07s\n- openstack-tox-pep8 http://logs.openstack.org/65/663265/1/check/openstack-tox-pep8/76fcf4c/ : SUCCESS in 6m 55s\n- openstack-tox-py27 http://logs.openstack.org/65/663265/1/check/openstack-tox-py27/52b692b/ : SUCCESS in 6m 09s\n- openstack-tox-py36 http://logs.openstack.org/65/663265/1/check/openstack-tox-py36/d9e446f/ : SUCCESS in 5m 50s\n- openstack-tox-py37 http://logs.openstack.org/65/663265/1/check/openstack-tox-py37/ccaa509/ : SUCCESS in 5m 25s\n- openstack-tox-docs http://logs.openstack.org/65/663265/1/check/openstack-tox-docs/c7c0c71/html/ : SUCCESS in 6m 56s\n- openstack-tox-functional http://logs.openstack.org/65/663265/1/check/openstack-tox-functional/8db5be3/ : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/663265/1/check/openstack-tox-functional-py36/55877ed/ : SUCCESS in 7m 02s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/65/663265/1/check/octavia-v2-dsvm-noop-api/78f8fc3/ : SUCCESS in 46m 20s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/65/663265/1/check/octavia-v2-dsvm-noop-py2-api/d267b7f/ : SUCCESS in 47m 49s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/65/663265/1/check/octavia-v2-dsvm-scenario/b43ccbc/ : FAILURE in 30m 09s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/65/663265/1/check/octavia-v2-dsvm-py2-scenario/f4b758c/ : FAILURE in 25m 20s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/65/663265/1/check/octavia-v2-dsvm-py2-scenario-centos-7/804f5dc/ : FAILURE in 2h 07m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/65/663265/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/add72af/ : SUCCESS in 1h 50m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/65/663265/1/check/octavia-v2-dsvm-scenario-two-node/913267f/ : SUCCESS in 1h 20m 12s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/65/663265/1/check/octavia-v2-dsvm-py2-scenario-two-node/98e12cc/ : SUCCESS in 1h 23m 58s (non-voting)\n- octavia-grenade http://logs.openstack.org/65/663265/1/check/octavia-grenade/e1d082c/ : SUCCESS in 1h 17m 10s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/65/663265/1/check/octavia-v2-dsvm-tls-barbican/791d0b4/ : SUCCESS in 1h 00m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"12ba1a240205bf5e6c5130747250d441546c1c14","author":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"date":"2019-10-29 15:17:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"bbc784c8e93aeaeec33651051f5e7a3dcc44ae9b","author":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"date":"2019-10-29 15:18:27.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"9dda100e9e7e6add994b68798453495c1a263d72","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-29 15:35:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"699b89baed66f656e29afcdf4fb75f4cb990aabb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 18:58:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f29b94221cc34b0ca18b712a99c16652 : SUCCESS in 6m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86ba418dcd9945c1bb809416842a06d8 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73ceccd7b5c144fa816adf293c18e9d9 : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f3f678ea84f4466a151f4eb8d2c095b : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8786489ce4f4bc2b3a488da6001eae7 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8476b3b18074de6ae8256edcff28305 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d4e0562b2094c4ba6bd370e6bc17eb9 : SUCCESS in 10m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fa74a2503290429ca0217ab3fa7efa25 : SUCCESS in 4m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cff1f813a730465bae434cd902697cd6 : SUCCESS in 5m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6de476f05ae743498b656e4e1c59a268 : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/508df18bd6e54a6e879ebd8635bab7d2 : SUCCESS in 8m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c11276aa88b4e0f949cc6b61a322aac : SUCCESS in 48m 43s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1c3d5533acc14710873cd263168a5d75 : SUCCESS in 54m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c39c3c0262f4d34b8c5911e0f49f1bd : SUCCESS in 1h 38m 52s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/12f25887130a4c0cbdb66c69b6c55413 : FAILURE in 1h 48m 24s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/be6c21c08c5b46e4a6f8fcf77a8b5531 : SUCCESS in 1h 48m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dd8c3458eb6e42c7be9f380f4b4a6ba4 : SUCCESS in 1h 42m 21s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/63986c8bc7ab4e69bbc2f0d257b32bcd : TIMED_OUT in 2h 33m 28s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/87d5191abe7d4d5f94122f518fc8626b : SUCCESS in 1h 29m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c4265f2f7029410b9609e447658a9777 : SUCCESS in 59m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cb3bb4ca13014d9a9d207b9a5bea87bd : SUCCESS in 1h 45m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca9b7d4b112242ccacadc24ad8ef5785 : SUCCESS in 1h 00m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fd5adcb386134534b61f75d43a7c8d60 : SUCCESS in 1h 21m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b88ffe5c1deb4778ad422732b9f59f46 : FAILURE in 1h 22m 11s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fd3ac363382e6c869d792440bee76b222588f0ac","revisions":{"7d18dca7e1656f27c69ba493901980d9ecc8f1a6":{"kind":"REWORK","_number":1,"created":"2019-06-05 07:59:03.000000000","uploader":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"ref":"refs/changes/65/663265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/663265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/663265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/663265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/663265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/663265/1"}}},"commit":{"parents":[{"commit":"59660fb3650d3729375571afc2b85cf9074bca9e","subject":"Merge \"Force amp-agent communication to TLSv1.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/59660fb3650d3729375571afc2b85cf9074bca9e"}]}],"author":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2018-10-30 09:54:45.000000000","tz":60},"committer":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2019-06-05 07:58:42.000000000","tz":120},"subject":"housekeeping: Log exception with error level and exc_info","message":"housekeeping: Log exception with error level and exc_info\n\nChange-Id: Ie810610cc7819c9ac70d4394e2f434ac11283e12\nSigned-off-by: Sven Wegener \u003csven.wegener@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d18dca7e1656f27c69ba493901980d9ecc8f1a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d18dca7e1656f27c69ba493901980d9ecc8f1a6"}]},"branch":"refs/heads/master"},"fd3ac363382e6c869d792440bee76b222588f0ac":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-10-29 15:17:58.000000000","uploader":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"ref":"refs/changes/65/663265/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/663265/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/663265/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/663265/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/663265/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/663265/2"}}},"commit":{"parents":[{"commit":"43577a6c047d9e76c690c001f81c45020d554c27","subject":"Merge \"Fix batch member update error on empty change list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43577a6c047d9e76c690c001f81c45020d554c27"}]}],"author":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2018-10-30 09:54:45.000000000","tz":60},"committer":{"name":"Sven Wegener","email":"sven.wegener@stealer.net","date":"2019-10-29 15:17:58.000000000","tz":0},"subject":"housekeeping: Log exception with error level and exc_info","message":"housekeeping: Log exception with error level and exc_info\n\nChange-Id: Ie810610cc7819c9ac70d4394e2f434ac11283e12\nSigned-off-by: Sven Wegener \u003csven.wegener@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd3ac363382e6c869d792440bee76b222588f0ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd3ac363382e6c869d792440bee76b222588f0ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
