)]}'
{"id":"openstack%2Foctavia~663277","triplet_id":"openstack%2Foctavia~master~I4230feae5b6f0778b738bddbc69c5a0cd3e6930c","project":"openstack/octavia","branch":"master","topic":"re-add-FailoverPreparationForAmphora","hashtags":[],"change_id":"I4230feae5b6f0778b738bddbc69c5a0cd3e6930c","subject":"worker: Re-add FailoverPreparationForAmphora","status":"MERGED","created":"2019-06-05 08:53:17.000000000","updated":"2019-08-22 21:53:44.000000000","submitted":"2019-08-22 21:50:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663277-1566510636820-8367d9e9","meta_rev_id":"5e5f6dc56a143fd0e45a7fc7ecb5802176b45d77","_number":663277,"virtual_id_number":663277,"owner":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-08-22 21:50:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},{"value":2,"date":"2019-08-22 18:00:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-08-22 19:15:49.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-08-22 19:15:49.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-08-22 19:15:49.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-22 17:59:35.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 17:59:35.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-08-22 18:00:42.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-08-22 19:15:49.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-08-22 21:50:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5fb04f4e04e4ca89df2925625e313e214d21b9bc","author":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"date":"2019-06-05 08:53:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cddf9c01816fa68e0556432bcf8527026a668a8d","author":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"date":"2019-06-05 08:53:34.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"11501bd8e4f18a8b21340b70395253aaad9ae0cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 10:59:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/77/663277/1/check/openstack-tox-cover/395c8d9/cover/ : SUCCESS in 9m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/663277/1/check/openstack-tox-lower-constraints/5aa1b14/ : SUCCESS in 9m 10s\n- openstack-tox-pep8 http://logs.openstack.org/77/663277/1/check/openstack-tox-pep8/801f550/ : SUCCESS in 5m 28s\n- openstack-tox-py27 http://logs.openstack.org/77/663277/1/check/openstack-tox-py27/097e6b4/ : SUCCESS in 5m 48s\n- openstack-tox-py36 http://logs.openstack.org/77/663277/1/check/openstack-tox-py36/f40c0d8/ : SUCCESS in 4m 44s\n- openstack-tox-py37 http://logs.openstack.org/77/663277/1/check/openstack-tox-py37/fb52da3/ : SUCCESS in 5m 24s\n- openstack-tox-docs http://logs.openstack.org/77/663277/1/check/openstack-tox-docs/a88d161/html/ : SUCCESS in 6m 44s\n- openstack-tox-functional http://logs.openstack.org/77/663277/1/check/openstack-tox-functional/3487d86/ : SUCCESS in 7m 29s\n- openstack-tox-functional-py36 http://logs.openstack.org/77/663277/1/check/openstack-tox-functional-py36/ff29b68/ : SUCCESS in 7m 17s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/77/663277/1/check/octavia-v2-dsvm-noop-api/c79b3ee/ : SUCCESS in 38m 46s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/77/663277/1/check/octavia-v2-dsvm-noop-py2-api/d8644bc/ : SUCCESS in 37m 34s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/77/663277/1/check/octavia-v2-dsvm-scenario/d319cd1/ : SUCCESS in 1h 34m 37s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/77/663277/1/check/octavia-v2-dsvm-py2-scenario/57c8789/ : SUCCESS in 1h 32m 50s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/77/663277/1/check/octavia-v2-dsvm-py2-scenario-centos-7/b0813e4/ : FAILURE in 1h 50m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/77/663277/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/012bc36/ : SUCCESS in 1h 43m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/77/663277/1/check/octavia-v2-dsvm-scenario-two-node/4e82c81/ : SUCCESS in 1h 09m 42s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/77/663277/1/check/octavia-v2-dsvm-py2-scenario-two-node/7bf7fbf/ : SUCCESS in 1h 54m 50s (non-voting)\n- octavia-grenade http://logs.openstack.org/77/663277/1/check/octavia-grenade/2112928/ : SUCCESS in 1h 16m 11s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/77/663277/1/check/octavia-v2-dsvm-tls-barbican/96c8682/ : SUCCESS in 56m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"09a3d2a955ecf8c6b594fd3175afaecd688c28bd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-22 17:59:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e929a86a0942477190258c8963b691e7430d7ee7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-22 18:00:42.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks for pointing this out -- I assumed we were still doing this... Might actually be what broke failovers in my deployment. \u003c_\u003c","accounts_in_message":[],"_revision_number":2},{"id":"b12c87525a129210beb70a6c570ad273a2d75a50","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-22 19:15:49.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"58ebe30274b5d2841bbd16c1c2097b2aad605fe7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 20:06:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c189dad50037461e8d2933bb632966d5 : SUCCESS in 9m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7fb631c57f04347b0a515837c603168 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/925ea25e5cb34cc98e1caaf37af1b7f7 : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21f82d038f2240c89fabf2279a406e30 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd71a126e16645b5bdb3ab4e9581227a : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d9c62953fcb447381744106d03199ab : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8d93512f77445c5954ea77ad58d0dd7 : SUCCESS in 7m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ba2efd44d2ed4d6e9f9ec6ffe58ec6b6 : SUCCESS in 5m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1b8a7a9bfa8d49449077e215cde3b59c : SUCCESS in 7m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6808ecda4bac41f4b7841e1cd037a2e2 : SUCCESS in 7m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/25a34d4dc10a4ca38c90f4865d7dd55c : SUCCESS in 6m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee92c869465b40c8b8fecd44953be70d : SUCCESS in 48m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a6eda558342a429d9f29b11555e82322 : SUCCESS in 57m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e86a360526ac419493c77c52a4b85295 : SUCCESS in 1h 39m 10s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f51d110daffb4f6f830818cd7c26c029 : SUCCESS in 1h 46m 22s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8ffa73713f63434999e9edc28d899aac : FAILURE in 2h 02m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/96adcceb9bcc43af96306bb890168366 : SUCCESS in 1h 55m 19s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/904e154f036f4f2ba71f25341cbd6bc6 : SUCCESS in 1h 17m 20s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4fe5bee28e914c468afde7c6d4bac605 : SUCCESS in 1h 17m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6dac5d9165f247f78dd947c60a1fc1bd : SUCCESS in 58m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/de05e9fe188e455cba6def7be726901c : SUCCESS in 1h 44m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7b2254d6f12474bbd5f5567b3987d6e : SUCCESS in 1h 01m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/36c91085c39141c69cbc14bfaa9624e3 : SUCCESS in 1h 17m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/57162820c290444db9d0d16788a0a08e : FAILURE in 1h 39m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ce6f79ad3669ed8d9c549b7d43b67260f356df32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 20:06:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"350ddb64115d6e9499c44c453cc46039abb4a0d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 21:50:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0642d22b7dc6bbe4fed2a8eeee9d91ab8cb08d7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 21:50:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18dc4803be444935a88d781f9719ff9c : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f85d09bb08847ceadfe749815c2c8d0 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cdf5af648a824f468ca0c9dff3f70c4c : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e80780b7ddc0418c9ebdf5fc43c2a13c : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b601421ae8bf4d12a1c1a7da04dde077 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ea58cb77e474793b40bfb59e0c60d15 : SUCCESS in 8m 34s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fa2adcca3e884effa7d715598c59558b : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/569eb5b882f74724ae2321aab7a585b8 : SUCCESS in 7m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/885c279040e146b5b70eea9a76bbb6b9 : SUCCESS in 52m 15s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/075ae24cdb0b434eb15d16cb57c366f0 : SUCCESS in 47m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/654d6c9d092e45beadc489d24d0cbc61 : SUCCESS in 1h 37m 27s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/25ae2c29709640478102129776a649ba : SUCCESS in 1h 38m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b2666c5c983a46809e6972e6d9c04652 : SUCCESS in 1h 17m 57s","accounts_in_message":[],"_revision_number":2},{"id":"c94e23a76356e3d299a51649dfe57d581230857e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 21:53:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92d5e5424eab4544aa3a49f0e5a738e7 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c4c7387f8904075b961612b5a08aff47c51c081c","revisions":{"8b8afccfd5a34e3315a00db265d6d1bbb5c19b86":{"kind":"REWORK","_number":1,"created":"2019-06-05 08:53:17.000000000","uploader":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"ref":"refs/changes/77/663277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/663277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/663277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/663277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/663277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/663277/1"}}},"commit":{"parents":[{"commit":"59660fb3650d3729375571afc2b85cf9074bca9e","subject":"Merge \"Force amp-agent communication to TLSv1.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/59660fb3650d3729375571afc2b85cf9074bca9e"}]}],"author":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2019-01-16 12:44:34.000000000","tz":60},"committer":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2019-06-05 08:53:06.000000000","tz":120},"subject":"worker: Re-add FailoverPreparationForAmphora","message":"worker: Re-add FailoverPreparationForAmphora\n\nThis tasks removes the dns_name option from ports, so that they can be\nplugged into a new amphora, which has a different name. This was lost in\nchange I04d267bd3cdedca11f0350c5255086233cba14ec.\n\nChange-Id: I4230feae5b6f0778b738bddbc69c5a0cd3e6930c\nSigned-off-by: Sven Wegener \u003csven.wegener@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b8afccfd5a34e3315a00db265d6d1bbb5c19b86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b8afccfd5a34e3315a00db265d6d1bbb5c19b86"}]},"branch":"refs/heads/master"},"c4c7387f8904075b961612b5a08aff47c51c081c":{"kind":"REWORK","_number":2,"created":"2019-08-22 17:59:35.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/77/663277/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/663277/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/663277/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/663277/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/663277/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/663277/2"}}},"commit":{"parents":[{"commit":"f6b91c180d29a2c06f1c27a89cbca48ae6222eec","subject":"Merge \"Add Octavia tox \"tips\" jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f6b91c180d29a2c06f1c27a89cbca48ae6222eec"}]}],"author":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2019-01-16 12:44:34.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-22 17:59:34.000000000","tz":-420},"subject":"worker: Re-add FailoverPreparationForAmphora","message":"worker: Re-add FailoverPreparationForAmphora\n\nThis tasks removes the dns_name option from ports, so that they can be\nplugged into a new amphora, which has a different name. This was lost in\nchange I04d267bd3cdedca11f0350c5255086233cba14ec.\n\nStory: 2006205\nTask: 35791\n\nChange-Id: I4230feae5b6f0778b738bddbc69c5a0cd3e6930c\nSigned-off-by: Sven Wegener \u003csven.wegener@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4c7387f8904075b961612b5a08aff47c51c081c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4c7387f8904075b961612b5a08aff47c51c081c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
