)]}'
{"id":"openstack%2Foctavia~663459","triplet_id":"openstack%2Foctavia~master~I8f509677da463bc5d0a7649f2f609045cf9b2dd7","project":"openstack/octavia","branch":"master","topic":"663459","hashtags":[],"change_id":"I8f509677da463bc5d0a7649f2f609045cf9b2dd7","subject":"Add project_id to all of the provider objects","status":"MERGED","created":"2019-06-05 23:35:44.000000000","updated":"2019-07-22 22:38:56.000000000","submitted":"2019-07-22 22:38:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663459-1563835136831-cc061412","meta_rev_id":"5695ebcfc2d8ab79389732d20c50c50c1e6d1d26","_number":663459,"virtual_id_number":663459,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":2,"date":"2019-07-22 22:38:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-07-22 20:23:01.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-07-22 20:56:06.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-07-22 20:56:06.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-2,"date":"2019-07-22 20:23:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 20:23:01.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-07-22 20:56:06.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-07-22 22:38:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1a7bcd600b33eb305ab24789654d30266b50fa0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-05 23:35:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f018ba476d8aaf0d38e1361f0e012727902b462","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-05 23:37:54.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"53ffec05c2479c0bbdab5f0108bde86b86244b6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 01:50:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/59/663459/1/check/openstack-tox-cover/ab65813/ : FAILURE in 7m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/663459/1/check/openstack-tox-lower-constraints/16a4e75/ : FAILURE in 6m 03s\n- openstack-tox-pep8 http://logs.openstack.org/59/663459/1/check/openstack-tox-pep8/76ae190/ : SUCCESS in 8m 14s\n- openstack-tox-py27 http://logs.openstack.org/59/663459/1/check/openstack-tox-py27/75cc8f1/ : FAILURE in 5m 41s\n- openstack-tox-py36 http://logs.openstack.org/59/663459/1/check/openstack-tox-py36/b9f9fbe/ : FAILURE in 7m 29s\n- openstack-tox-py37 http://logs.openstack.org/59/663459/1/check/openstack-tox-py37/8def515/ : FAILURE in 5m 42s\n- openstack-tox-docs http://logs.openstack.org/59/663459/1/check/openstack-tox-docs/5937619/html/ : SUCCESS in 5m 44s\n- openstack-tox-functional http://logs.openstack.org/59/663459/1/check/openstack-tox-functional/7a1e787/ : FAILURE in 7m 25s\n- openstack-tox-functional-py36 http://logs.openstack.org/59/663459/1/check/openstack-tox-functional-py36/590bb21/ : FAILURE in 8m 09s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/59/663459/1/check/octavia-v2-dsvm-noop-api/939cdde/ : FAILURE in 38m 11s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/59/663459/1/check/octavia-v2-dsvm-noop-py2-api/979577e/ : FAILURE in 36m 36s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/59/663459/1/check/octavia-v2-dsvm-scenario/c2878ce/ : FAILURE in 1h 23m 50s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/59/663459/1/check/octavia-v2-dsvm-py2-scenario/801bfcb/ : FAILURE in 1h 22m 05s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/59/663459/1/check/octavia-v2-dsvm-py2-scenario-centos-7/5846f72/ : FAILURE in 2h 06m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/59/663459/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/c410d06/ : FAILURE in 1h 20m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/59/663459/1/check/octavia-v2-dsvm-scenario-two-node/fb0c691/ : FAILURE in 1h 43m 32s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/59/663459/1/check/octavia-v2-dsvm-py2-scenario-two-node/453f608/ : FAILURE in 1h 19m 49s (non-voting)\n- octavia-grenade http://logs.openstack.org/59/663459/1/check/octavia-grenade/f8c86da/ : FAILURE in 57m 16s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/59/663459/1/check/octavia-v2-dsvm-tls-barbican/f18e274/ : FAILURE in 48m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"761774b5e2f5b93dc36821289412f18f27384890","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-06 13:14:02.000000000","message":"Patch Set 1:\n\nThanks for this... If you want me to try and address the test failure in this patch just let me know.","accounts_in_message":[],"_revision_number":1},{"id":"b9db5e8b9689238006fe6aebc542d21f776fd089","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-06 14:59:24.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8bbc0d5917dfcc46fbcb15b87116ccb4b1615de5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 17:08:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/59/663459/2/check/openstack-tox-cover/3050d06/ : FAILURE in 5m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/663459/2/check/openstack-tox-lower-constraints/a396c77/ : FAILURE in 4m 57s\n- openstack-tox-pep8 http://logs.openstack.org/59/663459/2/check/openstack-tox-pep8/c032a84/ : SUCCESS in 6m 43s\n- openstack-tox-py27 http://logs.openstack.org/59/663459/2/check/openstack-tox-py27/adfbd9f/ : FAILURE in 5m 12s\n- openstack-tox-py36 http://logs.openstack.org/59/663459/2/check/openstack-tox-py36/e918c6f/ : FAILURE in 5m 09s\n- openstack-tox-py37 http://logs.openstack.org/59/663459/2/check/openstack-tox-py37/737fc2f/ : FAILURE in 5m 44s\n- openstack-tox-docs http://logs.openstack.org/59/663459/2/check/openstack-tox-docs/cd550b8/html/ : SUCCESS in 6m 01s\n- openstack-tox-functional http://logs.openstack.org/59/663459/2/check/openstack-tox-functional/65d7d80/ : FAILURE in 6m 23s\n- openstack-tox-functional-py36 http://logs.openstack.org/59/663459/2/check/openstack-tox-functional-py36/893d77b/ : FAILURE in 7m 09s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/59/663459/2/check/octavia-v2-dsvm-noop-api/54b29fa/ : SUCCESS in 58m 56s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/59/663459/2/check/octavia-v2-dsvm-noop-py2-api/f65faa0/ : SUCCESS in 40m 47s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/59/663459/2/check/octavia-v2-dsvm-scenario/fcee6b4/ : SUCCESS in 1h 33m 24s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/59/663459/2/check/octavia-v2-dsvm-py2-scenario/4aeda4d/ : SUCCESS in 1h 48m 41s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/59/663459/2/check/octavia-v2-dsvm-py2-scenario-centos-7/96054f0/ : SUCCESS in 2h 04m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/59/663459/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/396bfbb/ : SUCCESS in 1h 40m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/59/663459/2/check/octavia-v2-dsvm-scenario-two-node/48edfe2/ : SUCCESS in 1h 19m 22s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/59/663459/2/check/octavia-v2-dsvm-py2-scenario-two-node/8d7b9ba/ : SUCCESS in 1h 57m 09s (non-voting)\n- octavia-grenade http://logs.openstack.org/59/663459/2/check/octavia-grenade/5ef5f82/ : FAILURE in 1h 10m 41s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/59/663459/2/check/octavia-v2-dsvm-tls-barbican/86038a0/ : SUCCESS in 49m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/59/663459/2/check/octavia-v2-dsvm-scenario-amphora-v2/35e2c2b/ : SUCCESS in 1h 15m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bb3a37ff59c0d1c518d047d9dda4dad313bd0f9d","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-06 18:46:15.000000000","message":"Patch Set 2:\n\nVerified with vmware-nsx LBs, Listeners + Pools; looks good.","accounts_in_message":[],"_revision_number":2},{"id":"49c9a06e4f129ecfef72174a824cccde9ffe7e1c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-11 18:14:00.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"fe6a5db1052d013866ca4230e8011ace23ad494b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 20:26:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/59/663459/3/check/openstack-tox-cover/f5f8830/ : FAILURE in 5m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/663459/3/check/openstack-tox-lower-constraints/a80062e/ : FAILURE in 5m 17s\n- openstack-tox-pep8 http://logs.openstack.org/59/663459/3/check/openstack-tox-pep8/3359580/ : SUCCESS in 6m 38s\n- openstack-tox-py27 http://logs.openstack.org/59/663459/3/check/openstack-tox-py27/95d467f/ : FAILURE in 8m 09s\n- openstack-tox-py36 http://logs.openstack.org/59/663459/3/check/openstack-tox-py36/6d0b4b7/ : FAILURE in 6m 56s\n- openstack-tox-py37 http://logs.openstack.org/59/663459/3/check/openstack-tox-py37/0f543fb/ : FAILURE in 7m 43s\n- openstack-tox-docs http://logs.openstack.org/59/663459/3/check/openstack-tox-docs/8771232/html/ : SUCCESS in 6m 16s\n- openstack-tox-functional http://logs.openstack.org/59/663459/3/check/openstack-tox-functional/e242d9a/ : FAILURE in 7m 36s\n- openstack-tox-functional-py36 http://logs.openstack.org/59/663459/3/check/openstack-tox-functional-py36/d31d9fe/ : FAILURE in 6m 07s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-noop-api/ad199de/ : SUCCESS in 43m 34s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-noop-py2-api/42fbccb/ : SUCCESS in 44m 42s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-scenario/e7d0e00/ : SUCCESS in 1h 31m 56s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-py2-scenario/78d92cb/ : FAILURE in 28m 00s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-py2-scenario-centos-7/1ec43c3/ : FAILURE in 2h 00m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/497515b/ : SUCCESS in 1h 40m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-scenario-two-node/f4e291e/ : SUCCESS in 1h 33m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-py2-scenario-two-node/5e86d65/ : RETRY_LIMIT in 3m 36s (non-voting)\n- octavia-grenade http://logs.openstack.org/59/663459/3/check/octavia-grenade/badf58c/ : FAILURE in 1h 01m 53s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-tls-barbican/00e1215/ : SUCCESS in 51m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-scenario-amphora-v2/12ec3e0/ : SUCCESS in 1h 26m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0768f486eed294cb3e378a4ada6266125cc4c5c1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-14 18:00:51.000000000","message":"Patch Set 3:\n\nrecheck\nUC updated","accounts_in_message":[],"_revision_number":3},{"id":"adb3cfa7162caab344472722d9202c75db6af454","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 20:46:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/59/663459/3/check/openstack-tox-cover/656bea5/cover/ : SUCCESS in 8m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/663459/3/check/openstack-tox-lower-constraints/fc79a97/ : SUCCESS in 7m 28s\n- openstack-tox-pep8 http://logs.openstack.org/59/663459/3/check/openstack-tox-pep8/bcb9155/ : SUCCESS in 6m 42s\n- openstack-tox-py27 http://logs.openstack.org/59/663459/3/check/openstack-tox-py27/bcc9cab/ : SUCCESS in 5m 42s\n- openstack-tox-py36 http://logs.openstack.org/59/663459/3/check/openstack-tox-py36/d4cc70f/ : SUCCESS in 5m 26s\n- openstack-tox-py37 http://logs.openstack.org/59/663459/3/check/openstack-tox-py37/25295e9/ : SUCCESS in 10m 23s\n- openstack-tox-docs http://logs.openstack.org/59/663459/3/check/openstack-tox-docs/8762cae/html/ : SUCCESS in 6m 14s\n- openstack-tox-functional http://logs.openstack.org/59/663459/3/check/openstack-tox-functional/f523456/ : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 http://logs.openstack.org/59/663459/3/check/openstack-tox-functional-py36/bf0b679/ : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-noop-api/5cd873f/ : SUCCESS in 43m 58s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-noop-py2-api/f816cfa/ : SUCCESS in 43m 11s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-scenario/07aa5c4/ : SUCCESS in 1h 35m 36s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-py2-scenario/a5a61f5/ : SUCCESS in 1h 44m 19s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-py2-scenario-centos-7/1c18d8d/ : FAILURE in 2h 03m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/51494d4/ : SUCCESS in 1h 51m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-scenario-two-node/b2af9f1/ : SUCCESS in 1h 30m 33s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-py2-scenario-two-node/6aae2c7/ : SUCCESS in 2h 00m 47s (non-voting)\n- octavia-grenade http://logs.openstack.org/59/663459/3/check/octavia-grenade/276b677/ : SUCCESS in 1h 15m 54s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-tls-barbican/20cffe0/ : SUCCESS in 1h 07m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/59/663459/3/check/octavia-v2-dsvm-scenario-amphora-v2/5c8cd93/ : SUCCESS in 1h 36m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"62bdafcd05659c704603192f20cd4bc44d333c2b","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-22 12:30:14.000000000","message":"Patch Set 3:\n\nIs there a reason this hasn\u0027t landed yet?\nAs mentioned some time ago; it seems to work as expected based on my testing.","accounts_in_message":[],"_revision_number":3},{"id":"67a4ccc3297e776b8d377445b2ad155e0ca22604","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-22 12:36:34.000000000","message":"Patch Set 3: Backport-Candidate-2 Code-Review+2\n\nNo reason, I think. Review bandwidth is our current bottleneck.","accounts_in_message":[],"_revision_number":3},{"id":"0e2ada1a166d3e524aa62a149c2d335ea41e9037","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-22 12:47:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"924e784fe752691ae670af0fe8e39076b25e0df1","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-22 13:34:22.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c22f2a20fbe5f2a6249e86a3400774357eb6bd83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 13:34:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"67ba201a5a7e4d251e84fa410d2dec8070caa095","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 13:44:02.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/59/663459/3/gate/openstack-tox-lower-constraints/8e90230/ : FAILURE in 6m 32s\n- openstack-tox-pep8 finger://ze09.openstack.org/1b797693d87e4768a43b39a5c6ce40fc : CANCELED in 4m 24s\n- openstack-tox-py27 http://logs.openstack.org/59/663459/3/gate/openstack-tox-py27/4dfc1bd/ : SUCCESS in 3m 54s\n- openstack-tox-py36 finger://ze10.openstack.org/7c034bfdc8ab43f28dcf8ec5c176e6c1 : CANCELED in 1m 56s\n- openstack-tox-py37 finger://ze01.openstack.org/ec150c06e2064b48bd4e03adc70da5de : CANCELED in 4m 31s\n- openstack-tox-docs finger://ze02.openstack.org/e0b9a063868548ba83c9f4a1b00a4a96 : CANCELED in 5m 45s\n- openstack-tox-functional finger://ze05.openstack.org/7e386d7405b54990a8c03696480159f0 : CANCELED in 3m 23s\n- openstack-tox-functional-py36 http://logs.openstack.org/59/663459/3/gate/openstack-tox-functional-py36/4793942/ : SUCCESS in 6m 34s\n- octavia-v2-dsvm-noop-api finger://ze10.openstack.org/3c7d86f6db154396b1ce5c262bd2a84a : CANCELED in 3m 06s\n- octavia-v2-dsvm-noop-py2-api finger://ze03.openstack.org/a04fda7328434d3ca85e27a2bd887740 : CANCELED in 5m 46s\n- octavia-v2-dsvm-scenario finger://ze05.openstack.org/f36457b7033e407e96ea98f547b58674 : CANCELED in 6m 24s\n- octavia-v2-dsvm-py2-scenario finger://ze01.openstack.org/f3188730b6624289a12f29fef42544f7 : CANCELED in 5m 49s\n- octavia-grenade finger://ze01.openstack.org/23c848a30fff4ad58a8b8c76b80332d0 : CANCELED in 5m 44s","accounts_in_message":[],"_revision_number":3},{"id":"39cac41d35c364a5736ff7a180c8b4cfe3db1b44","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-22 15:16:11.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"25def5a5ba4f3c0f80fe8d683e765f4eac4937a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 17:37:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/59/663459/4/check/openstack-tox-cover/7f10579/cover/ : SUCCESS in 7m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/663459/4/check/openstack-tox-lower-constraints/402cf44/ : FAILURE in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/59/663459/4/check/openstack-tox-pep8/cf6bd03/ : SUCCESS in 5m 12s\n- openstack-tox-py27 http://logs.openstack.org/59/663459/4/check/openstack-tox-py27/a3d7367/ : SUCCESS in 7m 12s\n- openstack-tox-py36 http://logs.openstack.org/59/663459/4/check/openstack-tox-py36/6c980d0/ : SUCCESS in 4m 44s\n- openstack-tox-py37 http://logs.openstack.org/59/663459/4/check/openstack-tox-py37/242ec11/ : SUCCESS in 4m 57s\n- openstack-tox-docs http://logs.openstack.org/59/663459/4/check/openstack-tox-docs/e6b7434/html/ : SUCCESS in 5m 07s\n- openstack-tox-functional http://logs.openstack.org/59/663459/4/check/openstack-tox-functional/1a59ed3/ : SUCCESS in 7m 52s\n- openstack-tox-functional-py36 http://logs.openstack.org/59/663459/4/check/openstack-tox-functional-py36/1c186a2/ : SUCCESS in 7m 55s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/59/663459/4/check/octavia-v2-dsvm-noop-api/cac945e/ : SUCCESS in 52m 18s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/59/663459/4/check/octavia-v2-dsvm-noop-py2-api/001d38b/ : SUCCESS in 53m 11s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/59/663459/4/check/octavia-v2-dsvm-scenario/f8abf36/ : SUCCESS in 1h 37m 15s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/59/663459/4/check/octavia-v2-dsvm-py2-scenario/bf6bc45/ : SUCCESS in 1h 38m 54s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/59/663459/4/check/octavia-v2-dsvm-py2-scenario-centos-7/033a55b/ : FAILURE in 1h 45m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/59/663459/4/check/octavia-v2-dsvm-scenario-ubuntu-bionic/8d89c0a/ : SUCCESS in 1h 56m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/59/663459/4/check/octavia-v2-dsvm-scenario-two-node/6150501/ : SUCCESS in 2h 05m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/59/663459/4/check/octavia-v2-dsvm-py2-scenario-two-node/8812929/ : FAILURE in 1h 25m 55s (non-voting)\n- octavia-grenade http://logs.openstack.org/59/663459/4/check/octavia-grenade/2e393a2/ : SUCCESS in 1h 16m 23s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/59/663459/4/check/octavia-v2-dsvm-tls-barbican/c6fce68/ : SUCCESS in 1h 01m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/59/663459/4/check/octavia-v2-dsvm-scenario-amphora-v2/cebe172/ : SUCCESS in 1h 39m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e1fca00e62df1d0d568d207f21c6d03e0d86e106","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-22 17:44:29.000000000","message":"Patch Set 4:\n\nguess we need a new release of octavia-lib for the depends on patch herein","accounts_in_message":[],"_revision_number":4},{"id":"85e5b93124c6dfd86b27ca5ed397fbd1443e49be","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-22 17:50:08.000000000","message":"Patch Set 4: Code-Review-1\n\nAh, octavia-lib 1.2.0 includes https://review.opendev.org/#/c/663434/. The problem is this patch needs to bump the lower constraint version of octavia-lib to 1.2.0","accounts_in_message":[],"_revision_number":4},{"id":"99c2291d2125352d761015571028c3a23209af8e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-22 18:31:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6ac8643e6a6e5fdfaa1219d1ebd58d19441e9b51","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-22 18:49:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1d86f7ad139d44748bebbe3dfec944ef5120c6bd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-22 20:23:01.000000000","message":"Patch Set 6: Backport-Candidate-2 Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"a0d2d5f4adabcc7d025b48bd362ee3858a6cfa02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 20:54:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/59/663459/6/check/requirements-check/6c613b7/ : SUCCESS in 3m 50s\n- openstack-tox-cover http://logs.openstack.org/59/663459/6/check/openstack-tox-cover/0257db9/cover/ : SUCCESS in 8m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/663459/6/check/openstack-tox-lower-constraints/6c24461/ : SUCCESS in 7m 15s\n- openstack-tox-pep8 http://logs.openstack.org/59/663459/6/check/openstack-tox-pep8/83a6a26/ : SUCCESS in 6m 12s\n- openstack-tox-py27 http://logs.openstack.org/59/663459/6/check/openstack-tox-py27/a1f3ffe/ : SUCCESS in 4m 20s\n- openstack-tox-py36 http://logs.openstack.org/59/663459/6/check/openstack-tox-py36/d044d0c/ : SUCCESS in 4m 26s\n- openstack-tox-py37 http://logs.openstack.org/59/663459/6/check/openstack-tox-py37/b6f6cde/ : SUCCESS in 6m 32s\n- openstack-tox-docs http://logs.openstack.org/59/663459/6/check/openstack-tox-docs/82380e8/html/ : SUCCESS in 6m 53s\n- openstack-tox-functional http://logs.openstack.org/59/663459/6/check/openstack-tox-functional/309fbd1/ : SUCCESS in 7m 12s\n- openstack-tox-functional-py36 http://logs.openstack.org/59/663459/6/check/openstack-tox-functional-py36/f88f7c9/ : SUCCESS in 5m 51s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/59/663459/6/check/octavia-v2-dsvm-noop-api/4403b1f/ : SUCCESS in 46m 02s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/59/663459/6/check/octavia-v2-dsvm-noop-py2-api/4c7d965/ : SUCCESS in 45m 03s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/59/663459/6/check/octavia-v2-dsvm-scenario/5aa0ada/ : SUCCESS in 1h 19m 52s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/59/663459/6/check/octavia-v2-dsvm-py2-scenario/88ff890/ : SUCCESS in 1h 26m 57s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/59/663459/6/check/octavia-v2-dsvm-py2-scenario-centos-7/ab8718c/ : FAILURE in 2h 00m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/59/663459/6/check/octavia-v2-dsvm-scenario-ubuntu-bionic/4cae069/ : SUCCESS in 1h 42m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/59/663459/6/check/octavia-v2-dsvm-scenario-two-node/6f3425e/ : SUCCESS in 1h 33m 21s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/59/663459/6/check/octavia-v2-dsvm-py2-scenario-two-node/73d3a78/ : SUCCESS in 1h 41m 29s (non-voting)\n- octavia-grenade http://logs.openstack.org/59/663459/6/check/octavia-grenade/5f3b78c/ : SUCCESS in 1h 18m 39s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/59/663459/6/check/octavia-v2-dsvm-tls-barbican/4e3fb86/ : SUCCESS in 58m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/59/663459/6/check/octavia-v2-dsvm-scenario-amphora-v2/2a8a85c/ : SUCCESS in 1h 28m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2e0eafbb075300f6d23c1b63f32ac2eb5981aded","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-22 20:56:06.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nMy bad; I should\u0027ve seen that.\nThanks for handling it.","accounts_in_message":[],"_revision_number":6},{"id":"a82617e6807279140d630057d77b1ddebaec8609","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 20:56:19.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8a97b3ccda0525d82bfdfc46ba1ae309bee12956","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 22:38:56.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/59/663459/6/gate/requirements-check/6f0bc81/ : SUCCESS in 3m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/663459/6/gate/openstack-tox-lower-constraints/0f312f2/ : SUCCESS in 6m 55s\n- openstack-tox-pep8 http://logs.openstack.org/59/663459/6/gate/openstack-tox-pep8/e55f538/ : SUCCESS in 6m 08s\n- openstack-tox-py27 http://logs.openstack.org/59/663459/6/gate/openstack-tox-py27/2694f4f/ : SUCCESS in 5m 00s\n- openstack-tox-py36 http://logs.openstack.org/59/663459/6/gate/openstack-tox-py36/6e17afb/ : SUCCESS in 4m 04s\n- openstack-tox-py37 http://logs.openstack.org/59/663459/6/gate/openstack-tox-py37/7319cef/ : SUCCESS in 4m 16s\n- openstack-tox-docs http://logs.openstack.org/59/663459/6/gate/openstack-tox-docs/7d83270/html/ : SUCCESS in 5m 57s\n- openstack-tox-functional http://logs.openstack.org/59/663459/6/gate/openstack-tox-functional/e286e8c/ : SUCCESS in 4m 55s\n- openstack-tox-functional-py36 http://logs.openstack.org/59/663459/6/gate/openstack-tox-functional-py36/01a7d54/ : SUCCESS in 5m 34s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/59/663459/6/gate/octavia-v2-dsvm-noop-api/97065b6/ : SUCCESS in 46m 43s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/59/663459/6/gate/octavia-v2-dsvm-noop-py2-api/a51d59d/ : SUCCESS in 41m 30s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/59/663459/6/gate/octavia-v2-dsvm-scenario/07936f1/ : SUCCESS in 1h 27m 11s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/59/663459/6/gate/octavia-v2-dsvm-py2-scenario/1a43e79/ : SUCCESS in 1h 37m 06s\n- octavia-grenade http://logs.openstack.org/59/663459/6/gate/octavia-grenade/a1fdf8f/ : SUCCESS in 1h 15m 47s","accounts_in_message":[],"_revision_number":6},{"id":"27bd11d617264ecd1b1b82e6146c00504c640b49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 22:38:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"45cf2f328f7410955934cbd8d9f2adc21944c52b","revisions":{"38b7fd169522cb7befa96aa1ba300ee1fd64ee21":{"kind":"REWORK","_number":1,"created":"2019-06-05 23:35:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/59/663459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/663459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/663459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/663459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/663459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/663459/1"}}},"commit":{"parents":[{"commit":"6d2e2be86af84bc117e439e6ec13e55360fecdc9","subject":"Remove references to OpenStack Anchor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d2e2be86af84bc117e439e6ec13e55360fecdc9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-05 23:34:38.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-05 23:35:43.000000000","tz":-420},"subject":"Add project_id to all of the provider objects","message":"Add project_id to all of the provider objects\n\nThis patch adds a project_id field to all of the provider driver data\nmodel objects.\n\nThis is useful for drivers to track/associate objects on the backend.\n\nChange-Id: I8f509677da463bc5d0a7649f2f609045cf9b2dd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38b7fd169522cb7befa96aa1ba300ee1fd64ee21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38b7fd169522cb7befa96aa1ba300ee1fd64ee21"}]},"branch":"refs/heads/master"},"f7eeee0fb570b7ad77cda6e507506024f170b541":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-06 14:59:24.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/59/663459/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/663459/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/663459/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/663459/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/663459/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/663459/2"}}},"commit":{"parents":[{"commit":"6d2e2be86af84bc117e439e6ec13e55360fecdc9","subject":"Remove references to OpenStack Anchor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d2e2be86af84bc117e439e6ec13e55360fecdc9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-05 23:34:38.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-06 14:59:21.000000000","tz":0},"subject":"Add project_id to all of the provider objects","message":"Add project_id to all of the provider objects\n\nThis patch adds a project_id field to all of the provider driver data\nmodel objects.\n\nThis is useful for drivers to track/associate objects on the backend.\n\nDepends-On: https://review.opendev.org/663434\n\nChange-Id: I8f509677da463bc5d0a7649f2f609045cf9b2dd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f7eeee0fb570b7ad77cda6e507506024f170b541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f7eeee0fb570b7ad77cda6e507506024f170b541"}]},"branch":"refs/heads/master"},"2b51ed5bfe7a27eda54a8c339566a1a3b490666b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-11 18:14:00.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/59/663459/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/663459/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/663459/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/663459/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/663459/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/663459/3"}}},"commit":{"parents":[{"commit":"620c19804cd518af807839c877a01fe56db7462d","subject":"Merge \"Add RHEL 8 amphora support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/620c19804cd518af807839c877a01fe56db7462d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-05 23:34:38.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-11 18:14:00.000000000","tz":0},"subject":"Add project_id to all of the provider objects","message":"Add project_id to all of the provider objects\n\nThis patch adds a project_id field to all of the provider driver data\nmodel objects.\n\nThis is useful for drivers to track/associate objects on the backend.\n\nDepends-On: https://review.opendev.org/663434\n\nChange-Id: I8f509677da463bc5d0a7649f2f609045cf9b2dd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2b51ed5bfe7a27eda54a8c339566a1a3b490666b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2b51ed5bfe7a27eda54a8c339566a1a3b490666b"}]},"branch":"refs/heads/master"},"a15a54b5a245005e25cb72268768188f8d4a1e4d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-22 15:16:11.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/59/663459/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/663459/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/663459/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/663459/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/663459/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/663459/4"}}},"commit":{"parents":[{"commit":"2bb2f918ee4fb38b7a17fb6f8781042d204b05e8","subject":"Fix cryptsetup --pbkdf-memory failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2bb2f918ee4fb38b7a17fb6f8781042d204b05e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-05 23:34:38.000000000","tz":-420},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-07-22 15:16:10.000000000","tz":-360},"subject":"Add project_id to all of the provider objects","message":"Add project_id to all of the provider objects\n\nThis patch adds a project_id field to all of the provider driver data\nmodel objects.\n\nThis is useful for drivers to track/associate objects on the backend.\n\nDepends-On: https://review.opendev.org/663434\n\nChange-Id: I8f509677da463bc5d0a7649f2f609045cf9b2dd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a15a54b5a245005e25cb72268768188f8d4a1e4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a15a54b5a245005e25cb72268768188f8d4a1e4d"}]},"branch":"refs/heads/master"},"846a4aebaaa64e6c2eca7757b3503b1ee0775f16":{"kind":"REWORK","_number":5,"created":"2019-07-22 18:31:10.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/59/663459/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/663459/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/663459/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/663459/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/663459/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/663459/5"}}},"commit":{"parents":[{"commit":"2bb2f918ee4fb38b7a17fb6f8781042d204b05e8","subject":"Fix cryptsetup --pbkdf-memory failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2bb2f918ee4fb38b7a17fb6f8781042d204b05e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-05 23:34:38.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-22 18:30:43.000000000","tz":120},"subject":"Add project_id to all of the provider objects","message":"Add project_id to all of the provider objects\n\nThis patch adds a project_id field to all of the provider driver data\nmodel objects.\n\nThis is useful for drivers to track/associate objects on the backend.\n\nChange-Id: I8f509677da463bc5d0a7649f2f609045cf9b2dd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/846a4aebaaa64e6c2eca7757b3503b1ee0775f16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/846a4aebaaa64e6c2eca7757b3503b1ee0775f16"}]},"branch":"refs/heads/master"},"45cf2f328f7410955934cbd8d9f2adc21944c52b":{"kind":"REWORK","_number":6,"created":"2019-07-22 18:49:29.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/59/663459/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/663459/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/663459/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/663459/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/663459/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/663459/6"}}},"commit":{"parents":[{"commit":"2bb2f918ee4fb38b7a17fb6f8781042d204b05e8","subject":"Fix cryptsetup --pbkdf-memory failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2bb2f918ee4fb38b7a17fb6f8781042d204b05e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-05 23:34:38.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-22 18:49:19.000000000","tz":120},"subject":"Add project_id to all of the provider objects","message":"Add project_id to all of the provider objects\n\nThis patch adds a project_id field to all of the provider driver data\nmodel objects.\n\nThis is useful for drivers to track/associate objects on the backend.\n\nChange-Id: I8f509677da463bc5d0a7649f2f609045cf9b2dd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45cf2f328f7410955934cbd8d9f2adc21944c52b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45cf2f328f7410955934cbd8d9f2adc21944c52b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
