)]}'
{"id":"openstack%2Foctavia~663784","triplet_id":"openstack%2Foctavia~master~I018e36f69a9c0b48a6651a01cc9a64abfc04d4de","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I018e36f69a9c0b48a6651a01cc9a64abfc04d4de","subject":"Limit cryptsetup key RAM usage","status":"MERGED","created":"2019-06-06 22:10:37.000000000","updated":"2019-06-07 21:46:18.000000000","submitted":"2019-06-07 15:35:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663784-1559921745419-d9d96933","meta_rev_id":"cffa845dd8c2efa9553743f7d5ca102ac4f8a9b7","_number":663784,"virtual_id_number":663784,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-06-07 15:35:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"all":[{"value":2,"date":"2019-06-07 13:47:09.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":2,"date":"2019-06-07 06:32:41.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"all":[{"value":1,"date":"2019-06-07 13:47:09.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":2,"date":"2019-06-07 06:32:41.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 06:32:41.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-06-07 13:47:09.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-06-07 15:35:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c30eb23d7a3fed4769792e66c5dcb831b1c4759b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-06 22:10:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abbbeae76f2af057cc935ea5aefdc15bb165ad1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 00:45:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/84/663784/1/check/openstack-tox-cover/db78308/cover/ : SUCCESS in 8m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/84/663784/1/check/openstack-tox-lower-constraints/abe3bdb/ : SUCCESS in 7m 04s\n- openstack-tox-pep8 http://logs.openstack.org/84/663784/1/check/openstack-tox-pep8/d31cabe/ : SUCCESS in 6m 08s\n- openstack-tox-py27 http://logs.openstack.org/84/663784/1/check/openstack-tox-py27/a2dc688/ : SUCCESS in 5m 33s\n- openstack-tox-py36 http://logs.openstack.org/84/663784/1/check/openstack-tox-py36/031fc61/ : SUCCESS in 5m 44s\n- openstack-tox-py37 http://logs.openstack.org/84/663784/1/check/openstack-tox-py37/c149362/ : SUCCESS in 7m 37s\n- openstack-tox-docs http://logs.openstack.org/84/663784/1/check/openstack-tox-docs/3c2dc12/html/ : SUCCESS in 6m 46s\n- openstack-tox-functional http://logs.openstack.org/84/663784/1/check/openstack-tox-functional/a317e44/ : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 http://logs.openstack.org/84/663784/1/check/openstack-tox-functional-py36/e368aa3/ : SUCCESS in 6m 35s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/84/663784/1/check/octavia-v2-dsvm-noop-api/41d36a6/ : SUCCESS in 44m 54s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/84/663784/1/check/octavia-v2-dsvm-noop-py2-api/4b7b93d/ : SUCCESS in 42m 38s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/84/663784/1/check/octavia-v2-dsvm-scenario/ce49b13/ : SUCCESS in 1h 52m 10s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/84/663784/1/check/octavia-v2-dsvm-py2-scenario/a701432/ : SUCCESS in 1h 43m 05s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/84/663784/1/check/octavia-v2-dsvm-py2-scenario-centos-7/d201524/ : FAILURE in 2h 23m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/84/663784/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/84e9a4f/ : SUCCESS in 1h 46m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/84/663784/1/check/octavia-v2-dsvm-scenario-two-node/9a6d356/ : SUCCESS in 1h 16m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/84/663784/1/check/octavia-v2-dsvm-py2-scenario-two-node/8cf7242/ : SUCCESS in 1h 51m 46s (non-voting)\n- octavia-grenade http://logs.openstack.org/84/663784/1/check/octavia-grenade/bcf3eeb/ : SUCCESS in 1h 07m 04s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/84/663784/1/check/octavia-v2-dsvm-tls-barbican/09ac652/ : SUCCESS in 49m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/84/663784/1/check/octavia-v2-dsvm-scenario-amphora-v2/b309a48/ : SUCCESS in 1h 19m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e49786b4e1e2442313929da24012b82dcc5bd3ce","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-07 06:32:41.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5972be3c751f6c69a11b7449ec6002495b9bde63","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-06-07 13:47:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e93ca1ceac772e9862d2b3499fc30e7738529ccf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:47:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23f21ffc553d0deb9dc14b8313f2c43b0cd87796","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 15:35:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"425131460779b640812597a2161bfde920251869","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 15:35:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/84/663784/1/gate/openstack-tox-lower-constraints/870c5a5/ : SUCCESS in 6m 39s\n- openstack-tox-pep8 http://logs.openstack.org/84/663784/1/gate/openstack-tox-pep8/7b47f99/ : SUCCESS in 6m 37s\n- openstack-tox-py27 http://logs.openstack.org/84/663784/1/gate/openstack-tox-py27/84174cf/ : SUCCESS in 5m 17s\n- openstack-tox-py36 http://logs.openstack.org/84/663784/1/gate/openstack-tox-py36/40ec78e/ : SUCCESS in 5m 55s\n- openstack-tox-py37 http://logs.openstack.org/84/663784/1/gate/openstack-tox-py37/5640e0d/ : SUCCESS in 7m 32s\n- openstack-tox-docs http://logs.openstack.org/84/663784/1/gate/openstack-tox-docs/d3afd87/html/ : SUCCESS in 5m 43s\n- openstack-tox-functional http://logs.openstack.org/84/663784/1/gate/openstack-tox-functional/839fe14/ : SUCCESS in 6m 19s\n- openstack-tox-functional-py36 http://logs.openstack.org/84/663784/1/gate/openstack-tox-functional-py36/a97f626/ : SUCCESS in 6m 26s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/84/663784/1/gate/octavia-v2-dsvm-noop-api/c813ca0/ : SUCCESS in 39m 56s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/84/663784/1/gate/octavia-v2-dsvm-noop-py2-api/69803d5/ : SUCCESS in 45m 19s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/84/663784/1/gate/octavia-v2-dsvm-scenario/cdb21ef/ : SUCCESS in 1h 39m 50s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/84/663784/1/gate/octavia-v2-dsvm-py2-scenario/1529604/ : SUCCESS in 1h 32m 12s\n- octavia-grenade http://logs.openstack.org/84/663784/1/gate/octavia-grenade/7058b2b/ : SUCCESS in 1h 05m 10s","accounts_in_message":[],"_revision_number":1},{"id":"cffa845dd8c2efa9553743f7d5ca102ac4f8a9b7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-07 21:46:18.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 4b912e48295372cab31bd9f0b4da99411dd5b6e8","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4007d56b3eda356a476ac583fbf3f92cb0f179ef","revisions":{"4007d56b3eda356a476ac583fbf3f92cb0f179ef":{"kind":"REWORK","_number":1,"created":"2019-06-06 22:10:37.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/84/663784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/663784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/663784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/663784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/663784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/663784/1"}}},"commit":{"parents":[{"commit":"ff4680eb71cf03e4eae1a58e7a66321ddadcbead","subject":"Merge \"Create Amphora V2 provider driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff4680eb71cf03e4eae1a58e7a66321ddadcbead"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-06 22:06:54.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-06 22:10:36.000000000","tz":-420},"subject":"Limit cryptsetup key RAM usage","message":"Limit cryptsetup key RAM usage\n\nThis patch limits cryptsetup to 256MB of RAM during the amphora\nstartup. Recent distros have changed to LUKS2 with Argon2\nkey derivation which defaults to using up to 1GB of RAM.\nTypically our amphora are built with only 1GB of RAM for the whole\nsystem.\n\nChange-Id: I018e36f69a9c0b48a6651a01cc9a64abfc04d4de\nStory: 2005837\nTask: 33606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4007d56b3eda356a476ac583fbf3f92cb0f179ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4007d56b3eda356a476ac583fbf3f92cb0f179ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
