)]}'
{"id":"openstack%2Foctavia~664068","triplet_id":"openstack%2Foctavia~stable%2Fstein~I018e36f69a9c0b48a6651a01cc9a64abfc04d4de","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I018e36f69a9c0b48a6651a01cc9a64abfc04d4de","subject":"Limit cryptsetup key RAM usage","status":"MERGED","created":"2019-06-07 21:46:18.000000000","updated":"2019-06-09 13:25:59.000000000","submitted":"2019-06-09 13:25:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664068-1560086759634-95369c51","meta_rev_id":"967d1bc32c4dd287e36df0a5ef288fbc48ed5de0","_number":664068,"virtual_id_number":664068,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-06-09 13:25:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"all":[{"value":2,"date":"2019-06-09 11:39:43.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":2,"date":"2019-06-07 21:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"all":[{"value":1,"date":"2019-06-09 11:39:43.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 21:47:03.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-06-09 11:39:43.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-06-09 13:25:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"52c1c2ffad06654670a303f6011f410a688b5f79","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-07 21:46:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8e8715eafd24cddc76be3230692963f4ae458815","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-07 21:47:03.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"4a1fafe8d00ea56260f7b649025776568b86787e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 00:06:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/68/664068/1/check/openstack-tox-cover/2f5befa/cover/ : SUCCESS in 12m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/664068/1/check/openstack-tox-lower-constraints/2202568/ : SUCCESS in 7m 49s\n- openstack-tox-pep8 http://logs.openstack.org/68/664068/1/check/openstack-tox-pep8/aa3eeb9/ : SUCCESS in 6m 38s\n- openstack-tox-py27 http://logs.openstack.org/68/664068/1/check/openstack-tox-py27/c594abc/ : SUCCESS in 5m 21s\n- openstack-tox-py35 http://logs.openstack.org/68/664068/1/check/openstack-tox-py35/82c4338/ : SUCCESS in 5m 24s\n- openstack-tox-py36 http://logs.openstack.org/68/664068/1/check/openstack-tox-py36/2615544/ : SUCCESS in 5m 52s\n- openstack-tox-docs http://logs.openstack.org/68/664068/1/check/openstack-tox-docs/53244b9/html/ : SUCCESS in 7m 17s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/68/664068/1/check/neutron-lbaasv2-dsvm-api/7706403/ : SUCCESS in 1h 16m 02s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/68/664068/1/check/neutron-lbaasv2-dsvm-py3x-api/f2682ce/ : SUCCESS in 1h 17m 43s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/68/664068/1/check/neutron-lbaasv2-dsvm-scenario/2b4c7e0/ : FAILURE in 1h 12m 52s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/68/664068/1/check/neutron-lbaasv2-dsvm-py3x-scenario/16c1a29/ : FAILURE in 1h 41m 24s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/68/664068/1/check/openstack-tox-functional/e45cc50/ : SUCCESS in 7m 25s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/664068/1/check/openstack-tox-functional-py35/0ffeedf/ : SUCCESS in 10m 10s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/664068/1/check/openstack-tox-functional-py36/c541d9a/ : SUCCESS in 7m 35s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/68/664068/1/check/octavia-v1-dsvm-scenario/881a9a3/ : SUCCESS in 59m 59s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/68/664068/1/check/octavia-v1-dsvm-py3x-scenario/a249e44/ : SUCCESS in 1h 05m 29s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/664068/1/check/octavia-v2-dsvm-noop-api/c364c80/ : SUCCESS in 45m 15s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/664068/1/check/octavia-v2-dsvm-noop-py2-api/e709961/ : SUCCESS in 50m 41s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/664068/1/check/octavia-v2-dsvm-scenario/f8f55b5/ : SUCCESS in 1h 39m 26s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/664068/1/check/octavia-v2-dsvm-py2-scenario/c8cadd3/ : SUCCESS in 1h 38m 51s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/664068/1/check/octavia-v2-dsvm-py2-scenario-centos-7/884c2c0/ : FAILURE in 2h 03m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/664068/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/890a3c2/ : SUCCESS in 1h 39m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/664068/1/check/octavia-v2-dsvm-scenario-two-node/30bd3af/ : SUCCESS in 1h 49m 21s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/664068/1/check/octavia-v2-dsvm-py2-scenario-two-node/e570941/ : SUCCESS in 1h 51m 25s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/664068/1/check/octavia-grenade/de9a06d/ : SUCCESS in 1h 14m 11s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/664068/1/check/octavia-v2-dsvm-tls-barbican/435c52d/ : SUCCESS in 52m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d51ca27a45e97e076736d758ed06d4a24e34fd1","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-06-09 11:39:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport. LGTM","accounts_in_message":[],"_revision_number":1},{"id":"8774fd93c7b4486c941d009f7a4ee97a49fb5020","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 11:39:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"471f19dc85c682d465ea30d77fb763a1086272a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 13:25:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/68/664068/1/gate/openstack-tox-lower-constraints/d28b636/ : SUCCESS in 7m 58s\n- openstack-tox-pep8 http://logs.openstack.org/68/664068/1/gate/openstack-tox-pep8/a2c5471/ : SUCCESS in 6m 24s\n- openstack-tox-py27 http://logs.openstack.org/68/664068/1/gate/openstack-tox-py27/58f1ca9/ : SUCCESS in 5m 45s\n- openstack-tox-py35 http://logs.openstack.org/68/664068/1/gate/openstack-tox-py35/24f2819/ : SUCCESS in 5m 40s\n- openstack-tox-py36 http://logs.openstack.org/68/664068/1/gate/openstack-tox-py36/ec41e0c/ : SUCCESS in 8m 19s\n- openstack-tox-docs http://logs.openstack.org/68/664068/1/gate/openstack-tox-docs/2777e89/html/ : SUCCESS in 6m 50s\n- openstack-tox-functional http://logs.openstack.org/68/664068/1/gate/openstack-tox-functional/60b8e67/ : SUCCESS in 9m 25s\n- openstack-tox-functional-py35 http://logs.openstack.org/68/664068/1/gate/openstack-tox-functional-py35/f227fc8/ : SUCCESS in 7m 34s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/664068/1/gate/openstack-tox-functional-py36/81f9494/ : SUCCESS in 9m 43s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/68/664068/1/gate/octavia-v1-dsvm-scenario/c9cf1bd/ : SUCCESS in 59m 21s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/68/664068/1/gate/octavia-v1-dsvm-py3x-scenario/716524d/ : SUCCESS in 1h 15m 07s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/68/664068/1/gate/neutron-lbaasv2-dsvm-api/6855e9b/ : SUCCESS in 1h 28m 37s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/68/664068/1/gate/neutron-lbaasv2-dsvm-py3x-api/66eb839/ : SUCCESS in 1h 27m 54s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/664068/1/gate/octavia-v2-dsvm-noop-api/fc890e8/ : SUCCESS in 54m 20s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/664068/1/gate/octavia-v2-dsvm-noop-py2-api/d542a40/ : SUCCESS in 58m 06s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/664068/1/gate/octavia-v2-dsvm-scenario/b9928bb/ : SUCCESS in 1h 33m 15s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/664068/1/gate/octavia-v2-dsvm-py2-scenario/3d82239/ : SUCCESS in 1h 41m 15s\n- octavia-grenade http://logs.openstack.org/68/664068/1/gate/octavia-grenade/6dcbdb1/ : SUCCESS in 1h 11m 59s","accounts_in_message":[],"_revision_number":1},{"id":"967d1bc32c4dd287e36df0a5ef288fbc48ed5de0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 13:25:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b912e48295372cab31bd9f0b4da99411dd5b6e8","revisions":{"4b912e48295372cab31bd9f0b4da99411dd5b6e8":{"kind":"REWORK","_number":1,"created":"2019-06-07 21:46:18.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/68/664068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/664068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/664068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/664068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/664068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/664068/1"}}},"commit":{"parents":[{"commit":"3ac8f296953d5bf4951d3278428f51b29118aa91","subject":"Performance improvement for non-udp health checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ac8f296953d5bf4951d3278428f51b29118aa91"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-06 22:06:54.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-07 21:46:18.000000000","tz":0},"subject":"Limit cryptsetup key RAM usage","message":"Limit cryptsetup key RAM usage\n\nThis patch limits cryptsetup to 256MB of RAM during the amphora\nstartup. Recent distros have changed to LUKS2 with Argon2\nkey derivation which defaults to using up to 1GB of RAM.\nTypically our amphora are built with only 1GB of RAM for the whole\nsystem.\n\nChange-Id: I018e36f69a9c0b48a6651a01cc9a64abfc04d4de\nStory: 2005837\nTask: 33606\n(cherry picked from commit 4007d56b3eda356a476ac583fbf3f92cb0f179ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b912e48295372cab31bd9f0b4da99411dd5b6e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b912e48295372cab31bd9f0b4da99411dd5b6e8"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
