)]}'
{"id":"openstack%2Foctavia~665032","triplet_id":"openstack%2Foctavia~master~I3eebcf374ca6350a530ce44a3ecccac926773186","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I3eebcf374ca6350a530ce44a3ecccac926773186","subject":"DNM: experimental element for centos debug","status":"ABANDONED","created":"2019-06-12 23:00:32.000000000","updated":"2019-06-13 17:44:00.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b1448cfb82aaf0c32aba39f8c25d742c2a8052dd","_number":665032,"virtual_id_number":665032,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":-1,"date":"2019-06-13 15:42:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 01:24:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-13 07:02:44.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"6cb8929e5ca2a49b22bc0eee9ba6f80a84db0828","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-12 23:00:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c8466332fb4a3abeffd4df24c3fc3255bbb98c6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-12 23:00:54.000000000","message":"Patch Set 1: Workflow-1\n\nThis is an experiment.","accounts_in_message":[],"_revision_number":1},{"id":"91cc6ea8b26c50b0a2b016ca7a2bc385923ab8c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 01:24:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/32/665032/1/check/openstack-tox-cover/1437973/cover/ : SUCCESS in 11m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/665032/1/check/openstack-tox-lower-constraints/0c6cb7a/ : SUCCESS in 7m 27s\n- openstack-tox-pep8 http://logs.openstack.org/32/665032/1/check/openstack-tox-pep8/2ed99d5/ : SUCCESS in 6m 40s\n- openstack-tox-py27 http://logs.openstack.org/32/665032/1/check/openstack-tox-py27/f754e90/ : SUCCESS in 5m 08s\n- openstack-tox-py36 http://logs.openstack.org/32/665032/1/check/openstack-tox-py36/c2ac05c/ : SUCCESS in 6m 23s\n- openstack-tox-py37 http://logs.openstack.org/32/665032/1/check/openstack-tox-py37/63aba31/ : SUCCESS in 8m 31s\n- openstack-tox-docs http://logs.openstack.org/32/665032/1/check/openstack-tox-docs/48a76bd/html/ : SUCCESS in 6m 28s\n- openstack-tox-functional http://logs.openstack.org/32/665032/1/check/openstack-tox-functional/e5b8c99/ : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 http://logs.openstack.org/32/665032/1/check/openstack-tox-functional-py36/fe4f94b/ : SUCCESS in 6m 36s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/32/665032/1/check/octavia-v2-dsvm-noop-api/326d9b0/ : SUCCESS in 42m 04s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/32/665032/1/check/octavia-v2-dsvm-noop-py2-api/5fda757/ : SUCCESS in 40m 16s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/32/665032/1/check/octavia-v2-dsvm-scenario/8bf6235/ : SUCCESS in 1h 25m 51s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/32/665032/1/check/octavia-v2-dsvm-py2-scenario/5c0358a/ : SUCCESS in 1h 33m 57s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/32/665032/1/check/octavia-v2-dsvm-py2-scenario-centos-7/47a0ba1/ : FAILURE in 2h 05m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/32/665032/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/ca66bdb/ : SUCCESS in 1h 36m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/32/665032/1/check/octavia-v2-dsvm-scenario-two-node/33e573b/ : SUCCESS in 1h 46m 16s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/32/665032/1/check/octavia-v2-dsvm-py2-scenario-two-node/6b1d9e4/ : SUCCESS in 1h 47m 18s (non-voting)\n- octavia-grenade http://logs.openstack.org/32/665032/1/check/octavia-grenade/1bfb42d/ : SUCCESS in 1h 17m 55s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/32/665032/1/check/octavia-v2-dsvm-tls-barbican/68dca36/ : SUCCESS in 53m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/32/665032/1/check/octavia-v2-dsvm-scenario-amphora-v2/bfade4e/ : SUCCESS in 1h 40m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73d5a2f143328fc1511774a2876e7df1e65bddd0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-13 07:02:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nKdump service failing to start appears not to be the problem we\u0027re trying to fix -- the centos job still failed. Although, I concur with the last part of the last sentence in the comment message. So, instead of disabling, why not removing kdump (package kexec-tools)? It would also slightly reduce the image size.\n\nhttps://access.redhat.com/documentation/en-us/red_hat_enterprise_linux/7/html/kernel_administration_guide/kernel_crash_dump_guide","accounts_in_message":[],"_revision_number":1},{"id":"7a2a238236807a289e0e5c4f570b42ede7d53603","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-13 08:05:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a158b775ddd24423baf27e39598166bad42fae6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-13 13:56:11.000000000","message":"Abandoned\n\nThis did not seem to address the problem","accounts_in_message":[],"_revision_number":1},{"id":"1f193e6230f9c00323540bedc83792945261e5e9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-13 15:35:59.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"37549270c8ce4468901e9ecff91068ff818d4a46","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-13 15:40:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bf86759d132270c5db88dab214b99e72afd0ff4a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-13 15:42:38.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"b1448cfb82aaf0c32aba39f8c25d742c2a8052dd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-13 17:44:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0bf67b543230b0335873d76945ac4ef26398165c","revisions":{"05233a7b06f1f535946c488ccc72718c358488d2":{"kind":"REWORK","_number":1,"created":"2019-06-12 23:00:32.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/32/665032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/665032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/665032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/665032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/665032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/665032/1"}}},"commit":{"parents":[{"commit":"620c19804cd518af807839c877a01fe56db7462d","subject":"Merge \"Add RHEL 8 amphora support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/620c19804cd518af807839c877a01fe56db7462d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-12 22:59:39.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-12 23:00:30.000000000","tz":-420},"subject":"Disable the kdump service in the amphora","message":"Disable the kdump service in the amphora\n\nWe have also seen instances where kdump is causing the boot to fail\nbecause the amphora image requires so little RAM the auto memory\nallocation for kdump reserves zero RAM.\n\nThis element disables the kdump service in the amphora as it requires\nRAM and since we actively manage the amphora it is unlikely the crash file\nwill be around long enough to be analyzed.\n\nChange-Id: I3eebcf374ca6350a530ce44a3ecccac926773186\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05233a7b06f1f535946c488ccc72718c358488d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05233a7b06f1f535946c488ccc72718c358488d2"}]},"branch":"refs/heads/master"},"0bf67b543230b0335873d76945ac4ef26398165c":{"kind":"REWORK","_number":2,"created":"2019-06-13 15:40:24.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/32/665032/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/665032/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/665032/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/665032/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/665032/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/665032/2"}}},"commit":{"parents":[{"commit":"620c19804cd518af807839c877a01fe56db7462d","subject":"Merge \"Add RHEL 8 amphora support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/620c19804cd518af807839c877a01fe56db7462d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-12 22:59:39.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-13 15:39:51.000000000","tz":-420},"subject":"DNM: experimental element for centos debug","message":"DNM: experimental element for centos debug\n\nChange-Id: I3eebcf374ca6350a530ce44a3ecccac926773186\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0bf67b543230b0335873d76945ac4ef26398165c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0bf67b543230b0335873d76945ac4ef26398165c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
