)]}'
{"id":"openstack%2Foctavia~665089","triplet_id":"openstack%2Foctavia~master~I349a646b54233f35392ecbd1085f75c6e2e795c6","project":"openstack/octavia","branch":"master","topic":"665089","hashtags":[],"change_id":"I349a646b54233f35392ecbd1085f75c6e2e795c6","subject":"DNM: systemd-219-64.el7","status":"ABANDONED","created":"2019-06-13 08:00:56.000000000","updated":"2019-06-13 17:51:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4cb59285334a2723ed067fda9756e8be84f20030","_number":665089,"virtual_id_number":665089,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-06-13 14:50:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 14:50:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fec5eac934fd2499cdfe79ce708b0696012b8a26","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-13 08:00:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15e045325283dd1ea06307edfc66e391ce7f9300","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-13 08:07:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7f4a7a5f178611e0ec848ecc5fd177546c4c15e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 10:04:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/89/665089/2/check/openstack-tox-cover/d025ebd/cover/ : SUCCESS in 8m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/665089/2/check/openstack-tox-lower-constraints/5646fe4/ : SUCCESS in 7m 36s\n- openstack-tox-pep8 http://logs.openstack.org/89/665089/2/check/openstack-tox-pep8/ebf77d5/ : SUCCESS in 6m 36s\n- openstack-tox-py27 http://logs.openstack.org/89/665089/2/check/openstack-tox-py27/e831c9c/ : SUCCESS in 5m 19s\n- openstack-tox-py36 http://logs.openstack.org/89/665089/2/check/openstack-tox-py36/3297db5/ : SUCCESS in 7m 45s\n- openstack-tox-py37 http://logs.openstack.org/89/665089/2/check/openstack-tox-py37/49898ce/ : SUCCESS in 7m 40s\n- openstack-tox-docs http://logs.openstack.org/89/665089/2/check/openstack-tox-docs/2ffc0c5/html/ : SUCCESS in 8m 48s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/89/665089/2/check/octavia-v2-dsvm-py2-scenario-centos-7/cdcaf10/ : FAILURE in 1h 56m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"16fe16ecd307fd8493348d2b9deffd8212c863f2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-13 12:35:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c92fd25f44df502c2ee319dbb0dad5022710509a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 14:50:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/89/665089/3/check/openstack-tox-cover/51edd89/cover/ : SUCCESS in 10m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/665089/3/check/openstack-tox-lower-constraints/406c479/ : SUCCESS in 7m 07s\n- openstack-tox-pep8 http://logs.openstack.org/89/665089/3/check/openstack-tox-pep8/86b6d65/ : SUCCESS in 7m 48s\n- openstack-tox-py27 http://logs.openstack.org/89/665089/3/check/openstack-tox-py27/50c169d/ : SUCCESS in 5m 00s\n- openstack-tox-py36 http://logs.openstack.org/89/665089/3/check/openstack-tox-py36/6881ea6/ : SUCCESS in 5m 44s\n- openstack-tox-py37 http://logs.openstack.org/89/665089/3/check/openstack-tox-py37/6b9ab5e/ : SUCCESS in 5m 34s\n- openstack-tox-docs http://logs.openstack.org/89/665089/3/check/openstack-tox-docs/b87f05c/html/ : SUCCESS in 7m 26s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/89/665089/3/check/octavia-v2-dsvm-py2-scenario-centos-7/d41289e/ : FAILURE in 1h 55m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4832bda02b78352759c6effff3f3c46da36c42cb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-13 17:51:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4c1ac72c578ca8b8e1c7a9875dbb69472aaee719","revisions":{"554ab792797e0c5a7b16f7bed7d839e2752ec2b4":{"kind":"REWORK","_number":1,"created":"2019-06-13 08:00:56.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/89/665089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/665089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/665089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/665089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/665089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/665089/1"}}},"commit":{"parents":[{"commit":"620c19804cd518af807839c877a01fe56db7462d","subject":"Merge \"Add RHEL 8 amphora support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/620c19804cd518af807839c877a01fe56db7462d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-13 07:59:50.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-13 07:59:50.000000000","tz":120},"subject":"DNM: systemd-219-64.el7","message":"DNM: systemd-219-64.el7\n\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1666612\n\nChange-Id: I349a646b54233f35392ecbd1085f75c6e2e795c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/554ab792797e0c5a7b16f7bed7d839e2752ec2b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/554ab792797e0c5a7b16f7bed7d839e2752ec2b4"}]},"branch":"refs/heads/master"},"510c5da601e5fc4631cbc0948bba31a03e11f794":{"kind":"REWORK","_number":2,"created":"2019-06-13 08:07:03.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/89/665089/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/665089/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/665089/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/665089/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/665089/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/665089/2"}}},"commit":{"parents":[{"commit":"620c19804cd518af807839c877a01fe56db7462d","subject":"Merge \"Add RHEL 8 amphora support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/620c19804cd518af807839c877a01fe56db7462d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-13 07:59:50.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-13 08:06:56.000000000","tz":120},"subject":"DNM: systemd-219-64.el7","message":"DNM: systemd-219-64.el7\n\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1666612\n\nChange-Id: I349a646b54233f35392ecbd1085f75c6e2e795c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/510c5da601e5fc4631cbc0948bba31a03e11f794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/510c5da601e5fc4631cbc0948bba31a03e11f794"}]},"branch":"refs/heads/master"},"4c1ac72c578ca8b8e1c7a9875dbb69472aaee719":{"kind":"REWORK","_number":3,"created":"2019-06-13 12:35:50.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/89/665089/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/665089/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/665089/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/665089/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/665089/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/665089/3"}}},"commit":{"parents":[{"commit":"620c19804cd518af807839c877a01fe56db7462d","subject":"Merge \"Add RHEL 8 amphora support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/620c19804cd518af807839c877a01fe56db7462d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-13 07:59:50.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-13 12:32:15.000000000","tz":120},"subject":"DNM: systemd-219-64.el7","message":"DNM: systemd-219-64.el7\n\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1666612\n\nChange-Id: I349a646b54233f35392ecbd1085f75c6e2e795c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c1ac72c578ca8b8e1c7a9875dbb69472aaee719"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c1ac72c578ca8b8e1c7a9875dbb69472aaee719"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
