)]}'
{"id":"openstack%2Foctavia~665103","triplet_id":"openstack%2Foctavia~master~I87917440565ea953ec0d762a7e219d912c56c418","project":"openstack/octavia","branch":"master","topic":"IPv6-support-and-testing","hashtags":[],"change_id":"I87917440565ea953ec0d762a7e219d912c56c418","subject":"Allow IPv6 health network in devstack","status":"MERGED","created":"2019-06-13 10:12:00.000000000","updated":"2019-10-23 09:23:18.000000000","submitted":"2019-10-23 09:20:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665103-1571822433082-ed58b23b","meta_rev_id":"f9fd92becc6e30d6cf56d6af1417437d815fe4b3","_number":665103,"virtual_id_number":665103,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-23 09:20:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-10-23 07:15:41.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-10-21 22:47:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-21 22:47:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-10-23 07:15:41.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-19 05:25:44.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-10-18 23:15:31.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-10-18 23:54:35.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-23 07:15:41.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-23 09:20:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"97fabc59a82f5cdff6e9f37fa68f565847b1872f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-13 10:12:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed75709f8a0a51974292e30093a0ca20b5fbed0f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-13 11:22:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b5b517de976a85c55be244691e67879a636ff6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 13:45:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/03/665103/2/check/openstack-tox-cover/2c22787/cover/ : SUCCESS in 8m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/665103/2/check/openstack-tox-lower-constraints/fc42f92/ : SUCCESS in 6m 58s\n- openstack-tox-pep8 http://logs.openstack.org/03/665103/2/check/openstack-tox-pep8/cfef479/ : SUCCESS in 6m 46s\n- openstack-tox-py27 http://logs.openstack.org/03/665103/2/check/openstack-tox-py27/eedba9e/ : SUCCESS in 5m 33s\n- openstack-tox-py36 http://logs.openstack.org/03/665103/2/check/openstack-tox-py36/5ea3a7c/ : SUCCESS in 6m 14s\n- openstack-tox-py37 http://logs.openstack.org/03/665103/2/check/openstack-tox-py37/832a24d/ : SUCCESS in 6m 00s\n- openstack-tox-docs http://logs.openstack.org/03/665103/2/check/openstack-tox-docs/fc455a6/html/ : SUCCESS in 6m 24s\n- openstack-tox-functional http://logs.openstack.org/03/665103/2/check/openstack-tox-functional/b07a0c6/ : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 http://logs.openstack.org/03/665103/2/check/openstack-tox-functional-py36/c39ae86/ : SUCCESS in 6m 28s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/03/665103/2/check/octavia-v2-dsvm-noop-api/209e88b/ : SUCCESS in 42m 46s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/03/665103/2/check/octavia-v2-dsvm-noop-py2-api/758d3f1/ : SUCCESS in 47m 46s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/03/665103/2/check/octavia-v2-dsvm-scenario/fc06b28/ : SUCCESS in 1h 32m 28s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/03/665103/2/check/octavia-v2-dsvm-py2-scenario/2e34a3c/ : SUCCESS in 1h 39m 40s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/03/665103/2/check/octavia-v2-dsvm-py2-scenario-centos-7/2ab0e4b/ : FAILURE in 2h 03m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/03/665103/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/2a6a5b9/ : SUCCESS in 1h 35m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/03/665103/2/check/octavia-v2-dsvm-scenario-two-node/96b2620/ : SUCCESS in 2h 01m 55s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/03/665103/2/check/octavia-v2-dsvm-py2-scenario-two-node/c478b04/ : SUCCESS in 1h 58m 56s (non-voting)\n- octavia-grenade http://logs.openstack.org/03/665103/2/check/octavia-grenade/5dd6c7e/ : SUCCESS in 1h 23m 39s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/03/665103/2/check/octavia-v2-dsvm-tls-barbican/8fce106/ : SUCCESS in 53m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/03/665103/2/check/octavia-v2-dsvm-scenario-amphora-v2/926a096/ : SUCCESS in 1h 32m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"57525c3e01c0c04ec2b247381b841a2d8e11a2a8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-18 11:02:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"14d7bc960afcbec2c36de8e6a7fe2d3be9381dc6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 14:22:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/03/665103/3/check/openstack-tox-cover/d82d3f3/cover/ : SUCCESS in 8m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/665103/3/check/openstack-tox-lower-constraints/df8d158/ : SUCCESS in 6m 46s\n- openstack-tox-pep8 http://logs.openstack.org/03/665103/3/check/openstack-tox-pep8/6a4e723/ : SUCCESS in 6m 19s\n- openstack-tox-py27 http://logs.openstack.org/03/665103/3/check/openstack-tox-py27/d86eb64/ : SUCCESS in 7m 34s\n- openstack-tox-py36 http://logs.openstack.org/03/665103/3/check/openstack-tox-py36/e3fca08/ : SUCCESS in 8m 42s\n- openstack-tox-py37 http://logs.openstack.org/03/665103/3/check/openstack-tox-py37/12f9659/ : SUCCESS in 8m 14s\n- openstack-tox-docs http://logs.openstack.org/03/665103/3/check/openstack-tox-docs/229aecb/html/ : SUCCESS in 9m 10s\n- openstack-tox-functional http://logs.openstack.org/03/665103/3/check/openstack-tox-functional/a792f0a/ : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 http://logs.openstack.org/03/665103/3/check/openstack-tox-functional-py36/0a9b989/ : SUCCESS in 7m 57s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/03/665103/3/check/octavia-v2-dsvm-noop-api/e418a67/ : SUCCESS in 44m 39s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/03/665103/3/check/octavia-v2-dsvm-noop-py2-api/31f51b7/ : RETRY_LIMIT in 3m 46s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/03/665103/3/check/octavia-v2-dsvm-scenario/7ccefc6/ : FAILURE in 1h 29m 04s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/03/665103/3/check/octavia-v2-dsvm-py2-scenario/9502270/ : FAILURE in 1h 38m 58s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/03/665103/3/check/octavia-v2-dsvm-py2-scenario-centos-7/e68712d/ : FAILURE in 1h 53m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/03/665103/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/e2ab73b/ : FAILURE in 2h 09m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/03/665103/3/check/octavia-v2-dsvm-scenario-two-node/b280940/ : FAILURE in 2h 23m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/03/665103/3/check/octavia-v2-dsvm-py2-scenario-two-node/408ab85/ : FAILURE in 1h 47m 55s (non-voting)\n- octavia-grenade http://logs.openstack.org/03/665103/3/check/octavia-grenade/551ed6c/ : FAILURE in 39m 30s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/03/665103/3/check/octavia-v2-dsvm-tls-barbican/e885be0/ : SUCCESS in 54m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/03/665103/3/check/octavia-v2-dsvm-scenario-amphora-v2/d0984ba/ : FAILURE in 1h 44m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6b159eee1e81a4f77ab8fe894c18adf9b99be0f5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-18 16:02:31.000000000","message":"Patch Set 3: Code-Review-1\n\nWhy aren\u0027t you adopting Dustin\u0027s patch we have been sitting on?\nhttps://review.opendev.org/#/c/391204\nHe did good work on it and we should leverage/give credit.","accounts_in_message":[],"_revision_number":3},{"id":"d85aebdd849cdad43bd7520fb690cca5a2837a0c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-18 16:08:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d90d78f62006088acb5ba58cba9f6c132cda03ef","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-19 07:59:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5b6b1e85382b8ada8eedae15022f743f22072929","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-19 14:33:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c42769f258ba6c56af19169aef83addb4aa2d2bd","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-20 20:52:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7c2214b10b49a6a9465af39be1a1306c0137eb45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 23:33:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/03/665103/4/check/openstack-tox-cover/2276f49/cover/ : SUCCESS in 19m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/665103/4/check/openstack-tox-lower-constraints/d7dbc67/ : SUCCESS in 19m 11s\n- openstack-tox-pep8 http://logs.openstack.org/03/665103/4/check/openstack-tox-pep8/53a0bad/ : SUCCESS in 17m 40s\n- openstack-tox-py27 http://logs.openstack.org/03/665103/4/check/openstack-tox-py27/5cf6944/ : SUCCESS in 5m 37s\n- openstack-tox-py36 http://logs.openstack.org/03/665103/4/check/openstack-tox-py36/bc33205/ : SUCCESS in 5m 52s\n- openstack-tox-py37 http://logs.openstack.org/03/665103/4/check/openstack-tox-py37/a48b51d/ : SUCCESS in 8m 20s\n- openstack-tox-docs http://logs.openstack.org/03/665103/4/check/openstack-tox-docs/34e1f88/html/ : SUCCESS in 9m 04s\n- openstack-tox-functional http://logs.openstack.org/03/665103/4/check/openstack-tox-functional/630401e/ : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 http://logs.openstack.org/03/665103/4/check/openstack-tox-functional-py36/aa9ab94/ : SUCCESS in 18m 00s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/03/665103/4/check/octavia-v2-dsvm-noop-api/8b42303/ : SUCCESS in 1h 42m 32s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/03/665103/4/check/octavia-v2-dsvm-noop-py2-api/6f41b0a/ : SUCCESS in 44m 34s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/03/665103/4/check/octavia-v2-dsvm-scenario/1296ede/ : TIMED_OUT in 2h 33m 20s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/03/665103/4/check/octavia-v2-dsvm-py2-scenario/1e0ca69/ : FAILURE in 1h 24m 19s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/03/665103/4/check/octavia-v2-dsvm-py2-scenario-centos-7/57957d5/ : FAILURE in 2h 08m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/03/665103/4/check/octavia-v2-dsvm-scenario-ubuntu-bionic/2b8beb7/ : FAILURE in 2h 15m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/03/665103/4/check/octavia-v2-dsvm-scenario-two-node/1517d54/ : FAILURE in 2h 03m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/03/665103/4/check/octavia-v2-dsvm-py2-scenario-two-node/feb8cb2/ : FAILURE in 1h 53m 38s (non-voting)\n- octavia-grenade http://logs.openstack.org/03/665103/4/check/octavia-grenade/ea96d52/ : SUCCESS in 1h 13m 48s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/03/665103/4/check/octavia-v2-dsvm-tls-barbican/c1c5f7e/ : SUCCESS in 52m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/03/665103/4/check/octavia-v2-dsvm-scenario-amphora-v2/0cf50a4/ : FAILURE in 1h 41m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f5e9c9be014c52bb898a0cdec181002452906ea7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 18:00:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ff79c6035bcdb99b70e884f490878f15886e663f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 18:07:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f009802371a5efb00de72bc6867d4f79d0f2a430","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 19:49:15.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"651aa067b8496d0c06ef7b57f88db07b359cf626","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 20:08:38.000000000","message":"Topic changed from IPv6-support-and-testing to ipv6-only-deployment-and-testing","accounts_in_message":[],"_revision_number":7},{"id":"808b8da70529472ca690b88b977bc68acf1f8ad5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 22:19:03.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b5be072c3cc564a3f740d0f67df17b7596d6be60","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 23:29:51.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ee75fcbfb3bd5b12cea27f9f2392f83d024b139a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 00:16:26.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"17c8bdcb53c87a7fcf55993c067920115ac71312","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 00:20:27.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"79b452a6fc51e6bdba929ecd0defb14071307dca","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 01:42:43.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"0c3b163a172dddadd6a04bd2b64650975f869ab7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 03:03:21.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"35d8bca2635de10cb71bf97a238cfbb0bd7c6311","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 03:52:35.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"e186051ac9504fd22bbeae3a80f6769775ec46be","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 04:18:03.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"50163b8c0ab74fea007f0c1b5156068ae3284ef8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 04:28:35.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c0248f6f7db4ee8be7102750435d8b0 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d526b4ce505e4c32916d5b52b4b353b8 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":15},{"id":"11a160a6121fb0355e571b587d96b5022f682dcb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 04:29:08.000000000","message":"Patch Set 15: Workflow-1","accounts_in_message":[],"_revision_number":15},{"id":"7bcbcd678aa3c59e56c6fcd34eedc5a64a57d7c6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 05:07:14.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"3906a4c01351609dfe467352a1ba8ce56be4a629","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 05:12:24.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ed0bab2e5684db8a0ea5f76032802f5 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ef922e7f2ed4e1d872400c824b71d87 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":16},{"id":"a1a2f9827fbdb1bce30cf29724f75828a8b806bf","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 05:52:54.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"299f7b285ab20027fbd9d945069d1f55f7388353","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 06:02:08.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02a9e1f319924b2a8e6c476a016fadf7 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7caf36f155b24f22b4f42f6c96b8eea1 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":17},{"id":"a6e5644b3a89e522d72a32b052438d5aed408813","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 17:36:37.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"2628b3430fcb4b2d640001077baae08596b4419c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 17:48:55.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2677ddf8754e48c39dfa7d39f8c5bdfa : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5ae89b07d544c939d09dd2b33af80e2 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":18},{"id":"f217e07dc08fa99c81e14d80715312a08061ade0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 18:35:04.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"c4f180ffeb60b9ce5fd7597978aca55b56435768","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 19:07:04.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75ab149f628242aa86494830808a67d3 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97bba96901dd40c89f6f1ce64b508758 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":19},{"id":"cd2f7d3b1cf2a411c8fd6ae7324d44748202a545","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 19:27:11.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"0fbe5bbd24bc40bd562d21f43d8a64e4fd6b0d2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 19:56:04.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/291c74d94776401caa884769a4f1fb6d : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b2db0807dc4a45a69085b82d382a92ba : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":20},{"id":"4eccf7993b7a3f2cd7e66f6f0897601a196579b7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 21:13:55.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"eb17e2481a34d9c0b39a1d1d450bb2df0c777357","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 21:25:18.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de4d5644576e4040af78a0a4855f7c7b : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/baa3bfd5577c4fa69fb02ce3a5501efa : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":21},{"id":"a4558e61c6f7179cd6bb4b3196a58276d398478d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 23:55:48.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"071be70c015112c7922a8217dc0feaeb2ae5db40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 00:06:38.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a5680babda64f2da366d775d81c68b6 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6577ed15dba34da0831c83f160715c0b : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":22},{"id":"afce68ac7e2ec0193e08fe2aa453864db3186fae","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-19 04:57:55.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"83affdf851ec78ef94a70e0800a193c36eb842f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 05:25:04.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be2afc6945034920aedcfcd75cf63c94 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/650d743ef1534bbbbe0348be4090405e : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":23},{"id":"260b3bb5eca1269e65eba998c625b77d4a3ee4d8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-19 13:07:43.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"fee2ebeaa281570ccf368d8270deed9791f3f623","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 16:38:17.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e69aad0462114843a2769ef8b4362d46 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/662fc658458844ab8c0a889399761f59 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":24},{"id":"97bc9cc33c85a83bacee03e4d06dad4a809b27b4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-20 04:05:13.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"9b81395a5073daef3ddd918ffefb3296486c2262","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 06:41:53.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d654aaf31af4d348ab98ebc3fad7094 : SUCCESS in 7m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8704c335bb2e40328fd04553266891ed : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3ea38e2ab144ba7948812b600fd8d28 : SUCCESS in 6m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/90d926b6ebfe4ca7a5a7ed8ac772c0e5 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4555f4756b6145b59b6bf68a4253bcee : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/950a1eb8d2d946089019687d22f41af8 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/504badf925174384ad807e0dc7e76eed : SUCCESS in 6m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/61ce724c39d74e3f8321716a158ca990 : SUCCESS in 5m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7facc177af694fe3a63779e986bb91bd : SUCCESS in 6m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/cce8f65e008a4362a58aac3ac8570cb5 : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9f6f9d26ffa84658932ce3870221d185 : SUCCESS in 6m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7f6a88fcecb1452997b838cf019dde07 : SUCCESS in 46m 10s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/45fe2b2fce194d888733031178e69cf1 : FAILURE in 27m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f5d9d2fcd88d4f4bbbdf9d990568dd3e : SUCCESS in 1h 42m 06s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/823fe6a254614c4299b652a8c1d1dfc3 : SUCCESS in 1h 44m 46s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e260b1c4dbfc4a518b129e83195ad7b1 : FAILURE in 27m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/051ed51538254718b364fbcfb2dc920d : FAILURE in 1h 58m 19s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5fb3625458cd4b8ea10ca5ec9726f7c7 : SUCCESS in 2h 01m 35s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2b3d924b019d47ad918b20035cf99ef2 : SUCCESS in 1h 29m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/412dc03bc91c458aa9f05fd0caaa6ebf : SUCCESS in 55m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8b670f020b964a9da8f4a4527790ba8a : SUCCESS in 1h 46m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc6d15bf21254a53bceb7368b3cfd15f : FAILURE in 1h 44m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9e7a6e47455e40a79650e2c5663c271d : SUCCESS in 1h 14m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fc17a94f1ab6440588ad1063fff0ae13 : FAILURE in 26m 38s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"cc5bc6a5ffb05a133466ae6fa2f389ea870dac4e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-09 22:12:04.000000000","message":"Patch Set 25:\n\nrecheck","accounts_in_message":[],"_revision_number":25},{"id":"dbf8741bc1de1891c43442a4110983c082b4247d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 00:34:31.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76cb241bff144b1eabd4446601a160ec : SUCCESS in 7m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84861e53661c4e4fb6b369db6ad55aa3 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bbe556d75594d53978d07ea8d0cf439 : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/28197c4dbce1480d831b16139794b591 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94c0fcce99314795b498d04978d919fa : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ffe578bcbc54578b4330142e9901f59 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6406ce6808ea46c19d44e71941b3890b : SUCCESS in 10m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/195e676140ef41a39994c2caae622a89 : SUCCESS in 4m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/14420106ec8c4b979cac75db1f9bfe15 : SUCCESS in 6m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ac6e6bbd912740ea9572126233b7c752 : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6875f608a2a04ae6884309f1bd2799e4 : SUCCESS in 6m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/49f82bc643384f7a9832df95e33019e1 : SUCCESS in 45m 14s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d98c4a9cac4d478ea07f0834c7ac79e6 : SUCCESS in 46m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b4f543e195034e6ebea9f058570409f7 : SUCCESS in 1h 32m 16s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f041df6431e04b229e8b390c12bde360 : SUCCESS in 1h 47m 55s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c36d1c325255452ea616edf7f40eb3ba : SUCCESS in 2h 16m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a92e2125f4f4434797273a2b0b4ee64e : SUCCESS in 1h 48m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/acaefe633f194a21b552faee52139866 : SUCCESS in 1h 55m 25s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d3be613ab82b45afbd91c44e7e7d4ba8 : SUCCESS in 1h 26m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c1315af9b70b47d6a0126de902ec2437 : SUCCESS in 53m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/778bfcbe44354bd2b7120a1fda05f729 : SUCCESS in 1h 27m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7dd6488a8fa48dfbb89f152f0df89e8 : SUCCESS in 1h 03m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7c9865e8e2c64aa19bc7dabfeac5e54a : SUCCESS in 1h 08m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/525c1be975794282b07a1376b86799d6 : SUCCESS in 1h 29m 00s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"07e0f502f1ea24747b1bc4ad74186787ce0cd3e2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-18 23:14:00.000000000","message":"Patch Set 25:\n\nIs this good now? I feel like it might be?","accounts_in_message":[],"_revision_number":25},{"id":"0647d8e771e34ef23bf69f2779dde53e057e9754","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-18 23:14:04.000000000","message":"Patch Set 26: Patch Set 25 was rebased","accounts_in_message":[],"_revision_number":26},{"id":"c48cffeb8be330c0d80f81ca0738b18a9619ea6f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-18 23:15:31.000000000","message":"Patch Set 26: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":26},{"id":"91879d91f70fb02bbe7deb72b99fd84629d7a990","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-18 23:54:35.000000000","message":"Patch Set 26: Code-Review+1","accounts_in_message":[],"_revision_number":26},{"id":"8ac1668cc52ff77df426a0deb978df67f1ed4e98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 01:36:43.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ec446eec336418094cbc033a5f431b0 : FAILURE in 4m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32373d0f052b4af9a0b9a59b65bdb307 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2edc9254d28240c785321e7999a131a4 : SUCCESS in 6m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4659b9f902cc4c9aa36ebb44d534c1aa : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aeae560318114fcb91c1f94c884e0c15 : FAILURE in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a931a58d35c475a971b4f135388d041 : FAILURE in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b16c92dccf8e42e5b57918e9a73b529d : SUCCESS in 9m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9ea63d3028eb45ae9c68b54b3e21676f : FAILURE in 5m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/68cf4dcb36e34b4cad953eecb2b4c3e8 : SUCCESS in 8m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/acc86774b7834ceaabccf472076d50af : SUCCESS in 4m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b138239e26a24105b5dcc0633330de82 : SUCCESS in 6m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/092a49d0cbcb49d88c2a7b735b8285b1 : SUCCESS in 50m 40s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0604dd053a4b4852bdce9f1960032193 : SUCCESS in 57m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78ba52aabe3f48208652967151d95ebc : SUCCESS in 1h 41m 18s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b7979c87b85b47c8b58fb1392d2be82b : SUCCESS in 1h 50m 54s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3f43162e0dee432d9919bfe6a23e3960 : SUCCESS in 2h 19m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5d6c4b41060d4905b97057d1b8ff8935 : SUCCESS in 1h 20m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a43d8b1d64224b88996e58148ed90ec0 : SUCCESS in 1h 19m 09s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/eaff76e94aa8423aa4077e98ecc08b7d : SUCCESS in 1h 16m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/685cdcc8b34140f187261803c0818290 : SUCCESS in 1h 03m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/603f5ca6573c491c8c4578bb45c5dfb5 : SUCCESS in 1h 45m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1f82ed9250147a4b8ade251ef01ea14 : FAILURE in 1h 48m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f5e70feea50343f38cef09cfa29461ce : SUCCESS in 1h 21m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8bcaf7c5d75c4f0aac32f84428607666 : SUCCESS in 1h 35m 33s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"4e5bb61273d50f9e209cf509f6745bb81563c15a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-21 22:47:28.000000000","message":"Patch Set 27: Patch Set 26 was rebased","accounts_in_message":[],"_revision_number":27},{"id":"e59fdd2c9707c78607e00bab0b8dd146e77aa391","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 00:55:55.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96bc3afde55f4fc49f0df6f010f75bf9 : SUCCESS in 7m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a71f5863c31c4161a6b012c7080b6ddf : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0666cfc0ca243a88a5181f170fcbf42 : SUCCESS in 9m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a2c83d7040964a2daab1bb4ecfdf5ebb : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f17ab60ef12439fb04aa095d5e2416f : SUCCESS in 9m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/92aad6228f7b4a7aa0bdf3723c2b7ffb : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/053524e18b8345718475f748bf69d7ce : SUCCESS in 10m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/50ef971d4aae43dd88626cb80c808249 : SUCCESS in 9m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/43f43d6a5e8c4456b6fa299a16cccf72 : SUCCESS in 5m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8a3e6c332050489bbf47085bff5fb57e : SUCCESS in 7m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5fe738a5afb64fbab2ffe7e7ade48c9e : SUCCESS in 5m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aaffd0b8489b42d785286d2d75975c11 : SUCCESS in 39m 09s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c2c01edca17149f6b8e87c2ac3d1d43f : SUCCESS in 40m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b06d16f7cf44015af1495d0c49bcb6b : SUCCESS in 1h 33m 57s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/dfe3a64448ec40159c88d145173a1e0c : SUCCESS in 1h 33m 47s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/cb59cb157b2f4d768b14743e511f9c70 : FAILURE in 2h 06m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/abd0b7fa0a69449aa86cb40eb2b9980f : FAILURE in 1h 51m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/dfa209ba8c9746fcac01b53f23346809 : SUCCESS in 1h 06m 25s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3292a3fbff06486a92d65b86b9315300 : FAILURE in 1h 38m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e7294fb3ce274c1687753de2c98e01c6 : SUCCESS in 43m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/82f1066713a64df0a2a79b4a652ee0ad : SUCCESS in 1h 26m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f86a47195db94a78828bdea40d7bff66 : SUCCESS in 56m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/44c1cd902da449659f3f82f5451ccb98 : SUCCESS in 53m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/08be10cf945044c8b32a0adc62b3dcba : POST_FAILURE in 1h 23m 59s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"1a4996257dffca760c83445d32a1d5e83630f9cd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-22 18:10:00.000000000","message":"Patch Set 27:\n\nrecheck","accounts_in_message":[],"_revision_number":27},{"id":"edcd10ab30133db0f565b852197741050cecbb77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 20:23:37.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f77a42d93794db8ba00badf440555aa : SUCCESS in 7m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf1333e0b96841e8ac35702cdd1317b8 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f06d068903b744bd8455fb2f8c4486ca : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ed7538b955084dba834dab09c93e1889 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/228adcf4db594aeeac98b766de64549c : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e83b31d139e347279dbd850de1b219ee : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae9fa5b4d3714e2aaba2853154d2fa80 : SUCCESS in 9m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/34a0d3c534f14c87a508c9131476b0dd : SUCCESS in 6m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e40a8bbfab45466c8a6c3f4c54ce69d4 : SUCCESS in 6m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/04e847fb75d44d43b450dc3bed317d3f : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cdf324eb778f43288339c381d908100a : SUCCESS in 5m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/08aaf861141143a39f7263e37b888cbb : SUCCESS in 44m 05s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b496ce819c6c43ddbe12de5a29b5afac : SUCCESS in 47m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02d6d362289c431ca4430bb9c2bbf38c : SUCCESS in 1h 31m 43s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fffe1aab94624a0bab7ec8090e2a4229 : SUCCESS in 1h 40m 14s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/897ac3b8c8e8409090794b1bc255de3b : FAILURE in 18m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/41fb51289d374f69b581c3c4c36478da : SUCCESS in 2h 07m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5a929711a1c3426b81ab4eb582f69791 : FAILURE in 1h 28m 43s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/51ac58e7162949feabaf795d98b90388 : SUCCESS in 1h 25m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8e1ccca890364d90be6732ec396fa75f : SUCCESS in 49m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/eb5063bd36f54675b64108b143641183 : SUCCESS in 1h 37m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a52a9b9b97a3478ab18558b283d34350 : SUCCESS in 1h 11m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/75d15f00d78d4055a72d542b2f0b76b1 : FAILURE in 1h 05m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d75710cc362f437ba54a12494bd7df9b : FAILURE in 17m 21s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"dbbf49a2b434223beeba5fc74bb8d149d0dd2eb3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-23 07:15:41.000000000","message":"Patch Set 27: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":27},{"id":"bfe4ec473ac1ae2df9201fb9a82b47f59e916ef7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 07:15:53.000000000","message":"Patch Set 27: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":27},{"id":"10286d6e810d65ed350387d816f78068b5471b61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 09:20:32.000000000","message":"Patch Set 27: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b422da137ec476d84fbf4b497fd8cef : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c1f984913ce418293c1e969e2a78abe : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f1e9a52a12fc4023b6f2bf99d6aba231 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c315e585006341c9b7b9328011230db5 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee35b4dd2ed44f309a902fbb0aacae6e : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30dd6275073441d290f312dcb4c4cc52 : SUCCESS in 12m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/63259fdf85df485994ea4fa0f7c8a3c4 : SUCCESS in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b9bb46766db143cab016c76ca4212f12 : SUCCESS in 5m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2efe13eff3c9415d9f3f4c5d6539ac15 : SUCCESS in 46m 50s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/70f5869a1edd4f4e88386d366afd34f0 : SUCCESS in 47m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aa95edbda53340a4b969ebd1f0d9da7e : SUCCESS in 1h 49m 22s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e29184f45f1c48869c4c1d1fb0a432d5 : SUCCESS in 1h 56m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b5256a1435584149935558820c364fdd : SUCCESS in 1h 38m 39s","accounts_in_message":[],"_revision_number":27},{"id":"8ff21ea85d3904376650a95eef26ad5ca09b2572","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 09:20:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":27},{"id":"52d311c7e973486677041fd973943e3dabbeaf73","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 09:23:18.000000000","message":"Patch Set 27:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73e5d3e6277f4e25aa9a93df74ddf8f9 : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":27}],"current_revision_number":27,"current_revision":"2c6d8fc5ecec2f774b63ef645cb417f4a071785b","revisions":{"b0322c1ed9c5c7b6b41838b0f1d1ea7996ffafea":{"kind":"REWORK","_number":1,"created":"2019-06-13 10:12:00.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/03/665103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/1"}}},"commit":{"parents":[{"commit":"620c19804cd518af807839c877a01fe56db7462d","subject":"Merge \"Add RHEL 8 amphora support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/620c19804cd518af807839c877a01fe56db7462d"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"subject":"WIP: Allow IPv6 health network in devstack","message":"WIP: Allow IPv6 health network in devstack\n\nFor testing, see: https://review.opendev.org/#/c/594078/\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0322c1ed9c5c7b6b41838b0f1d1ea7996ffafea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0322c1ed9c5c7b6b41838b0f1d1ea7996ffafea"}]},"branch":"refs/heads/master"},"798bae0601a5509624cb94794db9f143a2c8926b":{"kind":"REWORK","_number":2,"created":"2019-06-13 11:22:45.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/03/665103/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/2"}}},"commit":{"parents":[{"commit":"620c19804cd518af807839c877a01fe56db7462d","subject":"Merge \"Add RHEL 8 amphora support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/620c19804cd518af807839c877a01fe56db7462d"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 11:22:32.000000000","tz":-420},"subject":"WIP: Allow IPv6 health network in devstack","message":"WIP: Allow IPv6 health network in devstack\n\nFor testing, see: https://review.opendev.org/#/c/594078/\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/798bae0601a5509624cb94794db9f143a2c8926b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/798bae0601a5509624cb94794db9f143a2c8926b"}]},"branch":"refs/heads/master"},"9c38c812f12a117692a5ea897dc3f3df8edacaa1":{"kind":"REWORK","_number":3,"created":"2019-06-18 11:02:41.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/03/665103/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/3"}}},"commit":{"parents":[{"commit":"954025cbc32b5286509985062aff17900e04743d","subject":"Merge \"Fix a python3 issue in the amphora-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/954025cbc32b5286509985062aff17900e04743d"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-18 11:02:31.000000000","tz":-420},"subject":"WIP: Allow IPv6 health network in devstack","message":"WIP: Allow IPv6 health network in devstack\n\nFor testing, see: https://review.opendev.org/#/c/594078/\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c38c812f12a117692a5ea897dc3f3df8edacaa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c38c812f12a117692a5ea897dc3f3df8edacaa1"}]},"branch":"refs/heads/master"},"283f979f0dcc91be5afc5f556e82a52c55a0bb37":{"kind":"REWORK","_number":4,"created":"2019-06-20 20:52:44.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/03/665103/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/4"}}},"commit":{"parents":[{"commit":"a485934de3e5226a1604c649a6210efb79987617","subject":"Merge \"Clarify that the certificate guide should be used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a485934de3e5226a1604c649a6210efb79987617"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-20 20:52:31.000000000","tz":-420},"subject":"WIP: Allow IPv6 health network in devstack","message":"WIP: Allow IPv6 health network in devstack\n\nFor testing, see: https://review.opendev.org/#/c/594078/\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/283f979f0dcc91be5afc5f556e82a52c55a0bb37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/283f979f0dcc91be5afc5f556e82a52c55a0bb37"}]},"branch":"refs/heads/master"},"54c46ac123c2d01b53463fbd375433fbf7ab1d9f":{"kind":"REWORK","_number":5,"created":"2019-09-17 18:00:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/5"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-17 18:00:51.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54c46ac123c2d01b53463fbd375433fbf7ab1d9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54c46ac123c2d01b53463fbd375433fbf7ab1d9f"}]},"branch":"refs/heads/master"},"ed0bd3583d099526dd751b2e2a19baa1bf48d221":{"kind":"REWORK","_number":6,"created":"2019-09-17 18:07:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/6"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-17 18:07:20.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed0bd3583d099526dd751b2e2a19baa1bf48d221"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed0bd3583d099526dd751b2e2a19baa1bf48d221"}]},"branch":"refs/heads/master"},"6c012519873756c57c66ca03d6ecc3acbd59c286":{"kind":"REWORK","_number":7,"created":"2019-09-17 19:49:15.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/7"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-17 19:49:09.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c012519873756c57c66ca03d6ecc3acbd59c286"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c012519873756c57c66ca03d6ecc3acbd59c286"}]},"branch":"refs/heads/master"},"48cc0d4cce105de2a0ddfd8fcb316d38c5215dd6":{"kind":"REWORK","_number":8,"created":"2019-09-17 22:19:03.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/8"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-17 22:18:57.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48cc0d4cce105de2a0ddfd8fcb316d38c5215dd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48cc0d4cce105de2a0ddfd8fcb316d38c5215dd6"}]},"branch":"refs/heads/master"},"69df93479578527c8459d833ead3d27581444385":{"kind":"REWORK","_number":9,"created":"2019-09-17 23:29:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/9"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-17 23:29:45.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69df93479578527c8459d833ead3d27581444385"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69df93479578527c8459d833ead3d27581444385"}]},"branch":"refs/heads/master"},"acad5929109255cfc054f6f69d989edcc29331ba":{"kind":"REWORK","_number":10,"created":"2019-09-18 00:16:26.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/10"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-18 00:16:19.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/acad5929109255cfc054f6f69d989edcc29331ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/acad5929109255cfc054f6f69d989edcc29331ba"}]},"branch":"refs/heads/master"},"aa2914fb9a08cbf343a0d23111f95703e53a3dfd":{"kind":"REWORK","_number":11,"created":"2019-09-18 00:20:27.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/11"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-18 00:20:22.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa2914fb9a08cbf343a0d23111f95703e53a3dfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa2914fb9a08cbf343a0d23111f95703e53a3dfd"}]},"branch":"refs/heads/master"},"29904934ee01e347998185114fdf1e28ce1ba42f":{"kind":"REWORK","_number":12,"created":"2019-09-18 01:42:43.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/12"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-18 01:42:38.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29904934ee01e347998185114fdf1e28ce1ba42f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29904934ee01e347998185114fdf1e28ce1ba42f"}]},"branch":"refs/heads/master"},"55602930af3df8eaba078c1dba305456ddc1870d":{"kind":"REWORK","_number":13,"created":"2019-09-18 03:03:21.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/13"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-18 03:03:13.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55602930af3df8eaba078c1dba305456ddc1870d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55602930af3df8eaba078c1dba305456ddc1870d"}]},"branch":"refs/heads/master"},"1667112f74791bd4b01bc0b942219c9ad0cfc77f":{"kind":"REWORK","_number":14,"created":"2019-09-18 03:52:35.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/14"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-18 03:52:29.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1667112f74791bd4b01bc0b942219c9ad0cfc77f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1667112f74791bd4b01bc0b942219c9ad0cfc77f"}]},"branch":"refs/heads/master"},"c62483a24952be2ba5f936767c9c97739048af24":{"kind":"REWORK","_number":15,"created":"2019-09-18 04:18:03.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/15"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-18 04:17:55.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c62483a24952be2ba5f936767c9c97739048af24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c62483a24952be2ba5f936767c9c97739048af24"}]},"branch":"refs/heads/master"},"9cd510829cd1cc075f93969e20eb5b32a7fcc933":{"kind":"REWORK","_number":16,"created":"2019-09-18 05:07:14.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/16"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-18 05:07:07.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9cd510829cd1cc075f93969e20eb5b32a7fcc933"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9cd510829cd1cc075f93969e20eb5b32a7fcc933"}]},"branch":"refs/heads/master"},"6bcaf8ac7029917dd773d7a9149459c739c413ab":{"kind":"REWORK","_number":17,"created":"2019-09-18 05:52:54.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/17"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-18 05:52:47.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6bcaf8ac7029917dd773d7a9149459c739c413ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6bcaf8ac7029917dd773d7a9149459c739c413ab"}]},"branch":"refs/heads/master"},"09dbb095325925a18746041f1ccc98e7f638c4cd":{"kind":"REWORK","_number":18,"created":"2019-09-18 17:36:37.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/18"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-18 17:36:31.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09dbb095325925a18746041f1ccc98e7f638c4cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09dbb095325925a18746041f1ccc98e7f638c4cd"}]},"branch":"refs/heads/master"},"99d490427cc5b1880a6cc1a75d3656bc1cce044c":{"kind":"REWORK","_number":19,"created":"2019-09-18 18:35:04.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/19"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-18 18:34:58.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/99d490427cc5b1880a6cc1a75d3656bc1cce044c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/99d490427cc5b1880a6cc1a75d3656bc1cce044c"}]},"branch":"refs/heads/master"},"eb56b75a9882d49fb7e95351f308b4b641c7f12e":{"kind":"REWORK","_number":20,"created":"2019-09-18 19:27:11.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/20"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-18 19:27:06.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eb56b75a9882d49fb7e95351f308b4b641c7f12e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eb56b75a9882d49fb7e95351f308b4b641c7f12e"}]},"branch":"refs/heads/master"},"82b1734cde61e12288fcb85ec90d4651fbb5cfff":{"kind":"REWORK","_number":21,"created":"2019-09-18 21:13:55.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/21"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-18 21:13:50.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/82b1734cde61e12288fcb85ec90d4651fbb5cfff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/82b1734cde61e12288fcb85ec90d4651fbb5cfff"}]},"branch":"refs/heads/master"},"033ca4f5d19f4ed3b7034408537f26fd2b37b2b2":{"kind":"REWORK","_number":22,"created":"2019-09-18 23:55:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/22"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-18 23:55:42.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/033ca4f5d19f4ed3b7034408537f26fd2b37b2b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/033ca4f5d19f4ed3b7034408537f26fd2b37b2b2"}]},"branch":"refs/heads/master"},"3c91b56cca78dda0a713c6667c712917e390e118":{"kind":"REWORK","_number":23,"created":"2019-09-19 04:57:55.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/23"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-19 04:57:42.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c91b56cca78dda0a713c6667c712917e390e118"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c91b56cca78dda0a713c6667c712917e390e118"}]},"branch":"refs/heads/master"},"5d1b6197323207de15fe76bf2d5454d5188b4a68":{"kind":"REWORK","_number":24,"created":"2019-09-19 13:07:43.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/24"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-19 13:07:37.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d1b6197323207de15fe76bf2d5454d5188b4a68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d1b6197323207de15fe76bf2d5454d5188b4a68"}]},"branch":"refs/heads/master"},"6ec2eb0e15d3b72f7499d2c5455da051df4bb319":{"kind":"REWORK","_number":25,"created":"2019-09-20 04:05:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/665103/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/25"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-20 04:05:03.000000000","tz":-420},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ec2eb0e15d3b72f7499d2c5455da051df4bb319"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ec2eb0e15d3b72f7499d2c5455da051df4bb319"}]},"branch":"refs/heads/master"},"3faae262c14951b39b574c636a4e73d70c481ba3":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2019-10-18 23:14:04.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/03/665103/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/26"}}},"commit":{"parents":[{"commit":"076c3adc36380113f16561280bcec6aed50a628a","subject":"Add backend re-encryption to the LB cookbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/076c3adc36380113f16561280bcec6aed50a628a"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-18 23:14:04.000000000","tz":0},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3faae262c14951b39b574c636a4e73d70c481ba3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3faae262c14951b39b574c636a4e73d70c481ba3"}]},"branch":"refs/heads/master"},"2c6d8fc5ecec2f774b63ef645cb417f4a071785b":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2019-10-21 22:47:28.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/03/665103/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/665103/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/665103/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/665103/27"}}},"commit":{"parents":[{"commit":"ae9895adccab4886187ddb31cbb9ff084db589a9","subject":"Merge \"Fix log offload file permissions in CentOS devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ae9895adccab4886187ddb31cbb9ff084db589a9"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-13 10:10:47.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-21 22:47:28.000000000","tz":0},"subject":"Allow IPv6 health network in devstack","message":"Allow IPv6 health network in devstack\n\nThis allows the Octavia devstack plugin to use IPv6 for the service\nendpoint as well as for the lb-mgmt-net.\n\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n\nChange-Id: I87917440565ea953ec0d762a7e219d912c56c418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c6d8fc5ecec2f774b63ef645cb417f4a071785b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c6d8fc5ecec2f774b63ef645cb417f4a071785b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
