)]}'
{"id":"openstack%2Foctavia~665402","triplet_id":"openstack%2Foctavia~master~I1384c6f52eec99e6573a8e83fe5a80a632804083","project":"openstack/octavia","branch":"master","topic":"multi-vip","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2022-08-31 17:44:04.000000000","reason":"Change was submitted"},"10273":{"account":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"last_update":"2022-08-31 17:44:04.000000000","reason":"Change was submitted"},"27256":{"account":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"last_update":"2022-08-31 17:44:04.000000000","reason":"Change was submitted"},"11290":{"account":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"last_update":"2022-03-21 09:20:25.000000000","reason":"Gregory Thiemonge replied on the change"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-08-31 17:44:04.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-08-31 15:00:00.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-08-29 15:36:34.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I1384c6f52eec99e6573a8e83fe5a80a632804083","subject":"Fix plugging member subnets on existing networks","status":"MERGED","created":"2019-06-14 14:13:52.000000000","updated":"2022-08-31 17:45:28.000000000","submitted":"2022-08-31 17:44:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":164,"unresolved_comment_count":13,"has_review_started":true,"submission_id":"665402-multi-vip","meta_rev_id":"66a4203d1d13f76e75609fdcac5abf938d102156","_number":665402,"virtual_id_number":665402,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-31 17:44:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-31 15:00:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-08-29 15:36:34.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-31 15:00:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-15 01:07:41.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-08-14 08:16:08.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-22 01:04:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-03 15:41:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-09-23 08:25:27.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-04-12 09:52:57.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-07-14 09:29:25.000000000","updated_by":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"reviewer":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"state":"CC"},{"updated":"2022-08-09 17:19:00.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"3ee4cf910e572d64caee5f30368934b26ecd7aff","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-14 14:13:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c73153ae90cf9c4ac923a19683070af934169e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 19:06:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/02/665402/1/check/openstack-tox-cover/57b7cab/ : FAILURE in 8m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/665402/1/check/openstack-tox-lower-constraints/0cd3c4f/ : FAILURE in 7m 34s\n- openstack-tox-pep8 http://logs.openstack.org/02/665402/1/check/openstack-tox-pep8/16928d2/ : FAILURE in 6m 01s\n- openstack-tox-py27 http://logs.openstack.org/02/665402/1/check/openstack-tox-py27/3b79db5/ : SUCCESS in 5m 39s\n- openstack-tox-py36 http://logs.openstack.org/02/665402/1/check/openstack-tox-py36/3b5678b/ : SUCCESS in 5m 54s\n- openstack-tox-py37 http://logs.openstack.org/02/665402/1/check/openstack-tox-py37/66f9208/ : SUCCESS in 7m 19s\n- openstack-tox-docs http://logs.openstack.org/02/665402/1/check/openstack-tox-docs/6401aac/html/ : SUCCESS in 6m 37s\n- openstack-tox-functional http://logs.openstack.org/02/665402/1/check/openstack-tox-functional/9def400/ : FAILURE in 6m 37s\n- openstack-tox-functional-py36 http://logs.openstack.org/02/665402/1/check/openstack-tox-functional-py36/573709b/ : FAILURE in 6m 50s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/02/665402/1/check/octavia-v2-dsvm-noop-api/1252e3d/ : FAILURE in 48m 16s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/02/665402/1/check/octavia-v2-dsvm-noop-py2-api/73e012a/ : FAILURE in 43m 06s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/02/665402/1/check/octavia-v2-dsvm-scenario/b8ddebf/ : SUCCESS in 1h 18m 09s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/02/665402/1/check/octavia-v2-dsvm-py2-scenario/1ae6b6b/ : SUCCESS in 1h 45m 38s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/02/665402/1/check/octavia-v2-dsvm-py2-scenario-centos-7/eed47f4/ : FAILURE in 2h 20m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/02/665402/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/7f8be57/ : SUCCESS in 1h 56m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/02/665402/1/check/octavia-v2-dsvm-scenario-two-node/42ff574/ : SUCCESS in 1h 56m 04s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/02/665402/1/check/octavia-v2-dsvm-py2-scenario-two-node/05dd0db/ : SUCCESS in 1h 52m 07s (non-voting)\n- octavia-grenade http://logs.openstack.org/02/665402/1/check/octavia-grenade/b839c64/ : SUCCESS in 1h 28m 22s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/02/665402/1/check/octavia-v2-dsvm-tls-barbican/4a1dc07/ : SUCCESS in 53m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/02/665402/1/check/octavia-v2-dsvm-scenario-amphora-v2/f362f6d/ : SUCCESS in 1h 31m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e237c764240eed606a0fe84e09b22d6e5d7e94c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-15 01:07:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)\n\nSome comments","accounts_in_message":[],"_revision_number":1},{"id":"0d31058fdff81734b280e363871932005fe01533","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-15 03:22:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"65480970f3d320baa4425378dbac8572483b4960","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-15 05:26:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c7cf46f35b959d6eec4188c24d0eaa31f7f7d988","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 07:30:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/02/665402/3/check/openstack-tox-cover/bb5d051/cover/ : SUCCESS in 8m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/665402/3/check/openstack-tox-lower-constraints/e5d2649/ : SUCCESS in 6m 56s\n- openstack-tox-pep8 http://logs.openstack.org/02/665402/3/check/openstack-tox-pep8/fa93094/ : SUCCESS in 6m 34s\n- openstack-tox-py27 http://logs.openstack.org/02/665402/3/check/openstack-tox-py27/bb0acd4/ : SUCCESS in 5m 25s\n- openstack-tox-py36 http://logs.openstack.org/02/665402/3/check/openstack-tox-py36/20c891e/ : SUCCESS in 5m 18s\n- openstack-tox-py37 http://logs.openstack.org/02/665402/3/check/openstack-tox-py37/7f99d36/ : SUCCESS in 6m 05s\n- openstack-tox-docs http://logs.openstack.org/02/665402/3/check/openstack-tox-docs/4b9dba1/html/ : SUCCESS in 6m 29s\n- openstack-tox-functional http://logs.openstack.org/02/665402/3/check/openstack-tox-functional/e9920f6/ : SUCCESS in 6m 08s\n- openstack-tox-functional-py36 http://logs.openstack.org/02/665402/3/check/openstack-tox-functional-py36/8324da6/ : SUCCESS in 8m 08s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/02/665402/3/check/octavia-v2-dsvm-noop-api/9f20fdd/ : FAILURE in 45m 51s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/02/665402/3/check/octavia-v2-dsvm-noop-py2-api/499d3e3/ : FAILURE in 41m 41s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/02/665402/3/check/octavia-v2-dsvm-scenario/53de1fa/ : SUCCESS in 1h 35m 36s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/02/665402/3/check/octavia-v2-dsvm-py2-scenario/55c649e/ : SUCCESS in 1h 31m 03s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/02/665402/3/check/octavia-v2-dsvm-py2-scenario-centos-7/2a5f6b5/ : FAILURE in 1h 58m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/02/665402/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/e1e23cf/ : SUCCESS in 1h 43m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/02/665402/3/check/octavia-v2-dsvm-scenario-two-node/5ec28f1/ : SUCCESS in 1h 46m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/02/665402/3/check/octavia-v2-dsvm-py2-scenario-two-node/54afed6/ : FAILURE in 1h 43m 15s (non-voting)\n- octavia-grenade http://logs.openstack.org/02/665402/3/check/octavia-grenade/f14023f/ : SUCCESS in 1h 23m 29s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/02/665402/3/check/octavia-v2-dsvm-tls-barbican/345f070/ : SUCCESS in 52m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/02/665402/3/check/octavia-v2-dsvm-scenario-amphora-v2/49edc6e/ : SUCCESS in 1h 23m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2e8d136df7b0e0f72b1619411d1805a454b7b120","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-16 00:40:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"196aebce6bbc723ed2fc6b85649c55b6c9a5590d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-16 02:42:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/02/665402/4/check/openstack-tox-cover/2254a0e/cover/ : SUCCESS in 8m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/665402/4/check/openstack-tox-lower-constraints/d3920a5/ : SUCCESS in 6m 36s\n- openstack-tox-pep8 http://logs.openstack.org/02/665402/4/check/openstack-tox-pep8/3da7ac8/ : SUCCESS in 6m 09s\n- openstack-tox-py27 http://logs.openstack.org/02/665402/4/check/openstack-tox-py27/8c94a5d/ : SUCCESS in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/02/665402/4/check/openstack-tox-py36/b5aaf28/ : SUCCESS in 4m 58s\n- openstack-tox-py37 http://logs.openstack.org/02/665402/4/check/openstack-tox-py37/3b1e7e0/ : SUCCESS in 5m 39s\n- openstack-tox-docs http://logs.openstack.org/02/665402/4/check/openstack-tox-docs/23e62c2/html/ : SUCCESS in 5m 55s\n- openstack-tox-functional http://logs.openstack.org/02/665402/4/check/openstack-tox-functional/ba47441/ : SUCCESS in 6m 02s\n- openstack-tox-functional-py36 http://logs.openstack.org/02/665402/4/check/openstack-tox-functional-py36/76af36d/ : SUCCESS in 6m 14s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/02/665402/4/check/octavia-v2-dsvm-noop-api/dd92fc2/ : FAILURE in 42m 59s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/02/665402/4/check/octavia-v2-dsvm-noop-py2-api/e608ebf/ : FAILURE in 45m 47s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/02/665402/4/check/octavia-v2-dsvm-scenario/8b2da04/ : FAILURE in 1h 55m 58s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/02/665402/4/check/octavia-v2-dsvm-py2-scenario/d0ea443/ : SUCCESS in 1h 34m 43s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/02/665402/4/check/octavia-v2-dsvm-py2-scenario-centos-7/ad10644/ : FAILURE in 1h 55m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/02/665402/4/check/octavia-v2-dsvm-scenario-ubuntu-bionic/ce77059/ : SUCCESS in 1h 39m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/02/665402/4/check/octavia-v2-dsvm-scenario-two-node/e86f261/ : SUCCESS in 1h 53m 16s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/02/665402/4/check/octavia-v2-dsvm-py2-scenario-two-node/f7335a7/ : FAILURE in 1h 44m 19s (non-voting)\n- octavia-grenade http://logs.openstack.org/02/665402/4/check/octavia-grenade/0602f89/ : SUCCESS in 1h 12m 31s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/02/665402/4/check/octavia-v2-dsvm-tls-barbican/dee52a7/ : SUCCESS in 52m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/02/665402/4/check/octavia-v2-dsvm-scenario-amphora-v2/ecc7d18/ : SUCCESS in 1h 35m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5b578b36040c826e27ef6415f41ac792456cba82","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-16 06:35:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"16084de001b0eaad7883b296e5043de8b80c2107","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-16 06:37:51.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e2f079e7d3bf333eb7bff3fb3da12cbfc65ab5c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-16 08:51:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/02/665402/5/check/openstack-tox-cover/24b2a9d/cover/ : SUCCESS in 7m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/665402/5/check/openstack-tox-lower-constraints/7da658e/ : SUCCESS in 6m 26s\n- openstack-tox-pep8 http://logs.openstack.org/02/665402/5/check/openstack-tox-pep8/2c5ebba/ : SUCCESS in 6m 31s\n- openstack-tox-py27 http://logs.openstack.org/02/665402/5/check/openstack-tox-py27/49a6d56/ : SUCCESS in 5m 12s\n- openstack-tox-py36 http://logs.openstack.org/02/665402/5/check/openstack-tox-py36/1eb495c/ : SUCCESS in 5m 34s\n- openstack-tox-py37 http://logs.openstack.org/02/665402/5/check/openstack-tox-py37/b83bb71/ : SUCCESS in 5m 43s\n- openstack-tox-docs http://logs.openstack.org/02/665402/5/check/openstack-tox-docs/3581df5/html/ : SUCCESS in 6m 08s\n- openstack-tox-functional http://logs.openstack.org/02/665402/5/check/openstack-tox-functional/8e45052/ : SUCCESS in 6m 07s\n- openstack-tox-functional-py36 http://logs.openstack.org/02/665402/5/check/openstack-tox-functional-py36/a5afbee/ : SUCCESS in 6m 34s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/02/665402/5/check/octavia-v2-dsvm-noop-api/668d0c2/ : FAILURE in 43m 55s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/02/665402/5/check/octavia-v2-dsvm-noop-py2-api/1fb37ec/ : FAILURE in 42m 41s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/02/665402/5/check/octavia-v2-dsvm-scenario/04aac3b/ : SUCCESS in 1h 34m 28s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/02/665402/5/check/octavia-v2-dsvm-py2-scenario/cf032bf/ : FAILURE in 1h 42m 59s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/02/665402/5/check/octavia-v2-dsvm-py2-scenario-centos-7/fb492db/ : FAILURE in 2h 04m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/02/665402/5/check/octavia-v2-dsvm-scenario-ubuntu-bionic/6425450/ : SUCCESS in 1h 29m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/02/665402/5/check/octavia-v2-dsvm-scenario-two-node/c9e8ab3/ : FAILURE in 1h 29m 00s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/02/665402/5/check/octavia-v2-dsvm-py2-scenario-two-node/c55efca/ : SUCCESS in 1h 49m 12s (non-voting)\n- octavia-grenade http://logs.openstack.org/02/665402/5/check/octavia-grenade/c7e1de5/ : SUCCESS in 1h 19m 00s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/02/665402/5/check/octavia-v2-dsvm-tls-barbican/a6dada1/ : SUCCESS in 53m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/02/665402/5/check/octavia-v2-dsvm-scenario-amphora-v2/99bbd99/ : SUCCESS in 1h 33m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5e52b5c61f4367adc050374af4adde94825c9fbc","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-16 10:55:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"14bee4150020fe4a68941178864be873aec5938d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-16 13:08:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/02/665402/6/check/openstack-tox-cover/fa73b51/cover/ : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/665402/6/check/openstack-tox-lower-constraints/3cb7bf3/ : SUCCESS in 8m 34s\n- openstack-tox-pep8 http://logs.openstack.org/02/665402/6/check/openstack-tox-pep8/d01d5e9/ : SUCCESS in 6m 57s\n- openstack-tox-py27 http://logs.openstack.org/02/665402/6/check/openstack-tox-py27/7ed7d45/ : SUCCESS in 6m 28s\n- openstack-tox-py36 http://logs.openstack.org/02/665402/6/check/openstack-tox-py36/14f0ca8/ : SUCCESS in 6m 57s\n- openstack-tox-py37 http://logs.openstack.org/02/665402/6/check/openstack-tox-py37/ba8b57e/ : SUCCESS in 7m 20s\n- openstack-tox-docs http://logs.openstack.org/02/665402/6/check/openstack-tox-docs/a654ccf/html/ : SUCCESS in 6m 04s\n- openstack-tox-functional http://logs.openstack.org/02/665402/6/check/openstack-tox-functional/ecc6ca1/ : SUCCESS in 8m 09s\n- openstack-tox-functional-py36 http://logs.openstack.org/02/665402/6/check/openstack-tox-functional-py36/fef1ab4/ : SUCCESS in 7m 46s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/02/665402/6/check/octavia-v2-dsvm-noop-api/fdd269b/ : FAILURE in 45m 35s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/02/665402/6/check/octavia-v2-dsvm-noop-py2-api/b298154/ : FAILURE in 50m 09s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/02/665402/6/check/octavia-v2-dsvm-scenario/66a1cc7/ : SUCCESS in 1h 29m 19s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/02/665402/6/check/octavia-v2-dsvm-py2-scenario/2773a86/ : FAILURE in 1h 36m 37s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/02/665402/6/check/octavia-v2-dsvm-py2-scenario-centos-7/98e2897/ : FAILURE in 2h 10m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/02/665402/6/check/octavia-v2-dsvm-scenario-ubuntu-bionic/f71a63b/ : SUCCESS in 1h 59m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/02/665402/6/check/octavia-v2-dsvm-scenario-two-node/c6fe565/ : SUCCESS in 1h 28m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/02/665402/6/check/octavia-v2-dsvm-py2-scenario-two-node/425f1f5/ : FAILURE in 31m 42s (non-voting)\n- octavia-grenade http://logs.openstack.org/02/665402/6/check/octavia-grenade/2548846/ : SUCCESS in 1h 22m 17s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/02/665402/6/check/octavia-v2-dsvm-tls-barbican/682403e/ : SUCCESS in 1h 01m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/02/665402/6/check/octavia-v2-dsvm-scenario-amphora-v2/ff1235f/ : SUCCESS in 1h 37m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3e6b4002b6db55ac0c21410dd4fb5a97d27b7381","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-18 04:10:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"48e44da68e1dda4f61de07328173abf6d2056bfc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 06:40:18.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/02/665402/7/check/openstack-tox-cover/b4b643c/cover/ : SUCCESS in 7m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/665402/7/check/openstack-tox-lower-constraints/55133b3/ : SUCCESS in 8m 13s\n- openstack-tox-pep8 http://logs.openstack.org/02/665402/7/check/openstack-tox-pep8/7a130f8/ : SUCCESS in 6m 32s\n- openstack-tox-py27 http://logs.openstack.org/02/665402/7/check/openstack-tox-py27/ada67f8/ : SUCCESS in 5m 08s\n- openstack-tox-py36 http://logs.openstack.org/02/665402/7/check/openstack-tox-py36/9d0227c/ : SUCCESS in 7m 30s\n- openstack-tox-py37 http://logs.openstack.org/02/665402/7/check/openstack-tox-py37/55ed9ae/ : SUCCESS in 5m 54s\n- openstack-tox-docs http://logs.openstack.org/02/665402/7/check/openstack-tox-docs/7dc6035/html/ : SUCCESS in 6m 41s\n- openstack-tox-functional http://logs.openstack.org/02/665402/7/check/openstack-tox-functional/141ee9f/ : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 http://logs.openstack.org/02/665402/7/check/openstack-tox-functional-py36/358909b/ : SUCCESS in 8m 41s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/02/665402/7/check/octavia-v2-dsvm-noop-api/c9a3c27/ : FAILURE in 1h 26m 34s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/02/665402/7/check/octavia-v2-dsvm-noop-py2-api/7d6a60d/ : FAILURE in 1h 25m 45s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/02/665402/7/check/octavia-v2-dsvm-scenario/d7e7e79/ : SUCCESS in 1h 31m 20s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/02/665402/7/check/octavia-v2-dsvm-py2-scenario/a488954/ : FAILURE in 1h 39m 25s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/02/665402/7/check/octavia-v2-dsvm-py2-scenario-centos-7/20e1e47/ : FAILURE in 1h 56m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/02/665402/7/check/octavia-v2-dsvm-scenario-ubuntu-bionic/fe36e89/ : SUCCESS in 1h 53m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/02/665402/7/check/octavia-v2-dsvm-scenario-two-node/7212c81/ : SUCCESS in 1h 15m 22s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/02/665402/7/check/octavia-v2-dsvm-py2-scenario-two-node/b9e5e78/ : SUCCESS in 1h 19m 49s (non-voting)\n- octavia-grenade http://logs.openstack.org/02/665402/7/check/octavia-grenade/f68631e/ : SUCCESS in 1h 22m 16s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/02/665402/7/check/octavia-v2-dsvm-tls-barbican/bba2cb6/ : SUCCESS in 56m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/02/665402/7/check/octavia-v2-dsvm-scenario-amphora-v2/2f85f18/ : SUCCESS in 1h 46m 55s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f9f6e8487db2bc2c14518cff118b64606534f6dd","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-18 07:07:03.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e36ce07f0c64593e21f52bb33cedfde0f28e1178","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:18:44.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/02/665402/8/check/openstack-tox-cover/46428f3/ : FAILURE in 6m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/665402/8/check/openstack-tox-lower-constraints/bb2ee9e/ : FAILURE in 5m 22s\n- openstack-tox-pep8 http://logs.openstack.org/02/665402/8/check/openstack-tox-pep8/4c472a4/ : SUCCESS in 6m 38s\n- openstack-tox-py27 http://logs.openstack.org/02/665402/8/check/openstack-tox-py27/e1aa374/ : FAILURE in 7m 03s\n- openstack-tox-py36 http://logs.openstack.org/02/665402/8/check/openstack-tox-py36/17df2a5/ : FAILURE in 5m 46s\n- openstack-tox-py37 http://logs.openstack.org/02/665402/8/check/openstack-tox-py37/b4b27a4/ : FAILURE in 5m 47s\n- openstack-tox-docs http://logs.openstack.org/02/665402/8/check/openstack-tox-docs/0f6e4a1/html/ : SUCCESS in 7m 20s\n- openstack-tox-functional http://logs.openstack.org/02/665402/8/check/openstack-tox-functional/f3a11cb/ : SUCCESS in 7m 27s\n- openstack-tox-functional-py36 http://logs.openstack.org/02/665402/8/check/openstack-tox-functional-py36/64e2b01/ : SUCCESS in 6m 45s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/02/665402/8/check/octavia-v2-dsvm-noop-api/0413c22/ : SUCCESS in 49m 07s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/02/665402/8/check/octavia-v2-dsvm-noop-py2-api/fdc74e5/ : SUCCESS in 44m 40s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/02/665402/8/check/octavia-v2-dsvm-scenario/e1896d5/ : SUCCESS in 1h 26m 51s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/02/665402/8/check/octavia-v2-dsvm-py2-scenario/d848f49/ : SUCCESS in 1h 35m 21s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/02/665402/8/check/octavia-v2-dsvm-py2-scenario-centos-7/a163b61/ : FAILURE in 2h 05m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/02/665402/8/check/octavia-v2-dsvm-scenario-ubuntu-bionic/a318caa/ : SUCCESS in 1h 33m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/02/665402/8/check/octavia-v2-dsvm-scenario-two-node/0cf86d3/ : SUCCESS in 1h 29m 16s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/02/665402/8/check/octavia-v2-dsvm-py2-scenario-two-node/71fb3e0/ : SUCCESS in 1h 52m 03s (non-voting)\n- octavia-grenade http://logs.openstack.org/02/665402/8/check/octavia-grenade/6a14dd7/ : SUCCESS in 1h 15m 11s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/02/665402/8/check/octavia-v2-dsvm-tls-barbican/da06e62/ : SUCCESS in 45m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/02/665402/8/check/octavia-v2-dsvm-scenario-amphora-v2/c658ebf/ : SUCCESS in 1h 28m 35s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f658a9167ed8854e1bece3cd621ec091271ccd9f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-19 02:18:16.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"faaad95b70ed08ed734eb7fc68cab205a6319458","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 04:56:31.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/02/665402/9/check/openstack-tox-cover/8198ad8/cover/ : SUCCESS in 9m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/665402/9/check/openstack-tox-lower-constraints/ea46758/ : SUCCESS in 8m 54s\n- openstack-tox-pep8 http://logs.openstack.org/02/665402/9/check/openstack-tox-pep8/d526f80/ : SUCCESS in 6m 25s\n- openstack-tox-py27 http://logs.openstack.org/02/665402/9/check/openstack-tox-py27/9307085/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/02/665402/9/check/openstack-tox-py36/0f052e5/ : SUCCESS in 6m 08s\n- openstack-tox-py37 http://logs.openstack.org/02/665402/9/check/openstack-tox-py37/ec5570b/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/02/665402/9/check/openstack-tox-docs/62a7399/html/ : SUCCESS in 6m 45s\n- openstack-tox-functional http://logs.openstack.org/02/665402/9/check/openstack-tox-functional/1a3f162/ : SUCCESS in 6m 04s\n- openstack-tox-functional-py36 http://logs.openstack.org/02/665402/9/check/openstack-tox-functional-py36/be435be/ : SUCCESS in 8m 27s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/02/665402/9/check/octavia-v2-dsvm-noop-api/2d445b3/ : SUCCESS in 42m 33s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/02/665402/9/check/octavia-v2-dsvm-noop-py2-api/1fd66ef/ : SUCCESS in 43m 18s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/02/665402/9/check/octavia-v2-dsvm-scenario/e3359d7/ : SUCCESS in 1h 21m 36s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/02/665402/9/check/octavia-v2-dsvm-py2-scenario/f8bf28e/ : SUCCESS in 1h 39m 39s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/02/665402/9/check/octavia-v2-dsvm-py2-scenario-centos-7/e54c705/ : FAILURE in 1h 51m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/02/665402/9/check/octavia-v2-dsvm-scenario-ubuntu-bionic/99ba78f/ : SUCCESS in 1h 44m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/02/665402/9/check/octavia-v2-dsvm-scenario-two-node/823ed7e/ : SUCCESS in 1h 57m 02s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/02/665402/9/check/octavia-v2-dsvm-py2-scenario-two-node/25e8a1b/ : FAILURE in 2h 34m 56s (non-voting)\n- octavia-grenade http://logs.openstack.org/02/665402/9/check/octavia-grenade/9fb0108/ : SUCCESS in 1h 23m 39s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/02/665402/9/check/octavia-v2-dsvm-tls-barbican/f94eeee/ : SUCCESS in 53m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/02/665402/9/check/octavia-v2-dsvm-scenario-amphora-v2/6ca3263/ : SUCCESS in 1h 34m 04s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8cb9b0243b0a7c83979e675124f278dc25cb7989","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-05 16:16:53.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"bf5f19078421036b06fc32b2737c2fee9111ed68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 18:50:43.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/02/665402/10/check/openstack-tox-cover/cf3c72a/cover/ : SUCCESS in 7m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/665402/10/check/openstack-tox-lower-constraints/ce2fe72/ : SUCCESS in 6m 21s\n- openstack-tox-pep8 http://logs.openstack.org/02/665402/10/check/openstack-tox-pep8/8fb4e1b/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/02/665402/10/check/openstack-tox-py27/29c5923/ : SUCCESS in 3m 54s\n- openstack-tox-py36 http://logs.openstack.org/02/665402/10/check/openstack-tox-py36/a2db3eb/ : SUCCESS in 4m 26s\n- openstack-tox-py37 http://logs.openstack.org/02/665402/10/check/openstack-tox-py37/3d22a4b/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/02/665402/10/check/openstack-tox-docs/ce1f8fd/html/ : SUCCESS in 5m 26s\n- openstack-tox-functional http://logs.openstack.org/02/665402/10/check/openstack-tox-functional/550c6d4/ : SUCCESS in 5m 32s\n- openstack-tox-functional-py36 http://logs.openstack.org/02/665402/10/check/openstack-tox-functional-py36/0f313f8/ : SUCCESS in 6m 03s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/02/665402/10/check/octavia-v2-dsvm-noop-api/9bb6e6c/ : SUCCESS in 43m 37s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/02/665402/10/check/octavia-v2-dsvm-noop-py2-api/54bfebb/ : SUCCESS in 44m 56s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/02/665402/10/check/octavia-v2-dsvm-scenario/0084433/ : FAILURE in 1h 51m 39s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/02/665402/10/check/octavia-v2-dsvm-py2-scenario/370d204/ : SUCCESS in 1h 51m 20s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/02/665402/10/check/octavia-v2-dsvm-py2-scenario-centos-7/6ce6b23/ : SUCCESS in 2h 16m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/02/665402/10/check/octavia-v2-dsvm-scenario-ubuntu-bionic/48dd852/ : SUCCESS in 2h 07m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/02/665402/10/check/octavia-v2-dsvm-scenario-two-node/e5c9d50/ : SUCCESS in 2h 05m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/02/665402/10/check/octavia-v2-dsvm-py2-scenario-two-node/b32d094/ : SUCCESS in 1h 24m 27s (non-voting)\n- octavia-grenade http://logs.openstack.org/02/665402/10/check/octavia-grenade/100560d/ : SUCCESS in 1h 16m 38s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/02/665402/10/check/octavia-v2-dsvm-tls-barbican/d294ad7/ : SUCCESS in 48m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/02/665402/10/check/octavia-v2-dsvm-scenario-amphora-v2/1b6fe7e/ : SUCCESS in 1h 20m 04s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"2c6cecc6977d3e1e912ffc2a71c82acb55922dc6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-22 20:08:57.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"ecdd6b9c86586c2052883b8502b26d204ca1f948","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-22 20:55:47.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"504419ba805eb302f80c2421475c66ccb675f196","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 23:41:10.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8f9d27ab59345f6a3fd318bcfc785e4 : SUCCESS in 7m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c317244b7bf4db6bceb58f33ae6583c : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54a0c1e37bdd46d280b3bc7d8df8ebd6 : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c89721df829489ca696791165203bc9 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11b075c1f7e74942aad0b0264d377019 : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f614dbdcfa754dffa14e6dc3ddfef1e7 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5504eb9e03e14e61b54ac90951aae746 : SUCCESS in 8m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0c0e38dc189f4bb0ae2b31fbf2714b80 : SUCCESS in 5m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4adcf5ee2a8449eb86243a2d1e22f1ed : SUCCESS in 7m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d9d6b45f6c224f2390750e31c6aba419 : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7bc091e224ed40f2910bf8f32f97276e : SUCCESS in 6m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee43f7d1027a4e7f9d7b9f1955ec37f4 : SUCCESS in 44m 04s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e88a7db635b54bcf8b1ec0e3acf0f1d6 : SUCCESS in 49m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2b3a665ed4749efbadbb0c024b0c695 : FAILURE in 1h 43m 32s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/09467c630d2d45338224b64ecd2277ae : SUCCESS in 1h 52m 18s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fb7553ccab0e42caa266ce77ed9c6987 : TIMED_OUT in 2h 37m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/66e3d85eedd54a0b97e755851def5ada : SUCCESS in 1h 58m 48s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4fb8aadec7a84dc3bf345a1011ae1145 : SUCCESS in 1h 17m 35s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/be6b24e22d464f21b642276a35dcc898 : SUCCESS in 1h 18m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e83afef25da043f3bf09a67fe6945f6c : SUCCESS in 51m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2bcc598205ab4cbfbfda7661c38cc51c : SUCCESS in 1h 38m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8d04cba114446128c434296898027bc : SUCCESS in 1h 12m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/02725a95521748f1b8de70dfaac59efa : SUCCESS in 1h 18m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7dc94d08c60347e39cb41e938a672323 : SUCCESS in 1h 31m 03s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"5beae597cb7e2c748bb980c73b0af0d467c13ed8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-23 09:41:13.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"29c202daecd88dd88e52bfe069dcf5cd911081cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 12:15:58.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65128058a82d499ea3efd34cc91a179d : SUCCESS in 6m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de0ee88812744a5c86535a5cd86336c5 : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1f5632e591d481b9e9562ce7dea3764 : SUCCESS in 6m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c77a1fae8a9d49469fd7841eaf3d196d : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9646751637524c86ae8126fc4b43bd5e : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6948447d80404f00a0b04a4ac04fb61c : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/202ea3ab474d45a7b1b86451b8b11eeb : SUCCESS in 7m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/82bccc6214ce457190e74334cc3cedd6 : SUCCESS in 6m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/dbeb2cd0f15e4f0f8e4f9f1b82fd4260 : SUCCESS in 6m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9b771071c5aa45e6867873441938b836 : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef42c145128b49a18fd9c53c628fcebc : SUCCESS in 6m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b2b33c2e1d324991b8f6d68ad4332c2d : SUCCESS in 1h 00m 30s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/92d53fbb8b6c4e37b5608b62d902737c : SUCCESS in 43m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0c923c81aa7646e0957bafed459ad03a : SUCCESS in 1h 44m 35s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d9a28859633842eaa7fd342dac8a6af4 : SUCCESS in 1h 44m 49s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/df80914dc2a44a04beec74b0e77d96ba : SUCCESS in 2h 29m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3ccec72523c84c45b0fae3aa0ff12443 : SUCCESS in 2h 13m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/1b691aea16fb432788c824daad2e65ec : SUCCESS in 1h 51m 34s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9b25a30204274272b39017b5d343bb52 : SUCCESS in 1h 17m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1ef864cba67e46b49dd16da935df844d : SUCCESS in 53m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/629294a3486b4d2d9584c43162f48d06 : FAILURE in 1h 38m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de0e1bd2afd444beafbbe6acc158c28f : SUCCESS in 1h 16m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e4879f2a2021469db73f27125066096d : FAILURE in 1h 09m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d69feb42f5164b21a2ab8994cd2bd9ea : FAILURE in 52m 54s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c37dae899580a9f277898a6320a7c89665d35338","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-17 00:27:24.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"8a1285c7a5e1b0f134d2d43197a58cba5fe8632b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 01:52:37.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8463ac786e342959511473d1f51461f : SUCCESS in 8m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0be844591f2b4b54aecedb174abad4fd : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86afd6d244334462ac4f8099310b11ba : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dbb37260ddea4d679b5f19ea63512ce5 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc65971a86a14589bc2d3f103f453e23 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae49a7e042ad4e4e9dc0490241b682d0 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ae8bbb3e3f549bdb6002235df91679e : SUCCESS in 7m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e6c0e9ef6e7b4bd581c1d33345c06677 : SUCCESS in 5m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a735224305b84fce926573e5bef6e373 : SUCCESS in 6m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/178336f1b11c45ad80f1108465a79936 : SUCCESS in 5m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9e6c6cf3c000423684e7c0c9f96336c1 : SUCCESS in 5m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/16e742df7a4b48d0bab8c3d5e8310d57 : FAILURE in 26m 56s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ef6c695e8d324551a67dfbfbea71d99f : FAILURE in 26m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/df8d811857a4467fa29fe8f108c25766 : FAILURE in 26m 00s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/43da4149b9db40bab9e2324af5ba5e6c : FAILURE in 24m 45s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/54e5ae43eaab46379f0a6d7341a8ffe0 : FAILURE in 32m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/81f55288ad0541b098bbfebf45fcf859 : FAILURE in 32m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/50c5cb87b20041ee9e3840529b2f71cb : FAILURE in 27m 38s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b68e65396b0c4f8d9044d8ad90fc9576 : FAILURE in 1h 18m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e2fab9f3409741e0b260a909b180bea0 : FAILURE in 31m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/288d1f1c66424f06a86f97a0decdb8a9 : FAILURE in 26m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b56ee139d0db4a64a92226af6e5f995f : FAILURE in 25m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/01965cd22f47466ebc1ae63181ca09cc : FAILURE in 24m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7635d98171224195adf2c80429a4c10d : FAILURE in 30m 19s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"2935db02c7474f3fd92fc14ed785f3d55881d8fb","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-17 05:33:47.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"ebb5bf6784936913e613ad7f79411d9f963da890","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 06:51:26.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e02abba5ea34128a7d103cef9c5c917 : SUCCESS in 7m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1dfa969c726f4e7dacaba64dc8075d79 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8e48dd97ee04534a7ee2dc3a417f925 : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/16fe7ad1d2c74fdab45f15f4b2ef0c45 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ae57ae1dabc42149e1b38cccd0dd333 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe10ecee319e4dbda4bc6e174f0094d1 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fb3376e5eea4d078e0f254db4706a8a : SUCCESS in 7m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e34d852438944882bee8018b5a8c59fb : SUCCESS in 4m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/917cd40585764bf880aa2636c028d8f0 : SUCCESS in 5m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/53be24492c7945f3bdbe41b0fe353677 : SUCCESS in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d098b44b9d2a4f738e7af4d9208541db : SUCCESS in 6m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d0f5fa3dad964b6f853a49d270fe34b0 : FAILURE in 21m 28s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ea63ae2d1b92473a807bfc88485f5a9b : FAILURE in 24m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/57927b629db34267b7dcf0ce7a8bc2bb : FAILURE in 25m 16s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0d794f60f48d47f9a7f2b5155210db96 : FAILURE in 24m 05s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/adf63754989f403796693e7ba1862368 : FAILURE in 33m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5d8d9354991446b793c057e8f4376aa2 : FAILURE in 28m 50s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/027a61a4ce664a69acbf8ce3e0a19111 : FAILURE in 26m 28s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0f09372cb58e455dbd16a40491712a97 : FAILURE in 1h 11m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9d029cbd98824b4589836ab3404627c8 : FAILURE in 32m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4b57cd9a47ed470696fae393c80c9b21 : FAILURE in 28m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2f63341ff034f5d9e0250f514ced1a2 : FAILURE in 31m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/739675199d8e4cdbb28af695fd289e1d : FAILURE in 31m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/46239ae6ebb34196acf0aecf755ddb6d : FAILURE in 31m 04s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"d0fac0d23ab5669695b782d54aa5aa193b365215","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-21 23:54:06.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"0c766745248cc337c4b8618e9e9b81da48c6265e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 01:04:59.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d68fbbe411894760be973b7a3a397a4b : SUCCESS in 7m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/731abc7c396f416a943e041b551ac5ac : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac7cd433ef9344608fe56b57eaa61881 : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5abfa0805f3e4a6bad3e57f76030a3d7 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2f6be2d70634766ae731619e4cc0b69 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9a79f74e1844340a741b27c67feb2ab : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3552d599d5a6415e8b95699047c3b2b6 : SUCCESS in 11m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c6a943816485465e85b261cb40a6a980 : SUCCESS in 4m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/40c395eec35a4449ab3b64802c6f95be : SUCCESS in 7m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d21516e5d7584bcaaf0d05ba2fa3577a : SUCCESS in 5m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a10ba378033e45d995ab814ce902bb69 : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/57666022c39d4220af0d809bc6f457e7 : FAILURE in 27m 16s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b315046faec64c32a3b77f900ecf744d : FAILURE in 24m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f7861007c4b4e97864aa022b7677c8b : FAILURE in 23m 51s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/583234a8e7434bf7a372093b15f7b023 : FAILURE in 23m 21s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8bdb6eb33e3b4fd4a59e99b97ee298b7 : FAILURE in 28m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4305633130e54554ba8bd0a004cea391 : FAILURE in 28m 01s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/929a9f5b94aa4d50a450627faa7825ce : FAILURE in 22m 45s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a12839f685174877b3eca8e9e5987a94 : FAILURE in 1h 07m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4776e253a36c427eb07d9b40607192bd : FAILURE in 30m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a916e53c12074b74bee2b2345f94755b : FAILURE in 28m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/580705013fd649d8a181eb26c304beda : FAILURE in 21m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6a78e192a0b8480bb27beba7d20a7ab5 : FAILURE in 24m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e05b0796c12c4446bb1b74f8f355ff71 : FAILURE in 27m 25s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"0ec41420331928ae7e8afcc063a4da38fe7ec503","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-23 08:25:27.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"a63417283c512b3fe90a5d08798e6a92d0a9c659","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 11:32:04.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de953e1602b44220b236c79c701160be : FAILURE in 11m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/853fbcce91b841cda2dda48dd5e61c04 : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27b949cf6f6342a5a1afd7945b2030e1 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7ad0bafdcfe466095a40c33142b1676 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2bd40ecdcc0417ebbfc5a0dfe37059d : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fbe5c3666aa436cac0d9e1452cddb2f : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/583ec50128af43e895ee174eca736bf6 : SUCCESS in 14m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1ff056b85d6a4892bcf62063565049fd : SUCCESS in 6m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a15cb66df45e48529a395c4d04fefee3 : SUCCESS in 8m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2b294d592fbc4230903100932363db58 : SUCCESS in 6m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c2fd8b91ea3d4306b6c6d63c7920b2ea : SUCCESS in 10m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ddb10af781e94cf4b7310a94614571a8 : FAILURE in 2h 02m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c621a93f5ece4e36ad8a45f04854dcaa : FAILURE in 2h 18m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f3e20a37b76740328dd22d3dc25e5d1f : SUCCESS in 46m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e6d16be558a24749bb4ea8e2429b4480 : SUCCESS in 1h 14m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f16b3b44333d43088afa8d586a500d50 : FAILURE in 44m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/09d1365be47b4c508a9abcffbfc3d224 : FAILURE in 2h 21m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/52ac0ba2251e4b9b8412d094f2ce0ec0 : TIMED_OUT in 3h 04m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6e7907ee0e02491caddef0267fbe8754 : FAILURE in 2h 24m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3c84832519074cd4a43c2e09ef4cb742 : TIMED_OUT in 2h 34m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a0d3bc6f5b3d45e3bc15833032b59834 : TIMED_OUT in 2h 33m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6fa8c21c197440b4bf4d48bcbacabad3 : FAILURE in 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"722253a28449b54390e14dac1d391f8eb3e35a0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-23 12:58:36.000000000","message":"Uploaded patch set 16: Commit message was updated.","accounts_in_message":[],"_revision_number":16},{"id":"2c850645fedeb1a05056cd2f6dad3dfde5da0c4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 15:11:17.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41f63f4ecb72424da46f64d81a71f6df : FAILURE in 12m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8ed1075f95541fb91bd96bdf89f01cf : FAILURE in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f388423e44784756b644a93a32288ebb : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75726256b5614a2884b061f996601c47 : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dea83168bbaa4a5fa0552183039c3cab : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abf5b7307da44a3da942a67960cab23f : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65874ca876bf45759095f7e265ac64ad : SUCCESS in 13m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/10e4ec6e9898471d8dc7fd817a9569ce : SUCCESS in 8m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/21e48a09352e4278b7669956901e0df8 : SUCCESS in 9m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/149dc843ef5a4607b84c1716e2c92306 : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/244d3a9d8ad1449fa3061edd71b1e161 : SUCCESS in 9m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cd4632ae97524f2eb991c21ec4191884 : FAILURE in 2h 00m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/788a4a226a7842c5a4e31526abe1bff0 : SUCCESS in 1h 42m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/96fbb60bb0bf4b07969986b99538e81d : SUCCESS in 47m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4793b5791cf04170be7f8e09c0680f00 : SUCCESS in 1h 01m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8be37b021d641a88bb8c50d853fc9fc : FAILURE in 49m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0809857a7c4944ad944991b3a9a1e782 : SUCCESS in 1h 36m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5d1cede8e7bc4b318e49533ebdfb7592 : SUCCESS in 2h 08m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7d18b163e8964f30b28ee33e357e26ee : SUCCESS in 1h 59m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1234a5bf8fe64153b3c451a180bea8a2 : SUCCESS in 1h 47m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fe65264abedd4c87b895ae7d2b124de8 : SUCCESS in 1h 50m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/81dde6371899471a96720c5e91f8edbc : FAILURE in 11m 42s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"ae2ddcdfee799cabfe347c4e2e8f08e00178cc0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-27 12:03:59.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"9d9ea064eff138ca66c3ab80131a01103b7caf2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 14:37:20.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f26c7385b112489183f4630ac1ac9c14 : SUCCESS in 11m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70ce6b79e0044ce2ba584bbfb4da0ef4 : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ebc8a84133b412aa23fbf418a5bf2bd : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8475b5f00174398a62e0cc521aa7fc3 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d2747b34eaa493dbf61e971ebd01850 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c339270618ee48a88f0dae41949b7181 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a0573fa384c4000aca6fa42a15a356b : SUCCESS in 15m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/31e7b9b4d3414c8db3ad47ffe2ac5686 : SUCCESS in 6m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/40eb4df4493a4402aa9d6dca7b073c20 : SUCCESS in 10m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4c55947014594b9ea3110f2666f1ecac : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/64de0e4b5c304076bc835b4fcf00c548 : SUCCESS in 9m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/098cef28431b4589829a70cdc44e2269 : FAILURE in 2h 02m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dea435e407c1495baef811d4aa24efe6 : SUCCESS in 2h 04m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0d3fd04f4c694605837a77c64176f61a : SUCCESS in 38m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d622b6fdf9fe4e4481bfeddeb489d31c : SUCCESS in 1h 01m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/57a7110cc9e04e9cb57d426d71ac71b7 : SUCCESS in 37m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2966d688179540c8a7823fccca21bc28 : FAILURE in 1h 37m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/27e6bdaa9c684b9488c0e5a28708e056 : SUCCESS in 2h 31m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/631a47aad146416181de503a35264ed7 : SUCCESS in 1h 51m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/45afd855f4cf4fa887c48e97e41bd234 : SUCCESS in 1h 41m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f64dcefaf37043558372b1ee5a649cf5 : SUCCESS in 1h 36m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ac1856fb03184b7892561ff2bedd80e3 : FAILURE in 10m 12s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"0a4b1e28f95fb401eebe08e6a0f3dff16239ef45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-04 19:38:30.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"93cef7d8a65eeca514a1925f9f49cea05e9f463a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 22:46:27.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cca9d019ce94ccea56e66959e4db437 : FAILURE in 6m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1ba940cff384a6194bd09b1c42a191a : FAILURE in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64a0f9d2784348548bd75921b13d1d25 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03025c077b5f4e2ea62e71641958f1b2 : FAILURE in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b29e643509c14045b8a02e39626b49f2 : FAILURE in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83e7e0a159694827a1bc170300c7ef42 : FAILURE in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/719f66c0b52442cc8628db86eb8cd603 : SUCCESS in 12m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c82d5e73ca1f41a48db353b7d1e745e4 : FAILURE in 6m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1d5db193696d4d7ca9546abc7f021cb2 : SUCCESS in 10m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b11996f9cca847029b380ac5059c9831 : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e0807a6ba3f54929a7867cc77ba6e83a : SUCCESS in 9m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1327b59cf6734f12a48d896ef394f8de : FAILURE in 2h 00m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0dc47a1443a24a2e86a18240415ee204 : TIMED_OUT in 2h 34m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aa76b01025a5471db66e76b9d4313f25 : SUCCESS in 45m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7664f699270344ff8d6e4359527828cc : SUCCESS in 1h 19m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0bc4d7eda6be49c8abfd71f41e59a3b7 : FAILURE in 37m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4750edae45a646218ad773aa0dd971fc : TIMED_OUT in 2h 33m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7c56f702cc6544c1ba31d0cc39409edc : TIMED_OUT in 3h 03m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/21ad5fff15bb404ebbbd80f8b430a9cb : TIMED_OUT in 2h 33m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5ee84a249ab042559399ad19bcc09cc0 : TIMED_OUT in 2h 34m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5f61a234b8b8414fb6f8dd172258dc5c : TIMED_OUT in 2h 33m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7909880cc3a6406587a7187856d988d8 : FAILURE in 8m 22s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"7b65e01b6706c49e0a72abb87bb8b6ca1da621f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-05 07:16:34.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"844008b935b85b1853c98e1aa8719780374c9c9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 09:33:01.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6c59813a14a49ea82aaebfd01f41f8d : SUCCESS in 11m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8671bee5806482bb44208c7c3d8092c : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfaace9575c64bbb83d947a60ba3caf9 : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd8faab0d3bb484f8a61b621fb8f938e : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/804b28cd44714ec6a06dbe992ef65559 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/382f1b12322f44ec9976aaf8db7b50dd : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9410222bc48b4595bf23b6b474b74301 : SUCCESS in 12m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7db8f0649a2e45aca4c2d49e1819241d : SUCCESS in 7m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/dfbb58effb2c4bc48b3f6495744199cd : SUCCESS in 9m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d45529dba07046808d3b5c69ba30303e : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8321686109eb47deb8ec8c17860fe360 : SUCCESS in 8m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e12f6a6203ea44e8a23eb4ae815a638b : FAILURE in 2h 13m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ddd5c627908400aa6df51bb187e33a4 : SUCCESS in 1h 21m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b81c4df488e047ae95c7506771fb1bce : SUCCESS in 38m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ed07d2562f3a464dab99e022d6ffa148 : SUCCESS in 49m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/597c3e524fd143a392ceabc39d1b2235 : SUCCESS in 35m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/64218ab85c264d55943c730fa9b8fc66 : SUCCESS in 1h 41m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/58c0e9b4a55548a2a1b7e4949b2110e1 : SUCCESS in 1h 58m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/65ae61f232bc45dba3aecab889020042 : SUCCESS in 1h 40m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/35b4b1978e144b7b9981b9a03ab44952 : SUCCESS in 1h 35m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6e5a676715194be182876c88e8b18c14 : SUCCESS in 1h 25m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5f7a3c7326fe4c78aab05756ab1172cd : FAILURE in 14m 10s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"8fdb79376dcfd5facb3f7e0f6c80e7eb284ec276","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-25 14:23:50.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"674d57283937c8e8854a21959c2a75ab8ef905c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 17:05:24.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0a965de68244476bfbae69141ec752e : SUCCESS in 12m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/110d5667dea345e88ad6b7b8393addf9 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53a433faae1946bea7cb2f280517fcac : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0ff7b7c730a419eb99720b481dcabad : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f75876fa6f9043ec9c9d1190ca7130a1 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d8ab72e6a2044a4bc4a9f0a9371bb6f : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebf67b3ca6a14e20abca7851954a8e7e : SUCCESS in 13m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6d9fee4fe7f492e82da805ba9c6c7fd : SUCCESS in 6m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/df39d23f09bb48009ef65b177ebcb903 : SUCCESS in 7m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cef2f26c8e1842debfedf9f881c0a6a5 : SUCCESS in 12m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2bb4ef2fead74506b85b6630902668c5 : SUCCESS in 7m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3798b4b0d1504179ac06e5a6d52d4f4a : SUCCESS in 11m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/345205f7daca487f8a8db05fd87ed3a9 : SUCCESS in 2h 08m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c5d700b051694d2f83eb30b6fbdd3313 : TIMED_OUT in 2h 35m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6679134e769544c2b9180dc54eaff15d : SUCCESS in 38m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d908be9adc0640fb8fda1f1734b9c42b : SUCCESS in 1h 11m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15d562b5586749cdaaae7f6255e3f4f4 : SUCCESS in 34m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1ea7a38e18ba4b75b4789e3ba9828e80 : SUCCESS in 1h 58m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d23d7ab1e5904fdb996514ae4c913f88 : SUCCESS in 2h 37m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bf21a267624c430980a6798bd1fffc5a : TIMED_OUT in 2h 35m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/be4cc11d92274174892e5f9deb4edaa6 : FAILURE in 23m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/16e2f58e33354c14be6778fd3a4f1e53 : SUCCESS in 2h 22m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/39b94715289f4a468c0fde402641f973 : FAILURE in 8m 39s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"8d252147da8f3ff7cbe2a75abfa61532cb33fb8d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-26 13:18:54.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"d0d384356c3403867d8d2833fb59713896140d74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 15:51:36.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95c5e3a97f7c433f8dfe1e15e6e96906 : SUCCESS in 12m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/364f98e486444baea3fbd7d9c9d47559 : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bde94db93d9e4b249372ea31615ea3de : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89683c1438d5451aa2815a28e0da4393 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/805f82ee1e10473b9d0f8bfef8b17e9e : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fec75360a13b496db4130b4c3dafe086 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f531655e7634ba482d989548b2c9195 : SUCCESS in 12m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fb06632c7374f93afaff3eeac5402ee : SUCCESS in 6m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fe99b9c4cbc84329a514d904638f9032 : SUCCESS in 8m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4b7eb3174a2245ce9fb96ab685a55e05 : SUCCESS in 12m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/99ea996633e74260baf6b602c03279e1 : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ca475c909f3543b2bf9862b9ff427eae : SUCCESS in 10m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cda37cd58e8f4139ab410fc918867100 : SUCCESS in 2h 24m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2fc1748b62f640929ab93f83c7317318 : SUCCESS in 1h 59m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/89f893287fe2407fa7bd4736722cb8b8 : SUCCESS in 38m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/881fd2f339be47e2a407880c93e9cb61 : SUCCESS in 50m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f5e160737a36446488dcd5999624772e : SUCCESS in 36m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cdb0d41d344d4af98069c29bc8841bed : SUCCESS in 1h 56m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e96545e997fb45cfaf754b61aeb765a4 : SUCCESS in 2h 18m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/75757862e4f04183808deb0be97d3360 : SUCCESS in 1h 49m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8cc1338f1e5b4f5984e1109c0894c4fb : SUCCESS in 1h 53m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/55c2972259df4052b0c566c325d5ebc3 : SUCCESS in 1h 43m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/205ea4e47d9b40d6a25be4ca9c34e51b : FAILURE in 11m 10s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"a64d5fb93297032d5fbc0993400176ddfde3d6e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-27 08:39:17.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"0a4ba4ae6b9b9c20a343d53e0811d216199ffcc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 11:17:13.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e194652d68564fd69d345945710542c2 : SUCCESS in 11m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47d4023008f6442abb4e71352d39bb6c : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c164331ebf234fecbd9970c9cb6a359a : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a807f797e3a403580e746334014b514 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/792bf8859a984ff0919065ef1ae60199 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19398bbf534c4e8797fb5e5c5275e2e3 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97035a102c4642a5b2370949502e284a : SUCCESS in 13m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/811fd4fe07864c38934076d0a3d0677e : SUCCESS in 5m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bd8ba641bb83405082553c0a39879d81 : SUCCESS in 6m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a0822995e928489aa3c78dc4c106c212 : SUCCESS in 9m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/38f0c6f5121b435d97b159e793606b5f : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e4a1e3aace4545d6b8a7c1260d5e21c2 : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9d9499c5b8414f3b9dc5bbf576bc3df5 : SUCCESS in 2h 34m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/64d1cef6fd8640d2979f539d9325b663 : SUCCESS in 1h 33m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d17ebbc22b4e47b08681685de21bd833 : SUCCESS in 45m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/53f52ea6e058487c993b1a500d53d3cf : SUCCESS in 47m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8fa49fd47e6471186e9731b528353cb : SUCCESS in 34m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1a1ea21ba8ab4a43919faba20749684e : SUCCESS in 2h 23m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/027855df7d8642b4a9bec1613c67e518 : SUCCESS in 1h 53m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7fef8d917c824c94bdc07eec56c256be : SUCCESS in 1h 51m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a442e47acee545ea99b94bbf87bb1874 : SUCCESS in 2h 02m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/98ff460c659342d4b7a68414da9b2e62 : SUCCESS in 1h 36m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c5d36d867c2b46108a610fdbe697a53c : FAILURE in 11m 59s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"455a4c16bb4dacf778597635f5cb8d795f729d1a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-29 08:25:57.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"619e0ebc979aa898f531b21097db15021c2a8223","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-29 08:26:06.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"4f7324d87f655eb143066c3c2fe3a8d1d07a8346","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 11:17:34.000000000","message":"Patch Set 21:\n\nBuild succeeded (experimental pipeline).\n\n- octavia-v2-dsvm-scenario-nftables https://zuul.opendev.org/t/openstack/build/9960437b47a1457994036e9c8fb5efd5 : SUCCESS in 1h 29m 51s\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/04b8985bd2ed4c6eb6cc081d99c98092 : SUCCESS in 2h 44m 25s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/1fa70432e1724013b85bd59b0d1f9e70 : SUCCESS in 43m 27s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5d4808fb012046d18610cb97de47510c : SUCCESS in 41m 58s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/24c9b49f1d994aefb7fcb43bb006681f : SUCCESS in 47m 15s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/720885769c7247a4be9fbc18e5e4811a : SUCCESS in 2h 03m 22s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/9d1c425aef044097a93dab6b82c020b7 : SUCCESS in 2h 24m 23s","accounts_in_message":[],"_revision_number":21},{"id":"29a90660bcbe0892b93fbaa5779093e14600b9c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 11:34:57.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2991b11fd1df40a3acf1271d6c41b330 : SUCCESS in 11m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/856641f2c9a44226b216a366a6ecc317 : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ff8a6dc7053426ba41df82be2ef1fac : FAILURE in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9382c2b228de46659e7b1c77533e585b : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62752983d1974727aa0781fdb6219e82 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/803136828bb248d699e22fafa731a97c : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0381939b2a294dff82eaa8dbc52032e6 : SUCCESS in 11m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a48a859ff45a4be8971a589540bc2fb9 : SUCCESS in 6m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/39d383d5cbed46b3a33d18e67d582ee4 : SUCCESS in 6m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1e2a05be20a8421fbce6e005ba233678 : SUCCESS in 10m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/83a12b8cda3548ac9caad14acadea798 : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0d4b63df0853494692c897c7492caeff : SUCCESS in 8m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a48a91b4272847ea8d09eeb9e60ebad7 : SUCCESS in 2h 31m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1988889c6b53428db44df49016a826db : SUCCESS in 2h 04m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ccff4e2b198e44bdb391b3fdb867b43d : SUCCESS in 1h 02m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7e4f2b82335b45d9a409acc99aee8075 : SUCCESS in 47m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/321525401f1640b596982ef1383a7ca2 : SUCCESS in 32m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a21b6c5a33774e438fd82cf9e1e48feb : SUCCESS in 1h 39m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/16551bcf1c514e1da1b3840b1b7352e7 : TIMED_OUT in 3h 05m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/588f5a218a7f43298670ab671776363d : FAILURE in 2h 13m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7fcd111e1302413a8119540ceea51448 : TIMED_OUT in 2h 37m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3268baccdb974ca9813d4dafb5106f78 : SUCCESS in 2h 21m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8395a8a625844bb7a5f8e786c9fc1638 : FAILURE in 8m 16s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/api/v2/controllers/amphora.py","accounts_in_message":[],"_revision_number":21},{"id":"bcf263db354cef6cdb8385c49df92720806af90d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-06 10:21:05.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"a533e5f65b19bdb108602ab62d26c4754608112e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 12:44:28.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f231763c51bd4f64835b84d8b7f9e536 : SUCCESS in 11m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4c706efe56f402facb7d7d4a68ee886 : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b968069866a64d909743eb87dda05849 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/393a2824d22c4f32a094c141bca75d33 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3a6c7ba23664dd0b64c03c0f6cd9dcd : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/736353b430ba4709aa311aa1ccf44813 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecc00975b7bc4a34ae58a39b5a2aed07 : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31596803d44b4f0aaf7f0ca9abc284ae : SUCCESS in 6m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7769e6d173264791b5a275c47157a619 : SUCCESS in 7m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/52a487fc50d848c490568ab4b91ffd72 : SUCCESS in 10m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/440bf2e04d0b4eb1a5ff93c2a8764479 : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3ed9c3d1794a4f0baac852da41704970 : SUCCESS in 9m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d53a7cba502a4ca0974fa2adf87fb692 : SUCCESS in 2h 20m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f6393edcfdb4df4a0deac2d4edddb8f : SUCCESS in 1h 39m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6046373d0ca54b0b9d7fb048059be782 : SUCCESS in 32m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/465aa08c76124397950d5777a889e4d2 : SUCCESS in 47m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2d35b37a2bb4cb7a968498bfa211c45 : SUCCESS in 31m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bea2ccec664d4dcc9248f8fe7443ef9d : SUCCESS in 1h 25m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0a2789eef96747ce9f86245c6997f5e2 : SUCCESS in 1h 58m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f52f837919864ea5a9c8b7a90321391f : SUCCESS in 1h 51m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/55d05b8d0c65442d976bc3b81c832a75 : SUCCESS in 1h 30m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/442ed2fcbb0b4ddca392209a22892796 : SUCCESS in 1h 30m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3894d5bffd0e4995bb45183ee473575b : FAILURE in 11m 54s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"8774fc4fa5b8e0e64fef685bf3f238e158ee60b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2022-02-22 16:41:34.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"aa7ca13cc1ad4f38f598645ea337634d7a779bb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 16:46:58.000000000","message":"Patch Set 22: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":22},{"id":"21646696b39df01898a0fe256eb72ee5d33bf9a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2022-02-22 19:55:18.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"3adcc72c77dc4517c64802845fbb81716e30313b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:56:10.000000000","message":"Patch Set 23: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":23},{"id":"b9f5d4c50bfddfcfd81fa6d953c074f20ff368af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2022-02-23 08:00:07.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"9e3bba128271a53533bbf21ddfb7de18fc4870a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 11:09:13.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09ffc869de2b42a9a32d7245e41fefbc : SUCCESS in 9m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61ecddda4f974207b3867cbbcdd68366 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ff9cd65969e4a9782f559352ca45e1f : FAILURE in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70de38d893954a5381622936b80d3df9 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c85a5c60731b44cc905ceae558eef013 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69a4161d30d94b97a91c8f6321c290f2 : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b44b5e7903a948d98417ec33039e4301 : SUCCESS in 13m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48351453984f4f3b9f77500386ab4ce5 : SUCCESS in 5m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dbc15e109a4c409b88c49b8954c718c9 : SUCCESS in 6m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/32ed6c448f3a4396b50cdd31705f9131 : SUCCESS in 9m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/528790119da6468c833be6f97eb3f345 : SUCCESS in 6m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b3ef870587b04759a3dbda1fde535278 : SUCCESS in 8m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/725dd91477c44c4f9732d796cf29de70 : SUCCESS in 2h 39m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f417d795cbc04684912c72e4d38ef409 : FAILURE in 2h 17m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/909208e3713940a093146506a2f311ff : SUCCESS in 33m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a247d740ecde4b709a7c1f45c8cefa2b : SUCCESS in 1h 04m 35s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/595acdaddadb4c409126a54f4eccdd7b : SUCCESS in 1h 11m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c349dc3aaca4c6e9addf8b4c0d0de43 : SUCCESS in 38m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/69b17bb2cbd44a86ad487ed95c4310fa : FAILURE in 2h 01m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dfb6d6930aff4f9da4abc46893890c30 : TIMED_OUT in 3h 05m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e49d6913a8cd47818e70b773e416701f : TIMED_OUT in 2h 35m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3e8caa8350804036bf9ec78bf40a6a86 : TIMED_OUT in 2h 35m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/518d815e1d1948cdbcf8d1ff1969102e : FAILURE in 2h 05m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e888d1e3bd440c6a7558fd2a13e4e76 : TIMED_OUT in 2h 36m 47s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_interface.py","accounts_in_message":[],"_revision_number":24},{"id":"55a98b1d6cbbbec671dd1c4407bca2db45a92df7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2022-02-23 12:23:41.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"743cf94028892c26f15abe6b050b875fcda5dfc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 15:33:40.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ab80cfc2160475fb84fa77cef6ff918 : SUCCESS in 13m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abc014e6b67445b6938fd6257246b439 : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/861f015d7dec4ffe9f5a58dc6dba3839 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2eb8e71666646daae5961f2a00bf5ab : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4696204e18e14e87a545463893de227d : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47f2f1029ea947b2a4f0c480e5eef420 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67e038000dc44650ac8ef80ef171b4ae : SUCCESS in 11m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ffb8a23fd6a41799b3af2aca914b67d : SUCCESS in 5m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/09369f04c551481aafea13ca50b909b2 : SUCCESS in 7m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a425acbe78ae490ca3efa0a238804067 : SUCCESS in 8m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2a8ea8fe7ebe4906b55066ef862f8ce3 : SUCCESS in 5m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c9781d546194db99b652829b518272f : SUCCESS in 9m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/13f7da4e63754cfaad24945583a6337d : SUCCESS in 2h 23m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/60cf1042889c44fc98468e12bc2c6327 : FAILURE in 2h 28m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bbd65b15ce7a4730bda22d947c2096c0 : SUCCESS in 32m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f2929cf95fd84c4ea6eca241b7adbccc : SUCCESS in 51m 46s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/356a496952f14c44b110908d7625b26e : SUCCESS in 1h 09m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/07f750d77183499d8d43b9b21aded5a6 : SUCCESS in 34m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e14407a847844a868e1be43ed8e37562 : FAILURE in 2h 23m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6915211e0d0e4372a87becd1404488fd : TIMED_OUT in 3h 06m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5fde6610632c41a7af8e9c94dedd9b1f : FAILURE in 2h 23m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9051cd71d5ad447b9d36c34d5505eafa : TIMED_OUT in 2h 34m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/baa1b5a9f148454ebe2441247bdb1e6a : FAILURE in 2h 18m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/25946543d61649dfaa1e1708c6797017 : TIMED_OUT in 2h 36m 01s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"2dcf87f7581164ff0c17ba88d424e4d57b75da73","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-25 08:06:18.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"859aaa1f9379d1f83ae98df83b9b8f4cc8985559","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-16 10:48:17.000000000","message":"Patch Set 25: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"67aa39e78a190f5c646d0e3c3bedb3d22efd5d90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-21 09:16:03.000000000","message":"Uploaded patch set 26: Commit message was updated.","accounts_in_message":[],"_revision_number":26},{"id":"d3150b9bedf0125f8cc572bf1fee27e52be2438a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-21 09:20:25.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"5d21f98c326cbb7b6df6a122c8eefb1a31188f96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 11:59:25.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0eaa887695b84d289b820fa186cd2390 : SUCCESS in 12m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a2e687e5d4e4d429c975e5a1cde8db6 : SUCCESS in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df26f39670ae413d80d8e35a0181edfa : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad9d34ebd0464994b65f3daaabb5aa65 : SUCCESS in 7m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4de1a1689f834c70bd39e98b8547d8fc : SUCCESS in 7m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2ac09424a4c4d05812a3a1f825f0bb0 : SUCCESS in 10m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7c760c36fb64a3ba03075de200d89ac : FAILURE in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc56d24e890949b7bc0706c9d17a5f53 : SUCCESS in 6m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d446645a1c5944078e9c1d86e646b361 : SUCCESS in 7m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/32a7d579e3644ead9264d8a20ef9bc59 : SUCCESS in 11m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/715dc16b418741ca9799583a13f56364 : SUCCESS in 7m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bbd90afab7724b6187af718a96aa288a : SUCCESS in 10m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2e67c8c95d2348c78790fe1890143706 : SUCCESS in 2h 23m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/549e805abcef4dbb97a5267c41d56532 : FAILURE in 1h 49m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/676734b4a65c494f9b1e3be55bdb5df0 : SUCCESS in 36m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/83369344d4cb4350af3d4c1fde2519c7 : SUCCESS in 47m 08s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ef2f496475b84f2ea2326f1ffd997d1b : SUCCESS in 1h 32m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/708a42a6d3394cf3bb8c8ba917d2a6ee : SUCCESS in 34m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5fa0b8c91aab47d6b0db35113c122450 : FAILURE in 1h 59m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1a516bf59e5e4da48878646ab7040cee : FAILURE in 2h 28m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/552dc35e371e435ebcf802f230bace41 : FAILURE in 2h 35m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a58aff7950de45d0b97f9b935193ef62 : FAILURE in 1h 53m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a77a1dfb81e54d8b980f344cac873ba6 : FAILURE in 1h 58m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/2efdfad87f1e40a7be5615dcded96a51 : TIMED_OUT in 2h 37m 46s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"95b7086e85eeed4683a371c4f4db8e3279f73049","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2022-03-28 08:32:38.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"afbb42dfbbf502579b54860416b12734bdeb8877","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2022-03-28 10:28:27.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"dc4bcf916d51adee636a35dd45ed143b106224ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-28 12:29:55.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"67622795374323cf10af3257faa4c8e035b9e699","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-28 14:24:51.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.","accounts_in_message":[],"_revision_number":30},{"id":"88c9285e7c0a9597ad43a7c530b9b04808bce462","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 16:48:10.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c70eb5478a74e1c81c076a6865abb81 : SUCCESS in 12m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e42a813c1414207956b54eede9ab7fa : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08f7428dd3f74c50977df094b5825872 : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac187738701c4e27bca38fc8f269e248 : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc67c6dd1c144a0c8b8d2530fc2380fd : SUCCESS in 8m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef80ed2dea8545f29c0e4596aa598528 : SUCCESS in 10m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a0a2f429fd7414da79bc3366b34f358 : SUCCESS in 13m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4c9aa9570ea4b908249637c398351f5 : SUCCESS in 6m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/822c154b3b6744fe89d6ccd9cd1a3b7e : SUCCESS in 7m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0aa5da411ba64d8e8223271716d4004a : SUCCESS in 13m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/581f0fed41b74e4cb4e55c992223efd0 : SUCCESS in 7m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/805ccfa80da04e68a029ac2007ddaf77 : SUCCESS in 11m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/63d61bb2fc8e4b38873beb9c6d84f7a0 : SUCCESS in 2h 17m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/48add4c079fd48988beedfc936b18b9c : SUCCESS in 1h 37m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c721b93709004859b59405213adf4a54 : SUCCESS in 38m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d7ade8ee019446fe9a1c26319adac9cc : SUCCESS in 46m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c4e8969e7f13481897d78bb107f1b2fa : NODE_FAILURE in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6cbdeb1493d74d8a92af5a7512c61441 : SUCCESS in 37m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fa966f0a7fee4c3eb5cc9fd5c6d7a9e0 : SUCCESS in 1h 37m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/00bf743f5e804149ae81054bf287206f : SUCCESS in 2h 07m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/da050b48865241e391f0865560170156 : SUCCESS in 1h 39m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d732bb0d02294818a808e10077fd1006 : SUCCESS in 1h 33m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ec1a4c2bfef14d39bf7b907180767791 : SUCCESS in 1h 33m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/6fca099d4168429aae8f4c2f8356e6d9 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"d7fd44fb7cad5546981eda8d7002abb6a519f0a5","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-12 09:52:57.000000000","message":"Patch Set 30: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":30},{"id":"f85341153968ad20df8f70480018bc2515a0e092","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-15 11:09:10.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"b35f98cf803839534d6e7223dfbf05557b183acd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-15 11:10:28.000000000","message":"Patch Set 31:\n\n(6 comments)","accounts_in_message":[],"_revision_number":31},{"id":"604ce04a33f76595b7bab54ff3d4063b5a922fbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 13:53:35.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb9e662e2e7846c1b04ec1e27dc84c8b : SUCCESS in 10m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26887ad4e3544234952ebc8b27c8e1a2 : SUCCESS in 9m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8243e4303bad4af288ef6ee34ccfdebc : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32e091cdf06e48ceb29b32a710191602 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4abbd74063ce430f9545f6b77cc47060 : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ec6f08434a043e5b5f05ab8278eb7ac : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46fc4044b0bc4ee38432f5ca145c7f9c : SUCCESS in 14m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd4849b470fe49fa8afab0c03af9b974 : SUCCESS in 5m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8d9745aa53b347499e6e4e1aebe68591 : SUCCESS in 6m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0631808d64584aab84223549b1eea16e : SUCCESS in 10m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/62dad77e76174be581fdadea3957a846 : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4a1b7edf626e4f358410d558e5646297 : SUCCESS in 8m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c52a42a3d3544b6ebcb16da7b2f283e3 : SUCCESS in 1h 57m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9cfc350eba554552b1965039ae732ebb : SUCCESS in 1h 51m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7eecab231b0645499095ae5b03485d71 : SUCCESS in 36m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/16793216b2024e7ebbda295bdc891ff1 : FAILURE in 22m 28s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a7f13c9c5fc544d391c9816411566616 : SUCCESS in 56m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2af61299eca244e88af4f957998576bb : SUCCESS in 36m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0fc5249771ae451490cc773a35280e39 : SUCCESS in 1h 48m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5a5baaad2a7843e5ba82251257b2ea19 : SUCCESS in 2h 31m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d5a587c385f7453486c644e7941c49bc : SUCCESS in 2h 01m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9d8bf8f2b1414247b49f63cd5a78b5fc : SUCCESS in 2h 17m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7764cbfed12e4996a03c6249e6d88443 : SUCCESS in 1h 59m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c2c9847884a4e4cbeb0846ad43ab6a3 : SUCCESS in 2h 21m 31s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"ab81283735115352ff4f5fc24085423108081ed6","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-22 09:51:53.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"ae3a4bb88c064eeb2c08d65cd98122674f11b310","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 12:21:39.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1175bac3b0d8487a97764bfd4af39e57 : SUCCESS in 12m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffefb8172e5145ec9ad0954189feea52 : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2acfcda34d84d12885ef9cdc1a01956 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/987c854f184d44718f34edd0f2db6b19 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff89a1b12b1e4f53bf9454e9962fb4c3 : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5fa4d4f18c524fb0b6a7e85bafc56923 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/decbf49a892d4a53b3aa25be56a9bfd1 : SUCCESS in 11m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a96f78a52b9d4e21bbf5230589519a58 : SUCCESS in 5m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a2fd65578857460ba4ebd8de640b9cea : SUCCESS in 6m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c94b34882d4b4c39adae504b007767fc : SUCCESS in 10m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8ba0200111c440fbb3851251df313e6b : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d82bb77725284d8d89bcd33d441c2a51 : SUCCESS in 8m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9ead822a484c412292ce2d53fec9e0bc : SUCCESS in 2h 23m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7cbb0068f0a14c3cbf904904b281fabd : SUCCESS in 1h 35m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/acfa3d7927b8453a8a967978a5051cab : SUCCESS in 34m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d307e37bc06f41859c9837b1890aeef8 : SUCCESS in 51m 08s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/16cffe580a164d1e944554cbd5d86d17 : POST_FAILURE in 17m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e32b999ebc44d9badd05f7d848a563b : SUCCESS in 32m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/66ccd1f5693c4a6b8b2f4e82edce4935 : SUCCESS in 1h 34m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/83d0261da5284014b66a3ed352b77f7d : SUCCESS in 2h 21m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ba0426975f10490e809441d0576c1427 : SUCCESS in 1h 49m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bd76ba94db72428385ee8b14d1187342 : SUCCESS in 1h 27m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/315fea6495b744a8925429c2ac389aa3 : SUCCESS in 1h 51m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb99593834094b76ad1b84fe23f08831 : POST_FAILURE in 14m 33s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"ead6a663ccce8e1fed76f251c8cfecf4ece12c1f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-27 22:19:34.000000000","message":"Patch Set 31: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":31},{"id":"15c0b6571d2502e85433035f84e2a277c435988c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-29 13:04:23.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"51dd49f5c96f9764976d15ed1d37c7b8f86284d7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-29 13:04:28.000000000","message":"Patch Set 31:\n\n(8 comments)","accounts_in_message":[],"_revision_number":31},{"id":"e272bf0c335df969127eb282456d85033bea8080","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-29 13:05:01.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"019fd77a725816c560c44a68fb581ee7ae334202","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 15:42:28.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b89332b1582c4c7fa085d83879c1a72c : SUCCESS in 10m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b489880a1b17415f8f631b2b1101de9e : SUCCESS in 9m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/074b93821af64a2c890c0a6d680c2884 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a9ac31cce7e4c57a3015cd597f3f1ae : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4811b420719f488e93252a9f9300959d : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49104f47d6624bc59aab64ed6e511acf : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4805068c614843e2b2af7befb672ff7e : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25f09907cd9442869c2032c8dccaa2be : SUCCESS in 5m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d6f616ceec0243509a1f2809468b0144 : SUCCESS in 6m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/68ceb6e554b043f8a0d398f5c80065e8 : SUCCESS in 9m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8a79c5d899cf40cab6aa2cc3d09bf54b : SUCCESS in 6m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/79a10abe2a684ed48ba41c1ee0a89a2f : SUCCESS in 7m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/40e980a992c24072810486c5f77ab80b : SUCCESS in 2h 22m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c01c9bdd7f4a4d4ab4a246e7bf4cc0ff : TIMED_OUT in 2h 35m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8295b7cfea9f4b4ca64b2d30866dff57 : SUCCESS in 42m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fba4d5f7ec8e464696d3d3c2fba79606 : SUCCESS in 1h 10m 44s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/614cac0cb773414a9d7d31acb84cf84e : FAILURE in 19m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/41a8200284a24273b3e4fa824cacb00f : SUCCESS in 40m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4337d3a21dfc49ceab7eb0aba5f45266 : FAILURE in 19m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0f95a4b413cf445ca4f087aff8b9aadb : SUCCESS in 2h 24m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/40a2390e9b2b4c5bb9d01faa61fe6b02 : TIMED_OUT in 2h 34m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0523192639fd4c529e6e74976be65f20 : SUCCESS in 2h 18m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b6d1c0399e5c4a9980372a18304b4430 : SUCCESS in 1h 59m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/bcecd3a156574ac3a92433b45889d012 : FAILURE in 17m 57s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"78bb896f22dc7fdb2aba1e27dd1b5d2dbcc58ce8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-04 17:46:19.000000000","message":"Patch Set 32: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"d12228584cb70b95482f61705d4e34587407e3f8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-04 19:39:01.000000000","message":"Patch Set 32:\n\n(2 comments)","accounts_in_message":[],"_revision_number":32},{"id":"48af33f286459a9595474f74f959d483f98cc7ea","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-04 20:47:27.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"c54e1b5666ce81bcfb737c78228c62e6570716c3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-05 06:30:02.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"d043a7a000c923c97b8869c2454fc229e28ee17e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-05 06:56:55.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"7cfdd1c8f1e761586344d896e1a80e390aef1c9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-06 13:25:34.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"b85421f61b0086f1d3d81aee5f45632a23da9fbf","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-06 13:26:55.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"466448d555354b387825202ab3a0267e069915cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 15:48:05.000000000","message":"Patch Set 33: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bc07e7fb01642449722921b13547246 : SUCCESS in 13m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1912eb937ca84c5890bd5dc65f9984f9 : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dafed527d7d449a38bdc6fba0070783c : FAILURE in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f602fc65b39b4d07acacb47e0f7fe4a8 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/217051d204e342ca8bbd9aab53eb5f4e : SUCCESS in 6m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27ea7f7918344ac8911d85f009f33a82 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c25ce84f7def4b07a4a7cf7744932ac9 : SUCCESS in 13m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed3f26eeb7cf476e8f71b26ad29c926b : SUCCESS in 5m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1a7c995f06c346f191cb5f4516cd251f : SUCCESS in 6m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/46e28805b62141149c6c4282ff930ed3 : SUCCESS in 9m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/86f94ac1133640878e0f6d76cc58a80e : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4ae761bc3c6e4b99abfd728c891a555f : SUCCESS in 10m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/64fd6da388b34382b9d7612bba9906e3 : SUCCESS in 2h 19m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95a5e5f829fa4e019561d988ba2e20f1 : SUCCESS in 1h 40m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/20f70f40e8974195b66d31068a63622e : SUCCESS in 36m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/44015f2b0c4f42f4b7542a852d8b9d5e : SUCCESS in 55m 07s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/40edb80f5f714937ac8b893272950f06 : FAILURE in 23m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78c9773ff0194d8186c79792f2a59e6d : SUCCESS in 34m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/70ad82b73e0c405e8d07e51c570b68c6 : SUCCESS in 1h 41m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/adf33f5d46814224b8d9b06929b64bde : SUCCESS in 1h 58m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9999ccd468ad425ab9951de185615aeb : SUCCESS in 1h 40m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/aa0351f7d04e4e58b4f1b02c8142b55f : SUCCESS in 1h 48m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bfe7e40a34d84a30b713ab2dc7aeae22 : SUCCESS in 1h 37m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/010ab87df1e24dea96f04e100c63fcc4 : FAILURE in 19m 18s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"31ae4726a5aee7f32de8cb7a78ac14564f3b96f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-21 07:13:40.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"735146c20d7e16b84dd3b33527742283fb324545","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 10:05:20.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42e4871610124017be9b8a583309c184 : SUCCESS in 9m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9032149c7af462ebfc5ac5ec8f8af19 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b79902dfbe84bc9b6344eee67ec70be : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb19a65e50eb4f43a0fb65e9049e835c : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8558e5760874a31879077077a142c48 : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc2ea1bffe4b4b2bacbda1a5b855c1ff : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33db70e757a54bcda9bada7375341028 : SUCCESS in 12m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfbb63c46859420b92b0007524511b89 : SUCCESS in 5m 12s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9556379eaee549c2967a397a6ca7e13c : SUCCESS in 5m 37s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/adb1af7ee840498a9ae86553718f2781 : SUCCESS in 8m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f58d1363ca124789b05e6b07fd4166db : SUCCESS in 5m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3c1fdece4b57478ea2dfc2efdbffd34b : SUCCESS in 7m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fee1e26ae6834c1ea95fd214e9ae18af : SUCCESS in 8m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d81da7c6019b4d1b84872720934bbb7d : SUCCESS in 8m 52s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5e76f01e6da14e1f92beffebbf606d0c : SUCCESS in 2h 43m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2f76f7c2f9e7424d8403261fd372917c : SUCCESS in 2h 08m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/af0e5c66958d4819aa0f053971be8fa9 : SUCCESS in 37m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/29a35d765ef94ed8ade33ea82d28f800 : SUCCESS in 1h 11m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b76b3509dc1e4a1e8a82700337c608a2 : RETRY_LIMIT in 9m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8231251c606e40a9b11a73da3a52604a : SUCCESS in 40m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/47617f9b9a9041b19849ec525c59fe9e : SUCCESS in 2h 04m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b8b66feec279432b8ad9076f54559059 : SUCCESS in 2h 15m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8d21e113e5f94ed7874b0501593dce15 : SUCCESS in 2h 10m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4b5dcdba70e548debeef4efb52058a4f : SUCCESS in 2h 05m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2ba9c9668d7c46b2a45eaf6fe31296dd : SUCCESS in 2h 25m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/18a8a8d18494476c85db4f007af1c950 : FAILURE in 9m 47s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"fbd8155d562ff8a35f44f01b1e019d81e23fd506","author":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"date":"2022-07-14 09:29:25.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"29e5917a875ca57534b48f6bcfcfc4fd8ace82a3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-19 09:09:29.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"d61f67d13804481dd651f198942f3565c48a4284","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-08-09 17:19:00.000000000","message":"Patch Set 34: Code-Review-1\n\n(14 comments)","accounts_in_message":[],"_revision_number":34},{"id":"eb527c0678a003b9e5dc4ec60e74925a26cec0fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-11 13:49:07.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"3eef834542e5293e02e61381f419e2173040b864","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 16:41:21.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08668fb0b32d48759e76a67b0d9e6425\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f19563bb0c24c37a787ffa89db9105f : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7c8baa80b8344cfb9e843d46296a3da : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfddf0d60166419cb27a5554de921591 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3135588ab73e410c93a543db251d7780 : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61810e6f3cdc4437bd9a2c3355beedcf : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/193bbec5f4184779b7e4ce20aaa14a17 : SUCCESS in 12m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81bb61df751a4471b6a562e75b111b38 : SUCCESS in 5m 34s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/cfaa74a7d40b4a9f9761a3159914f36b : SUCCESS in 5m 11s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5f761434e463444d97bc14af78da5498 : SUCCESS in 8m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5bf1a0d4821b45d983124c72bf409d98 : SUCCESS in 5m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bd463abfdc5a4466b6277e96003f92fe : SUCCESS in 7m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/83a2dbb0699641a8aca899a7417169b4 : SUCCESS in 8m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bfb35ec9f5f842568c28d79c3f1be966 : SUCCESS in 8m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/23377586317e40e0aa79b8b2e0d96313 : SUCCESS in 2h 27m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2363f92fb16a414c9a4e71ee3009be90 : SUCCESS in 2h 35m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/088c259dae8948fc93c8446d4ddd8198 : SUCCESS in 56m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6952e316d1bf408684b7b28eed121bb8 : SUCCESS in 1h 23m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c787ef898ff64fcf8be321208259c618 : FAILURE in 27m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/486dbd6492e240eface0a276e61c3cc9 : SUCCESS in 54m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/70b0379d5dca43d6825cb6df8d0ca79e : SUCCESS in 2h 04m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8a564b5e0073443b8f1a9673c5e48ce8 : SUCCESS in 2h 32m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f0d2764f446f4155bffea04dc37343e1 : SUCCESS in 2h 31m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8d4097fb4af44744b4b16168b5aa83eb : TIMED_OUT in 2h 43m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d952ef0f6e2943f2a8ce756958154b6b : SUCCESS in 2h 08m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b364c6c92b042539fae16561cd22468 : FAILURE in 24m 16s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"338ddf013e56b771ed8ffc840e3c269e96dfa425","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-18 07:59:07.000000000","message":"Patch Set 35:\n\n(13 comments)","accounts_in_message":[],"_revision_number":35},{"id":"8cfaeed23788272b528403490ca227420a33031d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-19 15:06:25.000000000","message":"Uploaded patch set 36: Patch Set 35 was rebased.","accounts_in_message":[],"_revision_number":36},{"id":"113e09a1fe95ec8dbfcc8f1acd4a52dd46d1915a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-19 16:40:30.000000000","message":"Uploaded patch set 37: Patch Set 36 was rebased.","accounts_in_message":[],"_revision_number":37},{"id":"81de1ac76b7b5b9ed590ef6f31706a4e74e6c9fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 19:20:47.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c87fedad1e547938fb834d47dd6043f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84d6aff645e3417283461aa80f520f51 : SUCCESS in 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65ede9ac6753400ead3dc6f1301c5607 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1480972aeb464f04a9a8c5b3010f9dae : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8e4f02bf1df4e008380857170b42cc7 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f06f99bed9c0461a9544ebdef56095a7 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcd507126c164e35a3b90b478594127c : SUCCESS in 12m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e11d05bd07e64218b987dfa9c5d15d3f : SUCCESS in 4m 42s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a4f2c7b6dbc34a66904c3ce36a361585 : SUCCESS in 5m 21s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4139d7238a8b437aa1ecc41713579e7f : SUCCESS in 9m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/850d0716ed3a47038a4ce908f4a6d474 : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/81b461652d4d40c5af40cdb88472cc08 : SUCCESS in 7m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b7deafd69bb5402bb768746e50692792 : SUCCESS in 9m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f8316bfa0ce54e32abaed82f61385454 : SUCCESS in 8m 29s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0683f3a2dae54a2291e3780b9b5e7aa6 : SUCCESS in 2h 33m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/815f33f5bf9140d8a1bd1e31934a2414 : SUCCESS in 1h 39m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/289456a7e0854985bb605949b6433bf4 : SUCCESS in 43m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/47854cb052814d00b4b50867e0095528 : SUCCESS in 1h 06m 06s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1a9c64a5010742e5b089f02a873ace68 : SUCCESS in 1h 13m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe314574ee8f4a129d8245303ecfc8f7 : SUCCESS in 38m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/60ec12cb7ad54b4f8e466bfb194e3dd9 : SUCCESS in 1h 41m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c9a8daa73e3c496c8ba3a94949b7f6ba : SUCCESS in 2h 04m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/12c762688ecf475d9103adc1cf7a5d5b : SUCCESS in 1h 43m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/96705c64c94e4dcfa63eafc0964eb36c : SUCCESS in 2h 08m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/17956ae520c74ebcadba2f0783368c36 : SUCCESS in 1h 55m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a8fec6a165f4816b27eabe2c1bcb602 : SUCCESS in 2h 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"812bcad6437a03224739a06556ba402e322aaf5b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-19 23:40:57.000000000","message":"Patch Set 37: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":37},{"id":"3d00b4773403de4d67a3d8e48c470349ffac703b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-22 07:07:00.000000000","message":"Patch Set 37:\n\n(2 comments)","accounts_in_message":[],"_revision_number":37},{"id":"563e972797ed8a00df104608f0fd66eec70ffd70","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-22 14:56:54.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"d83a4cea9e47dda2d91cb7b0addd5635e58f8b7e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-23 07:42:30.000000000","message":"Patch Set 37:\n\n(4 comments)","accounts_in_message":[],"_revision_number":37},{"id":"a7825a7fdbe48104de36a185ea37e7fe61e864cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-23 07:42:33.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"c8b35b8a28fd928d161b0a5b5db97a7caf425132","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 10:22:17.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d28d2d6e9db24618b7fae00a0e72d10e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/105511c9b121458bb6f7100dcfd87f18 : SUCCESS in 11m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbe98c3efba34b41b311e51e99cf4149 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7aa8524db0934597903f6e399db2ca71 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d99f60107f440898595bfee0a45bb86 : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c79f7c7119a840219717b238e0992ece : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab0a07e045e94130aac4ba41df73f5ce : SUCCESS in 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43fa4caf7afe4f0eae2becf3cee38827 : SUCCESS in 5m 25s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b46715bbece34ea8bfe8aeaf6ed37819 : SUCCESS in 6m 06s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/35cdfed34f3b46ee8d1c48004b7b987f : SUCCESS in 8m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d344bccbb3c64686a1eda55146c320cf : SUCCESS in 5m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/98c8acb2d70140f5a91e23975f271dd4 : SUCCESS in 9m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/faf6a4fbfdcb47baa9828362782a4f40 : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a466ba2f5d064564bd9b52598b7b4bbe : SUCCESS in 7m 17s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5799ea55dd9f463e82575dd8c3386717 : SUCCESS in 2h 34m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9293990e7e9b48d9ab50ffa256e7e85a : SUCCESS in 1h 52m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e3c41f414359484dbb2e57b92eec7a23 : SUCCESS in 48m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f39d8859c3b246688e83f1931df3cd03 : SUCCESS in 58m 38s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c7fce3b982204300a9f14c32cf501c09 : FAILURE in 1h 09m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/93d30434887c4a51b9d2ca380a1e5b87 : SUCCESS in 42m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/077ee7076e50411285757a21c257f698 : SUCCESS in 2h 00m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6ab1953ffdff46c7a3c2a282032ade54 : SUCCESS in 2h 12m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dc71e637a51743e5bec885abefbb61b0 : SUCCESS in 1h 51m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/76e9287de1a84dd0ace9c988ecbc2ac9 : SUCCESS in 1h 48m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a158535eaf9f480f8f95285b8887da5e : SUCCESS in 1h 31m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d3d9caf1e064e03b90a997c91806075 : SUCCESS in 2h 25m 49s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"2981b033894951be9d1de7c9da7e2f74021b2c71","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-24 00:16:53.000000000","message":"Patch Set 38: Code-Review-1\n\n(23 comments)","accounts_in_message":[],"_revision_number":38},{"id":"a34e2c71c4a77ef6b77beddcb2341765aa7f7723","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-24 00:19:09.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"1d4cd4df88b877a17a16d57c01a9b147bdfa40ae","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-24 12:31:16.000000000","message":"Patch Set 38:\n\n(8 comments)","accounts_in_message":[],"_revision_number":38},{"id":"3e6dbc3c0b1d0b4cbfe3570c17222e7ec4ffe399","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-24 19:55:54.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"63c6a4b18554d8f9f53835b698280c5b22e4a05e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-24 20:03:57.000000000","message":"Patch Set 38:\n\n(2 comments)","accounts_in_message":[],"_revision_number":38},{"id":"b238b216c78becc5345d8f3940aec51a9bbe60d9","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-08-25 10:25:34.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"9dab50418a17871958a9fd407ad848df429540dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-25 12:48:23.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"a6e4fdda6c45486c77963081f44a5c272c70d865","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-25 12:48:34.000000000","message":"Patch Set 38:\n\n(30 comments)","accounts_in_message":[],"_revision_number":38},{"id":"9fd39eec1d2076eb374fd39a7e8eb2c7de7ac852","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 15:32:18.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2af176a483644641b0fc20877418f332\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c82db5f1fde542fba45f8cd9bd0fdfef : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1455ca03213643d89c76daa4be58de24 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12229832bc7c4ca2a52ab6e167e14086 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0909219cd556440e9c307c34eae24cde : SUCCESS in 6m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b399f82026442a4ab6b7b924c08c3e6 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9a59f3ed6b54e5f962ff48fad5e0acd : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d01f4815e01c4220be5461e2f8fd0827 : SUCCESS in 6m 30s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a47059e8d12f4ab5a5197c61a346b4a4 : SUCCESS in 6m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8ce2e5e00ad5470081b2e9318751566b : SUCCESS in 8m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/32337bfd37c74387b34b91ffd52d5e17 : SUCCESS in 7m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3bb6b98bab1441faaf3ba3f3f9275433 : SUCCESS in 9m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d9957b0f59ae4508918e3ce7357d27f6 : SUCCESS in 8m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eca28236ad92425e89cffcc63835b04e : SUCCESS in 7m 49s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eed4a8db41ad47eaaae6082205d805ae : SUCCESS in 2h 36m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a663dfac02654ea982f7f15d6d06d578 : SUCCESS in 2h 08m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c3803de29b7d4312a479ecd37a608b73 : SUCCESS in 42m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4d9da39818c24d9cb0ed91cb7b36699a : SUCCESS in 1h 08m 46s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d861dca4472747f48e600ebb75333da9 : SUCCESS in 1h 04m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/20695767717648f7b6a2daa31d3372af : SUCCESS in 39m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e02d066b9fa9439e92f22ea76366705a : SUCCESS in 1h 54m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d23d5fbebbfc4c428d572d992fb35fe0 : SUCCESS in 2h 20m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/89b9f3f07d0b4fae93a993367d0ed6ea : SUCCESS in 1h 51m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ca70109ad1f247d6a414c99e30eb68c3 : SUCCESS in 2h 19m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d1cf47639462416ab7d13eafd01adbc7 : SUCCESS in 2h 05m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/67eed6ff5ab0489ea05aaa6926f711d2 : SUCCESS in 2h 27m 25s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"0a9455ef05c824cf4315c7fb3ed2e43ad712140b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-29 15:36:34.000000000","message":"Patch Set 39: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":39},{"id":"ab3cd77b7892cb87e0e21ee35a961e02240e1a3e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-31 15:00:00.000000000","message":"Patch Set 39: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"d6859e5ee260b1da5883c13a5577724e9ae4dde5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 15:00:46.000000000","message":"Patch Set 39: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":39},{"id":"6555097ebc679d75ae0cb081f5578b702c970a03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 17:44:04.000000000","message":"Patch Set 39: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ba5dbd2487144d681245c14a5169e69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2810885aa6341ff8904f384fb1b40b4 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa63a726607a47d8bf1aa91578acaa3d : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cba918a6c9c47b6ba7ac5886e5ca624 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6637f1f6ad13440586b4612009f49f67 : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91cb4915224447de85740017b6053b9c : SUCCESS in 5m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3756f32e40534ffebb382bfdd98c2e4b : SUCCESS in 5m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fd0a8fc9b72542ae9f9b6c8645c952a1 : SUCCESS in 7m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3bf0e5f27d544dcaaff95b0f6b0bd6a7 : SUCCESS in 2h 37m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2bf5241b6dd64d8b96fe9d06f9e43d81 : SUCCESS in 1h 41m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1bac004b620e44378f78dcc59d927926 : SUCCESS in 48m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a732a1b8041f4d2fa16bfeb9b0abcc58 : SUCCESS in 1h 05m 09s","accounts_in_message":[],"_revision_number":39},{"id":"1adfa863f8dc509324f7a4a12fbe68a5e0e90c50","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 17:44:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":39},{"id":"66a4203d1d13f76e75609fdcac5abf938d102156","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 17:45:28.000000000","message":"Patch Set 39:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28f28249f03f4a22ab887d671759a0ac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e9061fe0824bfd9aa26ede547c256d : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d3ccb87270d4d57a5813003ef700488 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":39}],"current_revision_number":39,"current_revision":"4b0d4c1b591e41d42fba87ce4a50a899abee6fcf","revisions":{"e258ccdcd3fc62f1b9aef7da34ed1cc1b0a2e24a":{"kind":"REWORK","_number":1,"created":"2019-06-14 14:13:52.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/02/665402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/1"}}},"commit":{"parents":[{"commit":"8ea5d937e3cc6702b06155a5c8e7ae65a7f49078","subject":"Allow multiple VIPs per LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ea5d937e3cc6702b06155a5c8e7ae65a7f49078"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"subject":"WIP: handle plugging member subnets on existing networks","message":"WIP: handle plugging member subnets on existing networks\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e258ccdcd3fc62f1b9aef7da34ed1cc1b0a2e24a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e258ccdcd3fc62f1b9aef7da34ed1cc1b0a2e24a"}]},"branch":"refs/heads/master"},"7adb4e5ff47a1d52748cba3d64080b257abab7c9":{"kind":"REWORK","_number":2,"created":"2019-06-15 03:22:16.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/02/665402/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/2"}}},"commit":{"parents":[{"commit":"536018bee7ec2070094826bfad6a072fd8ab0d59","subject":"Allow multiple VIPs per LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/536018bee7ec2070094826bfad6a072fd8ab0d59"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-15 03:22:05.000000000","tz":-420},"subject":"WIP: handle plugging member subnets on existing networks","message":"WIP: handle plugging member subnets on existing networks\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7adb4e5ff47a1d52748cba3d64080b257abab7c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7adb4e5ff47a1d52748cba3d64080b257abab7c9"}]},"branch":"refs/heads/master"},"7792b5f8e93b81b7869f68e7e1be0f76dcc38afd":{"kind":"REWORK","_number":3,"created":"2019-06-15 05:26:20.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/02/665402/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/3"}}},"commit":{"parents":[{"commit":"7209ba58ad14773b9ba66173a41fdf33816459c1","subject":"Allow multiple VIPs per LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7209ba58ad14773b9ba66173a41fdf33816459c1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-15 05:23:58.000000000","tz":-420},"subject":"WIP: handle plugging member subnets on existing networks","message":"WIP: handle plugging member subnets on existing networks\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7792b5f8e93b81b7869f68e7e1be0f76dcc38afd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7792b5f8e93b81b7869f68e7e1be0f76dcc38afd"}]},"branch":"refs/heads/master"},"c18d017d020cb2771696529e688ee0890b74efdf":{"kind":"REWORK","_number":4,"created":"2019-06-16 00:40:31.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/02/665402/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/4"}}},"commit":{"parents":[{"commit":"7209ba58ad14773b9ba66173a41fdf33816459c1","subject":"Allow multiple VIPs per LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7209ba58ad14773b9ba66173a41fdf33816459c1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-16 00:39:32.000000000","tz":-420},"subject":"WIP: fix plugging member subnets on existing networks","message":"WIP: fix plugging member subnets on existing networks\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c18d017d020cb2771696529e688ee0890b74efdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c18d017d020cb2771696529e688ee0890b74efdf"}]},"branch":"refs/heads/master"},"cab233194367671b1f8cbe4175ed8bce23a5cfab":{"kind":"REWORK","_number":5,"created":"2019-06-16 06:35:08.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/02/665402/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/5"}}},"commit":{"parents":[{"commit":"7209ba58ad14773b9ba66173a41fdf33816459c1","subject":"Allow multiple VIPs per LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7209ba58ad14773b9ba66173a41fdf33816459c1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-16 06:34:56.000000000","tz":-420},"subject":"WIP: fix plugging member subnets on existing networks","message":"WIP: fix plugging member subnets on existing networks\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cab233194367671b1f8cbe4175ed8bce23a5cfab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cab233194367671b1f8cbe4175ed8bce23a5cfab"}]},"branch":"refs/heads/master"},"bd1f057d29d2a7261777da5b3e6786a6c2968c4c":{"kind":"REWORK","_number":6,"created":"2019-06-16 10:55:12.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/02/665402/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/6"}}},"commit":{"parents":[{"commit":"7209ba58ad14773b9ba66173a41fdf33816459c1","subject":"Allow multiple VIPs per LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7209ba58ad14773b9ba66173a41fdf33816459c1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-16 10:55:00.000000000","tz":-420},"subject":"WIP: fix plugging member subnets on existing networks","message":"WIP: fix plugging member subnets on existing networks\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd1f057d29d2a7261777da5b3e6786a6c2968c4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd1f057d29d2a7261777da5b3e6786a6c2968c4c"}]},"branch":"refs/heads/master"},"80b4a13e25e543aa6f9fcc7a8182bdc6e85e091a":{"kind":"REWORK","_number":7,"created":"2019-06-18 04:10:49.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/02/665402/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/7"}}},"commit":{"parents":[{"commit":"710d6665ebd34b5125f7fff150f805a49ccaa08a","subject":"Allow multiple VIPs per LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/710d6665ebd34b5125f7fff150f805a49ccaa08a"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-18 04:04:29.000000000","tz":-420},"subject":"WIP: fix plugging member subnets on existing networks","message":"WIP: fix plugging member subnets on existing networks\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/80b4a13e25e543aa6f9fcc7a8182bdc6e85e091a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/80b4a13e25e543aa6f9fcc7a8182bdc6e85e091a"}]},"branch":"refs/heads/master"},"003a9bbee608613921c112c5eb49f0cb9b3620fa":{"kind":"REWORK","_number":8,"created":"2019-06-18 07:07:03.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/02/665402/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/8"}}},"commit":{"parents":[{"commit":"710d6665ebd34b5125f7fff150f805a49ccaa08a","subject":"Allow multiple VIPs per LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/710d6665ebd34b5125f7fff150f805a49ccaa08a"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-18 07:06:40.000000000","tz":-420},"subject":"WIP: fix plugging member subnets on existing networks","message":"WIP: fix plugging member subnets on existing networks\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/003a9bbee608613921c112c5eb49f0cb9b3620fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/003a9bbee608613921c112c5eb49f0cb9b3620fa"}]},"branch":"refs/heads/master"},"9e215fe6e89360bb82b4d91c1b746713d24e2f1b":{"kind":"REWORK","_number":9,"created":"2019-06-19 02:18:16.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/02/665402/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/9"}}},"commit":{"parents":[{"commit":"bfa62dbb5911797cb045eb02f79bb174dbc6c51a","subject":"Allow multiple VIPs per LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bfa62dbb5911797cb045eb02f79bb174dbc6c51a"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-19 02:17:55.000000000","tz":-420},"subject":"WIP: fix plugging member subnets on existing networks","message":"WIP: fix plugging member subnets on existing networks\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e215fe6e89360bb82b4d91c1b746713d24e2f1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e215fe6e89360bb82b4d91c1b746713d24e2f1b"}]},"branch":"refs/heads/master"},"0a04eacc1240f224a2aea0c53484f5292f16136f":{"kind":"REWORK","_number":10,"created":"2019-07-05 16:16:53.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/02/665402/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/10"}}},"commit":{"parents":[{"commit":"ff28ff6a207bfd51b9a2ee233504cf32d7175d52","subject":"Allow multiple VIPs per LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff28ff6a207bfd51b9a2ee233504cf32d7175d52"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-07-05 16:16:17.000000000","tz":-420},"subject":"WIP: fix plugging member subnets on existing networks","message":"WIP: fix plugging member subnets on existing networks\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a04eacc1240f224a2aea0c53484f5292f16136f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a04eacc1240f224a2aea0c53484f5292f16136f"}]},"branch":"refs/heads/master"},"2d56fe0a53e7fb6b05f790aeee4181fdc346832d":{"kind":"REWORK","_number":11,"created":"2019-08-22 20:08:57.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/02/665402/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/11"}}},"commit":{"parents":[{"commit":"398829108d12d750f75abd09742f47312bc68113","subject":"Allow multiple VIPs per LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/398829108d12d750f75abd09742f47312bc68113"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-22 20:07:35.000000000","tz":-420},"subject":"WIP: fix plugging member subnets on existing networks","message":"WIP: fix plugging member subnets on existing networks\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d56fe0a53e7fb6b05f790aeee4181fdc346832d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d56fe0a53e7fb6b05f790aeee4181fdc346832d"}]},"branch":"refs/heads/master"},"036be464d45d0043d9b1bdb83d34ba13a4571b1b":{"kind":"REWORK","_number":12,"created":"2019-08-22 20:55:47.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/02/665402/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/12"}}},"commit":{"parents":[{"commit":"34fc584fc9194c324cb755614b84b5b7a694766c","subject":"Allow multiple VIPs per LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34fc584fc9194c324cb755614b84b5b7a694766c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-22 20:54:00.000000000","tz":-420},"subject":"WIP: fix plugging member subnets on existing networks","message":"WIP: fix plugging member subnets on existing networks\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/036be464d45d0043d9b1bdb83d34ba13a4571b1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/036be464d45d0043d9b1bdb83d34ba13a4571b1b"}]},"branch":"refs/heads/master"},"c22cccf1b1f39aee7dd1c0c0138c0171932f6fd5":{"kind":"REWORK","_number":13,"created":"2019-09-17 00:27:24.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/02/665402/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/13"}}},"commit":{"parents":[{"commit":"2a8ad34c66b3c149c51e1d8ebc5437ac04471633","subject":"Allow multiple VIPs per LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a8ad34c66b3c149c51e1d8ebc5437ac04471633"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-09-17 00:27:09.000000000","tz":-420},"subject":"WIP: fix plugging member subnets on existing networks","message":"WIP: fix plugging member subnets on existing networks\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c22cccf1b1f39aee7dd1c0c0138c0171932f6fd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c22cccf1b1f39aee7dd1c0c0138c0171932f6fd5"}]},"branch":"refs/heads/master"},"a0b206307d7102862a5c2ef4d3f796f3b0803bc3":{"kind":"REWORK","_number":14,"created":"2019-10-21 23:54:06.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/02/665402/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/14"}}},"commit":{"parents":[{"commit":"92b0daf945206a03760c98b014d19e576e854faf","subject":"Allow multiple VIPs per LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92b0daf945206a03760c98b014d19e576e854faf"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-10-21 23:52:34.000000000","tz":-420},"subject":"WIP: fix plugging member subnets on existing networks","message":"WIP: fix plugging member subnets on existing networks\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a0b206307d7102862a5c2ef4d3f796f3b0803bc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a0b206307d7102862a5c2ef4d3f796f3b0803bc3"}]},"branch":"refs/heads/master"},"c880e0584a3696e423e200fba3053d1b87c055dc":{"kind":"REWORK","_number":15,"created":"2021-09-23 08:25:27.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/15"}}},"commit":{"parents":[{"commit":"b8ddbc94a1ff1c8dcb3414afd1d5f5269a09f9da","subject":"Merge \"Disable conntrack for TCP flows in the amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8ddbc94a1ff1c8dcb3414afd1d5f5269a09f9da"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-23 08:25:13.000000000","tz":120},"subject":"WIP: fix plugging member subnets on existing networks","message":"WIP: fix plugging member subnets on existing networks\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c880e0584a3696e423e200fba3053d1b87c055dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c880e0584a3696e423e200fba3053d1b87c055dc"}]},"branch":"refs/heads/master"},"c38208beb1e0fdc29ce48169fc86a3c21a07e072":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2021-09-23 12:58:36.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/16"}}},"commit":{"parents":[{"commit":"b8ddbc94a1ff1c8dcb3414afd1d5f5269a09f9da","subject":"Merge \"Disable conntrack for TCP flows in the amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8ddbc94a1ff1c8dcb3414afd1d5f5269a09f9da"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-23 12:58:11.000000000","tz":120},"subject":"WIP: fix plugging member subnets on existing networks","message":"WIP: fix plugging member subnets on existing networks\n\nDepends-On: https://review.opendev.org/810660\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c38208beb1e0fdc29ce48169fc86a3c21a07e072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c38208beb1e0fdc29ce48169fc86a3c21a07e072"}]},"branch":"refs/heads/master"},"7c23db9a6f2dae874222c44fa32c89bab028cbd3":{"kind":"REWORK","_number":17,"created":"2021-09-27 12:03:59.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/17"}}},"commit":{"parents":[{"commit":"b41b275b8314f79cdafe1575d84aface2b5861c5","subject":"Fix management network selection when calculating deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b41b275b8314f79cdafe1575d84aface2b5861c5"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-27 12:02:45.000000000","tz":120},"subject":"WIP: fix plugging member subnets on existing networks","message":"WIP: fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnest instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nStill missing: v1, release note, remove some debug, check compatility\nwith old amphora.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/810660\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c23db9a6f2dae874222c44fa32c89bab028cbd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c23db9a6f2dae874222c44fa32c89bab028cbd3"}]},"branch":"refs/heads/master"},"89f0e6828a4941717e2c0a6891aa1ba35656ff33":{"kind":"REWORK","_number":18,"created":"2021-10-04 19:38:30.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/18"}}},"commit":{"parents":[{"commit":"53ce187a67b0841e184de647c79b422222ccaf25","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/53ce187a67b0841e184de647c79b422222ccaf25"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-04 19:36:24.000000000","tz":120},"subject":"WIP: fix plugging member subnets on existing networks","message":"WIP: fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnest instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nStill missing: v1, release note, remove some debug, check compatility\nwith old amphora.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/810660\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89f0e6828a4941717e2c0a6891aa1ba35656ff33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89f0e6828a4941717e2c0a6891aa1ba35656ff33"}]},"branch":"refs/heads/master"},"d576b3b2651959870d1bdf99bf3a295e51907195":{"kind":"REWORK","_number":19,"created":"2021-10-05 07:16:34.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/19"}}},"commit":{"parents":[{"commit":"53ce187a67b0841e184de647c79b422222ccaf25","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/53ce187a67b0841e184de647c79b422222ccaf25"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-05 07:16:23.000000000","tz":120},"subject":"WIP: fix plugging member subnets on existing networks","message":"WIP: fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnest instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nStill missing: v1, release note, remove some debug, check compatility\nwith old amphora.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/810660\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d576b3b2651959870d1bdf99bf3a295e51907195"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d576b3b2651959870d1bdf99bf3a295e51907195"}]},"branch":"refs/heads/master"},"599c8dd33a89ca9ba1a2ac3e98f066cebebec707":{"kind":"REWORK","_number":20,"created":"2021-10-25 14:23:50.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/20"}}},"commit":{"parents":[{"commit":"fb8cff934724fa1dafce0d9d2c63251fb9e2889d","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb8cff934724fa1dafce0d9d2c63251fb9e2889d"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-25 14:18:52.000000000","tz":120},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnest instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/815313\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/599c8dd33a89ca9ba1a2ac3e98f066cebebec707"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/599c8dd33a89ca9ba1a2ac3e98f066cebebec707"}]},"branch":"refs/heads/master"},"79898a49b7220b7de3b1ff3002c8da71287632ed":{"kind":"REWORK","_number":21,"created":"2021-10-27 08:39:17.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/21"}}},"commit":{"parents":[{"commit":"1a25289e2ea0501670ccfa87510a78b4c13b7e0c","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a25289e2ea0501670ccfa87510a78b4c13b7e0c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-27 08:39:01.000000000","tz":120},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnest instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/815313\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79898a49b7220b7de3b1ff3002c8da71287632ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79898a49b7220b7de3b1ff3002c8da71287632ed"}]},"branch":"refs/heads/master"},"cb54f7c74a062428306ce52689422d926a4463ce":{"kind":"REWORK","_number":22,"created":"2022-02-22 16:41:34.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/02/665402/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/22"}}},"commit":{"parents":[{"commit":"0bd35fbf8b843f52929ccbf0e0ebf5bf4f13704e","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0bd35fbf8b843f52929ccbf0e0ebf5bf4f13704e"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2022-02-22 16:41:16.000000000","tz":60},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnest instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2005608\nTask: 30847\n\nDepends-On: https://review.opendev.org/815313\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb54f7c74a062428306ce52689422d926a4463ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb54f7c74a062428306ce52689422d926a4463ce"}]},"branch":"refs/heads/master"},"7654f2917cf7d9d932cdb3e92063cf7924bf7c1f":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2022-02-22 19:55:18.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/02/665402/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/23"}}},"commit":{"parents":[{"commit":"76db5f73ea35af92c858476ecde6b0b6549fde6b","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76db5f73ea35af92c858476ecde6b0b6549fde6b"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2022-02-22 19:50:41.000000000","tz":60},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnest instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2005608\nTask: 30847\n\nDepends-On: https://review.opendev.org/815313\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7654f2917cf7d9d932cdb3e92063cf7924bf7c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7654f2917cf7d9d932cdb3e92063cf7924bf7c1f"}]},"branch":"refs/heads/master"},"d639d72fda9f8061b2dbebacc10a3233c292615f":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2022-02-23 08:00:07.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/02/665402/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/24"}}},"commit":{"parents":[{"commit":"108460673516322e27354324ac108c3f3f3cdccd","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/108460673516322e27354324ac108c3f3f3cdccd"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2022-02-23 07:59:57.000000000","tz":60},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnest instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2005608\nTask: 30847\n\nDepends-On: https://review.opendev.org/815313\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d639d72fda9f8061b2dbebacc10a3233c292615f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d639d72fda9f8061b2dbebacc10a3233c292615f"}]},"branch":"refs/heads/master"},"9729a2ff5f1683c68b2736f3d103b4d610b547fd":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2022-02-23 12:23:41.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/02/665402/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/25"}}},"commit":{"parents":[{"commit":"a8a99ab75e8f9a0a4e855b0d34f012157a56a14c","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8a99ab75e8f9a0a4e855b0d34f012157a56a14c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2022-02-23 12:23:06.000000000","tz":60},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnest instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2005608\nTask: 30847\n\nDepends-On: https://review.opendev.org/815313\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9729a2ff5f1683c68b2736f3d103b4d610b547fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9729a2ff5f1683c68b2736f3d103b4d610b547fd"}]},"branch":"refs/heads/master"},"163d0b282c08f0635afa793133d1341b75493974":{"kind":"NO_CODE_CHANGE","_number":26,"created":"2022-03-21 09:16:03.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/26"}}},"commit":{"parents":[{"commit":"a8a99ab75e8f9a0a4e855b0d34f012157a56a14c","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8a99ab75e8f9a0a4e855b0d34f012157a56a14c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-03-21 09:00:24.000000000","tz":60},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnest instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2005608\nTask: 30847\nStory: 2004112\nTask: 27533\n\nDepends-On: https://review.opendev.org/815313\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/163d0b282c08f0635afa793133d1341b75493974"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/163d0b282c08f0635afa793133d1341b75493974"}]},"branch":"refs/heads/master"},"171e276355b039ca51affb9be40d1d81899316c7":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":27,"created":"2022-03-28 08:32:38.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/02/665402/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/27"}}},"commit":{"parents":[{"commit":"edbc4d7b34b026778347bdbc7e6564b33e94f0dd","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/edbc4d7b34b026778347bdbc7e6564b33e94f0dd"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2022-03-28 08:31:51.000000000","tz":120},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnets instead of networks.\nHandleNetworkDelta plugs subnets on existing ports or removes subnets\nfrom ports.\n\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2005608\nTask: 30847\n\nDepends-On: https://review.opendev.org/815313\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/171e276355b039ca51affb9be40d1d81899316c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/171e276355b039ca51affb9be40d1d81899316c7"}]},"branch":"refs/heads/master"},"c2e3d17adfafbd80c24a6e95faf2cb357c5def1b":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":28,"created":"2022-03-28 10:28:27.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/02/665402/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/28"}}},"commit":{"parents":[{"commit":"fd4a368ddc05c2591572365d27620e20a7575f55","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd4a368ddc05c2591572365d27620e20a7575f55"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2022-03-28 10:27:01.000000000","tz":120},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnest instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2005608\nTask: 30847\nStory: 2004112\nTask: 27533\n\nDepends-On: https://review.opendev.org/815313\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2e3d17adfafbd80c24a6e95faf2cb357c5def1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2e3d17adfafbd80c24a6e95faf2cb357c5def1b"}]},"branch":"refs/heads/master"},"501e8627ac78f4728d80b859fefcfb31e4e5f798":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":29,"created":"2022-03-28 12:29:55.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/29"}}},"commit":{"parents":[{"commit":"4aa20417b851010fba52eadab5d73f5e133a67d4","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4aa20417b851010fba52eadab5d73f5e133a67d4"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-03-28 12:29:35.000000000","tz":120},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnest instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2005608\nTask: 30847\nStory: 2004112\nTask: 27533\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/501e8627ac78f4728d80b859fefcfb31e4e5f798"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/501e8627ac78f4728d80b859fefcfb31e4e5f798"}]},"branch":"refs/heads/master"},"ce2464d18e39b98c6f35f32f99cef3c28830f14b":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2022-03-28 14:24:51.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/30"}}},"commit":{"parents":[{"commit":"8e4d8be6a64ec4ba7e30f6f4c0b5c804e498afdf","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8e4d8be6a64ec4ba7e30f6f4c0b5c804e498afdf"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-03-28 14:24:40.000000000","tz":120},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnest instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2005608\nTask: 30847\nStory: 2004112\nTask: 27533\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce2464d18e39b98c6f35f32f99cef3c28830f14b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce2464d18e39b98c6f35f32f99cef3c28830f14b"}]},"branch":"refs/heads/master"},"025ac361cb235992f522cf8994326828ddd1b7ac":{"kind":"REWORK","_number":31,"created":"2022-04-15 11:09:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/31"}}},"commit":{"parents":[{"commit":"8e4d8be6a64ec4ba7e30f6f4c0b5c804e498afdf","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8e4d8be6a64ec4ba7e30f6f4c0b5c804e498afdf"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-15 11:04:41.000000000","tz":120},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnets instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2005608\nTask: 30847\nStory: 2004112\nTask: 27533\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/025ac361cb235992f522cf8994326828ddd1b7ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/025ac361cb235992f522cf8994326828ddd1b7ac"}]},"branch":"refs/heads/master"},"0c112ecde3006a3f743cb4851ea18016488b21a5":{"kind":"REWORK","_number":32,"created":"2022-04-29 13:04:23.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/32"}}},"commit":{"parents":[{"commit":"8e4d8be6a64ec4ba7e30f6f4c0b5c804e498afdf","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8e4d8be6a64ec4ba7e30f6f4c0b5c804e498afdf"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-29 13:04:15.000000000","tz":120},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnets instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2005608\nTask: 30847\nStory: 2004112\nTask: 27533\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c112ecde3006a3f743cb4851ea18016488b21a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c112ecde3006a3f743cb4851ea18016488b21a5"}]},"branch":"refs/heads/master"},"715bd100cdffbed0e9f5211ef558c6d8763dd2f5":{"kind":"REWORK","_number":33,"created":"2022-05-06 13:25:34.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/33"}}},"commit":{"parents":[{"commit":"8e4d8be6a64ec4ba7e30f6f4c0b5c804e498afdf","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8e4d8be6a64ec4ba7e30f6f4c0b5c804e498afdf"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-05-06 13:25:24.000000000","tz":120},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnets instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2005608\nTask: 30847\nStory: 2004112\nTask: 27533\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/715bd100cdffbed0e9f5211ef558c6d8763dd2f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/715bd100cdffbed0e9f5211ef558c6d8763dd2f5"}]},"branch":"refs/heads/master"},"5afea696b796ff122346ebb4e5d149dcad09fd82":{"kind":"REWORK","_number":34,"created":"2022-06-21 07:13:40.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/34"}}},"commit":{"parents":[{"commit":"f59355b3963695b7746fdce7e48333f0fb750545","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f59355b3963695b7746fdce7e48333f0fb750545"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-21 06:55:05.000000000","tz":120},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnets instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2005608\nTask: 30847\nStory: 2004112\nTask: 27533\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5afea696b796ff122346ebb4e5d149dcad09fd82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5afea696b796ff122346ebb4e5d149dcad09fd82"}]},"branch":"refs/heads/master"},"4f0f183a71734894381caa7606db14daee47f956":{"kind":"REWORK","_number":35,"created":"2022-08-11 13:49:07.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/35"}}},"commit":{"parents":[{"commit":"dffb3b3fee66f47a2de317731f3fc2964585218c","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dffb3b3fee66f47a2de317731f3fc2964585218c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-11 13:48:56.000000000","tz":120},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnets instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2005608\nTask: 30847\nStory: 2004112\nTask: 27533\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f0f183a71734894381caa7606db14daee47f956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f0f183a71734894381caa7606db14daee47f956"}]},"branch":"refs/heads/master"},"d435dc440720293cad0d67104e1acd4961cc8f64":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2022-08-19 15:06:25.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/36"}}},"commit":{"parents":[{"commit":"58630d947fd126ceaaf54237498a6d2d15f13e48","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58630d947fd126ceaaf54237498a6d2d15f13e48"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-19 14:55:40.000000000","tz":120},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnets instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2005608\nTask: 30847\nStory: 2004112\nTask: 27533\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d435dc440720293cad0d67104e1acd4961cc8f64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d435dc440720293cad0d67104e1acd4961cc8f64"}]},"branch":"refs/heads/master"},"e11a2eecde57b0ea30dfe449f3a7e941ecf93df9":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2022-08-19 16:40:30.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/37"}}},"commit":{"parents":[{"commit":"7118e8d56f76a99f19fd0696be3f09b6fb258cb2","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7118e8d56f76a99f19fd0696be3f09b6fb258cb2"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-19 16:38:42.000000000","tz":120},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnets instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2005608\nTask: 30847\nStory: 2004112\nTask: 27533\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e11a2eecde57b0ea30dfe449f3a7e941ecf93df9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e11a2eecde57b0ea30dfe449f3a7e941ecf93df9"}]},"branch":"refs/heads/master"},"1ce8218d426e4fb894480cbdb891aa3e51026fbf":{"kind":"REWORK","_number":38,"created":"2022-08-23 07:42:33.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/38"}}},"commit":{"parents":[{"commit":"7118e8d56f76a99f19fd0696be3f09b6fb258cb2","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7118e8d56f76a99f19fd0696be3f09b6fb258cb2"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-23 07:41:58.000000000","tz":120},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnets instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2004112\nTask: 27533\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ce8218d426e4fb894480cbdb891aa3e51026fbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ce8218d426e4fb894480cbdb891aa3e51026fbf"}]},"branch":"refs/heads/master"},"4b0d4c1b591e41d42fba87ce4a50a899abee6fcf":{"kind":"REWORK","_number":39,"created":"2022-08-25 12:48:23.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/665402/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/665402/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/665402/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/665402/39"}}},"commit":{"parents":[{"commit":"7118e8d56f76a99f19fd0696be3f09b6fb258cb2","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7118e8d56f76a99f19fd0696be3f09b6fb258cb2"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-25 12:31:46.000000000","tz":120},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnets instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2004112\nTask: 27533\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b0d4c1b591e41d42fba87ce4a50a899abee6fcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b0d4c1b591e41d42fba87ce4a50a899abee6fcf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
