)]}'
{"id":"openstack%2Foctavia~665405","triplet_id":"openstack%2Foctavia~stable%2Frocky~Ib3d5eb1f27ec1f10ed0f73b342927b78f1a237c7","project":"openstack/octavia","branch":"stable/rocky","topic":"2003733-stable/rocky","hashtags":[],"change_id":"Ib3d5eb1f27ec1f10ed0f73b342927b78f1a237c7","subject":"Fix for utils LB DM transformation function","status":"MERGED","created":"2019-06-14 14:32:19.000000000","updated":"2019-07-26 23:35:38.000000000","submitted":"2019-07-26 23:35:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665405-1564184138154-c60208eb","meta_rev_id":"4ff41fc199e8a33b4857546634292f86c875fe06","_number":665405,"virtual_id_number":665405,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-07-26 23:35:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8446,"name":"Evgeny Fedoruk","email":"evgenyf@radware.com","username":"evgenyfedoruk"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-07-26 21:36:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-07-23 21:56:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8446,"name":"Evgeny Fedoruk","email":"evgenyf@radware.com","username":"evgenyfedoruk"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-07-26 21:36:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8446,"name":"Evgeny Fedoruk","email":"evgenyf@radware.com","username":"evgenyfedoruk"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-14 14:32:19.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":8446,"name":"Evgeny Fedoruk","email":"evgenyf@radware.com","username":"evgenyfedoruk"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8446,"name":"Evgeny Fedoruk","email":"evgenyf@radware.com","username":"evgenyfedoruk"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-14 14:32:19.000000000","updated_by":{"_account_id":8446,"name":"Evgeny Fedoruk","email":"evgenyf@radware.com","username":"evgenyfedoruk"},"reviewer":{"_account_id":8446,"name":"Evgeny Fedoruk","email":"evgenyf@radware.com","username":"evgenyfedoruk"},"state":"REVIEWER"},{"updated":"2019-07-23 21:56:05.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-07-26 21:36:41.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-07-26 23:35:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"454cd802d4ee97b0d64c6398647b3e5124bed60b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-14 14:32:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"84b1e9caeb009412c167144422b12f7564c14ad5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 19:11:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/05/665405/1/check/openstack-tox-lower-constraints/999d876/ : SUCCESS in 8m 58s\n- openstack-tox-pep8 http://logs.openstack.org/05/665405/1/check/openstack-tox-pep8/882401c/ : SUCCESS in 8m 27s\n- openstack-tox-py27 http://logs.openstack.org/05/665405/1/check/openstack-tox-py27/e096fb8/ : SUCCESS in 6m 56s\n- openstack-tox-py35 http://logs.openstack.org/05/665405/1/check/openstack-tox-py35/be2e9bc/ : SUCCESS in 6m 26s\n- openstack-tox-docs http://logs.openstack.org/05/665405/1/check/openstack-tox-docs/a2695f9/html/ : SUCCESS in 6m 43s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/05/665405/1/check/neutron-lbaasv2-dsvm-api/618d4c9/ : SUCCESS in 1h 15m 09s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/05/665405/1/check/neutron-lbaasv2-dsvm-py3x-api/ff2e48a/ : SUCCESS in 1h 23m 09s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/05/665405/1/check/neutron-lbaasv2-dsvm-scenario/8b1a426/ : SUCCESS in 1h 38m 39s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/05/665405/1/check/neutron-lbaasv2-dsvm-py3x-scenario/6e33a3d/ : SUCCESS in 1h 29m 20s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/05/665405/1/check/openstack-tox-functional/c2dc0f9/ : SUCCESS in 8m 08s\n- openstack-tox-functional-py35 http://logs.openstack.org/05/665405/1/check/openstack-tox-functional-py35/d11b38a/ : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 http://logs.openstack.org/05/665405/1/check/openstack-tox-functional-py36/3d68a42/ : SUCCESS in 7m 31s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/05/665405/1/check/octavia-v1-dsvm-scenario/ecd22b3/ : SUCCESS in 58m 13s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/05/665405/1/check/octavia-v1-dsvm-py3x-scenario/722e31a/ : SUCCESS in 1h 02m 17s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/05/665405/1/check/octavia-v1-dsvm-scenario-multinode/cd51ebc/ : FAILURE in 1h 37m 54s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/05/665405/1/check/octavia-v1-dsvm-py3x-scenario-multinode/77537e0/ : FAILURE in 1h 28m 18s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/05/665405/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/66b82e2/ : FAILURE in 54m 48s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/05/665405/1/check/octavia-v2-dsvm-noop-api/547b2c1/ : SUCCESS in 43m 53s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/05/665405/1/check/octavia-v2-dsvm-noop-py35-api/549eeff/ : SUCCESS in 47m 32s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/05/665405/1/check/octavia-v2-dsvm-scenario/307d1db/ : SUCCESS in 1h 43m 34s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/05/665405/1/check/octavia-v2-dsvm-py35-scenario/4623875/ : SUCCESS in 1h 44m 57s (non-voting)\n- octavia-grenade http://logs.openstack.org/05/665405/1/check/octavia-grenade/27e6d10/ : SUCCESS in 1h 10m 58s\n- openstack-tox-cover http://logs.openstack.org/05/665405/1/check/openstack-tox-cover/1590e6e/cover/ : SUCCESS in 9m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e6c7792cbad6b7d3c34839f324de34c573922990","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-23 21:56:05.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"537f41920b732eb8a63cb4781e1d599b4c308bb0","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-26 21:36:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a15e9bf807108bd177b4c482d3213a2d361b5d95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 21:36:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d661d319efc133880d64f4d06741fa2318f6f69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 23:35:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/05/665405/1/gate/openstack-tox-lower-constraints/29e98a7/ : SUCCESS in 8m 34s\n- openstack-tox-pep8 http://logs.openstack.org/05/665405/1/gate/openstack-tox-pep8/d76c37d/ : SUCCESS in 6m 54s\n- openstack-tox-py27 http://logs.openstack.org/05/665405/1/gate/openstack-tox-py27/021acfd/ : SUCCESS in 5m 36s\n- openstack-tox-py35 http://logs.openstack.org/05/665405/1/gate/openstack-tox-py35/fea2e6f/ : SUCCESS in 7m 01s\n- openstack-tox-docs http://logs.openstack.org/05/665405/1/gate/openstack-tox-docs/7c071e2/html/ : SUCCESS in 6m 16s\n- openstack-tox-functional http://logs.openstack.org/05/665405/1/gate/openstack-tox-functional/1058b28/ : SUCCESS in 5m 36s\n- openstack-tox-functional-py35 http://logs.openstack.org/05/665405/1/gate/openstack-tox-functional-py35/32ce042/ : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 http://logs.openstack.org/05/665405/1/gate/openstack-tox-functional-py36/82ccd60/ : SUCCESS in 6m 09s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/05/665405/1/gate/octavia-v1-dsvm-scenario/9448304/ : SUCCESS in 1h 00m 33s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/05/665405/1/gate/octavia-v1-dsvm-py3x-scenario/7aaf716/ : SUCCESS in 59m 55s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/05/665405/1/gate/octavia-v1-dsvm-scenario-multinode/7e3242a/ : FAILURE in 1h 17m 32s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/05/665405/1/gate/octavia-v1-dsvm-py3x-scenario-multinode/5fb48d5/ : FAILURE in 1h 25m 19s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/05/665405/1/gate/neutron-lbaasv2-dsvm-py3x-api/02ee0cc/ : SUCCESS in 1h 18m 50s\n- octavia-grenade http://logs.openstack.org/05/665405/1/gate/octavia-grenade/84146f9/ : SUCCESS in 1h 12m 13s\n- openstack-tox-cover http://logs.openstack.org/05/665405/1/gate/openstack-tox-cover/8a34607/cover/ : SUCCESS in 7m 56s","accounts_in_message":[],"_revision_number":1},{"id":"3d5432e1095196efbb1bcd5ed6a1ea7d861551a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 23:35:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e865435de3004d4d2434d2302af4eeabd1b5c34","revisions":{"7e865435de3004d4d2434d2302af4eeabd1b5c34":{"kind":"REWORK","_number":1,"created":"2019-06-14 14:32:19.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/05/665405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/665405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/665405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/665405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/665405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/665405/1"}}},"commit":{"parents":[{"commit":"b12f0d9d6ce5e873b8e9b501e50c314ffb02e128","subject":"Merge \"Treat null admin_state_up as False\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b12f0d9d6ce5e873b8e9b501e50c314ffb02e128"}]}],"author":{"name":"Evgeny Fedoruk","email":"evgenyf@radware.com","date":"2018-09-26 09:25:00.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-14 14:32:19.000000000","tz":0},"subject":"Fix for utils LB DM transformation function","message":"Fix for utils LB DM transformation function\n\nFixing utils function for common LB data model transformation\nto provider driver data model which does not consider\na Vip data model object.\nUnit test which didn\u0027t cover this bug was modified.\n\nChange-Id: Ib3d5eb1f27ec1f10ed0f73b342927b78f1a237c7\nImplements Story: 2003733\n(cherry picked from commit a9b99d2b41173a112f8bed6ae08ea845daa29f37)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e865435de3004d4d2434d2302af4eeabd1b5c34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e865435de3004d4d2434d2302af4eeabd1b5c34"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
