)]}'
{"id":"openstack%2Foctavia~665433","triplet_id":"openstack%2Foctavia~master~I20572505e448f9b7ff6314b5167cc4c519c445fb","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I20572505e448f9b7ff6314b5167cc4c519c445fb","subject":"Fix allocate_and_associate DB deadlock","status":"MERGED","created":"2019-06-14 16:41:10.000000000","updated":"2019-06-16 19:32:15.000000000","submitted":"2019-06-15 02:08:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665433-1560564509207-ad3aa802","meta_rev_id":"d69243e1e1d1807f3cff94261f2efa5d03aaa7a5","_number":665433,"virtual_id_number":665433,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-06-15 02:08:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-06-15 00:11:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-06-14 16:49:30.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-06-15 00:11:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-06-14 16:49:30.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-06-14 16:41:21.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-14 16:49:30.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-06-15 00:11:34.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-06-15 02:08:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c8cd038a20b4267affa38388803b42096cd233d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-14 16:41:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc4e380f88c9625d03a778513b954231ea82b62b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-14 16:41:21.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"6f82c410069dc78f8f8d647436df14227fb12bb6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-14 16:49:30.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e01db085bfba5c95bccaa2bf77fae95a4bb9a2f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 20:19:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/33/665433/1/check/openstack-tox-cover/c4b8bf5/cover/ : SUCCESS in 8m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/33/665433/1/check/openstack-tox-lower-constraints/a8dd7b3/ : SUCCESS in 6m 52s\n- openstack-tox-pep8 http://logs.openstack.org/33/665433/1/check/openstack-tox-pep8/21fd3e5/ : SUCCESS in 6m 30s\n- openstack-tox-py27 http://logs.openstack.org/33/665433/1/check/openstack-tox-py27/7759daf/ : SUCCESS in 5m 25s\n- openstack-tox-py36 http://logs.openstack.org/33/665433/1/check/openstack-tox-py36/dadf6f1/ : SUCCESS in 5m 40s\n- openstack-tox-py37 http://logs.openstack.org/33/665433/1/check/openstack-tox-py37/1ba9769/ : SUCCESS in 6m 01s\n- openstack-tox-docs http://logs.openstack.org/33/665433/1/check/openstack-tox-docs/a52745a/html/ : SUCCESS in 7m 23s\n- build-openstack-releasenotes http://logs.openstack.org/33/665433/1/check/build-openstack-releasenotes/38c7158/html/ : SUCCESS in 8m 45s\n- openstack-tox-functional http://logs.openstack.org/33/665433/1/check/openstack-tox-functional/6d43b61/ : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 http://logs.openstack.org/33/665433/1/check/openstack-tox-functional-py36/a15a41c/ : SUCCESS in 6m 37s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/33/665433/1/check/octavia-v2-dsvm-noop-api/b915817/ : SUCCESS in 45m 31s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/33/665433/1/check/octavia-v2-dsvm-noop-py2-api/7f7b8c1/ : SUCCESS in 42m 35s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/33/665433/1/check/octavia-v2-dsvm-scenario/3190a0b/ : SUCCESS in 1h 32m 38s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/33/665433/1/check/octavia-v2-dsvm-py2-scenario/9e97e2b/ : SUCCESS in 1h 37m 55s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/33/665433/1/check/octavia-v2-dsvm-py2-scenario-centos-7/c496565/ : FAILURE in 1h 55m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/33/665433/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/95fa213/ : SUCCESS in 1h 43m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/33/665433/1/check/octavia-v2-dsvm-scenario-two-node/f6f80ed/ : SUCCESS in 1h 24m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/33/665433/1/check/octavia-v2-dsvm-py2-scenario-two-node/240ae74/ : SUCCESS in 1h 57m 38s (non-voting)\n- octavia-grenade http://logs.openstack.org/33/665433/1/check/octavia-grenade/e021938/ : SUCCESS in 1h 30m 59s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/33/665433/1/check/octavia-v2-dsvm-tls-barbican/2dc255d/ : SUCCESS in 1h 07m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/33/665433/1/check/octavia-v2-dsvm-scenario-amphora-v2/f4e328f/ : SUCCESS in 1h 23m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5378aaff4ee86bae2c9041dead577c36bd84b66","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-15 00:11:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b54c7601d8f917ee116dcf3b4b28663ab8cea55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 00:11:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb4fb666d35b182b8992c9ff14be19f793418ce2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 02:08:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/33/665433/1/gate/openstack-tox-lower-constraints/4e2637c/ : SUCCESS in 7m 12s\n- openstack-tox-pep8 http://logs.openstack.org/33/665433/1/gate/openstack-tox-pep8/ddc533d/ : SUCCESS in 6m 00s\n- openstack-tox-py27 http://logs.openstack.org/33/665433/1/gate/openstack-tox-py27/0d2e2b9/ : SUCCESS in 5m 45s\n- openstack-tox-py36 http://logs.openstack.org/33/665433/1/gate/openstack-tox-py36/ae6f46e/ : SUCCESS in 7m 45s\n- openstack-tox-py37 http://logs.openstack.org/33/665433/1/gate/openstack-tox-py37/ef036d7/ : SUCCESS in 5m 43s\n- openstack-tox-docs http://logs.openstack.org/33/665433/1/gate/openstack-tox-docs/70bba02/html/ : SUCCESS in 6m 45s\n- build-openstack-releasenotes http://logs.openstack.org/33/665433/1/gate/build-openstack-releasenotes/c156e76/html/ : SUCCESS in 5m 02s\n- openstack-tox-functional http://logs.openstack.org/33/665433/1/gate/openstack-tox-functional/089f164/ : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 http://logs.openstack.org/33/665433/1/gate/openstack-tox-functional-py36/733b240/ : SUCCESS in 6m 23s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/33/665433/1/gate/octavia-v2-dsvm-noop-api/336f715/ : SUCCESS in 42m 34s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/33/665433/1/gate/octavia-v2-dsvm-noop-py2-api/d4c53dc/ : SUCCESS in 40m 11s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/33/665433/1/gate/octavia-v2-dsvm-scenario/d23258d/ : SUCCESS in 1h 29m 31s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/33/665433/1/gate/octavia-v2-dsvm-py2-scenario/85023d4/ : SUCCESS in 1h 32m 51s\n- octavia-grenade http://logs.openstack.org/33/665433/1/gate/octavia-grenade/1269151/ : SUCCESS in 1h 30m 50s","accounts_in_message":[],"_revision_number":1},{"id":"270dee19da1503c21936163d1d31f622e834f947","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 02:08:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d69243e1e1d1807f3cff94261f2efa5d03aaa7a5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-16 19:32:15.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 32afefd6e4441bbac7a6442013714e5a7433c99a","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0eaab450c12e33f277c31098a6792421d3f5806c","revisions":{"0eaab450c12e33f277c31098a6792421d3f5806c":{"kind":"REWORK","_number":1,"created":"2019-06-14 16:41:10.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/33/665433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/665433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/665433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/665433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/665433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/665433/1"}}},"commit":{"parents":[{"commit":"620c19804cd518af807839c877a01fe56db7462d","subject":"Merge \"Add RHEL 8 amphora support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/620c19804cd518af807839c877a01fe56db7462d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-14 16:39:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-14 16:41:09.000000000","tz":-420},"subject":"Fix allocate_and_associate DB deadlock","message":"Fix allocate_and_associate DB deadlock\n\nIn testing we saw a DB deadlock in allocate_and_associate.\nThis patch corrects that with a DB retry wrapper.\n\nChange-Id: I20572505e448f9b7ff6314b5167cc4c519c445fb\nStory: 2005893\nTask: 33751\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0eaab450c12e33f277c31098a6792421d3f5806c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0eaab450c12e33f277c31098a6792421d3f5806c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
