)]}'
{"id":"openstack%2Foctavia~665557","triplet_id":"openstack%2Foctavia~stable%2Frocky~I20572505e448f9b7ff6314b5167cc4c519c445fb","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"I20572505e448f9b7ff6314b5167cc4c519c445fb","subject":"Fix allocate_and_associate DB deadlock","status":"MERGED","created":"2019-06-16 19:32:25.000000000","updated":"2019-06-21 13:10:31.000000000","submitted":"2019-06-21 13:10:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665557-1561122631058-f9ed76da","meta_rev_id":"fb6c5d1b7981cc14ca1a6cb2b42e66bde9235660","_number":665557,"virtual_id_number":665557,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2019-06-18 20:36:09.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-06-21 13:10:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-06-18 09:00:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-06-18 07:23:51.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-06-18 09:00:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 07:23:51.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-06-18 09:00:39.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-06-21 13:10:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5bbcbe97f14d9070df47e2e60febab0fd81a503e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-16 19:32:25.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"36e8bd753a522535ddbe4259d45250fd6d1d6ca4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-16 19:32:36.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit a96b00b4dfdd6ac3643149149ed5141a9fbc68cd","accounts_in_message":[],"_revision_number":1},{"id":"a32d0ed39c59026f7acd7d2c42e04b6500d1a285","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-16 21:13:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/57/665557/1/check/openstack-tox-lower-constraints/ac71667/ : SUCCESS in 8m 12s\n- openstack-tox-pep8 http://logs.openstack.org/57/665557/1/check/openstack-tox-pep8/a0e25ee/ : SUCCESS in 9m 07s\n- openstack-tox-py27 http://logs.openstack.org/57/665557/1/check/openstack-tox-py27/53cdfb9/ : SUCCESS in 6m 20s\n- openstack-tox-py35 http://logs.openstack.org/57/665557/1/check/openstack-tox-py35/a118954/ : SUCCESS in 6m 36s\n- openstack-tox-docs http://logs.openstack.org/57/665557/1/check/openstack-tox-docs/b26fbdb/html/ : SUCCESS in 6m 14s\n- build-openstack-releasenotes http://logs.openstack.org/57/665557/1/check/build-openstack-releasenotes/b2bb49d/html/ : SUCCESS in 5m 15s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-api/97c0ffb/ : SUCCESS in 1h 13m 38s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-py3x-api/d403ca8/ : SUCCESS in 1h 17m 09s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-scenario/abdefd9/ : SUCCESS in 1h 21m 20s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-py3x-scenario/29241b9/ : SUCCESS in 1h 28m 21s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/57/665557/1/check/openstack-tox-functional/b91d2a0/ : SUCCESS in 6m 21s\n- openstack-tox-functional-py35 http://logs.openstack.org/57/665557/1/check/openstack-tox-functional-py35/9efd843/ : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 http://logs.openstack.org/57/665557/1/check/openstack-tox-functional-py36/d4f8b4c/ : SUCCESS in 6m 55s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-scenario/4a8a1a8/ : SUCCESS in 58m 22s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-py3x-scenario/10addb5/ : SUCCESS in 59m 11s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-scenario-multinode/e2758d5/ : FAILURE in 1h 25m 03s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-py3x-scenario-multinode/0f4c249/ : FAILURE in 1h 20m 28s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/a8ad5dd/ : SUCCESS in 1h 16m 46s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-noop-api/642ac58/ : SUCCESS in 46m 31s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-noop-py35-api/f7aa318/ : SUCCESS in 44m 31s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-scenario/765cdfe/ : SUCCESS in 1h 34m 14s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-py35-scenario/305d1ee/ : SUCCESS in 1h 34m 52s (non-voting)\n- octavia-grenade http://logs.openstack.org/57/665557/1/check/octavia-grenade/bdd2931/ : SUCCESS in 1h 09m 01s\n- openstack-tox-cover http://logs.openstack.org/57/665557/1/check/openstack-tox-cover/7144a89/cover/ : SUCCESS in 8m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ca4b34f58cc303802296c5232f95749a81325691","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-18 07:23:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab93dbf57164c030eef60080f84ef007493e7e28","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-18 09:00:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9701df5f9337fb9e40e515f6c07423fb7108d8cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:00:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"afe77f8022641f57217a082c9cea2ce2d6133bd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 14:27:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/57/665557/1/gate/openstack-tox-lower-constraints/49f0eca/ : SUCCESS in 9m 01s\n- openstack-tox-pep8 http://logs.openstack.org/57/665557/1/gate/openstack-tox-pep8/0c9caff/ : SUCCESS in 8m 03s\n- openstack-tox-py27 http://logs.openstack.org/57/665557/1/gate/openstack-tox-py27/c43eb63/ : SUCCESS in 6m 34s\n- openstack-tox-py35 http://logs.openstack.org/57/665557/1/gate/openstack-tox-py35/27b9990/ : SUCCESS in 8m 43s\n- openstack-tox-docs http://logs.openstack.org/57/665557/1/gate/openstack-tox-docs/2493934/html/ : SUCCESS in 6m 43s\n- build-openstack-releasenotes http://logs.openstack.org/57/665557/1/gate/build-openstack-releasenotes/732f097/html/ : SUCCESS in 5m 52s\n- openstack-tox-functional http://logs.openstack.org/57/665557/1/gate/openstack-tox-functional/cf30e5e/ : SUCCESS in 6m 51s\n- openstack-tox-functional-py35 http://logs.openstack.org/57/665557/1/gate/openstack-tox-functional-py35/32eb7a8/ : SUCCESS in 7m 19s\n- openstack-tox-functional-py36 http://logs.openstack.org/57/665557/1/gate/openstack-tox-functional-py36/5186f33/ : SUCCESS in 7m 48s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/57/665557/1/gate/octavia-v1-dsvm-scenario/a52b55a/ : FAILURE in 53m 50s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/57/665557/1/gate/octavia-v1-dsvm-py3x-scenario/3ed1706/ : FAILURE in 54m 33s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/57/665557/1/gate/octavia-v1-dsvm-scenario-multinode/4c6f23e/ : FAILURE in 1h 25m 53s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/57/665557/1/gate/octavia-v1-dsvm-py3x-scenario-multinode/caa1a3c/ : FAILURE in 1h 29m 14s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/57/665557/1/gate/neutron-lbaasv2-dsvm-py3x-api/771aecd/ : SUCCESS in 1h 18m 33s\n- octavia-grenade http://logs.openstack.org/57/665557/1/gate/octavia-grenade/61cecfc/ : SUCCESS in 1h 09m 39s\n- openstack-tox-cover http://logs.openstack.org/57/665557/1/gate/openstack-tox-cover/ef1769f/cover/ : SUCCESS in 9m 35s","accounts_in_message":[],"_revision_number":1},{"id":"7d0e5c55460159fcaa6b0362bf05bbea7b10ad03","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-18 15:24:27.000000000","message":"Patch Set 1:\n\nrecheck\nnova failures","accounts_in_message":[],"_revision_number":1},{"id":"a59ad156189758ed40cfbd905138dc919f92b60b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 17:20:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/57/665557/1/check/openstack-tox-lower-constraints/a61c5bc/ : SUCCESS in 8m 29s\n- openstack-tox-pep8 http://logs.openstack.org/57/665557/1/check/openstack-tox-pep8/c15f64c/ : SUCCESS in 8m 47s\n- openstack-tox-py27 http://logs.openstack.org/57/665557/1/check/openstack-tox-py27/e4c7296/ : SUCCESS in 7m 41s\n- openstack-tox-py35 http://logs.openstack.org/57/665557/1/check/openstack-tox-py35/e9d9b03/ : SUCCESS in 7m 41s\n- openstack-tox-docs http://logs.openstack.org/57/665557/1/check/openstack-tox-docs/df68a82/html/ : SUCCESS in 6m 33s\n- build-openstack-releasenotes http://logs.openstack.org/57/665557/1/check/build-openstack-releasenotes/81d8194/html/ : SUCCESS in 5m 24s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-api/736c170/ : SUCCESS in 1h 15m 24s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-py3x-api/3080e3b/ : SUCCESS in 1h 20m 48s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-scenario/c24f0d9/ : FAILURE in 1h 01m 48s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-py3x-scenario/038ee6a/ : FAILURE in 1h 04m 23s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/57/665557/1/check/openstack-tox-functional/5d9c23e/ : SUCCESS in 8m 50s\n- openstack-tox-functional-py35 http://logs.openstack.org/57/665557/1/check/openstack-tox-functional-py35/777fbfa/ : SUCCESS in 7m 39s\n- openstack-tox-functional-py36 http://logs.openstack.org/57/665557/1/check/openstack-tox-functional-py36/c4166a4/ : SUCCESS in 7m 29s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-scenario/513e56b/ : FAILURE in 51m 14s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-py3x-scenario/d461de2/ : SUCCESS in 1h 11m 36s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-scenario-multinode/b03bf81/ : FAILURE in 1h 28m 05s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-py3x-scenario-multinode/955b168/ : FAILURE in 1h 26m 10s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/664f38a/ : SUCCESS in 1h 08m 29s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-noop-api/422dbfa/ : SUCCESS in 46m 12s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-noop-py35-api/7931c3b/ : SUCCESS in 47m 52s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-scenario/70be0fd/ : SUCCESS in 1h 21m 25s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-py35-scenario/3981750/ : SUCCESS in 1h 31m 49s (non-voting)\n- octavia-grenade http://logs.openstack.org/57/665557/1/check/octavia-grenade/0d00d5c/ : SUCCESS in 1h 08m 45s\n- openstack-tox-cover http://logs.openstack.org/57/665557/1/check/openstack-tox-cover/75cab84/cover/ : SUCCESS in 9m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f0de2def117adaa7ccb51b3a33810f377096ab7e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-18 20:36:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8932194dda979f9f9fa82889c1a32e2fd24401ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 22:40:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/57/665557/1/check/openstack-tox-lower-constraints/ab80d75/ : SUCCESS in 8m 51s\n- openstack-tox-pep8 http://logs.openstack.org/57/665557/1/check/openstack-tox-pep8/fedaabe/ : SUCCESS in 8m 54s\n- openstack-tox-py27 http://logs.openstack.org/57/665557/1/check/openstack-tox-py27/c35ef40/ : SUCCESS in 6m 44s\n- openstack-tox-py35 http://logs.openstack.org/57/665557/1/check/openstack-tox-py35/a87045e/ : SUCCESS in 7m 00s\n- openstack-tox-docs http://logs.openstack.org/57/665557/1/check/openstack-tox-docs/9cfdd8c/html/ : SUCCESS in 6m 33s\n- build-openstack-releasenotes http://logs.openstack.org/57/665557/1/check/build-openstack-releasenotes/d6b5e19/html/ : SUCCESS in 5m 38s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-api/0ff5510/ : SUCCESS in 1h 23m 11s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-py3x-api/8e3d3b5/ : SUCCESS in 1h 15m 27s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-scenario/2bdc0b4/ : FAILURE in 1h 48m 46s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-py3x-scenario/fff0a13/ : SUCCESS in 1h 22m 55s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/57/665557/1/check/openstack-tox-functional/00aff78/ : SUCCESS in 6m 35s\n- openstack-tox-functional-py35 http://logs.openstack.org/57/665557/1/check/openstack-tox-functional-py35/5e85e87/ : SUCCESS in 7m 56s\n- openstack-tox-functional-py36 http://logs.openstack.org/57/665557/1/check/openstack-tox-functional-py36/7b51ecd/ : SUCCESS in 8m 09s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-scenario/ad6cab4/ : SUCCESS in 55m 12s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-py3x-scenario/fbede3e/ : FAILURE in 54m 16s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-scenario-multinode/34f05ad/ : FAILURE in 1h 26m 44s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-py3x-scenario-multinode/b74aabf/ : FAILURE in 1h 25m 24s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/b0a758d/ : FAILURE in 1h 15m 28s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-noop-api/a7bdd5d/ : SUCCESS in 43m 36s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-noop-py35-api/1792788/ : SUCCESS in 49m 14s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-scenario/dead622/ : SUCCESS in 1h 24m 04s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-py35-scenario/094f880/ : SUCCESS in 1h 43m 00s (non-voting)\n- octavia-grenade http://logs.openstack.org/57/665557/1/check/octavia-grenade/0a4ae27/ : SUCCESS in 1h 12m 46s\n- openstack-tox-cover http://logs.openstack.org/57/665557/1/check/openstack-tox-cover/61b7b7f/cover/ : SUCCESS in 10m 24s","accounts_in_message":[],"_revision_number":1},{"id":"0a3d0d67194677d4bdf70c19e4b856a2ac73d6ee","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-20 21:45:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cc8976ae5d32812162e5c3e9759847490984e695","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 00:08:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/57/665557/1/check/openstack-tox-lower-constraints/4501384/ : SUCCESS in 9m 35s\n- openstack-tox-pep8 http://logs.openstack.org/57/665557/1/check/openstack-tox-pep8/1af79d3/ : SUCCESS in 8m 27s\n- openstack-tox-py27 http://logs.openstack.org/57/665557/1/check/openstack-tox-py27/3e673a3/ : SUCCESS in 21m 45s\n- openstack-tox-py35 http://logs.openstack.org/57/665557/1/check/openstack-tox-py35/5412d51/ : SUCCESS in 21m 45s\n- openstack-tox-docs http://logs.openstack.org/57/665557/1/check/openstack-tox-docs/59d5ed7/html/ : SUCCESS in 20m 08s\n- build-openstack-releasenotes http://logs.openstack.org/57/665557/1/check/build-openstack-releasenotes/e007ebc/html/ : SUCCESS in 7m 52s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-api/7350283/ : SUCCESS in 1h 15m 26s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-py3x-api/065badd/ : FAILURE in 2h 08m 24s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-scenario/34ea207/ : SUCCESS in 1h 27m 07s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-py3x-scenario/3c7032d/ : SUCCESS in 1h 01m 01s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/57/665557/1/check/openstack-tox-functional/40cce5e/ : SUCCESS in 22m 37s\n- openstack-tox-functional-py35 http://logs.openstack.org/57/665557/1/check/openstack-tox-functional-py35/af3dc3b/ : SUCCESS in 22m 16s\n- openstack-tox-functional-py36 http://logs.openstack.org/57/665557/1/check/openstack-tox-functional-py36/69c78f5/ : SUCCESS in 19m 06s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-scenario/7f57cd7/ : FAILURE in 2h 09m 30s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-py3x-scenario/8b76d57/ : FAILURE in 2h 09m 43s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-scenario-multinode/c98b975/ : FAILURE in 2h 09m 22s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-py3x-scenario-multinode/ee3daa1/ : FAILURE in 1h 30m 40s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/c4fc5a9/ : SUCCESS in 1h 12m 38s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-noop-api/4226617/ : SUCCESS in 52m 08s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-noop-py35-api/82b646d/ : SUCCESS in 50m 14s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-scenario/22c004d/ : SUCCESS in 1h 43m 24s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-py35-scenario/082a1df/ : SUCCESS in 1h 40m 24s (non-voting)\n- octavia-grenade http://logs.openstack.org/57/665557/1/check/octavia-grenade/7cb2590/ : SUCCESS in 1h 16m 50s\n- openstack-tox-cover http://logs.openstack.org/57/665557/1/check/openstack-tox-cover/ce1cf7d/cover/ : SUCCESS in 11m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a17acf3ee1f8d6db21a69c6db39d1a1f35cc96ba","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-21 08:12:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d1967dcccffd454d5cefdb45f8a0c5f0b607f9bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 10:51:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/57/665557/1/check/openstack-tox-lower-constraints/6f48c47/ : SUCCESS in 9m 20s\n- openstack-tox-pep8 http://logs.openstack.org/57/665557/1/check/openstack-tox-pep8/ce90ec7/ : SUCCESS in 8m 10s\n- openstack-tox-py27 http://logs.openstack.org/57/665557/1/check/openstack-tox-py27/8c34a6f/ : SUCCESS in 8m 22s\n- openstack-tox-py35 http://logs.openstack.org/57/665557/1/check/openstack-tox-py35/2029d67/ : SUCCESS in 7m 41s\n- openstack-tox-docs http://logs.openstack.org/57/665557/1/check/openstack-tox-docs/c1e5e2e/html/ : SUCCESS in 8m 24s\n- build-openstack-releasenotes http://logs.openstack.org/57/665557/1/check/build-openstack-releasenotes/e637731/html/ : SUCCESS in 7m 05s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-api/d06c21b/ : SUCCESS in 1h 10m 14s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-py3x-api/f527b4a/ : SUCCESS in 1h 17m 12s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-scenario/c225bef/ : SUCCESS in 51m 19s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/57/665557/1/check/neutron-lbaasv2-dsvm-py3x-scenario/026b490/ : SUCCESS in 1h 27m 25s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/57/665557/1/check/openstack-tox-functional/97c3f5a/ : SUCCESS in 10m 45s\n- openstack-tox-functional-py35 http://logs.openstack.org/57/665557/1/check/openstack-tox-functional-py35/fc91cb2/ : SUCCESS in 22m 23s\n- openstack-tox-functional-py36 http://logs.openstack.org/57/665557/1/check/openstack-tox-functional-py36/ccd44d5/ : SUCCESS in 11m 23s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-scenario/e4c528c/ : SUCCESS in 1h 00m 05s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-py3x-scenario/21a14c6/ : SUCCESS in 45m 32s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-scenario-multinode/5c228c3/ : FAILURE in 1h 21m 57s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-py3x-scenario-multinode/29adc42/ : FAILURE in 1h 18m 38s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/57/665557/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/d58f14c/ : FAILURE in 2h 08m 08s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-noop-api/7b5e014/ : SUCCESS in 44m 42s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-noop-py35-api/f99e44e/ : SUCCESS in 1h 57m 12s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-scenario/758d39f/ : SUCCESS in 1h 29m 26s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/57/665557/1/check/octavia-v2-dsvm-py35-scenario/4e0f44e/ : TIMED_OUT in 2h 33m 55s (non-voting)\n- octavia-grenade http://logs.openstack.org/57/665557/1/check/octavia-grenade/5eca5e2/ : SUCCESS in 1h 12m 36s\n- openstack-tox-cover http://logs.openstack.org/57/665557/1/check/openstack-tox-cover/cf5de86/cover/ : SUCCESS in 24m 34s","accounts_in_message":[],"_revision_number":1},{"id":"3b90f36e8a06c29aca4c8863d7999841ae8c58a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 10:52:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"acda55a48e198564b8a8c92edf5f21ff26f07cc6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 13:10:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/57/665557/1/gate/openstack-tox-lower-constraints/13f6090/ : SUCCESS in 8m 22s\n- openstack-tox-pep8 http://logs.openstack.org/57/665557/1/gate/openstack-tox-pep8/e9b81ca/ : SUCCESS in 8m 29s\n- openstack-tox-py27 http://logs.openstack.org/57/665557/1/gate/openstack-tox-py27/b051309/ : SUCCESS in 7m 11s\n- openstack-tox-py35 http://logs.openstack.org/57/665557/1/gate/openstack-tox-py35/8394ba1/ : SUCCESS in 8m 19s\n- openstack-tox-docs http://logs.openstack.org/57/665557/1/gate/openstack-tox-docs/3cddefe/html/ : SUCCESS in 6m 38s\n- build-openstack-releasenotes http://logs.openstack.org/57/665557/1/gate/build-openstack-releasenotes/33939b0/html/ : SUCCESS in 5m 45s\n- openstack-tox-functional http://logs.openstack.org/57/665557/1/gate/openstack-tox-functional/5f2c221/ : SUCCESS in 6m 35s\n- openstack-tox-functional-py35 http://logs.openstack.org/57/665557/1/gate/openstack-tox-functional-py35/d6b5f9b/ : SUCCESS in 7m 05s\n- openstack-tox-functional-py36 http://logs.openstack.org/57/665557/1/gate/openstack-tox-functional-py36/d161460/ : SUCCESS in 6m 43s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/57/665557/1/gate/octavia-v1-dsvm-scenario/a571f70/ : SUCCESS in 1h 02m 12s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/57/665557/1/gate/octavia-v1-dsvm-py3x-scenario/2c6101d/ : SUCCESS in 58m 29s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/57/665557/1/gate/octavia-v1-dsvm-scenario-multinode/32e60d0/ : FAILURE in 1h 21m 29s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/57/665557/1/gate/octavia-v1-dsvm-py3x-scenario-multinode/9b1eae4/ : FAILURE in 2h 09m 36s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/57/665557/1/gate/neutron-lbaasv2-dsvm-py3x-api/90e2189/ : SUCCESS in 1h 19m 30s\n- octavia-grenade http://logs.openstack.org/57/665557/1/gate/octavia-grenade/4d04a7b/ : SUCCESS in 1h 05m 40s\n- openstack-tox-cover http://logs.openstack.org/57/665557/1/gate/openstack-tox-cover/d3da94d/cover/ : SUCCESS in 8m 58s","accounts_in_message":[],"_revision_number":1},{"id":"fb6c5d1b7981cc14ca1a6cb2b42e66bde9235660","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 13:10:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac6ff98dfdbc705b26bdcda6ce45ed3c6dde04b0","revisions":{"ac6ff98dfdbc705b26bdcda6ce45ed3c6dde04b0":{"kind":"REWORK","_number":1,"created":"2019-06-16 19:32:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/665557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/665557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/665557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/665557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/665557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/665557/1"}}},"commit":{"parents":[{"commit":"b12f0d9d6ce5e873b8e9b501e50c314ffb02e128","subject":"Merge \"Treat null admin_state_up as False\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b12f0d9d6ce5e873b8e9b501e50c314ffb02e128"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-14 16:39:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-16 19:32:25.000000000","tz":0},"subject":"Fix allocate_and_associate DB deadlock","message":"Fix allocate_and_associate DB deadlock\n\nIn testing we saw a DB deadlock in allocate_and_associate.\nThis patch corrects that with a DB retry wrapper.\n\nChange-Id: I20572505e448f9b7ff6314b5167cc4c519c445fb\nStory: 2005893\nTask: 33751\n(cherry picked from commit 0eaab450c12e33f277c31098a6792421d3f5806c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac6ff98dfdbc705b26bdcda6ce45ed3c6dde04b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac6ff98dfdbc705b26bdcda6ce45ed3c6dde04b0"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
