)]}'
{"id":"openstack%2Foctavia~665683","triplet_id":"openstack%2Foctavia~stable%2Frocky~Ifc28b4991852e59c0d27b4ab3d1afc4e9965e88b","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"Ifc28b4991852e59c0d27b4ab3d1afc4e9965e88b","subject":"Update amphora-agent to report UDP listener health","status":"MERGED","created":"2019-06-17 14:14:33.000000000","updated":"2019-07-23 09:46:03.000000000","submitted":"2019-07-23 09:46:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665683-1563875163979-358cd872","meta_rev_id":"8b4366292402ad5acd3574d57f72f590c136020f","_number":665683,"virtual_id_number":665683,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-07-23 09:46:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-07-23 08:06:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-07-19 11:51:38.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-07-23 08:06:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2019-06-17 14:14:33.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 14:14:33.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-07-19 11:51:38.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-07-23 08:06:59.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-07-23 09:46:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fdf6ef8d9ef494b9bfc64b3cceeee1f26eab763f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-06-17 14:14:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"468f97c42311697b10c672af19be48211adf7674","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 15:51:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/83/665683/1/check/openstack-tox-lower-constraints/eda137d/ : SUCCESS in 8m 15s\n- openstack-tox-pep8 http://logs.openstack.org/83/665683/1/check/openstack-tox-pep8/69222f5/ : SUCCESS in 8m 52s\n- openstack-tox-py27 http://logs.openstack.org/83/665683/1/check/openstack-tox-py27/59c2335/ : SUCCESS in 6m 30s\n- openstack-tox-py35 http://logs.openstack.org/83/665683/1/check/openstack-tox-py35/537dad7/ : SUCCESS in 6m 14s\n- openstack-tox-docs http://logs.openstack.org/83/665683/1/check/openstack-tox-docs/8b078a8/html/ : SUCCESS in 6m 43s\n- build-openstack-releasenotes http://logs.openstack.org/83/665683/1/check/build-openstack-releasenotes/5077a22/html/ : SUCCESS in 5m 34s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/83/665683/1/check/neutron-lbaasv2-dsvm-api/645144e/ : SUCCESS in 1h 13m 15s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/83/665683/1/check/neutron-lbaasv2-dsvm-py3x-api/828dbd0/ : SUCCESS in 1h 17m 57s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/83/665683/1/check/neutron-lbaasv2-dsvm-scenario/a0df9e5/ : SUCCESS in 1h 18m 24s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/83/665683/1/check/neutron-lbaasv2-dsvm-py3x-scenario/8ef67cf/ : FAILURE in 1h 15m 11s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/83/665683/1/check/openstack-tox-functional/ef96c37/ : SUCCESS in 7m 55s\n- openstack-tox-functional-py35 http://logs.openstack.org/83/665683/1/check/openstack-tox-functional-py35/c917340/ : SUCCESS in 7m 16s\n- openstack-tox-functional-py36 http://logs.openstack.org/83/665683/1/check/openstack-tox-functional-py36/a743118/ : SUCCESS in 6m 24s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/83/665683/1/check/octavia-v1-dsvm-scenario/3bdf425/ : SUCCESS in 1h 03m 04s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/83/665683/1/check/octavia-v1-dsvm-py3x-scenario/cbc9729/ : SUCCESS in 59m 21s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/83/665683/1/check/octavia-v1-dsvm-scenario-multinode/b49613b/ : FAILURE in 1h 02m 21s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/83/665683/1/check/octavia-v1-dsvm-py3x-scenario-multinode/f9587a0/ : FAILURE in 43m 26s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/83/665683/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/cd8d930/ : SUCCESS in 1h 16m 19s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/83/665683/1/check/octavia-v2-dsvm-noop-api/6b542f3/ : SUCCESS in 42m 53s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/83/665683/1/check/octavia-v2-dsvm-noop-py35-api/b8c3367/ : SUCCESS in 46m 15s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/83/665683/1/check/octavia-v2-dsvm-scenario/38f8a2c/ : SUCCESS in 1h 25m 52s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/83/665683/1/check/octavia-v2-dsvm-py35-scenario/7a4317b/ : SUCCESS in 1h 23m 14s (non-voting)\n- octavia-grenade http://logs.openstack.org/83/665683/1/check/octavia-grenade/01de403/ : SUCCESS in 1h 07m 54s\n- openstack-tox-cover http://logs.openstack.org/83/665683/1/check/openstack-tox-cover/226d12a/cover/ : SUCCESS in 10m 08s","accounts_in_message":[],"_revision_number":1},{"id":"33848196a8cb95cd3d17cd869f96df2f20ded646","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-19 11:51:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0fc0154032c4ddf8f823b6c9532c2fd2f659e02e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-23 08:06:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"64fbd87b68abf31c5a8ce33bd87c3ad34adb56e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 08:07:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d9ff14c2b686c39939a2ea869eb6031ef405570a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 09:46:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8b4366292402ad5acd3574d57f72f590c136020f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 09:46:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/83/665683/1/gate/openstack-tox-lower-constraints/abf1f15/ : SUCCESS in 8m 30s\n- openstack-tox-pep8 http://logs.openstack.org/83/665683/1/gate/openstack-tox-pep8/d02a49a/ : SUCCESS in 7m 41s\n- openstack-tox-py27 http://logs.openstack.org/83/665683/1/gate/openstack-tox-py27/47c5cb7/ : SUCCESS in 4m 48s\n- openstack-tox-py35 http://logs.openstack.org/83/665683/1/gate/openstack-tox-py35/4b419e8/ : SUCCESS in 7m 41s\n- openstack-tox-docs http://logs.openstack.org/83/665683/1/gate/openstack-tox-docs/57ce397/html/ : SUCCESS in 5m 00s\n- build-openstack-releasenotes http://logs.openstack.org/83/665683/1/gate/build-openstack-releasenotes/82a6ce7/html/ : SUCCESS in 4m 27s\n- openstack-tox-functional http://logs.openstack.org/83/665683/1/gate/openstack-tox-functional/d39ecd6/ : SUCCESS in 6m 20s\n- openstack-tox-functional-py35 http://logs.openstack.org/83/665683/1/gate/openstack-tox-functional-py35/074edd1/ : SUCCESS in 6m 41s\n- openstack-tox-functional-py36 http://logs.openstack.org/83/665683/1/gate/openstack-tox-functional-py36/4d369a1/ : SUCCESS in 6m 49s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/83/665683/1/gate/octavia-v1-dsvm-scenario/c254acb/ : SUCCESS in 1h 03m 05s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/83/665683/1/gate/octavia-v1-dsvm-py3x-scenario/4707635/ : SUCCESS in 1h 02m 15s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/83/665683/1/gate/octavia-v1-dsvm-scenario-multinode/cd20638/ : FAILURE in 1h 24m 41s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/83/665683/1/gate/octavia-v1-dsvm-py3x-scenario-multinode/b8e4d92/ : FAILURE in 1h 32m 10s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/83/665683/1/gate/neutron-lbaasv2-dsvm-py3x-api/6361189/ : SUCCESS in 1h 22m 56s\n- octavia-grenade http://logs.openstack.org/83/665683/1/gate/octavia-grenade/9a0f473/ : SUCCESS in 1h 17m 12s\n- openstack-tox-cover http://logs.openstack.org/83/665683/1/gate/openstack-tox-cover/acfeceb/cover/ : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f8fcedcd0bfc21699628cda81f1348e4c8f1539","revisions":{"2f8fcedcd0bfc21699628cda81f1348e4c8f1539":{"kind":"REWORK","_number":1,"created":"2019-06-17 14:14:33.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/83/665683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/665683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/665683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/665683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/665683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/665683/1"}}},"commit":{"parents":[{"commit":"b12f0d9d6ce5e873b8e9b501e50c314ffb02e128","subject":"Merge \"Treat null admin_state_up as False\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b12f0d9d6ce5e873b8e9b501e50c314ffb02e128"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2018-08-30 14:42:27.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-06-17 14:14:33.000000000","tz":0},"subject":"Update amphora-agent to report UDP listener health","message":"Update amphora-agent to report UDP listener health\n\nCurrently the amphora-agent is not reporting UDP listener health\nwhen the UDP listener does not have a pool and members.\nThis patch changes that behavior to report the listener as healthy\nif the keepalived process is started and running in the amphora.\n\nThis patch also introduces message versioning for the health\nheartbeat messages.\n\nIt also corrects a few assertEqual tests that had the reference and\nactual values backwards.\n\nChange-Id: Ifc28b4991852e59c0d27b4ab3d1afc4e9965e88b\nStory: 2003592\nTask: 24911\n(cherry picked from commit 2170cc6c459b7ae8461a09a0b6fd754ecef9654e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f8fcedcd0bfc21699628cda81f1348e4c8f1539"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f8fcedcd0bfc21699628cda81f1348e4c8f1539"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
