)]}'
{"id":"openstack%2Foctavia~665915","triplet_id":"openstack%2Foctavia~master~I8f9c25e979a7e86ff0b8e19267e4525817802ba2","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I8f9c25e979a7e86ff0b8e19267e4525817802ba2","subject":"Workaround pyroute2 platform-dependent imports","status":"MERGED","created":"2019-06-18 08:48:15.000000000","updated":"2019-06-27 08:05:27.000000000","submitted":"2019-06-27 08:05:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665915-1561622727611-663f6094","meta_rev_id":"be4ea404154602ebb51b6c3ed1edaee07d7ceddf","_number":665915,"virtual_id_number":665915,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-19 00:46:06.000000000","_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-06-27 08:05:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-06-27 06:18:25.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2019-06-27 05:50:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-06-26 22:01:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-06-27 06:18:25.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-26 22:01:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-06-27 05:50:07.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-06-27 06:18:25.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-06-27 08:05:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58afcd166aa6531af30fc786611f0dc0724b6a91","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-18 08:48:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"acfe9a02dc1925b3d024c3534bfeb744129c472f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 11:09:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/15/665915/1/check/openstack-tox-cover/814cd37/cover/ : SUCCESS in 9m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/15/665915/1/check/openstack-tox-lower-constraints/4bed08f/ : SUCCESS in 6m 54s\n- openstack-tox-pep8 http://logs.openstack.org/15/665915/1/check/openstack-tox-pep8/104eaf8/ : SUCCESS in 6m 57s\n- openstack-tox-py27 http://logs.openstack.org/15/665915/1/check/openstack-tox-py27/13d652e/ : SUCCESS in 6m 11s\n- openstack-tox-py36 http://logs.openstack.org/15/665915/1/check/openstack-tox-py36/c2be9d6/ : SUCCESS in 5m 40s\n- openstack-tox-py37 http://logs.openstack.org/15/665915/1/check/openstack-tox-py37/d4d7580/ : SUCCESS in 5m 58s\n- openstack-tox-docs http://logs.openstack.org/15/665915/1/check/openstack-tox-docs/566e58d/html/ : SUCCESS in 7m 09s\n- openstack-tox-functional http://logs.openstack.org/15/665915/1/check/openstack-tox-functional/939fe47/ : SUCCESS in 6m 20s\n- openstack-tox-functional-py36 http://logs.openstack.org/15/665915/1/check/openstack-tox-functional-py36/55165dc/ : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-noop-api/fc926ef/ : SUCCESS in 45m 10s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-noop-py2-api/17f4b03/ : SUCCESS in 47m 00s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-scenario/2f8f6c0/ : SUCCESS in 1h 40m 39s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-py2-scenario/06ae535/ : SUCCESS in 1h 35m 06s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-py2-scenario-centos-7/1edf9a6/ : FAILURE in 2h 10m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/fb039a2/ : POST_FAILURE in 2h 14m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-scenario-two-node/fa0622d/ : POST_FAILURE in 2h 11m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-py2-scenario-two-node/fda8276/ : SUCCESS in 1h 44m 04s (non-voting)\n- octavia-grenade http://logs.openstack.org/15/665915/1/check/octavia-grenade/ef063ed/ : FAILURE in 49m 19s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-tls-barbican/2bdc661/ : SUCCESS in 58m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-scenario-amphora-v2/ead5b49/ : SUCCESS in 1h 41m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ed5e897356a9876f6f9a5cff42dc6bbfe31eb76","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-19 00:46:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"de66c88db396885d5c5c805b38c1023bbffcb0eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 03:05:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/15/665915/1/check/openstack-tox-cover/416172b/cover/ : SUCCESS in 7m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/15/665915/1/check/openstack-tox-lower-constraints/e823e7b/ : SUCCESS in 8m 16s\n- openstack-tox-pep8 http://logs.openstack.org/15/665915/1/check/openstack-tox-pep8/96a9a97/ : SUCCESS in 6m 35s\n- openstack-tox-py27 http://logs.openstack.org/15/665915/1/check/openstack-tox-py27/7da3d36/ : SUCCESS in 5m 28s\n- openstack-tox-py36 http://logs.openstack.org/15/665915/1/check/openstack-tox-py36/7cfb0f9/ : SUCCESS in 6m 03s\n- openstack-tox-py37 http://logs.openstack.org/15/665915/1/check/openstack-tox-py37/1d7e4dc/ : SUCCESS in 5m 37s\n- openstack-tox-docs http://logs.openstack.org/15/665915/1/check/openstack-tox-docs/9fa44d3/html/ : SUCCESS in 6m 40s\n- openstack-tox-functional http://logs.openstack.org/15/665915/1/check/openstack-tox-functional/5df50f8/ : SUCCESS in 6m 54s\n- openstack-tox-functional-py36 http://logs.openstack.org/15/665915/1/check/openstack-tox-functional-py36/c795116/ : SUCCESS in 6m 23s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-noop-api/b5fb336/ : SUCCESS in 40m 17s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-noop-py2-api/2b28fda/ : SUCCESS in 39m 22s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-scenario/770d3bf/ : SUCCESS in 2h 04m 52s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-py2-scenario/b00c945/ : SUCCESS in 1h 42m 22s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-py2-scenario-centos-7/ba222c1/ : FAILURE in 2h 08m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/3cffcee/ : SUCCESS in 1h 53m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-scenario-two-node/a6f2fa3/ : SUCCESS in 1h 44m 59s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-py2-scenario-two-node/0f0bb7c/ : SUCCESS in 1h 42m 54s (non-voting)\n- octavia-grenade http://logs.openstack.org/15/665915/1/check/octavia-grenade/1175cc6/ : SUCCESS in 1h 20m 18s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-tls-barbican/e683e38/ : SUCCESS in 57m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/15/665915/1/check/octavia-v2-dsvm-scenario-amphora-v2/f39a76c/ : SUCCESS in 1h 35m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1bcbca82cc3293519182ef22332d25d394acd75","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-26 22:01:25.000000000","message":"Patch Set 1: Code-Review+2\n\nI am ok with this, though it\u0027s kind of sad.","accounts_in_message":[],"_revision_number":1},{"id":"fb88351020038dc06dbddaced0e9eb361089b30f","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-06-27 05:50:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"36088a29ac55fb82aa4dec9992833ebb23dcf7ee","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-06-27 06:18:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"454e98b0a118caa7aa7283287a6c05498b9d5da0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 06:18:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27e86d893510bb97925206f003227687423b14d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 08:05:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"be4ea404154602ebb51b6c3ed1edaee07d7ceddf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 08:05:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/15/665915/1/gate/openstack-tox-lower-constraints/e8b7156/ : SUCCESS in 5m 37s\n- openstack-tox-pep8 http://logs.openstack.org/15/665915/1/gate/openstack-tox-pep8/32829f6/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/15/665915/1/gate/openstack-tox-py27/63f3318/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/15/665915/1/gate/openstack-tox-py36/2fd1754/ : SUCCESS in 4m 25s\n- openstack-tox-py37 http://logs.openstack.org/15/665915/1/gate/openstack-tox-py37/d003dfb/ : SUCCESS in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/15/665915/1/gate/openstack-tox-docs/bc0e0fd/html/ : SUCCESS in 5m 59s\n- openstack-tox-functional http://logs.openstack.org/15/665915/1/gate/openstack-tox-functional/d22db5a/ : SUCCESS in 5m 38s\n- openstack-tox-functional-py36 http://logs.openstack.org/15/665915/1/gate/openstack-tox-functional-py36/5442d76/ : SUCCESS in 6m 34s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/15/665915/1/gate/octavia-v2-dsvm-noop-api/a39197b/ : SUCCESS in 40m 45s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/15/665915/1/gate/octavia-v2-dsvm-noop-py2-api/640f1e9/ : SUCCESS in 42m 21s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/15/665915/1/gate/octavia-v2-dsvm-scenario/f29be83/ : SUCCESS in 1h 31m 15s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/15/665915/1/gate/octavia-v2-dsvm-py2-scenario/0873074/ : SUCCESS in 1h 39m 17s\n- octavia-grenade http://logs.openstack.org/15/665915/1/gate/octavia-grenade/120bedf/ : SUCCESS in 1h 13m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"624a63dce01006ad8c158f1c7c0841d8a04c3664","revisions":{"624a63dce01006ad8c158f1c7c0841d8a04c3664":{"kind":"REWORK","_number":1,"created":"2019-06-18 08:48:15.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/15/665915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/665915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/665915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/665915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/665915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/665915/1"}}},"commit":{"parents":[{"commit":"954025cbc32b5286509985062aff17900e04743d","subject":"Merge \"Fix a python3 issue in the amphora-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/954025cbc32b5286509985062aff17900e04743d"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-18 08:34:32.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-18 08:34:32.000000000","tz":-420},"subject":"Workaround pyroute2 platform-dependent imports","message":"Workaround pyroute2 platform-dependent imports\n\nPyroute2 recently changed the way they import their base namespace\nojects. Now they are somewhat platform dependent, so on OSX unit tests\nwill fail for lack of linux-specific libs. Using `create\u003dtrue` allows\nthe mocks to be created even if the library doesn\u0027t think they exist.\nhttps://github.com/svinota/pyroute2/commit/b69999ec82667828816fa2a7b19d0a4f2fe06a83\n\nChange-Id: I8f9c25e979a7e86ff0b8e19267e4525817802ba2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/624a63dce01006ad8c158f1c7c0841d8a04c3664"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/624a63dce01006ad8c158f1c7c0841d8a04c3664"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
