)]}'
{"id":"openstack%2Foctavia~667065","triplet_id":"openstack%2Foctavia~master~I8acde313e7e295cdecaa593dbeb41e8ee49afed5","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I8acde313e7e295cdecaa593dbeb41e8ee49afed5","subject":"Switch python3 versions of test jobs to match Train PTI","status":"MERGED","created":"2019-06-24 11:02:26.000000000","updated":"2019-06-27 12:15:07.000000000","submitted":"2019-06-27 12:15:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667065-1561637707276-f9ed76da","meta_rev_id":"3dadcff1d929d3a09c8b5f158a18c5be13fdd9a4","_number":667065,"virtual_id_number":667065,"owner":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":0,"date":"2019-06-24 11:02:37.000000000","_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-06-27 12:15:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"recommended":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"all":[{"value":1,"date":"2019-06-27 10:18:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":2,"date":"2019-06-27 10:30:41.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-06-24 18:37:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":1,"date":"2019-06-27 10:30:41.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":-2,"date":"2019-06-24 18:37:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 11:02:37.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2019-06-24 18:37:30.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-06-27 10:18:14.000000000","updated_by":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"reviewer":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"state":"REVIEWER"},{"updated":"2019-06-27 10:30:41.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-06-27 12:15:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9144774b2234c24fa2768ba006e99b3f5f6edc50","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-06-24 11:02:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04673795b1dbd8e90e6d5847a87bad77b9405775","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-06-24 11:02:37.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"6bb6cee2a9e1b78c179402979000915856c4b650","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 13:54:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/65/667065/1/check/openstack-tox-cover/9ae7b5b/cover/ : SUCCESS in 8m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/667065/1/check/openstack-tox-lower-constraints/b40bed4/ : SUCCESS in 6m 49s\n- openstack-tox-pep8 http://logs.openstack.org/65/667065/1/check/openstack-tox-pep8/c148936/ : SUCCESS in 6m 23s\n- openstack-tox-py27 http://logs.openstack.org/65/667065/1/check/openstack-tox-py27/45e32b1/ : SUCCESS in 4m 57s\n- openstack-tox-py36 http://logs.openstack.org/65/667065/1/check/openstack-tox-py36/abb18db/ : SUCCESS in 7m 22s\n- openstack-tox-py37 http://logs.openstack.org/65/667065/1/check/openstack-tox-py37/967e25c/ : SUCCESS in 8m 25s\n- openstack-tox-docs http://logs.openstack.org/65/667065/1/check/openstack-tox-docs/e7e6f6c/html/ : SUCCESS in 8m 35s\n- openstack-tox-functional http://logs.openstack.org/65/667065/1/check/openstack-tox-functional/aad90e0/ : SUCCESS in 8m 16s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/667065/1/check/openstack-tox-functional-py36/a52fec8/ : SUCCESS in 6m 19s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/65/667065/1/check/octavia-v2-dsvm-noop-api/faaebaf/ : SUCCESS in 46m 33s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/65/667065/1/check/octavia-v2-dsvm-noop-py2-api/86bad50/ : SUCCESS in 37m 31s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/65/667065/1/check/octavia-v2-dsvm-scenario/48f676e/ : SUCCESS in 1h 38m 58s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/65/667065/1/check/octavia-v2-dsvm-py2-scenario/8783303/ : SUCCESS in 1h 34m 20s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/65/667065/1/check/octavia-v2-dsvm-py2-scenario-centos-7/0532dee/ : FAILURE in 1h 58m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/65/667065/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/e5d0d70/ : FAILURE in 1h 42m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/65/667065/1/check/octavia-v2-dsvm-scenario-two-node/6ddb2be/ : SUCCESS in 1h 24m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/65/667065/1/check/octavia-v2-dsvm-py2-scenario-two-node/014d753/ : SUCCESS in 1h 43m 08s (non-voting)\n- octavia-grenade http://logs.openstack.org/65/667065/1/check/octavia-grenade/9405fd2/ : SUCCESS in 1h 10m 41s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/65/667065/1/check/octavia-v2-dsvm-tls-barbican/e96feb4/ : SUCCESS in 57m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/65/667065/1/check/octavia-v2-dsvm-scenario-amphora-v2/80bdd2d/ : SUCCESS in 1h 41m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b454360d682d3ada1846347d4db4bca39930456","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-24 18:37:30.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2\n\nSeems to run the same tests, so sure.","accounts_in_message":[],"_revision_number":1},{"id":"14e03956e7fbcffd9781b732c9e5320e27ff5f85","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-06-27 10:18:14.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM!","accounts_in_message":[],"_revision_number":1},{"id":"428e69c6bf30ed6969345f183cb5834bd88284e7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-27 10:30:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"695301c75758864495fb0548b6605106d8e706ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 10:30:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b34906bc854a0f7cf95cf8c5446987677db53344","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 12:15:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3dadcff1d929d3a09c8b5f158a18c5be13fdd9a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 12:15:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/667065/1/gate/openstack-tox-lower-constraints/e22c8d2/ : SUCCESS in 5m 47s\n- openstack-tox-pep8 http://logs.openstack.org/65/667065/1/gate/openstack-tox-pep8/2830cde/ : SUCCESS in 5m 24s\n- openstack-tox-py27 http://logs.openstack.org/65/667065/1/gate/openstack-tox-py27/bfe8932/ : SUCCESS in 4m 15s\n- openstack-tox-py36 http://logs.openstack.org/65/667065/1/gate/openstack-tox-py36/94201a6/ : SUCCESS in 4m 27s\n- openstack-tox-py37 http://logs.openstack.org/65/667065/1/gate/openstack-tox-py37/c1a18bf/ : SUCCESS in 4m 14s\n- openstack-tox-docs http://logs.openstack.org/65/667065/1/gate/openstack-tox-docs/7d18d26/html/ : SUCCESS in 5m 16s\n- openstack-tox-functional http://logs.openstack.org/65/667065/1/gate/openstack-tox-functional/08a0a97/ : SUCCESS in 5m 25s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/667065/1/gate/openstack-tox-functional-py36/da36ebb/ : SUCCESS in 5m 48s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/65/667065/1/gate/octavia-v2-dsvm-noop-api/2a73f49/ : SUCCESS in 42m 56s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/65/667065/1/gate/octavia-v2-dsvm-noop-py2-api/cee5de4/ : SUCCESS in 42m 56s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/65/667065/1/gate/octavia-v2-dsvm-scenario/fbcd37b/ : SUCCESS in 1h 37m 37s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/65/667065/1/gate/octavia-v2-dsvm-py2-scenario/69cba61/ : SUCCESS in 1h 30m 34s\n- octavia-grenade http://logs.openstack.org/65/667065/1/gate/octavia-grenade/816dca1/ : SUCCESS in 1h 16m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a144e512cfd5a99c5e0c30a0dc71972939acc0b9","revisions":{"a144e512cfd5a99c5e0c30a0dc71972939acc0b9":{"kind":"REWORK","_number":1,"created":"2019-06-24 11:02:26.000000000","uploader":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"ref":"refs/changes/65/667065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/667065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/667065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/667065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/667065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/667065/1"}}},"commit":{"parents":[{"commit":"3a9df30a9a4efd2a47919001dcf38d3377996ed7","subject":"Merge \"Fix pool API handling of None/null updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3a9df30a9a4efd2a47919001dcf38d3377996ed7"}]}],"author":{"name":"AjayKumar333","email":"ajay.k@india.nec.com","date":"2019-06-24 10:59:24.000000000","tz":330},"committer":{"name":"AjayKumar333","email":"ajay.k@india.nec.com","date":"2019-06-24 11:02:21.000000000","tz":330},"subject":"Switch python3 versions of test jobs to match Train PTI","message":"Switch python3 versions of test jobs to match Train PTI\n\nIn Train, we will use python 3.6 and 3.7 for python3 runtime\nin our gate jobs [1]. This commit also adds python 3.7.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I8acde313e7e295cdecaa593dbeb41e8ee49afed5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a144e512cfd5a99c5e0c30a0dc71972939acc0b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a144e512cfd5a99c5e0c30a0dc71972939acc0b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
