)]}'
{"id":"openstack%2Foctavia~667200","triplet_id":"openstack%2Foctavia~stable%2Fqueens~Ib0fb3db3b2d56345eeeda2014c15e35281bcfec8","project":"openstack/octavia","branch":"stable/queens","topic":"story/2005925","hashtags":[],"change_id":"Ib0fb3db3b2d56345eeeda2014c15e35281bcfec8","subject":"Validate certificate content at API level","status":"MERGED","created":"2019-06-24 19:21:23.000000000","updated":"2019-06-27 03:03:57.000000000","submitted":"2019-06-27 03:03:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667200-1561604637854-f790302f","meta_rev_id":"f90d2936853c240cb1ee64a630b2607b5f1e197b","_number":667200,"virtual_id_number":667200,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"date":"2019-06-25 18:21:30.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2019-06-27 03:03:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"all":[{"value":2,"date":"2019-06-26 11:37:41.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-06-27 01:15:25.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-06-24 23:55:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2019-06-25 13:08:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-06-27 01:15:25.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 23:55:19.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-06-25 13:08:52.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-06-26 11:37:41.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-06-27 01:15:25.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-06-27 03:03:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c46acb740b22c2215d57cc3989cc3c976a7b4318","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-24 19:21:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"caf31c939e8281ebe9eae2833ca073bfa706b6ba","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-24 19:22:10.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"bc52bea4e5cc8153a7b5f88cc90863610332a94c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-24 20:23:18.000000000","message":"Patch Set 1:\n\nMisses a release note","accounts_in_message":[],"_revision_number":1},{"id":"838adb5c0a415c88845e5cc0176bcbc14e92a625","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 21:17:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/00/667200/1/check/openstack-tox-pep8/b3047e2/ : SUCCESS in 9m 05s\n- openstack-tox-py27 http://logs.openstack.org/00/667200/1/check/openstack-tox-py27/98e2012/ : SUCCESS in 4m 51s\n- openstack-tox-py35 http://logs.openstack.org/00/667200/1/check/openstack-tox-py35/6e9533f/ : SUCCESS in 5m 05s\n- openstack-tox-docs http://logs.openstack.org/00/667200/1/check/openstack-tox-docs/135aade/ : FAILURE in 3m 40s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/00/667200/1/check/neutron-lbaasv2-dsvm-api/a36eff4/ : SUCCESS in 1h 17m 43s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/00/667200/1/check/neutron-lbaasv2-dsvm-py3x-api/b179ad4/ : SUCCESS in 1h 16m 14s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/00/667200/1/check/neutron-lbaasv2-dsvm-scenario/fe76ead/ : SUCCESS in 1h 11m 07s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/00/667200/1/check/neutron-lbaasv2-dsvm-py3x-scenario/70db0bf/ : SUCCESS in 1h 16m 48s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/00/667200/1/check/openstack-tox-cover/e523605/cover/ : SUCCESS in 8m 16s\n- openstack-tox-functional http://logs.openstack.org/00/667200/1/check/openstack-tox-functional/117325b/ : SUCCESS in 6m 36s\n- openstack-tox-functional-py35 http://logs.openstack.org/00/667200/1/check/openstack-tox-functional-py35/07dceff/ : SUCCESS in 6m 38s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/00/667200/1/check/octavia-v1-dsvm-scenario/d2a4952/ : SUCCESS in 59m 43s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/00/667200/1/check/octavia-v1-dsvm-py3x-scenario/065cd08/ : SUCCESS in 54m 20s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/00/667200/1/check/octavia-v1-dsvm-scenario-multinode/5d6bd81/ : SUCCESS in 1h 02m 26s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/00/667200/1/check/octavia-v1-dsvm-py3x-scenario-multinode/b22ae7b/ : SUCCESS in 1h 09m 01s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/00/667200/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/835f01d/ : SUCCESS in 1h 18m 22s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/00/667200/1/check/octavia-v2-dsvm-noop-api/b12c3cf/ : SUCCESS in 48m 19s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/00/667200/1/check/octavia-v2-dsvm-noop-py35-api/c5810af/ : SUCCESS in 43m 13s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/00/667200/1/check/octavia-v2-dsvm-scenario/f2a74e0/ : SUCCESS in 1h 49m 18s\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/00/667200/1/check/octavia-v2-dsvm-py35-scenario/6767d72/ : FAILURE in 1h 28m 09s","accounts_in_message":[],"_revision_number":1},{"id":"fac5970c37090dd349a3a9dfa32371076e5d9a04","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-24 22:21:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3ac3f99fe94fac9a31791c49eee8e59fd8a19763","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-24 22:29:59.000000000","message":"Patch Set 3: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"56d26268e128c8beda5e4e6693eea723cbda2b1a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-24 22:33:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f9da64cd37ad0a7c7d08b391c01dcb0e91081841","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-24 22:40:19.000000000","message":"Patch Set 4:\n\n[vagrant@localhost ~]$ openstack loadbalancer show lb1\n+---------------------+--------------------------------------+\n| Field               | Value                                |\n+---------------------+--------------------------------------+\n| admin_state_up      | True                                 |\n| created_at          | 2019-06-24T22:37:25                  |\n| description         |                                      |\n| flavor_id           |                                      |\n| id                  | f7afb7a6-64af-4aca-9bd6-536900a57f90 |\n| listeners           |                                      |\n| name                | lb1                                  |\n| operating_status    | ONLINE                               |\n| pools               |                                      |\n| project_id          | 5b785708ec424f898c7940d6e01b44af     |\n| provider            | octavia                              |\n| provisioning_status | ACTIVE                               |\n| updated_at          | 2019-06-24T22:39:19                  |\n| vip_address         | 10.0.0.7                             |\n| vip_network_id      | ba4cd315-9f36-411b-96af-6151acdf0147 |\n| vip_port_id         | 7ce24883-1e34-4501-bbc6-1a55d5c9db29 |\n| vip_qos_policy_id   | None                                 |\n| vip_subnet_id       | 1560efed-db95-4c10-a12e-b6b4fe1365cb |\n+---------------------+--------------------------------------+\n[vagrant@localhost ~]$ openstack loadbalancer listener create --protocol-port 443 --protocol TERMINATED_HTTPS --name listener_bad_cn --default-tls-container\u003d$(openstack secret list | awk \u0027/ tls_secret_cg_bad_cn / {print $2}\u0027) lb1\nCould not retrieve certificate: [\u0027http://192.168.121.22/key-manager/v1/secrets/80a409e3-941d-429b-9569-739fbb5cb105\u0027] (HTTP 400) (Request-ID: req-5925af45-5b84-49f6-a9af-cce61203efaa)","accounts_in_message":[],"_revision_number":4},{"id":"75c86627d3ac83d841004c4a0730d89725f83a18","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-24 22:40:56.000000000","message":"Topic changed from story/2005925-queens to story/2005925","accounts_in_message":[],"_revision_number":4},{"id":"6b37ce2680c8754fc782368aba2c569242c3ada6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-24 23:55:19.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"78c80eb1cd1df4b85a4dcf54b5ac1d3f19c568b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 02:32:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/00/667200/4/check/openstack-tox-pep8/c90405a/ : SUCCESS in 8m 25s\n- openstack-tox-py27 http://logs.openstack.org/00/667200/4/check/openstack-tox-py27/f5ef525/ : SUCCESS in 5m 02s\n- openstack-tox-py35 http://logs.openstack.org/00/667200/4/check/openstack-tox-py35/c71d00b/ : SUCCESS in 5m 32s\n- openstack-tox-docs http://logs.openstack.org/00/667200/4/check/openstack-tox-docs/d237db2/html/ : SUCCESS in 5m 41s\n- build-openstack-releasenotes http://logs.openstack.org/00/667200/4/check/build-openstack-releasenotes/1a2b176/html/ : SUCCESS in 3m 45s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/00/667200/4/check/neutron-lbaasv2-dsvm-api/5c14bfe/ : SUCCESS in 1h 18m 56s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/00/667200/4/check/neutron-lbaasv2-dsvm-py3x-api/a28dd47/ : SUCCESS in 1h 22m 17s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/00/667200/4/check/neutron-lbaasv2-dsvm-scenario/10d1147/ : SUCCESS in 1h 15m 14s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/00/667200/4/check/neutron-lbaasv2-dsvm-py3x-scenario/94ddce1/ : SUCCESS in 1h 14m 35s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/00/667200/4/check/openstack-tox-cover/2a1d177/cover/ : SUCCESS in 8m 38s\n- openstack-tox-functional http://logs.openstack.org/00/667200/4/check/openstack-tox-functional/6748984/ : SUCCESS in 5m 35s\n- openstack-tox-functional-py35 http://logs.openstack.org/00/667200/4/check/openstack-tox-functional-py35/0953c05/ : SUCCESS in 6m 19s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-scenario/ccd005e/ : SUCCESS in 55m 31s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-py3x-scenario/320ce41/ : SUCCESS in 1h 12m 33s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-scenario-multinode/5798801/ : SUCCESS in 1h 16m 35s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-py3x-scenario-multinode/370046f/ : SUCCESS in 1h 17m 13s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-scenario-kvm-centos.7/9290f44/ : SUCCESS in 1h 13m 59s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/00/667200/4/check/octavia-v2-dsvm-noop-api/61346a0/ : SUCCESS in 48m 52s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/00/667200/4/check/octavia-v2-dsvm-noop-py35-api/4fd3543/ : SUCCESS in 45m 29s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/00/667200/4/check/octavia-v2-dsvm-scenario/295eb97/ : FAILURE in 1h 31m 59s\n- octavia-v2-dsvm-py35-scenario finger://ze01.openstack.org/746ab12226064c8c91e1de9e4844b1bb : RETRY_LIMIT in 1h 15m 38s","accounts_in_message":[],"_revision_number":4},{"id":"adf50881cf7b5103ded5bfd9128765756295a202","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-06-25 08:41:56.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"56d64e87e7b66c105669b42e3e5938a790fb105f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-06-25 13:08:52.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"01439fe776939c9297717d2b3f47df9ebba420e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 13:12:55.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/00/667200/4/check/openstack-tox-pep8/fb0c645/ : SUCCESS in 9m 00s\n- openstack-tox-py27 http://logs.openstack.org/00/667200/4/check/openstack-tox-py27/09fde2e/ : SUCCESS in 4m 46s\n- openstack-tox-py35 http://logs.openstack.org/00/667200/4/check/openstack-tox-py35/9df2723/ : SUCCESS in 6m 52s\n- openstack-tox-docs http://logs.openstack.org/00/667200/4/check/openstack-tox-docs/3b3d29b/html/ : SUCCESS in 6m 43s\n- build-openstack-releasenotes http://logs.openstack.org/00/667200/4/check/build-openstack-releasenotes/20e29ca/html/ : SUCCESS in 4m 44s\n- neutron-lbaasv2-dsvm-api finger://ze11.openstack.org/a6f4b060f0f64f6fb32dc5910a99fc76 : RETRY_LIMIT in 52m 20s\n- neutron-lbaasv2-dsvm-py3x-api finger://ze01.openstack.org/b739bf60871a463a80e87ad5e575bf64 : RETRY_LIMIT in 54m 01s\n- neutron-lbaasv2-dsvm-scenario finger://ze04.openstack.org/2f40caf39714460691ddf53a76fc71de : RETRY_LIMIT in 29m 32s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario finger://ze11.openstack.org/f16da8df1c524f7782e8cf7106b13ebe : RETRY_LIMIT in 1h 01m 43s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/00/667200/4/check/openstack-tox-cover/48f2c65/cover/ : SUCCESS in 8m 16s\n- openstack-tox-functional http://logs.openstack.org/00/667200/4/check/openstack-tox-functional/7f25de8/ : SUCCESS in 6m 05s\n- openstack-tox-functional-py35 http://logs.openstack.org/00/667200/4/check/openstack-tox-functional-py35/415c77a/ : SUCCESS in 6m 16s\n- octavia-v1-dsvm-scenario finger://ze10.openstack.org/a3f0358912d54e90a312bff567b4f596 : RETRY_LIMIT in 59m 28s\n- octavia-v1-dsvm-py3x-scenario finger://ze06.openstack.org/c3410f085e3a4a8aaeec4fe70736f65a : RETRY_LIMIT in 46m 56s\n- octavia-v1-dsvm-scenario-multinode finger://ze01.openstack.org/712f36bee52345ecb5cc1a42d4076772 : RETRY_LIMIT in 47m 07s\n- octavia-v1-dsvm-py3x-scenario-multinode finger://ze02.openstack.org/5773516ed4ef49d7a761a92e7d82cf73 : RETRY_LIMIT in 36m 00s\n- octavia-v1-dsvm-scenario-kvm-centos.7 finger://ze01.openstack.org/e25010e25d2743a6bca7f479b9ecce9e : RETRY_LIMIT in 45m 07s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/00/667200/4/check/octavia-v2-dsvm-noop-api/70f8395/ : POST_FAILURE in 51m 10s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/00/667200/4/check/octavia-v2-dsvm-noop-py35-api/bb6dc9c/ : SUCCESS in 43m 18s\n- octavia-v2-dsvm-scenario finger://ze06.openstack.org/6f329fc0c57148acb737db4a5d6b9199 : RETRY_LIMIT in 59m 26s\n- octavia-v2-dsvm-py35-scenario finger://ze02.openstack.org/c3e29f3943454d4c97a238373619cb68 : RETRY_LIMIT in 51m 53s","accounts_in_message":[],"_revision_number":4},{"id":"2a001f08b5a75075bcca9339d1ab20a2f655084c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-25 13:19:59.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"cedd4474af949200456a9ec0055f3bc3270ef8f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 18:19:13.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/00/667200/4/check/openstack-tox-pep8/c4434ed/ : SUCCESS in 9m 54s\n- openstack-tox-py27 http://logs.openstack.org/00/667200/4/check/openstack-tox-py27/939a2b3/ : SUCCESS in 4m 46s\n- openstack-tox-py35 http://logs.openstack.org/00/667200/4/check/openstack-tox-py35/460afbd/ : SUCCESS in 5m 33s\n- openstack-tox-docs http://logs.openstack.org/00/667200/4/check/openstack-tox-docs/918e9b9/html/ : SUCCESS in 5m 33s\n- build-openstack-releasenotes http://logs.openstack.org/00/667200/4/check/build-openstack-releasenotes/92c7674/html/ : SUCCESS in 4m 13s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/00/667200/4/check/neutron-lbaasv2-dsvm-api/629b5dd/ : SUCCESS in 1h 21m 43s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/00/667200/4/check/neutron-lbaasv2-dsvm-py3x-api/7f1f7da/ : SUCCESS in 1h 21m 08s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/00/667200/4/check/neutron-lbaasv2-dsvm-scenario/0fb0dcc/ : SUCCESS in 1h 17m 07s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/00/667200/4/check/neutron-lbaasv2-dsvm-py3x-scenario/bddc678/ : SUCCESS in 1h 18m 06s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/00/667200/4/check/openstack-tox-cover/0e90756/cover/ : SUCCESS in 7m 30s\n- openstack-tox-functional http://logs.openstack.org/00/667200/4/check/openstack-tox-functional/ccd3837/ : SUCCESS in 6m 00s\n- openstack-tox-functional-py35 http://logs.openstack.org/00/667200/4/check/openstack-tox-functional-py35/0bba631/ : SUCCESS in 6m 37s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-scenario/07230ab/ : SUCCESS in 1h 04m 25s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-py3x-scenario/96f997e/ : SUCCESS in 57m 22s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-scenario-multinode/254f65f/ : SUCCESS in 1h 10m 43s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-py3x-scenario-multinode/7464cb9/ : SUCCESS in 1h 09m 53s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-scenario-kvm-centos.7/be988b6/ : SUCCESS in 1h 09m 55s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/00/667200/4/check/octavia-v2-dsvm-noop-api/42db9eb/ : SUCCESS in 45m 32s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/00/667200/4/check/octavia-v2-dsvm-noop-py35-api/f79a92b/ : SUCCESS in 43m 10s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/00/667200/4/check/octavia-v2-dsvm-scenario/8251dd3/ : FAILURE in 1h 26m 33s\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/00/667200/4/check/octavia-v2-dsvm-py35-scenario/d57cff2/ : SUCCESS in 1h 21m 25s","accounts_in_message":[],"_revision_number":4},{"id":"472b43f76a5babd1a384848baa06bf68c99c2cbf","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-25 18:21:30.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c6c3d95f888dbd6df540e0794c29d875c27bd67a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 20:16:01.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/00/667200/4/check/openstack-tox-pep8/afb34fb/ : SUCCESS in 9m 00s\n- openstack-tox-py27 http://logs.openstack.org/00/667200/4/check/openstack-tox-py27/f97afe1/ : SUCCESS in 5m 30s\n- openstack-tox-py35 http://logs.openstack.org/00/667200/4/check/openstack-tox-py35/b3871da/ : SUCCESS in 5m 18s\n- openstack-tox-docs http://logs.openstack.org/00/667200/4/check/openstack-tox-docs/3298795/html/ : SUCCESS in 5m 51s\n- build-openstack-releasenotes http://logs.openstack.org/00/667200/4/check/build-openstack-releasenotes/5d2df9d/html/ : SUCCESS in 5m 12s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/00/667200/4/check/neutron-lbaasv2-dsvm-api/ff1ed57/ : SUCCESS in 1h 27m 19s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/00/667200/4/check/neutron-lbaasv2-dsvm-py3x-api/0a61cfe/ : SUCCESS in 1h 21m 17s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/00/667200/4/check/neutron-lbaasv2-dsvm-scenario/866c8e7/ : SUCCESS in 1h 34m 04s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/00/667200/4/check/neutron-lbaasv2-dsvm-py3x-scenario/bb80fb1/ : SUCCESS in 1h 20m 02s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/00/667200/4/check/openstack-tox-cover/2046ff9/cover/ : SUCCESS in 8m 50s\n- openstack-tox-functional http://logs.openstack.org/00/667200/4/check/openstack-tox-functional/50e4220/ : SUCCESS in 6m 12s\n- openstack-tox-functional-py35 http://logs.openstack.org/00/667200/4/check/openstack-tox-functional-py35/2c60cd6/ : SUCCESS in 6m 51s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-scenario/1628cee/ : SUCCESS in 1h 02m 52s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-py3x-scenario/4fc9ba2/ : SUCCESS in 1h 06m 29s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-scenario-multinode/e9f9aac/ : SUCCESS in 1h 16m 45s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-py3x-scenario-multinode/cb347f6/ : SUCCESS in 1h 11m 27s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-scenario-kvm-centos.7/162036a/ : SUCCESS in 1h 22m 11s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/00/667200/4/check/octavia-v2-dsvm-noop-api/d725967/ : SUCCESS in 42m 30s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/00/667200/4/check/octavia-v2-dsvm-noop-py35-api/771b1d5/ : SUCCESS in 42m 54s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/00/667200/4/check/octavia-v2-dsvm-scenario/271cac9/ : SUCCESS in 1h 21m 21s\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/00/667200/4/check/octavia-v2-dsvm-py35-scenario/15a647a/ : FAILURE in 1h 38m 42s","accounts_in_message":[],"_revision_number":4},{"id":"db3504138734db7e21a670bcdaffc63df0c436d2","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-06-26 11:37:41.000000000","message":"Patch Set 4: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"8b3301781251f1776b7f40fe106b7c1d34d1c408","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 13:15:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/00/667200/4/check/openstack-tox-pep8/e2bd5ff/ : SUCCESS in 9m 42s\n- openstack-tox-py27 http://logs.openstack.org/00/667200/4/check/openstack-tox-py27/0421610/ : SUCCESS in 5m 07s\n- openstack-tox-py35 http://logs.openstack.org/00/667200/4/check/openstack-tox-py35/eca5f78/ : SUCCESS in 5m 10s\n- openstack-tox-docs http://logs.openstack.org/00/667200/4/check/openstack-tox-docs/268b72b/html/ : SUCCESS in 5m 23s\n- build-openstack-releasenotes http://logs.openstack.org/00/667200/4/check/build-openstack-releasenotes/6869790/html/ : SUCCESS in 3m 57s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/00/667200/4/check/neutron-lbaasv2-dsvm-api/579ed2f/ : SUCCESS in 1h 19m 18s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/00/667200/4/check/neutron-lbaasv2-dsvm-py3x-api/782199c/ : SUCCESS in 1h 16m 04s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/00/667200/4/check/neutron-lbaasv2-dsvm-scenario/6c4a6ef/ : SUCCESS in 1h 30m 49s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/00/667200/4/check/neutron-lbaasv2-dsvm-py3x-scenario/57f79ca/ : SUCCESS in 1h 18m 14s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/00/667200/4/check/openstack-tox-cover/aaf03b1/cover/ : SUCCESS in 8m 02s\n- openstack-tox-functional http://logs.openstack.org/00/667200/4/check/openstack-tox-functional/9abfb44/ : SUCCESS in 6m 04s\n- openstack-tox-functional-py35 http://logs.openstack.org/00/667200/4/check/openstack-tox-functional-py35/679a49c/ : SUCCESS in 6m 40s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-scenario/2ca43ec/ : SUCCESS in 1h 00m 21s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-py3x-scenario/f33592b/ : SUCCESS in 1h 05m 36s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-scenario-multinode/2e3d70e/ : SUCCESS in 1h 12m 07s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-py3x-scenario-multinode/3a725cf/ : SUCCESS in 1h 09m 51s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/00/667200/4/check/octavia-v1-dsvm-scenario-kvm-centos.7/674db49/ : SUCCESS in 1h 17m 30s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/00/667200/4/check/octavia-v2-dsvm-noop-api/57f74f2/ : SUCCESS in 46m 00s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/00/667200/4/check/octavia-v2-dsvm-noop-py35-api/9af967a/ : SUCCESS in 43m 33s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/00/667200/4/check/octavia-v2-dsvm-scenario/fb32ae8/ : SUCCESS in 1h 29m 37s\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/00/667200/4/check/octavia-v2-dsvm-py35-scenario/a29b7d9/ : SUCCESS in 1h 26m 51s","accounts_in_message":[],"_revision_number":4},{"id":"5df745303b4c2b56152e48823101302a27028cc9","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-06-27 01:15:25.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"664e8fef18ac5324769e6bf39d08e9c31f9469ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 01:15:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a2bd6c56f84a52ffffe88cdb65876fb4ca8980f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 03:03:57.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/00/667200/4/gate/openstack-tox-pep8/4bdddc1/ : SUCCESS in 9m 26s\n- openstack-tox-py27 http://logs.openstack.org/00/667200/4/gate/openstack-tox-py27/12f3d18/ : SUCCESS in 4m 51s\n- openstack-tox-py35 http://logs.openstack.org/00/667200/4/gate/openstack-tox-py35/bf60f78/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/00/667200/4/gate/openstack-tox-docs/f522424/html/ : SUCCESS in 4m 48s\n- build-openstack-releasenotes http://logs.openstack.org/00/667200/4/gate/build-openstack-releasenotes/ac2aef9/html/ : SUCCESS in 5m 02s\n- openstack-tox-cover http://logs.openstack.org/00/667200/4/gate/openstack-tox-cover/d58a4cb/cover/ : SUCCESS in 7m 52s\n- openstack-tox-functional http://logs.openstack.org/00/667200/4/gate/openstack-tox-functional/6eba14e/ : SUCCESS in 5m 28s\n- openstack-tox-functional-py35 http://logs.openstack.org/00/667200/4/gate/openstack-tox-functional-py35/18bf12a/ : SUCCESS in 6m 26s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/00/667200/4/gate/neutron-lbaasv2-dsvm-api/07118de/ : SUCCESS in 1h 17m 27s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/00/667200/4/gate/neutron-lbaasv2-dsvm-py3x-api/0a29905/ : SUCCESS in 1h 22m 41s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/00/667200/4/gate/octavia-v1-dsvm-scenario/9a6d43d/ : SUCCESS in 58m 45s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/00/667200/4/gate/octavia-v1-dsvm-py3x-scenario/94ee5c7/ : SUCCESS in 1h 01m 48s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/00/667200/4/gate/octavia-v1-dsvm-scenario-multinode/0bb5155/ : SUCCESS in 1h 08m 56s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/00/667200/4/gate/octavia-v1-dsvm-py3x-scenario-multinode/84ccce8/ : SUCCESS in 1h 11m 25s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/00/667200/4/gate/octavia-v2-dsvm-noop-api/f50c27e/ : SUCCESS in 44m 10s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/00/667200/4/gate/octavia-v2-dsvm-noop-py35-api/29c3cdc/ : SUCCESS in 44m 28s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/00/667200/4/gate/octavia-v2-dsvm-scenario/b95d031/ : SUCCESS in 1h 28m 10s\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/00/667200/4/gate/octavia-v2-dsvm-py35-scenario/8cec3fa/ : SUCCESS in 1h 41m 55s","accounts_in_message":[],"_revision_number":4},{"id":"f90d2936853c240cb1ee64a630b2607b5f1e197b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 03:03:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"26d8fde69fca888e90f7ea3292f3f1d34d7c002b","revisions":{"0e9e4e4f3dec41b30000789172850cb20f696306":{"kind":"REWORK","_number":1,"created":"2019-06-24 19:21:23.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/00/667200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/667200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/667200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/667200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/667200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/667200/1"}}},"commit":{"parents":[{"commit":"a96b00b4dfdd6ac3643149149ed5141a9fbc68cd","subject":"Fix allocate_and_associate DB deadlock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a96b00b4dfdd6ac3643149149ed5141a9fbc68cd"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-24 18:56:50.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-24 19:21:01.000000000","tz":120},"subject":"Validate certificate content at API level","message":"Validate certificate content at API level\n\nStarting from Rocky, certificates are loaded still at API level when\nconverting objects to provider data models. The act of loading the\ncertificate provides validation as to its content. For example, it\nchecks if a value in the Common Name field is set. When the content is\npassed in on create and update actions via reference, it is checked at\nAPI level. If it\u0027s invalid, it fails right there and an error is\nreturned to the user.\n\nAltough, certificate content is not checked at API level in Queens.\nShould an invalid certificate be passed in, the API accepts but it will\nlater fail at provisioning -- the listener and loadbalancer go into\nERROR. The problem starts when the health manager runs the periodic\nupdate health check. It calculates the expected number of listeners and\nsees the listener in ERROR. In an attempt to self-heal it, an amphora\nfailover is triggered. As it runs the failover, it tries, again, to load\nup the invalid certificate. Amphora failover goes on in a loop.\n\nThis patch is a Queens-only patch.\n\nTask: 34261\nStory: 2005925\n\nChange-Id: Ib0fb3db3b2d56345eeeda2014c15e35281bcfec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e9e4e4f3dec41b30000789172850cb20f696306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e9e4e4f3dec41b30000789172850cb20f696306"}]},"branch":"refs/heads/stable/queens"},"2127c33b7be0ead62e5c757aac44aebefe415b10":{"kind":"REWORK","_number":2,"created":"2019-06-24 22:21:38.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/00/667200/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/667200/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/667200/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/667200/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/667200/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/667200/2"}}},"commit":{"parents":[{"commit":"a96b00b4dfdd6ac3643149149ed5141a9fbc68cd","subject":"Fix allocate_and_associate DB deadlock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a96b00b4dfdd6ac3643149149ed5141a9fbc68cd"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-24 18:56:50.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-24 22:19:37.000000000","tz":120},"subject":"Validate certificate content at API level","message":"Validate certificate content at API level\n\nStarting from Rocky, certificates are loaded still at API level when\nconverting objects to provider data models. The act of loading the\ncertificate provides validation as to its content. For example, it\nchecks if a value in the Common Name field is set. When the content is\npassed in on create and update actions via reference, it is checked at\nAPI level. If it\u0027s invalid, it fails right there and an error is\nreturned to the user.\n\nAlthough, certificate content is not checked at API level in Queens.\nShould an invalid certificate be passed in, the API accepts but it will\nlater fail at provisioning -- the listener and loadbalancer go into\nERROR. The problem starts when the health manager runs the periodic\nupdate health check. It calculates the expected number of listeners and\nsees the listener in ERROR. In an attempt to heal it, an amphora\nfailover is triggered. As it runs the failover, it tries, again, to load\nup the invalid certificate. Amphora failover goes on in a loop.\n\nThis patch is a Queens-only patch.\n\nTask: 34261\nStory: 2005925\n\nChange-Id: Ib0fb3db3b2d56345eeeda2014c15e35281bcfec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2127c33b7be0ead62e5c757aac44aebefe415b10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2127c33b7be0ead62e5c757aac44aebefe415b10"}]},"branch":"refs/heads/stable/queens"},"59fee85e30c20b0dfa70ebc42a890c02c47d3082":{"kind":"REWORK","_number":3,"created":"2019-06-24 22:29:59.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/00/667200/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/667200/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/667200/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/667200/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/667200/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/667200/3"}}},"commit":{"parents":[{"commit":"ae505577aaca4ceed2790a1d939b42c658d07809","subject":"Add bindep.txt for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ae505577aaca4ceed2790a1d939b42c658d07809"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-24 18:56:50.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-24 22:29:59.000000000","tz":0},"subject":"Validate certificate content at API level","message":"Validate certificate content at API level\n\nStarting from Rocky, certificates are loaded still at API level when\nconverting objects to provider data models. The act of loading the\ncertificate provides validation as to its content. For example, it\nchecks if a value in the Common Name field is set. When the content is\npassed in on create and update actions via reference, it is checked at\nAPI level. If it\u0027s invalid, it fails right there and an error is\nreturned to the user.\n\nAltough, certificate content is not checked at API level in Queens.\nShould an invalid certificate be passed in, the API accepts but it will\nlater fail at provisioning -- the listener and loadbalancer go into\nERROR. The problem starts when the health manager runs the periodic\nupdate health check. It calculates the expected number of listeners and\nsees the listener in ERROR. In an attempt to self-heal it, an amphora\nfailover is triggered. As it runs the failover, it tries, again, to load\nup the invalid certificate. Amphora failover goes on in a loop.\n\nThis patch is a Queens-only patch.\n\nTask: 34261\nStory: 2005925\n\nChange-Id: Ib0fb3db3b2d56345eeeda2014c15e35281bcfec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/59fee85e30c20b0dfa70ebc42a890c02c47d3082"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/59fee85e30c20b0dfa70ebc42a890c02c47d3082"}]},"branch":"refs/heads/stable/queens"},"26d8fde69fca888e90f7ea3292f3f1d34d7c002b":{"kind":"REWORK","_number":4,"created":"2019-06-24 22:33:08.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/00/667200/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/667200/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/667200/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/667200/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/667200/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/667200/4"}}},"commit":{"parents":[{"commit":"d9c459a8686817bdaf4967ffb45b2c180a5e4265","subject":"Add bindep.txt for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d9c459a8686817bdaf4967ffb45b2c180a5e4265"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-24 18:56:50.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-24 22:32:42.000000000","tz":120},"subject":"Validate certificate content at API level","message":"Validate certificate content at API level\n\nStarting from Rocky, certificates are loaded still at API level when\nconverting objects to provider data models. The act of loading the\ncertificate provides validation as to its content. For example, it\nchecks if a value in the Common Name field is set. When the content is\npassed in on create and update actions via reference, it is checked at\nAPI level. If it\u0027s invalid, it fails right there and an error is\nreturned to the user.\n\nAlthough, certificate content is not checked at API level in Queens.\nShould an invalid certificate be passed in, the API accepts but it will\nlater fail at provisioning -- the listener and loadbalancer go into\nERROR. The problem starts when the health manager runs the periodic\nupdate health check. It calculates the expected number of listeners and\nsees the listener in ERROR. In an attempt to heal it, an amphora\nfailover is triggered. As it runs the failover, it tries, again, to load\nup the invalid certificate. Amphora failover goes on in a loop.\n\nThis patch is a Queens-only patch.\n\nTask: 34261\nStory: 2005925\n\nChange-Id: Ib0fb3db3b2d56345eeeda2014c15e35281bcfec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26d8fde69fca888e90f7ea3292f3f1d34d7c002b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26d8fde69fca888e90f7ea3292f3f1d34d7c002b"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
