)]}'
{"id":"openstack%2Foctavia~667208","triplet_id":"openstack%2Foctavia~master~I48e82f39e3980d6d50e3b710326b8e437eddb08b","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I48e82f39e3980d6d50e3b710326b8e437eddb08b","subject":"Add bindep.txt for Octavia","status":"MERGED","created":"2019-06-24 19:59:13.000000000","updated":"2019-06-25 01:39:05.000000000","submitted":"2019-06-25 01:39:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667208-1561426745423-de56cdb0","meta_rev_id":"e5f3c1cd2ae9e48474964ede68dbafcac59c2a2d","_number":667208,"virtual_id_number":667208,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-06-25 01:39:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-06-24 22:35:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-06-24 22:34:58.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-06-24 22:35:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-06-24 22:35:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 22:34:58.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-06-24 22:35:35.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-06-25 01:39:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"27af78fbf03c1bb7b354a864a8599438dd8541db","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-24 19:59:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc288d2a996f6c98fd92f90ec3d44ba3ea784025","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-24 22:28:12.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit d53010dc423867bc965069206b007a62fbb02475","accounts_in_message":[],"_revision_number":1},{"id":"5748c9bf05d06d5b99f7b4b2aa599fc9626b63d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 22:29:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/08/667208/1/check/build-openstack-api-ref/4c871d5/html/ : SUCCESS in 4m 20s\n- openstack-tox-cover http://logs.openstack.org/08/667208/1/check/openstack-tox-cover/984dc73/cover/ : SUCCESS in 6m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/667208/1/check/openstack-tox-lower-constraints/6898b0d/ : SUCCESS in 6m 24s\n- openstack-tox-pep8 http://logs.openstack.org/08/667208/1/check/openstack-tox-pep8/87bef36/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/08/667208/1/check/openstack-tox-py27/29c32b3/ : SUCCESS in 4m 17s\n- openstack-tox-py36 http://logs.openstack.org/08/667208/1/check/openstack-tox-py36/efb342e/ : SUCCESS in 4m 33s\n- openstack-tox-py37 http://logs.openstack.org/08/667208/1/check/openstack-tox-py37/8141768/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/08/667208/1/check/openstack-tox-docs/ff60d76/html/ : SUCCESS in 6m 14s\n- build-openstack-releasenotes http://logs.openstack.org/08/667208/1/check/build-openstack-releasenotes/4571245/html/ : SUCCESS in 4m 17s\n- openstack-tox-functional http://logs.openstack.org/08/667208/1/check/openstack-tox-functional/2a118dd/ : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 http://logs.openstack.org/08/667208/1/check/openstack-tox-functional-py36/3b2bccc/ : SUCCESS in 6m 43s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/08/667208/1/check/octavia-v2-dsvm-noop-api/5f957ac/ : SUCCESS in 43m 44s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/08/667208/1/check/octavia-v2-dsvm-noop-py2-api/f339f2f/ : SUCCESS in 44m 00s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/08/667208/1/check/octavia-v2-dsvm-scenario/48b7044/ : SUCCESS in 1h 34m 10s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/08/667208/1/check/octavia-v2-dsvm-py2-scenario/33f6042/ : SUCCESS in 1h 27m 52s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/08/667208/1/check/octavia-v2-dsvm-py2-scenario-centos-7/2950ee2/ : FAILURE in 2h 20m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/08/667208/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/903cce7/ : SUCCESS in 1h 56m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/08/667208/1/check/octavia-v2-dsvm-scenario-two-node/e8ee91e/ : SUCCESS in 1h 45m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/08/667208/1/check/octavia-v2-dsvm-py2-scenario-two-node/f87fa20/ : SUCCESS in 1h 56m 34s (non-voting)\n- octavia-grenade http://logs.openstack.org/08/667208/1/check/octavia-grenade/ad11cbf/ : SUCCESS in 1h 22m 47s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/08/667208/1/check/octavia-v2-dsvm-tls-barbican/27500f0/ : SUCCESS in 53m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/08/667208/1/check/octavia-v2-dsvm-scenario-amphora-v2/3411a6f/ : SUCCESS in 1h 30m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92f211542c06c67f93902ba1fdf197b0de81d8b5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-24 22:34:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"05bf42297931d209b7dbe3543d3aaa7958be74b7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-24 22:35:35.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"44cb288b0339e01bd9189362927d7cc0a27f8933","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 22:35:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"05ba281062c6a7af0cd5cd92852ba83b2b46b235","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 01:39:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/08/667208/1/gate/build-openstack-api-ref/df5607e/html/ : SUCCESS in 3m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/667208/1/gate/openstack-tox-lower-constraints/5cda022/ : SUCCESS in 5m 39s\n- openstack-tox-pep8 http://logs.openstack.org/08/667208/1/gate/openstack-tox-pep8/88e9ca9/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/08/667208/1/gate/openstack-tox-py27/a78eaef/ : SUCCESS in 4m 27s\n- openstack-tox-py36 http://logs.openstack.org/08/667208/1/gate/openstack-tox-py36/119b3b9/ : SUCCESS in 4m 45s\n- openstack-tox-py37 http://logs.openstack.org/08/667208/1/gate/openstack-tox-py37/28353e9/ : SUCCESS in 6m 04s\n- openstack-tox-docs http://logs.openstack.org/08/667208/1/gate/openstack-tox-docs/b3f2867/html/ : SUCCESS in 4m 57s\n- build-openstack-releasenotes http://logs.openstack.org/08/667208/1/gate/build-openstack-releasenotes/ac4204d/html/ : SUCCESS in 4m 16s\n- openstack-tox-functional http://logs.openstack.org/08/667208/1/gate/openstack-tox-functional/ad5c561/ : SUCCESS in 5m 20s\n- openstack-tox-functional-py36 http://logs.openstack.org/08/667208/1/gate/openstack-tox-functional-py36/74b53c7/ : SUCCESS in 5m 13s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/08/667208/1/gate/octavia-v2-dsvm-noop-api/624cb03/ : SUCCESS in 46m 33s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/08/667208/1/gate/octavia-v2-dsvm-noop-py2-api/0e7ee10/ : SUCCESS in 42m 35s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/08/667208/1/gate/octavia-v2-dsvm-scenario/2ea8c49/ : SUCCESS in 1h 32m 32s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/08/667208/1/gate/octavia-v2-dsvm-py2-scenario/13b149f/ : SUCCESS in 1h 36m 11s\n- octavia-grenade http://logs.openstack.org/08/667208/1/gate/octavia-grenade/9d2f914/ : SUCCESS in 1h 15m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e5f3c1cd2ae9e48474964ede68dbafcac59c2a2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 01:39:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"09733a0c39ad394a319f7b807518a5439ec0826e","revisions":{"09733a0c39ad394a319f7b807518a5439ec0826e":{"kind":"REWORK","_number":1,"created":"2019-06-24 19:59:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/667208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/667208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/667208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/667208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/667208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/667208/1"}}},"commit":{"parents":[{"commit":"3a9df30a9a4efd2a47919001dcf38d3377996ed7","subject":"Merge \"Fix pool API handling of None/null updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3a9df30a9a4efd2a47919001dcf38d3377996ed7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 19:57:55.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 19:59:12.000000000","tz":-420},"subject":"Add bindep.txt for Octavia","message":"Add bindep.txt for Octavia\n\nSome base packages are no longer automatically included in zuul\njob runs. This is good as we need to call out our platform package\nrequirements.\nThis patch adds a bindep.txt to Octavia to capture those\npackage requirements.\n\nChange-Id: I48e82f39e3980d6d50e3b710326b8e437eddb08b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09733a0c39ad394a319f7b807518a5439ec0826e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09733a0c39ad394a319f7b807518a5439ec0826e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
