)]}'
{"id":"openstack%2Foctavia~667210","triplet_id":"openstack%2Foctavia~master~Ida0d41c63bc6e57ee40b8eeef0c799c2f9c15427","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ida0d41c63bc6e57ee40b8eeef0c799c2f9c15427","subject":"DNM: bindep.txt test patch","status":"ABANDONED","created":"2019-06-24 20:00:16.000000000","updated":"2019-08-28 19:22:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5c3da8e707e2fb651f2c177ad86a18c0ddd603d6","_number":667210,"virtual_id_number":667210,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-06-24 22:23:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 22:23:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a26361028fd76cd957f2b19063f02c41e8aca989","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-24 20:00:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0204196f0bbda54c57b54b492714b8cdf440084c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 22:23:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/10/667210/1/check/openstack-tox-cover/df96859/cover/ : SUCCESS in 6m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/667210/1/check/openstack-tox-lower-constraints/a8a617f/ : SUCCESS in 5m 50s\n- openstack-tox-pep8 http://logs.openstack.org/10/667210/1/check/openstack-tox-pep8/5c077be/ : SUCCESS in 5m 20s\n- openstack-tox-py27 http://logs.openstack.org/10/667210/1/check/openstack-tox-py27/8e874ad/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/10/667210/1/check/openstack-tox-py36/005afb4/ : SUCCESS in 4m 10s\n- openstack-tox-py37 http://logs.openstack.org/10/667210/1/check/openstack-tox-py37/cb83358/ : SUCCESS in 4m 49s\n- openstack-tox-docs http://logs.openstack.org/10/667210/1/check/openstack-tox-docs/fd5dc65/html/ : SUCCESS in 5m 20s\n- openstack-tox-functional http://logs.openstack.org/10/667210/1/check/openstack-tox-functional/3dddb8d/ : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 http://logs.openstack.org/10/667210/1/check/openstack-tox-functional-py36/c449553/ : SUCCESS in 5m 36s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/10/667210/1/check/octavia-v2-dsvm-noop-api/4917954/ : SUCCESS in 44m 00s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/10/667210/1/check/octavia-v2-dsvm-noop-py2-api/5028869/ : SUCCESS in 44m 49s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/10/667210/1/check/octavia-v2-dsvm-scenario/11dde01/ : SUCCESS in 1h 27m 50s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/10/667210/1/check/octavia-v2-dsvm-py2-scenario/c799c5a/ : SUCCESS in 1h 38m 28s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/10/667210/1/check/octavia-v2-dsvm-py2-scenario-centos-7/de62a9c/ : FAILURE in 2h 03m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/10/667210/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/a96a5e6/ : SUCCESS in 1h 49m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/10/667210/1/check/octavia-v2-dsvm-scenario-two-node/ecd8929/ : SUCCESS in 1h 59m 49s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/10/667210/1/check/octavia-v2-dsvm-py2-scenario-two-node/aec5a55/ : SUCCESS in 1h 50m 12s (non-voting)\n- octavia-grenade http://logs.openstack.org/10/667210/1/check/octavia-grenade/947c239/ : SUCCESS in 1h 20m 07s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/10/667210/1/check/octavia-v2-dsvm-tls-barbican/86a2b5d/ : SUCCESS in 56m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/10/667210/1/check/octavia-v2-dsvm-scenario-amphora-v2/34cab37/ : SUCCESS in 1h 30m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c3da8e707e2fb651f2c177ad86a18c0ddd603d6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-28 19:22:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"595c08871cf68a9f078f1cb768dbee43f5f553e5","revisions":{"595c08871cf68a9f078f1cb768dbee43f5f553e5":{"kind":"REWORK","_number":1,"created":"2019-06-24 20:00:16.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/10/667210/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/10/667210/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/667210/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/667210/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/667210/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/10/667210/1"}}},"commit":{"parents":[{"commit":"09733a0c39ad394a319f7b807518a5439ec0826e","subject":"Add bindep.txt for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09733a0c39ad394a319f7b807518a5439ec0826e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 19:59:59.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 19:59:59.000000000","tz":-420},"subject":"DNM: bindep.txt test patch","message":"DNM: bindep.txt test patch\n\nChange-Id: Ida0d41c63bc6e57ee40b8eeef0c799c2f9c15427\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/595c08871cf68a9f078f1cb768dbee43f5f553e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/595c08871cf68a9f078f1cb768dbee43f5f553e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
