)]}'
{"id":"openstack%2Foctavia~667237","triplet_id":"openstack%2Foctavia~stable%2Fstein~I48e82f39e3980d6d50e3b710326b8e437eddb08b","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I48e82f39e3980d6d50e3b710326b8e437eddb08b","subject":"Add bindep.txt for Octavia","status":"MERGED","created":"2019-06-24 22:28:12.000000000","updated":"2019-06-26 21:20:35.000000000","submitted":"2019-06-26 21:20:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667237-1561584035608-328d8c3f","meta_rev_id":"ffee59e4c5b9b5c1f5900b27fe657007ad4afc5d","_number":667237,"virtual_id_number":667237,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"date":"2019-06-26 17:11:16.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-06-26 21:20:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-06-24 22:52:36.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-06-26 07:06:43.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-06-26 07:06:43.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 22:52:36.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-06-26 07:06:43.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-06-26 17:11:16.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2019-06-26 21:20:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fcb91a57e5258acc08624472a1410a845524635f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-24 22:28:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e22555f39c90a49bac9a1dd203887ed8b9991acf","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-24 22:28:24.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 5bc9a788d953d96adafd297f81691e07a2597e15","accounts_in_message":[],"_revision_number":1},{"id":"817fcd5f5b704fdcdfc6ea83ef7ac0e01ae0dfd3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-24 22:52:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e68be3f92cee359cffd5cf87b8f7dc7379c74a1c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 02:10:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/37/667237/1/check/openstack-tox-cover/8944ad1/cover/ : SUCCESS in 8m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/37/667237/1/check/openstack-tox-lower-constraints/6bcdef3/ : SUCCESS in 5m 46s\n- openstack-tox-pep8 http://logs.openstack.org/37/667237/1/check/openstack-tox-pep8/ec8ab5d/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/37/667237/1/check/openstack-tox-py27/400713e/ : SUCCESS in 3m 59s\n- openstack-tox-py35 http://logs.openstack.org/37/667237/1/check/openstack-tox-py35/f7094c8/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/37/667237/1/check/openstack-tox-py36/2030916/ : SUCCESS in 4m 44s\n- openstack-tox-docs http://logs.openstack.org/37/667237/1/check/openstack-tox-docs/980b1c4/html/ : SUCCESS in 4m 46s\n- build-openstack-releasenotes http://logs.openstack.org/37/667237/1/check/build-openstack-releasenotes/8622350/html/ : SUCCESS in 4m 27s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/37/667237/1/check/neutron-lbaasv2-dsvm-api/64d459d/ : SUCCESS in 1h 14m 39s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/37/667237/1/check/neutron-lbaasv2-dsvm-py3x-api/0b4b289/ : SUCCESS in 1h 26m 59s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/37/667237/1/check/neutron-lbaasv2-dsvm-scenario/7c66566/ : SUCCESS in 1h 25m 13s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/37/667237/1/check/neutron-lbaasv2-dsvm-py3x-scenario/b1640d6/ : SUCCESS in 1h 24m 50s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/37/667237/1/check/openstack-tox-functional/8c83019/ : SUCCESS in 5m 20s\n- openstack-tox-functional-py35 http://logs.openstack.org/37/667237/1/check/openstack-tox-functional-py35/e4cb3a8/ : SUCCESS in 8m 15s\n- openstack-tox-functional-py36 http://logs.openstack.org/37/667237/1/check/openstack-tox-functional-py36/906e44d/ : SUCCESS in 6m 11s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/37/667237/1/check/octavia-v1-dsvm-scenario/7751488/ : SUCCESS in 1h 02m 31s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/37/667237/1/check/octavia-v1-dsvm-py3x-scenario/05b6498/ : SUCCESS in 1h 02m 33s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-noop-api/2440d91/ : POST_FAILURE in 1h 10m 24s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-noop-py2-api/98db1ba/ : SUCCESS in 46m 25s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-scenario/b8c5e96/ : SUCCESS in 1h 39m 23s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-py2-scenario/cecef9c/ : SUCCESS in 1h 33m 17s\n- octavia-v2-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-scenario-ubuntu-xenial/dd2d09e/ : SUCCESS in 1h 38m 50s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-py2-scenario-centos-7/0c511f6/ : FAILURE in 11m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/b5c71ed/ : SUCCESS in 1h 35m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-scenario-two-node/6936818/ : SUCCESS in 1h 57m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-py2-scenario-two-node/311b162/ : SUCCESS in 2h 09m 35s (non-voting)\n- octavia-grenade http://logs.openstack.org/37/667237/1/check/octavia-grenade/12793bb/ : SUCCESS in 1h 19m 12s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-tls-barbican/433a238/ : SUCCESS in 54m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72f8b9e30f066c5ad7e2d8af98fb1562f1aa78c9","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-06-26 07:06:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81dec8d6f350f7fa3bdaa1461f620aa2e573ae43","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-06-26 07:06:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"633fa43ebe4c813d5c718b02f74f87a1f2e6d6be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 09:28:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/37/667237/1/check/openstack-tox-cover/ecdb295/cover/ : SUCCESS in 9m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/37/667237/1/check/openstack-tox-lower-constraints/29c30ee/ : SUCCESS in 7m 50s\n- openstack-tox-pep8 http://logs.openstack.org/37/667237/1/check/openstack-tox-pep8/930b4e3/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/37/667237/1/check/openstack-tox-py27/cb9d7c7/ : SUCCESS in 3m 58s\n- openstack-tox-py35 http://logs.openstack.org/37/667237/1/check/openstack-tox-py35/9bfcf35/ : SUCCESS in 4m 31s\n- openstack-tox-py36 http://logs.openstack.org/37/667237/1/check/openstack-tox-py36/23a4b14/ : SUCCESS in 4m 08s\n- openstack-tox-docs http://logs.openstack.org/37/667237/1/check/openstack-tox-docs/c761b9a/html/ : SUCCESS in 5m 09s\n- build-openstack-releasenotes http://logs.openstack.org/37/667237/1/check/build-openstack-releasenotes/b0e3244/html/ : SUCCESS in 4m 08s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/37/667237/1/check/neutron-lbaasv2-dsvm-api/ea47d79/ : SUCCESS in 1h 18m 29s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/37/667237/1/check/neutron-lbaasv2-dsvm-py3x-api/bec249d/ : SUCCESS in 1h 18m 15s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/37/667237/1/check/neutron-lbaasv2-dsvm-scenario/d94064b/ : SUCCESS in 1h 25m 18s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/37/667237/1/check/neutron-lbaasv2-dsvm-py3x-scenario/f7e71c6/ : SUCCESS in 1h 24m 32s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/37/667237/1/check/openstack-tox-functional/a556e15/ : SUCCESS in 6m 09s\n- openstack-tox-functional-py35 http://logs.openstack.org/37/667237/1/check/openstack-tox-functional-py35/a5acde1/ : SUCCESS in 8m 10s\n- openstack-tox-functional-py36 http://logs.openstack.org/37/667237/1/check/openstack-tox-functional-py36/f9ee5ee/ : SUCCESS in 6m 24s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/37/667237/1/check/octavia-v1-dsvm-scenario/5d975fa/ : SUCCESS in 1h 06m 44s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/37/667237/1/check/octavia-v1-dsvm-py3x-scenario/ecee39a/ : SUCCESS in 1h 06m 25s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-noop-api/144634b/ : SUCCESS in 44m 05s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-noop-py2-api/7385ec0/ : SUCCESS in 42m 16s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-scenario/d5d308f/ : SUCCESS in 1h 27m 10s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-py2-scenario/3755278/ : SUCCESS in 1h 31m 08s\n- octavia-v2-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-scenario-ubuntu-xenial/1959325/ : SUCCESS in 1h 36m 30s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-py2-scenario-centos-7/7fc49a9/ : FAILURE in 2h 00m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/5fc3208/ : SUCCESS in 1h 36m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-scenario-two-node/285d7e7/ : SUCCESS in 1h 22m 36s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-py2-scenario-two-node/2dc1f32/ : SUCCESS in 1h 50m 11s (non-voting)\n- octavia-grenade http://logs.openstack.org/37/667237/1/check/octavia-grenade/e6d7d70/ : SUCCESS in 1h 08m 32s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-tls-barbican/93936a8/ : SUCCESS in 48m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c7cbc68fb0edcf3000df4c8bb3696ba66987b3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 09:28:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4cc124c9f921b588b5400691063ed140d42b32fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 12:37:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/37/667237/1/gate/openstack-tox-lower-constraints/be10692/ : SUCCESS in 6m 20s\n- openstack-tox-pep8 http://logs.openstack.org/37/667237/1/gate/openstack-tox-pep8/1124157/ : SUCCESS in 6m 11s\n- openstack-tox-py27 http://logs.openstack.org/37/667237/1/gate/openstack-tox-py27/ddc0fc6/ : SUCCESS in 3m 45s\n- openstack-tox-py35 http://logs.openstack.org/37/667237/1/gate/openstack-tox-py35/75b79d5/ : FAILURE in 17m 12s\n- openstack-tox-py36 http://logs.openstack.org/37/667237/1/gate/openstack-tox-py36/a6de43b/ : SUCCESS in 4m 07s\n- openstack-tox-docs http://logs.openstack.org/37/667237/1/gate/openstack-tox-docs/ded16c0/html/ : SUCCESS in 5m 18s\n- build-openstack-releasenotes http://logs.openstack.org/37/667237/1/gate/build-openstack-releasenotes/953664a/html/ : SUCCESS in 5m 12s\n- openstack-tox-functional http://logs.openstack.org/37/667237/1/gate/openstack-tox-functional/fbf75e6/ : SUCCESS in 5m 07s\n- openstack-tox-functional-py35 http://logs.openstack.org/37/667237/1/gate/openstack-tox-functional-py35/49a422c/ : SUCCESS in 7m 04s\n- openstack-tox-functional-py36 http://logs.openstack.org/37/667237/1/gate/openstack-tox-functional-py36/9757784/ : SUCCESS in 6m 05s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/37/667237/1/gate/octavia-v1-dsvm-scenario/a069b76/ : SUCCESS in 59m 38s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/37/667237/1/gate/octavia-v1-dsvm-py3x-scenario/c7bf64f/ : SUCCESS in 1h 06m 49s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/37/667237/1/gate/neutron-lbaasv2-dsvm-api/ceadc6d/ : SUCCESS in 1h 21m 22s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/37/667237/1/gate/neutron-lbaasv2-dsvm-py3x-api/209e968/ : SUCCESS in 1h 20m 22s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/37/667237/1/gate/octavia-v2-dsvm-noop-api/b5e81d4/ : SUCCESS in 44m 56s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/37/667237/1/gate/octavia-v2-dsvm-noop-py2-api/f39f600/ : SUCCESS in 41m 58s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/37/667237/1/gate/octavia-v2-dsvm-scenario/8f42d21/ : SUCCESS in 1h 36m 37s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/37/667237/1/gate/octavia-v2-dsvm-py2-scenario/9cec27b/ : SUCCESS in 1h 28m 57s\n- octavia-v2-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/37/667237/1/gate/octavia-v2-dsvm-scenario-ubuntu-xenial/d55c8d2/ : SUCCESS in 1h 56m 08s\n- octavia-grenade http://logs.openstack.org/37/667237/1/gate/octavia-grenade/d554245/ : SUCCESS in 1h 14m 04s","accounts_in_message":[],"_revision_number":1},{"id":"39149d8755cf45f318880bea91f55bc9f560641a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-26 17:11:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e8b6192009663e089d8f864fc2c54594757259c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 19:31:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/37/667237/1/check/openstack-tox-cover/94f7ed8/cover/ : SUCCESS in 7m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/37/667237/1/check/openstack-tox-lower-constraints/d3d1233/ : SUCCESS in 6m 33s\n- openstack-tox-pep8 http://logs.openstack.org/37/667237/1/check/openstack-tox-pep8/120515d/ : SUCCESS in 6m 39s\n- openstack-tox-py27 http://logs.openstack.org/37/667237/1/check/openstack-tox-py27/05ac271/ : SUCCESS in 4m 03s\n- openstack-tox-py35 http://logs.openstack.org/37/667237/1/check/openstack-tox-py35/6319d99/ : SUCCESS in 5m 26s\n- openstack-tox-py36 http://logs.openstack.org/37/667237/1/check/openstack-tox-py36/b5047cf/ : SUCCESS in 4m 50s\n- openstack-tox-docs http://logs.openstack.org/37/667237/1/check/openstack-tox-docs/c9de863/html/ : SUCCESS in 4m 52s\n- build-openstack-releasenotes http://logs.openstack.org/37/667237/1/check/build-openstack-releasenotes/8a61fef/html/ : SUCCESS in 4m 45s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/37/667237/1/check/neutron-lbaasv2-dsvm-api/9963660/ : SUCCESS in 1h 19m 36s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/37/667237/1/check/neutron-lbaasv2-dsvm-py3x-api/b59641f/ : SUCCESS in 1h 18m 36s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/37/667237/1/check/neutron-lbaasv2-dsvm-scenario/69682a4/ : SUCCESS in 1h 48m 30s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/37/667237/1/check/neutron-lbaasv2-dsvm-py3x-scenario/fa04ad2/ : SUCCESS in 1h 22m 12s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/37/667237/1/check/openstack-tox-functional/5a63edc/ : SUCCESS in 5m 25s\n- openstack-tox-functional-py35 http://logs.openstack.org/37/667237/1/check/openstack-tox-functional-py35/23ff006/ : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 http://logs.openstack.org/37/667237/1/check/openstack-tox-functional-py36/ea05da8/ : SUCCESS in 6m 06s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/37/667237/1/check/octavia-v1-dsvm-scenario/403a4cb/ : SUCCESS in 1h 06m 17s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/37/667237/1/check/octavia-v1-dsvm-py3x-scenario/bf9ee9e/ : SUCCESS in 1h 01m 41s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-noop-api/f350ae0/ : SUCCESS in 46m 22s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-noop-py2-api/420c4b1/ : SUCCESS in 49m 27s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-scenario/0506d81/ : SUCCESS in 1h 31m 38s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-py2-scenario/761203b/ : SUCCESS in 1h 47m 01s\n- octavia-v2-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-scenario-ubuntu-xenial/4b505a8/ : SUCCESS in 1h 39m 08s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-py2-scenario-centos-7/821873b/ : FAILURE in 2h 02m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/c001e6d/ : SUCCESS in 1h 39m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-scenario-two-node/8841d9f/ : SUCCESS in 1h 46m 34s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-py2-scenario-two-node/8077930/ : SUCCESS in 1h 46m 08s (non-voting)\n- octavia-grenade http://logs.openstack.org/37/667237/1/check/octavia-grenade/820c3a6/ : SUCCESS in 1h 17m 22s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/37/667237/1/check/octavia-v2-dsvm-tls-barbican/f3193c8/ : SUCCESS in 51m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3486f4b5849869130c5c73e7f0294d5117df5189","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 19:31:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8b2bc4b0c4b4e93b45869c1c3b19a1b721b3ce6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 21:20:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ffee59e4c5b9b5c1f5900b27fe657007ad4afc5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 21:20:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/37/667237/1/gate/openstack-tox-lower-constraints/b557daa/ : SUCCESS in 5m 54s\n- openstack-tox-pep8 http://logs.openstack.org/37/667237/1/gate/openstack-tox-pep8/986023f/ : SUCCESS in 4m 51s\n- openstack-tox-py27 http://logs.openstack.org/37/667237/1/gate/openstack-tox-py27/2f1eb7f/ : SUCCESS in 4m 41s\n- openstack-tox-py35 http://logs.openstack.org/37/667237/1/gate/openstack-tox-py35/8e5c259/ : SUCCESS in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/37/667237/1/gate/openstack-tox-py36/e79bd84/ : SUCCESS in 4m 53s\n- openstack-tox-docs http://logs.openstack.org/37/667237/1/gate/openstack-tox-docs/995483c/html/ : SUCCESS in 5m 34s\n- build-openstack-releasenotes http://logs.openstack.org/37/667237/1/gate/build-openstack-releasenotes/5db5dac/html/ : SUCCESS in 4m 27s\n- openstack-tox-functional http://logs.openstack.org/37/667237/1/gate/openstack-tox-functional/8b76dce/ : SUCCESS in 5m 40s\n- openstack-tox-functional-py35 http://logs.openstack.org/37/667237/1/gate/openstack-tox-functional-py35/a7a3060/ : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 http://logs.openstack.org/37/667237/1/gate/openstack-tox-functional-py36/6bae7e2/ : SUCCESS in 6m 14s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/37/667237/1/gate/octavia-v1-dsvm-scenario/33315e3/ : SUCCESS in 1h 05m 04s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/37/667237/1/gate/octavia-v1-dsvm-py3x-scenario/5fdcc3c/ : SUCCESS in 55m 52s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/37/667237/1/gate/neutron-lbaasv2-dsvm-api/530cc60/ : SUCCESS in 1h 19m 36s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/37/667237/1/gate/neutron-lbaasv2-dsvm-py3x-api/d832fa9/ : SUCCESS in 1h 19m 29s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/37/667237/1/gate/octavia-v2-dsvm-noop-api/25baaf8/ : SUCCESS in 42m 36s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/37/667237/1/gate/octavia-v2-dsvm-noop-py2-api/e587e16/ : SUCCESS in 41m 08s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/37/667237/1/gate/octavia-v2-dsvm-scenario/e397e7c/ : SUCCESS in 1h 43m 15s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/37/667237/1/gate/octavia-v2-dsvm-py2-scenario/ff81b11/ : SUCCESS in 1h 41m 34s\n- octavia-v2-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/37/667237/1/gate/octavia-v2-dsvm-scenario-ubuntu-xenial/71ef799/ : SUCCESS in 1h 41m 54s\n- octavia-grenade http://logs.openstack.org/37/667237/1/gate/octavia-grenade/8fca7c0/ : SUCCESS in 1h 17m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d53010dc423867bc965069206b007a62fbb02475","revisions":{"d53010dc423867bc965069206b007a62fbb02475":{"kind":"REWORK","_number":1,"created":"2019-06-24 22:28:12.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/37/667237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/667237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/667237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/667237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/667237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/667237/1"}}},"commit":{"parents":[{"commit":"33776e159dcf50d990221a3714d58ae15010fe87","subject":"Merge \"Fix allocate_and_associate DB deadlock\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/33776e159dcf50d990221a3714d58ae15010fe87"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 19:57:55.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 22:28:12.000000000","tz":0},"subject":"Add bindep.txt for Octavia","message":"Add bindep.txt for Octavia\n\nSome base packages are no longer automatically included in zuul\njob runs. This is good as we need to call out our platform package\nrequirements.\nThis patch adds a bindep.txt to Octavia to capture those\npackage requirements.\n\nChange-Id: I48e82f39e3980d6d50e3b710326b8e437eddb08b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d53010dc423867bc965069206b007a62fbb02475"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d53010dc423867bc965069206b007a62fbb02475"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
