)]}'
{"id":"openstack%2Foctavia~667249","triplet_id":"openstack%2Foctavia~master~I076335d08d5411fd629c3e8860f14980b0dbcb5a","project":"openstack/octavia","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"I076335d08d5411fd629c3e8860f14980b0dbcb5a","subject":"Generate PDF documentation","status":"MERGED","created":"2019-06-24 23:46:35.000000000","updated":"2019-09-20 07:07:09.000000000","submitted":"2019-09-20 07:05:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667249-1568963115862-4a385b24","meta_rev_id":"2723f49706ec460c7ed7c777acfccb89d1a72d79","_number":667249,"virtual_id_number":667249,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-20 07:05:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-19 22:03:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-09-19 20:16:15.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-09-19 22:03:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":-2,"date":"2019-09-19 20:16:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-12 07:45:41.000000000","updated_by":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"reviewer":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"state":"REVIEWER"},{"updated":"2019-09-14 15:51:38.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-09-14 16:03:10.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-19 20:16:15.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-19 22:03:08.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-20 07:05:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d185774bc1b36df182ed7dd9115b071cb30654b9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-24 23:46:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f0cd03643866ee2da3235eb82cc8b5848af516a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-24 23:46:46.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b042b733acaf68325e94dd29daa596f79b28aaae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 04:16:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/49/667249/1/check/build-openstack-api-ref/2f61f82/html/ : SUCCESS in 5m 55s\n- requirements-check http://logs.openstack.org/49/667249/1/check/requirements-check/88e1eb6/ : FAILURE in 3m 17s\n- openstack-tox-cover http://logs.openstack.org/49/667249/1/check/openstack-tox-cover/4399a85/cover/ : SUCCESS in 9m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/667249/1/check/openstack-tox-lower-constraints/eb0b6db/ : SUCCESS in 8m 18s\n- openstack-tox-pep8 http://logs.openstack.org/49/667249/1/check/openstack-tox-pep8/17f4882/ : SUCCESS in 7m 18s\n- openstack-tox-py27 http://logs.openstack.org/49/667249/1/check/openstack-tox-py27/91e03cf/ : SUCCESS in 6m 54s\n- openstack-tox-py36 http://logs.openstack.org/49/667249/1/check/openstack-tox-py36/b0ffce1/ : SUCCESS in 6m 27s\n- openstack-tox-py37 http://logs.openstack.org/49/667249/1/check/openstack-tox-py37/50a8fe4/ : SUCCESS in 6m 01s\n- openstack-tox-docs http://logs.openstack.org/49/667249/1/check/openstack-tox-docs/e015bc2/html/ : SUCCESS in 9m 52s\n- build-openstack-releasenotes http://logs.openstack.org/49/667249/1/check/build-openstack-releasenotes/694ce09/html/ : SUCCESS in 7m 07s\n- openstack-tox-functional http://logs.openstack.org/49/667249/1/check/openstack-tox-functional/8ce88a5/ : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 http://logs.openstack.org/49/667249/1/check/openstack-tox-functional-py36/2fc02f4/ : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/49/667249/1/check/octavia-v2-dsvm-noop-api/a118c14/ : SUCCESS in 48m 23s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/49/667249/1/check/octavia-v2-dsvm-noop-py2-api/277716e/ : SUCCESS in 48m 20s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/49/667249/1/check/octavia-v2-dsvm-scenario/fabed76/ : SUCCESS in 1h 37m 13s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/49/667249/1/check/octavia-v2-dsvm-py2-scenario/6730198/ : SUCCESS in 1h 35m 01s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/49/667249/1/check/octavia-v2-dsvm-py2-scenario-centos-7/4dfbbbf/ : FAILURE in 2h 19m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/49/667249/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/024e69e/ : SUCCESS in 1h 42m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/49/667249/1/check/octavia-v2-dsvm-scenario-two-node/9def2db/ : SUCCESS in 1h 44m 21s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/49/667249/1/check/octavia-v2-dsvm-py2-scenario-two-node/cee057d/ : SUCCESS in 1h 51m 07s (non-voting)\n- octavia-grenade http://logs.openstack.org/49/667249/1/check/octavia-grenade/af09323/ : SUCCESS in 1h 18m 18s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/49/667249/1/check/octavia-v2-dsvm-tls-barbican/b7ed183/ : SUCCESS in 54m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/49/667249/1/check/octavia-v2-dsvm-scenario-amphora-v2/d83c02a/ : SUCCESS in 1h 27m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d00d651a42905acd41998bee3844682647945d2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-23 19:07:17.000000000","message":"Patch Set 1:\n\nThe job for building PDFs has merged and you can use it.\n\nFor this, you need a pdf-docs environment and then the normal openstack-tox-docs job will do all the work.\n\nFor details, see https://review.opendev.org/#/c/664555/","accounts_in_message":[],"_revision_number":1},{"id":"282c1865f53d90ce2002c3b27df2df8b7db4ec2d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-23 21:44:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"275fbb4c00a1c30bbf0bb3c0e301b3f8a86581df","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-23 21:44:34.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"ffef15bd9ab469688c35724ed740a20df750f1c8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-23 23:18:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3e25f0b3554ae463c1c89b4b3f9e20c385913b87","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-23 23:18:12.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"bbf897bb9afdc42ec80c00bbe62bd328bf250bbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 01:24:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d7b02d5800a44f26997f9ae470871da7 : SUCCESS in 10m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7890325ad3d848b3a3d9785271c7e62a : FAILURE in 4m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70b2bf5efabe4c64919e0067c028b7db : SUCCESS in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38f2c6e433a94f76b41c8b76cce2fd42 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4f6dc581dce417f859a589c3c064168 : SUCCESS in 7m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/39628d7dc0f846e7aa90a8f41600a39c : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0a35f4cbba447e6824fa4a6459f82d7 : SUCCESS in 6m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/07522b8f75954b31aa1e9c6e42dfcad3 : SUCCESS in 11m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/808d1a0bcb3745d3b02faba97c1b87d5 : SUCCESS in 17m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/448128b4e90f4136983e953e4705f0b3 : SUCCESS in 5m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1a3973d116234b1e908b0b6cdcbd2a88 : SUCCESS in 9m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/407c6cf5f43746fe93090ae22aab316a : SUCCESS in 8m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/71cf824aa75e41859387df6109c135dc : SUCCESS in 7m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b783053e34274a64b438b094d077f446 : SUCCESS in 8m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/47d3e8ad5d60486d997559042cc8e65f : SUCCESS in 45m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3c7714ba53444df09ada80ae2584d358 : SUCCESS in 48m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/76c8244aba2a47cbb0707d8075663823 : SUCCESS in 1h 45m 42s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/83fae31bdb364991991256d2e410444e : SUCCESS in 1h 27m 13s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/25a3554a29e44ef79602e26a50298966 : SUCCESS in 2h 01m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7bd763d0f22b4e5e9b5e008747ec14d4 : FAILURE in 1h 44m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/69fa6e7511e84b36ae0368c1a306d2c2 : SUCCESS in 1h 53m 28s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/99d2c95a4c014cd485dfda1572ef7998 : SUCCESS in 1h 08m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bf1aa324b8ba41e2ab19707fecbfa4c4 : SUCCESS in 55m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/06a38ec6ab5244cc9a5e3c2aaf2c63e8 : SUCCESS in 1h 31m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/98dd731aa8374b828b328cc45f49568d : SUCCESS in 1h 09m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f959e08014e94be48e1e9605345ebac7 : SUCCESS in 1h 06m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/63dcacb8befd486fa6f4593214819844 : SUCCESS in 1h 44m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b14dd05b874c002d18a8f8664d5b8dca1656bc2b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-25 05:47:19.000000000","message":"Patch Set 3:\n\nrecheck - fixes in","accounts_in_message":[],"_revision_number":3},{"id":"dffeef45b57e13d1d287476008302cc7c44edff4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-25 05:56:49.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3762f120206769068778d02b5a4d647b8cae3ba2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-25 05:57:25.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"422802d9b7c4896fdc3ca71667faf25cf4a5f2ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 08:06:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0214e28a9a964cc286ebc5f328688475 : SUCCESS in 5m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/af1afaf8e90e47d1b2d9e87b3a4b7a32 : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f727533a9bf4c5bb33e7e02f9c5a854 : SUCCESS in 7m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea9a895e6536474f9d49525c681165c2 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da58080095b3470796dbacc31a1e1eec : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7294c80e31ef4c0881e008b7a3468844 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68d89e8bcab145238c969adeaab5dc18 : SUCCESS in 22m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce9420761bd44535a47eb972d4c665c7 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/594548e8908f4f918bf5f06fe6be3776 : SUCCESS in 15m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8167e5ff94c24578873265a1519c2a05 : SUCCESS in 6m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3c5be7e892ad43819714d751eb764dc6 : SUCCESS in 5m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/28ee4eaa71ab4f869ad5fe71c62d104c : SUCCESS in 6m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6edb9dc7b71545df8cad4511ad2ed42b : SUCCESS in 5m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/164a8085c1d04b1c991fe00d51699a54 : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/609d892f224b4b9b93c0135bda6f39e5 : SUCCESS in 42m 59s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b210709cbbd44f2c8f07840ede102219 : SUCCESS in 43m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/839f3c1ad519441aa19d95d17af12639 : FAILURE in 27m 49s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/03580ef8b9ac40f582459bcb88415d53 : SUCCESS in 1h 53m 39s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fae1f75f75404f5589d48aa9fa16939f : SUCCESS in 1h 47m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ea45c2181f994fc1943d1570da10ae1e : SUCCESS in 1h 10m 52s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/aa69762cb5914225aa319e31621f2f67 : SUCCESS in 1h 42m 54s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/26448da945df44ffa93a41ed51eadc1b : SUCCESS in 1h 12m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5461f298b6484c8c9ac2f8fb537ed835 : SUCCESS in 59m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/061388cd79b14086aef54e4adab93f8c : SUCCESS in 1h 33m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b31b56d5e7924d8288fd23d8d9013d26 : SUCCESS in 1h 11m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3b355502d4134b0e95f88de486dd74c1 : SUCCESS in 1h 08m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/13daf6ac6a2f4b51989d74e73e8e1f58 : SUCCESS in 1h 26m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"34d42988e9a84a8b815bc0963ff54776c0074fc4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-04 04:59:43.000000000","message":"Patch Set 4: Code-Review-1\n\nNote that https://review.opendev.org/679777 changed the pdf file name to doc-PROJECT.pdf, please adjust doc/source/conf.py","accounts_in_message":[],"_revision_number":4},{"id":"b852ac168912c71972ef8c0ad1bba800aa8308d0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-04 05:00:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"85d422cdd768b454500e1d0c613c3405ba568bf6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-04 05:18:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"12f0cb8f07e21d22c5413968531fbef08492f7cb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-04 05:19:01.000000000","message":"Patch Set 4:\n\nUpdated doc file name and removed obsolete content from conf.py","accounts_in_message":[],"_revision_number":4},{"id":"ef0e628df229cf4dd43aefe4c62e9c26267d7fe0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 08:35:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f0980ce78b9043918aff71d730b1e5f5 : SUCCESS in 6m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b37689466cb44998b655ab98aa31714f : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed54835fd3764f53955cfa0e01c19ae0 : SUCCESS in 7m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8a9120df9c3478e8aadad5143108e5a : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc1e6661e84c49c8a207e390189db57d : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2bf5c33327e844fb89306bebff49a30a : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30032adc441a48579fbbcdd603a7502d : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e11b55285094086a5faddf7edb7735c : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc4b3a79664c4316901ca0a35470fcad : FAILURE in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39103d1c09b0482580265122d238008d : SUCCESS in 9m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8b731f0c12ac48a886728a0ed2daa93a : SUCCESS in 5m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9e3aecc5319d486d803a08db451d1252 : SUCCESS in 6m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/684d85f4d7b44e21bb8fb68c4876c846 : SUCCESS in 5m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/49638fa016c543ba984c7e05e74fbb0a : SUCCESS in 6m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4f4e31ca299f4b4b87fa54b699bdac27 : SUCCESS in 49m 14s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4054279426f24abd9652a472bcda84b8 : SUCCESS in 44m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3219627bc5ce44649317743e52ea02f6 : FAILURE in 29m 52s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/75456a7984f84363a57a0b6659d268c0 : SUCCESS in 1h 47m 23s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8404aba7bb574178bbf373c01798219c : TIMED_OUT in 2h 37m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c5c4146d013a478ba2380b98693aa69c : FAILURE in 1h 38m 14s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ceafaab4314942dca39492dd68bae04d : FAILURE in 28m 08s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d48c7cdad1c34d12aa7d7acad690f33b : SUCCESS in 1h 29m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/250c01373bbf4bc1ad80b7523111b652 : SUCCESS in 48m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/423afe6325e2440eb4572adc7d6f57c0 : SUCCESS in 1h 34m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/03a7cb51e2e1449e8753681e0f722a1e : SUCCESS in 1h 01m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fc9cad41c84e4b9d9163b15ada2c82f4 : FAILURE in 26m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e3495f6ab58a41eb93103fd053834d5a : SUCCESS in 1h 26m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"38cd3656bb8d06570e6168b4d4b0d23c8bb86f94","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-04 08:39:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4ecdd0bbf76e0ec1a42905bbb7647447196be411","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 12:50:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c99efed10bab4d95ae6576c19a610043 : SUCCESS in 10m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6964c98dd30c4be6b4580fb4f42561f8 : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6904ae43800e46db82368663b2aa61bc : SUCCESS in 10m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecc8b86c17f54ceebd4c29cc4ec703ec : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5307007225dc4950a0fd3488b21757d2 : SUCCESS in 9m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/88b611caf8124939b6762ee1dbf1b217 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7840570d448747af81a00ec8e92aa59e : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3e0b24ecf6b4d23acdda3298c7baa44 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f85d7fb052ab46b6bb361b9cf0f70e9c : SUCCESS in 24m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04986781dbdf41af880355fcd0441259 : SUCCESS in 5m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ebd8574354544f02928f084c1ce8cf6f : SUCCESS in 6m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/aa82af330eb64a53b3e800a6672f172f : SUCCESS in 9m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b78b733f4c134b08a57cca318ba7c557 : SUCCESS in 5m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fc7edbf4742f4b399eaa8a9edfbd0c4e : SUCCESS in 6m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3700b7e771f949a9af8574416f3e9d37 : SUCCESS in 53m 50s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b0af653f5c934cff9a3847c8fb7b8cd8 : SUCCESS in 51m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee526eee387448559c466197089c832c : SUCCESS in 1h 39m 46s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d5afd1f02d6a44438adca672a2af6feb : RETRY_LIMIT in 1h 16m 14s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2dc3bc373ac9477e94cad03f638d1d83 : SUCCESS in 2h 16m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/26ef330f5bf34b2c8d49032cd9bae642 : SUCCESS in 2h 01m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f434303147724b6d81734bd37196fe99 : SUCCESS in 1h 14m 36s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1503e70e345c4ed9bad966ee29214631 : SUCCESS in 1h 10m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8f29b932b001410d9fd0b6ebbe718df6 : SUCCESS in 1h 04m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d75c4b5390534242a0d4fd3d0bd58994 : FAILURE in 26m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/98686a5fc12a47b697676d07d31233ad : SUCCESS in 1h 22m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2eade2b25a0646b58362c3606f7fb687 : SUCCESS in 1h 21m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a60681e5712847eb9a97d8092edd7c22 : FAILURE in 1h 18m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"792ead0e21a786805a595b24049e6a6a6efc8357","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-11 04:23:05.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6b68d8f3769182b39307b6ba304478813e037233","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-11 04:24:10.000000000","message":"Patch Set 7: Workflow-1\n\nThis still has some work to do.\n\nhttps://etherpad.openstack.org/p/pdf-goal-train-octavia","accounts_in_message":[],"_revision_number":7},{"id":"379594cdde6f0cdb8e03b58080a3011a54fd5b75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 06:45:32.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bb29b164276e467195f81bfa8fa12368 : SUCCESS in 4m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5ce69754486e4f488ccd946d15a206f2 : SUCCESS in 1m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e316349e9234f99baea22539830e53c : SUCCESS in 7m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41fc70f8a8e3437ea4607b3c9f5e7b3c : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1289f034dac748258d49da43f77e1c93 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2aff187e2a1647b98e58c01d1f599cc0 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fadbcf3a4a8b44808f30412ad2aa0932 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d96fe46767c74287a46ddc97318dd866 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac630ecf833d47c49c38097d717c769c : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/beeabbf1e5b645ed9b3bf8a63f3ee40e : SUCCESS in 4m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1298144a13234f81b847589e7dd1155e : SUCCESS in 5m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/82097d59dbf6479685b7d0c8be94befd : SUCCESS in 6m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8d67119c42974b17908fc3abcd0269a6 : SUCCESS in 4m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2193b5f1c33741e78da41bb32dbff5b6 : SUCCESS in 6m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c16862613a142e7beaccd753e02bd18 : SUCCESS in 46m 04s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1fed2cf3933948b68ffd2c7395051a00 : SUCCESS in 1h 25m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e2011b743544dd3a96e5b393452f2b5 : SUCCESS in 1h 44m 23s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1e461039c4704f69b73e55842323a44b : SUCCESS in 1h 59m 36s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e00faa18cb564f3aa69d89374f0ce88f : SUCCESS in 2h 00m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f39a7873b11c4803abadeffcb58ac648 : SUCCESS in 2h 16m 50s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/506845c72a334dfb829109bed93b1e82 : SUCCESS in 1h 41m 48s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c6a79a8671c745c0a4395779ecfb4305 : SUCCESS in 1h 25m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/598133f1a6fa4e8b9574b80c83c6059f : SUCCESS in 57m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ef96ccdb7b074b32aafa237ffed4785b : SUCCESS in 1h 28m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db039f06e2a24ae487315216b7c2a272 : SUCCESS in 1h 11m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/56c1332ab5bf4a628423e0c5a64cb2d3 : SUCCESS in 1h 17m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/976f86d49be7433fa66026552bbb604d : SUCCESS in 1h 48m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fd330a921f624bcc029e3bc1ee3348f3665d2c04","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-09-12 07:45:41.000000000","message":"Patch Set 7:\n\n(3 comments)\n\nthanks for the patch. please find some comments inline.","accounts_in_message":[],"_revision_number":7},{"id":"acaa1dd75d7a90b6926f012c9136228b728b9d25","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-12 21:44:26.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"fc1353af8c3fabf872c5d9383a605e3a91b1b185","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-12 22:02:13.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f9685b6f72a300fc5c9a072c5355c0a73bc3fe5a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-12 23:23:05.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9a2fa9c4c5bc7ba40e619f6fa0f33122a5463537","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-12 23:57:55.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"754d028ac3d6803c251bb4f9646355b133bad401","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 02:26:04.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4d7fb40a573f41d1887dea432ed75f63 : SUCCESS in 5m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/227ce579369a4050a199aaecf4904cc8 : SUCCESS in 2m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e62a10c15be14746a5c63b95df3811de : SUCCESS in 7m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4be1745a321c479f9bd8d66115195b28 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/582d7544a21f4e3eb2ba27dbf3e3ed6a : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/59afeefa4fe04ce19ec99f70c27c6a56 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3227ce2446a408db1a3f402459c4d54 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/907ea262f2ff4ea291eee957d8f152dd : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bffeae948f04f6998835d531d91f340 : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e44d9eb910ae45628b6e652b2bc2cf50 : SUCCESS in 5m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a9c2ebdb2bf1478b9071b74c6c6ffbec : SUCCESS in 5m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e8621639a3e44fa8a2f5a6b9927b8f4f : SUCCESS in 6m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/cafed74a97ab48aaa6e75e228186800f : SUCCESS in 5m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9e497d42d43d4fd29eeb65d1ff89dbef : SUCCESS in 5m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/308bf02a91bf46a7b1f07e37d881cd2c : SUCCESS in 43m 35s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/bd46442df723456ba117058655853bc5 : SUCCESS in 45m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2ef8219ab6547c1bfe2b116fbd70134 : SUCCESS in 1h 33m 13s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6117b8bea86d49c5aad94486ce63b709 : SUCCESS in 1h 42m 58s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/47dcf16ab49c4d848ac3762302aaa907 : SUCCESS in 2h 14m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/280bb999ad0a4a11b2c77990a82e2725 : SUCCESS in 2h 02m 13s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ffececc0572b41ec903a18ac9881146f : FAILURE in 1h 55m 10s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4a0dda40a9114c4da04db1a97d3e0a10 : SUCCESS in 1h 19m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a6ae49807b3e4df29510e904bd6f3fe3 : SUCCESS in 59m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b711da9d82fe442fb59524040e9dc4e8 : SUCCESS in 1h 31m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bd13878bc5c044838b3970bad0e2bbe9 : SUCCESS in 1h 03m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/880b4b223be146778784dd9469276597 : FAILURE in 46m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ead6bbdb19684553b2e9fe3e8a1679e4 : FAILURE in 1h 09m 33s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"2c86d7c3d324b78aee68b4370476d904ca551b17","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-09-14 15:51:38.000000000","message":"Patch Set 11: Backport-Candidate+1","accounts_in_message":[],"_revision_number":11},{"id":"0776c1d0c07ebf171e48f4808bea737cc834042a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-14 16:03:10.000000000","message":"Patch Set 11: Backport-Candidate-1\n\nThis is NOT a backport candidate, this is for train only.","accounts_in_message":[],"_revision_number":11},{"id":"6ee2ebeb0f205532d483b76a4768bf76c8156dae","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-14 19:17:13.000000000","message":"Patch Set 11: Backport-Candidate-2\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"c140fba2c53f912305e29dc52dd47146852e2e48","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-19 17:58:39.000000000","message":"Patch Set 11:\n\nnote: this does have a bug where some sections are not rendering.\nI am tracking that here: https://storyboard.openstack.org/#!/story/2006585 and on the PDF goal problems etherpad here: https://etherpad.openstack.org/p/pdf-goal-train-common-problems\nI think this is good enough for a first patch. We can work on fixing the bug(s) in future patches.","accounts_in_message":[],"_revision_number":11},{"id":"3a4fdfdcaf4687b40f1d977fc358450e251296b9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-19 18:58:11.000000000","message":"Patch Set 11: Backport-Candidate-2 Code-Review-1\n\n(1 comment)\n\nA small request for clarification.","accounts_in_message":[],"_revision_number":11},{"id":"85fec230bfd54a0711423fb41fa76143a6869b53","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-19 20:12:43.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"aeb551019c7b20374e15d1eedfd1e7b51d0a54af","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-19 20:13:31.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"59a388929da7c36d129135a566b6b961eeb2aa00","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-19 20:16:15.000000000","message":"Patch Set 12: Backport-Candidate-2 Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"9fb84e7d88c12f513b83fd7d656ff895ad6f94b8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-19 22:03:08.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\nI don\u0027t see anything glaring here. The output format is still painful for me, but I guess it\u0027s not really FOR me, so that\u0027s fine. We\u0027ll tweak moving forward.","accounts_in_message":[],"_revision_number":12},{"id":"8365cc8de085c332d677d64ed4780bc563b61ec7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 23:18:20.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/db1bdecefba142a78f2d168be7a247ff : SUCCESS in 4m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/65b4d2c7cb494d7a8e32122ad378aa32 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e02f5eda9e642169683c783ef3778e0 : SUCCESS in 8m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67151275d90a406a972c9fc090d81c70 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40884b3ac9074667a21daa189b8402eb : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d96055ac84ff4d77af35e192d3e2470d : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88118702ad904ad6970018553bcbefd8 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5feea411003d426887b77c4fcf564439 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4d1f384ccc741e6ad4889df7d9854b5 : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66068d90690c4f758d345bc69761b2fc : SUCCESS in 6m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3f68e38221a5479da6a4a5f6ac2fdf06 : SUCCESS in 4m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/400e4a9fc3294581a697f23d4ac2c6c4 : SUCCESS in 7m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6df40ff6f5564547b001cf7c31c1465a : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ccddaa6a1a1f4950bc85f4e1b3222424 : SUCCESS in 8m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0a79135362044e76b0fb5e9492422fa2 : SUCCESS in 49m 17s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/98b344e23af94bfabcdd37211fe1a5e5 : SUCCESS in 46m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/44aa04cce80d423486041c2de32e3073 : SUCCESS in 1h 40m 39s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d48d1b6fe2ba4fc58af6a424ff216123 : SUCCESS in 1h 59m 09s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/93c8244501fd4e1d816fc42ac33f0f98 : FAILURE in 20m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1d6349b15efa48dc95fa3c015f5f9381 : SUCCESS in 2h 03m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/9d055a02c576401e885770855a2c2e58 : SUCCESS in 1h 57m 33s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ef44528fa5534a2cbb3565c2d7bb3003 : SUCCESS in 1h 22m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ddf38577652741b3af2a8035ea490e18 : SUCCESS in 53m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5f22a557c15443dfa3e20fa3c36cf80c : SUCCESS in 1h 31m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc73b49e9cd5430cb1f6fb27ebaf6c31 : SUCCESS in 1h 06m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/92eefde99cdb4c8b8edbbf09de0095c1 : SUCCESS in 1h 10m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c53f427e738f4fee91a873e01d338ee0 : FAILURE in 20m 45s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f4772f4fa9a98bafd3c37280a4e917e250bb7c4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 23:18:33.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"c6bbefa4a79c8fc8c86d838de3def04f7fd6e9b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 00:49:37.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cf77820f3917430eb62964efa1194653 : SUCCESS in 6m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/59679190456f46bd8129adb90b6412c4 : SUCCESS in 4m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bfaf4285ce2412d829ad1b2ebb6d4ae : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1139b532e6a4bee9b8bf53f58da5e31 : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/295ce7a197d24e67a27e7ce333151727 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53b08a4e3cd74e42a528616a5cf22047 : SUCCESS in 6m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5eceff909e44478f8bc13b762ba84348 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1191c0fc3a8440c78c140984154bb1f9 : SUCCESS in 11m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36bded38bfe245379e354db449013526 : SUCCESS in 6m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6299a2dc20054495a3d692649a5b6b8b : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/db40edd13ea04505bb10c0083fba3dfc : SUCCESS in 6m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/50fff55dedb748d2a235aa0389990024 : SUCCESS in 52m 59s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/7f75a77682c64abfb6077af02519b36b : SUCCESS in 45m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4da7a20228474eb7a4a8d424b8db9eaa : CANCELED in 1h 23m 38s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/11431b0a24cc4d85a3a49af04acb3991 : CANCELED in 1h 23m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7bdfb21b0c3f49988df99ddde4b3aecb : FAILURE in 1h 27m 56s","accounts_in_message":[],"_revision_number":12},{"id":"6680ddd8ddd7da121e465c5e49c2f8bff7c10b07","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-20 02:34:49.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"74d6226968eec2634744b2b8f3c6a065b6815967","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 04:58:18.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b8b9b58732b04251b0bac1f87c9c851e : SUCCESS in 6m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/16c70eecb48349a29df1e25101708322 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6777004abb244b1493b76abc6eb8dd8e : SUCCESS in 8m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcd669d98a5640e99a6afe86037f2993 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c010c0287cbd4e4ca50b179dbf9a128c : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/305911df122349ed996de853ccb64490 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d531528a020641749262098b3108b5c8 : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc99b243ca174ec88fbb2cff51f0e26f : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a21c5d9853f1416ba1e5eaa0ddcc7301 : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d1dcb9c4bad46ceb80ced4ff438bf33 : SUCCESS in 7m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ebd2c9852abe4aaf9ea0d0a363dcfb1b : SUCCESS in 4m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/18379a0ad98545c0bd6ce171c88b9ff6 : SUCCESS in 6m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1913271516f2424e9640b833d3566529 : SUCCESS in 4m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d43aeb8742e548f7896b4e54c76d6e5d : SUCCESS in 5m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0da64edbf72949f297cfb7c07c2b4e62 : SUCCESS in 52m 32s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/eb75dc8b45da4d7cb53f62c5b0bd15f8 : SUCCESS in 46m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c160cd935454a5981474468210cc75d : SUCCESS in 1h 50m 10s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/693b1e6d4e50422cba3b05622dc6ed58 : SUCCESS in 1h 38m 23s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/40a24ab950514e4cabd52057a97ee998 : FAILURE in 20m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b29e8d700ec54211a9964330e0ef2649 : FAILURE in 43m 32s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/be48e136998c4a9cb60860ab8a513f53 : SUCCESS in 2h 04m 04s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4262d336b10641f58c82ba3624c31b80 : SUCCESS in 1h 31m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8d66da41edc345c398992b39a08fc3d0 : SUCCESS in 52m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/94941543233049bbac69fbd634a1179e : SUCCESS in 1h 58m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3d567f10eee4e52b6361c22af0e480a : SUCCESS in 1h 25m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/57ee092208a44fb8adcff799a3b2f45d : FAILURE in 59m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/38a9b60a2533460d861367e591e59a46 : FAILURE in 22m 12s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f19c7f719b632c2614af7e5a9f220bf74d784b05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 04:58:31.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"da620ded38529c6bb497d74c5bd2efbb9caf940a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 07:05:15.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b44ec1d0d3424eae93117a06161aa80f : SUCCESS in 5m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a3d9483f63aa4ec9bd98ddd5db7a3162 : SUCCESS in 2m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/913353e3f99c47d0a136e80211b86052 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5db32b19f1474274a243e36d25f79040 : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/747621e18d494a45ba871e5c4dd58efb : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a7941f8f8e74d9cbd3f71ed9ad21a07 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47c69858cdd848729a5a992751404cc5 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/913a2e43ccde428ab89307a078e019f3 : SUCCESS in 10m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05768a1094184925bd7355807274bc92 : SUCCESS in 5m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4fd54b20214b48bf95681b8cf9411f6a : SUCCESS in 5m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/40674cebb47f486caf03fb1c96f08949 : SUCCESS in 5m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9a34be54be414ee2b1d3b32fbc2aca5a : SUCCESS in 43m 55s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/833c9f71653342d3bb579b2b9fd82cc7 : SUCCESS in 48m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3afacb8ef8574ebb8abf18ce606bd3a1 : SUCCESS in 1h 57m 10s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a4fca7437a5d4eaea04a6404e95fdbd9 : SUCCESS in 1h 48m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/acb2699504a74c50ac16675e034b1598 : SUCCESS in 1h 18m 40s","accounts_in_message":[],"_revision_number":12},{"id":"f15673bcda8940507fb9fc97c96538c328d1ea79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 07:05:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"a3ce7ef59c8ebfb2561d93ee85597d63dcc99b46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 07:07:09.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/4a0574815e994a3b89d9c6a18903f4a1 : SUCCESS in 56s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b32841f4d635452ea0eb58178f3965b7 : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bc6d9bb7214491ea5b281e948f7c4e4 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"97058e925b2f301b68f2b4df1e181a43c268bf1b","revisions":{"23f13abfbe96e69d8bb6d853e6494ad8011ae8e8":{"kind":"REWORK","_number":1,"created":"2019-06-24 23:46:35.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/49/667249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/667249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/667249/1"}}},"commit":{"parents":[{"commit":"09733a0c39ad394a319f7b807518a5439ec0826e","subject":"Add bindep.txt for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09733a0c39ad394a319f7b807518a5439ec0826e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 23:45:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 23:46:24.000000000","tz":-420},"subject":"WIP: Generate PDF documentation","message":"WIP: Generate PDF documentation\n\nThis patch sets up PDF document generation for Octavia.\n\nChange-Id: I076335d08d5411fd629c3e8860f14980b0dbcb5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23f13abfbe96e69d8bb6d853e6494ad8011ae8e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23f13abfbe96e69d8bb6d853e6494ad8011ae8e8"}]},"branch":"refs/heads/master"},"6075e567c1a9fb5633d94c9b8360347b882ed9fe":{"kind":"REWORK","_number":2,"created":"2019-08-23 21:44:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/49/667249/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/667249/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/667249/2"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 23:45:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-23 21:44:07.000000000","tz":-420},"subject":"WIP: Generate PDF documentation","message":"WIP: Generate PDF documentation\n\nThis patch sets up PDF document generation for Octavia.\n\nDepends-On: https://review.opendev.org/#/c/664555/\nChange-Id: I076335d08d5411fd629c3e8860f14980b0dbcb5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6075e567c1a9fb5633d94c9b8360347b882ed9fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6075e567c1a9fb5633d94c9b8360347b882ed9fe"}]},"branch":"refs/heads/master"},"ce2377be0d8ba5b142baa8ca5f5a31b81e97a474":{"kind":"REWORK","_number":3,"created":"2019-08-23 23:18:00.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/49/667249/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/667249/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/667249/3"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 23:45:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-23 23:17:47.000000000","tz":-420},"subject":"WIP: Generate PDF documentation","message":"WIP: Generate PDF documentation\n\nThis patch sets up PDF document generation for Octavia.\n\nDepends-On: https://review.opendev.org/#/c/664555/\nChange-Id: I076335d08d5411fd629c3e8860f14980b0dbcb5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce2377be0d8ba5b142baa8ca5f5a31b81e97a474"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce2377be0d8ba5b142baa8ca5f5a31b81e97a474"}]},"branch":"refs/heads/master"},"50efe7941bfe5424b2fd4bc23321f0e9f32831e8":{"kind":"REWORK","_number":4,"created":"2019-08-25 05:57:25.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/49/667249/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/667249/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/667249/4"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 23:45:13.000000000","tz":-420},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-08-25 05:57:20.000000000","tz":0},"subject":"WIP: Generate PDF documentation","message":"WIP: Generate PDF documentation\n\nThis patch sets up PDF document generation for Octavia.\n\nDepends-On: https://review.opendev.org/#/c/664555/\nChange-Id: I076335d08d5411fd629c3e8860f14980b0dbcb5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50efe7941bfe5424b2fd4bc23321f0e9f32831e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50efe7941bfe5424b2fd4bc23321f0e9f32831e8"}]},"branch":"refs/heads/master"},"8e24ab4a5434f7f6222277499358c772f160d212":{"kind":"REWORK","_number":5,"created":"2019-09-04 05:18:48.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/49/667249/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/667249/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/667249/5"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 23:45:13.000000000","tz":-420},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-04 05:18:25.000000000","tz":120},"subject":"WIP: Generate PDF documentation","message":"WIP: Generate PDF documentation\n\nThis patch sets up PDF document generation for Octavia.\n\nDepends-On: https://review.opendev.org/#/c/664555/\nChange-Id: I076335d08d5411fd629c3e8860f14980b0dbcb5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8e24ab4a5434f7f6222277499358c772f160d212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8e24ab4a5434f7f6222277499358c772f160d212"}]},"branch":"refs/heads/master"},"5a79844d4d5556f497be4e4e71fa54990507bd03":{"kind":"REWORK","_number":6,"created":"2019-09-04 08:39:12.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/49/667249/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/667249/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/667249/6"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 23:45:13.000000000","tz":-420},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-04 08:38:44.000000000","tz":120},"subject":"WIP: Generate PDF documentation","message":"WIP: Generate PDF documentation\n\nThis patch sets up PDF document generation for Octavia.\n\nDepends-On: https://review.opendev.org/#/c/664555/\nChange-Id: I076335d08d5411fd629c3e8860f14980b0dbcb5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a79844d4d5556f497be4e4e71fa54990507bd03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a79844d4d5556f497be4e4e71fa54990507bd03"}]},"branch":"refs/heads/master"},"2aadedb9366b2ab595e16f7300dec828cad737f1":{"kind":"REWORK","_number":7,"created":"2019-09-11 04:23:05.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/49/667249/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/667249/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/667249/7"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 23:45:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-11 04:23:03.000000000","tz":-420},"subject":"Generate PDF documentation","message":"Generate PDF documentation\n\nThis patch sets up PDF document generation for Octavia.\n\nChange-Id: I076335d08d5411fd629c3e8860f14980b0dbcb5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2aadedb9366b2ab595e16f7300dec828cad737f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2aadedb9366b2ab595e16f7300dec828cad737f1"}]},"branch":"refs/heads/master"},"8fc82d313818a5862db2e70c3fe4134c0af12df0":{"kind":"REWORK","_number":8,"created":"2019-09-12 21:44:26.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/49/667249/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/667249/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/667249/8"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 23:45:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-12 21:44:15.000000000","tz":-420},"subject":"Generate PDF documentation","message":"Generate PDF documentation\n\nThis patch sets up PDF document generation for Octavia.\n\nChange-Id: I076335d08d5411fd629c3e8860f14980b0dbcb5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8fc82d313818a5862db2e70c3fe4134c0af12df0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8fc82d313818a5862db2e70c3fe4134c0af12df0"}]},"branch":"refs/heads/master"},"ecd45660e11078dd16a484012515d21dd76eb2ed":{"kind":"REWORK","_number":9,"created":"2019-09-12 22:02:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/49/667249/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/667249/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/667249/9"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 23:45:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-12 21:59:53.000000000","tz":-420},"subject":"Generate PDF documentation","message":"Generate PDF documentation\n\nThis patch sets up PDF document generation for Octavia.\n\nStory: 2006101\nTask: 35146\nChange-Id: I076335d08d5411fd629c3e8860f14980b0dbcb5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ecd45660e11078dd16a484012515d21dd76eb2ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ecd45660e11078dd16a484012515d21dd76eb2ed"}]},"branch":"refs/heads/master"},"01d06913ba3c1ae4f2a8549f02e407cf6d054939":{"kind":"REWORK","_number":10,"created":"2019-09-12 23:23:05.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/49/667249/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/667249/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/667249/10"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 23:45:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-12 23:22:59.000000000","tz":-420},"subject":"Generate PDF documentation","message":"Generate PDF documentation\n\nThis patch sets up PDF document generation for Octavia.\n\nStory: 2006101\nTask: 35146\nChange-Id: I076335d08d5411fd629c3e8860f14980b0dbcb5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01d06913ba3c1ae4f2a8549f02e407cf6d054939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01d06913ba3c1ae4f2a8549f02e407cf6d054939"}]},"branch":"refs/heads/master"},"bdece421dc1b126d328417c22afdb1b78d2cfb31":{"kind":"REWORK","_number":11,"created":"2019-09-12 23:57:55.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/49/667249/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/667249/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/667249/11"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 23:45:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-12 23:57:47.000000000","tz":-420},"subject":"Generate PDF documentation","message":"Generate PDF documentation\n\nThis patch sets up PDF document generation for Octavia.\n\nStory: 2006101\nTask: 35146\nChange-Id: I076335d08d5411fd629c3e8860f14980b0dbcb5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bdece421dc1b126d328417c22afdb1b78d2cfb31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bdece421dc1b126d328417c22afdb1b78d2cfb31"}]},"branch":"refs/heads/master"},"97058e925b2f301b68f2b4df1e181a43c268bf1b":{"kind":"REWORK","_number":12,"created":"2019-09-19 20:12:43.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/49/667249/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/667249/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/667249/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/667249/12"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 23:45:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-19 20:12:37.000000000","tz":-420},"subject":"Generate PDF documentation","message":"Generate PDF documentation\n\nThis patch sets up PDF document generation for Octavia.\n\nStory: 2006101\nTask: 35146\nChange-Id: I076335d08d5411fd629c3e8860f14980b0dbcb5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97058e925b2f301b68f2b4df1e181a43c268bf1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97058e925b2f301b68f2b4df1e181a43c268bf1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
