)]}'
{"id":"openstack%2Foctavia~667940","triplet_id":"openstack%2Foctavia~master~I323ca1a136c45d1fb629156d4ed54824cf64feac","project":"openstack/octavia","branch":"master","topic":"story/2006060","hashtags":[],"change_id":"I323ca1a136c45d1fb629156d4ed54824cf64feac","subject":"Fix listener stats and amp health DB deadlock","status":"ABANDONED","created":"2019-06-27 14:29:46.000000000","updated":"2020-06-05 14:26:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"607019d85285e0cf5043bd691876bdf9cb7216ee","_number":667940,"virtual_id_number":667940,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-06-27 16:47:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-06-27 15:17:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 15:17:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-06-27 16:47:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c59fba3ce33fe881f8602b80b93f2e63bdb5a67","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-27 14:29:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3449e88ba987b5d1c2c2e78e912c46bcd9706927","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-27 15:17:20.000000000","message":"Patch Set 1: Code-Review-1\n\nWe are handling this exception in the code, but it is being logged at an error level. I think this should just be dropping the log level for the DBDeadlock messages as opposed to adding retries here that could cause old data to be inserted.","accounts_in_message":[],"_revision_number":1},{"id":"a8070a35f82897e3034ec0b2c6894eac1c23fc4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 16:47:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/40/667940/1/check/openstack-tox-cover/7e233d4/cover/ : SUCCESS in 7m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/667940/1/check/openstack-tox-lower-constraints/f3d8949/ : SUCCESS in 5m 36s\n- openstack-tox-pep8 http://logs.openstack.org/40/667940/1/check/openstack-tox-pep8/0e43a20/ : SUCCESS in 5m 23s\n- openstack-tox-py27 http://logs.openstack.org/40/667940/1/check/openstack-tox-py27/bbe8dcf/ : SUCCESS in 5m 33s\n- openstack-tox-py36 http://logs.openstack.org/40/667940/1/check/openstack-tox-py36/e22a900/ : SUCCESS in 4m 15s\n- openstack-tox-py37 http://logs.openstack.org/40/667940/1/check/openstack-tox-py37/bc8239d/ : SUCCESS in 5m 10s\n- openstack-tox-docs http://logs.openstack.org/40/667940/1/check/openstack-tox-docs/2a53ad3/html/ : SUCCESS in 7m 21s\n- build-openstack-releasenotes http://logs.openstack.org/40/667940/1/check/build-openstack-releasenotes/c4cd7cb/html/ : SUCCESS in 4m 51s\n- openstack-tox-functional http://logs.openstack.org/40/667940/1/check/openstack-tox-functional/d7a6526/ : SUCCESS in 5m 34s\n- openstack-tox-functional-py36 http://logs.openstack.org/40/667940/1/check/openstack-tox-functional-py36/61e367e/ : SUCCESS in 5m 28s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/40/667940/1/check/octavia-v2-dsvm-noop-api/b96df9d/ : SUCCESS in 45m 56s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/40/667940/1/check/octavia-v2-dsvm-noop-py2-api/16b1911/ : SUCCESS in 44m 32s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/40/667940/1/check/octavia-v2-dsvm-scenario/36cc1b5/ : SUCCESS in 1h 33m 21s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/40/667940/1/check/octavia-v2-dsvm-py2-scenario/7b7f3dc/ : SUCCESS in 1h 42m 20s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/40/667940/1/check/octavia-v2-dsvm-py2-scenario-centos-7/13bb6d9/ : FAILURE in 2h 07m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/40/667940/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/56131dc/ : SUCCESS in 1h 40m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/40/667940/1/check/octavia-v2-dsvm-scenario-two-node/dc0587b/ : FAILURE in 38m 40s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/40/667940/1/check/octavia-v2-dsvm-py2-scenario-two-node/2443fb5/ : SUCCESS in 1h 47m 29s (non-voting)\n- octavia-grenade http://logs.openstack.org/40/667940/1/check/octavia-grenade/df74baf/ : SUCCESS in 1h 15m 52s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/40/667940/1/check/octavia-v2-dsvm-tls-barbican/bd14f6c/ : SUCCESS in 54m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/40/667940/1/check/octavia-v2-dsvm-scenario-amphora-v2/4a4a1ae/ : SUCCESS in 1h 23m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"870d32bc21b88f528b977688e1a370f1079021f3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-05 14:26:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1b65f44f8b819ccc28e1009b0c684e809f62bd0","revisions":{"d1b65f44f8b819ccc28e1009b0c684e809f62bd0":{"kind":"REWORK","_number":1,"created":"2019-06-27 14:29:46.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/40/667940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/40/667940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/667940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/667940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/667940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/40/667940/1"}}},"commit":{"parents":[{"commit":"fbe8368779073ff3110fc52c7ae11baef0b66b0c","subject":"Merge \"Workaround pyroute2 platform-dependent imports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fbe8368779073ff3110fc52c7ae11baef0b66b0c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-27 12:24:13.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-27 14:29:21.000000000","tz":120},"subject":"Fix listener stats and amp health DB deadlock","message":"Fix listener stats and amp health DB deadlock\n\nSimilar to what was observed in story 2005893 and fixed in\nI20572505e448f9b7ff6314b5167cc4c519c445fb, a DBDeadLock was also\nobserved when updating the health of an amphora.\n\nhttp://logs.openstack.org/51/618751/13/check/ \\\nkuryr-kubernetes-tempest-openshift/49a4d64/controller/ \\\nlogs/screen-o-hm.txt.gz?level\u003dERROR\n\nThis patch adds deadlock retry wrapper to the listener stats and amphora\nhealth replace methods.\n\nTask: 34773\nStory: 2006060\n\nChange-Id: I323ca1a136c45d1fb629156d4ed54824cf64feac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1b65f44f8b819ccc28e1009b0c684e809f62bd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1b65f44f8b819ccc28e1009b0c684e809f62bd0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
