)]}'
{"id":"openstack%2Foctavia~668032","triplet_id":"openstack%2Foctavia~master~I728a9aa82986edef8128907a48c9df87d119fa8f","project":"openstack/octavia","branch":"master","topic":"fix/rollback-when-able","hashtags":[],"change_id":"I728a9aa82986edef8128907a48c9df87d119fa8f","subject":"only rollback DB when we have a connection to the DB","status":"MERGED","created":"2019-06-27 19:34:37.000000000","updated":"2019-08-26 12:49:48.000000000","submitted":"2019-07-03 05:52:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668032-1562133152380-cbb5daaf","meta_rev_id":"2fefea17c3fab218692575f01fcbb06c306482ad","_number":668032,"virtual_id_number":668032,"owner":{"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-07-03 05:52:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":2,"date":"2019-07-03 04:14:12.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-06-27 20:07:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":1,"date":"2019-07-03 04:14:12.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-06-27 20:07:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 20:07:48.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-07-03 04:14:12.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-07-03 05:52:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"634d837db0053134de6df131c9419eca92fe6495","author":{"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"},"date":"2019-06-27 19:34:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0abb998ccf04cf49af9e8f2a8c5c84092953808d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-27 19:41:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nLooks good code wise. We need a story and release note with it. This makes it easier to backport, etc.","accounts_in_message":[],"_revision_number":1},{"id":"cf53a1a08075c8902b79aac8425714852613f83d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-27 19:41:58.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"0dca5c8bf01df0596ac9e3847c9b8d85abfa8de1","author":{"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"},"date":"2019-06-27 19:48:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cbb1b3f27155d13d782c6b0ca550846a142294f7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-27 20:06:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"95ad8a6bd8fbc3f634ed5a756784438a53d94ed6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-27 20:07:48.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"1f8ce5d51363f35f137a2afe9dbac4da21624376","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 22:15:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/32/668032/3/check/openstack-tox-cover/61f70b8/cover/ : SUCCESS in 7m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/668032/3/check/openstack-tox-lower-constraints/018e499/ : SUCCESS in 5m 42s\n- openstack-tox-pep8 http://logs.openstack.org/32/668032/3/check/openstack-tox-pep8/d150032/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/32/668032/3/check/openstack-tox-py27/9c341ca/ : SUCCESS in 4m 36s\n- openstack-tox-py36 http://logs.openstack.org/32/668032/3/check/openstack-tox-py36/c2165ef/ : SUCCESS in 3m 57s\n- openstack-tox-py37 http://logs.openstack.org/32/668032/3/check/openstack-tox-py37/8241aa6/ : SUCCESS in 4m 38s\n- openstack-tox-docs http://logs.openstack.org/32/668032/3/check/openstack-tox-docs/797441c/html/ : SUCCESS in 6m 05s\n- build-openstack-releasenotes http://logs.openstack.org/32/668032/3/check/build-openstack-releasenotes/cde721c/html/ : SUCCESS in 4m 14s\n- openstack-tox-functional http://logs.openstack.org/32/668032/3/check/openstack-tox-functional/f399fe9/ : SUCCESS in 5m 30s\n- openstack-tox-functional-py36 http://logs.openstack.org/32/668032/3/check/openstack-tox-functional-py36/abff68b/ : SUCCESS in 6m 37s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-noop-api/d1fb974/ : SUCCESS in 42m 25s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-noop-py2-api/b64c46b/ : SUCCESS in 40m 55s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-scenario/9d2ba7d/ : FAILURE in 1h 45m 55s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-py2-scenario/1b24ec0/ : SUCCESS in 1h 32m 07s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-py2-scenario-centos-7/3a9cd07/ : FAILURE in 2h 04m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/ee94440/ : SUCCESS in 1h 38m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-scenario-two-node/686dee1/ : SUCCESS in 1h 49m 08s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-py2-scenario-two-node/1deef32/ : SUCCESS in 1h 47m 00s (non-voting)\n- octavia-grenade http://logs.openstack.org/32/668032/3/check/octavia-grenade/ed93e7e/ : SUCCESS in 1h 09m 50s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-tls-barbican/45a1ea6/ : SUCCESS in 52m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-scenario-amphora-v2/8d3d085/ : SUCCESS in 1h 26m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3d05cd365e5793293d8e668e087e5f16ce970387","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-02 15:41:45.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b99f3e462a350473e15baf45d4ebe478067f7d19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 17:49:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/32/668032/3/check/openstack-tox-cover/80b1b77/cover/ : SUCCESS in 7m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/668032/3/check/openstack-tox-lower-constraints/2a55983/ : SUCCESS in 9m 25s\n- openstack-tox-pep8 http://logs.openstack.org/32/668032/3/check/openstack-tox-pep8/49cadf3/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/32/668032/3/check/openstack-tox-py27/a4fe9b7/ : SUCCESS in 3m 31s\n- openstack-tox-py36 http://logs.openstack.org/32/668032/3/check/openstack-tox-py36/01fda8f/ : SUCCESS in 4m 09s\n- openstack-tox-py37 http://logs.openstack.org/32/668032/3/check/openstack-tox-py37/dc9e0fe/ : SUCCESS in 5m 54s\n- openstack-tox-docs http://logs.openstack.org/32/668032/3/check/openstack-tox-docs/d9b6c7d/html/ : SUCCESS in 5m 47s\n- build-openstack-releasenotes http://logs.openstack.org/32/668032/3/check/build-openstack-releasenotes/22eb505/html/ : SUCCESS in 4m 11s\n- openstack-tox-functional http://logs.openstack.org/32/668032/3/check/openstack-tox-functional/490134e/ : SUCCESS in 4m 42s\n- openstack-tox-functional-py36 http://logs.openstack.org/32/668032/3/check/openstack-tox-functional-py36/3fd31cf/ : SUCCESS in 5m 22s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-noop-api/a445d95/ : SUCCESS in 43m 58s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-noop-py2-api/317fea5/ : SUCCESS in 40m 51s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-scenario/cbfba5e/ : SUCCESS in 1h 47m 27s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-py2-scenario/e815ea8/ : SUCCESS in 1h 26m 38s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-py2-scenario-centos-7/c896d3b/ : FAILURE in 2h 00m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/6928b82/ : SUCCESS in 1h 37m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-scenario-two-node/b4200c9/ : FAILURE in 1h 43m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-py2-scenario-two-node/78db368/ : FAILURE in 1h 34m 12s (non-voting)\n- octavia-grenade http://logs.openstack.org/32/668032/3/check/octavia-grenade/c237f74/ : SUCCESS in 1h 04m 44s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-tls-barbican/8c81f11/ : SUCCESS in 51m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/32/668032/3/check/octavia-v2-dsvm-scenario-amphora-v2/a0939ee/ : SUCCESS in 1h 40m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"23178f56112ff897869c4632b6ddbe0d44d20067","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-07-03 04:14:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ac76e5ca917607d72b2b3e8c1dc951e4a3ab1350","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 04:14:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ec1b0b2e201bd5cb3f3faa0092d3fcf8c21ac067","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 05:52:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"27fea26e5edd4976ce96f0f931584cab713f34f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 05:52:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/32/668032/3/gate/openstack-tox-lower-constraints/e6c6d68/ : SUCCESS in 10m 02s\n- openstack-tox-pep8 http://logs.openstack.org/32/668032/3/gate/openstack-tox-pep8/8f47239/ : SUCCESS in 8m 51s\n- openstack-tox-py27 http://logs.openstack.org/32/668032/3/gate/openstack-tox-py27/b1bd38b/ : SUCCESS in 6m 08s\n- openstack-tox-py36 http://logs.openstack.org/32/668032/3/gate/openstack-tox-py36/6ee2c52/ : SUCCESS in 6m 54s\n- openstack-tox-py37 http://logs.openstack.org/32/668032/3/gate/openstack-tox-py37/3c1299c/ : SUCCESS in 5m 17s\n- openstack-tox-docs http://logs.openstack.org/32/668032/3/gate/openstack-tox-docs/b2e5d0f/html/ : SUCCESS in 4m 42s\n- build-openstack-releasenotes http://logs.openstack.org/32/668032/3/gate/build-openstack-releasenotes/ed6d965/html/ : SUCCESS in 5m 26s\n- openstack-tox-functional http://logs.openstack.org/32/668032/3/gate/openstack-tox-functional/e1f95a4/ : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 http://logs.openstack.org/32/668032/3/gate/openstack-tox-functional-py36/655a45c/ : SUCCESS in 4m 52s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/32/668032/3/gate/octavia-v2-dsvm-noop-api/56f945e/ : SUCCESS in 39m 28s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/32/668032/3/gate/octavia-v2-dsvm-noop-py2-api/4ea3b6e/ : SUCCESS in 52m 59s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/32/668032/3/gate/octavia-v2-dsvm-scenario/18614f7/ : SUCCESS in 1h 32m 35s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/32/668032/3/gate/octavia-v2-dsvm-py2-scenario/dc5a938/ : SUCCESS in 1h 24m 05s\n- octavia-grenade http://logs.openstack.org/32/668032/3/gate/octavia-grenade/58bc860/ : SUCCESS in 1h 17m 46s","accounts_in_message":[],"_revision_number":3},{"id":"2a1b8b9d469a79ed6e2b33163de76a1fa5b59f9c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 12:49:48.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 15a40d31b7dbd1660826d7d078b94cca6f3dbd09","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"feaa1ce02170cd2ff3eddfe28b1e61a6b07e15a6","revisions":{"9263defe5f1b36c2d8a1e1314f07902d2a2d985d":{"kind":"REWORK","_number":1,"created":"2019-06-27 19:34:37.000000000","uploader":{"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"},"ref":"refs/changes/32/668032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/668032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/668032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/668032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/668032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/668032/1"}}},"commit":{"parents":[{"commit":"520f1d0c7d69074535d86138acece59ca3e8db0d","subject":"Merge \"Switch python3 versions of test jobs to match Train PTI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/520f1d0c7d69074535d86138acece59ca3e8db0d"}]}],"author":{"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","date":"2019-06-27 19:28:52.000000000","tz":-240},"committer":{"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","date":"2019-06-27 19:34:31.000000000","tz":-240},"subject":"only rollback DB when we have a connection to the DB","message":"only rollback DB when we have a connection to the DB\n\nChange-Id: I728a9aa82986edef8128907a48c9df87d119fa8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9263defe5f1b36c2d8a1e1314f07902d2a2d985d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9263defe5f1b36c2d8a1e1314f07902d2a2d985d"}]},"branch":"refs/heads/master"},"d069cbfa69858de7a84bf86f67c2732188d45994":{"kind":"REWORK","_number":2,"created":"2019-06-27 19:48:52.000000000","uploader":{"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"},"ref":"refs/changes/32/668032/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/668032/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/668032/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/668032/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/668032/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/668032/2"}}},"commit":{"parents":[{"commit":"520f1d0c7d69074535d86138acece59ca3e8db0d","subject":"Merge \"Switch python3 versions of test jobs to match Train PTI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/520f1d0c7d69074535d86138acece59ca3e8db0d"}]}],"author":{"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","date":"2019-06-27 19:28:52.000000000","tz":-240},"committer":{"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","date":"2019-06-27 19:48:47.000000000","tz":-240},"subject":"only rollback DB when we have a connection to the DB","message":"only rollback DB when we have a connection to the DB\n\nChange-Id: I728a9aa82986edef8128907a48c9df87d119fa8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d069cbfa69858de7a84bf86f67c2732188d45994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d069cbfa69858de7a84bf86f67c2732188d45994"}]},"branch":"refs/heads/master"},"feaa1ce02170cd2ff3eddfe28b1e61a6b07e15a6":{"kind":"REWORK","_number":3,"created":"2019-06-27 20:06:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/32/668032/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/668032/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/668032/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/668032/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/668032/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/668032/3"}}},"commit":{"parents":[{"commit":"520f1d0c7d69074535d86138acece59ca3e8db0d","subject":"Merge \"Switch python3 versions of test jobs to match Train PTI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/520f1d0c7d69074535d86138acece59ca3e8db0d"}]}],"author":{"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","date":"2019-06-27 19:28:52.000000000","tz":-240},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 20:06:50.000000000","tz":-420},"subject":"only rollback DB when we have a connection to the DB","message":"only rollback DB when we have a connection to the DB\n\nStory: 2006062\nTask: 34775\n\nChange-Id: I728a9aa82986edef8128907a48c9df87d119fa8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/feaa1ce02170cd2ff3eddfe28b1e61a6b07e15a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/feaa1ce02170cd2ff3eddfe28b1e61a6b07e15a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
