)]}'
{"id":"openstack%2Foctavia~668068","triplet_id":"openstack%2Foctavia~master~Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b","project":"openstack/octavia","branch":"master","topic":"single-process","hashtags":[],"change_id":"Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b","subject":"Fix multi-listener load balancers","status":"MERGED","created":"2019-06-27 23:03:07.000000000","updated":"2019-07-29 22:14:04.000000000","submitted":"2019-07-29 22:14:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":111,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668068-1564438444650-bedbf1ba","meta_rev_id":"c1aa735418cb001024d7f0d76ffce69f0c5c928a","_number":668068,"virtual_id_number":668068,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-07-29 22:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-07-26 21:32:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":2,"date":"2019-07-29 20:21:34.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":1,"date":"2019-07-29 20:21:34.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":2,"date":"2019-07-29 20:21:34.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-09 13:29:08.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-07-11 13:03:56.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-07-26 21:32:08.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-07-29 20:21:34.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-07-29 22:14:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e911ae76dae533dd892f333c08511a179178980d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-27 23:03:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46c38d7743241b3c20e593ddaaf10472bbc9ea5e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-27 23:03:27.000000000","message":"Patch Set 1: Workflow-1\n\nJust a start","accounts_in_message":[],"_revision_number":1},{"id":"0a5161515349408493cf23d12c1a1061b61280fd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-27 23:19:22.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3cb487360a63d4741de7ca0ff7e36e62cfbfac16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 01:05:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/1/check/openstack-tox-cover/d73e9bd/ : FAILURE in 4m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/1/check/openstack-tox-lower-constraints/39268f4/ : FAILURE in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/1/check/openstack-tox-pep8/4b8fa1d/ : SUCCESS in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/1/check/openstack-tox-py27/633474f/ : FAILURE in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/1/check/openstack-tox-py36/7076c94/ : FAILURE in 4m 27s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/1/check/openstack-tox-py37/f7bd0ce/ : FAILURE in 4m 02s\n- openstack-tox-docs http://logs.openstack.org/68/668068/1/check/openstack-tox-docs/059bbc5/html/ : SUCCESS in 5m 41s\n- openstack-tox-functional http://logs.openstack.org/68/668068/1/check/openstack-tox-functional/e8e8cde/ : SUCCESS in 7m 20s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/1/check/openstack-tox-functional-py36/160af03/ : SUCCESS in 6m 48s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/1/check/octavia-v2-dsvm-noop-api/9e795f5/ : FAILURE in 49m 18s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/1/check/octavia-v2-dsvm-noop-py2-api/77413c0/ : FAILURE in 38m 37s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/1/check/octavia-v2-dsvm-scenario/f6d38b4/ : FAILURE in 1h 24m 00s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/1/check/octavia-v2-dsvm-py2-scenario/af02595/ : FAILURE in 1h 20m 33s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/1/check/octavia-v2-dsvm-py2-scenario-centos-7/2fd00ad/ : FAILURE in 1h 54m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/390fd37/ : FAILURE in 1h 28m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/1/check/octavia-v2-dsvm-scenario-two-node/55380be/ : FAILURE in 1h 09m 32s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/1/check/octavia-v2-dsvm-py2-scenario-two-node/8fd0ada/ : FAILURE in 1h 08m 07s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/1/check/octavia-grenade/6912e68/ : SUCCESS in 1h 16m 30s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/1/check/octavia-v2-dsvm-tls-barbican/e65b1dc/ : FAILURE in 1h 06m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/1/check/octavia-v2-dsvm-scenario-amphora-v2/4f590df/ : FAILURE in 1h 19m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"81d789b63135d3418aa1194d49b7e0f3d0da5dbc","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-28 06:30:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b704f7c12805a0aedf86b30ccb538bc20075361e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-28 06:32:18.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b028bf91088a7576ad610fe05cc944a10dce81b2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-28 06:36:53.000000000","message":"Patch Set 2: Workflow-1\n\n(5 comments)\n\nI had to make a bunch of assumptions about how to handle some \"listener specific\" stuff that goes on pools, now that there\u0027s no guarantee of it being a 1:1 relationship. I don\u0027t know a way to make these all behave *exactly* the way they\u0027re intended to work unless we just give up and duplicate every pool as a pool+listener combo. :/","accounts_in_message":[],"_revision_number":2},{"id":"9adbff6e56ca4f6cb5aad51a872402597e6b94a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 08:54:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/2/check/openstack-tox-cover/47ac9c0/cover/ : SUCCESS in 8m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/2/check/openstack-tox-lower-constraints/187024f/ : SUCCESS in 7m 05s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/2/check/openstack-tox-pep8/9205179/ : SUCCESS in 5m 28s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/2/check/openstack-tox-py27/e6f65ac/ : SUCCESS in 4m 35s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/2/check/openstack-tox-py36/78ed264/ : SUCCESS in 5m 04s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/2/check/openstack-tox-py37/bd2c775/ : SUCCESS in 4m 26s\n- openstack-tox-docs http://logs.openstack.org/68/668068/2/check/openstack-tox-docs/093dad3/html/ : SUCCESS in 5m 38s\n- openstack-tox-functional http://logs.openstack.org/68/668068/2/check/openstack-tox-functional/b6e2223/ : SUCCESS in 5m 53s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/2/check/openstack-tox-functional-py36/6f3e35e/ : SUCCESS in 5m 48s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/2/check/octavia-v2-dsvm-noop-api/f20cf73/ : FAILURE in 41m 56s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/2/check/octavia-v2-dsvm-noop-py2-api/8c1e5f6/ : FAILURE in 42m 37s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/2/check/octavia-v2-dsvm-scenario/c07a722/ : FAILURE in 1h 35m 31s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/2/check/octavia-v2-dsvm-py2-scenario/39c466d/ : FAILURE in 1h 41m 40s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/2/check/octavia-v2-dsvm-py2-scenario-centos-7/1fddb07/ : FAILURE in 2h 18m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/efa0027/ : FAILURE in 1h 31m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/2/check/octavia-v2-dsvm-scenario-two-node/241c339/ : FAILURE in 1h 55m 13s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/2/check/octavia-v2-dsvm-py2-scenario-two-node/4bc0698/ : FAILURE in 1h 46m 18s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/2/check/octavia-grenade/2b1831d/ : SUCCESS in 1h 12m 01s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/2/check/octavia-v2-dsvm-tls-barbican/db00c5f/ : SUCCESS in 53m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/2/check/octavia-v2-dsvm-scenario-amphora-v2/d157cbe/ : FAILURE in 1h 38m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"06469b1fc0c8fce2ce92806c0323593dfbdbdfb2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-28 16:22:59.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c5e0e4b391fe625825a90db565b4b5f9429efc8d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-29 00:57:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e3693c97553ebd2f09ef24e8bff9d0e1cd82d984","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-29 01:30:34.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0b9902b361f3f7f14037d142c0637f6f6f3134a5","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-29 01:52:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"99e05ca19639b5144ef9cf96a8ce5ab4919c047e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-29 01:57:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1024df7696e22e9006d79f6936aed2ad2c6ebcea","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-29 01:59:02.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b09af062b55f40ab86f5a6b447c8093a672b455d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 04:05:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/5/check/openstack-tox-cover/45e6856/cover/ : SUCCESS in 6m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/5/check/openstack-tox-lower-constraints/6528fe4/ : SUCCESS in 5m 35s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/5/check/openstack-tox-pep8/83a8ab7/ : SUCCESS in 5m 06s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/5/check/openstack-tox-py27/58e0730/ : SUCCESS in 5m 12s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/5/check/openstack-tox-py36/c1b2c4e/ : SUCCESS in 4m 58s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/5/check/openstack-tox-py37/526ed41/ : SUCCESS in 4m 36s\n- openstack-tox-docs http://logs.openstack.org/68/668068/5/check/openstack-tox-docs/1e78dfd/html/ : SUCCESS in 5m 58s\n- openstack-tox-functional http://logs.openstack.org/68/668068/5/check/openstack-tox-functional/07f565c/ : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/5/check/openstack-tox-functional-py36/0a560fc/ : SUCCESS in 6m 02s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/5/check/octavia-v2-dsvm-noop-api/98f37fe/ : SUCCESS in 36m 40s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/5/check/octavia-v2-dsvm-noop-py2-api/3ccd89e/ : SUCCESS in 38m 53s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/5/check/octavia-v2-dsvm-scenario/bd8d501/ : FAILURE in 1h 31m 23s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/5/check/octavia-v2-dsvm-py2-scenario/1fa9ee2/ : FAILURE in 1h 16m 59s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/5/check/octavia-v2-dsvm-py2-scenario-centos-7/b061867/ : FAILURE in 2h 01m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/5/check/octavia-v2-dsvm-scenario-ubuntu-bionic/9db23c7/ : FAILURE in 1h 35m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/5/check/octavia-v2-dsvm-scenario-two-node/1732952/ : FAILURE in 1h 43m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/5/check/octavia-v2-dsvm-py2-scenario-two-node/53b7cf4/ : FAILURE in 1h 50m 00s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/5/check/octavia-grenade/454ba5c/ : SUCCESS in 1h 15m 10s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/5/check/octavia-v2-dsvm-tls-barbican/bfe612f/ : SUCCESS in 43m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/5/check/octavia-v2-dsvm-scenario-amphora-v2/b19a888/ : FAILURE in 1h 14m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bd52e957b769749973b65acb02579134047cdf1e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-30 05:49:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c8d60f8d95719eb291b298676bb6effce000b293","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-30 09:21:59.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/6/check/openstack-tox-cover/95f20e6/cover/ : SUCCESS in 6m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/6/check/openstack-tox-lower-constraints/754338a/ : SUCCESS in 7m 22s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/6/check/openstack-tox-pep8/f0f6926/ : SUCCESS in 5m 05s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/6/check/openstack-tox-py27/ece42e2/ : SUCCESS in 4m 31s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/6/check/openstack-tox-py36/47e4003/ : SUCCESS in 5m 23s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/6/check/openstack-tox-py37/efaf1c3/ : SUCCESS in 6m 20s\n- openstack-tox-docs http://logs.openstack.org/68/668068/6/check/openstack-tox-docs/76e5746/html/ : SUCCESS in 6m 18s\n- openstack-tox-functional http://logs.openstack.org/68/668068/6/check/openstack-tox-functional/d3a5f54/ : SUCCESS in 5m 33s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/6/check/openstack-tox-functional-py36/ecf7f2e/ : SUCCESS in 5m 27s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/6/check/octavia-v2-dsvm-noop-api/8fb4567/ : SUCCESS in 47m 34s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/6/check/octavia-v2-dsvm-noop-py2-api/eaf1f78/ : SUCCESS in 48m 03s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/6/check/octavia-v2-dsvm-scenario/3f3d0ee/ : FAILURE in 1h 31m 08s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/6/check/octavia-v2-dsvm-py2-scenario/42776dc/ : FAILURE in 1h 30m 57s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/6/check/octavia-v2-dsvm-py2-scenario-centos-7/912f070/ : FAILURE in 1h 52m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/6/check/octavia-v2-dsvm-scenario-ubuntu-bionic/bea0ab4/ : FAILURE in 1h 47m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/6/check/octavia-v2-dsvm-scenario-two-node/7a1b9ea/ : FAILURE in 1h 58m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/6/check/octavia-v2-dsvm-py2-scenario-two-node/10664b2/ : FAILURE in 1h 24m 50s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/6/check/octavia-grenade/8cb695f/ : SUCCESS in 1h 21m 12s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/6/check/octavia-v2-dsvm-tls-barbican/5b2b07f/ : SUCCESS in 54m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/6/check/octavia-v2-dsvm-scenario-amphora-v2/73d1c11/ : FAILURE in 1h 29m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b02425841b54c72bb6a10bcb461cb75079b99379","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-30 11:28:15.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b968f71b9ec813eaa987d42711fc8d5523ead31e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-30 13:57:19.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/7/check/openstack-tox-cover/d62132d/ : FAILURE in 3m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/7/check/openstack-tox-lower-constraints/f7676cd/ : FAILURE in 3m 38s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/7/check/openstack-tox-pep8/2300166/ : FAILURE in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/7/check/openstack-tox-py27/f17482e/ : SUCCESS in 4m 12s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/7/check/openstack-tox-py36/4269e13/ : FAILURE in 3m 57s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/7/check/openstack-tox-py37/d17bd08/ : FAILURE in 5m 14s\n- openstack-tox-docs http://logs.openstack.org/68/668068/7/check/openstack-tox-docs/87e14d0/html/ : SUCCESS in 6m 46s\n- openstack-tox-functional http://logs.openstack.org/68/668068/7/check/openstack-tox-functional/4073303/ : SUCCESS in 5m 45s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/7/check/openstack-tox-functional-py36/00eafee/ : SUCCESS in 6m 18s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/7/check/octavia-v2-dsvm-noop-api/1c83884/ : SUCCESS in 51m 35s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/7/check/octavia-v2-dsvm-noop-py2-api/b458b70/ : SUCCESS in 52m 42s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/7/check/octavia-v2-dsvm-scenario/0060046/ : FAILURE in 1h 36m 09s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/7/check/octavia-v2-dsvm-py2-scenario/3b4359a/ : FAILURE in 1h 27m 31s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/7/check/octavia-v2-dsvm-py2-scenario-centos-7/213ce4f/ : FAILURE in 2h 28m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/7/check/octavia-v2-dsvm-scenario-ubuntu-bionic/f2dcc2b/ : FAILURE in 1h 35m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/7/check/octavia-v2-dsvm-scenario-two-node/fe9838b/ : FAILURE in 1h 09m 59s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/7/check/octavia-v2-dsvm-py2-scenario-two-node/1d3f8ae/ : FAILURE in 1h 36m 16s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/7/check/octavia-grenade/dbab17d/ : SUCCESS in 1h 23m 50s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/7/check/octavia-v2-dsvm-tls-barbican/4262328/ : SUCCESS in 59m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/7/check/octavia-v2-dsvm-scenario-amphora-v2/45d2541/ : FAILURE in 1h 41m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"37fd7ede4819ccb9cefce6e4dc5be0d83ecb6d78","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-30 16:47:04.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"11fb935e392f44ab77cfd50ada1bddc45e1bc66b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-30 16:54:52.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2d2a1b7e2754deff4e1c4b2b725d7bcb4372259b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-30 18:55:08.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/9/check/openstack-tox-cover/b6a0531/cover/ : SUCCESS in 7m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/9/check/openstack-tox-lower-constraints/3952693/ : SUCCESS in 7m 19s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/9/check/openstack-tox-pep8/9bbe666/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/9/check/openstack-tox-py27/a208d78/ : SUCCESS in 4m 03s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/9/check/openstack-tox-py36/b709037/ : SUCCESS in 3m 57s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/9/check/openstack-tox-py37/0ee4300/ : SUCCESS in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/68/668068/9/check/openstack-tox-docs/174857e/html/ : SUCCESS in 4m 54s\n- openstack-tox-functional http://logs.openstack.org/68/668068/9/check/openstack-tox-functional/e25ed2a/ : SUCCESS in 4m 56s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/9/check/openstack-tox-functional-py36/c700acc/ : SUCCESS in 5m 55s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/9/check/octavia-v2-dsvm-noop-api/44b5f29/ : FAILURE in 42m 14s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/9/check/octavia-v2-dsvm-noop-py2-api/c396cb7/ : FAILURE in 44m 26s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/9/check/octavia-v2-dsvm-scenario/1bd488d/ : SUCCESS in 1h 13m 15s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/9/check/octavia-v2-dsvm-py2-scenario/acccc50/ : SUCCESS in 1h 28m 09s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/9/check/octavia-v2-dsvm-py2-scenario-centos-7/5a0c445/ : FAILURE in 1h 54m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/9/check/octavia-v2-dsvm-scenario-ubuntu-bionic/c03fe38/ : SUCCESS in 1h 29m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/9/check/octavia-v2-dsvm-scenario-two-node/97e825d/ : SUCCESS in 1h 45m 47s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/9/check/octavia-v2-dsvm-py2-scenario-two-node/76c069b/ : FAILURE in 1h 28m 26s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/9/check/octavia-grenade/6fc80d7/ : SUCCESS in 1h 12m 26s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/9/check/octavia-v2-dsvm-tls-barbican/a5eea7f/ : SUCCESS in 48m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/9/check/octavia-v2-dsvm-scenario-amphora-v2/4d71371/ : SUCCESS in 1h 15m 40s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e898f2d32b4b3cf1daa2dab2d6b822549c1ff1f9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-02 00:27:29.000000000","message":"Patch Set 9: Workflow-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"305774db14ff3d9fa4e8b7a8c36bb8a68706a730","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-02 14:10:03.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"94f80598c55e3f90dc262b03252d526689cfe635","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-02 16:45:08.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"2d5a8df03e511c21dc6b2d15977fd546ffafb0ac","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-02 18:02:00.000000000","message":"Patch Set 10: Workflow-1","accounts_in_message":[],"_revision_number":10},{"id":"0ad5cd0cab171ccb886fc260823661edf8d64324","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 19:18:34.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/10/check/openstack-tox-cover/560409a/ : FAILURE in 6m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/10/check/openstack-tox-lower-constraints/60cc5b4/ : FAILURE in 5m 42s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/10/check/openstack-tox-pep8/e2334bc/ : SUCCESS in 5m 28s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/10/check/openstack-tox-py27/ae48767/ : FAILURE in 6m 13s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/10/check/openstack-tox-py36/2a70b20/ : FAILURE in 3m 57s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/10/check/openstack-tox-py37/30af648/ : FAILURE in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/68/668068/10/check/openstack-tox-docs/0633dca/html/ : SUCCESS in 5m 36s\n- openstack-tox-functional http://logs.openstack.org/68/668068/10/check/openstack-tox-functional/87ea401/ : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/10/check/openstack-tox-functional-py36/3427e7f/ : SUCCESS in 5m 48s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/10/check/octavia-v2-dsvm-noop-api/8fb02ec/ : FAILURE in 49m 04s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/10/check/octavia-v2-dsvm-noop-py2-api/e7324fc/ : SUCCESS in 47m 36s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/10/check/octavia-v2-dsvm-scenario/586eb4f/ : SUCCESS in 1h 41m 02s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/10/check/octavia-v2-dsvm-py2-scenario/5d7eb43/ : SUCCESS in 1h 34m 09s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/10/check/octavia-v2-dsvm-py2-scenario-centos-7/ba1b660/ : FAILURE in 2h 12m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/10/check/octavia-v2-dsvm-scenario-ubuntu-bionic/61210b3/ : SUCCESS in 1h 34m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/10/check/octavia-v2-dsvm-scenario-two-node/810ed75/ : FAILURE in 1h 23m 00s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/10/check/octavia-v2-dsvm-py2-scenario-two-node/02fec56/ : SUCCESS in 1h 43m 52s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/10/check/octavia-grenade/b9adb4d/ : SUCCESS in 1h 21m 02s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/10/check/octavia-v2-dsvm-tls-barbican/a77a245/ : SUCCESS in 52m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/10/check/octavia-v2-dsvm-scenario-amphora-v2/6b70733/ : SUCCESS in 1h 22m 00s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d810053ac858233aafd2528c23a2b3e681e675af","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-03 00:51:00.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"11df76d1f84658baf8ea65b877aed1f67fbc7771","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-03 00:51:12.000000000","message":"Patch Set 11: Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"579f0fcfc7c29ce88255f3e045bb3762e81a3725","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 03:03:27.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/11/check/openstack-tox-cover/cc8fa92/ : FAILURE in 3m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/11/check/openstack-tox-lower-constraints/c566629/ : FAILURE in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/11/check/openstack-tox-pep8/711c4c4/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/11/check/openstack-tox-py27/d17379e/ : FAILURE in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/11/check/openstack-tox-py36/ffb1446/ : FAILURE in 5m 26s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/11/check/openstack-tox-py37/458dd55/ : FAILURE in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/68/668068/11/check/openstack-tox-docs/b03ccdd/html/ : SUCCESS in 5m 03s\n- openstack-tox-functional http://logs.openstack.org/68/668068/11/check/openstack-tox-functional/ded8a1b/ : SUCCESS in 5m 48s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/11/check/openstack-tox-functional-py36/4b6d8d4/ : SUCCESS in 4m 57s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/11/check/octavia-v2-dsvm-noop-api/cba9597/ : FAILURE in 46m 22s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/11/check/octavia-v2-dsvm-noop-py2-api/ddbdb21/ : SUCCESS in 47m 26s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/11/check/octavia-v2-dsvm-scenario/464c6c7/ : FAILURE in 1h 24m 50s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/11/check/octavia-v2-dsvm-py2-scenario/1f26518/ : FAILURE in 1h 31m 37s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/11/check/octavia-v2-dsvm-py2-scenario-centos-7/17d2db8/ : FAILURE in 2h 06m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/11/check/octavia-v2-dsvm-scenario-ubuntu-bionic/989281e/ : FAILURE in 1h 33m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/11/check/octavia-v2-dsvm-scenario-two-node/4e2988b/ : FAILURE in 1h 36m 08s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/11/check/octavia-v2-dsvm-py2-scenario-two-node/2a5d06f/ : FAILURE in 1h 53m 28s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/11/check/octavia-grenade/130a447/ : SUCCESS in 1h 10m 53s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/11/check/octavia-v2-dsvm-tls-barbican/73f3002/ : SUCCESS in 52m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/11/check/octavia-v2-dsvm-scenario-amphora-v2/0a540cb/ : FAILURE in 1h 21m 10s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6d6d45ea3a2dc1b9659e16f3b0e1742aeabf1634","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-03 13:34:11.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"af36fa27b9bdaca2d7acb45c165f743682a4c91e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-03 15:43:34.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"2122e596438cd9f5692d6e4b76bbc718d912a192","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 18:08:19.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/12/check/openstack-tox-cover/d6349a0/cover/ : SUCCESS in 6m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/12/check/openstack-tox-lower-constraints/9f107f4/ : SUCCESS in 6m 19s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/12/check/openstack-tox-pep8/fc39eae/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/12/check/openstack-tox-py27/01a419b/ : SUCCESS in 4m 06s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/12/check/openstack-tox-py36/5b54a70/ : SUCCESS in 3m 55s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/12/check/openstack-tox-py37/9f3368c/ : SUCCESS in 4m 49s\n- openstack-tox-docs http://logs.openstack.org/68/668068/12/check/openstack-tox-docs/5d8952c/html/ : SUCCESS in 6m 29s\n- openstack-tox-functional http://logs.openstack.org/68/668068/12/check/openstack-tox-functional/a63ac34/ : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/12/check/openstack-tox-functional-py36/0398ce1/ : SUCCESS in 4m 46s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/12/check/octavia-v2-dsvm-noop-api/fdc15d2/ : FAILURE in 42m 13s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/12/check/octavia-v2-dsvm-noop-py2-api/302b7c9/ : SUCCESS in 52m 55s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/12/check/octavia-v2-dsvm-scenario/5d5b9f0/ : SUCCESS in 1h 28m 49s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/12/check/octavia-v2-dsvm-py2-scenario/3800792/ : SUCCESS in 2h 02m 37s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/12/check/octavia-v2-dsvm-py2-scenario-centos-7/8d240f5/ : FAILURE in 2h 15m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/12/check/octavia-v2-dsvm-scenario-ubuntu-bionic/3106ef0/ : SUCCESS in 1h 25m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/12/check/octavia-v2-dsvm-scenario-two-node/bbdaaab/ : SUCCESS in 1h 53m 51s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/12/check/octavia-v2-dsvm-py2-scenario-two-node/82d3d90/ : FAILURE in 1h 20m 59s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/12/check/octavia-grenade/af10347/ : SUCCESS in 1h 30m 55s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/12/check/octavia-v2-dsvm-tls-barbican/edc3b17/ : SUCCESS in 52m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/12/check/octavia-v2-dsvm-scenario-amphora-v2/79b7820/ : SUCCESS in 1h 33m 26s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"66d28cd27b7e16bd9ac5cf50635568cf1f0cf1ad","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-04 00:30:25.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"fd776403257af4cfd515ca3b212d04f10af5259c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-04 00:30:34.000000000","message":"Patch Set 13: Workflow-1","accounts_in_message":[],"_revision_number":13},{"id":"e203696d80e87e906c0172c8ed43acc69345ffa0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 02:43:01.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/13/check/openstack-tox-cover/85d0257/cover/ : SUCCESS in 6m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/13/check/openstack-tox-lower-constraints/dbba8a8/ : SUCCESS in 6m 26s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/13/check/openstack-tox-pep8/5e94403/ : SUCCESS in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/13/check/openstack-tox-py27/9e91eda/ : SUCCESS in 3m 28s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/13/check/openstack-tox-py36/d0cdd2f/ : SUCCESS in 4m 07s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/13/check/openstack-tox-py37/429c241/ : SUCCESS in 3m 53s\n- openstack-tox-docs http://logs.openstack.org/68/668068/13/check/openstack-tox-docs/72a585c/html/ : SUCCESS in 5m 19s\n- openstack-tox-functional http://logs.openstack.org/68/668068/13/check/openstack-tox-functional/f7f056d/ : SUCCESS in 4m 41s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/13/check/openstack-tox-functional-py36/fa4dff9/ : SUCCESS in 5m 23s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/13/check/octavia-v2-dsvm-noop-api/86bbe7a/ : SUCCESS in 42m 46s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/13/check/octavia-v2-dsvm-noop-py2-api/470f12b/ : SUCCESS in 47m 57s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/13/check/octavia-v2-dsvm-scenario/18e748c/ : SUCCESS in 1h 37m 13s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/13/check/octavia-v2-dsvm-py2-scenario/484a1db/ : SUCCESS in 1h 35m 27s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/13/check/octavia-v2-dsvm-py2-scenario-centos-7/e56b3db/ : FAILURE in 2h 10m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/13/check/octavia-v2-dsvm-scenario-ubuntu-bionic/fc04441/ : SUCCESS in 1h 40m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/13/check/octavia-v2-dsvm-scenario-two-node/c63ef51/ : SUCCESS in 1h 47m 50s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/13/check/octavia-v2-dsvm-py2-scenario-two-node/bebcd22/ : SUCCESS in 1h 48m 18s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/13/check/octavia-grenade/15c8abd/ : SUCCESS in 1h 16m 51s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/13/check/octavia-v2-dsvm-tls-barbican/29a4b97/ : SUCCESS in 51m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/13/check/octavia-v2-dsvm-scenario-amphora-v2/86f2f90/ : SUCCESS in 1h 20m 13s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"f218d59a0a608df388f5b6127961c11d6d7a762a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-04 17:37:58.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"eb54330c0cbd50babb4ff68290e12e9019534992","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-07-08 15:27:27.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"5af51f4b8ba485c67b8a13808a916424c0bd5521","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-08 23:24:00.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"b1237e8ae29fb87681d114d34bea36a41ad12483","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-08 23:24:24.000000000","message":"Patch Set 13: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"e7359669e39538fd92dfc080f487be8ea3734df0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 01:58:33.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/14/check/openstack-tox-cover/65acec0/ : FAILURE in 4m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/14/check/openstack-tox-lower-constraints/ed3faf7/ : FAILURE in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/14/check/openstack-tox-pep8/bb31a05/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/14/check/openstack-tox-py27/013b04c/ : FAILURE in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/14/check/openstack-tox-py36/b231127/ : FAILURE in 4m 06s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/14/check/openstack-tox-py37/0175f2b/ : FAILURE in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/68/668068/14/check/openstack-tox-docs/e2b5107/html/ : SUCCESS in 5m 34s\n- openstack-tox-functional http://logs.openstack.org/68/668068/14/check/openstack-tox-functional/b96775c/ : SUCCESS in 6m 20s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/14/check/openstack-tox-functional-py36/59bae32/ : SUCCESS in 4m 57s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/14/check/octavia-v2-dsvm-noop-api/dce71e3/ : SUCCESS in 41m 45s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/14/check/octavia-v2-dsvm-noop-py2-api/e6b34fc/ : SUCCESS in 49m 04s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/14/check/octavia-v2-dsvm-scenario/d691ea0/ : SUCCESS in 1h 18m 31s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/14/check/octavia-v2-dsvm-py2-scenario/f48c7c3/ : SUCCESS in 1h 42m 18s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/14/check/octavia-v2-dsvm-py2-scenario-centos-7/f4e3c5e/ : FAILURE in 2h 17m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/14/check/octavia-v2-dsvm-scenario-ubuntu-bionic/16a2b6e/ : SUCCESS in 1h 44m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/14/check/octavia-v2-dsvm-scenario-two-node/ef9b7f0/ : FAILURE in 2h 22m 35s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/14/check/octavia-v2-dsvm-py2-scenario-two-node/d20cc1f/ : SUCCESS in 1h 42m 35s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/14/check/octavia-grenade/7d562a5/ : SUCCESS in 1h 16m 49s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/14/check/octavia-v2-dsvm-tls-barbican/24e4eac/ : SUCCESS in 53m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/14/check/octavia-v2-dsvm-scenario-amphora-v2/3eb46f9/ : SUCCESS in 1h 14m 25s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e0259ca7330dbda110de3c5c72928e9888cd8c64","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-07-09 08:12:10.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"1ec3401d1bdd3d25a98fd479278ce13ab17c7a42","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-09 15:19:04.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"18a279dd2a84a230c3d9835a5d98729900e1b6d4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-09 16:09:51.000000000","message":"Patch Set 14:\n\n(2 comments)\n\nnext PS","accounts_in_message":[],"_revision_number":14},{"id":"259572d986bf6f072e727c980160d88439977b33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 17:39:54.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/15/check/openstack-tox-cover/ca60811/ : FAILURE in 4m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/15/check/openstack-tox-lower-constraints/0cf0f10/ : FAILURE in 4m 15s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/15/check/openstack-tox-pep8/d6483b8/ : FAILURE in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/15/check/openstack-tox-py27/3fcf02a/ : FAILURE in 5m 11s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/15/check/openstack-tox-py36/b23c95d/ : FAILURE in 3m 53s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/15/check/openstack-tox-py37/a80ba0f/ : FAILURE in 5m 05s\n- openstack-tox-docs http://logs.openstack.org/68/668068/15/check/openstack-tox-docs/0d5aac2/html/ : SUCCESS in 5m 00s\n- openstack-tox-functional http://logs.openstack.org/68/668068/15/check/openstack-tox-functional/6733528/ : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/15/check/openstack-tox-functional-py36/481757e/ : SUCCESS in 4m 51s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/15/check/octavia-v2-dsvm-noop-api/ebe047a/ : SUCCESS in 42m 54s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/15/check/octavia-v2-dsvm-noop-py2-api/40fc183/ : SUCCESS in 50m 11s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/15/check/octavia-v2-dsvm-scenario/f883666/ : FAILURE in 1h 36m 52s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/15/check/octavia-v2-dsvm-py2-scenario/4bf8800/ : SUCCESS in 1h 39m 02s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/15/check/octavia-v2-dsvm-py2-scenario-centos-7/e5c216f/ : FAILURE in 2h 13m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/15/check/octavia-v2-dsvm-scenario-ubuntu-bionic/991efc2/ : FAILURE in 1h 33m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/15/check/octavia-v2-dsvm-scenario-two-node/f9a595c/ : FAILURE in 1h 49m 28s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/15/check/octavia-v2-dsvm-py2-scenario-two-node/151743d/ : SUCCESS in 1h 13m 17s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/15/check/octavia-grenade/99cbe4c/ : SUCCESS in 1h 13m 47s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/15/check/octavia-v2-dsvm-tls-barbican/c1afe2d/ : FAILURE in 1h 07m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/15/check/octavia-v2-dsvm-scenario-amphora-v2/deaace6/ : FAILURE in 1h 15m 08s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"130dd6be94a2e648a4c3c6c235afd8cced7789ee","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-09 18:03:29.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"ff3553b252797f4633412a144311a509c73942ec","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-09 18:17:51.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"c4faff7ab367b3eeeb3600111c3f6f6b606464e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 20:59:15.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/17/check/openstack-tox-cover/9b6ea63/ : FAILURE in 5m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/17/check/openstack-tox-lower-constraints/06ed635/ : FAILURE in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/17/check/openstack-tox-pep8/299ac53/ : FAILURE in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/17/check/openstack-tox-py27/3fd3ad7/ : FAILURE in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/17/check/openstack-tox-py36/05ba98f/ : FAILURE in 4m 30s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/17/check/openstack-tox-py37/9e64b25/ : FAILURE in 5m 02s\n- openstack-tox-docs http://logs.openstack.org/68/668068/17/check/openstack-tox-docs/df8a4e2/html/ : SUCCESS in 5m 48s\n- openstack-tox-functional http://logs.openstack.org/68/668068/17/check/openstack-tox-functional/43cd659/ : FAILURE in 5m 37s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/17/check/openstack-tox-functional-py36/b38a870/ : FAILURE in 6m 19s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/17/check/octavia-v2-dsvm-noop-api/066f1c5/ : SUCCESS in 40m 31s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/17/check/octavia-v2-dsvm-noop-py2-api/082eaba/ : SUCCESS in 40m 51s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/17/check/octavia-v2-dsvm-scenario/4a75693/ : FAILURE in 1h 23m 02s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/17/check/octavia-v2-dsvm-py2-scenario/f7c20c4/ : FAILURE in 1h 27m 54s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/17/check/octavia-v2-dsvm-py2-scenario-centos-7/1b4f902/ : TIMED_OUT in 2h 34m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/17/check/octavia-v2-dsvm-scenario-ubuntu-bionic/a531e67/ : FAILURE in 1h 23m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/17/check/octavia-v2-dsvm-scenario-two-node/6b7ad54/ : FAILURE in 1h 24m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/17/check/octavia-v2-dsvm-py2-scenario-two-node/0d24474/ : FAILURE in 1h 17m 43s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/17/check/octavia-grenade/b145e2f/ : SUCCESS in 1h 15m 46s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/17/check/octavia-v2-dsvm-tls-barbican/30bdfc1/ : FAILURE in 48m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/17/check/octavia-v2-dsvm-scenario-amphora-v2/9c16a09/ : FAILURE in 1h 04m 12s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"b768ec765e8eb2304ffc2465b23ee5a77c0909c5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-10 01:38:30.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"cdc8802042a0946c32193ca8de29d0932ba1d330","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 03:39:29.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/18/check/openstack-tox-cover/98f0e7b/ : FAILURE in 4m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/18/check/openstack-tox-lower-constraints/19f393e/ : FAILURE in 4m 43s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/18/check/openstack-tox-pep8/c174797/ : SUCCESS in 4m 53s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/18/check/openstack-tox-py27/0630269/ : FAILURE in 4m 16s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/18/check/openstack-tox-py36/2483e62/ : FAILURE in 5m 31s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/18/check/openstack-tox-py37/5e2acb4/ : FAILURE in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/68/668068/18/check/openstack-tox-docs/c04d667/html/ : SUCCESS in 5m 08s\n- openstack-tox-functional http://logs.openstack.org/68/668068/18/check/openstack-tox-functional/846bcc1/ : FAILURE in 5m 13s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/18/check/openstack-tox-functional-py36/1a1d2ec/ : FAILURE in 5m 01s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/18/check/octavia-v2-dsvm-noop-api/37be068/ : SUCCESS in 44m 02s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/18/check/octavia-v2-dsvm-noop-py2-api/07f7fcc/ : SUCCESS in 46m 38s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/18/check/octavia-v2-dsvm-scenario/c3c03a3/ : SUCCESS in 1h 28m 58s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/18/check/octavia-v2-dsvm-py2-scenario/6cdb342/ : SUCCESS in 1h 25m 59s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/18/check/octavia-v2-dsvm-py2-scenario-centos-7/12f5cdf/ : SUCCESS in 1h 53m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/18/check/octavia-v2-dsvm-scenario-ubuntu-bionic/b6f5091/ : SUCCESS in 1h 32m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/18/check/octavia-v2-dsvm-scenario-two-node/da89d73/ : SUCCESS in 1h 32m 35s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/18/check/octavia-v2-dsvm-py2-scenario-two-node/c3b99df/ : SUCCESS in 1h 44m 25s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/18/check/octavia-grenade/1d3818d/ : SUCCESS in 1h 07m 28s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/18/check/octavia-v2-dsvm-tls-barbican/5189649/ : SUCCESS in 43m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/18/check/octavia-v2-dsvm-scenario-amphora-v2/c9e5eed/ : SUCCESS in 1h 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"fc6876f25218fb99120722de9f0323582dd39b49","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-10 03:45:40.000000000","message":"Patch Set 18: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"2b858a330e0d4b3cb91f6da7ce415e2b5195d2ba","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-10 03:56:08.000000000","message":"Patch Set 18:\n\n(4 comments)","accounts_in_message":[],"_revision_number":18},{"id":"19a2ebc6e360c5f0daa534269023f0b00cc9cbdb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-10 04:05:45.000000000","message":"Patch Set 18: Workflow-1","accounts_in_message":[],"_revision_number":18},{"id":"51b7f0e4715cdbabc69fa62d1de0503db44cd641","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-10 04:08:20.000000000","message":"Patch Set 18:\n\n(3 comments)","accounts_in_message":[],"_revision_number":18},{"id":"c6b9289ea2b11fca169be926167f637d3dbb9265","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-10 13:41:29.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"05eca1527489c534adeb3f46b1550c8e63cf2849","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 15:22:16.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/19/check/openstack-tox-cover/ae28c77/ : FAILURE in 7m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/19/check/openstack-tox-lower-constraints/e9673e8/ : FAILURE in 5m 00s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/19/check/openstack-tox-pep8/9eca795/ : FAILURE in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/19/check/openstack-tox-py27/b3eb376/ : FAILURE in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/19/check/openstack-tox-py36/8fe829b/ : FAILURE in 4m 02s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/19/check/openstack-tox-py37/1472805/ : FAILURE in 7m 51s\n- openstack-tox-docs http://logs.openstack.org/68/668068/19/check/openstack-tox-docs/73a28a1/html/ : SUCCESS in 4m 57s\n- openstack-tox-functional http://logs.openstack.org/68/668068/19/check/openstack-tox-functional/fb539e3/ : FAILURE in 4m 50s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/19/check/openstack-tox-functional-py36/22e2651/ : FAILURE in 6m 04s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/19/check/octavia-v2-dsvm-noop-api/9217f03/ : SUCCESS in 45m 30s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/19/check/octavia-v2-dsvm-noop-py2-api/4da109b/ : SUCCESS in 45m 47s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/19/check/octavia-v2-dsvm-scenario/bf2418c/ : FAILURE in 50m 34s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/19/check/octavia-v2-dsvm-py2-scenario/c8741cf/ : FAILURE in 50m 45s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/19/check/octavia-v2-dsvm-py2-scenario-centos-7/facd8fc/ : FAILURE in 1h 05m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/19/check/octavia-v2-dsvm-scenario-ubuntu-bionic/28ef730/ : FAILURE in 55m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/19/check/octavia-v2-dsvm-scenario-two-node/8832cbb/ : FAILURE in 1h 22m 15s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/19/check/octavia-v2-dsvm-py2-scenario-two-node/096c306/ : FAILURE in 1h 19m 51s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/19/check/octavia-grenade/0c7b7fc/ : SUCCESS in 1h 30m 50s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/19/check/octavia-v2-dsvm-tls-barbican/db4493e/ : FAILURE in 49m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/19/check/octavia-v2-dsvm-scenario-amphora-v2/75bf9a2/ : FAILURE in 47m 14s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"1ae93a9f8bba7e8914b81019126658d58f239fa8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-10 16:56:26.000000000","message":"Patch Set 19: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"a9f61b7d3ce613c221f2815355e7e7a2ffbbff25","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-10 22:01:54.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"3bf74da2844c79ad23bcd9076881d4f21f224717","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-11 00:33:26.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"abd7a202611de15a1805005ca99a5007b3c3bc23","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-11 00:35:27.000000000","message":"Patch Set 20: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"5b11d20a0266ff41badbf928d87868cdd68bc79a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 02:34:57.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/20/check/openstack-tox-cover/4733821/ : FAILURE in 4m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/20/check/openstack-tox-lower-constraints/8ac3612/ : FAILURE in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/20/check/openstack-tox-pep8/b1430ee/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/20/check/openstack-tox-py27/fe5ed41/ : FAILURE in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/20/check/openstack-tox-py36/a001d44/ : FAILURE in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/20/check/openstack-tox-py37/4a40c30/ : FAILURE in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/68/668068/20/check/openstack-tox-docs/7fccdb1/html/ : SUCCESS in 5m 10s\n- openstack-tox-functional http://logs.openstack.org/68/668068/20/check/openstack-tox-functional/33eb515/ : FAILURE in 4m 53s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/20/check/openstack-tox-functional-py36/44b0c85/ : FAILURE in 5m 09s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/20/check/octavia-v2-dsvm-noop-api/1d3ae83/ : SUCCESS in 49m 46s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/20/check/octavia-v2-dsvm-noop-py2-api/fc25205/ : SUCCESS in 41m 32s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/20/check/octavia-v2-dsvm-scenario/94a14b1/ : SUCCESS in 1h 37m 49s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/20/check/octavia-v2-dsvm-py2-scenario/766233e/ : SUCCESS in 1h 54m 47s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/20/check/octavia-v2-dsvm-py2-scenario-centos-7/8e28f1d/ : FAILURE in 1h 46m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/20/check/octavia-v2-dsvm-scenario-ubuntu-bionic/7547e2a/ : SUCCESS in 1h 46m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/20/check/octavia-v2-dsvm-scenario-two-node/acd8815/ : SUCCESS in 1h 39m 48s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/20/check/octavia-v2-dsvm-py2-scenario-two-node/c35d053/ : SUCCESS in 1h 46m 27s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/20/check/octavia-grenade/870c73f/ : SUCCESS in 1h 12m 23s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/20/check/octavia-v2-dsvm-tls-barbican/a755486/ : SUCCESS in 49m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/20/check/octavia-v2-dsvm-scenario-amphora-v2/e6c7f47/ : SUCCESS in 1h 23m 58s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"acfbd103d2313955792dfe8f6cbce18d2c301243","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-07-11 13:03:56.000000000","message":"Patch Set 20:\n\n(9 comments)\n\nsmall missing things (function renames, removals, etc..) in tests","accounts_in_message":[],"_revision_number":20},{"id":"a2ff6f31f52c7ec75a82931ed8081585e923286b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-11 23:48:07.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"5560224bc26ab3563e48d16ab2b9c6736d58fe55","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-11 23:48:21.000000000","message":"Patch Set 21: Workflow-1","accounts_in_message":[],"_revision_number":21},{"id":"782624c1811f2c0672bc29342a4390b92c3a507a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-11 23:57:23.000000000","message":"Patch Set 21:\n\n(6 comments)","accounts_in_message":[],"_revision_number":21},{"id":"61b1ef95d316e1797a7399fe4c30fd5274cde89d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 01:48:32.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/21/check/openstack-tox-cover/11db863/ : FAILURE in 4m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/21/check/openstack-tox-lower-constraints/f7d9f79/ : FAILURE in 5m 16s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/21/check/openstack-tox-pep8/1b3b092/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/21/check/openstack-tox-py27/8dcb758/ : FAILURE in 4m 12s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/21/check/openstack-tox-py36/63784c3/ : FAILURE in 3m 53s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/21/check/openstack-tox-py37/823049f/ : FAILURE in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/68/668068/21/check/openstack-tox-docs/88c9ea9/html/ : SUCCESS in 5m 06s\n- openstack-tox-functional http://logs.openstack.org/68/668068/21/check/openstack-tox-functional/9a02fbb/ : FAILURE in 5m 07s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/21/check/openstack-tox-functional-py36/b70b552/ : FAILURE in 5m 25s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/21/check/octavia-v2-dsvm-noop-api/39e8d29/ : SUCCESS in 45m 40s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/21/check/octavia-v2-dsvm-noop-py2-api/d89259c/ : SUCCESS in 46m 35s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/21/check/octavia-v2-dsvm-scenario/cb4004a/ : FAILURE in 1h 31m 03s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/21/check/octavia-v2-dsvm-py2-scenario/642657b/ : SUCCESS in 1h 31m 08s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/21/check/octavia-v2-dsvm-py2-scenario-centos-7/a917bc0/ : FAILURE in 1h 54m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/21/check/octavia-v2-dsvm-scenario-ubuntu-bionic/412257d/ : SUCCESS in 1h 44m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/21/check/octavia-v2-dsvm-scenario-two-node/2f9267a/ : SUCCESS in 1h 08m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/21/check/octavia-v2-dsvm-py2-scenario-two-node/a9e1898/ : SUCCESS in 1h 52m 27s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/21/check/octavia-grenade/ad032fe/ : SUCCESS in 1h 14m 22s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/21/check/octavia-v2-dsvm-tls-barbican/e0f2cc8/ : SUCCESS in 1h 01m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/21/check/octavia-v2-dsvm-scenario-amphora-v2/78a3f74/ : SUCCESS in 1h 39m 12s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"d27748770cf058a55085da8f27bc084cd9d3bbb6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-12 07:38:06.000000000","message":"Patch Set 20:\n\n(11 comments)","accounts_in_message":[],"_revision_number":20},{"id":"c1b0d5015bc2145fa9cb1be6700fdf32445b3609","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-12 09:10:25.000000000","message":"Patch Set 21:\n\n(6 comments)","accounts_in_message":[],"_revision_number":21},{"id":"41612e162d3e9fe26f5bc313e0447a9356d1b16d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-12 09:10:35.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"1fb0a768e4292fa58a723297bf582d739d026cfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 11:22:53.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/22/check/openstack-tox-cover/d60caa1/ : FAILURE in 7m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/22/check/openstack-tox-lower-constraints/12c1ccb/ : FAILURE in 7m 04s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/22/check/openstack-tox-pep8/fcf73f1/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/22/check/openstack-tox-py27/e9e7996/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/22/check/openstack-tox-py36/7e44e23/ : SUCCESS in 3m 54s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/22/check/openstack-tox-py37/4dc7e2f/ : SUCCESS in 6m 13s\n- openstack-tox-docs http://logs.openstack.org/68/668068/22/check/openstack-tox-docs/88c9f01/html/ : SUCCESS in 5m 53s\n- openstack-tox-functional http://logs.openstack.org/68/668068/22/check/openstack-tox-functional/bf99d53/ : FAILURE in 4m 55s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/22/check/openstack-tox-functional-py36/a5fd770/ : FAILURE in 5m 09s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/22/check/octavia-v2-dsvm-noop-api/f0626c0/ : SUCCESS in 43m 06s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/22/check/octavia-v2-dsvm-noop-py2-api/3f124a5/ : SUCCESS in 49m 51s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/22/check/octavia-v2-dsvm-scenario/1151fad/ : SUCCESS in 1h 29m 36s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/22/check/octavia-v2-dsvm-py2-scenario/1371a5c/ : SUCCESS in 1h 26m 01s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/22/check/octavia-v2-dsvm-py2-scenario-centos-7/57ea2a2/ : SUCCESS in 2h 06m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/22/check/octavia-v2-dsvm-scenario-ubuntu-bionic/cc2728d/ : SUCCESS in 1h 38m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/22/check/octavia-v2-dsvm-scenario-two-node/d847e13/ : SUCCESS in 1h 46m 11s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/22/check/octavia-v2-dsvm-py2-scenario-two-node/fa9b6fe/ : SUCCESS in 2h 01m 32s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/22/check/octavia-grenade/39b3f90/ : SUCCESS in 1h 25m 16s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/22/check/octavia-v2-dsvm-tls-barbican/b344890/ : SUCCESS in 49m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/22/check/octavia-v2-dsvm-scenario-amphora-v2/a7909cd/ : SUCCESS in 1h 24m 30s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"bc91202a367fa1df90788ef359e0e536f4a7485f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-12 16:02:25.000000000","message":"Patch Set 22: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"bcfae399a50e5c06330688cecc679d1fe7685494","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-12 16:12:05.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"334d934b1278b636fc20a9cdfdc6a7bd4d776235","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-12 18:57:06.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"a6d9ebabfd7d959cc7a6f5d26a430a63d5014c31","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-12 19:34:09.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"36c1b9eea0ef9a87d5d42f238c33b48c9e459d77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 21:28:18.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/24/check/openstack-tox-cover/3b29410/cover/ : SUCCESS in 7m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/24/check/openstack-tox-lower-constraints/18f6c86/ : SUCCESS in 6m 50s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/24/check/openstack-tox-pep8/163ad9e/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/24/check/openstack-tox-py27/959cfc2/ : SUCCESS in 3m 59s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/24/check/openstack-tox-py36/03915c6/ : SUCCESS in 4m 02s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/24/check/openstack-tox-py37/e7ea363/ : SUCCESS in 6m 07s\n- openstack-tox-docs http://logs.openstack.org/68/668068/24/check/openstack-tox-docs/89d53d6/html/ : SUCCESS in 7m 40s\n- openstack-tox-functional http://logs.openstack.org/68/668068/24/check/openstack-tox-functional/4d26814/ : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/24/check/openstack-tox-functional-py36/591fdf8/ : SUCCESS in 6m 41s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/24/check/octavia-v2-dsvm-noop-api/182c96e/ : SUCCESS in 57m 12s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/24/check/octavia-v2-dsvm-noop-py2-api/a7a71fc/ : SUCCESS in 42m 23s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/24/check/octavia-v2-dsvm-scenario/f93576c/ : FAILURE in 51m 09s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/24/check/octavia-v2-dsvm-py2-scenario/fca8df1/ : FAILURE in 42m 07s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/24/check/octavia-v2-dsvm-py2-scenario-centos-7/fcbcf88/ : FAILURE in 59m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/24/check/octavia-v2-dsvm-scenario-ubuntu-bionic/50adf66/ : FAILURE in 47m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/24/check/octavia-v2-dsvm-scenario-two-node/a0b527e/ : FAILURE in 1h 53m 23s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/24/check/octavia-v2-dsvm-py2-scenario-two-node/fadfbf7/ : FAILURE in 1h 19m 46s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/24/check/octavia-grenade/791751e/ : SUCCESS in 1h 11m 30s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/24/check/octavia-v2-dsvm-tls-barbican/cc8f9ff/ : FAILURE in 1h 17m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/24/check/octavia-v2-dsvm-scenario-amphora-v2/5942546/ : FAILURE in 1h 20m 20s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"d55fb0d9f0a3d4d1086755605355f3bdf870a1fe","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-12 21:34:51.000000000","message":"Patch Set 24: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":24},{"id":"a1d06f275af10b9fafb7b89148bcfa8e244a9557","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-12 21:53:23.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"fb505844eacbde83b7098359af9b577e1a35dab6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-13 01:22:05.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"c96a365a75554a607342b584946fb88b77026bae","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-13 01:22:35.000000000","message":"Patch Set 25: Workflow-1\n\nAmp agent is broken somehow.","accounts_in_message":[],"_revision_number":25},{"id":"96429fdc63c7807b982a2976a5e6fd3e893c41b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 03:33:09.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/25/check/openstack-tox-cover/ad7f08c/cover/ : SUCCESS in 6m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/25/check/openstack-tox-lower-constraints/1667f07/ : SUCCESS in 6m 17s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/25/check/openstack-tox-pep8/1fd4a65/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/25/check/openstack-tox-py27/af84c86/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/25/check/openstack-tox-py36/5005954/ : SUCCESS in 3m 47s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/25/check/openstack-tox-py37/9343bc8/ : SUCCESS in 3m 53s\n- openstack-tox-docs http://logs.openstack.org/68/668068/25/check/openstack-tox-docs/4fe926a/html/ : SUCCESS in 5m 27s\n- openstack-tox-functional http://logs.openstack.org/68/668068/25/check/openstack-tox-functional/9c83bae/ : SUCCESS in 4m 53s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/25/check/openstack-tox-functional-py36/65bbd81/ : SUCCESS in 5m 09s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/25/check/octavia-v2-dsvm-noop-api/09abddc/ : SUCCESS in 44m 45s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/25/check/octavia-v2-dsvm-noop-py2-api/6df7e91/ : SUCCESS in 47m 01s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/25/check/octavia-v2-dsvm-scenario/554ec1b/ : SUCCESS in 1h 30m 24s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/25/check/octavia-v2-dsvm-py2-scenario/f7d60b2/ : SUCCESS in 1h 43m 14s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/25/check/octavia-v2-dsvm-py2-scenario-centos-7/f3f955f/ : FAILURE in 2h 06m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/25/check/octavia-v2-dsvm-scenario-ubuntu-bionic/d3793ed/ : SUCCESS in 1h 45m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/25/check/octavia-v2-dsvm-scenario-two-node/36240a9/ : SUCCESS in 1h 49m 56s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/25/check/octavia-v2-dsvm-py2-scenario-two-node/a6e19b5/ : SUCCESS in 1h 07m 56s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/25/check/octavia-grenade/53efb7c/ : SUCCESS in 1h 14m 07s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/25/check/octavia-v2-dsvm-tls-barbican/59a08d3/ : SUCCESS in 47m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/25/check/octavia-v2-dsvm-scenario-amphora-v2/c957f2e/ : SUCCESS in 1h 31m 13s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"978cc796c04a68c3f6c9b72c7393419f3f1d2c17","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-14 07:55:10.000000000","message":"Patch Set 25: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"ab17b74044d62b10728aaa972462d4daf83d872b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-16 22:41:25.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"ccbaea54a0982637795076cc3abf9cd370db9ced","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-16 22:41:34.000000000","message":"Patch Set 26: Workflow-1","accounts_in_message":[],"_revision_number":26},{"id":"bccada7dbd22cbc40def6b486f2bb9feb6d7ea3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 00:59:35.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/26/check/openstack-tox-cover/76cffb4/cover/ : SUCCESS in 6m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/26/check/openstack-tox-lower-constraints/dbf0048/ : SUCCESS in 6m 32s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/26/check/openstack-tox-pep8/b366c49/ : FAILURE in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/26/check/openstack-tox-py27/9022d91/ : SUCCESS in 3m 53s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/26/check/openstack-tox-py36/98ba464/ : SUCCESS in 4m 15s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/26/check/openstack-tox-py37/9006ef9/ : SUCCESS in 6m 08s\n- openstack-tox-docs http://logs.openstack.org/68/668068/26/check/openstack-tox-docs/61f669c/html/ : SUCCESS in 7m 07s\n- openstack-tox-functional http://logs.openstack.org/68/668068/26/check/openstack-tox-functional/03ed49a/ : SUCCESS in 7m 03s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/26/check/openstack-tox-functional-py36/30b6a92/ : SUCCESS in 5m 10s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/26/check/octavia-v2-dsvm-noop-api/f5db843/ : SUCCESS in 46m 52s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/26/check/octavia-v2-dsvm-noop-py2-api/dc97f72/ : SUCCESS in 48m 26s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/26/check/octavia-v2-dsvm-scenario/58dccc6/ : SUCCESS in 1h 36m 49s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/26/check/octavia-v2-dsvm-py2-scenario/0bb99cc/ : SUCCESS in 1h 42m 44s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/26/check/octavia-v2-dsvm-py2-scenario-centos-7/9a75e0d/ : SUCCESS in 1h 51m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/26/check/octavia-v2-dsvm-scenario-ubuntu-bionic/35be678/ : FAILURE in 1h 38m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/26/check/octavia-v2-dsvm-scenario-two-node/753a3a6/ : SUCCESS in 1h 58m 04s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/26/check/octavia-v2-dsvm-py2-scenario-two-node/4e97831/ : SUCCESS in 1h 31m 59s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/26/check/octavia-grenade/aac88ca/ : SUCCESS in 1h 23m 45s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/26/check/octavia-v2-dsvm-tls-barbican/f41d709/ : SUCCESS in 57m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/26/check/octavia-v2-dsvm-scenario-amphora-v2/ec4fdfa/ : SUCCESS in 1h 23m 53s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"112937ab0357c578f003047782cd6ae46ddc7b00","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-17 22:40:10.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"a65fb4911ada8e3df5109e14ddd14a133671239a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-17 23:19:37.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"b970bc4d2e408c721bd5e447574e62ba13bf9bd7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-17 23:34:05.000000000","message":"Uploaded patch set 29: Commit message was updated.","accounts_in_message":[],"_revision_number":29},{"id":"d70af2e0eb2b06e4ace7eabf0ac562ad59cdaeca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 02:08:17.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/29/check/openstack-tox-cover/c591c89/cover/ : SUCCESS in 7m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/29/check/openstack-tox-lower-constraints/aea92d9/ : SUCCESS in 6m 30s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/29/check/openstack-tox-pep8/395b142/ : SUCCESS in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/29/check/openstack-tox-py27/93138bc/ : SUCCESS in 3m 52s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/29/check/openstack-tox-py36/9aaaa6c/ : SUCCESS in 3m 44s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/29/check/openstack-tox-py37/119664f/ : SUCCESS in 4m 17s\n- openstack-tox-docs http://logs.openstack.org/68/668068/29/check/openstack-tox-docs/709cb54/html/ : SUCCESS in 5m 05s\n- build-openstack-releasenotes http://logs.openstack.org/68/668068/29/check/build-openstack-releasenotes/27af637/html/ : SUCCESS in 4m 08s\n- openstack-tox-functional http://logs.openstack.org/68/668068/29/check/openstack-tox-functional/2910830/ : SUCCESS in 4m 45s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/29/check/openstack-tox-functional-py36/d91e29c/ : SUCCESS in 5m 14s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/29/check/octavia-v2-dsvm-noop-api/3a1c188/ : SUCCESS in 48m 07s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/29/check/octavia-v2-dsvm-noop-py2-api/3c31fc4/ : SUCCESS in 41m 32s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/29/check/octavia-v2-dsvm-scenario/ce94926/ : SUCCESS in 1h 32m 41s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/29/check/octavia-v2-dsvm-py2-scenario/84628ac/ : SUCCESS in 1h 34m 02s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/29/check/octavia-v2-dsvm-py2-scenario-centos-7/7c30e7a/ : FAILURE in 2h 12m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/29/check/octavia-v2-dsvm-scenario-ubuntu-bionic/6fe0187/ : SUCCESS in 1h 34m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/29/check/octavia-v2-dsvm-scenario-two-node/f348f10/ : SUCCESS in 2h 10m 57s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/29/check/octavia-v2-dsvm-py2-scenario-two-node/bbbca5f/ : SUCCESS in 1h 16m 02s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/29/check/octavia-grenade/3e2af7b/ : SUCCESS in 1h 14m 46s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/29/check/octavia-v2-dsvm-tls-barbican/81a39d1/ : SUCCESS in 55m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/29/check/octavia-v2-dsvm-scenario-amphora-v2/1e0ee50/ : SUCCESS in 1h 27m 47s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"eef5563a5d6dc0d92abd4e32a2dcb4cb89fd948a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-18 21:05:32.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"86ed2449fd8674ff637ca53e13267d24938b6c4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 23:17:07.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/30/check/openstack-tox-cover/9e6707a/cover/ : SUCCESS in 7m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/30/check/openstack-tox-lower-constraints/22472d4/ : SUCCESS in 6m 11s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/30/check/openstack-tox-pep8/e8303cc/ : SUCCESS in 5m 21s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/30/check/openstack-tox-py27/5628f95/ : SUCCESS in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/30/check/openstack-tox-py36/8317f25/ : SUCCESS in 4m 25s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/30/check/openstack-tox-py37/d31dd18/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/68/668068/30/check/openstack-tox-docs/f8c99cf/html/ : SUCCESS in 5m 25s\n- build-openstack-releasenotes http://logs.openstack.org/68/668068/30/check/build-openstack-releasenotes/d6b9fe2/html/ : SUCCESS in 4m 14s\n- openstack-tox-functional http://logs.openstack.org/68/668068/30/check/openstack-tox-functional/7d922a4/ : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/30/check/openstack-tox-functional-py36/960be11/ : SUCCESS in 5m 51s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/30/check/octavia-v2-dsvm-noop-api/10a2ddb/ : SUCCESS in 42m 21s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/30/check/octavia-v2-dsvm-noop-py2-api/63c8dcb/ : SUCCESS in 44m 27s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/30/check/octavia-v2-dsvm-scenario/a690450/ : SUCCESS in 1h 35m 04s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/30/check/octavia-v2-dsvm-py2-scenario/e2fa0a6/ : SUCCESS in 1h 32m 24s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/30/check/octavia-v2-dsvm-py2-scenario-centos-7/3d6756d/ : FAILURE in 2h 03m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/30/check/octavia-v2-dsvm-scenario-ubuntu-bionic/79f7b9a/ : SUCCESS in 1h 36m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/30/check/octavia-v2-dsvm-scenario-two-node/1e133aa/ : SUCCESS in 2h 04m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/30/check/octavia-v2-dsvm-py2-scenario-two-node/8fb2630/ : SUCCESS in 1h 48m 49s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/30/check/octavia-grenade/fe437a0/ : SUCCESS in 1h 21m 43s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/30/check/octavia-v2-dsvm-tls-barbican/84609d6/ : SUCCESS in 1h 00m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/30/check/octavia-v2-dsvm-scenario-amphora-v2/2ed349d/ : SUCCESS in 1h 33m 36s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"c8da957a600fe83caad0d95372ba12eb93127d0a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-19 15:02:19.000000000","message":"Patch Set 30: Code-Review-1\n\nSimple LB create (CentOS 7, haproxy 1.5, agent API v0.5):\n\nJul 19 14:51:46 localhost.localdomain octavia-worker[11642]: ERROR oslo_messaging.rpc.server   File \"/opt/stack/octavia/octavia/amphorae/drivers/haproxy/rest_api_driver.py\", line 665, in get_api_version\nJul 19 14:51:46 localhost.localdomain octavia-worker[11642]: ERROR oslo_messaging.rpc.server     raise driver_except.NotFound()\nJul 19 14:51:46 localhost.localdomain octavia-worker[11642]: ERROR oslo_messaging.rpc.server AttributeError: \u0027module\u0027 object has no attribute \u0027NotFound\u0027\n\n\nhttp://logs.openstack.org/68/668068/30/check/openstack-tox-cover/9e6707a/cover/octavia_amphorae_drivers_haproxy_rest_api_driver_py.html#n665","accounts_in_message":[],"_revision_number":30},{"id":"5eac6129ec134b58e2c427ead84617df47cd0e0c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-19 15:04:08.000000000","message":"Patch Set 30:\n\nShould raise exc.NotFound","accounts_in_message":[],"_revision_number":30},{"id":"a0ca8d9a7895be6efa44edcfc80187b3ff9b9eb9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-19 15:16:28.000000000","message":"Patch Set 31: Published edit on patch set 30.","accounts_in_message":[],"_revision_number":31},{"id":"d470edb50b0b15cd82cb749356632ddc937faf78","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-19 16:16:27.000000000","message":"Patch Set 31: Code-Review-1\n\nUsing http://tarballs.openstack.org/octavia/test-images/test-only-amphora-x64-haproxy-ubuntu-bionic.qcow2 (2019-07-19 06:20 420M). No tracebacks in o-cw. I have not checked what might have gone wrong in the amp yet.\n\n\nCaptured traceback:\n~~~~~~~~~~~~~~~~~~~\n    Traceback (most recent call last):\n      File \"/opt/stack/tempest/.tox/tempest/lib/python2.7/site-packages/octavia_tempest_plugin/tests/scenario/v2/test_listener.py\", line 227, in test_listener_CRUD\n        CONF.load_balancer.build_timeout)\n      File \"/opt/stack/tempest/.tox/tempest/lib/python2.7/site-packages/octavia_tempest_plugin/tests/waiters.py\", line 94, in wait_for_status\n        raise exceptions.TimeoutException(message)\n    tempest.lib.exceptions.TimeoutException: Request timed out\n    Details: (ListenerScenarioTest:test_listener_CRUD) show_listener operating_status failed to update to ONLINE within the required time 300. Current status of show_listener: OFFLINE\n   \n\n\nCaptured traceback:\n~~~~~~~~~~~~~~~~~~~\n    Traceback (most recent call last):\n      File \"/opt/stack/tempest/.tox/tempest/lib/python2.7/site-packages/octavia_tempest_plugin/tests/scenario/v2/test_traffic_ops.py\", line 295, in test_healthmonitor_traffic\n        pool_id\u003dself.pool_id)\n      File \"/opt/stack/tempest/.tox/tempest/lib/python2.7/site-packages/octavia_tempest_plugin/tests/waiters.py\", line 94, in wait_for_status\n        raise exceptions.TimeoutException(message)\n    tempest.lib.exceptions.TimeoutException: Request timed out\n    Details: (TrafficOperationsScenarioTest:test_healthmonitor_traffic) show_member operating_status failed to update to OFFLINE within the required time 300. Current status of show_member: NO_MONITOR","accounts_in_message":[],"_revision_number":31},{"id":"f1411e3c647ef70feb81b9cd6201118d07c723ac","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-19 16:23:10.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"00957b0209f98d754f2cdc9740f41c44183eda22","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-19 16:33:06.000000000","message":"Patch Set 32:\n\ntest_listener_CRUD and test_healthmonitor_traffic failed the first two runs but passed four times afterward, so ignore my previous comment.","accounts_in_message":[],"_revision_number":32},{"id":"5efa4f08e58f9a815e35b692d4eaa6a83aab8f78","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-19 17:52:47.000000000","message":"Patch Set 32: Backport-Candidate+2 Code-Review+2\n\nWorks for me in testing. Thank you, Michael and Adam!\n\nCentOS 7 controller. Amphora images:\n- CentOS 7, HAproxy 1.8, agent v1.0\n- CentOS 7, HAproxy 1.5, agent v1.0\n- CentOS 7, HAproxy 1.5, agent v0.5\n- Ubuntu Bionic, HAproxy 1.8, agent v0.5\n\nTests:\n- basic Tempest scenario\n- API Tempest (w/ amphora driver instead of noop driver tested at gate)\n- active-standby\n- barbican\n- spare pool","accounts_in_message":[],"_revision_number":32},{"id":"6adf72482440eddba4676e6e1655e21980fa00f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:48:55.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/32/check/openstack-tox-cover/7afd972/cover/ : SUCCESS in 8m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/32/check/openstack-tox-lower-constraints/7854718/ : SUCCESS in 9m 08s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/32/check/openstack-tox-pep8/662ca1b/ : SUCCESS in 6m 49s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/32/check/openstack-tox-py27/e38da11/ : SUCCESS in 6m 13s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/32/check/openstack-tox-py36/a1a0359/ : SUCCESS in 5m 05s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/32/check/openstack-tox-py37/fe7a6cd/ : SUCCESS in 6m 15s\n- openstack-tox-docs http://logs.openstack.org/68/668068/32/check/openstack-tox-docs/426f505/html/ : SUCCESS in 5m 12s\n- build-openstack-releasenotes http://logs.openstack.org/68/668068/32/check/build-openstack-releasenotes/a77ad31/html/ : SUCCESS in 4m 54s\n- openstack-tox-functional http://logs.openstack.org/68/668068/32/check/openstack-tox-functional/c985617/ : SUCCESS in 5m 50s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/32/check/openstack-tox-functional-py36/0604b1b/ : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/32/check/octavia-v2-dsvm-noop-api/e994a5c/ : SUCCESS in 49m 34s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/32/check/octavia-v2-dsvm-noop-py2-api/ce3b7fa/ : SUCCESS in 53m 44s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/32/check/octavia-v2-dsvm-scenario/d343387/ : SUCCESS in 1h 31m 51s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/32/check/octavia-v2-dsvm-py2-scenario/78a0a7f/ : SUCCESS in 1h 34m 47s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/32/check/octavia-v2-dsvm-py2-scenario-centos-7/0f86bce/ : FAILURE in 1h 57m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/32/check/octavia-v2-dsvm-scenario-ubuntu-bionic/e7f100c/ : SUCCESS in 1h 33m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/32/check/octavia-v2-dsvm-scenario-two-node/c747ecb/ : SUCCESS in 2h 00m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/32/check/octavia-v2-dsvm-py2-scenario-two-node/794c95a/ : SUCCESS in 1h 49m 15s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/32/check/octavia-grenade/adbe12d/ : SUCCESS in 1h 19m 07s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/32/check/octavia-v2-dsvm-tls-barbican/6be9efc/ : SUCCESS in 53m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/32/check/octavia-v2-dsvm-scenario-amphora-v2/478458b/ : FAILURE in 2h 21m 36s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"5874d32a3cb6864a2fb99c932f346c53737e13e0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-19 22:16:48.000000000","message":"Patch Set 32: Code-Review-1\n\nPlease hold this for Adam to review as well.","accounts_in_message":[],"_revision_number":32},{"id":"a8d920d5866d5a6112dfa5de2c6d68a1c9077116","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-23 05:06:35.000000000","message":"Patch Set 32: Code-Review-1\n\n(12 comments)\n\nInitial comments. Mostly cleanup -- but probably we should do it?\nNeed to check coverage still.","accounts_in_message":[],"_revision_number":32},{"id":"40a7d61b2c9c874abffc05fc8b2260095dd704c8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-23 05:23:43.000000000","message":"Patch Set 32:\n\n13 more uncovered lines. Probably OK?\n... Hopefully those 13 lines don\u0027t wreck us. :D","accounts_in_message":[],"_revision_number":32},{"id":"b8116391abfd426907711186b3f9f882f66027ae","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-23 20:58:53.000000000","message":"Patch Set 32:\n\n(11 comments)","accounts_in_message":[],"_revision_number":32},{"id":"38272aea3caefbff07d2a571cace227afb6feb71","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-23 21:28:51.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"a7676c5581d61ed6010ad29493877b750dc8dd61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 23:42:49.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/68/668068/33/check/openstack-tox-cover/73cda79/cover/ : SUCCESS in 8m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/33/check/openstack-tox-lower-constraints/eec7655/ : SUCCESS in 6m 51s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/33/check/openstack-tox-pep8/b7ca937/ : SUCCESS in 6m 54s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/33/check/openstack-tox-py27/1df195b/ : SUCCESS in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/33/check/openstack-tox-py36/7567256/ : SUCCESS in 4m 55s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/33/check/openstack-tox-py37/828413c/ : SUCCESS in 5m 51s\n- openstack-tox-docs http://logs.openstack.org/68/668068/33/check/openstack-tox-docs/1055a29/html/ : SUCCESS in 6m 44s\n- build-openstack-releasenotes http://logs.openstack.org/68/668068/33/check/build-openstack-releasenotes/2894fc8/html/ : SUCCESS in 4m 27s\n- openstack-tox-functional http://logs.openstack.org/68/668068/33/check/openstack-tox-functional/399da4a/ : SUCCESS in 5m 49s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/33/check/openstack-tox-functional-py36/27da9c0/ : SUCCESS in 5m 58s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/33/check/octavia-v2-dsvm-noop-api/bb12f2e/ : SUCCESS in 53m 16s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/33/check/octavia-v2-dsvm-noop-py2-api/d439f77/ : SUCCESS in 40m 33s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/33/check/octavia-v2-dsvm-scenario/db16ba0/ : SUCCESS in 1h 21m 53s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/33/check/octavia-v2-dsvm-py2-scenario/e1c4c22/ : SUCCESS in 1h 38m 34s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/668068/33/check/octavia-v2-dsvm-py2-scenario-centos-7/d280d13/ : FAILURE in 2h 01m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/668068/33/check/octavia-v2-dsvm-scenario-ubuntu-bionic/9f8329a/ : FAILURE in 1h 48m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/668068/33/check/octavia-v2-dsvm-scenario-two-node/3655437/ : SUCCESS in 1h 21m 35s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/668068/33/check/octavia-v2-dsvm-py2-scenario-two-node/809545a/ : FAILURE in 1h 26m 25s (non-voting)\n- octavia-grenade http://logs.openstack.org/68/668068/33/check/octavia-grenade/f94a91f/ : SUCCESS in 1h 36m 02s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/668068/33/check/octavia-v2-dsvm-tls-barbican/3303800/ : SUCCESS in 47m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/68/668068/33/check/octavia-v2-dsvm-scenario-amphora-v2/250a3a6/ : SUCCESS in 1h 26m 06s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"5f07c30292d6d7b36a05e57985a1d129610bb285","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-26 21:32:08.000000000","message":"Patch Set 33: Code-Review+2\n\nI think this is as good as it\u0027s going to get until we get it out into the wild and time-tested. This is a complex change... I am confident that it should work in most cases, I just hope I can be *more* confident in it by the time we release Train. :D","accounts_in_message":[],"_revision_number":33},{"id":"d5c02d0962a754b7eaaf4b184ba40bcbede98a23","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-26 21:46:49.000000000","message":"Patch Set 33: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"9e5463a42df25c372dc35284f7ea3d8373be4757","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-29 20:21:34.000000000","message":"Patch Set 33: Backport-Candidate+2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"911b394623d45027d362291c38136eb0c2eb3faf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 20:21:47.000000000","message":"Patch Set 33: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":33},{"id":"7aa886bb91b7944af32d7a2ca10c2aa2e5f81620","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 22:14:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":33},{"id":"2d351dbd817759613baa39a62d90ba406239a79a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 22:14:04.000000000","message":"Patch Set 33: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/68/668068/33/gate/openstack-tox-lower-constraints/1a07aa3/ : SUCCESS in 7m 44s\n- openstack-tox-pep8 http://logs.openstack.org/68/668068/33/gate/openstack-tox-pep8/d6ac426/ : SUCCESS in 6m 42s\n- openstack-tox-py27 http://logs.openstack.org/68/668068/33/gate/openstack-tox-py27/5522d44/ : SUCCESS in 5m 18s\n- openstack-tox-py36 http://logs.openstack.org/68/668068/33/gate/openstack-tox-py36/107155f/ : SUCCESS in 5m 42s\n- openstack-tox-py37 http://logs.openstack.org/68/668068/33/gate/openstack-tox-py37/ac6654f/ : SUCCESS in 5m 22s\n- openstack-tox-docs http://logs.openstack.org/68/668068/33/gate/openstack-tox-docs/804587d/html/ : SUCCESS in 5m 56s\n- build-openstack-releasenotes http://logs.openstack.org/68/668068/33/gate/build-openstack-releasenotes/e704039/html/ : SUCCESS in 5m 29s\n- openstack-tox-functional http://logs.openstack.org/68/668068/33/gate/openstack-tox-functional/ab310fa/ : SUCCESS in 6m 43s\n- openstack-tox-functional-py36 http://logs.openstack.org/68/668068/33/gate/openstack-tox-functional-py36/c966444/ : SUCCESS in 5m 55s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/668068/33/gate/octavia-v2-dsvm-noop-api/28484e1/ : SUCCESS in 52m 20s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/668068/33/gate/octavia-v2-dsvm-noop-py2-api/e225b57/ : SUCCESS in 51m 20s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/668068/33/gate/octavia-v2-dsvm-scenario/280c886/ : SUCCESS in 1h 45m 45s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/668068/33/gate/octavia-v2-dsvm-py2-scenario/e22e655/ : SUCCESS in 1h 43m 53s\n- octavia-grenade http://logs.openstack.org/68/668068/33/gate/octavia-grenade/ebb951d/ : SUCCESS in 1h 22m 53s","accounts_in_message":[],"_revision_number":33}],"current_revision_number":33,"current_revision":"06ce4777c3bad2868e906101bcd2cd7c753b4859","revisions":{"1102d81067119f919fbd87e7b9d02153f2b8447d":{"kind":"REWORK","_number":1,"created":"2019-06-27 23:03:07.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/668068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/1"}}},"commit":{"parents":[{"commit":"09733a0c39ad394a319f7b807518a5439ec0826e","subject":"Add bindep.txt for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09733a0c39ad394a319f7b807518a5439ec0826e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:03:06.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\nStory: 2005412\nTask: 34744\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1102d81067119f919fbd87e7b9d02153f2b8447d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1102d81067119f919fbd87e7b9d02153f2b8447d"}]},"branch":"refs/heads/master"},"7f7eacd3879976f5dc79a9cfab18be3bec0d941e":{"kind":"REWORK","_number":2,"created":"2019-06-28 06:30:45.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/68/668068/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/2"}}},"commit":{"parents":[{"commit":"09733a0c39ad394a319f7b807518a5439ec0826e","subject":"Add bindep.txt for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09733a0c39ad394a319f7b807518a5439ec0826e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-28 06:30:00.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f7eacd3879976f5dc79a9cfab18be3bec0d941e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f7eacd3879976f5dc79a9cfab18be3bec0d941e"}]},"branch":"refs/heads/master"},"70558b8050e306f75c54274f933915e2e16a0397":{"kind":"REWORK","_number":3,"created":"2019-06-29 00:57:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/668068/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/3"}}},"commit":{"parents":[{"commit":"09733a0c39ad394a319f7b807518a5439ec0826e","subject":"Add bindep.txt for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09733a0c39ad394a319f7b807518a5439ec0826e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-29 00:57:19.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70558b8050e306f75c54274f933915e2e16a0397"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70558b8050e306f75c54274f933915e2e16a0397"}]},"branch":"refs/heads/master"},"c7b651589de9fb22ed5ab8125a17d8db2da0713c":{"kind":"REWORK","_number":4,"created":"2019-06-29 01:52:50.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/68/668068/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/4"}}},"commit":{"parents":[{"commit":"09733a0c39ad394a319f7b807518a5439ec0826e","subject":"Add bindep.txt for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09733a0c39ad394a319f7b807518a5439ec0826e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-29 01:46:27.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7b651589de9fb22ed5ab8125a17d8db2da0713c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7b651589de9fb22ed5ab8125a17d8db2da0713c"}]},"branch":"refs/heads/master"},"ab5b348bfe4d100d689d58b5918279b3839642fa":{"kind":"REWORK","_number":5,"created":"2019-06-29 01:59:02.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/68/668068/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/5"}}},"commit":{"parents":[{"commit":"520f1d0c7d69074535d86138acece59ca3e8db0d","subject":"Merge \"Switch python3 versions of test jobs to match Train PTI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/520f1d0c7d69074535d86138acece59ca3e8db0d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-29 01:58:28.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ab5b348bfe4d100d689d58b5918279b3839642fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ab5b348bfe4d100d689d58b5918279b3839642fa"}]},"branch":"refs/heads/master"},"1f4306ecf6287823a58eb149b42a0bb7e83cf3b5":{"kind":"REWORK","_number":6,"created":"2019-06-30 05:49:04.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/68/668068/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/6"}}},"commit":{"parents":[{"commit":"520f1d0c7d69074535d86138acece59ca3e8db0d","subject":"Merge \"Switch python3 versions of test jobs to match Train PTI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/520f1d0c7d69074535d86138acece59ca3e8db0d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-30 05:48:11.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f4306ecf6287823a58eb149b42a0bb7e83cf3b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f4306ecf6287823a58eb149b42a0bb7e83cf3b5"}]},"branch":"refs/heads/master"},"9cbf04df05d259b25c5d651c20ca55349b5b6049":{"kind":"REWORK","_number":7,"created":"2019-06-30 11:28:15.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/68/668068/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/7"}}},"commit":{"parents":[{"commit":"520f1d0c7d69074535d86138acece59ca3e8db0d","subject":"Merge \"Switch python3 versions of test jobs to match Train PTI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/520f1d0c7d69074535d86138acece59ca3e8db0d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-30 11:27:54.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9cbf04df05d259b25c5d651c20ca55349b5b6049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9cbf04df05d259b25c5d651c20ca55349b5b6049"}]},"branch":"refs/heads/master"},"df14e3c8cc423825f5eb96010e6084beb02711cf":{"kind":"REWORK","_number":8,"created":"2019-06-30 16:47:04.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/68/668068/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/8"}}},"commit":{"parents":[{"commit":"520f1d0c7d69074535d86138acece59ca3e8db0d","subject":"Merge \"Switch python3 versions of test jobs to match Train PTI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/520f1d0c7d69074535d86138acece59ca3e8db0d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-30 16:46:17.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df14e3c8cc423825f5eb96010e6084beb02711cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df14e3c8cc423825f5eb96010e6084beb02711cf"}]},"branch":"refs/heads/master"},"f59bc3e4ddd329f8544611d6becc344066747e74":{"kind":"REWORK","_number":9,"created":"2019-06-30 16:54:52.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/68/668068/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/9"}}},"commit":{"parents":[{"commit":"520f1d0c7d69074535d86138acece59ca3e8db0d","subject":"Merge \"Switch python3 versions of test jobs to match Train PTI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/520f1d0c7d69074535d86138acece59ca3e8db0d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-30 16:54:28.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f59bc3e4ddd329f8544611d6becc344066747e74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f59bc3e4ddd329f8544611d6becc344066747e74"}]},"branch":"refs/heads/master"},"a80ef84c95064a49597bf92c32d6cd6679379e72":{"kind":"REWORK","_number":10,"created":"2019-07-02 16:45:08.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/68/668068/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/10"}}},"commit":{"parents":[{"commit":"520f1d0c7d69074535d86138acece59ca3e8db0d","subject":"Merge \"Switch python3 versions of test jobs to match Train PTI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/520f1d0c7d69074535d86138acece59ca3e8db0d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-07-02 16:44:43.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a80ef84c95064a49597bf92c32d6cd6679379e72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a80ef84c95064a49597bf92c32d6cd6679379e72"}]},"branch":"refs/heads/master"},"cbaffd26ca2c6a5adb3ef73b8c3b3d89cedf58ff":{"kind":"REWORK","_number":11,"created":"2019-07-03 00:51:00.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/668068/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/11"}}},"commit":{"parents":[{"commit":"520f1d0c7d69074535d86138acece59ca3e8db0d","subject":"Merge \"Switch python3 versions of test jobs to match Train PTI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/520f1d0c7d69074535d86138acece59ca3e8db0d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-03 00:50:51.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbaffd26ca2c6a5adb3ef73b8c3b3d89cedf58ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbaffd26ca2c6a5adb3ef73b8c3b3d89cedf58ff"}]},"branch":"refs/heads/master"},"05a7f9c7b7d006377c403e5f25ba5167187ac666":{"kind":"REWORK","_number":12,"created":"2019-07-03 15:43:34.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/68/668068/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/12"}}},"commit":{"parents":[{"commit":"520f1d0c7d69074535d86138acece59ca3e8db0d","subject":"Merge \"Switch python3 versions of test jobs to match Train PTI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/520f1d0c7d69074535d86138acece59ca3e8db0d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-07-03 15:43:10.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05a7f9c7b7d006377c403e5f25ba5167187ac666"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05a7f9c7b7d006377c403e5f25ba5167187ac666"}]},"branch":"refs/heads/master"},"eb41bbe340d1347b2ec13489c98cefc8e417d70b":{"kind":"REWORK","_number":13,"created":"2019-07-04 00:30:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/668068/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/13"}}},"commit":{"parents":[{"commit":"520f1d0c7d69074535d86138acece59ca3e8db0d","subject":"Merge \"Switch python3 versions of test jobs to match Train PTI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/520f1d0c7d69074535d86138acece59ca3e8db0d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-04 00:30:18.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eb41bbe340d1347b2ec13489c98cefc8e417d70b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eb41bbe340d1347b2ec13489c98cefc8e417d70b"}]},"branch":"refs/heads/master"},"dd234b19a4aa90a63787d6ada5e188b0b499c0f0":{"kind":"REWORK","_number":14,"created":"2019-07-08 23:24:00.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/668068/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/14"}}},"commit":{"parents":[{"commit":"520f1d0c7d69074535d86138acece59ca3e8db0d","subject":"Merge \"Switch python3 versions of test jobs to match Train PTI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/520f1d0c7d69074535d86138acece59ca3e8db0d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-08 23:23:59.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd234b19a4aa90a63787d6ada5e188b0b499c0f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd234b19a4aa90a63787d6ada5e188b0b499c0f0"}]},"branch":"refs/heads/master"},"3e237e6acb2971886a2d17513fe84ddae9902831":{"kind":"REWORK","_number":15,"created":"2019-07-09 15:19:04.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/68/668068/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/15"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-07-09 11:07:52.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3e237e6acb2971886a2d17513fe84ddae9902831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3e237e6acb2971886a2d17513fe84ddae9902831"}]},"branch":"refs/heads/master"},"9cbbe29703121946909914b7a896fb342d5fc78b":{"kind":"REWORK","_number":16,"created":"2019-07-09 18:03:29.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/68/668068/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/16"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-07-09 17:55:52.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9cbbe29703121946909914b7a896fb342d5fc78b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9cbbe29703121946909914b7a896fb342d5fc78b"}]},"branch":"refs/heads/master"},"4b8adb5a3dbd5dd980f0eac7b5c047b6a8e7202a":{"kind":"REWORK","_number":17,"created":"2019-07-09 18:17:51.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/68/668068/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/17"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-07-09 18:17:28.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b8adb5a3dbd5dd980f0eac7b5c047b6a8e7202a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b8adb5a3dbd5dd980f0eac7b5c047b6a8e7202a"}]},"branch":"refs/heads/master"},"5da549fa4519d9059f412dd14119509cd3a666f9":{"kind":"REWORK","_number":18,"created":"2019-07-10 01:38:30.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/668068/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/18"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-10 01:38:23.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5da549fa4519d9059f412dd14119509cd3a666f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5da549fa4519d9059f412dd14119509cd3a666f9"}]},"branch":"refs/heads/master"},"2cdecf5fa91215c99f939d95db90d963912a420b":{"kind":"REWORK","_number":19,"created":"2019-07-10 13:41:29.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/68/668068/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/19"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-07-10 13:41:03.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2cdecf5fa91215c99f939d95db90d963912a420b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2cdecf5fa91215c99f939d95db90d963912a420b"}]},"branch":"refs/heads/master"},"f47ca9e9e22f0f9f2b130dcd10145ab99499d0b9":{"kind":"REWORK","_number":20,"created":"2019-07-11 00:33:26.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/668068/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/20"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-11 00:33:21.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f47ca9e9e22f0f9f2b130dcd10145ab99499d0b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f47ca9e9e22f0f9f2b130dcd10145ab99499d0b9"}]},"branch":"refs/heads/master"},"0a375af40691b5034272cb87c5562974f9893a15":{"kind":"REWORK","_number":21,"created":"2019-07-11 23:48:07.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/668068/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/21"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-11 23:47:59.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a375af40691b5034272cb87c5562974f9893a15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a375af40691b5034272cb87c5562974f9893a15"}]},"branch":"refs/heads/master"},"382aa98c0a07c95b0fa40c40cc8af32f91488a98":{"kind":"REWORK","_number":22,"created":"2019-07-12 09:10:35.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/68/668068/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/22"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-07-12 09:10:09.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/382aa98c0a07c95b0fa40c40cc8af32f91488a98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/382aa98c0a07c95b0fa40c40cc8af32f91488a98"}]},"branch":"refs/heads/master"},"89ecb7cce7d79f3078c7c736b1773780380d11c8":{"kind":"REWORK","_number":23,"created":"2019-07-12 18:57:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/668068/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/23"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-12 18:56:58.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89ecb7cce7d79f3078c7c736b1773780380d11c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89ecb7cce7d79f3078c7c736b1773780380d11c8"}]},"branch":"refs/heads/master"},"74d0421421353c148ecba30ca2b64aae54567685":{"kind":"REWORK","_number":24,"created":"2019-07-12 19:34:09.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/68/668068/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/24"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-07-12 19:33:45.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/74d0421421353c148ecba30ca2b64aae54567685"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/74d0421421353c148ecba30ca2b64aae54567685"}]},"branch":"refs/heads/master"},"e59de49c19c5fc75faf072818e78e13009bdc2d2":{"kind":"REWORK","_number":25,"created":"2019-07-13 01:22:05.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/668068/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/25"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-13 01:21:57.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e59de49c19c5fc75faf072818e78e13009bdc2d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e59de49c19c5fc75faf072818e78e13009bdc2d2"}]},"branch":"refs/heads/master"},"86a9cc0d701954d210c9d539ee30ca404e28a53d":{"kind":"REWORK","_number":26,"created":"2019-07-16 22:41:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/668068/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/26"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-16 22:41:23.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/86a9cc0d701954d210c9d539ee30ca404e28a53d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/86a9cc0d701954d210c9d539ee30ca404e28a53d"}]},"branch":"refs/heads/master"},"93daafdc627947734cce3951f378694135e7cdc0":{"kind":"REWORK","_number":27,"created":"2019-07-17 22:40:10.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/668068/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/27"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-17 22:40:00.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93daafdc627947734cce3951f378694135e7cdc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93daafdc627947734cce3951f378694135e7cdc0"}]},"branch":"refs/heads/master"},"a19c4cdec22f79cb9a2020be36129f48516e24bc":{"kind":"REWORK","_number":28,"created":"2019-07-17 23:19:37.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/668068/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/28"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-17 23:19:31.000000000","tz":-420},"subject":"Make amphora use a single HAProxy instance","message":"Make amphora use a single HAProxy instance\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a19c4cdec22f79cb9a2020be36129f48516e24bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a19c4cdec22f79cb9a2020be36129f48516e24bc"}]},"branch":"refs/heads/master"},"51b9d919f585458837bbd9c0098e8f079d74a6a1":{"kind":"NO_CODE_CHANGE","_number":29,"created":"2019-07-17 23:34:05.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/668068/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/29"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-17 23:30:49.000000000","tz":-420},"subject":"Fix multi-listener load balancers","message":"Fix multi-listener load balancers\n\nLoad balancers with multiple listeners, running on an amphora image\nwith HAProxy 1.8 or newer can experience excessive memory usage that\nmay lead to an ERROR provisioning_status.\nThis patch resolves this issue by consolidating the listeners into\na single haproxy process inside the amphora.\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51b9d919f585458837bbd9c0098e8f079d74a6a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51b9d919f585458837bbd9c0098e8f079d74a6a1"}]},"branch":"refs/heads/master"},"960c422e21fc18ef90ddc56abe43e3403b8e0c55":{"kind":"REWORK","_number":30,"created":"2019-07-18 21:05:32.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/668068/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/30"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-18 21:05:26.000000000","tz":-420},"subject":"Fix multi-listener load balancers","message":"Fix multi-listener load balancers\n\nLoad balancers with multiple listeners, running on an amphora image\nwith HAProxy 1.8 or newer can experience excessive memory usage that\nmay lead to an ERROR provisioning_status.\nThis patch resolves this issue by consolidating the listeners into\na single haproxy process inside the amphora.\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/960c422e21fc18ef90ddc56abe43e3403b8e0c55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/960c422e21fc18ef90ddc56abe43e3403b8e0c55"}]},"branch":"refs/heads/master"},"9c770ad91f7abeb2d6bf8ecb4247d27cfaa4bc30":{"kind":"REWORK","_number":31,"created":"2019-07-19 15:16:28.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/668068/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/31"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-19 15:16:26.000000000","tz":0},"subject":"Fix multi-listener load balancers","message":"Fix multi-listener load balancers\n\nLoad balancers with multiple listeners, running on an amphora image\nwith HAProxy 1.8 or newer can experience excessive memory usage that\nmay lead to an ERROR provisioning_status.\nThis patch resolves this issue by consolidating the listeners into\na single haproxy process inside the amphora.\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c770ad91f7abeb2d6bf8ecb4247d27cfaa4bc30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c770ad91f7abeb2d6bf8ecb4247d27cfaa4bc30"}]},"branch":"refs/heads/master"},"cf9b14f9b3b5bd35d1b70d5e2a9c74d1c0f7af0e":{"kind":"REWORK","_number":32,"created":"2019-07-19 16:23:10.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/668068/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/32"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-19 16:23:04.000000000","tz":-420},"subject":"Fix multi-listener load balancers","message":"Fix multi-listener load balancers\n\nLoad balancers with multiple listeners, running on an amphora image\nwith HAProxy 1.8 or newer can experience excessive memory usage that\nmay lead to an ERROR provisioning_status.\nThis patch resolves this issue by consolidating the listeners into\na single haproxy process inside the amphora.\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf9b14f9b3b5bd35d1b70d5e2a9c74d1c0f7af0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf9b14f9b3b5bd35d1b70d5e2a9c74d1c0f7af0e"}]},"branch":"refs/heads/master"},"06ce4777c3bad2868e906101bcd2cd7c753b4859":{"kind":"REWORK","_number":33,"created":"2019-07-23 21:28:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/668068/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/668068/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/668068/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/668068/33"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-23 21:28:49.000000000","tz":-420},"subject":"Fix multi-listener load balancers","message":"Fix multi-listener load balancers\n\nLoad balancers with multiple listeners, running on an amphora image\nwith HAProxy 1.8 or newer can experience excessive memory usage that\nmay lead to an ERROR provisioning_status.\nThis patch resolves this issue by consolidating the listeners into\na single haproxy process inside the amphora.\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/06ce4777c3bad2868e906101bcd2cd7c753b4859"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/06ce4777c3bad2868e906101bcd2cd7c753b4859"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
