)]}'
{"id":"openstack%2Foctavia~668069","triplet_id":"openstack%2Foctavia~master~I61fe9502665e325ee39b6ef51e703f36f9a360f2","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I61fe9502665e325ee39b6ef51e703f36f9a360f2","subject":"Use ubuntu-minimal for periodic image build jobs","status":"ABANDONED","created":"2019-06-28 00:14:24.000000000","updated":"2019-07-01 23:44:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"00043899a21bc29b7338f027edc83421480d55dc","_number":668069,"virtual_id_number":668069,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-06-28 02:29:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-1,"date":"2019-06-30 18:12:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-28 02:29:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-30 18:12:58.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"647379cfeb8b0386257f150d7215dd916fe9d0f1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-28 00:14:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7dda5985772dc7e6d03a70bf2b15840b6a6aabc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 02:29:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/69/668069/1/check/openstack-tox-cover/4586c32/cover/ : SUCCESS in 7m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/668069/1/check/openstack-tox-lower-constraints/70ca226/ : SUCCESS in 7m 29s\n- openstack-tox-pep8 http://logs.openstack.org/69/668069/1/check/openstack-tox-pep8/c06fceb/ : SUCCESS in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/69/668069/1/check/openstack-tox-py27/0c1766c/ : SUCCESS in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/69/668069/1/check/openstack-tox-py36/d0205b3/ : SUCCESS in 5m 28s\n- openstack-tox-py37 http://logs.openstack.org/69/668069/1/check/openstack-tox-py37/24edea3/ : SUCCESS in 6m 58s\n- openstack-tox-docs http://logs.openstack.org/69/668069/1/check/openstack-tox-docs/1dcd7f6/html/ : SUCCESS in 7m 22s\n- openstack-tox-functional http://logs.openstack.org/69/668069/1/check/openstack-tox-functional/721b2d9/ : SUCCESS in 7m 04s\n- openstack-tox-functional-py36 http://logs.openstack.org/69/668069/1/check/openstack-tox-functional-py36/9ea6fef/ : SUCCESS in 6m 57s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/69/668069/1/check/octavia-v2-dsvm-noop-api/338e41e/ : SUCCESS in 54m 56s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/69/668069/1/check/octavia-v2-dsvm-noop-py2-api/8628457/ : SUCCESS in 53m 52s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/69/668069/1/check/octavia-v2-dsvm-scenario/30dac78/ : SUCCESS in 1h 49m 06s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/69/668069/1/check/octavia-v2-dsvm-py2-scenario/df7b045/ : SUCCESS in 1h 53m 09s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/69/668069/1/check/octavia-v2-dsvm-py2-scenario-centos-7/0d4ee62/ : FAILURE in 2h 06m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/69/668069/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/6215268/ : SUCCESS in 2h 01m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/69/668069/1/check/octavia-v2-dsvm-scenario-two-node/1ffe9fd/ : SUCCESS in 1h 41m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/69/668069/1/check/octavia-v2-dsvm-py2-scenario-two-node/5a7488b/ : SUCCESS in 1h 44m 33s (non-voting)\n- octavia-grenade http://logs.openstack.org/69/668069/1/check/octavia-grenade/7a18caa/ : SUCCESS in 1h 18m 05s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/69/668069/1/check/octavia-v2-dsvm-tls-barbican/5ffc850/ : SUCCESS in 59m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/69/668069/1/check/octavia-v2-dsvm-scenario-amphora-v2/bf94311/ : SUCCESS in 1h 30m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"164a09712d346db085c68b3d3d46feb780f79305","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-30 18:12:58.000000000","message":"Patch Set 1: Code-Review-1\n\nWill this patch actually have impact on the image? The diskimage-create.sh switches to ubuntu-minimal when AMP_BASEOS\u003dubuntu.\n\nhttps://opendev.org/openstack/octavia/src/commit/de30dbb62f98e89219231fbc282083f45417774c/diskimage-create/diskimage-create.sh#L137","accounts_in_message":[],"_revision_number":1},{"id":"00043899a21bc29b7338f027edc83421480d55dc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-01 23:44:55.000000000","message":"Abandoned\n\nHmm, I see that now. So I wonder why these jobs aren\u0027t using the new kvm kernel.... It must be setting the variable but DIB isn\u0027t yet picking it up.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf7da0c7e66c8f01ac5edae6de92dd0f3defd7ef","revisions":{"cf7da0c7e66c8f01ac5edae6de92dd0f3defd7ef":{"kind":"REWORK","_number":1,"created":"2019-06-28 00:14:24.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/69/668069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/668069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/668069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/668069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/668069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/668069/1"}}},"commit":{"parents":[{"commit":"520f1d0c7d69074535d86138acece59ca3e8db0d","subject":"Merge \"Switch python3 versions of test jobs to match Train PTI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/520f1d0c7d69074535d86138acece59ca3e8db0d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-28 00:12:01.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-28 00:14:23.000000000","tz":-420},"subject":"Use ubuntu-minimal for periodic image build jobs","message":"Use ubuntu-minimal for periodic image build jobs\n\nPreviously the periodic jobs that build Ubuntu images for\ntarballs.openstack.org was using the \"ubuntu\" base element even though\nwe had changed the default to be ubuntu-minimal.\nThis patch updates the Ubuntu image build jobs to use the\nubuntu-minimal base element.\n\nChange-Id: I61fe9502665e325ee39b6ef51e703f36f9a360f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf7da0c7e66c8f01ac5edae6de92dd0f3defd7ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf7da0c7e66c8f01ac5edae6de92dd0f3defd7ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
