)]}'
{"id":"openstack%2Foctavia~668173","triplet_id":"openstack%2Foctavia~master~Id94a293a88d27546c0fe7b45367922fa4cca260a","project":"openstack/octavia","branch":"master","topic":"jobboard","hashtags":[],"change_id":"Id94a293a88d27546c0fe7b45367922fa4cca260a","subject":"Transition l7rule flows to dicts","status":"MERGED","created":"2019-06-28 13:14:18.000000000","updated":"2020-01-26 18:21:57.000000000","submitted":"2020-01-26 18:19:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668173-1580062767163-845097a9","meta_rev_id":"cad2e17357df12ad377cf3491abd28fff899dd22","_number":668173,"virtual_id_number":668173,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-26 18:19:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-01-24 13:02:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-01-24 13:02:06.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":1,"date":"2020-01-24 13:02:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-01-26 16:32:34.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":-2,"date":"2020-01-24 13:02:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-14 13:35:53.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2020-01-06 19:03:39.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-01-06 20:24:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-01-26 16:32:34.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-01-26 18:19:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f88e43c00dbb08ac01ddc0dfc195a595c088424","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-06-28 13:14:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65ad17706f7a50938941fe83c2a32d85d1c8c6ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 15:53:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/73/668173/1/check/openstack-tox-cover/2460ebe/ : FAILURE in 5m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/668173/1/check/openstack-tox-lower-constraints/70e07f3/ : FAILURE in 4m 22s\n- openstack-tox-pep8 http://logs.openstack.org/73/668173/1/check/openstack-tox-pep8/9cc5c25/ : FAILURE in 4m 31s\n- openstack-tox-py27 http://logs.openstack.org/73/668173/1/check/openstack-tox-py27/38d4509/ : FAILURE in 4m 03s\n- openstack-tox-py36 http://logs.openstack.org/73/668173/1/check/openstack-tox-py36/ebd6363/ : FAILURE in 4m 23s\n- openstack-tox-py37 http://logs.openstack.org/73/668173/1/check/openstack-tox-py37/3935cb4/ : FAILURE in 5m 20s\n- openstack-tox-docs http://logs.openstack.org/73/668173/1/check/openstack-tox-docs/0aed00a/html/ : SUCCESS in 5m 21s\n- openstack-tox-functional http://logs.openstack.org/73/668173/1/check/openstack-tox-functional/807160c/ : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 http://logs.openstack.org/73/668173/1/check/openstack-tox-functional-py36/9568e7e/ : SUCCESS in 5m 46s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/73/668173/1/check/octavia-v2-dsvm-noop-api/0ef3ab6/ : SUCCESS in 44m 55s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/73/668173/1/check/octavia-v2-dsvm-noop-py2-api/d21cfa6/ : SUCCESS in 45m 18s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/73/668173/1/check/octavia-v2-dsvm-scenario/7a2314e/ : SUCCESS in 1h 28m 57s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/73/668173/1/check/octavia-v2-dsvm-py2-scenario/43c5d4a/ : SUCCESS in 1h 31m 12s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/73/668173/1/check/octavia-v2-dsvm-py2-scenario-centos-7/8cfcec7/ : FAILURE in 1h 52m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/73/668173/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/ef1ad0b/ : SUCCESS in 1h 34m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/73/668173/1/check/octavia-v2-dsvm-scenario-two-node/4ab930f/ : SUCCESS in 1h 54m 28s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/73/668173/1/check/octavia-v2-dsvm-py2-scenario-two-node/316341d/ : SUCCESS in 1h 44m 15s (non-voting)\n- octavia-grenade http://logs.openstack.org/73/668173/1/check/octavia-grenade/b16af81/ : SUCCESS in 1h 17m 29s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/73/668173/1/check/octavia-v2-dsvm-tls-barbican/aee345b/ : SUCCESS in 54m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/73/668173/1/check/octavia-v2-dsvm-scenario-amphora-v2/de0bdf1/ : TIMED_OUT in 2h 33m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d8e5d2611c9c146344270c980b3b1c47f225b38","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-07-02 11:33:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e087e52b0dfb674cacf99fc63cc4598f0ccffed8","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-07-02 11:34:41.000000000","message":"Topic changed from 665381 to jobboard_dicts","accounts_in_message":[],"_revision_number":2},{"id":"5000ce0b7665bc80a5573eb46eeae5d6e97de1fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 14:46:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/73/668173/2/check/openstack-tox-cover/7551553/cover/ : SUCCESS in 6m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/668173/2/check/openstack-tox-lower-constraints/5427a10/ : SUCCESS in 6m 51s\n- openstack-tox-pep8 http://logs.openstack.org/73/668173/2/check/openstack-tox-pep8/4a1d88a/ : SUCCESS in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/73/668173/2/check/openstack-tox-py27/e45fda3/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/73/668173/2/check/openstack-tox-py36/f6b255a/ : SUCCESS in 4m 19s\n- openstack-tox-py37 http://logs.openstack.org/73/668173/2/check/openstack-tox-py37/e5215d7/ : SUCCESS in 4m 32s\n- openstack-tox-docs http://logs.openstack.org/73/668173/2/check/openstack-tox-docs/5266018/html/ : SUCCESS in 5m 20s\n- openstack-tox-functional http://logs.openstack.org/73/668173/2/check/openstack-tox-functional/e05a1d6/ : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 http://logs.openstack.org/73/668173/2/check/openstack-tox-functional-py36/1ffee63/ : SUCCESS in 5m 07s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/73/668173/2/check/octavia-v2-dsvm-noop-api/784ceaa/ : SUCCESS in 44m 13s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/73/668173/2/check/octavia-v2-dsvm-noop-py2-api/ecb65ef/ : SUCCESS in 45m 11s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/73/668173/2/check/octavia-v2-dsvm-scenario/56c3589/ : SUCCESS in 1h 31m 04s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/73/668173/2/check/octavia-v2-dsvm-py2-scenario/62d4b17/ : SUCCESS in 1h 27m 03s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/73/668173/2/check/octavia-v2-dsvm-py2-scenario-centos-7/d14ff22/ : FAILURE in 1h 54m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/73/668173/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/cbd0cf8/ : SUCCESS in 1h 28m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/73/668173/2/check/octavia-v2-dsvm-scenario-two-node/d60672e/ : SUCCESS in 1h 47m 31s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/73/668173/2/check/octavia-v2-dsvm-py2-scenario-two-node/275c39f/ : SUCCESS in 1h 34m 55s (non-voting)\n- octavia-grenade http://logs.openstack.org/73/668173/2/check/octavia-grenade/27bf372/ : SUCCESS in 1h 14m 43s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/73/668173/2/check/octavia-v2-dsvm-tls-barbican/fe921dc/ : SUCCESS in 52m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/73/668173/2/check/octavia-v2-dsvm-scenario-amphora-v2/1244e47/ : TIMED_OUT in 2h 33m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"981e3bc429f1b9767a8b577f3b6bdf507639068b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-07-03 09:25:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"86efceff46459036e44ed04537ac584bc10ec304","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 12:17:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/73/668173/3/check/openstack-tox-cover/4415393/cover/ : SUCCESS in 9m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/668173/3/check/openstack-tox-lower-constraints/3a34ba0/ : SUCCESS in 6m 19s\n- openstack-tox-pep8 http://logs.openstack.org/73/668173/3/check/openstack-tox-pep8/5fc640e/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/73/668173/3/check/openstack-tox-py27/8dbd74f/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/73/668173/3/check/openstack-tox-py36/a902244/ : SUCCESS in 5m 24s\n- openstack-tox-py37 http://logs.openstack.org/73/668173/3/check/openstack-tox-py37/324b8b7/ : SUCCESS in 4m 36s\n- openstack-tox-docs http://logs.openstack.org/73/668173/3/check/openstack-tox-docs/5be6b81/html/ : SUCCESS in 7m 32s\n- openstack-tox-functional http://logs.openstack.org/73/668173/3/check/openstack-tox-functional/04beea8/ : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 http://logs.openstack.org/73/668173/3/check/openstack-tox-functional-py36/c970ae5/ : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/73/668173/3/check/octavia-v2-dsvm-noop-api/9eeb93e/ : SUCCESS in 44m 41s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/73/668173/3/check/octavia-v2-dsvm-noop-py2-api/a87a509/ : SUCCESS in 1h 21m 26s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/73/668173/3/check/octavia-v2-dsvm-scenario/b673bbc/ : SUCCESS in 1h 31m 43s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/73/668173/3/check/octavia-v2-dsvm-py2-scenario/040cae4/ : SUCCESS in 1h 20m 52s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/73/668173/3/check/octavia-v2-dsvm-py2-scenario-centos-7/d2942d2/ : FAILURE in 2h 12m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/73/668173/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/93a7204/ : SUCCESS in 1h 23m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/73/668173/3/check/octavia-v2-dsvm-scenario-two-node/6869d56/ : RETRY_LIMIT in 8m 12s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/73/668173/3/check/octavia-v2-dsvm-py2-scenario-two-node/e7edc99/ : FAILURE in 1h 35m 01s (non-voting)\n- octavia-grenade http://logs.openstack.org/73/668173/3/check/octavia-grenade/40895c0/ : SUCCESS in 1h 16m 26s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/73/668173/3/check/octavia-v2-dsvm-tls-barbican/80ce5e0/ : SUCCESS in 1h 28m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/73/668173/3/check/octavia-v2-dsvm-scenario-amphora-v2/201ff0f/ : TIMED_OUT in 2h 33m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6fd90d82aa7731e7279455127d21e450b5b99cad","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-07-14 13:35:53.000000000","message":"Patch Set 3: Code-Review+1\n\nwait for the others review","accounts_in_message":[],"_revision_number":3},{"id":"141a4df67d395232b8a82993cf60fd570313b8f4","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-09-02 13:00:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"51d2414a42e7e65c380d688ee2097a370a1fda05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 15:35:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd3e71d77ff742c4bccf102184cc6833 : SUCCESS in 9m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/785f9a6502f54486bc41d776b4103071 : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f2918bbb4c547f795a56404b2b1ef12 : SUCCESS in 9m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7741df69321b45799bf1231d8766aa7d : SUCCESS in 8m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c370b9fe8dfa43df91dcddb740d4bba6 : SUCCESS in 7m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20bdfe5952ef4086a8935f41e43ca587 : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7aae78d80894b278d1cf8fb00f20bc1 : SUCCESS in 10m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0bc80f2e17cf4e2e829190d5c3145086 : SUCCESS in 6m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/38fcd12e4c96454598b7c768b653eb17 : SUCCESS in 10m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4364d7567e6f4d5a9f886f8e1f57104e : SUCCESS in 7m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/de1e2ec35c8d418e8f2af5ed23589a11 : SUCCESS in 8m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/64a3203b223944b89eabaa509638289e : SUCCESS in 53m 52s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/52b919cd1d3541b48be4124ac42d4e01 : SUCCESS in 57m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f572f721177c49eb872e2fe8cf1dfac4 : SUCCESS in 1h 33m 54s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/79aabad840dc448b9f68ca5c8a269aa9 : SUCCESS in 1h 38m 39s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ff22c2ab1ecc4b199be12a4c5c34de12 : SUCCESS in 2h 24m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b0943cb2f1f449328e7a7ae510760328 : FAILURE in 2h 01m 55s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/39d0395adadb414eb239f7dd9e2cf90c : SUCCESS in 1h 18m 26s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/da9c241426c2493ea56ad42cd2a9f94b : SUCCESS in 1h 32m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/49c4dc6b5e0b433dafcd8e1bf54e9c0f : SUCCESS in 58m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/88cb51b9d9b4450f98fd457736793061 : TIMED_OUT in 2h 34m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/20fc81393b6b4f26a5f15b1ef281cf83 : SUCCESS in 1h 12m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ce4a80bcc6d94a86b1d54ac2b0191bfb : SUCCESS in 1h 20m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5cfb73d295f946c88882bc285d41f95c : FAILURE in 1h 19m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c1fb7c3056e4dd208bde0231a03e4de02eb00679","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-09-12 10:47:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f2d525e1dc80e6804c4201041113542531313b20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 14:29:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dfce54f3b7d4103b8a98bd64f0865d2 : SUCCESS in 7m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7709539cc34b47c1a46fef2cf910b0cc : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79e9a2a3fcc0455ebbf7c7118b66b135 : FAILURE in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b917670bbede49ab9a136934b213aad9 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0b17bd6f8414589bfd7952e11c81b4e : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a989a041d08140698c53a796d170bc6b : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b7d6f71d2d54554a32d1f7d1e951dea : SUCCESS in 6m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8beec707e7534928aec096525bb2e839 : SUCCESS in 5m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8681bd5b332049eb96dcc27f6ed781bc : SUCCESS in 7m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8ffb330f8eac40ed85f728ac5381172d : SUCCESS in 5m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/177559559fc943fca99079a9301b68eb : SUCCESS in 5m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/202dde9c860a433a96ca4b623a41dbd5 : SUCCESS in 44m 06s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/63d5d5d5b001457c8dc2081357acdeaf : SUCCESS in 56m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b408202625a94988b80f349f61c3ca18 : SUCCESS in 1h 44m 53s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/da6977efe68b4062960018c2e67c39e5 : SUCCESS in 1h 51m 18s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7fbe1fb20b854445b9cc20d06af0c673 : SUCCESS in 2h 20m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7088a5b1bcc14b61af38086d5186ed0c : SUCCESS in 2h 14m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ab5dd03f01084035a1bac3feb4f68d87 : FAILURE in 1h 26m 29s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f45fd7f9b11a4561bf0fd841c52db8c7 : SUCCESS in 1h 19m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4b3f4bd2aed143f0a2091853a6969d0a : SUCCESS in 57m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b149395608ed4947b7b6d7c11b60f35f : TIMED_OUT in 2h 33m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a9200e8e5e14c2aaa96430dba392b22 : SUCCESS in 58m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d2053b0de25144128b6dccc087409d93 : SUCCESS in 1h 11m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/89bf1b1e381d4846931bc75096a8f44e : SUCCESS in 1h 32m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"aa77c65c68881187408614190aa1494969364ea3","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-09-20 11:14:42.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"19aa04d614f7e9cc9086b69a197312826104d52b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 14:04:53.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99aa007652d1479a83c8ac94dec1a748 : SUCCESS in 8m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9a6bcdcc3de4276b9cbfed1a989ba4d : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/032efe463c8e4e87846fd158807dc720 : SUCCESS in 7m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d87ffd1b355f4d76a70582ede786c505 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d1d0582f03644708be8bc9c2a6221c4 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b42380dc4d29424090732421d65f2474 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99eff8d3fe374d7e99bc561eb097df99 : SUCCESS in 12m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/eb3c46eb6561483db4bea47c36dd65c0 : SUCCESS in 5m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d32b6cc3f7984755aef9242cce2618d2 : SUCCESS in 7m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/46f8b710bfa7433b9fb6a8b0c34dcbe9 : SUCCESS in 5m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f012c949488142688ce9c3380a628b07 : SUCCESS in 5m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/06a42425d1a447a5a1c0a29419eedbd9 : SUCCESS in 47m 43s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d4796504840345ab9298bd423cbd9864 : SUCCESS in 51m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/77ce51aee91c4d3f945708e8f1f4c947 : SUCCESS in 1h 54m 33s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4cc6182f0fb74376834683dc9bd1abed : SUCCESS in 2h 05m 59s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9ee37e38f05249588d28f1bc56535aa4 : FAILURE in 24m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/29adae0a076e4e98824a312e979e5f93 : SUCCESS in 2h 03m 59s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/fb954e10687e440eae2ad21219fad977 : SUCCESS in 2h 18m 48s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6d294bc7b6d14a219d886a852280406c : FAILURE in 1h 00m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/98204123dcf04adc97785df2c06d4add : SUCCESS in 51m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/59ba79cc6a8544d3810f99b8311fa5c8 : FAILURE in 2h 21m 40s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d805ff9984eb406897c8d76b6071a9c7 : SUCCESS in 1h 12m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/88b4efa39e6d405985b88f3d7b89e21c : FAILURE in 1h 40m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/cbf3d8baf2d24b5ebd67e2e5715a84f7 : FAILURE in 1h 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"132f4e2f4d49851deff18399da5d1f3be9439665","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-09-24 07:01:55.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"6816e043b965482cd1860bab269115d7e0b71969","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 09:41:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34e517f157e64e8eb6407997b6bc9d0b : SUCCESS in 7m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd9c2d20558148f5a02a26cebccba304 : SUCCESS in 11m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b88d154da3b147cfb0ad0c0078907b7d : SUCCESS in 6m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/643596b31fab4852b444d08fd5167006 : SUCCESS in 9m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abba952dfbd4491e857c283b1316daef : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/927c1eb65157401882e793c45d5176b6 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/888e9d9329ba4abfbc51721748faee07 : SUCCESS in 12m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/88c45e5506774ada8a5b058119eca523 : SUCCESS in 5m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/309ba0281cb64795ad4b44bf35353e81 : SUCCESS in 7m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/75b24546903141a5b23e94cadb62f3eb : SUCCESS in 5m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b6887fccf01d42bda2dbf8eca73052ed : SUCCESS in 5m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c9fa342073de454bbfe2313283cf6f53 : SUCCESS in 53m 29s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5ec5171c5b0b474bb2e7c2696dd982ae : SUCCESS in 49m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bce1eb82116c401b80ecc70339bb6b0a : SUCCESS in 1h 39m 22s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5d7d77c825f7464b96d49d58f97be271 : SUCCESS in 1h 39m 02s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/363e72bbad3b407f9c4a47b1752ef0f5 : SUCCESS in 2h 19m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ddd5d1993bd04479980da3a09a57faa9 : SUCCESS in 2h 09m 21s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/855eeec615fb441bbe0b6d1e3c33fa93 : SUCCESS in 1h 27m 40s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e6140c4450954a8caa0106e5de930f54 : SUCCESS in 1h 25m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/20520b5d46144aa6b55825967ccf17a9 : SUCCESS in 1h 00m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/45256ddfac3a49be9ff696e1cede830e : FAILURE in 2h 26m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fcaf1e2229544bb9a69086dedaaea628 : SUCCESS in 1h 03m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b2ed5d8e03a54bfb97eefd94ae3816b5 : FAILURE in 1h 00m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4f696d4c693345e090228882a926f909 : SUCCESS in 1h 26m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7544f295d0e10e916e59b0ec8313577149b28986","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-10-22 13:14:40.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5e8c2013f1e9b3e0f68efd0245717aae6046cc32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 15:33:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d999deb1b79489d8a1d42ea89032e94 : SUCCESS in 9m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ada03d89d0b24b549a8b66f57949a906 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8adcbf017974a61b86f9e1f1dbfc3d9 : SUCCESS in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/548857b6906e4b4c99e7ccae49d4b1b5 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a10cad6c0094b5eb09c72c8d755b7dc : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5bda285e06af487b953b14d2492c7500 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05dd3af18f44a4198827d2af91a3cfb : SUCCESS in 11m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/44612bee8b9c4d309b372f227ab68b5f : SUCCESS in 5m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7cc61e6d9f3442ed908deead812c7eb4 : SUCCESS in 6m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/234d3ffdb24145fb8e61c5c21633492a : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2879233f34074315a42e542df218685b : SUCCESS in 6m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/42ac9118c1844243b28336c7acb4a23d : SUCCESS in 48m 12s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/da632d1e298e4837ab22b3a989dee4c1 : SUCCESS in 44m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/52c7cc68ffa3411aa85e29299714c7a0 : SUCCESS in 1h 54m 55s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e0f8b4d5fbcf40dc8e2d63ccb7671b5d : FAILURE in 1h 45m 05s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2f94029a9bd8487691e86ab8702c9074 : SUCCESS in 2h 05m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/297a2a9a36c34e568fa933aae93c058a : SUCCESS in 1h 59m 41s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f81b7a0abcac4bc0a54e902fab44ade6 : SUCCESS in 2h 09m 50s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2eb0cbd317dc4a5598db8f69401a045c : SUCCESS in 1h 13m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5a45dfb516fe4f53ae93837ac6197a1e : SUCCESS in 44m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8e76f01480dc4656b06454b42406955b : FAILURE in 1h 39m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4f05c39f24c4ccf854b242baae3439f : SUCCESS in 1h 12m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/20aa234501f241ef8ee93712af486505 : SUCCESS in 1h 03m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/806989541fdd43f2b24d79e7569e88cb : FAILURE in 17m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3fa1f27b9f2bd21b25270ff2b4d2f3d8dbf21e55","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-10-23 12:44:34.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"545683b587a32a4afd5880d784ce5d699617d9f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 15:30:05.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf737dd8c8aa4ea196912441dafd2a37 : SUCCESS in 9m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21b70f953afc4307a2b51d52e7e83ac7 : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90836029929449caad3c5a1f3bc15c09 : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a60c5f18d0da47de985a51d6ed69a500 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ffec1d9ca534a6ea0e9ef2850f26942 : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af6f76ada2b44e66879285347322b30c : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb5727ab6b5a433aa350a11975256948 : SUCCESS in 13m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bef087bb30bc42a6890f8331732f27f2 : SUCCESS in 6m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2be76b75fb0241edb83edb3ba7e56322 : SUCCESS in 8m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/42e354116b494455a0a17c53c285a9fa : SUCCESS in 7m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1339a1a5bac34df481f22c67790bdd08 : SUCCESS in 5m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a2422bb7ff894a079d83d1ae395bc77e : SUCCESS in 43m 15s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/60534742523e4f489d00cd0b22a8152f : SUCCESS in 41m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1a5d59716e31486f840f03ba7661aebb : SUCCESS in 1h 57m 33s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/722d75b332be4de7922bfbf8fa43b818 : FAILURE in 1h 44m 32s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ad75c1725c6b4d9ba50b1b1d0c7660a2 : SUCCESS in 2h 17m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4db8fb89ff6843e0b36c229c54e0bfe3 : SUCCESS in 1h 57m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0b35387d15fe4db0b0a24cc39a8c0f0e : SUCCESS in 1h 17m 52s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/57ca3d29d5a34de4a4199b8a7ee16e68 : SUCCESS in 1h 39m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c8a0815dcec74012a9d8e0a89c613692 : SUCCESS in 48m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/15154b27b30f4fbaae2e05e6025c99a6 : FAILURE in 1h 20m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ff165cb8fcc4afd927decaace0777bb : SUCCESS in 1h 06m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ffa3cfd706a94197977865217458d949 : SUCCESS in 1h 19m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c0e6710ab28948e9835f89f9dda6ad1b : FAILURE in 1h 09m 29s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1c3d50290ab99c22c2e6c35b7eec117b52f12831","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-10-25 10:55:13.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"2afa0405eebf7a36eddd6eea0c3d0957e851f732","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 13:29:34.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67b27da2bd9b43c182823c3935e991e1 : SUCCESS in 7m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/383f26b8c9064558b76bf61d9df7fea7 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9149f79a95044168e3f121c2df2d0b0 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1335cac7b93a413ca616ca60bc2fb275 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4549952d03846c2be4a6afc0ae95293 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d1ed888f06734b7c84a09f06d17287dc : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fceccd0bbe442eda8338d62bbf73482 : SUCCESS in 10m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a358146d239b4cc380bb266c5e231025 : SUCCESS in 5m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/29caa8348a204a2296267d0148be1d8a : SUCCESS in 6m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f64a620aab2e4e809b88a52b42448d15 : SUCCESS in 5m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/969d5fd4ff1648629e86d4cff20a8105 : SUCCESS in 6m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/55c3bcb8240a4f019d218818ea9b7bb9 : SUCCESS in 42m 09s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f4321881d2ab4267b556a5aca50db0bf : SUCCESS in 41m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/643416493d0d4925a9adb1676188bf9a : SUCCESS in 1h 53m 15s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9d579e8c12394fd38646d4e6397f6d4e : SUCCESS in 1h 31m 20s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/932204c5da264d868e4cb810e7aec712 : SUCCESS in 2h 27m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/662cadf22fb945429c4b946f93a2e821 : FAILURE in 1h 31m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5c1a7d100edc4a8180e3f1dbf92e3d57 : FAILURE in 1h 39m 29s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6866c41b91874f4c8a49fa92a3a6c775 : SUCCESS in 1h 16m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/12b10cab88b9402cb083813918c60ff5 : SUCCESS in 1h 02m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/460c5c75e36545c6b4bc4853d29647ae : FAILURE in 1h 47m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bd38978775ae46b28cfc163584fa55ba : SUCCESS in 1h 01m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d720c8c991a04c29aad05c7d73e0ca9a : SUCCESS in 1h 20m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a4b7aebecaf940f0b8b75b10bb3af3ea : SUCCESS in 1h 32m 43s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6be83f1127f5870c3d9194946616274d70fc87e0","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-10-28 12:34:55.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"33d1157de98ae789e3454dc25452694a7bec3ecc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 14:46:29.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f04e366526c4c6fb1d23edd0bf62206 : SUCCESS in 8m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df78af45174f4fb4bd51ef5eab4042cc : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baf24b81511443ed9c3e7f53c88b178b : SUCCESS in 8m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ecc28622db747fca69577b071fe92e0 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7730df0e1f04570bb5382c3b2902b5a : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c4400e35fd9348b6885750863ce6d925 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa8d1369520413c8ba6127fa0f63455 : SUCCESS in 10m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dcfb5cb255cf416c92523c1b525e5ac1 : SUCCESS in 4m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bbba7d6dab704eb1be03f0e9c68866f9 : SUCCESS in 6m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/578bfa591df4480d95f8aff0c46ab40f : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/092cd8e77f474564937c6bf6ea792a6d : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d0dec237903044f8adcc80095b39e4d6 : SUCCESS in 49m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d304b82106524878837bcbeff0eac1da : SUCCESS in 47m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/05e6bba8dd4c4aae9b3b7c0a670e9f01 : SUCCESS in 1h 36m 21s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/622455e2ff984f59914b55c2d1ad38cc : SUCCESS in 1h 43m 42s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6c77d8e9746e4c68a47e1637250bf02e : SUCCESS in 1h 51m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c345773874ca425a9612d6091fa39b02 : SUCCESS in 2h 04m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/8f90ce37a54e4677b57b5689feb6212a : SUCCESS in 1h 45m 15s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/83dc09144b7549d7a475c3ff990f472d : FAILURE in 1h 27m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/59bd81e0ee9e42b18624e0c03fc8c496 : SUCCESS in 53m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/52056b4478d946bcb5218c8978fb4960 : FAILURE in 1h 12m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1dc9fc15408f4f91bec8b56b5cb61e8f : SUCCESS in 1h 04m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2122f657e8a14e2e89dec3d02345ecf6 : SUCCESS in 1h 17m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/657e3791f95f4562aeae359d8784cc5f : FAILURE in 1h 12m 51s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"abc68a97a1a70cbbcd460629ab57e22b9a598d53","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-10-29 07:17:47.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"5e34390beaa4d179b1c8441f4e78e602bf01f85b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 09:35:29.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/756348f0021f4f928c0e4fa0c9d0a10f : SUCCESS in 8m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ece900988f7448aab12d42faf4f55a0a : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6a89fd7acab4a86ad791c91f5b5a564 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd2f99313c044b02afec080a5a0ce0eb : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1293203d2ab4cf5b00da984508d46c3 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52bf29c097754fe8973e70b54e06631a : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/193764aaa0274e50a5f5f0b2428c71b6 : SUCCESS in 12m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cc775928892144739b369b7dcb414967 : SUCCESS in 6m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bff618d39f4a469393f5d1570019f141 : SUCCESS in 8m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ef31acaea0b34e17aefdd5efadce0048 : SUCCESS in 7m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3abc0767a26c4bfea87e84712ad9d163 : SUCCESS in 6m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2dfb8940317a41fd9f2f14d622574e98 : SUCCESS in 44m 29s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8332aab0ebf541ca9106e55364d383f8 : SUCCESS in 44m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/88ae74acce8d4b4186f5d7f8169c94ad : SUCCESS in 1h 41m 48s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7289a94071ba4b02ac3a1f9461d4edc4 : SUCCESS in 1h 49m 05s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/eb0712c9277f4c14a41ca0ecef9f91f9 : SUCCESS in 2h 09m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9dad02b9e59745b68420fa87f10b879b : SUCCESS in 1h 18m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ae9ed109cf1b4b97aafe7e3b0b139d3b : SUCCESS in 1h 15m 34s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3a3a1c03968a41f9a3b2647f8dcd5023 : SUCCESS in 1h 28m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/edf6029d1f3c4763a767a361436e1c24 : SUCCESS in 47m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fa6aae8afef74068b5f714c0ac5c12e6 : FAILURE in 1h 31m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9833006de6b4eb789f8f2a194601c5c : SUCCESS in 1h 09m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c00ef2faee624143bc3d4ec41aeba8bf : SUCCESS in 1h 05m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e9dc2d6473a14f739fb48c42f4b82d9c : FAILURE in 1h 11m 28s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"378fff13d2cfd8e2fd83f4022fc102fb143d5106","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-10-31 12:16:13.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"7d239797fe49ea99bdd4fbec77a4b35e9b602e5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 18:25:18.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c4c2943f8324e1aaa62174cb19df9e0 : SUCCESS in 6m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/853b9db14e1749fba35b0b718b282bba : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a893bc07f0a546bba0b4155cabcc4bc1 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2a6d2da13c674c5d876ad78516825591 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86e895e4fd2341cd84c30abe61c9bf4a : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/28b64033e909467d9eb43ede7fd6ccae : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cd5a0d051354d459fe1cf0a6efd7df0 : SUCCESS in 11m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f3a05810b2bc4d5e8056e97f35e60ac5 : SUCCESS in 4m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/25705a1042504114a1096fa5034e8ce1 : SUCCESS in 6m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0c8ebc3f965c4cc0a59e41834f89a272 : SUCCESS in 6m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b45a633f2d54451b3594cd8fe9f2aa6 : SUCCESS in 5m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fa0fb386c07645a1b6981c63e4a3c34e : SUCCESS in 44m 11s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5b947358b95a4e339174f7d897f7668a : SUCCESS in 48m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/92361b1c6a7d408a937dd6aed1d3b9ae : SUCCESS in 1h 31m 29s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4faf9a01b9284a298fa2d11bb6b37ac7 : SUCCESS in 1h 47m 02s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/85c7912b93714fe3a7406eacacc1224b : SUCCESS in 1h 45m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/782e344ac5e74db7af3c7b8538b7fc0f : SUCCESS in 1h 42m 43s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/b5f706bec9854c1c9ce94585986ea5de : SUCCESS in 1h 42m 50s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d4584995a431491eb721de31ed2285dc : SUCCESS in 1h 10m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5212553bf1654b28b053c4f0a5d65df4 : SUCCESS in 51m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1ae0c5f7b0d240259f8e89bc0334214e : FAILURE in 1h 23m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ee2d6a719c944c0abb0f5615276bf46 : SUCCESS in 1h 08m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e99b9064597e455fa4d962ed39b244b2 : SUCCESS in 1h 01m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/49600516c99a4163b4732b517cbc1444 : FAILURE in 1h 03m 57s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"0fb21c290abec2fae3441ca7070f69eaea2c89d5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-01 20:41:44.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"8b55d4bc941f44c34f7e40a25d732f89961f0b9a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-11-05 11:08:41.000000000","message":"Topic changed from jobboard to jobboard_dicts","accounts_in_message":[],"_revision_number":12},{"id":"0e27513d4c800d0b6b63e62d2fbf1a6f9972aff4","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-11-06 12:01:48.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"186067c179f6bbb6b3b69479f3a6e1247d3ec89d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-11-06 12:04:29.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"fe9fd2d161947b0bd233973ccf44c77c1f3fd054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 14:23:24.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/381059c3a62549f7bb69428b2da05adb : SUCCESS in 8m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b571473685b4110b8c17b87f262f193 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76669376815e4dccb2ade8d1b38dc545 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3d3ceba872234048988933adb433a94a : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45586ed3ff8f4ba7a5c04c43bbd9012c : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/618bb07004424779b4eed41b834a1455 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/729729e123eb4ef9b93e7774701fdd37 : SUCCESS in 10m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e9657362be464e97ba505d0903cc6937 : SUCCESS in 4m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5108203acdd643fa8cad6bd2ae584010 : SUCCESS in 6m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b0b40d9ce9244617a07266e84dcf61d0 : SUCCESS in 5m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e1698adc3c4f44f59c91bf5434a028ca : SUCCESS in 5m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c87be2e846644f66b4e46e4545adb361 : SUCCESS in 45m 58s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ed3671c03988461082f33f26fcc5aec0 : SUCCESS in 46m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5de050b9fad246db97f297433dc8ddff : SUCCESS in 1h 41m 13s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/dbaaf9049e354ee5b3678e8aa2c38288 : SUCCESS in 1h 41m 19s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/43b12f80fe0e453f916bc488b46f90f0 : SUCCESS in 2h 13m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d4bc34af56814564a87bba46ebdfb3fa : FAILURE in 35m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/bdf95fba8ffa4df3afeecfef40c11638 : POST_FAILURE in 1h 53m 58s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b145f491f9694de0b6a4672b9e648dab : SUCCESS in 1h 19m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8425d8eb4b194007b1e09ba256487bd0 : SUCCESS in 51m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b3861cb1a94240658b96f97daaf2d1e8 : FAILURE in 2h 05m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4062d7a4bed645eab124f21ebbdb7af0 : SUCCESS in 1h 12m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5705f15213024bb1921d97040d57057c : SUCCESS in 1h 15m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5a33844df4b14d6195c3c3381629a615 : FAILURE in 1h 28m 26s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c75fc1a33f4c676153e187657ff676dc6d02eb7c","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-11-07 07:53:55.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"42457fdcc477779bdfab7fe369d2e66b10b9f0dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 10:30:24.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/182e288c6cb140e091087d5d5f9213b7 : SUCCESS in 7m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a41b81840434e30aa2a85c108d930fb : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d29825612184bc8b6edcde53c8cac6b : SUCCESS in 6m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c8579b3067e48fb9f98d652b5565f57 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/744c437b7a004e888671ced50a335c4c : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d3cf764494748abb4d75ce0beb41f1e : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8b72de8278746949ddda102559d370e : SUCCESS in 11m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0b80d6561aef48fea43a56737c31dab3 : SUCCESS in 6m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4e068b25d8b641f5b46483ef2f87992b : SUCCESS in 7m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/740840780c004d61af8bc642869cdfdc : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b31fad96c2f247558fb0fe0dff2e9133 : SUCCESS in 8m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e6bab57cdfa24421ae10fc506d6cfa8e : SUCCESS in 45m 02s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5a97bfb9859d4826b41974ece097b9bd : SUCCESS in 46m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a03592e6a02f4b0da9d4c82e0eb8fadc : SUCCESS in 1h 33m 16s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c4aa34673a8f4560a86da8c34879c2a8 : SUCCESS in 1h 40m 38s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d1f7a861296d413b8838fdc9be061e29 : SUCCESS in 2h 14m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cab316431a444f1a8ceef02605eb9a4a : FAILURE in 34m 21s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a4f526df4f204fefa71764e85e61014e : SUCCESS in 1h 50m 43s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cae3967a94b14204ae77f78f3468d542 : SUCCESS in 1h 23m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/10cd6fd4a39c45c284225f2a090f2f3d : SUCCESS in 54m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/15f24b914f254f16854d00b57cadc2da : FAILURE in 2h 31m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/56cb98b7583649dca09fcf9e5cd604f8 : SUCCESS in 1h 17m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d320cdb6260944e0b8d9113a3b4784dd : SUCCESS in 1h 15m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b05d4c1f85274419ac1e43fa2a80d1f2 : SUCCESS in 1h 41m 38s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"7d18946c8944532baab9ddab17731914bab0667f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-11-08 06:22:45.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"70d095cb90ba32e6b9ea90ce7d83c76968fd141f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 08:58:45.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78ce279f1f8348cd830aa37221ed978c : SUCCESS in 7m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de6ef7674150403ebc82237afceb7642 : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d63153e5501e4cee9c6311c68f7a4330 : SUCCESS in 7m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce8f75b012e14239a8d15c4f0d24c716 : SUCCESS in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03ae16be71cf4372a2658056fbf21d34 : SUCCESS in 7m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d1a1a68e5de049f4a481b785dc208a74 : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe41410df46042b6a8c94234f9a116e5 : SUCCESS in 18m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e5cb360084a54c93867cb1d222440026 : SUCCESS in 9m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7c67d0a7097a44048b528277a136d909 : SUCCESS in 6m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/893fcfbf19874aefbfa8ab553bdfec20 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3acf79afe98846e19565bfcf7f789f1f : SUCCESS in 8m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a61cbb116f5f43b4b0a466a5133eb837 : SUCCESS in 1h 26m 49s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0f282732b8494b999d1d097e171c50c4 : SUCCESS in 1h 12m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2bbfe54a0930453e91eae7f7280fc97d : SUCCESS in 1h 50m 38s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f35b6b3c7f53481d9bd84ba844f42579 : SUCCESS in 1h 59m 06s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/92f1bf37bf1447919b5d1bbad5b02a5f : FAILURE in 1h 05m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/19d696fdf9104925a254d6795a0433f3 : FAILURE in 36m 11s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0891db324b3948ddbb087a528a977d58 : FAILURE in 1h 09m 20s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5e6f112843394f8db69ac5af7cea1729 : FAILURE in 1h 18m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8f21ed8b14a04cdc96e6f3e2acf7ea44 : SUCCESS in 1h 33m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ae7f19692065468186675a18f1c84ed2 : FAILURE in 2h 24m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d7e44b05f60469ca609acf50725fbd2 : SUCCESS in 1h 05m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4e460e59446a4d408ac367f601b11ff4 : SUCCESS in 1h 02m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ea57e062a1a04067bfbe5fe89af4a065 : FAILURE in 1h 06m 26s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"6e29c4d66a9dc6072a6b4df409fa8cb7c808a805","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-11-18 20:29:13.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"979378973053a5437b0caefc97d8db3d7a3eefd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 22:46:34.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffc06cce18734d43b8be0cd2216581a4 : SUCCESS in 10m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e17c41f5fa164d619b26aaf0b3b11f08 : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98478f1dbe954b2c900bb061185f84ab : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a50b140b3d224d4d8d2e2a2e373d3cd5 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ce389baefd54ab2a99978df9f3e8484 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b06033643b87422483ee50b26b4a84ac : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/127661b91cdb49c1b07077b571e6d192 : SUCCESS in 12m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6a4a0fd0b0024b5f9a34546f9fc990bb : SUCCESS in 6m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/296390d0ff2e48ef9062afe31f34aa32 : SUCCESS in 7m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4ccf490dab144167b22e112bf23541c2 : SUCCESS in 6m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d185b7efcb3644f5a30c969f58cf8e70 : SUCCESS in 6m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/48442437b7d54e60add4bb0fbff6f951 : SUCCESS in 55m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/68e28902304a477c8a0e9ec8efe95865 : SUCCESS in 43m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7baee2db802d4c45bfe80d16d642d36a : SUCCESS in 1h 39m 37s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/11c85d958db6462786a577e5e16a29d4 : SUCCESS in 1h 45m 17s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/276d0f98009d40bc911a3ddee434eedb : SUCCESS in 2h 06m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fc2475e2649f4998b41e62fd90e75e88 : FAILURE in 38m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ab927054e5904cd6868fe25a0c713297 : SUCCESS in 1h 14m 32s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b72f925266b9438cb14624a73feb2622 : FAILURE in 1h 16m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2a04d0f178c64badb67dc5d6841c6f09 : SUCCESS in 47m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e09837b88ac542edb219dc54cc3e4ec4 : FAILURE in 1h 31m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b20351973fb4501890e8f38be8b0f64 : SUCCESS in 1h 12m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/672704348b0f4908a7386924f8251eba : SUCCESS in 1h 04m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/af209e42af7f43f0a39733e44bf5b701 : FAILURE in 1h 13m 25s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"7b84e9b5ed34693890f720ed7e7b56d3086802d8","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-11-27 17:52:07.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"282d35f9ed811176d3cda97aaf5cb8e568d21422","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 21:07:36.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e117e9cc5e34c3d8823fb34f507a280 : SUCCESS in 8m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42a33eb2347f4164aaf4c183a116fed6 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e4705af4e944fc1b1b1df142a1d6894 : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/39be8e3e618f4d42aadacefbe1931b0c : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6753865acd24dc4b000637ce3d89f42 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d3a8538116f34c89a35974cfb5070dd3 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9717c3b08be4a29bb592762c1872ae9 : SUCCESS in 9m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e09a5f1f930b47b3962e2c3acff6fbc9 : SUCCESS in 4m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0489e637251f47698f231b5276ff4b50 : SUCCESS in 5m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2dfae65b63e54cfc9e5dfd3e7a5883ce : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a1e87427b76f4ceea6be56beb1eef925 : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9d8c91fce0df4cab87ba83a1d06bded9 : SUCCESS in 54m 10s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/49d8cecbb66e44068533f32eaa83578b : SUCCESS in 50m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bc91d18c39bf4b78b53be37981a19360 : SUCCESS in 1h 36m 37s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9a2e29bc5f214dc4a71db7a5d346e011 : SUCCESS in 1h 49m 55s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5a4203d7efe0494e9682da2bc8bb73a4 : SUCCESS in 1h 56m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4d31cd6d8ac2429cb93ea819b750cfe2 : FAILURE in 32m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3a54e92d445548698efbb88c423f5ad8 : SUCCESS in 2h 11m 21s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fe5b0a25e3534345aeb67e0ea8f15b6c : FAILURE in 1h 01m 43s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c16773c1aa6443ba9090967a5ece8d32 : SUCCESS in 1h 02m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7aa1979ce82f466096b57658f8938bb4 : FAILURE in 1h 50m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4871f29db99d42e58547329fdd48a449 : SUCCESS in 1h 00m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b733fe882a2f40bd92911a3ec5574fb5 : FAILURE in 1h 09m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/777f64b53244492f8750f867fe94f845 : FAILURE in 1h 19m 08s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"1257eeaeaa5c2066dcded68fcc4893509d24aedf","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-11-28 17:33:25.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"be78f195475e10840cddfc582b4645df868b6655","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 20:06:12.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79cd484776f84142a5a84622ee380c77 : SUCCESS in 8m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9d5cb751e7849388b46af8c1672675a : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf1327d029414755895b0e9df9176d73 : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/03c6d762c801452399c2fd9c09971a08 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b445c03e1f4242fbac0a291d97194fa7 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/868b56d999114c1aad56394dc302ae56 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d40c0cb5d784324922dd2c87a36e754 : SUCCESS in 9m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/11548672a94849598086a9ec912d6d1f : SUCCESS in 5m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a49807c21a3b4b088755a1eb545f8aca : SUCCESS in 8m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bc7228cba69e468981b093e2d56feb52 : SUCCESS in 6m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4cbb9f42fb454b60b864996f3413e81a : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8e48b7f3ef034ebda5abda1a8a31713e : SUCCESS in 51m 13s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/849cf7f3bef4450b8f32bd41ac1a12da : SUCCESS in 46m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02ce843d50684748946e95e9fab179eb : SUCCESS in 1h 33m 08s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c6c1ebd2e1c9422f8452a45a537c8bb2 : SUCCESS in 1h 38m 44s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8ccf6d2255d74a00b98f2df5d54f0bb3 : SUCCESS in 1h 57m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2850ffdf160a42dc822790d60547168f : FAILURE in 40m 14s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/86ee86994bff44dea151ef45292b0056 : SUCCESS in 2h 13m 48s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/37293f5970e84c748ab25020c79d760c : FAILURE in 1h 03m 37s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/81ed3a11b7074a1fa441edadbd374852 : SUCCESS in 58m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cf8c8c757a2f4cd392ebd3192c6d2912 : FAILURE in 1h 46m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f84cf1fd2a2c44e9a8c1aa196a0deba8 : FAILURE in 1h 25m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/756931ebe6e049259d99b37ab6d7fbd3 : SUCCESS in 1h 11m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9fd72d6a0df34165b91172964d7433ec : FAILURE in 52m 50s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"1b27b822f8ac2df3091b8921b8f76ce5676a7c51","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-11-29 02:55:37.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"deefbb1adef8f3d7d446feeda699dc8029f4e903","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 04:56:06.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f3bf8932efa4d199d039fcabdd899bf : SUCCESS in 7m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed69e7a943d8411e837840b5ffd2e76f : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad46d95b437b4afd90e8d2db36816e94 : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a6ce67bc36b48f090839d144f6f4ef6 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52d9cd817f1444478f0c5ef43b86ed51 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4be0c503e2884c8997fc52fbdec0b765 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11c4e33b96cf470bbabf630e1cf0f149 : SUCCESS in 10m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a0872ac1a5734a1f921dd909fa76c095 : SUCCESS in 4m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/634d024326e34609a932d4c68129356f : SUCCESS in 5m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c995c735de604daaabceef9f2fa37aa0 : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/399e40e76af14a28b6eaa2249f510347 : SUCCESS in 5m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3fc2148ccc984d3992327622d9738a11 : SUCCESS in 39m 01s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1117461df5b44401852cf209a6440f8f : SUCCESS in 42m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ddbd534b2244a7984ec9487c6b4f3f0 : SUCCESS in 1h 32m 16s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7521e684d9e44565a3cf0894a636a72f : SUCCESS in 1h 34m 38s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/542b6b9975ac40efb3b5fca1e5b73454 : SUCCESS in 1h 58m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d1512c31968a4e45aeab5ed06cbde0e9 : FAILURE in 31m 46s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/1b2eaff5f16f44ab9cc5c791390fe8d7 : SUCCESS in 1h 48m 27s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cab20a0bf00d4315a9537b74f08aad14 : FAILURE in 1h 02m 52s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ce00621622f64d4999cd48aeede52aeb : SUCCESS in 52m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/06ebec69880e41088e5a76cd36d58a03 : FAILURE in 1h 33m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/93589dfa0215433fb525354cd52a693e : SUCCESS in 1h 02m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/cf706ca8df784d36941a4f3ea73b94b3 : SUCCESS in 1h 13m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ed11f09b21c94ff4a12407bf2754dc6f : SUCCESS in 1h 24m 44s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"dc15a7d8109a0ad8f4605bd3c1d74131650299b3","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-12-03 04:38:25.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"d93132ac376dbcf78ebba5015d0fea67a67ccc64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 06:29:33.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18f1e368e2fb49809de05bd92488d702 : SUCCESS in 6m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/105a91d139a6457e8fed0bc3d4f642c9 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/130b36e0d744431eb3003252c573ec85 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afab1ca8b53144d59166f8f19366083c : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d46463a601ad46758df449a0ab7c7baa : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48a95f1d94a44c31a98878e88e3891c5 : SUCCESS in 9m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6ee50acace9a47afa2f2d1328d40d11e : SUCCESS in 4m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b48431fa81a04ef8b407eef2ca12e369 : SUCCESS in 6m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/be4f9f21c32a42e7a63f796176c031ad : SUCCESS in 5m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3d188872dca447c3b61ecb15ceb2a3d0 : SUCCESS in 42m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc58450424834283b6fc606fa7119c0e : SUCCESS in 1h 25m 33s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4e75912a97c94e10a76519776d78dec9 : FAILURE in 30m 58s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c6ba2d4678ed4464884b63f344ba1452 : FAILURE in 59m 58s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ed1936fe90ec45409978f46ae639e9f0 : SUCCESS in 50m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/16e05165bfb8409c889ada476b611786 : FAILURE in 1h 41m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/071a13c0914841e4b4f342ef0fe4c3c9 : SUCCESS in 1h 04m 11s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"49318112ec4947ce2e05d542d3f4e720db1a67dd","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-12-04 01:07:59.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"89e2fdf75347d1c80a94a2c4c3bcf34343d7950e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 02:58:35.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebfbf43cc9eb4e98b7b55ceab6989c76 : SUCCESS in 7m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14eb29d803cc4d5f95082d316f492d8d : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e120bee757384396bb8d024c67921221 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/719835f9b41b4b4da7e5a82fef11420b : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/593e4270b0194605a261cadc92ceacd6 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d6c140163ae4b11b4b7ee4e7781c4f8 : SUCCESS in 11m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/47a4cc0a9a9b4edf8dd5b2aa04516ab3 : SUCCESS in 4m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9bf0e4ccc75c4a6ba9790ebeba02f18c : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/25e5b1b0eefd477aa285ec16cba9c4ce : SUCCESS in 6m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f5f322d6922e4ec08bcee4df3c7b8665 : SUCCESS in 46m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b8792873ff1945b49b75036ea0b41047 : SUCCESS in 1h 45m 21s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/100b29baf9394d10b8085a25d62a485b : FAILURE in 32m 53s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/937f1c605e4d4948a109750ec61935dc : FAILURE in 57m 25s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b43d29f632e0455b84dcbb6077a85413 : SUCCESS in 51m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/50702432c39c4ba0bb03b79aa139639c : FAILURE in 1h 30m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7cbf6895aa7c4c3ca09236741602bf1d : SUCCESS in 1h 06m 09s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"fa56d08e49c4c523bc0ab94dc88631bb6c523a68","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-04 03:12:58.000000000","message":"Patch Set 21: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"8445f27a58dce2639830d3e04442ba57f5a22e76","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-04 19:43:50.000000000","message":"Patch Set 21: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"a6f8f19af61e544fe72946092b8c2b257e3e3bbc","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-12-05 01:42:31.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"7c3f19a7d4d82f95aa82d963a06a24648a88d058","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-12-05 01:42:36.000000000","message":"Patch Set 21:\n\n(3 comments)","accounts_in_message":[],"_revision_number":21},{"id":"9cb84369b44821f117d1296948068e092e9fa94a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 03:38:26.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6784d26de8f24d9f88369de513f68933 : SUCCESS in 7m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62d254ee2e454ec5aeec0c20b13a9784 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c32e6b66691456aa4395b357c1e7bdc : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1573c8bac6274b0587abf6dc9e035e81 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a00570afd5744c239a4df50388791a10 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/642a37c78f0a4c98b0cd729cd09b802b : SUCCESS in 11m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ca52a569b04844f791ff9e6d78cc642a : SUCCESS in 4m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cf36b378f2774231b64b282c5125f5a1 : FAILURE in 7m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/72e00119992b4b0d93315790aa3fc2a5 : SUCCESS in 7m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b1dd61e8a7b243259c669d86cfbb7b64 : SUCCESS in 45m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aaf7db10f4e84894b8d97e9154b868a1 : SUCCESS in 1h 48m 57s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cf3342f765c8469994bcf102ad9b099f : FAILURE in 36m 55s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/151d761dd6824d06b6573446b76bdb80 : FAILURE in 56m 34s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/54fbe8c1a5e34ab0b793ca0103404825 : SUCCESS in 1h 41m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/70b9a6479657430abe7de36df40f595a : FAILURE in 1h 30m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8ff7946460845dd9d2dd6f867331d3e : SUCCESS in 1h 02m 10s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"f15fb8a8478de2507ca840481eac9bba9b29d5d3","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-12-12 04:14:10.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"0f47e25a33bf38e54b36c1ffebf723874218c949","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 06:02:23.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92bf9564d8604263840afe6cd267602b : SUCCESS in 6m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffb4e40279894909918dda0935a5c505 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b964d00f6a544595b81922217cc07913 : FAILURE in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efdfc5c5d24845a591d021c9504b5b63 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e67789413cb4c3dbe74da3a09d3f132 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b7172b488644f1ebd89a5f0e1c61bff : SUCCESS in 12m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d169befcefbf47a28dc9975b526db249 : SUCCESS in 5m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e14315cec4684c69b4c4636df763c9d7 : SUCCESS in 6m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6ba83b72172b4110b3e67878441e241a : SUCCESS in 7m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4207134ffb074866a40e18713b4932d8 : SUCCESS in 43m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/66ca98220638496fa6e1e3c60e844ae6 : SUCCESS in 1h 38m 23s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a961c630b1c74d0eb326d3bf69d439c8 : FAILURE in 11m 54s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0b26052ae1c541ce8b5064591917b06a : FAILURE in 1h 10m 18s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/943f4663f07a4989a51099225ff108e6 : SUCCESS in 50m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/75e5a6ee386f4a0ca5af64815b914cee : FAILURE in 1h 38m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/34d40ffcbb8d4117b42dce1f2ed97e4d : SUCCESS in 1h 06m 22s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"43c4d79ccc83444cd22df72e190708ad6b29cc26","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-12-12 12:34:23.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"998faecd48e74da10278285263262de47b9310c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 14:35:54.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92012085f26f40a4a5449ec13fcfa3ef : SUCCESS in 8m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a8384406cc9455588a7554ac4479263 : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27ec35b11f3f4f56932e8d33317a1b3f : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/557cddd3d7cd47f9900e2eb4ee6f956a : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9db3254ff3ed4fbe938720d891db84e7 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df810af11f664498a6ae9c6b16f4d25f : SUCCESS in 13m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/15c845ef4ec64412970aa2528e7d4ec4 : SUCCESS in 6m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d58dbd4ea73d4716b4146808e9c7f76b : SUCCESS in 9m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f74af85d537d44949bbbf0a224879c8b : SUCCESS in 7m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/96f42efe578a416a98146867c9db13d1 : SUCCESS in 54m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f053692186d84837981f14f5533d5f30 : SUCCESS in 1h 58m 52s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/eb99f23e96ab4934bca43f5dddc9d782 : FAILURE in 10m 10s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ad2f05d7169d4b18bea2a5c9173e30a0 : FAILURE in 1h 23m 30s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/093b8c6872cc4eb5aff89fe07b28701a : SUCCESS in 1h 19m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/caa66f1f414441378b05ef559797a4f8 : FAILURE in 1h 50m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eef5a59087bd487583787c14ddf9a4e8 : SUCCESS in 1h 18m 11s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"804d9f91e0e4fb64ab7c446c69302aa49dfc3919","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-12-24 12:23:43.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"f536ffc1114de864e3d0988d95f799cca41e6e8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-24 14:08:55.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dde466becae94342a6af148866e2a593 : SUCCESS in 7m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/953bfd7ba1c44d4bbd248eb1b912002c : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68cc3d867ba14ede8e047b3ae165af69 : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2907f9c525764fa1b0e5f9307754287f : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86abecd58ea5458e9f633f947245831c : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c8f7e13cb044dcb126c61d54a45d51 : SUCCESS in 11m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1e1c850affa74a009f55b359f41512aa : SUCCESS in 6m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0ca1c43484504586b41240a715b3b67d : SUCCESS in 7m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0f274ba0b9f34514b94e607f36dd0429 : SUCCESS in 7m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e3792ddfa91b401d9190878cc5ed32ce : SUCCESS in 44m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8317bd95b8504120b95ca11d99b37282 : SUCCESS in 1h 37m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/015693b450e64c168d1cb0b936238eb7 : SUCCESS in 1h 05m 34s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dabe8f19e2e2486380d6970f7fb256bc : FAILURE in 15m 27s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cca6eeb136084ab1b4407aa829cf5e8c : FAILURE in 1h 04m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3bb5f7f1dec042afb58d813bfa01d78d : FAILURE in 1h 39m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/171c50c36cc74f52b60642a3877fa2a3 : SUCCESS in 1h 05m 34s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"79abb5fb960b58aa5a980d17ba2726afb7f16410","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-12-27 14:38:36.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"d69e88a448ecb48dcc4b5e60488c7977a692dcc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-27 16:10:47.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d70ab95911974c95a2087c4dad225c8f : SUCCESS in 6m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b15c3842b044b57acf34ad8e9f6f96c : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a482aab490d341a2a636672d65749e62 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f105d2c39a5e423b942f555f346d331f : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e4c4ee5c3d74b88912fb1f37a09057d : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7be3c28240aa4a948717ac8ddb73a654 : SUCCESS in 10m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/26cb5695685a4b57ad663ba2e179cdc7 : SUCCESS in 4m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/928169a39dc846c7960fc77c4f9f95c0 : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/76b41be050884d0eaafd6f76808f294e : SUCCESS in 5m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c04b87df28f0410c8281ea94327a2cf1 : SUCCESS in 41m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9b0f53f6cf8408eb34e5f8aba72fb27 : SUCCESS in 1h 30m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8465451b4b624664b97881fd5e4c6cac : SUCCESS in 57m 35s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/396e79fcefbf4c30bd37fee3a672d0af : FAILURE in 7m 02s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bb73da6a8e184156a29ad1e8d8300361 : FAILURE in 1h 00m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f66ce54f8a6348ac9130c69cc18535c9 : SUCCESS in 1h 29m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/00043199fc594054b033d1129ea2cb18 : SUCCESS in 1h 02m 01s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"fa2bc944f07c088d73b4596ab4aec30330bb40b0","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-01-06 19:03:39.000000000","message":"Patch Set 26: Code-Review+2","accounts_in_message":[],"_revision_number":26},{"id":"5e0d8c06e5f4c4df6e6d8821848afd21e3d5c920","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-06 20:24:55.000000000","message":"Patch Set 26: Code-Review+1","accounts_in_message":[],"_revision_number":26},{"id":"af1d55c66b031cc5045a73e06bb46ae1f23760fa","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-01-09 10:43:39.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.","accounts_in_message":[],"_revision_number":27},{"id":"773c50989f5238605e30f9c572556a25608e9a1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 12:28:20.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3848f8703ece4e33baa9d28f1140b524 : SUCCESS in 9m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd0b34841f794d22a306e5fc1ac77412 : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db2d2e084a9244e98c1bfbac5ef3741e : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74d4fba75d8a4a4597e4a64b71477566 : SUCCESS in 7m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9f4ec9d71184073a8e0d680e3cb5847 : SUCCESS in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70425594f8bc4706a7b48f17b7b39523 : SUCCESS in 13m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8f932bbf8f10480eaccf1ecc1c9fc8f6 : SUCCESS in 7m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/069054c89e9b4fa59820a2906cb8672b : SUCCESS in 9m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fefa729f92f24dfdb04392cdf7cc5ea2 : SUCCESS in 12m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/100401adfcef4d12953dc34f8125e032 : SUCCESS in 47m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d08a241d31ed4b92b4c7e7d4316a6139 : SUCCESS in 1h 40m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ad7b54859a9b4701a4a6739930df3d49 : SUCCESS in 58m 52s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dd9eceb7cf2f4298bd4d33448803a2af : FAILURE in 11m 52s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/596f95ae2162497ea15422ed24c38683 : FAILURE in 59m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9dd4ec5d065340ee913d04e99d8048d4 : SUCCESS in 1h 40m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a0506f6944ab456b801aec1ce12bfc7e : SUCCESS in 1h 02m 30s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"41f9eb6af939e68d39fdbbb56843e9f415380c79","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-01-17 08:27:47.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.","accounts_in_message":[],"_revision_number":28},{"id":"70e34e424663dff6ddd2320303f4fb79e4f6d7ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 10:15:39.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bf8da38243c49568eb1534475d5b2af : SUCCESS in 9m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad9acb9ea46d4d099ee544ca48bb8cb3 : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f15758f5e64b4d34a52da818b6b88e30 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2a08abc46d44b0ca942f17ce3d703bc : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73666370605e4fe59cd2f95feeec8970 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f848234d69f4fe2b4defcf5e839bc6a : SUCCESS in 10m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6b5bad975479414482b122b9c6c4cc2d : SUCCESS in 6m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/701796588ad9410986d6fe19e14f5b56 : SUCCESS in 7m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/441d9c060bab4c2b8e58aa5bfa3695c2 : SUCCESS in 7m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2b701ad707934eb1bdc24df1f5466e9a : SUCCESS in 50m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/201025e75a7b4350b8a3f566c3b61e8e : SUCCESS in 1h 41m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5f2cbbbd789244138b4a8d0d835f2355 : SUCCESS in 1h 02m 58s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/96d506ad475646c298b311165600dc65 : FAILURE in 12m 17s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6cf969d09b9747faba86330db3d38c7e : FAILURE in 1h 01m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/86f95dd7e6b54fb5b8d2f6aa0d8fb5fe : SUCCESS in 1h 35m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/936cadae4a664b868728222d237a1826 : SUCCESS in 1h 25m 36s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"8e22108c00564c15f266409e6eb504dd779219e1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-24 09:39:50.000000000","message":"Patch Set 28: Backport-Candidate-2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":28},{"id":"2c6e7fe744548fc5ede78d44158685e37bb810cd","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-01-24 13:02:06.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.","accounts_in_message":[],"_revision_number":29},{"id":"983d12a034524acc56bf82982e9186bfa849f24e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 14:32:44.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86f834bd8da44d54b304daeef117b790 : SUCCESS in 11m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4c0a91c67fb4be2a87a09a20808ea4c : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8ef385d8d9d4a4f95077c3b957b85e9 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fa3c246d73d4e44a344515bee340878 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb15f95dff1742d5a1bec6b7e7e62575 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a63c758a32494838a4684086e6b9a30a : SUCCESS in 12m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f0024f474cd74dd89d727b16afcf61e5 : SUCCESS in 5m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/72dbd532bcc941c1910d90e413fa5b32 : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/61e6616464154504802da9afd61e50b9 : SUCCESS in 8m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4870a3c2f6b546b5ab7fe5dfcfcb80bd : SUCCESS in 49m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15f7925dfc584558b3ff52597e6d1ebb : SUCCESS in 1h 26m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2ebb6b0839ee488d8fe14e7814d88d02 : SUCCESS in 1h 01m 53s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8e8e3caed282457f999fb293a10d2cb5 : FAILURE in 43m 35s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/202aaa80b509444385da9bd6978adf7f : FAILURE in 57m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2760f22f6e5e4280b7a558417fdebb13 : SUCCESS in 1h 28m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/29b6d784cb514c42a6930f8d64a6c079 : SUCCESS in 1h 03m 21s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"e1d4ce7cdae56e500700234a47a8cd07fc74205f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-26 16:32:34.000000000","message":"Patch Set 29: Workflow+1","accounts_in_message":[],"_revision_number":29},{"id":"80589ab72e561eefa1b17c7913dc5ab041ff067c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 16:32:47.000000000","message":"Patch Set 29: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":29},{"id":"f51c0cefa732dba0cb442d8f8738bd30682b9352","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 18:19:26.000000000","message":"Patch Set 29: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62d5a9fb0ac84fc985d9bd12e6aa5a70 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a75c5854f82d47aa84529eebc8f248b7 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cdf4ef52d73458386d20cd4fa8413de : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e10096abe9b4ddba8d6d6d6145ab4e8 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72d9d11027c04d86bace42327f3668d8 : SUCCESS in 9m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/36fb69d5004c403893cbc216491dd0dd : SUCCESS in 5m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bf320602ed2f4e268b8870d62d443c13 : SUCCESS in 43m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/393b18e355e44301bf72308639382066 : SUCCESS in 1h 29m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4bab8557a3514750a3f815877af7369c : SUCCESS in 55m 35s","accounts_in_message":[],"_revision_number":29},{"id":"1cba64fd7df5e32e5826cd219409b37344469308","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 18:19:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":29},{"id":"b15aa59ef590be4ab7e6414c8192575766dcc29b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 18:21:57.000000000","message":"Patch Set 29:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18ced1c64ebd4ead886e58bcd57ea156 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":29}],"current_revision_number":29,"current_revision":"ea6b065cb0e82b2949eed008121edded56ea3902","revisions":{"c013b61bf8310a27fa95cee2f2cd1814eb37162b":{"kind":"REWORK","_number":1,"created":"2019-06-28 13:14:18.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/1"}}},"commit":{"parents":[{"commit":"cf468bb91b64310f3dc87d61af151c5cac4ff234","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf468bb91b64310f3dc87d61af151c5cac4ff234"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"subject":"[WIP] Transition l7rule flows to dicts","message":"[WIP] Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c013b61bf8310a27fa95cee2f2cd1814eb37162b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c013b61bf8310a27fa95cee2f2cd1814eb37162b"}]},"branch":"refs/heads/master"},"22aa5e871a6e534d30c7cec99d2b2fb2a1415a43":{"kind":"REWORK","_number":2,"created":"2019-07-02 11:33:23.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/2"}}},"commit":{"parents":[{"commit":"cf468bb91b64310f3dc87d61af151c5cac4ff234","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf468bb91b64310f3dc87d61af151c5cac4ff234"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-07-02 11:32:57.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/22aa5e871a6e534d30c7cec99d2b2fb2a1415a43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/22aa5e871a6e534d30c7cec99d2b2fb2a1415a43"}]},"branch":"refs/heads/master"},"70936de85ee4f732125e6a79fc09ee56be0716cd":{"kind":"REWORK","_number":3,"created":"2019-07-03 09:25:52.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/3"}}},"commit":{"parents":[{"commit":"5c6b672391848a706a1901d9f5360cc30691ba02","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c6b672391848a706a1901d9f5360cc30691ba02"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-07-03 09:24:10.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70936de85ee4f732125e6a79fc09ee56be0716cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70936de85ee4f732125e6a79fc09ee56be0716cd"}]},"branch":"refs/heads/master"},"75106eba1483b600a93d47ed9ae822a85d8f7275":{"kind":"REWORK","_number":4,"created":"2019-09-02 13:00:10.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/4"}}},"commit":{"parents":[{"commit":"aeae70b74a2ea49e0967e50345a532928b8840f3","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aeae70b74a2ea49e0967e50345a532928b8840f3"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-09-02 12:59:52.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/75106eba1483b600a93d47ed9ae822a85d8f7275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/75106eba1483b600a93d47ed9ae822a85d8f7275"}]},"branch":"refs/heads/master"},"4ae2e9bc36ce69e83486ca48c2c4d5498879646a":{"kind":"REWORK","_number":5,"created":"2019-09-12 10:47:32.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/5"}}},"commit":{"parents":[{"commit":"b443c09b0d76e0c4c8618654e82178b05848328b","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b443c09b0d76e0c4c8618654e82178b05848328b"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-09-12 10:46:38.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4ae2e9bc36ce69e83486ca48c2c4d5498879646a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4ae2e9bc36ce69e83486ca48c2c4d5498879646a"}]},"branch":"refs/heads/master"},"c07f2bb39b7026570f88fd7ba1f3528080d46522":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-09-20 11:14:42.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/6"}}},"commit":{"parents":[{"commit":"4abb0c28e05d24ea5e340e2d2981753d1186ba44","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4abb0c28e05d24ea5e340e2d2981753d1186ba44"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-09-20 11:14:42.000000000","tz":0},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c07f2bb39b7026570f88fd7ba1f3528080d46522"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c07f2bb39b7026570f88fd7ba1f3528080d46522"}]},"branch":"refs/heads/master"},"ff6d1a0de083f8dce0bc4912828dad153b524032":{"kind":"REWORK","_number":7,"created":"2019-10-22 13:14:40.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/7"}}},"commit":{"parents":[{"commit":"6f4500ff067eaa02dd03dcbb2ff85791f82a454b","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f4500ff067eaa02dd03dcbb2ff85791f82a454b"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-10-22 12:49:19.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff6d1a0de083f8dce0bc4912828dad153b524032"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff6d1a0de083f8dce0bc4912828dad153b524032"}]},"branch":"refs/heads/master"},"9c52da52b93f742f408aadfff67b1721247b920e":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-10-23 12:44:34.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/8"}}},"commit":{"parents":[{"commit":"a1c3161dd6d6e3cee0e249e269959344a671a282","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1c3161dd6d6e3cee0e249e269959344a671a282"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-10-23 12:44:34.000000000","tz":0},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c52da52b93f742f408aadfff67b1721247b920e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c52da52b93f742f408aadfff67b1721247b920e"}]},"branch":"refs/heads/master"},"16ea9fd98033764336ba6de68757d6fa34aa09a2":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-10-25 10:55:13.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/9"}}},"commit":{"parents":[{"commit":"c045baa89020dddc6a66e28761b6b3a707ceb1cc","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c045baa89020dddc6a66e28761b6b3a707ceb1cc"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-10-25 10:55:13.000000000","tz":0},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16ea9fd98033764336ba6de68757d6fa34aa09a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16ea9fd98033764336ba6de68757d6fa34aa09a2"}]},"branch":"refs/heads/master"},"b6509e5870badc2e0da7d5ea43a849b4243d85a0":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-10-28 12:34:55.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/10"}}},"commit":{"parents":[{"commit":"069724ca2199062cc201de4d7cac35ea2b3876c7","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/069724ca2199062cc201de4d7cac35ea2b3876c7"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-10-28 12:34:04.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b6509e5870badc2e0da7d5ea43a849b4243d85a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b6509e5870badc2e0da7d5ea43a849b4243d85a0"}]},"branch":"refs/heads/master"},"4156bed77105f6e60d0ff6c2b01d1f77ced9fe22":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-10-29 07:17:47.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/11"}}},"commit":{"parents":[{"commit":"e9732098bdf8e8a73b9cfd35bf9be1adfee66019","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9732098bdf8e8a73b9cfd35bf9be1adfee66019"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-10-29 07:11:21.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4156bed77105f6e60d0ff6c2b01d1f77ced9fe22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4156bed77105f6e60d0ff6c2b01d1f77ced9fe22"}]},"branch":"refs/heads/master"},"b852c92c9e60df502d300eec51b4db2b9fad0075":{"kind":"REWORK","_number":12,"created":"2019-10-31 12:16:13.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/12"}}},"commit":{"parents":[{"commit":"42694e5375495cf63154beefeeec54b083b5ca80","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42694e5375495cf63154beefeeec54b083b5ca80"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-10-31 12:11:52.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b852c92c9e60df502d300eec51b4db2b9fad0075"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b852c92c9e60df502d300eec51b4db2b9fad0075"}]},"branch":"refs/heads/master"},"fddc672fd6cd7fe23071dd6d94526fde01586215":{"kind":"REWORK","_number":13,"created":"2019-11-06 12:04:29.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/13"}}},"commit":{"parents":[{"commit":"4e5101205fbc7d6b931ae0e32f669e8cb1c6c53b","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4e5101205fbc7d6b931ae0e32f669e8cb1c6c53b"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-11-06 12:04:09.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fddc672fd6cd7fe23071dd6d94526fde01586215"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fddc672fd6cd7fe23071dd6d94526fde01586215"}]},"branch":"refs/heads/master"},"87e5d5f7b99b44a255a9c088141ba6e02f24bd52":{"kind":"REWORK","_number":14,"created":"2019-11-07 07:53:55.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/14"}}},"commit":{"parents":[{"commit":"bed62c75fd9b3116c5b3fd10e366050c54b75407","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bed62c75fd9b3116c5b3fd10e366050c54b75407"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-11-07 07:53:31.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/87e5d5f7b99b44a255a9c088141ba6e02f24bd52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/87e5d5f7b99b44a255a9c088141ba6e02f24bd52"}]},"branch":"refs/heads/master"},"e4df0f8b45b0b4539b7ca81abfcb6b82ed277197":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-11-08 06:22:45.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/15"}}},"commit":{"parents":[{"commit":"3f6e39b12aec9f61b1cf142214a89ab620b48cb0","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f6e39b12aec9f61b1cf142214a89ab620b48cb0"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-11-08 06:22:45.000000000","tz":0},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4df0f8b45b0b4539b7ca81abfcb6b82ed277197"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4df0f8b45b0b4539b7ca81abfcb6b82ed277197"}]},"branch":"refs/heads/master"},"e64c845eb2dbe0953c706dbe6ad18fd6896306ed":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2019-11-18 20:29:13.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/16"}}},"commit":{"parents":[{"commit":"8cf5d59fa6f4b18465dda1b505a7ff6292e1a1b4","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8cf5d59fa6f4b18465dda1b505a7ff6292e1a1b4"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-11-18 20:27:41.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e64c845eb2dbe0953c706dbe6ad18fd6896306ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e64c845eb2dbe0953c706dbe6ad18fd6896306ed"}]},"branch":"refs/heads/master"},"de326e03dff5221f789bbcdd583b203eeb3218b1":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2019-11-27 17:52:07.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/17"}}},"commit":{"parents":[{"commit":"89d1d4339b4b2fcb0fe82269bf30550f9cfe3380","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89d1d4339b4b2fcb0fe82269bf30550f9cfe3380"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-11-27 17:20:14.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de326e03dff5221f789bbcdd583b203eeb3218b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de326e03dff5221f789bbcdd583b203eeb3218b1"}]},"branch":"refs/heads/master"},"548aa68a058b5bd28cb94ffd5dc928355169f682":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2019-11-28 17:33:25.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/18"}}},"commit":{"parents":[{"commit":"d2886fc162e9a38a16c6442205c0694f9fe74b17","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2886fc162e9a38a16c6442205c0694f9fe74b17"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-11-28 17:32:54.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/548aa68a058b5bd28cb94ffd5dc928355169f682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/548aa68a058b5bd28cb94ffd5dc928355169f682"}]},"branch":"refs/heads/master"},"355e2c025594f9cbbb91c70d8bf359bad6a95789":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2019-11-29 02:55:37.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/19"}}},"commit":{"parents":[{"commit":"5dfe76afe538688239653d4fca37552606ab0d85","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5dfe76afe538688239653d4fca37552606ab0d85"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-11-29 02:55:01.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/355e2c025594f9cbbb91c70d8bf359bad6a95789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/355e2c025594f9cbbb91c70d8bf359bad6a95789"}]},"branch":"refs/heads/master"},"5ef096067b2e35081cad3f8fd8ac22c3b7f044a5":{"kind":"REWORK","_number":20,"created":"2019-12-03 04:38:25.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/20"}}},"commit":{"parents":[{"commit":"7525f366889f6e780948d8b8b1675e1d8e908a87","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7525f366889f6e780948d8b8b1675e1d8e908a87"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-12-03 04:37:36.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ef096067b2e35081cad3f8fd8ac22c3b7f044a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ef096067b2e35081cad3f8fd8ac22c3b7f044a5"}]},"branch":"refs/heads/master"},"f81b003b9fd19210e56485e249201c3d5f3a9dfc":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2019-12-04 01:07:59.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/21"}}},"commit":{"parents":[{"commit":"8ed4510018d5d723a43f979faa6f7d3209c6d078","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ed4510018d5d723a43f979faa6f7d3209c6d078"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-12-04 01:06:37.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f81b003b9fd19210e56485e249201c3d5f3a9dfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f81b003b9fd19210e56485e249201c3d5f3a9dfc"}]},"branch":"refs/heads/master"},"8a38f9ca501b8a43b0774c1aa7a812e45b609ce8":{"kind":"REWORK","_number":22,"created":"2019-12-05 01:42:31.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/22"}}},"commit":{"parents":[{"commit":"5c7a41c444ce492811cd3991396d97c31da70856","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c7a41c444ce492811cd3991396d97c31da70856"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-12-05 01:42:21.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a38f9ca501b8a43b0774c1aa7a812e45b609ce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a38f9ca501b8a43b0774c1aa7a812e45b609ce8"}]},"branch":"refs/heads/master"},"e4580492089adf42ebeeaf8fffb2b2f49ef35702":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2019-12-12 04:14:10.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/23"}}},"commit":{"parents":[{"commit":"18ee790e60cacc8fb9ab6372854aaf91f9d3d9c6","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/18ee790e60cacc8fb9ab6372854aaf91f9d3d9c6"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-12-12 04:13:30.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4580492089adf42ebeeaf8fffb2b2f49ef35702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4580492089adf42ebeeaf8fffb2b2f49ef35702"}]},"branch":"refs/heads/master"},"35f71667e0441ab13a2d1c8b4046fa668a033bce":{"kind":"REWORK","_number":24,"created":"2019-12-12 12:34:23.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/24"}}},"commit":{"parents":[{"commit":"1c64df42d87d16ce7b751f8d5c280b8bc09ec1f4","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c64df42d87d16ce7b751f8d5c280b8bc09ec1f4"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-12-12 12:34:01.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35f71667e0441ab13a2d1c8b4046fa668a033bce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35f71667e0441ab13a2d1c8b4046fa668a033bce"}]},"branch":"refs/heads/master"},"490171fb6043230e2d3b2b10c23cae63dffced02":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2019-12-24 12:23:43.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/25"}}},"commit":{"parents":[{"commit":"40fcc6b95e58e3f889af5ddf64c18a17789d7324","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40fcc6b95e58e3f889af5ddf64c18a17789d7324"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-12-24 12:01:01.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/490171fb6043230e2d3b2b10c23cae63dffced02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/490171fb6043230e2d3b2b10c23cae63dffced02"}]},"branch":"refs/heads/master"},"5073afc1d7b807b876d4b25b97df6b7bf7e2ef1a":{"kind":"REWORK","_number":26,"created":"2019-12-27 14:38:36.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/26"}}},"commit":{"parents":[{"commit":"acb472a98e6ab62b8a03008bddd2b95958fdc8b1","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/acb472a98e6ab62b8a03008bddd2b95958fdc8b1"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-12-27 14:38:10.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5073afc1d7b807b876d4b25b97df6b7bf7e2ef1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5073afc1d7b807b876d4b25b97df6b7bf7e2ef1a"}]},"branch":"refs/heads/master"},"1ebb62f4c12ea17382d97cfb7913652258b884b7":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2020-01-09 10:43:39.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/27"}}},"commit":{"parents":[{"commit":"e93ba1885b4873c7be3b26efa6410f9d5453fab7","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e93ba1885b4873c7be3b26efa6410f9d5453fab7"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-01-09 08:30:39.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ebb62f4c12ea17382d97cfb7913652258b884b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ebb62f4c12ea17382d97cfb7913652258b884b7"}]},"branch":"refs/heads/master"},"79a583c6a5fda6beba45acd0de5deeea8fbadb89":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2020-01-17 08:27:47.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/28"}}},"commit":{"parents":[{"commit":"b61011fdf9c9318f57ab2120fc2a64f97de0cb20","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b61011fdf9c9318f57ab2120fc2a64f97de0cb20"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-01-17 08:27:27.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79a583c6a5fda6beba45acd0de5deeea8fbadb89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79a583c6a5fda6beba45acd0de5deeea8fbadb89"}]},"branch":"refs/heads/master"},"ea6b065cb0e82b2949eed008121edded56ea3902":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2020-01-24 13:02:06.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/73/668173/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/668173/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/668173/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/668173/29"}}},"commit":{"parents":[{"commit":"7d310c3985c3e0051f2e55cac21300774a1c6bdf","subject":"Transition l7policy flows to dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d310c3985c3e0051f2e55cac21300774a1c6bdf"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2019-06-28 13:13:13.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-01-24 12:58:07.000000000","tz":240},"subject":"Transition l7rule flows to dicts","message":"Transition l7rule flows to dicts\n\nChange-Id: Id94a293a88d27546c0fe7b45367922fa4cca260a\nStory: 2005072\nTask: 30813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea6b065cb0e82b2949eed008121edded56ea3902"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea6b065cb0e82b2949eed008121edded56ea3902"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
