)]}'
{"id":"openstack%2Foctavia~668216","triplet_id":"openstack%2Foctavia~stable%2Fstein~I9e0debcbfe6ceeff0012c827d70d80d938b5a2fb","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I9e0debcbfe6ceeff0012c827d70d80d938b5a2fb","subject":"Fix cryptsetup --pbkdf-memory failures","status":"MERGED","created":"2019-06-28 17:23:25.000000000","updated":"2019-07-05 19:27:00.000000000","submitted":"2019-07-05 19:26:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668216-1562354820122-0a1ee91c","meta_rev_id":"d3373f3892faa67178c8dcffa8922c19dde45a63","_number":668216,"virtual_id_number":668216,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2019-07-05 19:26:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-07-05 17:06:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-07-05 17:31:45.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-07-05 17:31:45.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-07-05 17:31:45.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 17:06:16.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-07-05 17:31:45.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2019-07-05 19:26:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"146fae0df1c24a60821cabbd49a750c40454726c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-28 17:23:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e6653f9b87f16f56611d5f14415050a04a2e1ec0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 19:55:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/16/668216/1/check/openstack-tox-cover/ef3f463/cover/ : SUCCESS in 7m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/668216/1/check/openstack-tox-lower-constraints/2f9830f/ : SUCCESS in 6m 27s\n- openstack-tox-pep8 http://logs.openstack.org/16/668216/1/check/openstack-tox-pep8/755425a/ : SUCCESS in 5m 24s\n- openstack-tox-py27 http://logs.openstack.org/16/668216/1/check/openstack-tox-py27/2c2de58/ : SUCCESS in 3m 56s\n- openstack-tox-py35 http://logs.openstack.org/16/668216/1/check/openstack-tox-py35/9dbf18b/ : SUCCESS in 4m 53s\n- openstack-tox-py36 http://logs.openstack.org/16/668216/1/check/openstack-tox-py36/305ccd3/ : SUCCESS in 4m 00s\n- openstack-tox-docs http://logs.openstack.org/16/668216/1/check/openstack-tox-docs/58e032c/html/ : SUCCESS in 5m 06s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/16/668216/1/check/neutron-lbaasv2-dsvm-api/c00a861/ : SUCCESS in 1h 29m 27s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/16/668216/1/check/neutron-lbaasv2-dsvm-py3x-api/47271ed/ : SUCCESS in 1h 26m 49s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/16/668216/1/check/neutron-lbaasv2-dsvm-scenario/664b7b0/ : FAILURE in 1h 20m 16s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/16/668216/1/check/neutron-lbaasv2-dsvm-py3x-scenario/fcb5cf6/ : FAILURE in 1h 32m 40s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/16/668216/1/check/openstack-tox-functional/b5c2fd8/ : SUCCESS in 6m 26s\n- openstack-tox-functional-py35 http://logs.openstack.org/16/668216/1/check/openstack-tox-functional-py35/727c7ae/ : SUCCESS in 6m 40s\n- openstack-tox-functional-py36 http://logs.openstack.org/16/668216/1/check/openstack-tox-functional-py36/47ea398/ : SUCCESS in 5m 17s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/16/668216/1/check/octavia-v1-dsvm-scenario/7d08c23/ : SUCCESS in 1h 05m 29s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/16/668216/1/check/octavia-v1-dsvm-py3x-scenario/a73d249/ : SUCCESS in 1h 09m 51s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/16/668216/1/check/octavia-v2-dsvm-noop-api/fc500e1/ : SUCCESS in 42m 56s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/16/668216/1/check/octavia-v2-dsvm-noop-py2-api/991e588/ : SUCCESS in 56m 29s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/16/668216/1/check/octavia-v2-dsvm-scenario/7453270/ : SUCCESS in 1h 32m 35s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/16/668216/1/check/octavia-v2-dsvm-py2-scenario/8a0b289/ : SUCCESS in 1h 37m 57s\n- octavia-v2-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/16/668216/1/check/octavia-v2-dsvm-scenario-ubuntu-xenial/90b79d0/ : SUCCESS in 1h 38m 20s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/16/668216/1/check/octavia-v2-dsvm-py2-scenario-centos-7/2cda56f/ : FAILURE in 2h 00m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/16/668216/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/468bb2f/ : SUCCESS in 1h 44m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/16/668216/1/check/octavia-v2-dsvm-scenario-two-node/dcd1aab/ : SUCCESS in 1h 48m 57s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/16/668216/1/check/octavia-v2-dsvm-py2-scenario-two-node/6db32d2/ : FAILURE in 1h 28m 16s (non-voting)\n- octavia-grenade http://logs.openstack.org/16/668216/1/check/octavia-grenade/d7d9937/ : SUCCESS in 1h 14m 44s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/16/668216/1/check/octavia-v2-dsvm-tls-barbican/37c54c9/ : SUCCESS in 54m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"33108d7455d20cab910b4d0ffd0c065db1c44d56","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-28 21:55:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"cff6e40a00f1dd53308a765add86df53b125ed7f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-28 21:56:20.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"38f63b53c8e9b3aa8dbca237264343176dc901b7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-28 21:56:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cc204e4ec31d1717e4c3f321085f943d356b2a19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 00:30:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/16/668216/4/check/openstack-tox-cover/fb945c3/cover/ : SUCCESS in 7m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/668216/4/check/openstack-tox-lower-constraints/5299d43/ : SUCCESS in 5m 44s\n- openstack-tox-pep8 http://logs.openstack.org/16/668216/4/check/openstack-tox-pep8/90d3032/ : SUCCESS in 5m 13s\n- openstack-tox-py27 http://logs.openstack.org/16/668216/4/check/openstack-tox-py27/2d4447f/ : SUCCESS in 4m 20s\n- openstack-tox-py35 http://logs.openstack.org/16/668216/4/check/openstack-tox-py35/9833a06/ : SUCCESS in 4m 10s\n- openstack-tox-py36 http://logs.openstack.org/16/668216/4/check/openstack-tox-py36/1de2e7f/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/16/668216/4/check/openstack-tox-docs/dc36f0d/html/ : SUCCESS in 4m 37s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/16/668216/4/check/neutron-lbaasv2-dsvm-api/6b7b9fc/ : SUCCESS in 1h 16m 10s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/16/668216/4/check/neutron-lbaasv2-dsvm-py3x-api/03afb00/ : SUCCESS in 1h 19m 55s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/16/668216/4/check/neutron-lbaasv2-dsvm-scenario/eb68abb/ : SUCCESS in 1h 25m 21s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/16/668216/4/check/neutron-lbaasv2-dsvm-py3x-scenario/d1cfa63/ : SUCCESS in 1h 22m 36s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/16/668216/4/check/openstack-tox-functional/dac287a/ : SUCCESS in 5m 25s\n- openstack-tox-functional-py35 http://logs.openstack.org/16/668216/4/check/openstack-tox-functional-py35/b36f104/ : SUCCESS in 6m 13s\n- openstack-tox-functional-py36 http://logs.openstack.org/16/668216/4/check/openstack-tox-functional-py36/fc655b4/ : SUCCESS in 5m 44s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/16/668216/4/check/octavia-v1-dsvm-scenario/f692715/ : SUCCESS in 1h 17m 54s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/16/668216/4/check/octavia-v1-dsvm-py3x-scenario/d014a44/ : FAILURE in 57m 49s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-noop-api/f3a8e24/ : SUCCESS in 44m 09s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-noop-py2-api/b7bfc2e/ : SUCCESS in 48m 56s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-scenario/ea2a7e4/ : SUCCESS in 1h 33m 56s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-py2-scenario/e99ee82/ : SUCCESS in 1h 39m 32s\n- octavia-v2-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-scenario-ubuntu-xenial/805a922/ : FAILURE in 2h 02m 03s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-py2-scenario-centos-7/d151e14/ : POST_FAILURE in 2h 12m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-scenario-ubuntu-bionic/068182a/ : SUCCESS in 1h 33m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-scenario-two-node/ff9456f/ : SUCCESS in 1h 18m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-py2-scenario-two-node/c3cf868/ : SUCCESS in 2h 23m 47s (non-voting)\n- octavia-grenade http://logs.openstack.org/16/668216/4/check/octavia-grenade/a48a19f/ : SUCCESS in 1h 19m 59s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-tls-barbican/770bdac/ : SUCCESS in 43m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ca5d915f04268f79d0b6565a0a4152085e35b92f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-29 00:59:38.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ec0b7dc4a2f61ca68f00de512d375fbda5008f14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 03:11:59.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/16/668216/4/check/openstack-tox-cover/e55058b/cover/ : SUCCESS in 7m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/668216/4/check/openstack-tox-lower-constraints/865109e/ : SUCCESS in 5m 49s\n- openstack-tox-pep8 http://logs.openstack.org/16/668216/4/check/openstack-tox-pep8/991ff89/ : SUCCESS in 5m 28s\n- openstack-tox-py27 http://logs.openstack.org/16/668216/4/check/openstack-tox-py27/06dcdbd/ : SUCCESS in 3m 50s\n- openstack-tox-py35 http://logs.openstack.org/16/668216/4/check/openstack-tox-py35/b14fab4/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/16/668216/4/check/openstack-tox-py36/5b07ab8/ : SUCCESS in 4m 08s\n- openstack-tox-docs http://logs.openstack.org/16/668216/4/check/openstack-tox-docs/f93de33/html/ : SUCCESS in 4m 51s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/16/668216/4/check/neutron-lbaasv2-dsvm-api/77c70bb/ : SUCCESS in 1h 16m 14s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/16/668216/4/check/neutron-lbaasv2-dsvm-py3x-api/3cf3830/ : SUCCESS in 1h 15m 41s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/16/668216/4/check/neutron-lbaasv2-dsvm-scenario/9c88201/ : FAILURE in 1h 15m 34s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/16/668216/4/check/neutron-lbaasv2-dsvm-py3x-scenario/58f2926/ : SUCCESS in 1h 30m 45s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/16/668216/4/check/openstack-tox-functional/901ade7/ : SUCCESS in 5m 22s\n- openstack-tox-functional-py35 http://logs.openstack.org/16/668216/4/check/openstack-tox-functional-py35/5dd54cf/ : SUCCESS in 6m 54s\n- openstack-tox-functional-py36 http://logs.openstack.org/16/668216/4/check/openstack-tox-functional-py36/543a56e/ : SUCCESS in 5m 45s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/16/668216/4/check/octavia-v1-dsvm-scenario/ef629ff/ : FAILURE in 56m 04s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/16/668216/4/check/octavia-v1-dsvm-py3x-scenario/c075219/ : SUCCESS in 1h 04m 17s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-noop-api/491eb7b/ : SUCCESS in 42m 37s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-noop-py2-api/3a2b77c/ : SUCCESS in 42m 14s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-scenario/8a25ea2/ : SUCCESS in 1h 31m 19s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-py2-scenario/c82f661/ : SUCCESS in 1h 33m 48s\n- octavia-v2-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-scenario-ubuntu-xenial/5746006/ : SUCCESS in 1h 34m 42s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-py2-scenario-centos-7/9fe22b2/ : FAILURE in 2h 06m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-scenario-ubuntu-bionic/2d59760/ : SUCCESS in 1h 21m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-scenario-two-node/df52ce2/ : SUCCESS in 1h 40m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-py2-scenario-two-node/32be399/ : FAILURE in 1h 37m 24s (non-voting)\n- octavia-grenade http://logs.openstack.org/16/668216/4/check/octavia-grenade/465dc7c/ : SUCCESS in 1h 08m 01s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-tls-barbican/ccb4d43/ : SUCCESS in 51m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"052649b87307a24246f07b8edee842fa1c9455b8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-01 23:45:56.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"cb7679bce5801ea5547af1893f00838ca411c875","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 01:59:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/16/668216/4/check/openstack-tox-cover/549da78/cover/ : SUCCESS in 6m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/668216/4/check/openstack-tox-lower-constraints/d207d2d/ : SUCCESS in 5m 34s\n- openstack-tox-pep8 http://logs.openstack.org/16/668216/4/check/openstack-tox-pep8/08298f8/ : SUCCESS in 6m 02s\n- openstack-tox-py27 http://logs.openstack.org/16/668216/4/check/openstack-tox-py27/76dfc80/ : SUCCESS in 4m 01s\n- openstack-tox-py35 http://logs.openstack.org/16/668216/4/check/openstack-tox-py35/ab4ea37/ : SUCCESS in 6m 57s\n- openstack-tox-py36 http://logs.openstack.org/16/668216/4/check/openstack-tox-py36/d2b1b0d/ : SUCCESS in 4m 01s\n- openstack-tox-docs http://logs.openstack.org/16/668216/4/check/openstack-tox-docs/5457f68/html/ : SUCCESS in 6m 25s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/16/668216/4/check/neutron-lbaasv2-dsvm-api/e7e0311/ : SUCCESS in 1h 26m 43s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/16/668216/4/check/neutron-lbaasv2-dsvm-py3x-api/9c5e78a/ : SUCCESS in 1h 22m 38s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/16/668216/4/check/neutron-lbaasv2-dsvm-scenario/a9b0ff7/ : SUCCESS in 1h 28m 21s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/16/668216/4/check/neutron-lbaasv2-dsvm-py3x-scenario/7acd538/ : SUCCESS in 1h 37m 28s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/16/668216/4/check/openstack-tox-functional/e9118a7/ : SUCCESS in 6m 16s\n- openstack-tox-functional-py35 http://logs.openstack.org/16/668216/4/check/openstack-tox-functional-py35/194acea/ : SUCCESS in 8m 09s\n- openstack-tox-functional-py36 http://logs.openstack.org/16/668216/4/check/openstack-tox-functional-py36/e91d2d9/ : SUCCESS in 5m 47s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/16/668216/4/check/octavia-v1-dsvm-scenario/c505498/ : SUCCESS in 1h 02m 50s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/16/668216/4/check/octavia-v1-dsvm-py3x-scenario/57b0ed9/ : SUCCESS in 1h 01m 50s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-noop-api/4ca37ca/ : SUCCESS in 50m 51s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-noop-py2-api/88adb6a/ : SUCCESS in 55m 51s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-scenario/752fb35/ : SUCCESS in 1h 36m 23s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-py2-scenario/d3df8e4/ : SUCCESS in 1h 46m 08s\n- octavia-v2-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-scenario-ubuntu-xenial/44ca644/ : SUCCESS in 1h 34m 58s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-py2-scenario-centos-7/1264241/ : FAILURE in 2h 05m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-scenario-ubuntu-bionic/b4796f2/ : SUCCESS in 1h 38m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-scenario-two-node/291371d/ : SUCCESS in 1h 51m 06s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-py2-scenario-two-node/bbaf735/ : SUCCESS in 1h 52m 29s (non-voting)\n- octavia-grenade http://logs.openstack.org/16/668216/4/check/octavia-grenade/cb30d7c/ : SUCCESS in 1h 27m 51s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/16/668216/4/check/octavia-v2-dsvm-tls-barbican/2af7766/ : SUCCESS in 55m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5f9bee0dd4c43126ef47d4d2155effe15dd44042","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-03 16:25:28.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"18000a129ca78588d8233a20aad14d56fc3b1415","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 18:37:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/16/668216/5/check/openstack-tox-cover/cec6648/cover/ : SUCCESS in 8m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/668216/5/check/openstack-tox-lower-constraints/56dc802/ : SUCCESS in 6m 59s\n- openstack-tox-pep8 http://logs.openstack.org/16/668216/5/check/openstack-tox-pep8/51b9bdc/ : SUCCESS in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/16/668216/5/check/openstack-tox-py27/014c9f3/ : SUCCESS in 4m 35s\n- openstack-tox-py35 http://logs.openstack.org/16/668216/5/check/openstack-tox-py35/67a0fe0/ : SUCCESS in 5m 18s\n- openstack-tox-py36 http://logs.openstack.org/16/668216/5/check/openstack-tox-py36/9a4852d/ : SUCCESS in 4m 17s\n- openstack-tox-docs http://logs.openstack.org/16/668216/5/check/openstack-tox-docs/83784cc/html/ : SUCCESS in 9m 35s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/16/668216/5/check/neutron-lbaasv2-dsvm-api/aee5a52/ : SUCCESS in 1h 22m 59s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/16/668216/5/check/neutron-lbaasv2-dsvm-py3x-api/e166bb5/ : SUCCESS in 1h 23m 00s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/16/668216/5/check/neutron-lbaasv2-dsvm-scenario/e4c255d/ : SUCCESS in 1h 29m 42s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/16/668216/5/check/neutron-lbaasv2-dsvm-py3x-scenario/57c3e43/ : SUCCESS in 1h 30m 17s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/16/668216/5/check/openstack-tox-functional/6f32add/ : SUCCESS in 8m 12s\n- openstack-tox-functional-py35 http://logs.openstack.org/16/668216/5/check/openstack-tox-functional-py35/8a007a8/ : SUCCESS in 10m 06s\n- openstack-tox-functional-py36 http://logs.openstack.org/16/668216/5/check/openstack-tox-functional-py36/d8e2a0d/ : SUCCESS in 8m 29s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/16/668216/5/check/octavia-v1-dsvm-scenario/ec6afa8/ : SUCCESS in 59m 50s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/16/668216/5/check/octavia-v1-dsvm-py3x-scenario/fc6213e/ : SUCCESS in 1h 03m 18s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/16/668216/5/check/octavia-v2-dsvm-noop-api/011a1a9/ : SUCCESS in 41m 19s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/16/668216/5/check/octavia-v2-dsvm-noop-py2-api/6c7aff4/ : SUCCESS in 46m 15s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/16/668216/5/check/octavia-v2-dsvm-scenario/e51ff8d/ : SUCCESS in 1h 42m 33s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/16/668216/5/check/octavia-v2-dsvm-py2-scenario/1face43/ : SUCCESS in 1h 41m 21s\n- octavia-v2-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/16/668216/5/check/octavia-v2-dsvm-scenario-ubuntu-xenial/9eea34e/ : SUCCESS in 1h 26m 48s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/16/668216/5/check/octavia-v2-dsvm-py2-scenario-centos-7/b1a18bb/ : FAILURE in 2h 01m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/16/668216/5/check/octavia-v2-dsvm-scenario-ubuntu-bionic/3bcaece/ : FAILURE in 42m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/16/668216/5/check/octavia-v2-dsvm-scenario-two-node/3338b3a/ : SUCCESS in 1h 58m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/16/668216/5/check/octavia-v2-dsvm-py2-scenario-two-node/a2b6fd1/ : SUCCESS in 1h 55m 33s (non-voting)\n- octavia-grenade http://logs.openstack.org/16/668216/5/check/octavia-grenade/446ba27/ : SUCCESS in 1h 05m 16s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/16/668216/5/check/octavia-v2-dsvm-tls-barbican/10a1fa5/ : SUCCESS in 47m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"46828bb57219d89717eae508660a6db33783619e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-05 17:06:16.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"36fc7acc059104f5b0442d7cd9886567b25aa9a6","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2019-07-05 17:31:45.000000000","message":"Patch Set 5: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"93ba37a041f05972d5c35de2b05035103a893fcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 17:31:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"db4540e64ec8e30e58d7c480892b7ce326fbe6bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 19:26:59.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/16/668216/5/gate/openstack-tox-lower-constraints/58d06e6/ : SUCCESS in 6m 14s\n- openstack-tox-pep8 http://logs.openstack.org/16/668216/5/gate/openstack-tox-pep8/effa3f4/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/16/668216/5/gate/openstack-tox-py27/0398818/ : SUCCESS in 3m 42s\n- openstack-tox-py35 http://logs.openstack.org/16/668216/5/gate/openstack-tox-py35/7769a4d/ : SUCCESS in 5m 28s\n- openstack-tox-py36 http://logs.openstack.org/16/668216/5/gate/openstack-tox-py36/c3c30d5/ : SUCCESS in 4m 18s\n- openstack-tox-docs http://logs.openstack.org/16/668216/5/gate/openstack-tox-docs/a266ead/html/ : SUCCESS in 5m 24s\n- openstack-tox-functional http://logs.openstack.org/16/668216/5/gate/openstack-tox-functional/18bdb20/ : SUCCESS in 5m 37s\n- openstack-tox-functional-py35 http://logs.openstack.org/16/668216/5/gate/openstack-tox-functional-py35/ee32d7d/ : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 http://logs.openstack.org/16/668216/5/gate/openstack-tox-functional-py36/0e4a698/ : SUCCESS in 5m 37s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/16/668216/5/gate/octavia-v1-dsvm-scenario/d31461d/ : SUCCESS in 58m 52s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/16/668216/5/gate/octavia-v1-dsvm-py3x-scenario/96f9ce3/ : SUCCESS in 1h 08m 38s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/16/668216/5/gate/neutron-lbaasv2-dsvm-api/59ab77c/ : SUCCESS in 1h 14m 00s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/16/668216/5/gate/neutron-lbaasv2-dsvm-py3x-api/4a539a5/ : SUCCESS in 1h 17m 19s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/16/668216/5/gate/octavia-v2-dsvm-noop-api/c7fb531/ : SUCCESS in 43m 30s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/16/668216/5/gate/octavia-v2-dsvm-noop-py2-api/6f83e07/ : SUCCESS in 43m 43s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/16/668216/5/gate/octavia-v2-dsvm-scenario/49a3354/ : SUCCESS in 1h 41m 39s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/16/668216/5/gate/octavia-v2-dsvm-py2-scenario/ebbb651/ : SUCCESS in 1h 28m 45s\n- octavia-v2-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/16/668216/5/gate/octavia-v2-dsvm-scenario-ubuntu-xenial/e158fe8/ : SUCCESS in 1h 49m 25s\n- octavia-grenade http://logs.openstack.org/16/668216/5/gate/octavia-grenade/488ab47/ : SUCCESS in 1h 17m 43s","accounts_in_message":[],"_revision_number":5},{"id":"d3373f3892faa67178c8dcffa8922c19dde45a63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 19:27:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3053676faed4bdbdba15e776bfccade35baa5631","revisions":{"9f5bc25dc26621d6c605d18939592d4cd5a927d0":{"kind":"REWORK","_number":1,"created":"2019-06-28 17:23:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/16/668216/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/668216/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/668216/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/668216/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/668216/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/668216/1"}}},"commit":{"parents":[{"commit":"d53010dc423867bc965069206b007a62fbb02475","subject":"Add bindep.txt for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d53010dc423867bc965069206b007a62fbb02475"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-28 17:18:10.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-28 17:23:25.000000000","tz":0},"subject":"Fix cryptsetup --pbkdf-memory failures","message":"Fix cryptsetup --pbkdf-memory failures\n\nA recent patch[1] added --pbkdf-memory to the cryptsetup command line\nto limit the memory cryptsetup is using. However, some distros use\nan older version of cryptsetup that does not need this setting.\nThis patch adds logic to detect this and run the commands without\n--pbkdf-memory.\n\n[1] https://review.opendev.org/663784\n\nChange-Id: I9e0debcbfe6ceeff0012c827d70d80d938b5a2fb\nStory: 2006066\nTask: 34782\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f5bc25dc26621d6c605d18939592d4cd5a927d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f5bc25dc26621d6c605d18939592d4cd5a927d0"}]},"branch":"refs/heads/stable/stein"},"652bbd90e48188a13c72afe7447fbcb32a6230ff":{"kind":"REWORK","_number":4,"created":"2019-06-28 21:56:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/16/668216/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/668216/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/668216/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/668216/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/668216/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/668216/4"}}},"commit":{"parents":[{"commit":"d53010dc423867bc965069206b007a62fbb02475","subject":"Add bindep.txt for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d53010dc423867bc965069206b007a62fbb02475"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-28 17:18:10.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-28 21:56:48.000000000","tz":0},"subject":"Fix cryptsetup --pbkdf-memory failures","message":"Fix cryptsetup --pbkdf-memory failures\n\nA recent patch[1] added --pbkdf-memory to the cryptsetup command line\nto limit the memory cryptsetup is using. However, some distros use\nan older version of cryptsetup that does not need this setting.\nThis patch adds logic to detect this and run the commands without\n--pbkdf-memory.\n\n[1] https://review.opendev.org/663784\n\nChange-Id: I9e0debcbfe6ceeff0012c827d70d80d938b5a2fb\nStory: 2006066\nTask: 34782\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/652bbd90e48188a13c72afe7447fbcb32a6230ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/652bbd90e48188a13c72afe7447fbcb32a6230ff"}]},"branch":"refs/heads/stable/stein"},"3053676faed4bdbdba15e776bfccade35baa5631":{"kind":"REWORK","_number":5,"created":"2019-07-03 16:25:28.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/16/668216/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/668216/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/668216/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/668216/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/668216/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/668216/5"}}},"commit":{"parents":[{"commit":"d53010dc423867bc965069206b007a62fbb02475","subject":"Add bindep.txt for Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d53010dc423867bc965069206b007a62fbb02475"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-28 17:18:10.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-03 16:25:26.000000000","tz":0},"subject":"Fix cryptsetup --pbkdf-memory failures","message":"Fix cryptsetup --pbkdf-memory failures\n\nA recent patch[1] added --pbkdf-memory to the cryptsetup command line\nto limit the memory cryptsetup is using. However, some distros use\nan older version of cryptsetup that does not need this setting.\nThis patch adds logic to detect this and run the commands without\n--pbkdf-memory.\n\n[1] https://review.opendev.org/663784\n\nChange-Id: I9e0debcbfe6ceeff0012c827d70d80d938b5a2fb\nStory: 2006066\nTask: 34782\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3053676faed4bdbdba15e776bfccade35baa5631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3053676faed4bdbdba15e776bfccade35baa5631"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
