)]}'
{"id":"openstack%2Foctavia~668258","triplet_id":"openstack%2Foctavia~stable%2Fqueens~I3db53821520e54747697808c7eb6988cc1556513","project":"openstack/octavia","branch":"stable/queens","topic":"constraints-updates","hashtags":[],"change_id":"I3db53821520e54747697808c7eb6988cc1556513","subject":"Update tox.ini for new upper constraints strategy","status":"MERGED","created":"2019-06-28 22:50:49.000000000","updated":"2019-07-08 04:42:56.000000000","submitted":"2019-07-08 04:42:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668258-1562560976074-eaa21e4c","meta_rev_id":"67623941b0ff376a712245b67a99dcaf10ffe60d","_number":668258,"virtual_id_number":668258,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"date":"2019-07-01 23:46:39.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-07-08 04:42:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-07-05 16:52:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-07-08 03:09:09.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-07-08 03:09:09.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 16:52:52.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-07-08 03:09:09.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-07-08 04:42:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d9f797940575d6ce1bad1d49be94371b69069fd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-28 22:50:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62c2613f186d5c042179fab8b76710630601cd7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 00:52:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/58/668258/1/check/openstack-tox-pep8/ff74697/ : SUCCESS in 8m 55s\n- openstack-tox-py27 http://logs.openstack.org/58/668258/1/check/openstack-tox-py27/543e328/ : SUCCESS in 4m 20s\n- openstack-tox-py35 http://logs.openstack.org/58/668258/1/check/openstack-tox-py35/3c748be/ : SUCCESS in 4m 46s\n- openstack-tox-docs http://logs.openstack.org/58/668258/1/check/openstack-tox-docs/034bf5b/html/ : SUCCESS in 5m 04s\n- build-openstack-releasenotes http://logs.openstack.org/58/668258/1/check/build-openstack-releasenotes/d4d2dbb/html/ : SUCCESS in 4m 17s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/58/668258/1/check/neutron-lbaasv2-dsvm-api/13afe11/ : SUCCESS in 1h 12m 54s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/58/668258/1/check/neutron-lbaasv2-dsvm-py3x-api/b5d41c2/ : SUCCESS in 1h 11m 35s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/58/668258/1/check/neutron-lbaasv2-dsvm-scenario/8abb74a/ : FAILURE in 1h 12m 26s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/58/668258/1/check/neutron-lbaasv2-dsvm-py3x-scenario/cca0112/ : SUCCESS in 1h 39m 53s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/58/668258/1/check/openstack-tox-cover/efe3374/cover/ : SUCCESS in 7m 07s\n- openstack-tox-functional http://logs.openstack.org/58/668258/1/check/openstack-tox-functional/cd483bd/ : SUCCESS in 6m 25s\n- openstack-tox-functional-py35 http://logs.openstack.org/58/668258/1/check/openstack-tox-functional-py35/572ec46/ : SUCCESS in 6m 31s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/58/668258/1/check/octavia-v1-dsvm-scenario/1abf79d/ : SUCCESS in 52m 35s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/58/668258/1/check/octavia-v1-dsvm-py3x-scenario/9e96082/ : SUCCESS in 46m 39s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/58/668258/1/check/octavia-v1-dsvm-scenario-multinode/a4671c7/ : SUCCESS in 1h 08m 01s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/58/668258/1/check/octavia-v1-dsvm-py3x-scenario-multinode/11fd9a9/ : SUCCESS in 1h 05m 46s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/58/668258/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/bc31e20/ : SUCCESS in 1h 07m 29s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/58/668258/1/check/octavia-v2-dsvm-noop-api/500f480/ : SUCCESS in 38m 24s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/58/668258/1/check/octavia-v2-dsvm-noop-py35-api/f3df0aa/ : SUCCESS in 41m 02s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/58/668258/1/check/octavia-v2-dsvm-scenario/a14068c/ : FAILURE in 1h 59m 55s\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/58/668258/1/check/octavia-v2-dsvm-py35-scenario/2f35ac7/ : SUCCESS in 1h 35m 01s","accounts_in_message":[],"_revision_number":1},{"id":"20b8fc4c5f1656a7b27e5b9bc3aae5933b2d06c5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-01 23:46:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d1f3d86fa5bd07606ffb6aac8a0f6bdbc28b1651","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 01:41:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/58/668258/1/check/openstack-tox-pep8/654fc87/ : SUCCESS in 9m 56s\n- openstack-tox-py27 http://logs.openstack.org/58/668258/1/check/openstack-tox-py27/4fe490a/ : SUCCESS in 4m 55s\n- openstack-tox-py35 http://logs.openstack.org/58/668258/1/check/openstack-tox-py35/304fedf/ : SUCCESS in 7m 15s\n- openstack-tox-docs http://logs.openstack.org/58/668258/1/check/openstack-tox-docs/d65c658/html/ : SUCCESS in 4m 36s\n- build-openstack-releasenotes http://logs.openstack.org/58/668258/1/check/build-openstack-releasenotes/7adbabe/html/ : SUCCESS in 5m 52s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/58/668258/1/check/neutron-lbaasv2-dsvm-api/c69b0d4/ : SUCCESS in 1h 13m 41s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/58/668258/1/check/neutron-lbaasv2-dsvm-py3x-api/efcb8ba/ : SUCCESS in 1h 18m 43s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/58/668258/1/check/neutron-lbaasv2-dsvm-scenario/1f1541f/ : SUCCESS in 1h 18m 52s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/58/668258/1/check/neutron-lbaasv2-dsvm-py3x-scenario/29a3bc7/ : SUCCESS in 1h 28m 27s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/58/668258/1/check/openstack-tox-cover/d68db1b/cover/ : SUCCESS in 7m 46s\n- openstack-tox-functional http://logs.openstack.org/58/668258/1/check/openstack-tox-functional/ec3abb5/ : SUCCESS in 6m 11s\n- openstack-tox-functional-py35 http://logs.openstack.org/58/668258/1/check/openstack-tox-functional-py35/374b043/ : SUCCESS in 7m 12s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/58/668258/1/check/octavia-v1-dsvm-scenario/43abe69/ : SUCCESS in 57m 31s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/58/668258/1/check/octavia-v1-dsvm-py3x-scenario/a2d50ab/ : SUCCESS in 1h 03m 59s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/58/668258/1/check/octavia-v1-dsvm-scenario-multinode/7c3abc0/ : SUCCESS in 1h 10m 29s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/58/668258/1/check/octavia-v1-dsvm-py3x-scenario-multinode/4e9d5cc/ : SUCCESS in 1h 15m 11s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/58/668258/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/c26e6b3/ : SUCCESS in 1h 10m 30s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/58/668258/1/check/octavia-v2-dsvm-noop-api/05e5251/ : SUCCESS in 48m 47s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/58/668258/1/check/octavia-v2-dsvm-noop-py35-api/833a29b/ : SUCCESS in 44m 35s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/58/668258/1/check/octavia-v2-dsvm-scenario/8cf2300/ : SUCCESS in 1h 31m 01s\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/58/668258/1/check/octavia-v2-dsvm-py35-scenario/3aaeecc/ : SUCCESS in 1h 35m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7501923a2fe8c4e5c4a93df39ae6f6e0baf107b0","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-05 16:52:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a4e7566d1b75568eb1e271e561e2d969947a2968","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-07-08 03:09:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aa5be08d2e6724933a8b9f9bd681000cb290a2f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 03:09:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77e4c9781f0241fab9eb50b1572fbb0326016908","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 04:42:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/58/668258/1/gate/openstack-tox-pep8/0748c58/ : SUCCESS in 8m 12s\n- openstack-tox-py27 http://logs.openstack.org/58/668258/1/gate/openstack-tox-py27/e9451f8/ : SUCCESS in 5m 20s\n- openstack-tox-py35 http://logs.openstack.org/58/668258/1/gate/openstack-tox-py35/4384ffa/ : SUCCESS in 5m 49s\n- openstack-tox-docs http://logs.openstack.org/58/668258/1/gate/openstack-tox-docs/54097b0/html/ : SUCCESS in 4m 34s\n- build-openstack-releasenotes http://logs.openstack.org/58/668258/1/gate/build-openstack-releasenotes/91d7141/html/ : SUCCESS in 6m 34s\n- openstack-tox-cover http://logs.openstack.org/58/668258/1/gate/openstack-tox-cover/1cfccf8/cover/ : SUCCESS in 6m 52s\n- openstack-tox-functional http://logs.openstack.org/58/668258/1/gate/openstack-tox-functional/9ed1795/ : SUCCESS in 5m 36s\n- openstack-tox-functional-py35 http://logs.openstack.org/58/668258/1/gate/openstack-tox-functional-py35/a0ecad8/ : SUCCESS in 5m 18s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/58/668258/1/gate/neutron-lbaasv2-dsvm-api/1305d9d/ : SUCCESS in 1h 24m 06s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/58/668258/1/gate/neutron-lbaasv2-dsvm-py3x-api/2f9702a/ : SUCCESS in 1h 16m 29s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/58/668258/1/gate/octavia-v1-dsvm-scenario/ec327d0/ : SUCCESS in 56m 14s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/58/668258/1/gate/octavia-v1-dsvm-py3x-scenario/1fcb058/ : SUCCESS in 54m 03s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/58/668258/1/gate/octavia-v1-dsvm-scenario-multinode/cafe059/ : SUCCESS in 1h 10m 52s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/58/668258/1/gate/octavia-v1-dsvm-py3x-scenario-multinode/a89270b/ : SUCCESS in 1h 19m 28s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/58/668258/1/gate/octavia-v2-dsvm-noop-api/15eedf6/ : SUCCESS in 43m 29s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/58/668258/1/gate/octavia-v2-dsvm-noop-py35-api/5a4b899/ : SUCCESS in 39m 48s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/58/668258/1/gate/octavia-v2-dsvm-scenario/3ae1556/ : SUCCESS in 1h 21m 55s\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/58/668258/1/gate/octavia-v2-dsvm-py35-scenario/33fd475/ : SUCCESS in 1h 20m 55s","accounts_in_message":[],"_revision_number":1},{"id":"96ef2d2c88e0c3cecb3bdb51276a1467b80b4856","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 04:42:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7eb83accd543535065c150c62219ad0567d715b2","revisions":{"7eb83accd543535065c150c62219ad0567d715b2":{"kind":"REWORK","_number":1,"created":"2019-06-28 22:50:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/58/668258/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/58/668258/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/668258/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/668258/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/668258/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/58/668258/1"}}},"commit":{"parents":[{"commit":"26d8fde69fca888e90f7ea3292f3f1d34d7c002b","subject":"Validate certificate content at API level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26d8fde69fca888e90f7ea3292f3f1d34d7c002b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-28 22:50:31.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-28 22:50:48.000000000","tz":-420},"subject":"Update tox.ini for new upper constraints strategy","message":"Update tox.ini for new upper constraints strategy\n\nThe requirements team has defined a new strategy for handling\nupper constraints[1]. This patch applies those strategies to\nOctavia.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/ \\\n    2019-May/006478.html\n\nChange-Id: I3db53821520e54747697808c7eb6988cc1556513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7eb83accd543535065c150c62219ad0567d715b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7eb83accd543535065c150c62219ad0567d715b2"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
