)]}'
{"id":"openstack%2Foctavia~668617","triplet_id":"openstack%2Foctavia~master~If6fb3fde9b43ac82af46eaddc48ec7a3a5b95602","project":"openstack/octavia","branch":"master","topic":"udp_mixed_ipv4_ipv6","hashtags":[],"change_id":"If6fb3fde9b43ac82af46eaddc48ec7a3a5b95602","subject":"Prevent UDP LBs to use different IP protocol versions in amphora driver","status":"MERGED","created":"2019-07-02 07:01:24.000000000","updated":"2019-08-07 19:53:00.000000000","submitted":"2019-08-07 19:52:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668617-1565207580117-ca213fc5","meta_rev_id":"b8b8da45f4b8afddb028c963774273691227e847","_number":668617,"virtual_id_number":668617,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-08-07 19:52:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-08-03 00:23:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-08-07 18:03:44.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-08-07 18:03:44.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-08-07 18:03:44.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 15:04:46.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-08-03 00:23:21.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-08-07 18:03:44.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-08-07 19:52:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b7f19195de319ced6f5b5ae9c68a83341654ec20","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-07-02 07:01:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d741f39362da0e257d0a29d4e3ec7e54c3859ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 09:26:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/17/668617/1/check/openstack-tox-cover/09339d3/ : FAILURE in 5m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/668617/1/check/openstack-tox-lower-constraints/a98e49a/ : FAILURE in 4m 21s\n- openstack-tox-pep8 http://logs.openstack.org/17/668617/1/check/openstack-tox-pep8/9d3cd11/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/17/668617/1/check/openstack-tox-py27/015bf30/ : FAILURE in 4m 03s\n- openstack-tox-py36 http://logs.openstack.org/17/668617/1/check/openstack-tox-py36/3131ad8/ : FAILURE in 7m 06s\n- openstack-tox-py37 http://logs.openstack.org/17/668617/1/check/openstack-tox-py37/bdbcd0e/ : FAILURE in 4m 10s\n- openstack-tox-docs http://logs.openstack.org/17/668617/1/check/openstack-tox-docs/4f215ac/html/ : SUCCESS in 5m 13s\n- build-openstack-releasenotes http://logs.openstack.org/17/668617/1/check/build-openstack-releasenotes/9c13704/html/ : SUCCESS in 5m 08s\n- openstack-tox-functional http://logs.openstack.org/17/668617/1/check/openstack-tox-functional/0e7551a/ : SUCCESS in 4m 53s\n- openstack-tox-functional-py36 http://logs.openstack.org/17/668617/1/check/openstack-tox-functional-py36/9fb5968/ : SUCCESS in 5m 13s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/17/668617/1/check/octavia-v2-dsvm-noop-api/ecc5859/ : SUCCESS in 43m 44s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/17/668617/1/check/octavia-v2-dsvm-noop-py2-api/c37f1a5/ : SUCCESS in 46m 33s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/17/668617/1/check/octavia-v2-dsvm-scenario/213cbc8/ : SUCCESS in 1h 27m 57s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/17/668617/1/check/octavia-v2-dsvm-py2-scenario/486b675/ : SUCCESS in 1h 25m 14s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/17/668617/1/check/octavia-v2-dsvm-py2-scenario-centos-7/2412b21/ : FAILURE in 2h 20m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/17/668617/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/f1658a9/ : SUCCESS in 1h 48m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/17/668617/1/check/octavia-v2-dsvm-scenario-two-node/04a343b/ : SUCCESS in 1h 47m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/17/668617/1/check/octavia-v2-dsvm-py2-scenario-two-node/725da2c/ : SUCCESS in 2h 06m 31s (non-voting)\n- octavia-grenade http://logs.openstack.org/17/668617/1/check/octavia-grenade/da3bd7d/ : SUCCESS in 1h 42m 11s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/17/668617/1/check/octavia-v2-dsvm-tls-barbican/835c510/ : SUCCESS in 1h 20m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/17/668617/1/check/octavia-v2-dsvm-scenario-amphora-v2/36ba9ce/ : SUCCESS in 1h 21m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f909bd68894e86d8104243725cea0fa3adfa9ed0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-07-02 09:50:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c51aede4a9724554386a31cfc0f03b05269ed961","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 12:04:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/17/668617/2/check/openstack-tox-cover/1bda842/cover/ : SUCCESS in 6m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/668617/2/check/openstack-tox-lower-constraints/5bd3fb7/ : SUCCESS in 7m 11s\n- openstack-tox-pep8 http://logs.openstack.org/17/668617/2/check/openstack-tox-pep8/7eea39c/ : SUCCESS in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/17/668617/2/check/openstack-tox-py27/1d4d24c/ : SUCCESS in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/17/668617/2/check/openstack-tox-py36/6e889f5/ : SUCCESS in 3m 52s\n- openstack-tox-py37 http://logs.openstack.org/17/668617/2/check/openstack-tox-py37/ebe1e2c/ : SUCCESS in 4m 49s\n- openstack-tox-docs http://logs.openstack.org/17/668617/2/check/openstack-tox-docs/7372b1f/html/ : SUCCESS in 5m 32s\n- build-openstack-releasenotes http://logs.openstack.org/17/668617/2/check/build-openstack-releasenotes/e0a06a3/html/ : SUCCESS in 3m 53s\n- openstack-tox-functional http://logs.openstack.org/17/668617/2/check/openstack-tox-functional/c0468a2/ : SUCCESS in 4m 48s\n- openstack-tox-functional-py36 http://logs.openstack.org/17/668617/2/check/openstack-tox-functional-py36/f107372/ : SUCCESS in 5m 45s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/17/668617/2/check/octavia-v2-dsvm-noop-api/490d545/ : SUCCESS in 40m 48s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/17/668617/2/check/octavia-v2-dsvm-noop-py2-api/e1f2f3f/ : SUCCESS in 46m 54s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/17/668617/2/check/octavia-v2-dsvm-scenario/39bfaf8/ : SUCCESS in 1h 38m 23s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/17/668617/2/check/octavia-v2-dsvm-py2-scenario/9c49daf/ : SUCCESS in 1h 29m 29s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/17/668617/2/check/octavia-v2-dsvm-py2-scenario-centos-7/90b9470/ : FAILURE in 1h 58m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/17/668617/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/4ef7da8/ : SUCCESS in 1h 38m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/17/668617/2/check/octavia-v2-dsvm-scenario-two-node/8130c94/ : SUCCESS in 1h 49m 04s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/17/668617/2/check/octavia-v2-dsvm-py2-scenario-two-node/e3c88ba/ : SUCCESS in 1h 42m 16s (non-voting)\n- octavia-grenade http://logs.openstack.org/17/668617/2/check/octavia-grenade/9ac3958/ : SUCCESS in 1h 10m 29s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/17/668617/2/check/octavia-v2-dsvm-tls-barbican/81a1c78/ : SUCCESS in 50m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/17/668617/2/check/octavia-v2-dsvm-scenario-amphora-v2/fd84400/ : SUCCESS in 1h 18m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"28aba6117b10577eb92fdd76ab810054f810975e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-02 14:58:14.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review-1\n\n(3 comments)\n\nLooks good, just needs better unit test coverage and a release notes update.","accounts_in_message":[],"_revision_number":2},{"id":"93e584ace9858f015d46c3097aedd48e99be0135","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-07-03 06:52:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"eef05bd577a33b24ab6e3a66fd21b679ce5204f4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-07-03 06:53:08.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c159b96fd0e4180cedb5c10391038f517917b544","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 08:53:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/17/668617/3/check/openstack-tox-cover/0cde7e5/cover/ : SUCCESS in 6m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/668617/3/check/openstack-tox-lower-constraints/85d7b4a/ : SUCCESS in 9m 15s\n- openstack-tox-pep8 http://logs.openstack.org/17/668617/3/check/openstack-tox-pep8/1daa3b6/ : SUCCESS in 7m 39s\n- openstack-tox-py27 http://logs.openstack.org/17/668617/3/check/openstack-tox-py27/f5d201f/ : SUCCESS in 4m 41s\n- openstack-tox-py36 http://logs.openstack.org/17/668617/3/check/openstack-tox-py36/f4acae8/ : SUCCESS in 4m 01s\n- openstack-tox-py37 http://logs.openstack.org/17/668617/3/check/openstack-tox-py37/19afa72/ : SUCCESS in 4m 34s\n- openstack-tox-docs http://logs.openstack.org/17/668617/3/check/openstack-tox-docs/95439af/html/ : SUCCESS in 8m 03s\n- build-openstack-releasenotes http://logs.openstack.org/17/668617/3/check/build-openstack-releasenotes/30e0b47/html/ : SUCCESS in 5m 04s\n- openstack-tox-functional http://logs.openstack.org/17/668617/3/check/openstack-tox-functional/0cf80d7/ : SUCCESS in 4m 57s\n- openstack-tox-functional-py36 http://logs.openstack.org/17/668617/3/check/openstack-tox-functional-py36/cdf7eb6/ : SUCCESS in 5m 26s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/17/668617/3/check/octavia-v2-dsvm-noop-api/f8ff786/ : SUCCESS in 41m 17s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/17/668617/3/check/octavia-v2-dsvm-noop-py2-api/de33982/ : SUCCESS in 42m 20s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/17/668617/3/check/octavia-v2-dsvm-scenario/98741e2/ : SUCCESS in 1h 36m 27s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/17/668617/3/check/octavia-v2-dsvm-py2-scenario/4cd0c82/ : SUCCESS in 1h 21m 37s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/17/668617/3/check/octavia-v2-dsvm-py2-scenario-centos-7/d07dc37/ : FAILURE in 1h 56m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/17/668617/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/36d5311/ : SUCCESS in 1h 36m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/17/668617/3/check/octavia-v2-dsvm-scenario-two-node/853ac84/ : SUCCESS in 1h 48m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/17/668617/3/check/octavia-v2-dsvm-py2-scenario-two-node/a1e970a/ : SUCCESS in 1h 32m 35s (non-voting)\n- octavia-grenade http://logs.openstack.org/17/668617/3/check/octavia-grenade/7f5b979/ : SUCCESS in 1h 29m 39s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/17/668617/3/check/octavia-v2-dsvm-tls-barbican/d9f5c28/ : SUCCESS in 56m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/17/668617/3/check/octavia-v2-dsvm-scenario-amphora-v2/1714f87/ : SUCCESS in 1h 14m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c42362a8b52dab8a327ef2a567d8024ada32d466","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-03 15:04:46.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"ae22fb6d868ef3d7b0bbe1d83cc814ec7e785759","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-05 16:22:25.000000000","message":"Patch Set 3: Code-Review-1\n\nI\u0027m dubious that this is a complete enough solution. This is one way for this to happen, but it\u0027s a significantly more complex issue than this patch makes it seem. You can still add a pool with existing mixed members to a new listener, and you can still batch-update a list of mixed members. IMO merging this just slightly confuses the situation without actually resolving it.","accounts_in_message":[],"_revision_number":3},{"id":"2a3369a87313ea59a67ed94072dc8f44aea17fb2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-07-08 08:16:23.000000000","message":"Patch Set 3: Workflow-1\n\n\u003e I\u0027m dubious that this is a complete enough solution. This is one\n \u003e way for this to happen, but it\u0027s a significantly more complex issue\n \u003e than this patch makes it seem. You can still add a pool with\n \u003e existing mixed members to a new listener, and you can still\n \u003e batch-update a list of mixed members. IMO merging this just\n \u003e slightly confuses the situation without actually resolving it.\n\nI agree for batch-update of members, but I don\u0027t find any other missing cases:\n- a pool is created with UDP protocol, it cannot have mixed members\n- an existing UDP pool might be added to a new listener but the listener have to be part of the same LB as the existing listener (so it uses the same subnet - for now), so we cannot mix IPv4 and IPv6","accounts_in_message":[],"_revision_number":3},{"id":"670eae997415fb0af2c14265870f0e41e4b7915b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-07-08 11:57:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1fc0306128bce5a3ccaa0f117c896216cf0b7886","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 16:24:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/17/668617/4/check/openstack-tox-cover/bca35f9/cover/ : SUCCESS in 6m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/668617/4/check/openstack-tox-lower-constraints/df5c2a0/ : SUCCESS in 6m 46s\n- openstack-tox-pep8 http://logs.openstack.org/17/668617/4/check/openstack-tox-pep8/6c7ced6/ : SUCCESS in 7m 14s\n- openstack-tox-py27 http://logs.openstack.org/17/668617/4/check/openstack-tox-py27/3449cc5/ : SUCCESS in 6m 04s\n- openstack-tox-py36 http://logs.openstack.org/17/668617/4/check/openstack-tox-py36/24e2b29/ : SUCCESS in 7m 30s\n- openstack-tox-py37 http://logs.openstack.org/17/668617/4/check/openstack-tox-py37/71482c7/ : SUCCESS in 4m 40s\n- openstack-tox-docs http://logs.openstack.org/17/668617/4/check/openstack-tox-docs/e34c57b/html/ : SUCCESS in 7m 04s\n- build-openstack-releasenotes http://logs.openstack.org/17/668617/4/check/build-openstack-releasenotes/871c27d/html/ : SUCCESS in 6m 14s\n- openstack-tox-functional http://logs.openstack.org/17/668617/4/check/openstack-tox-functional/b891abd/ : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 http://logs.openstack.org/17/668617/4/check/openstack-tox-functional-py36/cbb2318/ : SUCCESS in 7m 46s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/17/668617/4/check/octavia-v2-dsvm-noop-api/2dc27e3/ : SUCCESS in 46m 27s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/17/668617/4/check/octavia-v2-dsvm-noop-py2-api/14a8f11/ : SUCCESS in 44m 10s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/17/668617/4/check/octavia-v2-dsvm-scenario/53bf0c2/ : SUCCESS in 1h 19m 43s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/17/668617/4/check/octavia-v2-dsvm-py2-scenario/f089a1b/ : SUCCESS in 1h 34m 47s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/17/668617/4/check/octavia-v2-dsvm-py2-scenario-centos-7/53b7c3f/ : FAILURE in 1h 54m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/17/668617/4/check/octavia-v2-dsvm-scenario-ubuntu-bionic/9b65293/ : SUCCESS in 1h 36m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/17/668617/4/check/octavia-v2-dsvm-scenario-two-node/daa0960/ : SUCCESS in 1h 41m 04s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/17/668617/4/check/octavia-v2-dsvm-py2-scenario-two-node/5286fcb/ : SUCCESS in 1h 39m 30s (non-voting)\n- octavia-grenade http://logs.openstack.org/17/668617/4/check/octavia-grenade/72cf01f/ : SUCCESS in 1h 30m 02s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/17/668617/4/check/octavia-v2-dsvm-tls-barbican/d78724b/ : SUCCESS in 1h 11m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/17/668617/4/check/octavia-v2-dsvm-scenario-amphora-v2/f1234e1/ : SUCCESS in 1h 37m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e2d49c086fbce4db51a7252cc949809dee90125d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 00:23:21.000000000","message":"Patch Set 4: Code-Review+2\n\nOK, well, this is definitely better than what we have now. If there\u0027s other cases that still bypass this validation, we can fix them when we see them. :)","accounts_in_message":[],"_revision_number":4},{"id":"42e018f9ff63b73782099dad262d3eca54176cc7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-07 18:03:44.000000000","message":"Patch Set 4: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f5435730843a62bf0cc7a21acb5a09e732bb74e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 18:03:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ed37a86212dc122492a918e4a22069d8e5165432","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 19:52:59.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/17/668617/4/gate/openstack-tox-lower-constraints/19d65cb/ : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://logs.opendev.org/17/668617/4/gate/openstack-tox-pep8/589bd4d/ : SUCCESS in 6m 47s\n- openstack-tox-py27 https://logs.opendev.org/17/668617/4/gate/openstack-tox-py27/50b8463/ : SUCCESS in 5m 59s\n- openstack-tox-py36 https://logs.opendev.org/17/668617/4/gate/openstack-tox-py36/d5bd9db/ : SUCCESS in 5m 20s\n- openstack-tox-py37 https://logs.opendev.org/17/668617/4/gate/openstack-tox-py37/2fd418a/ : SUCCESS in 4m 52s\n- openstack-tox-docs https://logs.opendev.org/17/668617/4/gate/openstack-tox-docs/be9bf3b/html/ : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://logs.opendev.org/17/668617/4/gate/build-openstack-releasenotes/6912d58/html/ : SUCCESS in 4m 38s\n- openstack-tox-functional https://logs.opendev.org/17/668617/4/gate/openstack-tox-functional/4dc9219/ : SUCCESS in 7m 03s\n- openstack-tox-functional-py36 https://logs.opendev.org/17/668617/4/gate/openstack-tox-functional-py36/3da4ad9/ : SUCCESS in 6m 35s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/17/668617/4/gate/octavia-v2-dsvm-noop-api/bc6f777/ : SUCCESS in 53m 22s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/17/668617/4/gate/octavia-v2-dsvm-noop-py2-api/80f02c2/ : SUCCESS in 44m 36s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/17/668617/4/gate/octavia-v2-dsvm-scenario/ba37daf/ : SUCCESS in 1h 38m 26s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/17/668617/4/gate/octavia-v2-dsvm-py2-scenario/4250748/ : SUCCESS in 1h 44m 57s\n- octavia-grenade https://logs.opendev.org/17/668617/4/gate/octavia-grenade/ba29495/ : SUCCESS in 1h 32m 20s","accounts_in_message":[],"_revision_number":4},{"id":"3c62b9ff5caf3d85d9fb48fc1fd2804e1100e032","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 19:53:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"56bb1e134d670901bdfe0526495f5794a55f139a","revisions":{"f833c5be228fb13c305edde1e6b0a6b5b3cdc082":{"kind":"REWORK","_number":1,"created":"2019-07-02 07:01:24.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/17/668617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/668617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/668617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/668617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/668617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/668617/1"}}},"commit":{"parents":[{"commit":"de30dbb62f98e89219231fbc282083f45417774c","subject":"Update tox.ini for new upper constraints strategy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de30dbb62f98e89219231fbc282083f45417774c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-07-01 14:57:14.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-07-02 06:44:41.000000000","tz":120},"subject":"Prevent UDP LBs to use different IP protocol versions in amphora driver","message":"Prevent UDP LBs to use different IP protocol versions in amphora driver\n\nThe amphora doesn\u0027t support mixing IPv4 and IPv6 addresses for its\nmembers and its VIP when using UDP load balancers.\nThis commit adds a validation step to ensure that IP protocol versions\nare the same in a UDP LB in the amphora driver.\n\nStory: 2005876\nTask: 33689\n\nChange-Id: If6fb3fde9b43ac82af46eaddc48ec7a3a5b95602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f833c5be228fb13c305edde1e6b0a6b5b3cdc082"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f833c5be228fb13c305edde1e6b0a6b5b3cdc082"}]},"branch":"refs/heads/master"},"17c478c07fc816f8d795864b3b57c18efafb6945":{"kind":"REWORK","_number":2,"created":"2019-07-02 09:50:44.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/17/668617/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/668617/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/668617/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/668617/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/668617/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/668617/2"}}},"commit":{"parents":[{"commit":"de30dbb62f98e89219231fbc282083f45417774c","subject":"Update tox.ini for new upper constraints strategy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de30dbb62f98e89219231fbc282083f45417774c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-07-01 14:57:14.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-07-02 09:50:27.000000000","tz":120},"subject":"Prevent UDP LBs to use different IP protocol versions in amphora driver","message":"Prevent UDP LBs to use different IP protocol versions in amphora driver\n\nThe amphora doesn\u0027t support mixing IPv4 and IPv6 addresses for its\nmembers and its VIP when using UDP load balancers.\nThis commit adds a validation step to ensure that IP protocol versions\nare the same in a UDP LB in the amphora driver.\n\nStory: 2005876\nTask: 33689\n\nChange-Id: If6fb3fde9b43ac82af46eaddc48ec7a3a5b95602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/17c478c07fc816f8d795864b3b57c18efafb6945"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/17c478c07fc816f8d795864b3b57c18efafb6945"}]},"branch":"refs/heads/master"},"b12d1b6d9288fa1b2167b671269a98b59404bbbf":{"kind":"REWORK","_number":3,"created":"2019-07-03 06:52:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/17/668617/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/668617/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/668617/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/668617/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/668617/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/668617/3"}}},"commit":{"parents":[{"commit":"de30dbb62f98e89219231fbc282083f45417774c","subject":"Update tox.ini for new upper constraints strategy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de30dbb62f98e89219231fbc282083f45417774c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-07-01 14:57:14.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-07-03 06:49:26.000000000","tz":120},"subject":"Prevent UDP LBs to use different IP protocol versions in amphora driver","message":"Prevent UDP LBs to use different IP protocol versions in amphora driver\n\nThe amphora doesn\u0027t support mixing IPv4 and IPv6 addresses for its\nmembers and its VIP when using UDP load balancers.\nThis commit adds a validation step to ensure that IP protocol versions\nare the same in a UDP LB in the amphora driver.\n\nStory: 2005876\nTask: 33689\n\nChange-Id: If6fb3fde9b43ac82af46eaddc48ec7a3a5b95602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b12d1b6d9288fa1b2167b671269a98b59404bbbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b12d1b6d9288fa1b2167b671269a98b59404bbbf"}]},"branch":"refs/heads/master"},"56bb1e134d670901bdfe0526495f5794a55f139a":{"kind":"REWORK","_number":4,"created":"2019-07-08 11:57:18.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/17/668617/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/668617/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/668617/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/668617/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/668617/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/668617/4"}}},"commit":{"parents":[{"commit":"de30dbb62f98e89219231fbc282083f45417774c","subject":"Update tox.ini for new upper constraints strategy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de30dbb62f98e89219231fbc282083f45417774c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-07-01 14:57:14.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-07-08 11:55:51.000000000","tz":120},"subject":"Prevent UDP LBs to use different IP protocol versions in amphora driver","message":"Prevent UDP LBs to use different IP protocol versions in amphora driver\n\nThe amphora doesn\u0027t support mixing IPv4 and IPv6 addresses for its\nmembers and its VIP when using UDP load balancers.\nThis commit adds a validation step in the member_create and th e\nmember_batch_update functions of the amphora driver to ensure that IP\nprotocol versions are the same in a UDP load balancer.\n\nStory: 2005876\nTask: 33689\n\nChange-Id: If6fb3fde9b43ac82af46eaddc48ec7a3a5b95602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56bb1e134d670901bdfe0526495f5794a55f139a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56bb1e134d670901bdfe0526495f5794a55f139a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
